)]}'
{"id":"openstack%2Fneutron~667511","triplet_id":"openstack%2Fneutron~master~I33bd17c723b3e8d409415bda008440f8ed9cfa68","project":"openstack/neutron","branch":"master","topic":"bp/subnets-different-pools-same-net","hashtags":[],"change_id":"I33bd17c723b3e8d409415bda008440f8ed9cfa68","subject":"Relax subnet pool network affinity constraints","status":"MERGED","created":"2019-06-26 05:34:52.000000000","updated":"2019-08-02 01:28:09.000000000","submitted":"2019-08-02 01:28:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":63,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667511-1564709289283-ede0708f","meta_rev_id":"dbf68cf8684af58f0be184b9bd417ed3eae518bb","_number":667511,"virtual_id_number":667511,"owner":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-01 00:06:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-07-31 19:07:19.000000000","_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-08-02 01:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"date":"2019-08-01 16:16:29.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"date":"2019-08-01 17:21:06.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-31 23:53:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-01 08:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-07-31 15:07:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-31 21:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-01 08:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 14:08:22.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-07-09 17:59:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-30 14:15:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-07-31 02:28:28.000000000","updated_by":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"reviewer":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"state":"REVIEWER"},{"updated":"2019-07-31 15:07:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-07-31 19:07:19.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-31 21:44:34.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-31 23:53:43.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-01 00:06:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 08:31:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-01 16:16:29.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 17:21:06.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-02 01:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f3f6505d86abc3d1e0897e5f27e88e5a510e4ef","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 05:34:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"890a176e59387aaa9356f304c8473b9d6f470580","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-06-26 06:51:02.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/June-2019/Arista_Tempest_Testrefs/changes/11/667511/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d94d30a3c2be56e757391b764d4888ed88d77dd4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-26 07:12:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/1 : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5567aff445cfde0b4728d7082dcff21101a2617d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-26 15:02:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3db009d0368879822f796096d72515dbd4cb1734","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 15:07:00.000000000","message":"Topic changed from bug/1830240 to bp/subnets-different-pools-same-net","accounts_in_message":[],"_revision_number":1},{"id":"ffda20bdaa2c7c2fd726ce858d086108b9041505","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:59:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/1/check/neutron-tempest-plugin-api/04f469c/ : SUCCESS in 51m 00s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/1/check/neutron-tempest-plugin-designate-scenario/be97af8/ : SUCCESS in 40m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/1/check/neutron-tempest-plugin-dvr-multinode-scenario/1f7ebbb/ : FAILURE in 1h 41m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/1/check/neutron-tempest-plugin-scenario-linuxbridge/e4d1102/ : SUCCESS in 1h 11m 26s\n- openstack-tox-cover http://logs.openstack.org/11/667511/1/check/openstack-tox-cover/7f79b05/cover/ : SUCCESS in 26m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/1/check/openstack-tox-lower-constraints/cd20398/ : SUCCESS in 23m 35s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/1/check/openstack-tox-pep8/aa121e2/ : SUCCESS in 12m 59s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/1/check/openstack-tox-py27/d02b421/ : SUCCESS in 23m 29s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/1/check/openstack-tox-py36/e00170f/ : SUCCESS in 21m 23s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/1/check/openstack-tox-py37/676acd1/ : SUCCESS in 24m 35s\n- openstack-tox-docs http://logs.openstack.org/11/667511/1/check/openstack-tox-docs/78b4e4c/html/ : SUCCESS in 4m 12s\n- tempest-full http://logs.openstack.org/11/667511/1/check/tempest-full/5771d84/ : SUCCESS in 1h 48m 48s\n- neutron-grenade http://logs.openstack.org/11/667511/1/check/neutron-grenade/0979478/ : SUCCESS in 58m 10s\n- grenade-py3 http://logs.openstack.org/11/667511/1/check/grenade-py3/66f522f/ : SUCCESS in 58m 26s\n- tempest-full-py3 http://logs.openstack.org/11/667511/1/check/tempest-full-py3/0e3a7cf/ : SUCCESS in 1h 28m 17s\n- neutron-functional http://logs.openstack.org/11/667511/1/check/neutron-functional/7723555/ : SUCCESS in 49m 22s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/1/check/neutron-functional-python27/2ba5101/ : SUCCESS in 48m 11s\n- neutron-fullstack http://logs.openstack.org/11/667511/1/check/neutron-fullstack/a8b8537/ : SUCCESS in 55m 16s\n- neutron-rally-task http://logs.openstack.org/11/667511/1/check/neutron-rally-task/3d39a50/ : SUCCESS in 2h 13m 13s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/1/check/neutron-tempest-dvr/72ab799/ : SUCCESS in 1h 35m 19s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/1/check/neutron-tempest-linuxbridge/374548f/ : SUCCESS in 1h 39m 48s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/1/check/neutron-grenade-multinode/ca1c995/ : SUCCESS in 1h 12m 26s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/1/check/neutron-grenade-dvr-multinode/6a406b9/ : SUCCESS in 1h 07m 36s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/1/check/tempest-multinode-full-py3/c079dd7/ : SUCCESS in 1h 50m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/1/check/neutron-tempest-dvr-ha-multinode-full/ee205d2/ : SUCCESS in 2h 30m 09s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/1/check/neutron-tempest-iptables_hybrid/f3d63b0/ : SUCCESS in 1h 59m 26s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/1/check/neutron-tempest-iptables_hybrid-fedora/e500c29/ : SUCCESS in 2h 04m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5154c39/ : FAILURE in 38m 16s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/1/check/tempest-slow-py3/f51d639/ : SUCCESS in 2h 59m 02s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/1/check/networking-ovn-tempest-dsvm-ovs-release/b85c6cd/ : SUCCESS in 2h 11m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/1/check/networking-midonet-tempest-aio-ml2-centos-7/8ba13a8/ : SUCCESS in 1h 49m 18s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/1/check/openstacksdk-functional-devstack-networking/29d7d67/ : SUCCESS in 1h 32m 10s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/1/check/neutron-functional-with-uwsgi/3c9a1bb/ : SUCCESS in 58m 57s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/1/check/neutron-fullstack-with-uwsgi/925b51d/ : SUCCESS in 1h 20m 28s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/1/check/neutron-tempest-with-uwsgi/c2f9737/ : TIMED_OUT in 2h 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"501bb2ee60a5bf2da6d1c450c4f0d1a5dd91bd47","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 19:06:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f81ea7bc64f04df0e2744abd67f707ed4bacc198","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 19:10:42.000000000","message":"Patch Set 2: Workflow-1\n\nNeed to round out test cases and optimize a little bit. It doesn\u0027t appear there are any regressions, but needs coverage around address scope updates and cases where subnets from different pools are on the same network.","accounts_in_message":[],"_revision_number":2},{"id":"ac0ad536daddb38587022096f45e4eb89cdb159c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-26 20:49:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/2 : SUCCESS in 1h 43m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5e58ce84d06d67ddf44bd5e9913b298d2f0f6e10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 05:42:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/2/check/neutron-tempest-plugin-api/369492e/ : SUCCESS in 53m 07s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/2/check/neutron-tempest-plugin-designate-scenario/9c96487/ : SUCCESS in 36m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/2/check/neutron-tempest-plugin-dvr-multinode-scenario/864d169/ : FAILURE in 1h 56m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/2/check/neutron-tempest-plugin-scenario-linuxbridge/b5a5b9d/ : FAILURE in 1h 15m 05s\n- openstack-tox-cover http://logs.openstack.org/11/667511/2/check/openstack-tox-cover/0def618/cover/ : SUCCESS in 35m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/2/check/openstack-tox-lower-constraints/1cd0b0f/ : SUCCESS in 22m 17s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/2/check/openstack-tox-pep8/f3bfcda/ : SUCCESS in 11m 50s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/2/check/openstack-tox-py27/c24561a/ : SUCCESS in 20m 16s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/2/check/openstack-tox-py36/a044acc/ : SUCCESS in 23m 44s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/2/check/openstack-tox-py37/5babbc0/ : SUCCESS in 21m 43s\n- openstack-tox-docs http://logs.openstack.org/11/667511/2/check/openstack-tox-docs/94babf0/html/ : SUCCESS in 4m 41s\n- tempest-full http://logs.openstack.org/11/667511/2/check/tempest-full/01b003c/ : SUCCESS in 1h 41m 24s\n- neutron-grenade http://logs.openstack.org/11/667511/2/check/neutron-grenade/e4105e2/ : SUCCESS in 59m 20s\n- grenade-py3 http://logs.openstack.org/11/667511/2/check/grenade-py3/dd09659/ : SUCCESS in 1h 01m 27s\n- tempest-full-py3 http://logs.openstack.org/11/667511/2/check/tempest-full-py3/a0cab1a/ : SUCCESS in 1h 46m 39s\n- neutron-functional http://logs.openstack.org/11/667511/2/check/neutron-functional/813fdeb/ : SUCCESS in 42m 18s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/2/check/neutron-functional-python27/d0b186c/ : SUCCESS in 43m 58s\n- neutron-fullstack http://logs.openstack.org/11/667511/2/check/neutron-fullstack/dda30ff/ : SUCCESS in 52m 56s\n- neutron-rally-task http://logs.openstack.org/11/667511/2/check/neutron-rally-task/0c79771/ : SUCCESS in 1h 49m 54s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/2/check/neutron-tempest-dvr/2960553/ : SUCCESS in 1h 35m 54s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/2/check/neutron-tempest-linuxbridge/e6c1498/ : SUCCESS in 1h 28m 38s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/2/check/neutron-grenade-multinode/b9af8f0/ : SUCCESS in 1h 04m 15s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/2/check/neutron-grenade-dvr-multinode/d864726/ : SUCCESS in 1h 09m 45s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/2/check/tempest-multinode-full-py3/a948d0f/ : SUCCESS in 1h 49m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/2/check/neutron-tempest-dvr-ha-multinode-full/4c14bf4/ : SUCCESS in 2h 03m 52s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/2/check/neutron-tempest-iptables_hybrid/d2c1c3e/ : SUCCESS in 2h 05m 37s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/2/check/neutron-tempest-iptables_hybrid-fedora/a897f47/ : SUCCESS in 1h 58m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c350762/ : SUCCESS in 1h 07m 19s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/2/check/tempest-slow-py3/45d5e7f/ : SUCCESS in 2h 29m 32s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/2/check/networking-ovn-tempest-dsvm-ovs-release/f2ce08a/ : SUCCESS in 2h 24m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/2/check/networking-midonet-tempest-aio-ml2-centos-7/c4915f2/ : SUCCESS in 1h 28m 56s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/2/check/openstacksdk-functional-devstack-networking/fa631d7/ : SUCCESS in 1h 27m 11s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/2/check/neutron-functional-with-uwsgi/10f8bbd/ : SUCCESS in 40m 21s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/2/check/neutron-fullstack-with-uwsgi/15d8682/ : SUCCESS in 53m 50s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/2/check/neutron-tempest-with-uwsgi/71c81a2/ : FAILURE in 1h 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6df939cb6cb4eac490262ef15aaa93d002c1d02","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-27 09:39:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d6bb441c107fe108f70b6aca148ac01764dca38","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-27 14:45:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e5616274ea5836fbaf53260572cb22502955f2d","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-27 16:25:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d69d07ce72f360e3451a2230a0f488c7f3728fe","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-27 18:58:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efa99271c6159eb3490209f934b87359f9c6fd48","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-06-27 20:22:15.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/June-2019/Arista_Tempest_Testrefs/changes/11/667511/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a3672fbaaaf606d6ad741d9d84ff323cfc40de8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-28 02:35:48.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667511/3 : FAILURE in 13m 52s","accounts_in_message":[],"_revision_number":3},{"id":"498880ec0ae7a87b03e1cb4a71bb4fb4bdbd4b75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 03:00:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/3/check/neutron-tempest-plugin-api/e0aa99a/ : SUCCESS in 56m 26s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/3/check/neutron-tempest-plugin-designate-scenario/90fd31f/ : SUCCESS in 31m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/3/check/neutron-tempest-plugin-dvr-multinode-scenario/1a28596/ : FAILURE in 2h 01m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/3/check/neutron-tempest-plugin-scenario-linuxbridge/c3d989f/ : SUCCESS in 1h 14m 12s\n- openstack-tox-cover http://logs.openstack.org/11/667511/3/check/openstack-tox-cover/286394b/cover/ : SUCCESS in 29m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/3/check/openstack-tox-lower-constraints/dc4b455/ : SUCCESS in 25m 52s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/3/check/openstack-tox-pep8/f0b28f1/ : SUCCESS in 12m 41s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/3/check/openstack-tox-py27/07d9c68/ : SUCCESS in 26m 05s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/3/check/openstack-tox-py36/a91c376/ : SUCCESS in 23m 54s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/3/check/openstack-tox-py37/d04fc9a/ : SUCCESS in 22m 28s\n- openstack-tox-docs http://logs.openstack.org/11/667511/3/check/openstack-tox-docs/e26743c/html/ : SUCCESS in 4m 32s\n- tempest-full http://logs.openstack.org/11/667511/3/check/tempest-full/5543ae1/ : SUCCESS in 2h 14m 14s\n- neutron-grenade http://logs.openstack.org/11/667511/3/check/neutron-grenade/2af407d/ : SUCCESS in 57m 29s\n- grenade-py3 http://logs.openstack.org/11/667511/3/check/grenade-py3/f83b9c8/ : SUCCESS in 59m 06s\n- tempest-full-py3 http://logs.openstack.org/11/667511/3/check/tempest-full-py3/50f7e60/ : SUCCESS in 1h 41m 25s\n- neutron-functional http://logs.openstack.org/11/667511/3/check/neutron-functional/b66975e/ : SUCCESS in 42m 40s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/3/check/neutron-functional-python27/e9f6955/ : SUCCESS in 44m 58s\n- neutron-fullstack http://logs.openstack.org/11/667511/3/check/neutron-fullstack/366b28e/ : SUCCESS in 56m 43s\n- neutron-rally-task http://logs.openstack.org/11/667511/3/check/neutron-rally-task/7ee6a14/ : SUCCESS in 1h 49m 36s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/3/check/neutron-tempest-dvr/b50b54a/ : POST_FAILURE in 1h 28m 05s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/3/check/neutron-tempest-linuxbridge/228780c/ : SUCCESS in 1h 23m 58s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/3/check/neutron-grenade-multinode/a2ab7ff/ : SUCCESS in 1h 08m 34s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/3/check/neutron-grenade-dvr-multinode/14e0c25/ : SUCCESS in 1h 09m 41s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/3/check/tempest-multinode-full-py3/1036427/ : SUCCESS in 2h 08m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/3/check/neutron-tempest-dvr-ha-multinode-full/a0b5d66/ : SUCCESS in 2h 02m 00s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/3/check/neutron-tempest-iptables_hybrid/3347ca8/ : SUCCESS in 1h 49m 58s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/3/check/neutron-tempest-iptables_hybrid-fedora/d80cdef/ : SUCCESS in 1h 47m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d1e606d/ : SUCCESS in 50m 01s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/3/check/tempest-slow-py3/26f46a0/ : SUCCESS in 2h 25m 20s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/3/check/networking-ovn-tempest-dsvm-ovs-release/94878fb/ : FAILURE in 2h 23m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/3/check/networking-midonet-tempest-aio-ml2-centos-7/4eafc59/ : SUCCESS in 1h 31m 15s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/3/check/openstacksdk-functional-devstack-networking/f755e86/ : SUCCESS in 1h 30m 37s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/3/check/neutron-functional-with-uwsgi/7f909ee/ : SUCCESS in 46m 55s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/3/check/neutron-fullstack-with-uwsgi/87f76b7/ : SUCCESS in 53m 29s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/3/check/neutron-tempest-with-uwsgi/1769763/ : TIMED_OUT in 2h 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13562960ac5b428260f7d77c37f463bdf6d71840","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-30 19:34:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"472dbeffa8ef35959429881005da50ed2359dca4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-06-30 20:43:42.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/June-2019/Arista_Tempest_Testrefs/changes/11/667511/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f68c8f27d1c4ff4ef9c23fa109411b9336dc9077","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-06-30 21:09:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/4 : SUCCESS in 1h 34m 19s","accounts_in_message":[],"_revision_number":4},{"id":"dea0e3dedd17063362469e9f82452139698c9813","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 22:22:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/4/check/neutron-tempest-plugin-api/581e4cb/ : SUCCESS in 52m 20s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/4/check/neutron-tempest-plugin-designate-scenario/9cbc7fc/ : SUCCESS in 31m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/4/check/neutron-tempest-plugin-dvr-multinode-scenario/6439582/ : FAILURE in 1h 47m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/4/check/neutron-tempest-plugin-scenario-linuxbridge/7ca8bce/ : SUCCESS in 1h 00m 26s\n- openstack-tox-cover http://logs.openstack.org/11/667511/4/check/openstack-tox-cover/7e6cc0e/cover/ : SUCCESS in 27m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/4/check/openstack-tox-lower-constraints/be24076/ : SUCCESS in 25m 14s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/4/check/openstack-tox-pep8/021f7c0/ : SUCCESS in 11m 42s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/4/check/openstack-tox-py27/90d834e/ : SUCCESS in 22m 41s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/4/check/openstack-tox-py36/ca613b4/ : SUCCESS in 24m 02s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/4/check/openstack-tox-py37/4ff281d/ : SUCCESS in 22m 14s\n- openstack-tox-docs http://logs.openstack.org/11/667511/4/check/openstack-tox-docs/1c0036c/html/ : SUCCESS in 4m 42s\n- tempest-full http://logs.openstack.org/11/667511/4/check/tempest-full/c6eda0f/ : SUCCESS in 1h 43m 54s\n- neutron-grenade http://logs.openstack.org/11/667511/4/check/neutron-grenade/129e86a/ : SUCCESS in 55m 24s\n- grenade-py3 http://logs.openstack.org/11/667511/4/check/grenade-py3/9a3d9ac/ : SUCCESS in 52m 13s\n- tempest-full-py3 http://logs.openstack.org/11/667511/4/check/tempest-full-py3/1843375/ : SUCCESS in 1h 23m 25s\n- neutron-functional http://logs.openstack.org/11/667511/4/check/neutron-functional/6392dab/ : SUCCESS in 39m 49s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/4/check/neutron-functional-python27/bedb9f5/ : SUCCESS in 49m 22s\n- neutron-fullstack http://logs.openstack.org/11/667511/4/check/neutron-fullstack/22f4ae3/ : SUCCESS in 52m 22s\n- neutron-rally-task http://logs.openstack.org/11/667511/4/check/neutron-rally-task/9591bdd/ : SUCCESS in 1h 53m 54s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/4/check/neutron-tempest-dvr/d53ac6c/ : SUCCESS in 1h 34m 21s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/4/check/neutron-tempest-linuxbridge/4f51171/ : SUCCESS in 1h 26m 45s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/4/check/neutron-grenade-multinode/68e42c3/ : SUCCESS in 1h 02m 21s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/4/check/neutron-grenade-dvr-multinode/8ebb9e3/ : SUCCESS in 1h 02m 57s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/4/check/tempest-multinode-full-py3/3b1be98/ : SUCCESS in 1h 48m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/4/check/neutron-tempest-dvr-ha-multinode-full/d3ba336/ : SUCCESS in 2h 02m 42s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/4/check/neutron-tempest-iptables_hybrid/134240b/ : SUCCESS in 1h 36m 21s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/4/check/neutron-tempest-iptables_hybrid-fedora/f4ffa4e/ : SUCCESS in 1h 51m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4352385/ : SUCCESS in 46m 40s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/4/check/tempest-slow-py3/4b34c9c/ : SUCCESS in 2h 27m 40s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/4/check/networking-ovn-tempest-dsvm-ovs-release/4ab484d/ : SUCCESS in 2h 15m 01s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/4/check/networking-midonet-tempest-aio-ml2-centos-7/b94fbff/ : SUCCESS in 1h 16m 25s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/4/check/openstacksdk-functional-devstack-networking/7f76e98/ : SUCCESS in 1h 17m 59s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/4/check/neutron-functional-with-uwsgi/c339cb8/ : SUCCESS in 41m 41s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/4/check/neutron-fullstack-with-uwsgi/2dd5bf4/ : SUCCESS in 57m 28s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/4/check/neutron-tempest-with-uwsgi/83cefe3/ : FAILURE in 1h 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d96a3075c97b2b58fd39256c9b73c3f32c8b2f7","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-02 14:33:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"888cc27b2c738f15bb2d385e12de18a7f2cf8af6","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-08 14:12:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"da3ddaed88f449d8ff803017ce3bd7fcd69a1845","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-08 14:13:25.000000000","message":"Topic changed from bug/1830240 to bp/subnets-different-pools-same-net","accounts_in_message":[],"_revision_number":5},{"id":"0727a5b05e9b7b84b5b99dcbb215fdef130dbbf4","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-08 14:34:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aafe38185eb0b1a3abaea119304d6cc9cca56605","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-07-08 15:23:03.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/July-2019/Arista_Tempest_Testrefs/changes/11/667511/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5858c98f718b69353dd21dbb818986fdec5b8b36","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-08 16:12:23.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/6 : SUCCESS in 1h 37m 02s","accounts_in_message":[],"_revision_number":6},{"id":"985acab5784a71efe477b0e0a2b18b96f4c3ff5f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-08 20:15:21.000000000","message":"Patch Set 6:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/11/667511/6/check-mlnx/Neutron-ML2-Sriov/2a72a9e : FAILURE in 53m 27s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/11/667511/6/check-mlnx/Neutron-MACVTAP-ML2-Sriov/ea25dbf : SUCCESS in 1h 11m 37s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"e5956a31593a62e0ff7c63e11f120af4bc1b0121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 21:45:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/6/check/neutron-tempest-plugin-api/7f06290/ : SUCCESS in 52m 36s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/6/check/neutron-tempest-plugin-designate-scenario/620d9b9/ : SUCCESS in 33m 47s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/6/check/neutron-tempest-plugin-dvr-multinode-scenario/5eac543/ : SUCCESS in 1h 26m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/6/check/neutron-tempest-plugin-scenario-linuxbridge/a45c57f/ : SUCCESS in 56m 57s\n- openstack-tox-cover http://logs.openstack.org/11/667511/6/check/openstack-tox-cover/3b82134/cover/ : SUCCESS in 34m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/6/check/openstack-tox-lower-constraints/a18a2ad/ : SUCCESS in 23m 36s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/6/check/openstack-tox-pep8/b73ea94/ : SUCCESS in 12m 03s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/6/check/openstack-tox-py27/3fc9edb/ : SUCCESS in 22m 21s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/6/check/openstack-tox-py36/359d7da/ : SUCCESS in 23m 40s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/6/check/openstack-tox-py37/97e7d12/ : SUCCESS in 24m 52s\n- openstack-tox-docs http://logs.openstack.org/11/667511/6/check/openstack-tox-docs/404cb80/html/ : SUCCESS in 4m 39s\n- tempest-full http://logs.openstack.org/11/667511/6/check/tempest-full/b85c120/ : SUCCESS in 1h 51m 37s\n- neutron-grenade http://logs.openstack.org/11/667511/6/check/neutron-grenade/aeb5544/ : SUCCESS in 56m 37s\n- grenade-py3 http://logs.openstack.org/11/667511/6/check/grenade-py3/32162ed/ : SUCCESS in 1h 07m 16s\n- tempest-full-py3 http://logs.openstack.org/11/667511/6/check/tempest-full-py3/18da141/ : SUCCESS in 1h 32m 37s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/6/check/build-openstack-releasenotes/e44d047/html/ : SUCCESS in 18m 45s\n- neutron-functional http://logs.openstack.org/11/667511/6/check/neutron-functional/a8a5262/ : SUCCESS in 48m 00s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/6/check/neutron-functional-python27/cae251f/ : SUCCESS in 45m 05s\n- neutron-fullstack http://logs.openstack.org/11/667511/6/check/neutron-fullstack/424e3f4/ : SUCCESS in 1h 04m 52s\n- neutron-rally-task http://logs.openstack.org/11/667511/6/check/neutron-rally-task/d794ce1/ : SUCCESS in 1h 47m 04s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/6/check/neutron-tempest-dvr/2b46c99/ : SUCCESS in 1h 29m 46s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/6/check/neutron-tempest-linuxbridge/226edd2/ : SUCCESS in 1h 42m 12s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/6/check/neutron-grenade-multinode/4f2c5bd/ : SUCCESS in 1h 16m 00s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/6/check/neutron-grenade-dvr-multinode/64391f6/ : SUCCESS in 1h 11m 53s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/6/check/tempest-multinode-full-py3/e4943c0/ : SUCCESS in 1h 51m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/6/check/neutron-tempest-dvr-ha-multinode-full/8e4225c/ : SUCCESS in 2h 14m 50s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/6/check/neutron-tempest-iptables_hybrid/c6eadb9/ : SUCCESS in 1h 33m 49s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/6/check/neutron-tempest-iptables_hybrid-fedora/c98a0ee/ : SUCCESS in 1h 56m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/888b33b/ : SUCCESS in 56m 31s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/6/check/tempest-slow-py3/f9022e1/ : SUCCESS in 2h 14m 53s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/6/check/networking-ovn-tempest-dsvm-ovs-release/9c737ee/ : SUCCESS in 2h 20m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/6/check/networking-midonet-tempest-aio-ml2-centos-7/14d9c71/ : SUCCESS in 1h 36m 27s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/6/check/openstacksdk-functional-devstack-networking/63cc811/ : SUCCESS in 1h 24m 16s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/6/check/neutron-functional-with-uwsgi/862c4ef/ : SUCCESS in 42m 54s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/6/check/neutron-fullstack-with-uwsgi/d2ff74b/ : SUCCESS in 56m 44s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/6/check/neutron-tempest-with-uwsgi/780f3f2/ : TIMED_OUT in 2h 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2728b8d54a886d2e6760b7e6bd4990e034dc780a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-09 02:24:12.000000000","message":"Patch Set 6:\n\n(10 comments)","accounts_in_message":[],"_revision_number":6},{"id":"db9053631975190c284714283f1c8b88d18ba05a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-09 07:49:01.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nonly some nits. Great work Ryan :)","accounts_in_message":[],"_revision_number":6},{"id":"ee194618a49ea82fffff59213dbd28b325635982","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-09 13:14:00.000000000","message":"Patch Set 6: Workflow-1\n\n(13 comments)\n\nThanks for the good feedback, a few things to address now.","accounts_in_message":[],"_revision_number":6},{"id":"db3f29dd3efd147619398dabfc0cbb7db177f583","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-09 13:36:30.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"aa7abb62acbf96c124b333807d694a347a2122b3","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-09 15:52:32.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d689738ad582bcfaf2ad348bef9b83886eace823","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-09 16:07:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9593f56ecd666de2a072dce1b061365efcc3caf5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-09 16:53:24.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2b4ccd65243cf9a4c7bd9c643005cac8504bf57e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-09 17:49:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/7 : SUCCESS in 1h 41m 14s","accounts_in_message":[],"_revision_number":7},{"id":"5027db9c729e7699d12f37e2063537e1edde0eef","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-07-09 17:59:33.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/July-2019/Arista_Tempest_Testrefs/changes/11/667511/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"fc2617f158b59a807b366c6be3b7bc4b9c88df54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 21:09:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/7/check/neutron-tempest-plugin-api/d04048e/ : SUCCESS in 43m 51s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/7/check/neutron-tempest-plugin-designate-scenario/25c2019/ : SUCCESS in 31m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/7/check/neutron-tempest-plugin-dvr-multinode-scenario/3a2a74b/ : FAILURE in 1h 32m 20s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/7/check/neutron-tempest-plugin-scenario-linuxbridge/aae24be/ : SUCCESS in 1h 07m 40s\n- openstack-tox-cover http://logs.openstack.org/11/667511/7/check/openstack-tox-cover/6bcdc6c/cover/ : SUCCESS in 25m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/7/check/openstack-tox-lower-constraints/dfd97ea/ : SUCCESS in 23m 04s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/7/check/openstack-tox-pep8/eb04aa4/ : SUCCESS in 12m 32s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/7/check/openstack-tox-py27/46c5b57/ : SUCCESS in 24m 05s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/7/check/openstack-tox-py36/2e93b85/ : SUCCESS in 20m 25s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/7/check/openstack-tox-py37/d1010d2/ : SUCCESS in 21m 27s\n- openstack-tox-docs http://logs.openstack.org/11/667511/7/check/openstack-tox-docs/8cdd1f5/html/ : SUCCESS in 4m 24s\n- tempest-full http://logs.openstack.org/11/667511/7/check/tempest-full/66bdb05/ : SUCCESS in 1h 54m 43s\n- neutron-grenade http://logs.openstack.org/11/667511/7/check/neutron-grenade/90f11e6/ : SUCCESS in 1h 04m 46s\n- grenade-py3 http://logs.openstack.org/11/667511/7/check/grenade-py3/0771cfe/ : SUCCESS in 56m 45s\n- tempest-full-py3 http://logs.openstack.org/11/667511/7/check/tempest-full-py3/243c61c/ : SUCCESS in 1h 32m 28s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/7/check/build-openstack-releasenotes/94f672f/html/ : SUCCESS in 18m 05s\n- neutron-functional http://logs.openstack.org/11/667511/7/check/neutron-functional/c23b71e/ : SUCCESS in 46m 45s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/7/check/neutron-functional-python27/2c3ac21/ : SUCCESS in 43m 06s\n- neutron-fullstack http://logs.openstack.org/11/667511/7/check/neutron-fullstack/64de055/ : SUCCESS in 1h 00m 47s\n- neutron-rally-task http://logs.openstack.org/11/667511/7/check/neutron-rally-task/0eba648/ : SUCCESS in 2h 05m 30s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/7/check/neutron-tempest-dvr/dbc92bb/ : SUCCESS in 1h 55m 38s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/7/check/neutron-tempest-linuxbridge/b872a56/ : SUCCESS in 1h 35m 41s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/7/check/neutron-grenade-multinode/3ce7a91/ : SUCCESS in 1h 14m 45s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/7/check/neutron-grenade-dvr-multinode/b36a52c/ : SUCCESS in 1h 12m 05s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/7/check/tempest-multinode-full-py3/389944d/ : SUCCESS in 2h 02m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/7/check/neutron-tempest-dvr-ha-multinode-full/382996e/ : SUCCESS in 2h 23m 40s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/7/check/neutron-tempest-iptables_hybrid/0e1abc5/ : SUCCESS in 1h 42m 32s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/7/check/neutron-tempest-iptables_hybrid-fedora/64c649a/ : SUCCESS in 2h 04m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8e1db65/ : SUCCESS in 57m 25s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/7/check/tempest-slow-py3/b56ce69/ : SUCCESS in 2h 26m 46s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/7/check/networking-ovn-tempest-dsvm-ovs-release/4b30667/ : SUCCESS in 2h 06m 28s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/7/check/networking-midonet-tempest-aio-ml2-centos-7/b07d8c9/ : SUCCESS in 1h 35m 30s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/7/check/openstacksdk-functional-devstack-networking/6900c4b/ : SUCCESS in 1h 27m 58s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/7/check/neutron-functional-with-uwsgi/9a87dc2/ : SUCCESS in 48m 20s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/7/check/neutron-fullstack-with-uwsgi/bec2a63/ : SUCCESS in 57m 09s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/7/check/neutron-tempest-with-uwsgi/d22119e/ : TIMED_OUT in 2h 25m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"710e90d71f3b463df2d7e044950ddc310e0af650","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-10 07:44:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nonly this one nitty comment about adding comment in one place. Except that LGTM for me :)","accounts_in_message":[],"_revision_number":7},{"id":"0b2c4253dcb40147a9b672d47efee260b699bc48","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-10 16:13:28.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nI\u0027ll push a patch to update based on my comments, just trying to ping Ryan first so we don\u0027t step on each other.","accounts_in_message":[],"_revision_number":7},{"id":"a6d0ed3ed03b291bdaaf757eab93b76c61b00851","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-10 16:36:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2b9694f4d3d13792959c928c68c465055f6cb541","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-10 16:56:34.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667511/8 : FAILURE in 19m 27s","accounts_in_message":[],"_revision_number":8},{"id":"acf232b8e9a6b7a61ebd220a113cb7f8114b8fbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 22:50:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/8/check/neutron-tempest-plugin-api/371df1d/ : SUCCESS in 59m 03s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/8/check/neutron-tempest-plugin-designate-scenario/790058b/ : SUCCESS in 33m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/8/check/neutron-tempest-plugin-dvr-multinode-scenario/a86361c/ : SUCCESS in 1h 20m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/8/check/neutron-tempest-plugin-scenario-linuxbridge/63303d6/ : SUCCESS in 1h 05m 32s\n- openstack-tox-cover http://logs.openstack.org/11/667511/8/check/openstack-tox-cover/1bdae29/cover/ : SUCCESS in 24m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/8/check/openstack-tox-lower-constraints/bfdd46b/ : SUCCESS in 24m 08s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/8/check/openstack-tox-pep8/32244ff/ : SUCCESS in 12m 26s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/8/check/openstack-tox-py27/bc0987a/ : SUCCESS in 20m 41s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/8/check/openstack-tox-py36/b4d4829/ : SUCCESS in 23m 37s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/8/check/openstack-tox-py37/f59eeb5/ : SUCCESS in 22m 27s\n- openstack-tox-docs http://logs.openstack.org/11/667511/8/check/openstack-tox-docs/92c7c3d/html/ : SUCCESS in 6m 04s\n- tempest-full http://logs.openstack.org/11/667511/8/check/tempest-full/6095bbb/ : SUCCESS in 1h 48m 27s\n- neutron-grenade http://logs.openstack.org/11/667511/8/check/neutron-grenade/e5c5f03/ : SUCCESS in 1h 03m 07s\n- grenade-py3 http://logs.openstack.org/11/667511/8/check/grenade-py3/852b101/ : SUCCESS in 55m 35s\n- tempest-full-py3 http://logs.openstack.org/11/667511/8/check/tempest-full-py3/6e400a5/ : SUCCESS in 1h 34m 58s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/8/check/build-openstack-releasenotes/77e8941/html/ : SUCCESS in 20m 10s\n- neutron-functional http://logs.openstack.org/11/667511/8/check/neutron-functional/21e8071/ : SUCCESS in 50m 07s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/8/check/neutron-functional-python27/6517d6a/ : SUCCESS in 43m 55s\n- neutron-fullstack http://logs.openstack.org/11/667511/8/check/neutron-fullstack/0477132/ : SUCCESS in 52m 46s\n- neutron-rally-task http://logs.openstack.org/11/667511/8/check/neutron-rally-task/e38d6d7/ : SUCCESS in 2h 08m 55s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/8/check/neutron-tempest-dvr/d2fb307/ : SUCCESS in 1h 33m 00s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/8/check/neutron-tempest-linuxbridge/ad2adfd/ : SUCCESS in 1h 37m 43s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/8/check/neutron-grenade-multinode/1af7d03/ : SUCCESS in 1h 13m 41s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/8/check/neutron-grenade-dvr-multinode/5d80dc3/ : SUCCESS in 1h 15m 20s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/8/check/tempest-multinode-full-py3/3d94e81/ : SUCCESS in 1h 57m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/8/check/neutron-tempest-dvr-ha-multinode-full/474acd7/ : SUCCESS in 2h 10m 59s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/8/check/neutron-tempest-iptables_hybrid/091b6c9/ : SUCCESS in 1h 38m 54s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/8/check/neutron-tempest-iptables_hybrid-fedora/c63f609/ : SUCCESS in 1h 57m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9995d78/ : SUCCESS in 1h 01m 05s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/8/check/tempest-slow-py3/8a60aba/ : SUCCESS in 2h 35m 23s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/8/check/networking-ovn-tempest-dsvm-ovs-release/c80a690/ : SUCCESS in 2h 20m 06s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/8/check/networking-midonet-tempest-aio-ml2-centos-7/3a6c5e2/ : SUCCESS in 1h 43m 49s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/8/check/openstacksdk-functional-devstack-networking/828f8ec/ : SUCCESS in 1h 22m 17s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/8/check/neutron-functional-with-uwsgi/8589997/ : SUCCESS in 43m 18s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/8/check/neutron-fullstack-with-uwsgi/f5161fc/ : SUCCESS in 59m 07s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/8/check/neutron-tempest-with-uwsgi/07b516f/ : SUCCESS in 1h 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9615b43264d2d358414d82f5ce0598efe8b9a856","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-11 08:38:04.000000000","message":"Patch Set 8: Code-Review+2\n\nWould be also good to have some API tests for that maybe.","accounts_in_message":[],"_revision_number":8},{"id":"94965e353af43b0367cc136055d83bd16a6182ac","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-11 13:24:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a7d364e24a26844834e2c6a07400a9f3df6a58db","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-16 13:04:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6576ea82000044e7346f5cd826dd088fe1c5243c","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-16 13:05:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"342e987c50b37b362286b8fd4d1a54a2c783ecc6","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-16 13:05:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54a37203f77cbc45cd49569117155b7ad4575e7d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-16 14:11:44.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8b6157b5ebc0e747183e0a38aa4b40c6f11af47e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-16 14:14:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2e88e5d353d69117c4901c0797944ab91953a213","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-16 14:44:05.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/9 : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":9},{"id":"65b90d34c02ee03fa11b0623fefaf9cdb7e95bf1","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-16 15:06:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2a3fe7d3a04ec4d53aeb00648b49d08f7dbc42fb","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-16 15:08:49.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"3d6f2535192ee9ad5e45dcd724e5a09eb7161e92","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-16 15:46:39.000000000","message":"Patch Set 10: Code-Review+1\n\nSince I updated, will only +1","accounts_in_message":[],"_revision_number":10},{"id":"379ee32214ff3a733d2560f247fdfde64096abcd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-16 16:49:33.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/10 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":10},{"id":"11dbf58d36b7f31fc49121d2bef57e972bb73c1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:15:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-api/7cec481/ : SUCCESS in 58m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-designate-scenario/f38f419/ : SUCCESS in 34m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-dvr-multinode-scenario/9a6431a/ : FAILURE in 1h 34m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-scenario-linuxbridge/2c69a9d/ : SUCCESS in 1h 08m 59s\n- openstack-tox-cover http://logs.openstack.org/11/667511/10/check/openstack-tox-cover/9f41e2a/cover/ : SUCCESS in 31m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/10/check/openstack-tox-lower-constraints/d6f30d0/ : SUCCESS in 20m 07s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/10/check/openstack-tox-pep8/71900a7/ : SUCCESS in 13m 19s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/10/check/openstack-tox-py27/b5bfe80/ : SUCCESS in 29m 43s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/10/check/openstack-tox-py36/00668f9/ : SUCCESS in 26m 13s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/10/check/openstack-tox-py37/17b5b95/ : SUCCESS in 23m 30s\n- openstack-tox-docs http://logs.openstack.org/11/667511/10/check/openstack-tox-docs/ccc930f/html/ : SUCCESS in 5m 20s\n- tempest-full http://logs.openstack.org/11/667511/10/check/tempest-full/cc6085d/ : SUCCESS in 1h 29m 00s\n- neutron-grenade http://logs.openstack.org/11/667511/10/check/neutron-grenade/442df96/ : SUCCESS in 1h 02m 56s\n- grenade-py3 http://logs.openstack.org/11/667511/10/check/grenade-py3/82bb3fc/ : SUCCESS in 1h 03m 14s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/10/check/tempest-integrated-networking/3d52c64/ : SUCCESS in 1h 21m 29s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/10/check/build-openstack-releasenotes/9fed172/html/ : SUCCESS in 14m 36s\n- neutron-functional http://logs.openstack.org/11/667511/10/check/neutron-functional/54f61c6/ : SUCCESS in 39m 24s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/10/check/neutron-functional-python27/13253b9/ : SUCCESS in 49m 22s\n- neutron-fullstack http://logs.openstack.org/11/667511/10/check/neutron-fullstack/2187bb1/ : SUCCESS in 1h 03m 19s\n- neutron-rally-task http://logs.openstack.org/11/667511/10/check/neutron-rally-task/bba052a/ : SUCCESS in 2h 01m 56s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/10/check/neutron-tempest-dvr/0cce3e2/ : SUCCESS in 1h 32m 04s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/10/check/neutron-tempest-linuxbridge/fbf68f4/ : SUCCESS in 1h 43m 02s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/10/check/neutron-grenade-multinode/cd72a92/ : SUCCESS in 1h 20m 28s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/10/check/neutron-grenade-dvr-multinode/fd47072/ : SUCCESS in 1h 16m 17s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/10/check/tempest-multinode-full-py3/a14a1c1/ : SUCCESS in 1h 43m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/10/check/neutron-tempest-dvr-ha-multinode-full/89c421e/ : SUCCESS in 2h 03m 03s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/10/check/neutron-tempest-iptables_hybrid/a31241a/ : SUCCESS in 1h 24m 03s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/10/check/neutron-tempest-iptables_hybrid-fedora/f1c6254/ : SUCCESS in 1h 45m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/e9525e2/ : SUCCESS in 52m 00s (non-voting)\n- tempest-full-py3 http://logs.openstack.org/11/667511/10/check/tempest-full-py3/703918e/ : SUCCESS in 1h 20m 00s\n- tempest-slow-py3 http://logs.openstack.org/11/667511/10/check/tempest-slow-py3/699cda5/ : TIMED_OUT in 3h 06m 17s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/10/check/networking-ovn-tempest-dsvm-ovs-release/967ce63/ : SUCCESS in 2h 20m 32s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/10/check/networking-midonet-tempest-aio-ml2-centos-7/af942df/ : FAILURE in 1h 39m 24s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/10/check/openstacksdk-functional-devstack-networking/c57a155/ : SUCCESS in 1h 38m 09s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-functional-with-uwsgi/9021568/ : SUCCESS in 49m 34s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-fullstack-with-uwsgi/b3316cf/ : SUCCESS in 1h 15m 08s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-tempest-with-uwsgi/6750ac7/ : TIMED_OUT in 2h 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d2ab0c95c92807e01453519c640aa73ed560482b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-16 20:20:54.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"5a61daab4edc92823bd4e5ea7684dd41ebd27222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 00:39:36.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-api/ca54c32/ : SUCCESS in 1h 03m 57s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-designate-scenario/6fdcc54/ : SUCCESS in 40m 56s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-dvr-multinode-scenario/831b6e5/ : FAILURE in 1h 35m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-scenario-linuxbridge/6b8f08a/ : SUCCESS in 1h 25m 58s\n- openstack-tox-cover http://logs.openstack.org/11/667511/10/check/openstack-tox-cover/9510a13/cover/ : SUCCESS in 44m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/10/check/openstack-tox-lower-constraints/84a0b42/ : SUCCESS in 33m 52s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/10/check/openstack-tox-pep8/e77f382/ : SUCCESS in 12m 02s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/10/check/openstack-tox-py27/43cb0e1/ : SUCCESS in 25m 54s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/10/check/openstack-tox-py36/fe9705b/ : SUCCESS in 20m 53s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/10/check/openstack-tox-py37/a99c57f/ : SUCCESS in 21m 40s\n- openstack-tox-docs http://logs.openstack.org/11/667511/10/check/openstack-tox-docs/e140fb9/html/ : SUCCESS in 4m 25s\n- tempest-full http://logs.openstack.org/11/667511/10/check/tempest-full/8d41900/ : SUCCESS in 1h 52m 38s\n- neutron-grenade http://logs.openstack.org/11/667511/10/check/neutron-grenade/7b22cb4/ : FAILURE in 1h 14m 53s\n- grenade-py3 http://logs.openstack.org/11/667511/10/check/grenade-py3/78e0722/ : SUCCESS in 1h 07m 28s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/10/check/tempest-integrated-networking/4490225/ : SUCCESS in 1h 23m 53s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/10/check/build-openstack-releasenotes/806c9c4/html/ : SUCCESS in 19m 10s\n- neutron-functional http://logs.openstack.org/11/667511/10/check/neutron-functional/e3ef3a9/ : SUCCESS in 45m 19s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/10/check/neutron-functional-python27/0dcef9d/ : SUCCESS in 46m 21s\n- neutron-fullstack http://logs.openstack.org/11/667511/10/check/neutron-fullstack/fffbd23/ : SUCCESS in 57m 34s\n- neutron-rally-task http://logs.openstack.org/11/667511/10/check/neutron-rally-task/00c77e1/ : SUCCESS in 1h 55m 29s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/10/check/neutron-tempest-dvr/23fe31b/ : SUCCESS in 1h 18m 19s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/10/check/neutron-tempest-linuxbridge/3516644/ : SUCCESS in 1h 22m 04s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/10/check/neutron-grenade-multinode/eb478a4/ : SUCCESS in 1h 00m 58s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/10/check/neutron-grenade-dvr-multinode/e12523f/ : SUCCESS in 1h 23m 00s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/10/check/tempest-multinode-full-py3/438efef/ : SUCCESS in 1h 56m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/10/check/neutron-tempest-dvr-ha-multinode-full/e212a62/ : SUCCESS in 1h 55m 33s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/10/check/neutron-tempest-iptables_hybrid/bbffce0/ : SUCCESS in 1h 32m 29s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/10/check/neutron-tempest-iptables_hybrid-fedora/051e2ce/ : SUCCESS in 2h 19m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/61e22b3/ : SUCCESS in 1h 00m 58s (non-voting)\n- tempest-full-py3 http://logs.openstack.org/11/667511/10/check/tempest-full-py3/09d9693/ : SUCCESS in 1h 31m 05s\n- tempest-slow-py3 http://logs.openstack.org/11/667511/10/check/tempest-slow-py3/19b9e50/ : SUCCESS in 2h 24m 43s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/10/check/networking-ovn-tempest-dsvm-ovs-release/a18cafd/ : SUCCESS in 2h 16m 25s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/10/check/networking-midonet-tempest-aio-ml2-centos-7/d966b4b/ : FAILURE in 1h 43m 00s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/10/check/openstacksdk-functional-devstack-networking/59e179b/ : SUCCESS in 2h 15m 12s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-functional-with-uwsgi/e3abc0b/ : SUCCESS in 1h 12m 56s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-fullstack-with-uwsgi/663a25f/ : SUCCESS in 1h 13m 34s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-tempest-with-uwsgi/b401d55/ : SUCCESS in 1h 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e6a4759dc2bb3b1f2a91214de2f9a60668dba68f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-17 08:19:03.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"75ad9c9a183e9a620122f42619347cd49f2bff59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 11:35:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-api/a86e896/ : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-designate-scenario/b83e5b1/ : SUCCESS in 38m 34s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-dvr-multinode-scenario/942a8ff/ : SUCCESS in 1h 24m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/10/check/neutron-tempest-plugin-scenario-linuxbridge/e618136/ : SUCCESS in 1h 03m 46s\n- openstack-tox-cover http://logs.openstack.org/11/667511/10/check/openstack-tox-cover/abef084/cover/ : SUCCESS in 29m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/10/check/openstack-tox-lower-constraints/08952a5/ : SUCCESS in 23m 23s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/10/check/openstack-tox-pep8/66a4b24/ : SUCCESS in 13m 26s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/10/check/openstack-tox-py27/023cf45/ : SUCCESS in 24m 37s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/10/check/openstack-tox-py36/5403918/ : SUCCESS in 23m 40s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/10/check/openstack-tox-py37/873c9c4/ : SUCCESS in 22m 08s\n- openstack-tox-docs http://logs.openstack.org/11/667511/10/check/openstack-tox-docs/fd2d932/html/ : SUCCESS in 4m 24s\n- tempest-full http://logs.openstack.org/11/667511/10/check/tempest-full/a54e98d/ : SUCCESS in 1h 34m 28s\n- neutron-grenade http://logs.openstack.org/11/667511/10/check/neutron-grenade/67dffd3/ : SUCCESS in 1h 08m 44s\n- grenade-py3 http://logs.openstack.org/11/667511/10/check/grenade-py3/9ffba75/ : SUCCESS in 1h 05m 06s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/10/check/tempest-integrated-networking/552083e/ : SUCCESS in 1h 26m 54s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/10/check/build-openstack-releasenotes/b3501c4/html/ : SUCCESS in 19m 16s\n- neutron-functional http://logs.openstack.org/11/667511/10/check/neutron-functional/558ff11/ : SUCCESS in 50m 45s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/10/check/neutron-functional-python27/67f8938/ : SUCCESS in 51m 02s\n- neutron-fullstack http://logs.openstack.org/11/667511/10/check/neutron-fullstack/7f18e9f/ : SUCCESS in 53m 04s\n- neutron-rally-task http://logs.openstack.org/11/667511/10/check/neutron-rally-task/8468db3/ : SUCCESS in 2h 04m 40s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/10/check/neutron-tempest-dvr/7a73144/ : SUCCESS in 1h 24m 25s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/10/check/neutron-tempest-linuxbridge/fbb9044/ : SUCCESS in 1h 27m 07s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/10/check/neutron-grenade-multinode/0057006/ : SUCCESS in 1h 10m 56s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/10/check/neutron-grenade-dvr-multinode/8d9ba0a/ : SUCCESS in 1h 06m 31s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/10/check/tempest-multinode-full-py3/efec4c9/ : SUCCESS in 1h 40m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/10/check/neutron-tempest-dvr-ha-multinode-full/84f939c/ : SUCCESS in 1h 48m 59s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/10/check/neutron-tempest-iptables_hybrid/d5af878/ : SUCCESS in 1h 33m 17s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/10/check/neutron-tempest-iptables_hybrid-fedora/eca0650/ : SUCCESS in 1h 40m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/572e1a5/ : SUCCESS in 55m 12s (non-voting)\n- tempest-full-py3 http://logs.openstack.org/11/667511/10/check/tempest-full-py3/1a698c0/ : SUCCESS in 1h 21m 58s\n- tempest-slow-py3 http://logs.openstack.org/11/667511/10/check/tempest-slow-py3/f160d9a/ : SUCCESS in 2h 29m 54s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/10/check/networking-ovn-tempest-dsvm-ovs-release/e016dd5/ : SUCCESS in 2h 14m 34s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/10/check/networking-midonet-tempest-aio-ml2-centos-7/9b8ccab/ : FAILURE in 1h 41m 41s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/10/check/openstacksdk-functional-devstack-networking/f19776c/ : SUCCESS in 1h 30m 05s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-functional-with-uwsgi/e308d3a/ : SUCCESS in 49m 28s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-fullstack-with-uwsgi/f042fba/ : SUCCESS in 1h 02m 45s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/10/check/neutron-tempest-with-uwsgi/9167c28/ : TIMED_OUT in 2h 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1a616fa159d62235d539fdd4c7fae0e93daef59f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-07-17 12:28:01.000000000","message":"Patch Set 10:\n\n(4 comments)\n\nI wish we had some negative tests of _validate_network_subnetpools with IPv6","accounts_in_message":[],"_revision_number":10},{"id":"a9f942d914e5d03ad62ed358896d9a66f21a6d7b","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-17 12:59:51.000000000","message":"Patch Set 10:\n\n(3 comments)\n\nYou make a good point about negative tests for the prefix delegation case. We have good coverage on the happy path, I fought hard to get those tests passing :) . We could use some tests to assert things fail as we expect them to when using prefix delegation though.","accounts_in_message":[],"_revision_number":10},{"id":"241e7ae1b17df016dfa99ca7632dfdd48f573aec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-07-17 13:57:25.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4caac052e1337ed0010d7fdd4f05f55126e793c9","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-17 14:29:12.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c2216c2267e72ea887f9da073d01e8b07ce26fa1","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-18 14:25:19.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6d34b7755a29a2b3f2459024a7fc03385a156bf2","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-22 14:42:58.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7daece0d24dbda4c79e365e52d390c16d5922eae","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-22 14:44:10.000000000","message":"Patch Set 11:\n\nAdded a couple of tests to assert proper behavior of the prefix delegation paths.","accounts_in_message":[],"_revision_number":11},{"id":"d2bbe98968592382342279f4871a878a228b32d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-22 16:26:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/667511/11 : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":11},{"id":"2e667298479978065771957851eff2683b5d3c49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:43:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/11/check/neutron-tempest-plugin-api/dc41e92/ : SUCCESS in 1h 00m 05s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/11/check/neutron-tempest-plugin-designate-scenario/13d923d/ : SUCCESS in 34m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/11/check/neutron-tempest-plugin-dvr-multinode-scenario/27aa1b7/ : SUCCESS in 1h 33m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/11/check/neutron-tempest-plugin-scenario-linuxbridge/b2a9e94/ : SUCCESS in 1h 06m 16s\n- openstack-tox-cover http://logs.openstack.org/11/667511/11/check/openstack-tox-cover/0aea52d/cover/ : SUCCESS in 30m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/11/check/openstack-tox-lower-constraints/d0c3fe6/ : SUCCESS in 22m 52s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/11/check/openstack-tox-pep8/3bb3646/ : SUCCESS in 13m 39s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/11/check/openstack-tox-py27/c401706/ : SUCCESS in 22m 57s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/11/check/openstack-tox-py36/d8bb656/ : SUCCESS in 22m 15s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/11/check/openstack-tox-py37/643862f/ : SUCCESS in 24m 30s\n- openstack-tox-docs http://logs.openstack.org/11/667511/11/check/openstack-tox-docs/e61e4a6/html/ : SUCCESS in 6m 32s\n- grenade-py3 http://logs.openstack.org/11/667511/11/check/grenade-py3/1e266ca/ : SUCCESS in 1h 00m 10s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/11/check/tempest-integrated-networking/e7177e4/ : SUCCESS in 1h 17m 22s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/11/check/build-openstack-releasenotes/38f33b6/html/ : SUCCESS in 18m 24s\n- neutron-functional http://logs.openstack.org/11/667511/11/check/neutron-functional/3523812/ : SUCCESS in 56m 18s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/11/check/neutron-functional-python27/4f6acbd/ : SUCCESS in 47m 27s\n- neutron-fullstack http://logs.openstack.org/11/667511/11/check/neutron-fullstack/0a7ab19/ : SUCCESS in 58m 28s\n- neutron-rally-task http://logs.openstack.org/11/667511/11/check/neutron-rally-task/9ccca2f/ : SUCCESS in 1h 49m 41s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/11/check/neutron-tempest-dvr/e3a7bd1/ : SUCCESS in 1h 22m 59s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/11/check/neutron-tempest-linuxbridge/0fc2772/ : SUCCESS in 1h 26m 10s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/11/check/neutron-grenade-multinode/588f693/ : SUCCESS in 1h 22m 20s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/11/check/neutron-grenade-dvr-multinode/066b6a7/ : SUCCESS in 1h 19m 24s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/11/check/tempest-multinode-full-py3/b6a5d1e/ : SUCCESS in 1h 43m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/11/check/neutron-tempest-dvr-ha-multinode-full/d521528/ : SUCCESS in 1h 57m 27s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/11/check/neutron-tempest-iptables_hybrid/42989e0/ : SUCCESS in 1h 16m 42s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/11/check/neutron-tempest-iptables_hybrid-fedora/55b5c10/ : SUCCESS in 1h 30m 38s\n- neutron-grenade http://logs.openstack.org/11/667511/11/check/neutron-grenade/7ba4a9a/ : SUCCESS in 58m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/099260d/ : SUCCESS in 55m 35s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/11/check/tempest-slow-py3/dafac87/ : SUCCESS in 2h 22m 09s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/11/check/networking-ovn-tempest-dsvm-ovs-release/9077fd3/ : SUCCESS in 2h 03m 31s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/11/check/networking-midonet-tempest-aio-ml2-centos-7/360904c/ : FAILURE in 1h 48m 08s (non-voting)\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/11/check/openstacksdk-functional-devstack-networking/383a7af/ : SUCCESS in 1h 31m 13s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/11/check/neutron-functional-with-uwsgi/c9ea2aa/ : SUCCESS in 45m 52s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/11/check/neutron-fullstack-with-uwsgi/a831846/ : SUCCESS in 1h 00m 53s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/11/check/neutron-tempest-with-uwsgi/4969857/ : SUCCESS in 1h 24m 31s (non-voting)\n- neutron-centos-7-tripleo-standalone http://logs.openstack.org/11/667511/11/check/neutron-centos-7-tripleo-standalone/73bd2e0/ : FAILURE in 1h 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c31372f91a3e338da8a0bb07dcdb65ad7d36e3eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-30 01:09:51.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"ac0bcaad8bb61e7178177aa6cb32294885a9b0bf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-30 14:10:46.000000000","message":"Patch Set 11:\n\nbtw, the only reason I didn\u0027t +2 was because I made some of the changes","accounts_in_message":[],"_revision_number":11},{"id":"c4b1fa23e96d9724364774c72dbe04c15b8f9bdf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-07-30 14:15:46.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d48d2d09cbb8046de3fcc87084ddd753f168717c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-07-30 14:53:18.000000000","message":"Patch Set 11: Code-Review+1\n\nLooks good from what I can see","accounts_in_message":[],"_revision_number":11},{"id":"f525cd276b3d722849890ac014ab22ee3d62becf","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-07-30 23:34:14.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"aa78ad117bce7c522dc0740db344de93aa8d0b66","author":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"date":"2019-07-31 02:28:28.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d5bdec30c83d52a12586a5b5b9e1841acf414629","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-31 07:33:08.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"e4da3ed3f3d2af6ea2dddbe93fa21f1943306ad2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 07:33:34.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"46dc194c7c1b6fbcef1dc57fe3e35cd9a4788dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 10:54:57.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/11/gate/neutron-tempest-plugin-api/7cf2845/ : SUCCESS in 49m 05s\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/11/gate/neutron-tempest-plugin-scenario-linuxbridge/083b53d/ : SUCCESS in 1h 04m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/11/gate/openstack-tox-lower-constraints/f0f98dc/ : SUCCESS in 23m 50s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/11/gate/openstack-tox-pep8/4807050/ : FAILURE in 8m 13s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/11/gate/openstack-tox-py27/970bd59/ : SUCCESS in 25m 47s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/11/gate/openstack-tox-py36/6977a75/ : SUCCESS in 23m 58s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/11/gate/openstack-tox-py37/8889e88/ : SUCCESS in 25m 14s\n- openstack-tox-docs http://logs.openstack.org/11/667511/11/gate/openstack-tox-docs/7a2e73d/html/ : SUCCESS in 8m 08s\n- grenade-py3 http://logs.openstack.org/11/667511/11/gate/grenade-py3/85fe37d/ : SUCCESS in 59m 05s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/11/gate/tempest-integrated-networking/696e0bc/ : SUCCESS in 1h 21m 23s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/11/gate/build-openstack-releasenotes/cfdb234/html/ : SUCCESS in 20m 25s\n- neutron-functional http://logs.openstack.org/11/667511/11/gate/neutron-functional/b01aaac/ : SUCCESS in 42m 15s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/11/gate/neutron-functional-python27/6e05325/ : SUCCESS in 49m 24s\n- neutron-fullstack http://logs.openstack.org/11/667511/11/gate/neutron-fullstack/82f995c/ : SUCCESS in 1h 02m 57s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/11/gate/neutron-tempest-dvr/f3f86b9/ : SUCCESS in 1h 21m 19s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/11/gate/neutron-tempest-linuxbridge/c8682a8/ : SUCCESS in 1h 15m 09s\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/11/gate/neutron-tempest-iptables_hybrid/ed08825/ : SUCCESS in 1h 11m 10s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/11/gate/neutron-grenade-multinode/a20b43f/ : SUCCESS in 1h 12m 49s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/11/gate/neutron-grenade-dvr-multinode/71c720a/ : SUCCESS in 1h 11m 43s\n- neutron-grenade http://logs.openstack.org/11/667511/11/gate/neutron-grenade/6d1df98/ : SUCCESS in 1h 05m 39s\n- tempest-slow-py3 http://logs.openstack.org/11/667511/11/gate/tempest-slow-py3/075751d/ : SUCCESS in 2h 20m 21s\n- tempest-ipv6-only http://logs.openstack.org/11/667511/11/gate/tempest-ipv6-only/543cfc2/ : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":11},{"id":"1169a5155b9e2da16e6421ae1ddb16624deba658","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-31 14:50:34.000000000","message":"Patch Set 11:\n\nLooks like something changed related to pep8 enforcement on HEAD. Simple fix though, new patch set to clean this up coming shortly.","accounts_in_message":[],"_revision_number":11},{"id":"3a4c91518534ce6e52c8b0075bc8d6bbb26ff1c0","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-07-31 15:02:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c9ee58c4219fadf3a8964a065bcb3abd8717dff2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-31 15:07:14.000000000","message":"Patch Set 12: Code-Review+1\n\nThose E125 errors are all my doing :-o","accounts_in_message":[],"_revision_number":12},{"id":"f670de60a56474221b0ea5495d9731e8d8813f08","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 15:32:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-07-31/667511/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9cec88d : FAILURE in 29m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"099c3ce2a1c515f8329b9d610d668bc4f6c3c6c5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-31 16:52:12.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667511/12 : FAILURE in 1h 49m 26s","accounts_in_message":[],"_revision_number":12},{"id":"03950a75775e4a452af478a16d0a03e234b780dd","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-07-31 19:07:19.000000000","message":"Patch Set 12:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":12},{"id":"66726427db0c6febaf36d11217a36d7fe9fea887","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:13:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-api/77beaa0/ : SUCCESS in 53m 53s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-designate-scenario/784a720/ : SUCCESS in 39m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-dvr-multinode-scenario/ade0e10/ : FAILURE in 1h 47m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-scenario-linuxbridge/7cef6cb/ : SUCCESS in 1h 18m 26s\n- openstack-tox-cover http://logs.openstack.org/11/667511/12/check/openstack-tox-cover/d852114/cover/ : SUCCESS in 26m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/12/check/openstack-tox-lower-constraints/84a9467/ : SUCCESS in 25m 33s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/12/check/openstack-tox-pep8/b4cfcc5/ : SUCCESS in 13m 07s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/12/check/openstack-tox-py27/986f3d3/ : SUCCESS in 23m 27s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/12/check/openstack-tox-py36/1c6b54e/ : SUCCESS in 24m 46s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/12/check/openstack-tox-py37/663dab4/ : SUCCESS in 19m 55s\n- openstack-tox-docs http://logs.openstack.org/11/667511/12/check/openstack-tox-docs/3f4678e/html/ : SUCCESS in 4m 24s\n- grenade-py3 http://logs.openstack.org/11/667511/12/check/grenade-py3/982e0d0/ : SUCCESS in 1h 03m 12s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/12/check/tempest-integrated-networking/122729d/ : SUCCESS in 1h 22m 38s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/12/check/build-openstack-releasenotes/95a514a/html/ : SUCCESS in 20m 26s\n- neutron-functional http://logs.openstack.org/11/667511/12/check/neutron-functional/0c3d415/ : SUCCESS in 47m 10s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/12/check/neutron-functional-python27/9893309/ : SUCCESS in 47m 56s\n- neutron-fullstack http://logs.openstack.org/11/667511/12/check/neutron-fullstack/35c0012/ : SUCCESS in 57m 32s\n- neutron-rally-task http://logs.openstack.org/11/667511/12/check/neutron-rally-task/ba46b37/ : SUCCESS in 2h 04m 55s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/12/check/neutron-tempest-dvr/9ee88ef/ : SUCCESS in 1h 24m 31s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/12/check/neutron-tempest-linuxbridge/5f83b2e/ : SUCCESS in 1h 17m 54s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/12/check/neutron-grenade-multinode/0ed0d18/ : SUCCESS in 1h 00m 10s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/12/check/neutron-grenade-dvr-multinode/b198a98/ : SUCCESS in 1h 20m 31s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/12/check/tempest-multinode-full-py3/e77df51/ : FAILURE in 1h 22m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/12/check/neutron-tempest-dvr-ha-multinode-full/f1bbfe8/ : FAILURE in 1h 47m 16s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/12/check/neutron-tempest-iptables_hybrid/c44b5e9/ : SUCCESS in 1h 22m 41s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/12/check/neutron-tempest-iptables_hybrid-fedora/c952812/ : SUCCESS in 1h 36m 09s\n- neutron-grenade http://logs.openstack.org/11/667511/12/check/neutron-grenade/0a7382c/ : FAILURE in 9m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5054dca/ : SUCCESS in 54m 10s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/12/check/tempest-slow-py3/4cc1d44/ : SUCCESS in 2h 09m 35s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/12/check/networking-ovn-tempest-dsvm-ovs-release/46e9ce7/ : SUCCESS in 2h 09m 51s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/12/check/networking-midonet-tempest-aio-ml2-centos-7/0df8477/ : FAILURE in 18m 15s (non-voting)\n- tempest-ipv6-only http://logs.openstack.org/11/667511/12/check/tempest-ipv6-only/90f1fcf/ : SUCCESS in 1h 07m 10s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/12/check/openstacksdk-functional-devstack-networking/f021fcd/ : SUCCESS in 1h 26m 05s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/12/check/neutron-functional-with-uwsgi/966323d/ : SUCCESS in 47m 18s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/12/check/neutron-fullstack-with-uwsgi/60f2fdc/ : SUCCESS in 1h 13m 44s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/12/check/neutron-tempest-with-uwsgi/f0bea22/ : SUCCESS in 1h 45m 05s (non-voting)\n- neutron-centos-7-tripleo-standalone http://logs.openstack.org/11/667511/12/check/neutron-centos-7-tripleo-standalone/b1499e0/ : SUCCESS in 2h 06m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dca7165a48984b2c7c1d9b1fecefb81b72ae07bd","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-07-31 21:44:34.000000000","message":"Patch Set 12: Code-Review+2\n\nrecheck. grenade job failure doesn\u0027t seem related to the patch","accounts_in_message":[],"_revision_number":12},{"id":"18b1a9e20869f409b45a6d3a367d0af784183ff5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-07-31 23:23:53.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667511/12 : FAILURE in 1h 38m 50s","accounts_in_message":[],"_revision_number":12},{"id":"710c806caa71b7b1b2f6c89ce4c4dc7e7e3a8f14","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-07-31 23:53:43.000000000","message":"Patch Set 12: Code-Review+1\n\nand pep8 passed (side note we can probably call that new check the Brian detector?)","accounts_in_message":[],"_revision_number":12},{"id":"d88203edeab28c9f3a7b782313763eeb92591e9d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-01 00:06:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/11/667511/12/check-mlnx/Neutron-ML2-Sriov/df31e0c : SUCCESS in 1h 28m 47s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"83f9ddb162822e93e7d5ee210907fb18a8731af0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:20:08.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-api/fe7e9e3/ : SUCCESS in 51m 31s\n- neutron-tempest-plugin-designate-scenario http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-designate-scenario/0945f01/ : SUCCESS in 33m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-dvr-multinode-scenario/aead8c5/ : FAILURE in 1h 40m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge http://logs.openstack.org/11/667511/12/check/neutron-tempest-plugin-scenario-linuxbridge/002d455/ : SUCCESS in 1h 09m 55s\n- openstack-tox-cover http://logs.openstack.org/11/667511/12/check/openstack-tox-cover/f8cb340/cover/ : SUCCESS in 37m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/667511/12/check/openstack-tox-lower-constraints/b554af7/ : SUCCESS in 21m 18s\n- openstack-tox-pep8 http://logs.openstack.org/11/667511/12/check/openstack-tox-pep8/c44fe61/ : SUCCESS in 13m 25s\n- openstack-tox-py27 http://logs.openstack.org/11/667511/12/check/openstack-tox-py27/44ae339/ : SUCCESS in 25m 19s\n- openstack-tox-py36 http://logs.openstack.org/11/667511/12/check/openstack-tox-py36/7881804/ : SUCCESS in 32m 42s\n- openstack-tox-py37 http://logs.openstack.org/11/667511/12/check/openstack-tox-py37/8a766d5/ : SUCCESS in 22m 00s\n- openstack-tox-docs http://logs.openstack.org/11/667511/12/check/openstack-tox-docs/228a7c6/html/ : SUCCESS in 4m 06s\n- grenade-py3 http://logs.openstack.org/11/667511/12/check/grenade-py3/c3740b3/ : SUCCESS in 1h 06m 43s\n- tempest-integrated-networking http://logs.openstack.org/11/667511/12/check/tempest-integrated-networking/6453529/ : SUCCESS in 1h 24m 31s\n- build-openstack-releasenotes http://logs.openstack.org/11/667511/12/check/build-openstack-releasenotes/c92f939/html/ : SUCCESS in 19m 32s\n- neutron-functional http://logs.openstack.org/11/667511/12/check/neutron-functional/87081a5/ : TIMED_OUT in 2h 18m 24s\n- neutron-functional-python27 http://logs.openstack.org/11/667511/12/check/neutron-functional-python27/1200f3b/ : SUCCESS in 55m 04s\n- neutron-fullstack http://logs.openstack.org/11/667511/12/check/neutron-fullstack/746c911/ : SUCCESS in 59m 23s\n- neutron-rally-task http://logs.openstack.org/11/667511/12/check/neutron-rally-task/cb52926/ : SUCCESS in 2h 01m 13s\n- neutron-tempest-dvr http://logs.openstack.org/11/667511/12/check/neutron-tempest-dvr/e54b00d/ : SUCCESS in 1h 24m 50s\n- neutron-tempest-linuxbridge http://logs.openstack.org/11/667511/12/check/neutron-tempest-linuxbridge/3f92982/ : SUCCESS in 1h 10m 02s\n- neutron-grenade-multinode http://logs.openstack.org/11/667511/12/check/neutron-grenade-multinode/6df2887/ : SUCCESS in 1h 14m 37s\n- neutron-grenade-dvr-multinode http://logs.openstack.org/11/667511/12/check/neutron-grenade-dvr-multinode/1f1a5e0/ : SUCCESS in 1h 11m 17s\n- tempest-multinode-full-py3 http://logs.openstack.org/11/667511/12/check/tempest-multinode-full-py3/0d62e04/ : SUCCESS in 1h 44m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full http://logs.openstack.org/11/667511/12/check/neutron-tempest-dvr-ha-multinode-full/541ed8f/ : FAILURE in 1h 33m 37s (non-voting)\n- neutron-tempest-iptables_hybrid http://logs.openstack.org/11/667511/12/check/neutron-tempest-iptables_hybrid/f91e1ba/ : SUCCESS in 1h 33m 57s\n- neutron-tempest-iptables_hybrid-fedora http://logs.openstack.org/11/667511/12/check/neutron-tempest-iptables_hybrid-fedora/b808996/ : SUCCESS in 1h 36m 15s\n- neutron-grenade http://logs.openstack.org/11/667511/12/check/neutron-grenade/d592501/ : SUCCESS in 1h 11m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/667511/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a1796eb/ : SUCCESS in 1h 12m 24s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/11/667511/12/check/tempest-slow-py3/23c7416/ : SUCCESS in 2h 19m 02s\n- networking-ovn-tempest-dsvm-ovs-release http://logs.openstack.org/11/667511/12/check/networking-ovn-tempest-dsvm-ovs-release/292aa0a/ : SUCCESS in 2h 07m 47s\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/11/667511/12/check/networking-midonet-tempest-aio-ml2-centos-7/ace185d/ : FAILURE in 12m 36s (non-voting)\n- tempest-ipv6-only http://logs.openstack.org/11/667511/12/check/tempest-ipv6-only/84e0f17/ : SUCCESS in 1h 13m 18s\n- openstacksdk-functional-devstack-networking http://logs.openstack.org/11/667511/12/check/openstacksdk-functional-devstack-networking/f025e92/ : SUCCESS in 1h 25m 29s (non-voting)\n- neutron-functional-with-uwsgi http://logs.openstack.org/11/667511/12/check/neutron-functional-with-uwsgi/a87111c/ : SUCCESS in 45m 00s (non-voting)\n- neutron-fullstack-with-uwsgi http://logs.openstack.org/11/667511/12/check/neutron-fullstack-with-uwsgi/4471f1e/ : SUCCESS in 1h 00m 30s (non-voting)\n- neutron-tempest-with-uwsgi http://logs.openstack.org/11/667511/12/check/neutron-tempest-with-uwsgi/b31ef56/ : SUCCESS in 1h 13m 11s (non-voting)\n- neutron-centos-7-tripleo-standalone http://logs.openstack.org/11/667511/12/check/neutron-centos-7-tripleo-standalone/a49edff/ : SUCCESS in 3h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"03f84d277dc39b8c608b7a6f245f4291eb81b11a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-01 01:26:20.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"23e5cbd126fe40794bfe27d29d4ee534ca48c9da","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 01:57:09.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-01/667511/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/b91872a : FAILURE in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1ec30ca3ca06605c13f7128d9abeff501df1e15d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 04:26:27.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-api/be01d1a/ : SUCCESS in 49m 25s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-designate-scenario/ef9a38a/ : SUCCESS in 46m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-dvr-multinode-scenario/011f147/ : SUCCESS in 1h 29m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-scenario-linuxbridge/7ae6821/ : SUCCESS in 1h 04m 16s\n- openstack-tox-cover https://logs.opendev.org/11/667511/12/check/openstack-tox-cover/ead5bbe/cover/ : SUCCESS in 32m 33s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/667511/12/check/openstack-tox-lower-constraints/1027d96/ : SUCCESS in 20m 40s\n- openstack-tox-pep8 https://logs.opendev.org/11/667511/12/check/openstack-tox-pep8/f15f9d8/ : SUCCESS in 11m 23s\n- openstack-tox-py27 https://logs.opendev.org/11/667511/12/check/openstack-tox-py27/4f4a7f9/ : SUCCESS in 23m 08s\n- openstack-tox-py36 https://logs.opendev.org/11/667511/12/check/openstack-tox-py36/61dc8e5/ : SUCCESS in 25m 32s\n- openstack-tox-py37 https://logs.opendev.org/11/667511/12/check/openstack-tox-py37/2cb5f48/ : SUCCESS in 22m 08s\n- openstack-tox-docs https://logs.opendev.org/11/667511/12/check/openstack-tox-docs/017fa28/html/ : SUCCESS in 4m 15s\n- grenade-py3 https://logs.opendev.org/11/667511/12/check/grenade-py3/4a15c9b/ : SUCCESS in 57m 28s\n- tempest-integrated-networking https://logs.opendev.org/11/667511/12/check/tempest-integrated-networking/9467164/ : SUCCESS in 1h 21m 24s\n- build-openstack-releasenotes https://logs.opendev.org/11/667511/12/check/build-openstack-releasenotes/9e5eadb/html/ : SUCCESS in 15m 08s\n- neutron-functional https://logs.opendev.org/11/667511/12/check/neutron-functional/d2e50e5/ : SUCCESS in 54m 36s\n- neutron-functional-python27 https://logs.opendev.org/11/667511/12/check/neutron-functional-python27/2ba7b7f/ : SUCCESS in 45m 52s\n- neutron-fullstack https://logs.opendev.org/11/667511/12/check/neutron-fullstack/2e7a723/ : SUCCESS in 1h 10m 42s\n- neutron-rally-task https://logs.opendev.org/11/667511/12/check/neutron-rally-task/94e5ab1/ : SUCCESS in 2h 17m 42s\n- neutron-tempest-dvr https://logs.opendev.org/11/667511/12/check/neutron-tempest-dvr/7345ba8/ : SUCCESS in 1h 25m 56s\n- neutron-tempest-linuxbridge https://logs.opendev.org/11/667511/12/check/neutron-tempest-linuxbridge/0319e16/ : SUCCESS in 1h 17m 45s\n- neutron-grenade-multinode https://logs.opendev.org/11/667511/12/check/neutron-grenade-multinode/23d2c57/ : SUCCESS in 1h 15m 02s\n- neutron-grenade-dvr-multinode https://logs.opendev.org/11/667511/12/check/neutron-grenade-dvr-multinode/6419608/ : SUCCESS in 1h 07m 21s\n- tempest-multinode-full-py3 https://logs.opendev.org/11/667511/12/check/tempest-multinode-full-py3/672afdc/ : FAILURE in 1h 48m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://logs.opendev.org/11/667511/12/check/neutron-tempest-dvr-ha-multinode-full/98f71d8/ : FAILURE in 1h 52m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://logs.opendev.org/11/667511/12/check/neutron-tempest-iptables_hybrid/40102ca/ : SUCCESS in 1h 35m 23s\n- neutron-tempest-iptables_hybrid-fedora https://logs.opendev.org/11/667511/12/check/neutron-tempest-iptables_hybrid-fedora/9fbc5e2/ : SUCCESS in 1h 33m 46s\n- neutron-grenade https://logs.opendev.org/11/667511/12/check/neutron-grenade/50568cf/ : POST_FAILURE in 54m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://logs.opendev.org/11/667511/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f40cfca/ : SUCCESS in 47m 48s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/11/667511/12/check/tempest-slow-py3/ef79ab3/ : SUCCESS in 2h 57m 40s\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/11/667511/12/check/networking-ovn-tempest-dsvm-ovs-release/d842e4e/ : SUCCESS in 2h 22m 49s\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/11/667511/12/check/networking-midonet-tempest-aio-ml2-centos-7/229bc3d/ : FAILURE in 13m 54s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/11/667511/12/check/tempest-ipv6-only/5d164cc/ : SUCCESS in 59m 52s\n- openstacksdk-functional-devstack-networking https://logs.opendev.org/11/667511/12/check/openstacksdk-functional-devstack-networking/5543126/ : SUCCESS in 1h 20m 31s (non-voting)\n- neutron-functional-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-functional-with-uwsgi/82e90eb/ : SUCCESS in 45m 03s (non-voting)\n- neutron-fullstack-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-fullstack-with-uwsgi/95f196c/ : SUCCESS in 1h 02m 04s (non-voting)\n- neutron-tempest-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-tempest-with-uwsgi/d086bd2/ : SUCCESS in 1h 52m 51s (non-voting)\n- neutron-centos-7-tripleo-standalone https://logs.opendev.org/11/667511/12/check/neutron-centos-7-tripleo-standalone/0ca631b/ : SUCCESS in 2h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b9ec2e5e3f9506284d1f95dd6c265badc097831a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-01 06:58:33.000000000","message":"Patch Set 12:\n\nrecheck POST_FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"c5b49bfece2c1b9bcde58ce6ac6e05917b5af307","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-01 08:31:44.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"c7b1c839019e7cbb6e00dc11077b4290bc95b493","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-01 08:35:36.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667511/12 : FAILURE in 1h 36m 34s","accounts_in_message":[],"_revision_number":12},{"id":"c09911f1f8314850ad578aad15e2b63ded00b99b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 09:35:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-api/4befcd8/ : SUCCESS in 56m 17s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-designate-scenario/6192bcd/ : SUCCESS in 41m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-dvr-multinode-scenario/dbb09ab/ : SUCCESS in 1h 40m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-scenario-linuxbridge/58ec3da/ : SUCCESS in 1h 12m 35s\n- openstack-tox-cover https://logs.opendev.org/11/667511/12/check/openstack-tox-cover/b11dd21/cover/ : SUCCESS in 26m 29s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/667511/12/check/openstack-tox-lower-constraints/d991871/ : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://logs.opendev.org/11/667511/12/check/openstack-tox-pep8/f2a461c/ : SUCCESS in 11m 45s\n- openstack-tox-py27 https://logs.opendev.org/11/667511/12/check/openstack-tox-py27/3226436/ : SUCCESS in 22m 50s\n- openstack-tox-py36 https://logs.opendev.org/11/667511/12/check/openstack-tox-py36/763e4d5/ : SUCCESS in 22m 58s\n- openstack-tox-py37 https://logs.opendev.org/11/667511/12/check/openstack-tox-py37/6cc8465/ : SUCCESS in 24m 50s\n- openstack-tox-docs https://logs.opendev.org/11/667511/12/check/openstack-tox-docs/963d473/html/ : SUCCESS in 6m 02s\n- grenade-py3 https://logs.opendev.org/11/667511/12/check/grenade-py3/f79c53d/ : SUCCESS in 59m 52s\n- tempest-integrated-networking https://logs.opendev.org/11/667511/12/check/tempest-integrated-networking/e4ba7dd/ : SUCCESS in 1h 20m 28s\n- build-openstack-releasenotes https://logs.opendev.org/11/667511/12/check/build-openstack-releasenotes/7de7be6/html/ : SUCCESS in 17m 41s\n- neutron-functional https://logs.opendev.org/11/667511/12/check/neutron-functional/4790f64/ : SUCCESS in 42m 06s\n- neutron-functional-python27 https://logs.opendev.org/11/667511/12/check/neutron-functional-python27/a37fdb7/ : SUCCESS in 43m 41s\n- neutron-fullstack https://logs.opendev.org/11/667511/12/check/neutron-fullstack/999a06e/ : SUCCESS in 55m 58s\n- neutron-rally-task https://logs.opendev.org/11/667511/12/check/neutron-rally-task/d62a66d/ : SUCCESS in 1h 48m 56s\n- neutron-tempest-dvr https://logs.opendev.org/11/667511/12/check/neutron-tempest-dvr/9eb0913/ : SUCCESS in 1h 13m 07s\n- neutron-tempest-linuxbridge https://logs.opendev.org/11/667511/12/check/neutron-tempest-linuxbridge/871efd2/ : SUCCESS in 1h 19m 57s\n- neutron-grenade-multinode https://logs.opendev.org/11/667511/12/check/neutron-grenade-multinode/a3d608f/ : SUCCESS in 1h 05m 52s\n- neutron-grenade-dvr-multinode https://logs.opendev.org/11/667511/12/check/neutron-grenade-dvr-multinode/8a62a00/ : SUCCESS in 1h 05m 26s\n- tempest-multinode-full-py3 https://logs.opendev.org/11/667511/12/check/tempest-multinode-full-py3/0cdeb2f/ : SUCCESS in 1h 41m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://logs.opendev.org/11/667511/12/check/neutron-tempest-dvr-ha-multinode-full/14ef57d/ : SUCCESS in 1h 59m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://logs.opendev.org/11/667511/12/check/neutron-tempest-iptables_hybrid/b2a2d64/ : SUCCESS in 1h 16m 27s\n- neutron-tempest-iptables_hybrid-fedora https://logs.opendev.org/11/667511/12/check/neutron-tempest-iptables_hybrid-fedora/a6e4f56/ : SUCCESS in 1h 42m 13s\n- neutron-grenade https://logs.opendev.org/11/667511/12/check/neutron-grenade/d152627/ : SUCCESS in 58m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://logs.opendev.org/11/667511/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d68ca8f/ : SUCCESS in 51m 01s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/11/667511/12/check/tempest-slow-py3/db6a907/ : SUCCESS in 2h 32m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/11/667511/12/check/networking-ovn-tempest-dsvm-ovs-release/aadef4c/ : SUCCESS in 2h 03m 47s\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/11/667511/12/check/networking-midonet-tempest-aio-ml2-centos-7/928a9fe/ : FAILURE in 15m 40s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/11/667511/12/check/tempest-ipv6-only/26595fd/ : SUCCESS in 1h 08m 04s\n- openstacksdk-functional-devstack-networking https://logs.opendev.org/11/667511/12/check/openstacksdk-functional-devstack-networking/034792f/ : SUCCESS in 1h 19m 51s (non-voting)\n- neutron-functional-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-functional-with-uwsgi/1f514ae/ : SUCCESS in 53m 26s (non-voting)\n- neutron-fullstack-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-fullstack-with-uwsgi/f6cc96a/ : SUCCESS in 1h 00m 21s (non-voting)\n- neutron-tempest-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-tempest-with-uwsgi/92dabb6/ : SUCCESS in 1h 35m 27s (non-voting)\n- neutron-centos-7-tripleo-standalone https://logs.opendev.org/11/667511/12/check/neutron-centos-7-tripleo-standalone/316bdfc/ : SUCCESS in 1h 53m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"33dec5883ad866f0ad194593460375705a31f165","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 09:36:08.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"9f2d9b07811a784b01cbc8185ab7efad160fe5ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 12:14:19.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://logs.opendev.org/11/667511/12/gate/neutron-tempest-plugin-api/b8dfe42/ : SUCCESS in 50m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://logs.opendev.org/11/667511/12/gate/neutron-tempest-plugin-scenario-linuxbridge/a33e340/ : SUCCESS in 1h 08m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/667511/12/gate/openstack-tox-lower-constraints/8891aad/ : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://logs.opendev.org/11/667511/12/gate/openstack-tox-pep8/a9d57d7/ : SUCCESS in 14m 01s\n- openstack-tox-py27 https://logs.opendev.org/11/667511/12/gate/openstack-tox-py27/0dce501/ : SUCCESS in 22m 31s\n- openstack-tox-py36 https://logs.opendev.org/11/667511/12/gate/openstack-tox-py36/a4663cf/ : SUCCESS in 24m 57s\n- openstack-tox-py37 https://logs.opendev.org/11/667511/12/gate/openstack-tox-py37/c092782/ : SUCCESS in 24m 06s\n- openstack-tox-docs https://logs.opendev.org/11/667511/12/gate/openstack-tox-docs/da8fe0c/html/ : SUCCESS in 4m 51s\n- grenade-py3 https://logs.opendev.org/11/667511/12/gate/grenade-py3/0a56ab6/ : FAILURE in 26m 56s\n- tempest-integrated-networking https://logs.opendev.org/11/667511/12/gate/tempest-integrated-networking/d1dcba3/ : SUCCESS in 1h 30m 09s\n- build-openstack-releasenotes https://logs.opendev.org/11/667511/12/gate/build-openstack-releasenotes/fdab1d0/html/ : SUCCESS in 21m 02s\n- neutron-functional https://logs.opendev.org/11/667511/12/gate/neutron-functional/93f7b7a/ : SUCCESS in 44m 18s\n- neutron-functional-python27 https://logs.opendev.org/11/667511/12/gate/neutron-functional-python27/0c1a226/ : SUCCESS in 54m 40s\n- neutron-fullstack https://logs.opendev.org/11/667511/12/gate/neutron-fullstack/9b6b1b2/ : SUCCESS in 53m 20s\n- neutron-tempest-dvr https://logs.opendev.org/11/667511/12/gate/neutron-tempest-dvr/f7c3117/ : SUCCESS in 1h 32m 49s\n- neutron-tempest-linuxbridge https://logs.opendev.org/11/667511/12/gate/neutron-tempest-linuxbridge/6dbcbc4/ : SUCCESS in 1h 12m 49s\n- neutron-tempest-iptables_hybrid https://logs.opendev.org/11/667511/12/gate/neutron-tempest-iptables_hybrid/b3c8059/ : SUCCESS in 1h 31m 02s\n- neutron-grenade-multinode https://logs.opendev.org/11/667511/12/gate/neutron-grenade-multinode/bc22203/ : SUCCESS in 1h 02m 48s\n- neutron-grenade-dvr-multinode https://logs.opendev.org/11/667511/12/gate/neutron-grenade-dvr-multinode/f609cd3/ : SUCCESS in 59m 34s\n- neutron-grenade https://logs.opendev.org/11/667511/12/gate/neutron-grenade/b763a12/ : SUCCESS in 1h 10m 57s\n- tempest-slow-py3 https://logs.opendev.org/11/667511/12/gate/tempest-slow-py3/0458230/ : SUCCESS in 2h 31m 43s\n- tempest-ipv6-only https://logs.opendev.org/11/667511/12/gate/tempest-ipv6-only/6a26cf6/ : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":12},{"id":"bd29d455238cdca8e062fa2c1c506256dba5a08d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-01 15:27:52.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"747780498ee16ea8577c382d49cec8f7ac5a0ae4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 16:16:29.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-01/667511/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a3cadac : FAILURE in 46m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6120a77291f62ebfc27f388e2940b100b955dff2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-01 17:21:06.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/667511/12 : FAILURE in 1h 52m 48s","accounts_in_message":[],"_revision_number":12},{"id":"fcd3ee54e0d7682841cdc4f902abf550e6bad07d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:57:55.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-api/1572303/ : SUCCESS in 53m 43s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-designate-scenario/8b27e43/ : SUCCESS in 33m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-dvr-multinode-scenario/68db50a/ : SUCCESS in 1h 34m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://logs.opendev.org/11/667511/12/check/neutron-tempest-plugin-scenario-linuxbridge/bd64a84/ : SUCCESS in 1h 03m 38s\n- openstack-tox-cover https://logs.opendev.org/11/667511/12/check/openstack-tox-cover/6f9c172/cover/ : SUCCESS in 35m 22s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/667511/12/check/openstack-tox-lower-constraints/f299b39/ : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://logs.opendev.org/11/667511/12/check/openstack-tox-pep8/283f5fc/ : SUCCESS in 13m 17s\n- openstack-tox-py27 https://logs.opendev.org/11/667511/12/check/openstack-tox-py27/040916f/ : SUCCESS in 24m 53s\n- openstack-tox-py36 https://logs.opendev.org/11/667511/12/check/openstack-tox-py36/5a48007/ : SUCCESS in 21m 24s\n- openstack-tox-py37 https://logs.opendev.org/11/667511/12/check/openstack-tox-py37/23e94b9/ : SUCCESS in 19m 05s\n- openstack-tox-docs https://logs.opendev.org/11/667511/12/check/openstack-tox-docs/5549179/html/ : SUCCESS in 5m 51s\n- grenade-py3 https://logs.opendev.org/11/667511/12/check/grenade-py3/da6d475/ : SUCCESS in 58m 54s\n- tempest-integrated-networking https://logs.opendev.org/11/667511/12/check/tempest-integrated-networking/3e0d6f6/ : SUCCESS in 1h 09m 44s\n- build-openstack-releasenotes https://logs.opendev.org/11/667511/12/check/build-openstack-releasenotes/bc88235/html/ : SUCCESS in 18m 06s\n- neutron-functional https://logs.opendev.org/11/667511/12/check/neutron-functional/ac2cdf9/ : SUCCESS in 44m 50s\n- neutron-functional-python27 https://logs.opendev.org/11/667511/12/check/neutron-functional-python27/5faaedd/ : SUCCESS in 44m 23s\n- neutron-fullstack https://logs.opendev.org/11/667511/12/check/neutron-fullstack/f1fbe7a/ : SUCCESS in 1h 05m 24s\n- neutron-rally-task https://logs.opendev.org/11/667511/12/check/neutron-rally-task/37b4863/ : SUCCESS in 1h 45m 31s\n- neutron-tempest-dvr https://logs.opendev.org/11/667511/12/check/neutron-tempest-dvr/cf17b1b/ : SUCCESS in 1h 23m 54s\n- neutron-tempest-linuxbridge https://logs.opendev.org/11/667511/12/check/neutron-tempest-linuxbridge/84979e8/ : SUCCESS in 1h 18m 04s\n- neutron-grenade-multinode https://logs.opendev.org/11/667511/12/check/neutron-grenade-multinode/22935bc/ : SUCCESS in 1h 09m 12s\n- neutron-grenade-dvr-multinode https://logs.opendev.org/11/667511/12/check/neutron-grenade-dvr-multinode/74b1a38/ : SUCCESS in 1h 13m 40s\n- tempest-multinode-full-py3 https://logs.opendev.org/11/667511/12/check/tempest-multinode-full-py3/a062b3c/ : FAILURE in 1h 35m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://logs.opendev.org/11/667511/12/check/neutron-tempest-dvr-ha-multinode-full/2a324a9/ : FAILURE in 1h 38m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://logs.opendev.org/11/667511/12/check/neutron-tempest-iptables_hybrid/bb3da19/ : SUCCESS in 1h 31m 48s\n- neutron-tempest-iptables_hybrid-fedora https://logs.opendev.org/11/667511/12/check/neutron-tempest-iptables_hybrid-fedora/8ed9ae0/ : SUCCESS in 1h 49m 44s\n- neutron-grenade https://logs.opendev.org/11/667511/12/check/neutron-grenade/524ad00/ : SUCCESS in 1h 17m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://logs.opendev.org/11/667511/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/465d07f/ : SUCCESS in 54m 14s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/11/667511/12/check/tempest-slow-py3/47d5671/ : SUCCESS in 2h 26m 09s\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/11/667511/12/check/networking-ovn-tempest-dsvm-ovs-release/0a7d6dd/ : SUCCESS in 2h 10m 14s\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/11/667511/12/check/networking-midonet-tempest-aio-ml2-centos-7/d9d2c4c/ : FAILURE in 15m 56s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/11/667511/12/check/tempest-ipv6-only/f6f648d/ : SUCCESS in 1h 04m 06s\n- openstacksdk-functional-devstack-networking https://logs.opendev.org/11/667511/12/check/openstacksdk-functional-devstack-networking/2abe48d/ : SUCCESS in 1h 29m 32s (non-voting)\n- neutron-functional-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-functional-with-uwsgi/dbe77e2/ : SUCCESS in 47m 12s (non-voting)\n- neutron-fullstack-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-fullstack-with-uwsgi/684cedc/ : FAILURE in 59m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://logs.opendev.org/11/667511/12/check/neutron-tempest-with-uwsgi/8e098c9/ : SUCCESS in 1h 40m 14s (non-voting)\n- neutron-centos-7-tripleo-standalone https://logs.opendev.org/11/667511/12/check/neutron-centos-7-tripleo-standalone/514cf24/ : SUCCESS in 1h 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d20c970545c4cee8dd7a39207902368d1cc3112c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 22:58:15.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"eeea43f524d76167732f6618b4cfa921d5c9b763","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:28:08.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://logs.opendev.org/11/667511/12/gate/neutron-tempest-plugin-api/a217e87/ : SUCCESS in 52m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://logs.opendev.org/11/667511/12/gate/neutron-tempest-plugin-scenario-linuxbridge/aba6d33/ : SUCCESS in 1h 00m 48s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/667511/12/gate/openstack-tox-lower-constraints/7e8044e/ : SUCCESS in 33m 37s\n- openstack-tox-pep8 https://logs.opendev.org/11/667511/12/gate/openstack-tox-pep8/79313e8/ : SUCCESS in 13m 08s\n- openstack-tox-py27 https://logs.opendev.org/11/667511/12/gate/openstack-tox-py27/2948bfc/ : SUCCESS in 34m 50s\n- openstack-tox-py36 https://logs.opendev.org/11/667511/12/gate/openstack-tox-py36/71e25cf/ : SUCCESS in 22m 10s\n- openstack-tox-py37 https://logs.opendev.org/11/667511/12/gate/openstack-tox-py37/e5f9888/ : SUCCESS in 22m 25s\n- openstack-tox-docs https://logs.opendev.org/11/667511/12/gate/openstack-tox-docs/35edcf5/html/ : SUCCESS in 7m 19s\n- grenade-py3 https://logs.opendev.org/11/667511/12/gate/grenade-py3/6b3ac6d/ : SUCCESS in 1h 09m 46s\n- tempest-integrated-networking https://logs.opendev.org/11/667511/12/gate/tempest-integrated-networking/ffce51a/ : SUCCESS in 1h 21m 41s\n- build-openstack-releasenotes https://logs.opendev.org/11/667511/12/gate/build-openstack-releasenotes/0581b93/html/ : SUCCESS in 19m 58s\n- neutron-functional https://logs.opendev.org/11/667511/12/gate/neutron-functional/cce9026/ : SUCCESS in 47m 44s\n- neutron-functional-python27 https://logs.opendev.org/11/667511/12/gate/neutron-functional-python27/17be017/ : SUCCESS in 48m 27s\n- neutron-fullstack https://logs.opendev.org/11/667511/12/gate/neutron-fullstack/bf1d477/ : SUCCESS in 57m 02s\n- neutron-tempest-dvr https://logs.opendev.org/11/667511/12/gate/neutron-tempest-dvr/17a6642/ : SUCCESS in 1h 14m 10s\n- neutron-tempest-linuxbridge https://logs.opendev.org/11/667511/12/gate/neutron-tempest-linuxbridge/2d98875/ : SUCCESS in 1h 18m 21s\n- neutron-tempest-iptables_hybrid https://logs.opendev.org/11/667511/12/gate/neutron-tempest-iptables_hybrid/d189798/ : SUCCESS in 1h 18m 58s\n- neutron-grenade-multinode https://logs.opendev.org/11/667511/12/gate/neutron-grenade-multinode/a2774a0/ : SUCCESS in 1h 10m 02s\n- neutron-grenade-dvr-multinode https://logs.opendev.org/11/667511/12/gate/neutron-grenade-dvr-multinode/9480e16/ : SUCCESS in 1h 19m 22s\n- neutron-grenade https://logs.opendev.org/11/667511/12/gate/neutron-grenade/99f236d/ : SUCCESS in 57m 57s\n- tempest-slow-py3 https://logs.opendev.org/11/667511/12/gate/tempest-slow-py3/9af7b2d/ : SUCCESS in 2h 23m 35s\n- tempest-ipv6-only https://logs.opendev.org/11/667511/12/gate/tempest-ipv6-only/9c8c05f/ : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":12},{"id":"dbf68cf8684af58f0be184b9bd417ed3eae518bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:28:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"32182010c2c2050f37221824788a91d762081a0f","revisions":{"4dfa828ec3677ecdd8ce5996a5c5aa3c226bf161":{"kind":"REWORK","_number":1,"created":"2019-06-26 05:34:52.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/1"}}},"commit":{"parents":[{"commit":"f556ca703afebf250b0651f9ee24857c3dcc72d9","subject":"Merge \"Switch to new engine facade for BaseResourceFilter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f556ca703afebf250b0651f9ee24857c3dcc72d9"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:34:39.000000000","tz":-300},"subject":"[WIP] Relax subnet pool network affinity constraints","message":"[WIP] Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dfa828ec3677ecdd8ce5996a5c5aa3c226bf161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dfa828ec3677ecdd8ce5996a5c5aa3c226bf161"}]},"branch":"refs/heads/master"},"7a1a9444ad8c8aad5341312f5713fe9ba6a07485":{"kind":"REWORK","_number":2,"created":"2019-06-26 19:06:00.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/2"}}},"commit":{"parents":[{"commit":"44b0cae16bb049adcc5b086a5cbe79d9db07a03c","subject":"Merge \"Remove some synchronization decorators from ip_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44b0cae16bb049adcc5b086a5cbe79d9db07a03c"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 19:05:49.000000000","tz":-300},"subject":"[WIP] Relax subnet pool network affinity constraints","message":"[WIP] Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\nDepends-On: https://review.opendev.org/#/c/667644/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a1a9444ad8c8aad5341312f5713fe9ba6a07485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a1a9444ad8c8aad5341312f5713fe9ba6a07485"}]},"branch":"refs/heads/master"},"c0cadd7692101f4733601b05f9e79203de9f7e2c":{"kind":"REWORK","_number":3,"created":"2019-06-27 18:58:26.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/3"}}},"commit":{"parents":[{"commit":"1bcd2094369d30ab018ae8462e83b533302e2fda","subject":"Merge \"Retry trunk status updates failing with StaleDataError\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bcd2094369d30ab018ae8462e83b533302e2fda"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-27 18:58:05.000000000","tz":-300},"subject":"[WIP] Relax subnet pool network affinity constraints","message":"[WIP] Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0cadd7692101f4733601b05f9e79203de9f7e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0cadd7692101f4733601b05f9e79203de9f7e2c"}]},"branch":"refs/heads/master"},"bab7df8b5a725422836d7de362cfbd2aeeb3072e":{"kind":"REWORK","_number":4,"created":"2019-06-30 19:34:49.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/4"}}},"commit":{"parents":[{"commit":"ad36ba680f8dbfb1243835a51898707626594f6c","subject":"Merge \"BGP Documentation Updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ad36ba680f8dbfb1243835a51898707626594f6c"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-30 19:34:34.000000000","tz":-300},"subject":"[WIP] Relax subnet pool network affinity constraints","message":"[WIP] Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bab7df8b5a725422836d7de362cfbd2aeeb3072e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bab7df8b5a725422836d7de362cfbd2aeeb3072e"}]},"branch":"refs/heads/master"},"58ad2a35dc808d10094e40f307e96d7571cc21d5":{"kind":"REWORK","_number":5,"created":"2019-07-08 14:12:58.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/5"}}},"commit":{"parents":[{"commit":"e9dce503628d8dad1a4dc57d90e8019206fcab2b","subject":"Merge \"Minimizing L3 agent QoS extensions lock granularity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9dce503628d8dad1a4dc57d90e8019206fcab2b"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-08 14:12:08.000000000","tz":-300},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58ad2a35dc808d10094e40f307e96d7571cc21d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58ad2a35dc808d10094e40f307e96d7571cc21d5"}]},"branch":"refs/heads/master"},"81cdf2a4a121ed3407101458b56481747b6bd2b6":{"kind":"REWORK","_number":6,"created":"2019-07-08 14:34:51.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/6"}}},"commit":{"parents":[{"commit":"e9dce503628d8dad1a4dc57d90e8019206fcab2b","subject":"Merge \"Minimizing L3 agent QoS extensions lock granularity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9dce503628d8dad1a4dc57d90e8019206fcab2b"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-08 14:34:09.000000000","tz":-300},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81cdf2a4a121ed3407101458b56481747b6bd2b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81cdf2a4a121ed3407101458b56481747b6bd2b6"}]},"branch":"refs/heads/master"},"cc98d1f8c4b31528803a732670b5e6a2bc16cae3":{"kind":"REWORK","_number":7,"created":"2019-07-09 16:07:29.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/7"}}},"commit":{"parents":[{"commit":"a89244459a6cea1bc09af5cfd57be33ca12a53cb","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a89244459a6cea1bc09af5cfd57be33ca12a53cb"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-09 15:48:13.000000000","tz":-300},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nCloses: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc98d1f8c4b31528803a732670b5e6a2bc16cae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc98d1f8c4b31528803a732670b5e6a2bc16cae3"}]},"branch":"refs/heads/master"},"0740c33f50361c25b54579a5a118b0be0c3a7e12":{"kind":"REWORK","_number":8,"created":"2019-07-10 16:36:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/11/667511/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/8"}}},"commit":{"parents":[{"commit":"a89244459a6cea1bc09af5cfd57be33ca12a53cb","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a89244459a6cea1bc09af5cfd57be33ca12a53cb"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-10 16:33:59.000000000","tz":-240},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nCloses: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0740c33f50361c25b54579a5a118b0be0c3a7e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0740c33f50361c25b54579a5a118b0be0c3a7e12"}]},"branch":"refs/heads/master"},"ac8de36e93a67a64cd0e94c44b568de73736878d":{"kind":"REWORK","_number":9,"created":"2019-07-16 13:05:23.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/9"}}},"commit":{"parents":[{"commit":"64c8bfbbf0cf87942bccdde7efd73cf25187d362","subject":"Merge \"Improve \"OVSFirewallDriver.process_trusted_ports\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64c8bfbbf0cf87942bccdde7efd73cf25187d362"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-16 13:04:54.000000000","tz":-300},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nCloses: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac8de36e93a67a64cd0e94c44b568de73736878d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac8de36e93a67a64cd0e94c44b568de73736878d"}]},"branch":"refs/heads/master"},"2a750b349f9f76e7c7153652b98cdb79c960762b":{"kind":"REWORK","_number":10,"created":"2019-07-16 15:08:49.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/10"}}},"commit":{"parents":[{"commit":"64c8bfbbf0cf87942bccdde7efd73cf25187d362","subject":"Merge \"Improve \"OVSFirewallDriver.process_trusted_ports\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64c8bfbbf0cf87942bccdde7efd73cf25187d362"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-16 15:07:08.000000000","tz":0},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nCloses: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a750b349f9f76e7c7153652b98cdb79c960762b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a750b349f9f76e7c7153652b98cdb79c960762b"}]},"branch":"refs/heads/master"},"0d0982078ccb949bf77cca2eb43ad0db38d847f3":{"kind":"REWORK","_number":11,"created":"2019-07-22 14:42:58.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/11"}}},"commit":{"parents":[{"commit":"ff636e18a338a4afbd1153aa01a978f32326bc0c","subject":"Merge \"Switch test cases to sudo dependent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff636e18a338a4afbd1153aa01a978f32326bc0c"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-22 14:39:02.000000000","tz":-300},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nCloses: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d0982078ccb949bf77cca2eb43ad0db38d847f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d0982078ccb949bf77cca2eb43ad0db38d847f3"}]},"branch":"refs/heads/master"},"32182010c2c2050f37221824788a91d762081a0f":{"kind":"REWORK","_number":12,"created":"2019-07-31 15:02:16.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/11/667511/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/667511/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/667511/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/667511/12"}}},"commit":{"parents":[{"commit":"1421c63f4b48c2b0e8965d68c20e439638efcfa5","subject":"Merge \"Add RPC method to get networks for L3 and DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1421c63f4b48c2b0e8965d68c20e439638efcfa5"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 05:29:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-07-31 15:01:55.000000000","tz":-300},"subject":"Relax subnet pool network affinity constraints","message":"Relax subnet pool network affinity constraints\n\nThis change relaxes the constraint that all subnets of the same\naddress family on a network must be allocated from the same subnet\npool. It allows subnets that share the same address scope to\nco-exist on a network. If there is no address scope involved the\nsubnet pool / network affinity constraints continue to  enforced\nas done previously.\n\nChange-Id: I33bd17c723b3e8d409415bda008440f8ed9cfa68\nCloses: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32182010c2c2050f37221824788a91d762081a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32182010c2c2050f37221824788a91d762081a0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
