)]}'
{"id":"openstack%2Fneutron~674044","triplet_id":"openstack%2Fneutron~master~I6161a8b387812808c4d679f882a3193c93235647","project":"openstack/neutron","branch":"master","topic":"sec_gr_callback","hashtags":[],"change_id":"I6161a8b387812808c4d679f882a3193c93235647","subject":"use callback payloads for SECURITY_GROUP","status":"MERGED","created":"2019-08-01 14:14:44.000000000","updated":"2021-06-28 11:20:48.000000000","submitted":"2021-06-28 11:18:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"674044-sec_gr_callback","meta_rev_id":"8b782334236ae2e93a75b0e7f6e16e39205d4257","_number":674044,"virtual_id_number":674044,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 11:18:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-28 09:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-28 07:19:30.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-28 09:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 14:44:39.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 14:51:45.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 15:23:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 15:59:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-01 21:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 02:39:59.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:48:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-18 10:54:00.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"REVIEWER"},{"updated":"2021-05-25 13:10:03.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-06-18 03:31:32.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-06-28 09:07:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"c966266cbbb45d9b9f212271c29974cf7f5708e2","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-08-01 14:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b566b54da899ba5f169c8e96db0ab8975379e52f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-08-01 14:14:50.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f3e3fa26a5d66e484607462f93acf304bc07fc6c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 14:44:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-08-01/674044/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/784a086 : FAILURE in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"481109d752c2d4b3d7f1d8c45908916e54a60df5","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-08-01 14:51:45.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/674044/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d2e462917927ec94df180a64ae3c558b3a08375b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-08-01 15:23:40.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/August-2019/Arista_Tempest_Testrefs/changes/44/674044/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"afc76423ac0c5781d5f84ff6236d8bf06289c4b5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-01 15:59:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/1 : SUCCESS in 1h 44m 44s","accounts_in_message":[],"_revision_number":1},{"id":"569d2ace3dd05181dddf859e2cffb4d35e16bca6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:01:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://logs.opendev.org/44/674044/1/check/neutron-tempest-plugin-api/9983892/ : SUCCESS in 1h 02m 22s\n- neutron-tempest-plugin-designate-scenario https://logs.opendev.org/44/674044/1/check/neutron-tempest-plugin-designate-scenario/7e111a7/ : SUCCESS in 36m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario https://logs.opendev.org/44/674044/1/check/neutron-tempest-plugin-dvr-multinode-scenario/e7d2769/ : SUCCESS in 1h 33m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://logs.opendev.org/44/674044/1/check/neutron-tempest-plugin-scenario-linuxbridge/a87824e/ : SUCCESS in 1h 00m 49s\n- openstack-tox-cover https://logs.opendev.org/44/674044/1/check/openstack-tox-cover/2765793/cover/ : SUCCESS in 36m 57s\n- openstack-tox-lower-constraints https://logs.opendev.org/44/674044/1/check/openstack-tox-lower-constraints/471cfde/ : SUCCESS in 25m 14s\n- openstack-tox-pep8 https://logs.opendev.org/44/674044/1/check/openstack-tox-pep8/9addc72/ : SUCCESS in 14m 03s\n- openstack-tox-py27 https://logs.opendev.org/44/674044/1/check/openstack-tox-py27/d9d4a22/ : SUCCESS in 25m 37s\n- openstack-tox-py36 https://logs.opendev.org/44/674044/1/check/openstack-tox-py36/d9ba56a/ : SUCCESS in 23m 05s\n- openstack-tox-py37 https://logs.opendev.org/44/674044/1/check/openstack-tox-py37/3372a00/ : FAILURE in 21m 30s\n- openstack-tox-docs https://logs.opendev.org/44/674044/1/check/openstack-tox-docs/d10500b/html/ : SUCCESS in 4m 28s\n- grenade-py3 https://logs.opendev.org/44/674044/1/check/grenade-py3/75474f5/ : SUCCESS in 1h 03m 29s\n- tempest-integrated-networking https://logs.opendev.org/44/674044/1/check/tempest-integrated-networking/f2f7aa3/ : SUCCESS in 1h 20m 12s\n- neutron-functional https://logs.opendev.org/44/674044/1/check/neutron-functional/b4ee555/ : SUCCESS in 44m 11s\n- neutron-functional-python27 https://logs.opendev.org/44/674044/1/check/neutron-functional-python27/5a01f34/ : SUCCESS in 42m 40s\n- neutron-fullstack https://logs.opendev.org/44/674044/1/check/neutron-fullstack/0e3ad0d/ : FAILURE in 58m 55s\n- neutron-rally-task https://logs.opendev.org/44/674044/1/check/neutron-rally-task/fb013bd/ : SUCCESS in 1h 56m 08s\n- neutron-tempest-dvr https://logs.opendev.org/44/674044/1/check/neutron-tempest-dvr/54bde41/ : SUCCESS in 1h 10m 48s\n- neutron-tempest-linuxbridge https://logs.opendev.org/44/674044/1/check/neutron-tempest-linuxbridge/d377418/ : SUCCESS in 1h 20m 27s\n- neutron-grenade-multinode https://logs.opendev.org/44/674044/1/check/neutron-grenade-multinode/e958404/ : SUCCESS in 1h 03m 16s\n- neutron-grenade-dvr-multinode https://logs.opendev.org/44/674044/1/check/neutron-grenade-dvr-multinode/f4703ad/ : SUCCESS in 1h 13m 22s\n- tempest-multinode-full-py3 https://logs.opendev.org/44/674044/1/check/tempest-multinode-full-py3/2c56d70/ : SUCCESS in 1h 34m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://logs.opendev.org/44/674044/1/check/neutron-tempest-dvr-ha-multinode-full/2c3ffdb/ : FAILURE in 1h 27m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://logs.opendev.org/44/674044/1/check/neutron-tempest-iptables_hybrid/83fcb9d/ : SUCCESS in 1h 26m 44s\n- neutron-tempest-iptables_hybrid-fedora https://logs.opendev.org/44/674044/1/check/neutron-tempest-iptables_hybrid-fedora/b4fa645/ : SUCCESS in 1h 26m 19s\n- neutron-grenade https://logs.opendev.org/44/674044/1/check/neutron-grenade/53facd5/ : SUCCESS in 1h 02m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://logs.opendev.org/44/674044/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ad9c1b1/ : SUCCESS in 1h 01m 10s (non-voting)\n- tempest-slow-py3 https://logs.opendev.org/44/674044/1/check/tempest-slow-py3/547d27b/ : SUCCESS in 2h 17m 22s\n- networking-ovn-tempest-dsvm-ovs-release https://logs.opendev.org/44/674044/1/check/networking-ovn-tempest-dsvm-ovs-release/ffb9514/ : FAILURE in 30m 40s\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/44/674044/1/check/networking-midonet-tempest-aio-ml2-centos-7/53613c1/ : FAILURE in 15m 44s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/44/674044/1/check/tempest-ipv6-only/5ce6aa5/ : SUCCESS in 1h 06m 55s\n- openstacksdk-functional-devstack-networking https://logs.opendev.org/44/674044/1/check/openstacksdk-functional-devstack-networking/895ebd5/ : SUCCESS in 1h 24m 34s (non-voting)\n- neutron-functional-with-uwsgi https://logs.opendev.org/44/674044/1/check/neutron-functional-with-uwsgi/173ade1/ : SUCCESS in 42m 41s (non-voting)\n- neutron-fullstack-with-uwsgi https://logs.opendev.org/44/674044/1/check/neutron-fullstack-with-uwsgi/0c7407e/ : SUCCESS in 50m 51s (non-voting)\n- neutron-tempest-with-uwsgi https://logs.opendev.org/44/674044/1/check/neutron-tempest-with-uwsgi/845bd09/ : SUCCESS in 1h 31m 38s (non-voting)\n- neutron-centos-7-tripleo-standalone https://logs.opendev.org/44/674044/1/check/neutron-centos-7-tripleo-standalone/34ccec2/ : FAILURE in 1h 55m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bfce15498f36f6e29b9b481f9258dee9c365a61","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-02 02:39:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/44/674044/1/check-mlnx/Neutron-ML2-Sriov/f7d0db8 : SUCCESS in 1h 25m 13s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"e2cd7ccefd2f32c5dde6126ed0070116ec15a860","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:48:35.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"338a11a576ba60151cc2d0bcb31248799d23fe3e","tag":"autogenerated:gerrit:restore","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-14 12:41:23.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"93851e18a75737a17a777ed67e3de5aa24a365e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 12:41:54.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"05a6bb6b650f10b57880c1ae5b4cd43c81fcccfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 12:42:06.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"a5b41aed487cf5529db7df0a00b930910313627b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-18 10:54:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa87bc748530666f7e0de25b8ea7a1e8753aaace","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-05-18 10:56:55.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2021/Arista_Tempest_Testrefs/changes/44/674044/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"fdfa6096fe6551945b21dac7fe3cffc84098ee85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:37:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3651722d043a44f499de388a16786464 : FAILURE in 41m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d62c7ef3ca944338ad1c9d689c641132 : FAILURE in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05ba96561bbb1d42c1ab5bf1ac33518276476162","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-18 12:33:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/2 : SUCCESS in 1h 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b5f91eff499d6cad2381b10175c68455e9fc7a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:41:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e8be3b1900e4492b98a9bce8e1a2cf21 : SUCCESS in 30m 05s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e998e47947a543418488af1d55351e77 : SUCCESS in 2h 09m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5a1f6086d114406cb82e4a4b051332ef : SUCCESS in 2h 10m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00660a7cc0e7416ba7d6b0240fee2d8b : SUCCESS in 1h 32m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4d2e3206f10b42efbc9a6d02a12b20df : FAILURE in 27m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d2f2913eec14e659559a7af2d280536 : FAILURE in 38m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4e6d0c324fb4012ba8d75ec2a956e6e : FAILURE in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f085e8de9f64bc2b0ea80b8dbcca8ec : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5adf5d5ead6e42d5af5f86d522ee62f7 : FAILURE in 18m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b4f8fdb7e904daba674d3c99f2cbc1c : FAILURE in 29m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16904294d9ff4ee1b2e023d0e679fd1c : FAILURE in 29m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b650636dba84444b030716b51b0d0ea : SUCCESS in 8m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d87c5b0adf4452cb8a59a501124d6cd : RETRY_LIMIT in 11m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7393be4de1a64f8ab9dc172670057e7a : RETRY_LIMIT in 6m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e23d75aaed8544d8adc74050719710c4 : SUCCESS in 54m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7facc2edb3944b1e9e4e97178857e2c7 : SUCCESS in 1h 02m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b062ce4f8d9547ccb8504bc128f535b6 : SUCCESS in 1h 10m 58s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b28ec5bcab154cd88e69c17f0e2d06dc : SUCCESS in 1h 12m 02s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/708878700ba646f690d276edda97928d : FAILURE in 1h 21m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9f9319f24f64ad5845179d03bb52b07 : SUCCESS in 1h 36m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc2f7213f6c34467bd2ae569e832156e : SUCCESS in 57m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4892073d5d2e4931b32860b962f7b7c2 : FAILURE in 24m 27s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/39618781368a41e9a2c2e2e2505dd986 : FAILURE in 24m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd0e6b9101e0491ca336c25ac0ac1f86 : FAILURE in 24m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f8db19cb1b3c49c59df07d092ea99aed : FAILURE in 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"28cb7515b9809dd939bb52ac0f5b5be43fd33359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-19 14:17:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"261593b6a1cb4eccc66a78b26960b40dda024ddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 14:57:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/521e889ed16e402aa5db43463b2ec2e1 : FAILURE in 33m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3ebdf5b11614b23be7eda999f15dc8a : FAILURE in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3a260dd8f04a61cf4620ab67c1d8712ea55eccd0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-19 15:59:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/3 : SUCCESS in 1h 41m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f9cede5e556dd854cd5114d11c20a251802ec9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:10:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fe909b7478214354b9021b2a7ef3c8e7 : SUCCESS in 30m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e87da59e57174b51afa113aa0a30b129 : FAILURE in 1h 51m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/25f0f36e2270463e91f0f019c189141e : SUCCESS in 1h 47m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b204ff35b0b47e3a136eb60a50ef0d2 : SUCCESS in 1h 14m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cd47baf1135a4ee7b8b26045c5ecd441 : FAILURE in 18m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/596c08af6048498fa92e72a53d036ee6 : FAILURE in 36m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74232e5e197d41309bb2b87db4453656 : FAILURE in 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e375e35db1ca4e748fcca545b086fd77 : SUCCESS in 14m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/633b400b73e24279b2f18688ea206f26 : FAILURE in 24m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59d28b3ada664187a9781827262505d2 : FAILURE in 23m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb4f17a6b80e44f08bfb1a790d576527 : FAILURE in 24m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe902da977c401194716d196d5cd818 : FAILURE in 7m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d7fb20a1fd549f89dcacb9edacfe690 : FAILURE in 39m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28b9046a7bed4182977a1ebfa834c940 : FAILURE in 48m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1dfb9a12d7ce48d5b201e7b90f05bb2c : SUCCESS in 1h 10m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08c1f347383045c299b149c9b0d00c60 : SUCCESS in 58m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5e60028b29b1403a8cabc093c79de0eb : SUCCESS in 50m 46s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b41e63eb88b4f91a9aa5ece49bf155a : SUCCESS in 1h 15m 13s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9577cf15877c4608a753e80c90625797 : SUCCESS in 1h 20m 15s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c88b75b32c4e4d56a3c974b6b5e0448b : SUCCESS in 1h 37m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f9f34cfb8a04e4ba0eeb497b70db4e3 : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4f8a3bf168664c8f912711fa5c7818e5 : FAILURE in 25m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ac3f0bf35eed4f6a93ef6921286b4bb5 : FAILURE in 21m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efc66e9bc3d34efebeb7ac5c7e490115 : FAILURE in 22m 09s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c4855fa3e1d646c598bf44be66beab18 : FAILURE in 31m 52s","accounts_in_message":[],"_revision_number":3},{"id":"eaf739b237689f6d8edc3d3c7413065bb52272fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-22 07:42:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71ff26bf4c76689f00bc08c49d82a5015dda5a53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 08:27:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8e7f6b794ac40b0bdb7e5af1885e43f : FAILURE in 43m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40cf37eb659f44a0aae8892018dbf5e5 : FAILURE in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1d8a90b50e666e469281f9f39654a2469e59feef","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-22 09:14:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/4 : SUCCESS in 1h 31m 34s","accounts_in_message":[],"_revision_number":4},{"id":"b2099debcde489710ebfe9dc77cccfe4682f2682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-22 09:29:35.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a1b77f77ed8644e3bac7a75582160272 : SUCCESS in 37m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c1621a246dc14642ae75e9333843320a : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/06914dcf88b2409681bed637ea8551eb : SUCCESS in 1h 20m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ac54e784e2a4ce5be32835475826624 : SUCCESS in 1h 21m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/357ece1dee5a4ee6b895d68bd4cfcfba : SUCCESS in 1h 45m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fea76bacfde4e22a130904cb2934d20 : FAILURE in 36m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fbf19be82ef425baea0779ce77e71f8 : FAILURE in 38m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e688b4940e48929b853238c76a300f : FAILURE in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eecfe7fa000e4707a51c36f5ae86e7a6 : FAILURE in 23m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d361db1615ac4f90ae33672f23aa3f46 : FAILURE in 25m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df78f935fb154338beaf1bd4ea3af079 : FAILURE in 17m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76d93c9783c4acfbfd08cdec6420256 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfddead2a7174ede9e071bb523beeadc : FAILURE in 45m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/66db656fb4af4a33a8ec98427595fb77 : SUCCESS in 1h 16m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/248aaf400db74752bcc7f9a1b4471b58 : SUCCESS in 1h 11m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c2c807c1d184473a61b6cc69b5ee03e : SUCCESS in 54m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1c410b6007194d1a9122be3f769969f4 : SUCCESS in 57m 00s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/271ea584bead47318d578bf0ecca43f7 : SUCCESS in 1h 15m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85a3dbe519644b0585326d9daa85e86b : SUCCESS in 1h 16m 14s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72132599c3234bdf8584acf38ae7b389 : SUCCESS in 1h 22m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad886e717e29412cbb90c896b1dc81ef : SUCCESS in 47m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/77fd79fcbe584b44b56c81af1c0c521d : SUCCESS in 1h 02m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9895b4434a324f668cc83d26e96f996d : SUCCESS in 24m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d4008c7783474ae5809a60db3e3b0b58 : SUCCESS in 1h 06m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67c39777d59d48b4ba52fc0a4b542e1e : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":4},{"id":"00b7dbdd52d01f8087595f9cfda532bdbb285987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-23 06:10:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a94f3fbcda497e98fea126e75eec24689bf423b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 07:03:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/145522bd25374b2c8ad5c58e72acaa9e : SUCCESS in 39m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3411186dbc1c4fccbcf42fa4ce9976ae : SUCCESS in 43m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"edf6b1884b78ae809b17f833646aa69e69adabc8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-23 08:03:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/5 : SUCCESS in 1h 51m 54s","accounts_in_message":[],"_revision_number":5},{"id":"35d64a68dc8dab9717db10288f5ee1a11e8ef0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 08:27:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/339ff46aaa8a4b5ca5ccce981e69980d : SUCCESS in 44m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/54e18967ee694be6be4a3e45c2f62d45 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2e593f20b714e52b52e3e409257888c : SUCCESS in 1h 30m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac014b22de0d429d952d036b8b0552d7 : SUCCESS in 1h 29m 38s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/972f94ddc4c14d9483a3bd31cfd9e1c2 : SUCCESS in 2h 04m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcfa0af0159c44efb751ebfd4fd331e7 : SUCCESS in 38m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5559f4cfbf64d6e8af192bc09c815d4 : SUCCESS in 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a26a6ee8424675ae1f23938d54f86e : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4387820da96c4b13a062f9d3c3b8c5c4 : SUCCESS in 27m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebf29c191c00429fae8cb4eb1f6d97b2 : SUCCESS in 29m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc3a990838064e64ae0b9998b31baa71 : SUCCESS in 29m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c6e61ef230470283f2bc675e26e4b7 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb0f381603844ac0bdbe5d63d730c300 : SUCCESS in 47m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/beb8d804a6c042dc9b21e93d920d0b7b : SUCCESS in 1h 26m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ceb5fd66d03340e09b3fba240389deb4 : SUCCESS in 1h 17m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58d6c94a8ffa40c6b8eea46aff63ea3c : SUCCESS in 58m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d4003ed146594f09903b01c3f3a66da2 : SUCCESS in 1h 04m 43s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b8d48c4e1c2d44ebbb6a1b8eacb72bcb : SUCCESS in 1h 22m 28s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/06477326705b43778866936a9381cff7 : FAILURE in 1h 06m 17s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3190ec8fc9a543d3b71c164ad96d5065 : SUCCESS in 1h 11m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d841462be8f42d6995c96352b555138 : SUCCESS in 47m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/85981bc4bb38466e9513183d7c79d524 : SUCCESS in 47m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/95cffc6025d048d986b8fceec1c3d19e : SUCCESS in 27m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eac62d0a70c9488b8a420710a7661778 : SUCCESS in 1h 09m 49s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/809792341f9a4e329f4ef7408da62f94 : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":5},{"id":"baebaa6dbf9addd52cde4eb07b7bc9f4a602973a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-25 13:10:03.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0adc874a0b57673c4dc048a225d61954a259ae3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-25 13:37:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6895ed835a7a5ce2693e1af2649917961bc42141","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-25 13:38:00.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1cd5c8df3c0ee141580566ac5c20c64cf40d07b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 14:21:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49403050da4d4925b5e60f470f0a0325 : SUCCESS in 32m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8de2dfc65b714a54a92857465a6ee466 : SUCCESS in 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fd3640498b97c82df3a1ed0642da8e5f2697bf20","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-25 15:12:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/6 : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":6},{"id":"72ccd35fc2b356bbc1f2bf8330c2d42879f409ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:15:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b4c454b7a7f943e98b95a50511692853 : FAILURE in 42m 10s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e41bcdccaf6c401887727435abad5551 : SUCCESS in 1h 40m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3f6c0adb190d430b9c7a5d45b2ddb9c1 : SUCCESS in 1h 35m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a0aa9eaaa4b847e4b6bd625143788a82 : SUCCESS in 1h 13m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5f42e8ddfdb845f695d08c17d84a64a1 : SUCCESS in 2h 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8a3980b7a4a4e1d983c0d438a8ed8a0 : SUCCESS in 44m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a7f9998f3fa42929c48c0087ece504f : SUCCESS in 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad590113db8e45a4a0d789d7dbe731f3 : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e380c0b098684f0fa8edec74b163d020 : SUCCESS in 26m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22d1e0e2177742d898d14c2a57e60599 : SUCCESS in 25m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a58602218e6744e48580d862f779de9c : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b4b523a1a54da8a90855d5708327ce : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0d849f92be54406a5cf6b1d557946d0 : FAILURE in 40m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/899a41d8d9bb40c28828d753214b3d6e : SUCCESS in 1h 23m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/90c44576c4af410a9ddb686d14a75a1c : FAILURE in 1h 23m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/523bac575cae464bb81cc72f2cd5c3a8 : SUCCESS in 52m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae22fb7c670b40d190996c832011bfe2 : SUCCESS in 57m 32s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/19a6bedbae644aa6bc1e050a752250fc : SUCCESS in 1h 09m 45s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/395d2f26792145318d2047bd6c9d23dc : FAILURE in 24m 02s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9d087a1bd7848c1be8496943c5373e9 : SUCCESS in 1h 52m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fea6ae759094a18a3895fb24045a88f : SUCCESS in 1h 02m 44s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b8c9d1ff61544111807223ab007f343f : SUCCESS in 56m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/799104e60eba4946b41e32cfd0b5b1ff : SUCCESS in 27m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/66f6bccd919b48a4b91d7a0ebe9e6990 : SUCCESS in 1h 34m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/85b21ed773064276b65f5e3487e85375 : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":6},{"id":"dfa6941a8c42d554e3a84704faa5a803d353ed23","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-05-26 06:56:57.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b254d6ca42b96e853d086a6b1089e0da33e03c5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 08:13:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbb8a095c2fe426c93b58c7805a221ce : SUCCESS in 33m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ea48b2f68594ed6b70c007375481e3b : SUCCESS in 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5757f5e05667a481bceea6c752d8343b322d479f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-05-26 08:16:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"04105d475138a05cb26c2ce748a6929a8e36bad8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-05-26 08:37:56.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/674044/6 : FAILURE in 1h 40m 25s","accounts_in_message":[],"_revision_number":6},{"id":"5bb907662463fe83718f39265bf6e29f07625977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 09:25:55.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/99388603e7394285ba4817df48544b49 : FAILURE in 45m 15s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d175ad3659ee4fc7a3df655ba4648008 : SUCCESS in 1h 43m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4310a24838ea452e8e3eaa72755d1d47 : SUCCESS in 2h 17m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5220a003ac2f4d3a82ecb48287968aaa : SUCCESS in 1h 12m 19s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/77a6f3b3b36443839c0f671d89ac1de0 : SUCCESS in 2h 24m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/974baf8d22624e3a971caa0d1a908607 : SUCCESS in 37m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ba6079f77834698945710ce3194aa0c : SUCCESS in 44m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6057ba5253ea4862982399d3f51a2ec6 : SUCCESS in 14m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6690a4925f6e4015a822e47361e7acc0 : SUCCESS in 25m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a52f582fd6f4d83b33b7bac7bc96fac : SUCCESS in 26m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec75627b7cd740d48dbfed29e1e79188 : SUCCESS in 29m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b2b73db0ae4287b1eb3643359d0146 : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98800b22ae324478bbc887754e3c9466 : SUCCESS in 49m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0e31da175ab4ae59aaef4ea3e4244b3 : SUCCESS in 1h 23m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0c4fc19b642042a7bacb99b20f4c6625 : FAILURE in 1h 33m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8127da0d670845f0a6e9730c8bde5e4a : SUCCESS in 1h 00m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ade6cbd6938446c98981189e0fa728ce : SUCCESS in 58m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f05ef1a69c014c8bb1fe94dcf1622e15 : SUCCESS in 1h 26m 19s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8b175bcfb0824a3e82707946a747cf96 : FAILURE in 21m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50dd24e27cef4ffa951afe81a93acf05 : SUCCESS in 1h 52m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99687e0689d24b6786c6c03808884599 : SUCCESS in 1h 04m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/337ddf0833e742eb807e6e75c7577f25 : SUCCESS in 1h 07m 06s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8951b72879c04379bfe60370c12fe98b : SUCCESS in 30m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/777033bb5d6d4c63999557eca5b527b2 : SUCCESS in 1h 16m 14s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c11677a9e6914e9bae873968e14ef191 : SUCCESS in 1h 21m 26s","accounts_in_message":[],"_revision_number":6},{"id":"e88c20b2a4bf2f69e106b74b04b875c23661dd75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-03 07:44:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a86dcd9f6991aba6b47c87338c4df5c224b1bc04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 08:33:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35cfb76bfbb049e49d1db85e30cb16d3 : SUCCESS in 42m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6405ecc57adc44cea82112097dff1f11 : SUCCESS in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7c5ca0031cb9da53e840b063315c2b21d54acf10","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 09:28:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/7 : SUCCESS in 1h 43m 22s","accounts_in_message":[],"_revision_number":7},{"id":"cc4da121564a142de0b6a6e7b7d3f0effc8f1c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 09:48:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5886efff20446df9a8ae6d8bbf2574e : SUCCESS in 38m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b5f98c549ded45419744ad2cc5066805 : SUCCESS in 1h 46m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/74405671d32c4e499530121e9828aa97 : SUCCESS in 2h 02m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/59fa8d96508144b59d0e95cb0854c471 : SUCCESS in 1h 03m 49s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9f668bd2f7ea401ea8ceb03e58ae8bcb : SUCCESS in 1h 54m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f99b2fd141fd4f68beaee09cbd1ed31f : SUCCESS in 39m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31d67431295d43ab82646996545b3950 : FAILURE in 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0d7cea58424ab7942e36ecf0e736ea : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae0feffbb3494b46a703a116a333968c : SUCCESS in 32m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abfcfa9371fb4813a00d2d1a1059e2ad : SUCCESS in 23m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f855ee36f3a45bb8691bf51d59d6994 : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f83404dbdc4316928d5d01cd5280c1 : SUCCESS in 8m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e02f168ca3d4886b443f4c8b1c42392 : SUCCESS in 41m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f00c8f29fd714f42b07d41dc0125c68c : SUCCESS in 1h 20m 16s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c1d3a1ad86f54c288b28d223aebb74fe : FAILURE in 1h 32m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed871a4a832643b489757c1ec25550af : SUCCESS in 58m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/866a68658da947a0a36374eff83efaea : SUCCESS in 1h 02m 46s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c731c0ac455f484387eef2f3e3e0e76b : SUCCESS in 1h 17m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c02586558ec846998ba7d68897500be0 : FAILURE in 24m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e663e450e7b6409db3c1c98d8b8532cb : SUCCESS in 1h 46m 48s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd4b283ebbf04865bd92909a2e0c9d32 : SUCCESS in 1h 14m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f90f48bced0c4d27aeaefe6dfca1d774 : SUCCESS in 1h 09m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/076db3c2656c400386b00b173145c605 : SUCCESS in 29m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c8c3f4c1b2844c2e863481df2cac4bf3 : SUCCESS in 1h 24m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/85d07290e4ea4c7fbdbe307f30e81dea : FAILURE in 1h 43m 24s","accounts_in_message":[],"_revision_number":7},{"id":"d75a7bdb3f4e56c972f17e6a254681ea2087dc04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-03 18:41:20.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"9e8509d83ca0318e95fb0afdba41dbf1618c36f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:29:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cba45b5041eb4ca0a0678bf0c585bd16 : SUCCESS in 41m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0ada950c53d484cb211f85e7b68231d : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"757f2352018cb584ea510a26e6f0b482a3d3905a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 22:34:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9fc88e38244f46e3b5b74d46268e3067 : SUCCESS in 43m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3bf384bb58524b05814935e58df3909d : POST_FAILURE in 1h 52m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/070805829787406dbe99dd1cf42a1fa6 : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/797d9d617b7e4a1dbcab0cf14454286d : SUCCESS in 1h 28m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2d2c7543148a443394e5afb7b13a8c87 : TIMED_OUT in 2h 51m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d22e4209366d4d64998a67256c2f7f49 : SUCCESS in 39m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98aca0380091470eae23958526eebe1c : SUCCESS in 39m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1a3272b39a4d43977fd46de280fd9e : FAILURE in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51215cfae9ce43cda70a3ee2de46497e : SUCCESS in 25m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9524f712a2e46cdbd939c1069a3a594 : SUCCESS in 32m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5205c83352e14ba18f3b91337d17e3da : SUCCESS in 33m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a5e651017843e69d86a4dd1a35150c : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7370265087f4db0945b4da1137ca440 : SUCCESS in 49m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/be9846170e0f4cc2849bdbae4fc40c12 : SUCCESS in 1h 29m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c665516e3aec4101af851f6dee208c8b : SUCCESS in 1h 32m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c952f1204bd3425d9ed3b8a3e3a86310 : SUCCESS in 49m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d7e749fd3614e79bf63fa2ab8fc8600 : SUCCESS in 1h 02m 12s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86fdf982253f4bb0be9eee2efa7c84a1 : SUCCESS in 1h 19m 12s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/32f82599cdbd46f8bb952911a54665b5 : FAILURE in 27m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b350fccea04441658675d38b29bd06db : SUCCESS in 1h 47m 36s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/524519b3ddd64718a9f13b011de6537c : SUCCESS in 1h 07m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8e00ff69756548d0a44f57ae6c871c7e : SUCCESS in 1h 23m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f29e265df2ee4577a9efd49b4f584db4 : SUCCESS in 35m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90a5334333de4e54bed9d70bbc80a06d : SUCCESS in 1h 09m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5d90356481474818b174d0366364a788 : SUCCESS in 1h 08m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/bridge_lib.py\n  Comments left for invalid file neutron/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py","accounts_in_message":[],"_revision_number":8},{"id":"2c5c9bb19484c7c015048592c4bc8b3bc0cf803d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-03 23:48:42.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/8 : SUCCESS in 1h 39m 55s","accounts_in_message":[],"_revision_number":8},{"id":"d0d1ae33d09cbb51cf64a8f1e74b42d8a768e564","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-04 07:46:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8f3b7f97db288f46e4b6e4a9c0545f994a3b38e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-04 21:14:51.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/44/674044/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"8957a912ac8f7cac9386fb8e67a0cfd1b59bcfb7","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-05 04:59:42.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"dc498c914df3e79b0965e087ee973f6b2779012d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 06:09:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c96928f44194587b48cf0c694fd2a72 : SUCCESS in 1h 00m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51211edd722146deb349c7a06860ab6c : TIMED_OUT in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c450e581ab1085436b1a4320338b500979d9c820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 06:10:54.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/202edd825c61406ea2757ba6c2f657c7 : SUCCESS in 40m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779f503852834365b73f3ec973cebb00 : SUCCESS in 32m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eddf359057745d69809db2be9cee678 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49680eac1e6d4178b23a03ef26cdfe4b : SUCCESS in 35m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a70b537ddf5f4e1c8a61ff5e0cfa4c55 : SUCCESS in 33m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a64b4fa6baa46d7bf64ea07ece094ab : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c7aeb52039494b87729ce1b4243715 : POST_FAILURE in 15m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb8f5126a4af4b8995b05547c7ec132f : SUCCESS in 52m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e5bfd6bf4f764b3e929fe6593cbcc779 : SUCCESS in 1h 05m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fbf1375ff8578b2cd5bf77b2718a8f1fa973a8f2","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-05 07:05:30.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"09a49ac7bc0de93e69e5e715d7cad72175513ec9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-05 07:11:46.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/674044/8 : FAILURE in 2h 11m 28s","accounts_in_message":[],"_revision_number":8},{"id":"0fe1387a00d1b39319ab5606cd4ace22504a1451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 07:56:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29ecb76dd6be477cbac70c5098b87c93 : SUCCESS in 43m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1682a27572fc4d57b3864d4463e9941f : SUCCESS in 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d1357d0970d99dd15959f150cfce43b384e590e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 08:02:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e4cafdac6a7477ab037d75dc7f3cbe1 : SUCCESS in 37m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02c37a344c444a1c80ab8cca94f20583 : SUCCESS in 30m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c414a5062ef4e388be7a947f3782ffb : SUCCESS in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a392136bf7b4deda987fcdc551b4aa4 : SUCCESS in 32m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9771ed2dba1f44829689bab50bfda010 : SUCCESS in 22m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/790dece87ec94cd98992582768d1da8c : SUCCESS in 32m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b9743c94d641968dbff757466d700c : SUCCESS in 9m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/758946b800c8452d827460567bd975c4 : SUCCESS in 54m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c200702eaa8241419a752f080a5173d8 : SUCCESS in 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4a835777e604d8aa14cb98ccb71a0b709dfd36cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-07 09:23:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5790ed3d37a058d76d29db131c0cd8311b7bc2a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 10:13:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/254567fb0f2c4970b836a4d5f248e91a : SUCCESS in 45m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e690a11073274e6188cbc64d9333c2a4 : SUCCESS in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ae87c3256b8278dc165a40714b1ce9cfd1203c34","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-07 10:59:51.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/9 : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":9},{"id":"f0bc161112aef191c7d6f2db88fbfe8b34506651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 12:19:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8558c283ce147bdb5013f4c496a6f80 : SUCCESS in 41m 35s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c327394e04234846b78128235781a360 : SUCCESS in 1h 49m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/854ee0f8ee4f46b6a905ade15f5d359a : SUCCESS in 2h 05m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e88fa37db8a44963a340d0c69d02b4da : SUCCESS in 1h 39m 34s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/18c9ea41b29e41fe89f0ef440f69b7b6 : SUCCESS in 2h 49m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb2a5a2f2e274f048da20f05419ca033 : SUCCESS in 30m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be591e91f477442aad5e441e4219094e : SUCCESS in 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24b445939f947169c2cea731fcda15a : SUCCESS in 12m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6087f4a57ad5408ba8f1f0d6df00cb75 : SUCCESS in 25m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dca33a2fb6c47f88805910847c6d684 : SUCCESS in 29m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd9986ebe5174f96955687efe1b549a0 : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de871a5daa3f4c2495fd677f10ffe0f9 : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/98bb0c68e45e4ab3960c9a790805f11e : SUCCESS in 48m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/53c728e57dd44f8d94fa2e3a9c9df540 : SUCCESS in 1h 24m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f41c7054b584c20958e61928bb69592 : SUCCESS in 1h 17m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d881ae844e743299a6a069432a97803 : SUCCESS in 58m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5575b3d1d1954d26bff295a1ba3ba9e4 : SUCCESS in 1h 02m 21s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0f5c7929c903401aa9a8a53371df2247 : SUCCESS in 1h 18m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a069201c35944435bef43fe59a36db56 : FAILURE in 26m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9333602023fd44a28a262c178fc441ce : SUCCESS in 1h 37m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d3dcf1d3445436d9bbbe4ab044bbca8 : SUCCESS in 1h 02m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4a14303847c94a60ac4a33d6ea629fc4 : SUCCESS in 1h 02m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ffbf835bc4f94b079c2cfa0b82b6fc7a : SUCCESS in 33m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0db581783094be092a3b666a93d89e1 : SUCCESS in 1h 15m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3fce59ceaa944cac94a5984fb54dafe6 : SUCCESS in 1h 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"861d582f4244bfc0ae0f5984f0a99d9969e5e8b2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-07 12:21:00.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"428ff2d6e8a63e8e963bda0f182af9cac5a7c4d3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:11:00.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"4693ace4c65099199a99dae33db722f1a1ea6e65","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 05:36:33.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ Failed to upload log files ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"a0475c9b4af55f3cb3abab64146cf48b03ee2c5a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-08 07:47:28.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/44/674044/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"a74888fb4bf44492e3a1245784e14083eefa23d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-16 09:33:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"df918acbf6c83d235b71fac0b301fd8be36e5e0b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-16 12:07:08.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/10 : SUCCESS in 1h 58m 28s","accounts_in_message":[],"_revision_number":10},{"id":"f66c080092d7c3c7ab4be407a3e9504264542b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:00:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8a88ef695f294ad4b414bab009e8d4ec : SUCCESS in 44m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a63a87094c91462eb54f6f516f7f39f1 : SUCCESS in 1h 42m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e3495a2a11ab4e07a666ccd7551a033d : SUCCESS in 1h 42m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/acc7233fee8941608835806867c3cd6b : SUCCESS in 1h 22m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6228a7b804af467f80c48b2424bb0ae7 : SUCCESS in 1h 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/206a8fe01ec041229154f35427c157f1 : SUCCESS in 32m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69fddab697f042ae975f670f3e33eead : SUCCESS in 33m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4691d3019347869e7be30b518a9e05 : SUCCESS in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb939e265023422fb5154d31f25fb46d : SUCCESS in 25m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43c143085859414985f49b18c11ee4f1 : SUCCESS in 25m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1098bfd821204b86ba3941b0b955ea57 : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6768404c414df9a62edbd900e07c20 : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b74c8253fb04fc486658e4a41aeb0de : SUCCESS in 53m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3dcb95fc5efa4c11924fd7227b589e8e : SUCCESS in 1h 28m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/60d4d2e52909428598ee04fe2d7bd472 : SUCCESS in 1h 19m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b92dd35a790d4b36a52cc26b66ec872a : SUCCESS in 1h 03m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0c108e04801d4cd1891c45b8774439fc : FAILURE in 44m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f67e9c5cd6c649e682a08a280f6ce6cf : SUCCESS in 1h 17m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/79229d16b6a44b319f7b2134f3c99849 : SUCCESS in 1h 16m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55f16d2f256c4b3789abf13457babb0a : SUCCESS in 1h 24m 26s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/683dec29b58349b2a142e7ebd5e545ec : SUCCESS in 57m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c83e9d2f6f44605ab8b960ba2f9249a : SUCCESS in 57m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/b8843ad118404a59acfc91fdae045815 : SUCCESS in 34m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1cf3596d5c44402fa2d43fdd77f27b0f : SUCCESS in 1h 06m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/64092e2942ec456586751708869ae533 : SUCCESS in 1h 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ab8b95ea686285c26a2e7a6f76953a83a252604e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 21:02:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d69dd11647dc47cd82b4f04700c32531 : SUCCESS in 59m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9be4c6a2f34748719d04ba4ddda35e82 : SUCCESS in 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"edfb8b81d6e6268bdfa59ba934875924c8b66a36","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-17 08:08:55.000000000","message":"Patch Set 10: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"c166700b30724cf2e31916a91407cc21c6810232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-17 08:40:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"075080366c0e61d7f8a84790e9a4e23532694c41","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-17 08:40:36.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"466ce934f9f0bb0d4cd1b5cc1b8dc8082ba4753a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 09:20:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e56483af0a3c42489a6a4e0d89adc305 : SUCCESS in 38m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a08822773b88445a8ec0153a02638600 : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"18cfb2c263fbcb54c6704762f6836fc8e7eac095","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-17 09:24:46.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b737ad002950e586313eb230e62f51931df8d21c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-17 10:31:55.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/11 : SUCCESS in 1h 50m 48s","accounts_in_message":[],"_revision_number":11},{"id":"80bf82349073032d1f319b2f3b529720d100c2fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:03:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7683b03c043a483dbf610d860bc99e66 : SUCCESS in 41m 31s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ea4f23ce71af4309af25dce992d6c7e2 : SUCCESS in 2h 17m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b2255446475b42f0850dfd68186775fb : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66718b9c50c845bbb6a9a6a92e08fda6 : SUCCESS in 1h 31m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/754ae9f666f444769df78fdc630b854b : SUCCESS in 1h 49m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/204d9a5627c34aa5944b3439679185a1 : SUCCESS in 35m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca1813e2d2d54f74a53f01075807dcf1 : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e94bb28e1754c2d8c1528415648651f : SUCCESS in 15m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d92f2c50a8c467fa2b61efdcdf1e5f9 : SUCCESS in 26m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12c6bc66a814467aacf05dd3751e1c4e : SUCCESS in 24m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/819d1bf06e074c46975ab90e2124f180 : SUCCESS in 24m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b60ee2b6a54e5f839f57a1777c276f : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82a1179906204219bd283a3bf4fb41ca : SUCCESS in 46m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/971f2a5f7f74435cb9f10bdbaee12d34 : SUCCESS in 1h 02m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/074b7cda48f1493ea73e3db033efeb20 : SUCCESS in 1h 09m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d29950492379477fb4ed35c157ef5121 : SUCCESS in 1h 05m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0d7f3f1e55dc41bfa4ac75d458288dc0 : SUCCESS in 1h 21m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3b3d30b2b3e944c8bf3b6f3bcb4ccf15 : SUCCESS in 1h 16m 21s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bd838488cc984696bfc3a16b82f27422 : SUCCESS in 1h 15m 00s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbc21cb1b1cf44f18f0b2bc4af420b33 : SUCCESS in 1h 30m 21s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de7834d751fc4e4a816a691178991672 : SUCCESS in 47m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/72c45d22a1e94be386a91bce590716e7 : SUCCESS in 54m 21s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/2aea5f67b0c849c784881df0a23127c4 : SUCCESS in 26m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cefbf9ff82d745edadd7d68fd7a2420a : SUCCESS in 1h 06m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7c0455278b634bdc828f5c2a1d1689a4 : SUCCESS in 1h 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"48650c81f7240f43cb5f672ea6ff8d928ab05ae6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-17 23:28:36.000000000","message":"Patch Set 11:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/44/674044/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"5fc1cb074aae09305557b767f5faec7a7258edda","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-06-18 03:31:32.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"cb3e15099a5a0080c1f82b5a410b141140a46acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-18 12:09:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cea4dfc281512e17e781a048d1cb3cff8f955fb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:44:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dd82c732d214762b71fedf669189a20 : SUCCESS in 29m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11ae9decd5f74f7cabdc924a69e6995c : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4334dbcfa50b594c81162d3b675f43e60af11fe5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-18 14:06:49.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/12 : SUCCESS in 1h 56m 04s","accounts_in_message":[],"_revision_number":12},{"id":"2a6c77817a1aaad33d5e65ed6c66b1de5eae53f8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2021-06-18 14:14:02.000000000","message":"Patch Set 12:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2021/Arista_Tempest_Testrefs/changes/44/674044/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"ed56b91ed3bb20d1a599bac2838d3b39512407af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 15:49:13.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/447cc96a75a34ffb8d5c4fb75a59b429 : SUCCESS in 50m 58s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3491e93c245e4ce8ac8ce233d2f69142 : SUCCESS in 1h 53m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c57859fcd2a4a6f8d1117617e779431 : SUCCESS in 1h 45m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01d1c173e2b942ff96db85497f627b77 : SUCCESS in 1h 03m 27s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/99a2c68985854366a4e5fc6f7132d725 : SUCCESS in 1h 34m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162c54e7330c4cccb0f96a5ff995b0e5 : SUCCESS in 31m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4670613a094c4ba1b2629193ed185020 : SUCCESS in 42m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3592f9aafe4bb7bdcb35ef58aeba1a : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07892f4e247147e6b80fe6b1c000f5a4 : SUCCESS in 27m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dc1ebb9827849c59db6c22eca706b9e : SUCCESS in 24m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9eb06078a7a4f8fbf88a9fc348e56a8 : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa4253b766a47c0be60c418b5fbc4b1 : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4916a3205092445684480d98b768a6e1 : FAILURE in 47m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb8817580ad94fcebd86d7a6fbaefde8 : SUCCESS in 1h 22m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5f8bc7edcad5469682fbed58b35c95ba : SUCCESS in 1h 09m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d1d16ddc18a478ba8dff4b02060961e : SUCCESS in 1h 02m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4e2ac6c0eeb642a0b0c3661418854190 : SUCCESS in 1h 05m 19s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/af7bf4035f20426fa9f69761af1bd673 : SUCCESS in 1h 10m 27s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6f917beebdfa428ea900d1d5006605c9 : SUCCESS in 1h 22m 53s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98c048935a8f4ab0aac8d4ad68836139 : SUCCESS in 1h 25m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ceb6d2c71144741b575f5e2602198ff : SUCCESS in 52m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25be8dbeba2d40dfa472d7300b319ce0 : SUCCESS in 59m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e6aa5126e1474125a2e29dd41fc00aa8 : SUCCESS in 33m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99e370a8def844afb3d65fe0a1b0a30e : SUCCESS in 1h 13m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/86a9dc9b7f914ed3a359cf8ea6a4b74e : SUCCESS in 1h 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4ad4e22af25bfea238ca8b458fe8363845b80fbd","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-22 03:58:33.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ca862efda8661eb05c28ffabb003c68e74888f7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:42:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74a4c8867c7348f1a40e3570d8eb4b4c : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/227b5b9dd7ae45c2b5fe4960a52eda88 : SUCCESS in 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bef69fca4b08816312e25094594a54010e98268e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 05:33:22.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/674044/12 : FAILURE in 1h 34m 02s","accounts_in_message":[],"_revision_number":12},{"id":"efb49bbf603fd8b6d66c027726eaed39054e12f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 06:33:01.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ddf324dae8d947ef88dbc4c799b1f0ed : SUCCESS in 45m 13s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/36edb24c5ba04adcbc22b36705cada79 : SUCCESS in 1h 57m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d5c732b6578d4534895a5dbd1cb637fd : SUCCESS in 2h 05m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c96cc2a2e97b4886abc36f608b56d4c4 : SUCCESS in 1h 28m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/283965d77e5a43afb7b185e5d3363c34 : SUCCESS in 1h 47m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/841effa118dd40d6b61edba543475bf4 : SUCCESS in 38m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e989817279c4eca89d438a389525db0 : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f17f88bee044ac969e55eb1b6b316f : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ffdd412214d4178a5517d0ab78571bc : SUCCESS in 29m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a874841f2c849df85a1f4f2af9f0aa5 : SUCCESS in 28m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e79b0dfa849c49bd929a3721320a8f3a : SUCCESS in 30m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92502e490a8a4486b61f18893e3addc2 : SUCCESS in 12m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6bfe8e4d8c143d1898519ae0d4c287c : SUCCESS in 46m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7dad1f4a721a42e290f8559697863df2 : SUCCESS in 1h 40m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e43af948eefd4003b7e07119f0029ae8 : SUCCESS in 1h 11m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b10204fa282242d892e2281098ef590f : SUCCESS in 57m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/828b8ed9c76e48dfaf91eb445e225373 : SUCCESS in 59m 03s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/15202acf2d324273acb457b989ab48f1 : SUCCESS in 1h 09m 46s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e7a4b10c1fa4da99c7775746afdd954 : FAILURE in 2h 29m 15s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fdba39aaa7704f3f890f3a94572ea75f : SUCCESS in 1h 26m 15s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13e0cd7a6fc24412a51d0f4ba10a028b : SUCCESS in 54m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f4ee745188ba4fe286aacb7acd20c4f2 : SUCCESS in 57m 23s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9a9ffb46e1ab4937aaa8906dceddc80d : SUCCESS in 28m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05651bf511de448a95e61b2e0a4bb4e0 : SUCCESS in 1h 12m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/66606648df7f4bc0bb3ab1ca5e740566 : SUCCESS in 1h 08m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"318418a2ff0468e90ed5cfd0087ae9596a9cd10e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-22 07:26:36.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"be6fe2a917b5acef14fd874c7c5002448222525e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-22 08:40:01.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4ace3a3d73fe838f15346a67eb9f8fe6b6f91c0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:39:58.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e11928bc8d1c4ceea6ee3f9a0d1e98ce : SUCCESS in 37m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44f7b54381824a4eb086ddc8324d937c : SUCCESS in 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"74405ec7304fb13c97359447e0f26af84101af0f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 12:25:20.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/13 : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":13},{"id":"e52fd7f960007782d558bdc56cb4189cd788a496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 12:56:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e72f87a0f8b343b2b2930212d1d75e9f : LOST in 45m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e2fe16445b3745128421f4975cbc73b2 : SUCCESS in 1h 52m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f786a107ecf243768a056ca7d71c79d7 : SUCCESS in 2h 09m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8c66c9e6942482a9aafd80f5ed3ddf7 : SUCCESS in 1h 50m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e40f5ad618ce4e92bcf62e3d1d1af94b : SUCCESS in 2h 38m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2706de5b0b8944baa6084222718a8954 : SUCCESS in 40m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23e5eb73012e4461bca36c1e2976b6b0 : SUCCESS in 39m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32f9df7a2514c02b92b87286728aaf5 : SUCCESS in 14m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9995800e3c894202af37adddedf0eba5 : SUCCESS in 22m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1616dca3756465aa66bb8d8808b0534 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec96faf9e4c9427e873752487a2d5f06 : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b31aaffb48842c087f58fa5b23773bf : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcd8b231280c4db4af4abc0519ab9957 : SUCCESS in 47m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9334b78f63c4d51acbff5b2f2e1ef19 : FAILURE in 1h 12m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/37aaf2fa42ce432688144757c19e43bb : SUCCESS in 1h 15m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4c6bdce449b4fccad600f49ffe2dca3 : SUCCESS in 1h 06m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5632d0f8d12b4a91afb0e05b9c76ebab : SUCCESS in 58m 24s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c183fcb0150c4c3fb6077c85f686eec7 : SUCCESS in 1h 13m 25s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ecd0b0f9652442f2beabb98836671142 : LOST in 37m 25s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78b383496cbc4f4db5b55cda86481bbb : SUCCESS in 1h 31m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f40e412893074d218379ee3706be5857 : SUCCESS in 58m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f67affb795314b37b1f95ed21eb23f3b : SUCCESS in 1h 10m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a3d1ae7a39234c22a394ac7e531124b0 : SUCCESS in 29m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/729e4a5e3e1b4123918e0eaff15230f7 : SUCCESS in 1h 27m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9f86b5fb23f54867968837474c0f7247 : SUCCESS in 1h 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"eb2e699d12c122851908a9f734b91dc1cd7d1cee","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-22 13:05:01.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"f89e6ccd08b43d853d2be4c316e3b25a582356a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 13:48:11.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0304f9cb21874ec4ae8de0625c694293 : SUCCESS in 33m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd03ffc0533d4e749fa724b7c53d0671 : SUCCESS in 31m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7816a63af25e11b360c1c0f1b0e04281e808373c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 15:04:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/964acd1f5c3c4967b27cc24c74eb7547 : SUCCESS in 39m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6a82b43f20594ef7822cccbb08d2309c : SUCCESS in 1h 56m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c2ecce97dcb3457c864567a723e7d863 : SUCCESS in 1h 34m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7050cf550e9a401d8c18d6e07815c204 : SUCCESS in 1h 12m 07s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/27837e89417f4e48a5f4867f3d5654f6 : SUCCESS in 1h 40m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83bae71ab50414ab0e0978f57bebacc : SUCCESS in 28m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dca8d9d503c34fdf8393a57c2497255e : SUCCESS in 42m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce4506da5264c1a98f5bd08833bee54 : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07a983dc03224a6696537286293a9b1c : SUCCESS in 29m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f37732f2732e427e98c176276cbb4f3d : SUCCESS in 27m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18c52041901b4423928486e450137e38 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334a57986e4d463b8a5938690ec03467 : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fcbff0dd2924760bd9403283246f57d : SUCCESS in 52m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61905c45872a4407994d5ceb8db7a572 : SUCCESS in 1h 34m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/59189fea55e748889ab17ec6aa605027 : SUCCESS in 1h 16m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/691cf337d73a4e1685c06f822176bad6 : SUCCESS in 59m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d8aa9bcc9a0e4e0a915effbfb1d2a02e : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/87e73e27f6cd402cb1ffa0fdbddf639b : SUCCESS in 1h 12m 43s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c4588bc575c4fae91106a3553cc3244 : SUCCESS in 1h 34m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1812ad3a3fcd4fc9abb26f8bbf2e22d4 : SUCCESS in 1h 42m 47s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32c1a7bffd824b88b93c99a33d226040 : SUCCESS in 49m 55s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c0122955ae194397911eeb7654971e8a : SUCCESS in 56m 58s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0fa9a1f79e0d413aa5c2f18e06c25f11 : SUCCESS in 34m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efb0512bfca541e9969034659390bb5a : SUCCESS in 1h 25m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d755f929517c401ab44dbc30c26aab89 : SUCCESS in 1h 08m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1033ab4b5394fd17d42e47832c8ea671b1999e2a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-22 15:09:44.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"354bd7c43611d34dc042bb3af3a938ac1d0110f4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-22 15:42:49.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/13 : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":13},{"id":"c91e2a3b86fece589a4cc765be0a878c43b1253a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-23 21:17:05.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"a006da2a32b8b0ab89cadc78be19f14bfce1e1c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 21:53:12.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c88f732d9e724e709ee914016d59dff8 : SUCCESS in 30m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27e102e33b5d4273b514d2913152b4ad : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"394c5f802fd38d8d3f78fa0a926802bca3626f9c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-23 23:09:26.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/14 : SUCCESS in 1h 51m 35s","accounts_in_message":[],"_revision_number":14},{"id":"837701cf65ea2ed8c19c122244289665b86e7492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:18:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd3773aa765c4c28a6fce6c68aaa1216 : SUCCESS in 34m 47s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cd82b30205ca497bbccb92d4779db2d1 : SUCCESS in 2h 00m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/41f788cbce2f402fa68a6a5bc0a6d838 : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9e85f64e36544bb1bb936244d93af501 : SUCCESS in 1h 10m 10s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/76b54bcba62f4158aac30f98a6eae923 : SUCCESS in 1h 39m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb651a9638a4b458c180717cefe3dab : SUCCESS in 29m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5badfbf596148ceaf7faaf88af76b4a : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50042d35b27646778f0def449651953a : SUCCESS in 12m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d6edc8cf84e41beb5e7075ae47c8c7b : SUCCESS in 27m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db062daf0fc342c5afab88f5fcf42798 : SUCCESS in 20m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d380486a74c484b8fcd523fd29a1fa3 : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c22713881a499b9f6ffb2a3c97a9a6 : SUCCESS in 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17ee2e3b0c234f188f5ddc6081b8edfc : SUCCESS in 53m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/481f030de3f64b00981146d29bc5ce40 : SUCCESS in 1h 21m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4f2aa5a414e046b2b488473806eb5a05 : SUCCESS in 1h 13m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70420fc23f814d58a5932da14f5d25f7 : SUCCESS in 59m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0c983d595320410d869ccafc088f1d11 : SUCCESS in 58m 39s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ebfd9e969e56460f82d5434acea5df57 : SUCCESS in 1h 06m 26s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/da701ff35aa446cb90550a4795b02f31 : SUCCESS in 1h 21m 19s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/979f5cda457041a48276ddd25597fa8d : SUCCESS in 1h 24m 25s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ae83722725748a0a82b491da810fe70 : SUCCESS in 1h 04m 41s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/18e8e021d5da46a0914d007b0aed044e : SUCCESS in 1h 02m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1b853352ca0043d8b231b05bbad1eb70 : SUCCESS in 31m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7faa3a5cea884aefb17c074cdabea27f : SUCCESS in 1h 20m 11s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/42395ce918c24e93b92389f7e3dc93f7 : SUCCESS in 1h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5a812c45de2969e3d0ef257c6864b416d9243607","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-24 07:59:35.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"c62056ada9bb97faf1f7023d8d3fedbe6ce0459e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-25 13:07:48.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"f6325dc5d8532fc1dc80be2ae87afc5fb79f6324","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 13:49:14.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5456c1b1f3b7404a9aaa2c711bc6c75b : SUCCESS in 33m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4ce9aa717e440d692082f6067815c52 : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ce6b4db9ffb324f15087d1938eee0f52ab2231d1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-25 14:54:12.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/674044/15 : SUCCESS in 1h 44m 24s","accounts_in_message":[],"_revision_number":15},{"id":"245b812aa97055e9476bb4561f3f1fcbdfe8c45e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 15:00:45.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/36b93cdf9f6f49439f22c7d035a0954a : SUCCESS in 41m 08s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ac9f65ce8eae427b87e16866c2e4df7c : SUCCESS in 1h 41m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4473d240c8614821b60dfd7c6b5d5643 : SUCCESS in 1h 47m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7658f613372a489a92e41d8d33c949b8 : SUCCESS in 1h 26m 35s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/093c1460ff5f4566a2123bb624a2741b : SUCCESS in 1h 50m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49bb617cc1c4473794adb46ce9f44ecd : SUCCESS in 30m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18fba20086b942d2ae5c58270428ecac : SUCCESS in 30m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e878c2778d6421ea94078dfdce6550b : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e6237d99e84d7dbeb166536395d77f : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07b5f9313a8e40af9a332b041e834404 : SUCCESS in 29m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbb5f8a4887248d39d5bf00c51220eb1 : SUCCESS in 22m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abf15f76580a496c9189e21a69fef79c : SUCCESS in 23m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cabe862ff2b44069c318d3b5220f8d9 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5e8eb8a21684cecae7cd481a745f40c : SUCCESS in 43m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b00e62cb157f48478aeed0425bef94f5 : FAILURE in 1h 10m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/055ad1b148574caaa57ce32f7d1cb9fc : SUCCESS in 1h 06m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2529b662fa0f4c3da473df4351b73ad1 : SUCCESS in 59m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eac2c8f113f3471da024dcca758d8343 : SUCCESS in 1h 12m 13s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/10736e1f83434add9fbd2e9cd775f822 : SUCCESS in 1h 07m 52s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d8ad3140dad419b8a9790c358df27ac : FAILURE in 1h 08m 15s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12e87d0b699f40c487df8764c4eb8c06 : SUCCESS in 1h 28m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1e7daff297643558d8170a62bd71f13 : SUCCESS in 55m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f4a1beda34a04a96ba4eedafce2b8711 : SUCCESS in 53m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/75ce6007d7964c8dad8b51d9b4088677 : SUCCESS in 34m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06c9cd1454544ce5b07a22fe25d31b83 : SUCCESS in 1h 08m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/903c479a9b2d4ac6a50372684ff1c422 : SUCCESS in 1h 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d90d54038788976f94e6b540de545111bd93be80","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-06-25 15:04:39.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"32f1e5db23a8db557a2ff71311a21918916d9de7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 15:43:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/311b4431a3cf4665b9188b33184adaf2 : SUCCESS in 31m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/015fb5b6f209420894dc2dfd38fc2e30 : SUCCESS in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"16e713b356c27026e0964219dab050875c0d1151","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2021-06-25 16:49:40.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/674044/15 : FAILURE in 1h 43m 57s","accounts_in_message":[],"_revision_number":15},{"id":"faf6761bfcacfaa0159c53dcfcc7b3e533ea6778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 17:20:40.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/420feeac7a9c4f858fa82d8257819d92 : SUCCESS in 43m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/750c3a67cf234167b64a30d2aa513d94 : SUCCESS in 1h 39m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1cf21375f83b4e69a58a3207cd842134 : SUCCESS in 1h 26m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe89bbf2e7734f81a3cf5cfb907369e4 : SUCCESS in 1h 08m 37s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b16aa56bc52b48bcb98836234addb54e : SUCCESS in 2h 13m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ade1dbf172cc43cdb93b6da75eed841f : SUCCESS in 29m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3900e8356d6b429e86036f00d2021653 : SUCCESS in 31m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c08bf96d654a1dba38599ea30a39c1 : SUCCESS in 51m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fdc92eeee4d4b4c9732406a603144aa : SUCCESS in 15m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20d755f1b09b4c89b9ef6f4da629ea1e : SUCCESS in 25m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a6fdebc4d54bc7906f62d2798f0aaf : SUCCESS in 22m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3460bb0162f74404b566e55902fdaad8 : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52dbc4daf304aeaa9de604cc7081787 : SUCCESS in 11m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2db8f9acc5574065bd8be2b31c1d50fa : SUCCESS in 50m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/07cf20ebfd864b339b2d8160950f8440 : SUCCESS in 1h 21m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2c09833e78a040209a566cd83d6cf6b0 : SUCCESS in 1h 09m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1dbac38b35841a3a873b2c21b5e0ea1 : SUCCESS in 1h 17m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/824b48ebc7884c42970396cdedaaf6ab : SUCCESS in 59m 41s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/33fb8545af3647de98b575cf769ac2b3 : SUCCESS in 1h 07m 42s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9314bc92416f4cbbb8754045a858808b : FAILURE in 1h 11m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/61c67d44da154fedb9a4993f509ff0cb : SUCCESS in 1h 22m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee7774a360e844fc96f8a356812451a3 : SUCCESS in 53m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/fa4c173a928c4929be513277714253db : SUCCESS in 56m 05s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ab94698efc8a4b0cbb8359d14f63f98a : SUCCESS in 28m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/72829b33da6b4a46a1666458f10124d3 : SUCCESS in 1h 09m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9cd42c7bff354fa19e0ae4aed08cde76 : SUCCESS in 1h 09m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"251e0bb9a7fc7b5c9d8982007a705298eb5ab9bb","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-06-28 07:19:30.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"ca0cd9ba6ee72b62c665d75669aa6063cbfb11ea","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-28 09:07:58.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThese are used in networking-bigswitch (and some other project from x/) but those are not really active as I see","accounts_in_message":[],"_revision_number":15},{"id":"705bf8285833faecc258dea2f45f447c16e0f316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 09:08:27.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"355842aa1b24613e54568bebffb07562a4953e80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:18:37.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/178e08d1870f49efa31bc986032485ae : SUCCESS in 46m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d844db62d50c4d719d04d28d0e0b7cdc : SUCCESS in 2h 04m 39s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/172ba7e6b90e4182a0c32a52d363c2be : SUCCESS in 1h 53m 11s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/5f11d2504e39482fb358c32c91fba9e9 : SUCCESS in 1h 33m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bad449f8b09246a6b416477d367736dd : SUCCESS in 1h 25m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87bf86db32bb426693e76dd7927571e7 : SUCCESS in 39m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458def08c7d2498e9d59ab76d7e183b4 : SUCCESS in 16m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2773aa94dcf4dbdb4e442ef14b6882c : SUCCESS in 27m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da6340bd42b1443c91016a0291727bf2 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd02a262c02463287cf10e353c37164 : SUCCESS in 12m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c54fb2afc9d4aedab0c3e9bb3f618c3 : SUCCESS in 50m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3fe095d0dce540948cb08ec4f52fd5f2 : SUCCESS in 1h 25m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0539d7e9a5ed43cdaf62868e32a5aeb8 : SUCCESS in 1h 12m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebff8e99590c4bb7ab6b60c4dfdea07e : SUCCESS in 1h 05m 11s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/897055d57dfb44ef81fbfa41fa2b43f2 : SUCCESS in 1h 33m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da4aaef37fc44a79bfe99616ca77864d : SUCCESS in 1h 00m 13s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dd45b65b0e6e47bc93484ab0f9c0f0d7 : SUCCESS in 59m 14s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a9b2875a526449539bbbef57c4edab8f : SUCCESS in 35m 19s","accounts_in_message":[],"_revision_number":15},{"id":"b2598924c56b95a5e6c41135f27a530848a83213","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:18:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"8b782334236ae2e93a75b0e7f6e16e39205d4257","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 11:20:48.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51331903502347b88cb3393982fc4b46 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"b05a9186d1246460b21d3db26d5e7e70717e2d77","revisions":{"ab1db8ecef50e789f1373e3bd7996fd17dd5f19f":{"kind":"REWORK","_number":1,"created":"2019-08-01 14:14:44.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/44/674044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/1"}}},"commit":{"parents":[{"commit":"1421c63f4b48c2b0e8965d68c20e439638efcfa5","subject":"Merge \"Add RPC method to get networks for L3 and DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1421c63f4b48c2b0e8965d68c20e439638efcfa5"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab1db8ecef50e789f1373e3bd7996fd17dd5f19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab1db8ecef50e789f1373e3bd7996fd17dd5f19f"}]},"branch":"refs/heads/master"},"00cc8bc493aa86156d369322d5755ac11de7c4c7":{"kind":"REWORK","_number":2,"created":"2021-05-18 10:54:00.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/2"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-18 10:50:53.000000000","tz":180},"subject":"[WIP]use callback payloads for SECURITY_GROUP","message":"[WIP]use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00cc8bc493aa86156d369322d5755ac11de7c4c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00cc8bc493aa86156d369322d5755ac11de7c4c7"}]},"branch":"refs/heads/master"},"c365ff8d4b07152e252d65065758b5c1bd640ee3":{"kind":"REWORK","_number":3,"created":"2021-05-19 14:17:21.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/3"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-19 11:55:21.000000000","tz":180},"subject":"[WIP]use callback payloads for SECURITY_GROUP","message":"[WIP]use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c365ff8d4b07152e252d65065758b5c1bd640ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c365ff8d4b07152e252d65065758b5c1bd640ee3"}]},"branch":"refs/heads/master"},"d45fe01d0842a3649644c368f427b9221cfc505f":{"kind":"REWORK","_number":4,"created":"2021-05-22 07:42:01.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/4"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-22 07:41:39.000000000","tz":180},"subject":"[WIP]use callback payloads for SECURITY_GROUP","message":"[WIP]use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d45fe01d0842a3649644c368f427b9221cfc505f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d45fe01d0842a3649644c368f427b9221cfc505f"}]},"branch":"refs/heads/master"},"e3d604dd297922e6b059ad049b3f9df4e86e501b":{"kind":"REWORK","_number":5,"created":"2021-05-23 06:10:46.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/5"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-23 06:10:06.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3d604dd297922e6b059ad049b3f9df4e86e501b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3d604dd297922e6b059ad049b3f9df4e86e501b"}]},"branch":"refs/heads/master"},"7d553274969b057c24643fd368111c066253396e":{"kind":"REWORK","_number":6,"created":"2021-05-25 13:37:08.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/6"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-25 13:36:40.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d553274969b057c24643fd368111c066253396e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d553274969b057c24643fd368111c066253396e"}]},"branch":"refs/heads/master"},"ed46c6d30b3b05eb9d0e73ff3558816ef107dd2b":{"kind":"REWORK","_number":7,"created":"2021-06-03 07:44:12.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/7"}}},"commit":{"parents":[{"commit":"c12588c7be7ef6990284a5161975c2b81ff30f03","subject":"Merge \"Fix wrong assertion methods in a unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c12588c7be7ef6990284a5161975c2b81ff30f03"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-05-27 11:11:16.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed46c6d30b3b05eb9d0e73ff3558816ef107dd2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed46c6d30b3b05eb9d0e73ff3558816ef107dd2b"}]},"branch":"refs/heads/master"},"ffbb586dbfd679921eadcde357860632f2eb2581":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-03 18:41:20.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/8"}}},"commit":{"parents":[{"commit":"7e98d189270c408b4a5f5b69e9e479a458407b22","subject":"Merge \"Switch neutron-rally-task job to ML2/OVS and skip \"test_models_sync\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e98d189270c408b4a5f5b69e9e479a458407b22"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-03 18:41:20.000000000","tz":0},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffbb586dbfd679921eadcde357860632f2eb2581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffbb586dbfd679921eadcde357860632f2eb2581"}]},"branch":"refs/heads/master","description":"Rebase"},"7a45f853a994f5e92624d26d0bda25be2d556d8c":{"kind":"REWORK","_number":9,"created":"2021-06-07 09:23:51.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/9"}}},"commit":{"parents":[{"commit":"c6ebae439569a5f3aad33ebaa3119b87abb050f9","subject":"Merge \"Use payloads for ADDRESS_GROUP callback events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6ebae439569a5f3aad33ebaa3119b87abb050f9"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-07 09:23:14.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a45f853a994f5e92624d26d0bda25be2d556d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a45f853a994f5e92624d26d0bda25be2d556d8c"}]},"branch":"refs/heads/master"},"d85380b666fa7333a86e6d7138b6ea87cf690859":{"kind":"REWORK","_number":10,"created":"2021-06-16 09:33:06.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/10"}}},"commit":{"parents":[{"commit":"87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c","subject":"Merge \"Add support for OVN allow-stateless ACLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-16 09:30:44.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nNeutronLibImpact\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d85380b666fa7333a86e6d7138b6ea87cf690859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d85380b666fa7333a86e6d7138b6ea87cf690859"}]},"branch":"refs/heads/master"},"02e5c756316055bac0aec52b9905e4fe01841936":{"kind":"REWORK","_number":11,"created":"2021-06-17 08:40:22.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/11"}}},"commit":{"parents":[{"commit":"87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c","subject":"Merge \"Add support for OVN allow-stateless ACLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-17 08:37:37.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02e5c756316055bac0aec52b9905e4fe01841936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02e5c756316055bac0aec52b9905e4fe01841936"}]},"branch":"refs/heads/master"},"1eb8a211f49adcf8ee3772893d1f636b367ed487":{"kind":"REWORK","_number":12,"created":"2021-06-18 12:09:43.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/12"}}},"commit":{"parents":[{"commit":"87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c","subject":"Merge \"Add support for OVN allow-stateless ACLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-18 12:08:59.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1eb8a211f49adcf8ee3772893d1f636b367ed487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1eb8a211f49adcf8ee3772893d1f636b367ed487"}]},"branch":"refs/heads/master"},"d06ccef290d63a655f0243153f9c36ad8a64a2d6":{"kind":"REWORK","_number":13,"created":"2021-06-22 08:40:01.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/13"}}},"commit":{"parents":[{"commit":"87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c","subject":"Merge \"Add support for OVN allow-stateless ACLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87f7abb86cad13c8bc04b4e6165600ee6fd9ef7c"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-22 08:39:42.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d06ccef290d63a655f0243153f9c36ad8a64a2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d06ccef290d63a655f0243153f9c36ad8a64a2d6"}]},"branch":"refs/heads/master"},"e12d6ade646aa0ea719718074fd43219b312002b":{"kind":"REWORK","_number":14,"created":"2021-06-23 21:17:05.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/14"}}},"commit":{"parents":[{"commit":"9a760b9b14802eef015a5275853ec88a11d0b8ed","subject":"Merge \"[ovn] Clean-up unused ACL method for DHCP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a760b9b14802eef015a5275853ec88a11d0b8ed"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-23 21:16:48.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e12d6ade646aa0ea719718074fd43219b312002b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e12d6ade646aa0ea719718074fd43219b312002b"}]},"branch":"refs/heads/master"},"b05a9186d1246460b21d3db26d5e7e70717e2d77":{"kind":"REWORK","_number":15,"created":"2021-06-25 13:07:48.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/44/674044/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/674044/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/674044/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/674044/15"}}},"commit":{"parents":[{"commit":"7fd6d169f52096da195cdeb79da09acf0718c9d2","subject":"Merge \"use callback payloads for SUBNET\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7fd6d169f52096da195cdeb79da09acf0718c9d2"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-08-01 14:14:34.000000000","tz":-360},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-06-25 13:07:18.000000000","tz":180},"subject":"use callback payloads for SECURITY_GROUP","message":"use callback payloads for SECURITY_GROUP\n\nThis patch switches over to callback payloads for SECURITY_GROUP\nevents. To do so a few shims are put into place the handle both\npayload and kwarg style callbacks; these shims will be removed once\nall events use payloads. In addition a few UT updates are included to\nget the tests working properly with payloads.\n\nChange-Id: I6161a8b387812808c4d679f882a3193c93235647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b05a9186d1246460b21d3db26d5e7e70717e2d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b05a9186d1246460b21d3db26d5e7e70717e2d77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
