)]}'
{"id":"openstack%2Fneutron~675839","triplet_id":"openstack%2Fneutron~stable%2Fstein~I9b07255988f7ba6609af1961b3429c3ce12d5186","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1839595-stable/stein","hashtags":[],"change_id":"I9b07255988f7ba6609af1961b3429c3ce12d5186","subject":"Fix sort issue in test_dhcp_agent_scheduler.test_filter_bindings","status":"MERGED","created":"2019-08-12 08:16:43.000000000","updated":"2019-08-13 00:21:54.000000000","submitted":"2019-08-13 00:21:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675839-1565655714750-997bfbc5","meta_rev_id":"2f1225ce4a0a1fec28489f5ab27810b3ce77115e","_number":675839,"virtual_id_number":675839,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-08-13 00:21:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-12 09:59:12.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-12 20:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-08-12 13:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-12 20:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 09:59:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-12 13:31:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-08-12 20:32:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-13 00:21:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b9540c07eeeab3d13a868b89b387e3523e1c27d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-12 08:16:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"96b3dfe6b93d0590633bd7127a93549c51e3ed96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 09:43:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/39/675839/1/check/openstack-tox-cover/d81410e/cover/ : SUCCESS in 30m 09s\n- openstack-tox-lower-constraints https://logs.opendev.org/39/675839/1/check/openstack-tox-lower-constraints/f196859/ : SUCCESS in 32m 00s\n- openstack-tox-pep8 https://logs.opendev.org/39/675839/1/check/openstack-tox-pep8/08e775a/ : SUCCESS in 14m 38s\n- openstack-tox-py27 https://logs.opendev.org/39/675839/1/check/openstack-tox-py27/f9eb2b5/ : SUCCESS in 27m 08s\n- openstack-tox-py35 https://logs.opendev.org/39/675839/1/check/openstack-tox-py35/f4fc34c/ : SUCCESS in 26m 51s\n- openstack-tox-py36 https://logs.opendev.org/39/675839/1/check/openstack-tox-py36/906eed4/ : SUCCESS in 32m 04s\n- openstack-tox-py37 https://logs.opendev.org/39/675839/1/check/openstack-tox-py37/d870063/ : SUCCESS in 22m 56s\n- openstack-tox-docs https://logs.opendev.org/39/675839/1/check/openstack-tox-docs/2850041/html/ : SUCCESS in 5m 00s\n- neutron-functional https://logs.opendev.org/39/675839/1/check/neutron-functional/3d9b372/ : SUCCESS in 47m 57s\n- neutron-functional-python27 https://logs.opendev.org/39/675839/1/check/neutron-functional-python27/8fb7f93/ : SUCCESS in 43m 31s\n- neutron-fullstack https://logs.opendev.org/39/675839/1/check/neutron-fullstack/dfaff9a/ : SUCCESS in 1h 15m 58s\n- openstacksdk-functional-devstack-networking https://logs.opendev.org/39/675839/1/check/openstacksdk-functional-devstack-networking/a1be322/ : SUCCESS in 1h 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d496617ed658d716b2253f6e1630aee029c39a34","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-12 09:59:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/675839/1 : SUCCESS in 1h 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"470afd1b6053956977d3e7eb9460391e207f8ca0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-12 13:31:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa2914892ace730447e202eef9d59e8e23d6b5c1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-12 20:32:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to master.","accounts_in_message":[],"_revision_number":1},{"id":"0d1aef6746c12fde19eb61bac3e7e2e7aefcbded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:32:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3278a138b5622b24e61a1e6688137b539c3c806c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:21:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd5387bcea81cc648d5ab90f98d7cf6e39b8e998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:21:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3661e8668e04e80b0ae854725474cd0 : SUCCESS in 31m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba2de1ab37843c68c2acf19ecde778c : SUCCESS in 13m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38afc5fa7e4f4bde8402d2b136f1b06b : SUCCESS in 22m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f858fb2f4cb2493ebb4518ef83cda7fc : SUCCESS in 24m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cef8511d7e63478b8d408f2a016008fd : SUCCESS in 31m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f8dfc0e1d5646138b240480dede6511 : SUCCESS in 28m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183af156daf4489495297bca6edadf61 : SUCCESS in 4m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/60af35452fe9492e8520d2551e081867 : SUCCESS in 1h 41m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cebe86f55f124f30984e58763f1eeae1 : SUCCESS in 1h 28m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffe3a807d5e64fe7a105db0feeec6bfe : SUCCESS in 44m 36s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/a40af81b0224422b832aa77c286096de : SUCCESS in 50m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f0ad69018089463398c974c92276890b : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"750a8e8327388225f9bea549a7de61741739e937","revisions":{"750a8e8327388225f9bea549a7de61741739e937":{"kind":"REWORK","_number":1,"created":"2019-08-12 08:16:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/39/675839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/675839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/675839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/675839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/675839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/675839/1"}}},"commit":{"parents":[{"commit":"2e15b18679d3363c73cdc764ec28fc3d02930618","subject":"Merge \"Check for agent restarted after checking for DVR port\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e15b18679d3363c73cdc764ec28fc3d02930618"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-08-09 08:26:13.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-08-12 08:16:43.000000000","tz":0},"subject":"Fix sort issue in test_dhcp_agent_scheduler.test_filter_bindings","message":"Fix sort issue in test_dhcp_agent_scheduler.test_filter_bindings\n\nThe test creates a list of networks, and then acts on a list of\nNetworkDhcpAgentBindings obtained from get_objects() not guaranteed to\nfollow the original build order (based on the network_ids list)\n\nMake sure that returned list is sorted on network_id, and network_ids\nitself sorted so both lists match\n\nChange-Id: I9b07255988f7ba6609af1961b3429c3ce12d5186\nCloses-Bug: #1839595\n(cherry picked from commit f59b6a470620eaf53c98f4d98801789c95ed9a26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/750a8e8327388225f9bea549a7de61741739e937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/750a8e8327388225f9bea549a7de61741739e937"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
