)]}'
{"id":"openstack%2Fneutron~677055","triplet_id":"openstack%2Fneutron~stable%2Frocky~I41c02b0164537c5b1c766feab8117cc88487bc77","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1840443","hashtags":[],"change_id":"I41c02b0164537c5b1c766feab8117cc88487bc77","subject":"Initialize phys bridges before setup_rpc","status":"MERGED","created":"2019-08-17 07:09:55.000000000","updated":"2019-08-19 11:11:39.000000000","submitted":"2019-08-19 11:11:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677055-1566213099107-25fc65c6","meta_rev_id":"0d65573443956a4d6a5a95f1c1b1e89e6ac3260b","_number":677055,"virtual_id_number":677055,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-08-19 11:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-18 00:25:40.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2019-08-19 08:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-17 23:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2019-08-19 08:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-17 23:23:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-08-18 00:25:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-19 08:44:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-19 11:11:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4aa36aead27a2cdc35bb569f1ab6d98bb69d83d3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-17 07:09:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44a986df7538fb6a5f28e9b333aa04e136ab196a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-17 07:10:13.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 6618a917d86add7ee5c1005a21046d1bc57ac95c","accounts_in_message":[],"_revision_number":1},{"id":"7ca8aab8db0b30a1e85a8faea87d45d7ba88a72b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-17 08:11:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677055/1 : FAILURE in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cc1eccfff782f35c64a0121c2bd5d6936ec0791e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 09:14:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/4efb572658aa42abbdd108d7039c1852 : SUCCESS in 50m 27s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f25e1426b8274f1f8ceeec1070f02ebd : SUCCESS in 41m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/85ec0345db1b4591900c6125e525470b : FAILURE in 1h 34m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/2c5abed38b3e4ac1b5ca988fb9ecd50f : POST_FAILURE in 1h 07m 12s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/a791f9e6371c49108b6ac54af202c011 : SUCCESS in 1h 01m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/d3fe3c73796042b2984852b9e48f1623 : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e7ddcceec01410bae21ba3d14c34773 : SUCCESS in 11m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3720ec7dfeba4439944aa8616bc561f2 : SUCCESS in 26m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d3707d5ce72e47cd8ed50ced6c0c8d69 : SUCCESS in 22m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/db625e7544c641c591600cf07d6d64c0 : SUCCESS in 8m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f5a65a43d7c7431f916e7752357e27d1 : SUCCESS in 1h 27m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4c93e8b50bc14507ab5f4f88ca7c0469 : SUCCESS in 57m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cec3cbc3dcd48ca923e43f9ad0b4c94 : SUCCESS in 1h 31m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7cb89614f138442d91db2c0679c6ff77 : SUCCESS in 41m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/17cd71a42aeb425b993d910a800705ff : SUCCESS in 57m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2a65ce7d184447d89a432981c18128a0 : SUCCESS in 1h 46m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fa5ecc4f5a1e4e4ebaf28d27deabfc1a : SUCCESS in 1h 33m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bce563f36abc480590665c83ccfaa34c : SUCCESS in 1h 34m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ace5240aa6c24809bed635538d52f74a : SUCCESS in 1h 19m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa2c8b4ecd6f4ba3913101dadcb44407 : SUCCESS in 1h 09m 12s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/17a126a9ebf74429b5b9835432b4ec54 : SUCCESS in 1h 33m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/587f543561a84c96a4bc9acd96477952 : SUCCESS in 1h 57m 12s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba2ac70dff2141e0a5a4ba4e263c6ea7 : SUCCESS in 1h 25m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7c5b93b25264e6e9d2e3c14deeda468 : SUCCESS in 24m 30s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/819ac445220e477090e9132ba09662db : SUCCESS in 59m 54s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c49f6504e6184d5ca6346e171225d05a : SUCCESS in 49m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c140ac9af02e411bb65ba584b9a9d260 : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6927b75b548c6c15202d8e5464943dec9febd704","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-17 23:23:52.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e5ecd1a4d5feae238d37eb971cc93f085966b22a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-18 00:25:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677055/1 : FAILURE in 1h 01m 21s","accounts_in_message":[],"_revision_number":1},{"id":"911bc0ef4fe441173c1c1e9ef0207712c0aea3d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-18 01:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/b650ce85b9684e7ba176da0413b0e1f5 : SUCCESS in 59m 39s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f71c5c4b23714b7792a68fe513d509d4 : SUCCESS in 43m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/7a5532feb2b946748eaeec2055e25d6a : FAILURE in 1h 36m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/d001e0d8d21a4cdc894b24c7ae6cd492 : SUCCESS in 1h 11m 16s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/300b48a9bcd647c8811e56bb13da954a : SUCCESS in 1h 21m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/d3951cfc379f4accb54b8e3c26ee0350 : SUCCESS in 48m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba2291601894d06b213e807f1e574a6 : SUCCESS in 10m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12a342f0cb914ee1a8f8be1c1517f5be : SUCCESS in 22m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f7ab8b7bb494b1693c6aa18b9a04a4d : SUCCESS in 23m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f768ba9132ea4dd98d21af68b2f05662 : SUCCESS in 6m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6f090f4faa1d4da28d00132c8dead84c : SUCCESS in 1h 37m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a7f2bcc3ee50405885ed2760ee20c526 : SUCCESS in 1h 10m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e25700a9e824b2cb51b39a127efe9c1 : SUCCESS in 1h 25m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e6430dbd694743b7a06bf54c90013f7e : SUCCESS in 37m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f4f5d6e561724516b22b8c78b324ee2b : SUCCESS in 1h 02m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/407482847fbf47b98df62fd86bda7738 : SUCCESS in 1h 45m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/21d585d2bb7645b6ab6391bed3950b90 : SUCCESS in 1h 10m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d599e99a166b411ab04076dadf6e3343 : SUCCESS in 1h 19m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56760bf2bb0b4f8b99577cd7be2469d1 : SUCCESS in 1h 06m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a2af3af4dfa4842a74e115634d0a604 : SUCCESS in 1h 16m 02s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/35574740241d45b09201cec3471c39ab : SUCCESS in 1h 40m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4282b5ec2a7a4e61a1b8d9decc360aa3 : SUCCESS in 1h 35m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc2d2f6b3994440ba9cb79cb77770207 : SUCCESS in 1h 41m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f39740039f444a494945a2629b62925 : SUCCESS in 26m 52s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/8fd306f17c8a40c1bea687d067e32856 : SUCCESS in 1h 14m 47s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b1189015334f4ec8ad6e5356aa84439b : SUCCESS in 46m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/284463e3692f4185af1db90fca52d928 : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d512cd67100415843e99adfa6d68f5920807092e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-19 08:44:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf348732ecbfd0c5e50e6ac023d164faa0f1a9bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 08:45:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b73e2028d84b01b16dd1d01fcf99e3476cb1b638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:11:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/fbbd0e4757ca40e58a1b53d549af5226 : SUCCESS in 41m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef17749b3d14f12af5485b3044730cd : SUCCESS in 10m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cd99db07998415ab70305bd3c871296 : SUCCESS in 31m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7f0ba690a20649e3a9bb1eb8e527ac26 : SUCCESS in 29m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/66609d5eb1fd42b0923fcc386afbda9f : SUCCESS in 7m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/05867e58071042a08fef2c942d6d909a : SUCCESS in 1h 36m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e86f736b4cd4e0eaa03cf5b8f2d9ad9 : SUCCESS in 1h 01m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b0d099d4ae540edb59c10dc68e1c348 : SUCCESS in 1h 22m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a45ee5c9df2c4168be9b8ec76d7eb816 : SUCCESS in 46m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fd3bdc859f3e487b90e02e124e1388f3 : SUCCESS in 56m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e9fc99bfc75b4ff3b829eef372548e64 : SUCCESS in 1h 32m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/815346d0ebee4e488e5e67d42e82bf67 : SUCCESS in 1h 25m 57s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b7f3d6a24dae4bfca249aacb18b5dad0 : SUCCESS in 1h 24m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5d89a016c054f33a85f04f86774796e : SUCCESS in 1h 09m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b45796a0426349c8849be2b8c45c0ace : SUCCESS in 1h 08m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2daca5088fd64642b375ea1d092c91e8 : SUCCESS in 26m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/273aee22b6b94fa5b432f243cdebac7a : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4b29db831ff1cf07b47f448196b59344a750de3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:11:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9473566d559bf25c1baac2fcef505b0d2609fa1","revisions":{"f9473566d559bf25c1baac2fcef505b0d2609fa1":{"kind":"REWORK","_number":1,"created":"2019-08-17 07:09:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/55/677055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/677055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/677055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/677055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/677055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/677055/1"}}},"commit":{"parents":[{"commit":"ed3b41a0cd3a2178d0bcefeabf9d9596b6abd219","subject":"Merge \"[DVR] Add lock during creation of FIP agent gateway port\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed3b41a0cd3a2178d0bcefeabf9d9596b6abd219"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-08-16 13:44:09.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-08-17 07:09:19.000000000","tz":120},"subject":"Initialize phys bridges before setup_rpc","message":"Initialize phys bridges before setup_rpc\n\nNeutron-ovs-agent configures physical bridges that they works\nin fail_mode\u003dsecure. This means that only packets which match some\nOpenFlow rule in the bridge can be processed.\nThis may cause problem on hosts with only one physical NIC\nwhere same bridge is used to provide control plane connectivity\nlike connection to rabbitmq and data plane connectivity for VM.\nAfter e.g. host reboot bridge will still be in fail_mode\u003dsecure\nbut there will be no any OpenFlow rule on it thus there will be\nno communication to rabbitmq.\n\nWith current order of actions in __init__ method of OVSNeutronAgent\nclass it first tries to establish connection to rabbitmq and later\nconfigure physical bridges with some initial OpenFlow rules.\nAnd in case described above it will fail as there is no connectivity\nto rabbitmq through physical bridge.\n\nSo this patch changes order of actions in __init__ method that it first\nsetup physical bridges and than configure rpc connection.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I41c02b0164537c5b1c766feab8117cc88487bc77\nCloses-Bug: #1840443\n(cherry picked from commit d41bd58f31e259fe408c8c059b31299fdfe81127)\n(cherry picked from commit 3a2842bdd8d8d59e445393c7c7e7a9793357df08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9473566d559bf25c1baac2fcef505b0d2609fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9473566d559bf25c1baac2fcef505b0d2609fa1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
