)]}'
{"id":"openstack%2Fneutron~677056","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I41c02b0164537c5b1c766feab8117cc88487bc77","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1840443-stable/queens","hashtags":[],"change_id":"I41c02b0164537c5b1c766feab8117cc88487bc77","subject":"Initialize phys bridges before setup_rpc","status":"MERGED","created":"2019-08-17 07:10:13.000000000","updated":"2019-08-19 11:11:42.000000000","submitted":"2019-08-19 11:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677056-1566213102386-28760403","meta_rev_id":"25f79d6c5128a289e66350dd21d60501155a960f","_number":677056,"virtual_id_number":677056,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-08-19 11:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-17 07:50:40.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2019-08-19 08:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-17 23:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2019-08-19 08:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-17 07:50:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-17 23:23:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-08-19 08:44:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-19 11:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bea96691cd76e07271ebcf55dfa8ffb4d6574f0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-17 07:10:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"a1341675684199e1ec44cd5bf90d4465f46000ec","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-17 07:50:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677056/1 : FAILURE in 39m 55s","accounts_in_message":[],"_revision_number":1},{"id":"218787fb4747595fc9831ffa36c3f66472d81432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 08:53:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/b3726ee71f4f4a91876fa292b9b3dbbb : SUCCESS in 57m 19s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/fc7cdd01b9834c0d8defbbc93dbaa5e2 : SUCCESS in 27m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/f51d3df06544481c9f75a2eaf1c41779 : FAILURE in 1h 25m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/1d98facb0416438183ea50aae665718b : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/17480cc6f6064ce9b33961a5912b867a : SUCCESS in 58m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ceaec7bffdc485d9dc2c211f70bc69d : SUCCESS in 11m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09dfe6bf69654795942df22e886e3e8d : SUCCESS in 28m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eda1d6472d1c455089964b10df43382e : SUCCESS in 28m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ca51fdeb7d18446bbba5f1d3a0256fc8 : SUCCESS in 7m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fe12125bddd54db09c0fd565ebdbb88c : SUCCESS in 1h 32m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/782f7fee60c34eb09e4f7895dc336334 : SUCCESS in 1h 01m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d6f5fda02a641e7b09309bcf1fc6a24 : SUCCESS in 1h 30m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/148339dbf9a741ecb23c65f39a999287 : SUCCESS in 39m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2cc2f12b531940569ab08db50073db31 : FAILURE in 39m 34s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/56ee6acca55143339db573d2a61245d9 : SUCCESS in 1h 34m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/64f301540a7d44df9baf4094277b9ed1 : SUCCESS in 1h 33m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bf8f88f3207944bf899b199d08654d7d : SUCCESS in 1h 31m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4685098a6d1644a08066907a3babd332 : SUCCESS in 1h 06m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b8a095151314cb0ab2cb1b6550d7b6f : SUCCESS in 1h 01m 16s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b548c68f1bc94478baf09e14a7b809b3 : SUCCESS in 1h 33m 25s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5bd7f85312e84b639f1ed6c086595976 : SUCCESS in 1h 26m 25s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/30298762df544e628b1d4a29904b4f6e : SUCCESS in 1h 33m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7278d42c36d34d8cae70e0ae16a2725a : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bc612fbf57a40630238adf282bddbd0e07e2d4a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-08-17 23:23:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a91c05d516a6e47e58a157b5b54e6fcecad5a53","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-19 08:44:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ef1a6144f5da6c45b794e01d9a693d0bf8015ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 08:45:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97e99afd3e9836edb3101265854861f8ed71be93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:11:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3279e6e191034796ff5d063b23d6db6fec07cb6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:11:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d2af4a268f554030844c3ee443280499 : SUCCESS in 47m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c96503180045879453d81f5e4230d6 : SUCCESS in 11m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/704342f331364ef29ed551610a613f32 : SUCCESS in 24m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0d3243d76caf4c5abe52bc623b1b629e : SUCCESS in 24m 49s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/efb44f451c844e6e8c3bdd044fddab05 : SUCCESS in 7m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4fa969069da94003a9012b82f9390a3d : SUCCESS in 1h 27m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4fddbecbbdc149b5b3e458491f191f7b : SUCCESS in 55m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a12aab234f85455aafa92d5a9d0b8abd : SUCCESS in 1h 25m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f005efb864f4c41955bccc07b0ad986 : SUCCESS in 36m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3a995d83207b444c8052035474c60001 : SUCCESS in 1h 32m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7ce4d376ee3b44b9b846e18e67e69716 : SUCCESS in 1h 38m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17c01a700d3742dc9fe844ae4e6d4365 : SUCCESS in 1h 08m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a7aedf12e61040638fc0f14ec1415fd9 : SUCCESS in 1h 04m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2800972e1f14b1e8c7f0abbf2b61f35 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6618a917d86add7ee5c1005a21046d1bc57ac95c","revisions":{"6618a917d86add7ee5c1005a21046d1bc57ac95c":{"kind":"REWORK","_number":1,"created":"2019-08-17 07:10:13.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/677056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/677056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/677056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/677056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/677056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/677056/1"}}},"commit":{"parents":[{"commit":"a74539bfab92035af37bb6160a96f52013c158f6","subject":"Merge \"Make sure the port still in port map when prepare_port_filter\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a74539bfab92035af37bb6160a96f52013c158f6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-08-16 13:44:09.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-08-17 07:10:13.000000000","tz":0},"subject":"Initialize phys bridges before setup_rpc","message":"Initialize phys bridges before setup_rpc\n\nNeutron-ovs-agent configures physical bridges that they works\nin fail_mode\u003dsecure. This means that only packets which match some\nOpenFlow rule in the bridge can be processed.\nThis may cause problem on hosts with only one physical NIC\nwhere same bridge is used to provide control plane connectivity\nlike connection to rabbitmq and data plane connectivity for VM.\nAfter e.g. host reboot bridge will still be in fail_mode\u003dsecure\nbut there will be no any OpenFlow rule on it thus there will be\nno communication to rabbitmq.\n\nWith current order of actions in __init__ method of OVSNeutronAgent\nclass it first tries to establish connection to rabbitmq and later\nconfigure physical bridges with some initial OpenFlow rules.\nAnd in case described above it will fail as there is no connectivity\nto rabbitmq through physical bridge.\n\nSo this patch changes order of actions in __init__ method that it first\nsetup physical bridges and than configure rpc connection.\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n\nChange-Id: I41c02b0164537c5b1c766feab8117cc88487bc77\nCloses-Bug: #1840443\n(cherry picked from commit d41bd58f31e259fe408c8c059b31299fdfe81127)\n(cherry picked from commit 3a2842bdd8d8d59e445393c7c7e7a9793357df08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6618a917d86add7ee5c1005a21046d1bc57ac95c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6618a917d86add7ee5c1005a21046d1bc57ac95c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
