)]}'
{"id":"openstack%2Fneutron~677284","triplet_id":"openstack%2Fneutron~stable%2Fpike~I0d26d1b1d1c4ca64c7a1dd077752d191fd3a28fb","project":"openstack/neutron","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"I0d26d1b1d1c4ca64c7a1dd077752d191fd3a28fb","subject":"Remove experimental openSUSE 42.3 job","status":"MERGED","created":"2019-08-19 19:27:37.000000000","updated":"2019-08-21 05:19:05.000000000","submitted":"2019-08-21 05:19:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677284-1566364745652-18129b56","meta_rev_id":"f6dcbc11f79b48bc357021e3d861a0464e34ffe9","_number":677284,"virtual_id_number":677284,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-08-21 05:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 20:15:01.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2019-08-20 09:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-08-20 22:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-20 22:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 20:15:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-20 09:44:27.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-20 22:00:44.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-21 05:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9fb4d2148b049fad8e9b8743f8707f665770d015","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 19:27:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"516752dc19c771a80c8496cc63c4cc47413607f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-19 20:15:01.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677284/1 : FAILURE in 46m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e2c6cedbd809971158512235ac580e345d86be88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:18:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa483a4dd6474ef69aafd810c6533b3a : SUCCESS in 11m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21c203cbc8874c279718b624f3ea00e3 : SUCCESS in 24m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32b8b83612c74fb28dc09d6d52fc9075 : SUCCESS in 23m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b2822b4f41db4c62b67caf7729ae6170 : SUCCESS in 7m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a3204ab67e6342b69d283fee444aa4cb : SUCCESS in 1h 37m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a3fbed05b2b54d87a79a1c0e9fe707cc : SUCCESS in 49m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4fcd1c7ebd2e461fbabcaeec1d6639e7 : SUCCESS in 34m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1b70b6ffa5744f0ebd4d73904e6f2394 : FAILURE in 46m 52s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/a041a4782eee479ea5e3ab710216a9ef : SUCCESS in 1h 39m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a9e2fc9faeac46f5afa1d85e50593cf4 : SUCCESS in 1h 41m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/03be5cdee5354bf3a68ae08223cdb5a3 : SUCCESS in 1h 29m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/caac75cab90040b39a8cd62177af3386 : SUCCESS in 1h 04m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a969cedc5ca4f62b133466c2e54cde0 : SUCCESS in 1h 22m 27s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/02c4abe864254641b694de32ec48f594 : SUCCESS in 1h 43m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/00d0f7282f184f4e87f5c6edba5a2d7e : SUCCESS in 1h 58m 13s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/6f80cdeab8194c3da2f44a1773ae4fd2 : FAILURE in 1h 04m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2cf7f72b38a41d7a1d33b8fd71122ba : SUCCESS in 31m 43s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/24a5c411b2394c84964cbfc80f7be0a0 : SUCCESS in 44m 51s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9466a4a45cee4ff08b3313f0b55138c9 : SUCCESS in 2h 33m 15s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1d1d9524a208498e88653b2884589951 : FAILURE in 2h 05m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"505184aa03ff9370872a982b04bb7496f9a53885","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-20 09:44:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e481c31cbd0ae3768e6a2d4477c064973d0247ff","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-20 22:00:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b703f7c3c93d3ee118ab203190c51a76aca2c1e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 22:01:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcc968d328ac9228a42d1a38e8bd22044ec849f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 05:19:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5120d91b1ea64be1b4adbddf16f7009e : SUCCESS in 10m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f193d7ba42f41d3b92ab1bfa8f70060 : SUCCESS in 21m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a1a252998c1446ecaa505c3083bdad12 : SUCCESS in 24m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/32fc3d26d4c34c5d99bc3a2d9987571c : SUCCESS in 6m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e5b9580d6a24484380939086da5699b9 : SUCCESS in 1h 49m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ee21d5d79c6c47578a5aa6f60145472b : SUCCESS in 57m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d4066d7b6b046ef83cf8c9cc09f0d9b : SUCCESS in 33m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/60b5793ed7034c3da883097d84ccf785 : SUCCESS in 1h 38m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/794da9514e0d49d0b19451c97cf5d51b : SUCCESS in 1h 28m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a3ce1a9f65643a8b6f09685693ce2e1 : SUCCESS in 1h 00m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e957c5ade45b4e6f984575778e8dc1f3 : SUCCESS in 1h 12m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a9d95b8d109450cb2349cc29a744c94 : SUCCESS in 32m 35s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/9702897a562245b09da9b75211e6ccb6 : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d47c9dced84499b4d634336c36d7e993dcb29ee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 05:19:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fe32529f14d5e16e1ad1c04c293832a8e5ae354","revisions":{"6fe32529f14d5e16e1ad1c04c293832a8e5ae354":{"kind":"REWORK","_number":1,"created":"2019-08-19 19:27:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/84/677284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/677284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/677284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/677284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/677284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/677284/1"}}},"commit":{"parents":[{"commit":"fe403825ab08728e8b2d0fe15cfba3b452921f57","subject":"Yield control to other greenthreads while processing trusted ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe403825ab08728e8b2d0fe15cfba3b452921f57"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 19:24:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 19:27:16.000000000","tz":120},"subject":"Remove experimental openSUSE 42.3 job","message":"Remove experimental openSUSE 42.3 job\n\nThe openSUSE 42.3 distribution is eol, remove this experimental job so\nthat the job can be removed from Zuul.\n\nNote that master has a job for newer openSUSE running.\n\nChange-Id: I0d26d1b1d1c4ca64c7a1dd077752d191fd3a28fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fe32529f14d5e16e1ad1c04c293832a8e5ae354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fe32529f14d5e16e1ad1c04c293832a8e5ae354"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
