)]}'
{"id":"openstack%2Fneutron~677539","triplet_id":"openstack%2Fneutron~stable%2Fpike~I3f2905c2c4fca02406dfa3c801c166c14389ba41","project":"openstack/neutron","branch":"stable/pike","topic":"bug/1682094-stable/pike","hashtags":[],"change_id":"I3f2905c2c4fca02406dfa3c801c166c14389ba41","subject":"Allow first address in an IPv6 subnet as valid unicast","status":"MERGED","created":"2019-08-20 18:35:49.000000000","updated":"2019-08-27 12:26:52.000000000","submitted":"2019-08-27 12:26:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677539-1566908812076-727564a4","meta_rev_id":"81d0a41fa306df3bed535f8f8dff77101f4b7ade","_number":677539,"virtual_id_number":677539,"owner":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-27 12:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-27 08:30:24.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-08-20 22:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-08-25 07:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-21 03:24:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-08-25 07:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 22:17:12.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-21 03:24:08.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-25 07:33:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-27 08:30:24.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-27 12:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4290ebd1e84204ca3c6173891cbcc90668c2794e","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-08-20 18:35:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.","accounts_in_message":[],"_revision_number":1},{"id":"fe2474af7ace03b3d60ee17fc3b92a8f6e683051","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 21:57:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aeab6e6b42c4be4967f03a7539bd512 : SUCCESS in 10m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a5458ad97104bd395991f4a38623cde : SUCCESS in 22m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/016f0bc0e94e45a3a8b15a6181b32080 : SUCCESS in 22m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/06edfcb914cb494695c95e066229f860 : SUCCESS in 6m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2aa8b11c8f7e4a71beecd23826bbd651 : SUCCESS in 1h 44m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/702f980d3be6407c9ab55e4976d550b3 : SUCCESS in 54m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bbf8fedbcb14cc6bec0850e7b250669 : SUCCESS in 19m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/49d9e2a2821c4aaebf860099f4724bbf : SUCCESS in 38m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5993a88c2c554eab99cd0d5602810a1b : FAILURE in 50m 41s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/afcd0af404634e86b83d3fc42b82597c : SUCCESS in 1h 24m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2cc2d256840846008131c3c502e8d4f3 : SUCCESS in 1h 27m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7ecdacad392341018c973e8bea4c111e : SUCCESS in 1h 26m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/312a826727264782a9bc155b86846348 : SUCCESS in 1h 00m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f918e5eaf9e4880b413a6003cf43456 : SUCCESS in 1h 12m 48s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3eb6b526fb8849059469163588986f9c : SUCCESS in 1h 46m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3cb7bce5a2a4c0abd71b6a216bb16dc : SUCCESS in 1h 41m 07s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/a497251abdba4a999499ed225999cd4c : SUCCESS in 1h 27m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78be7b640c304f0a8a99b3cb42e5e6f4 : SUCCESS in 34m 11s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/329932ea41ed47e297f35399104cd655 : SUCCESS in 48m 30s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/dba427c4ffc045a8ad151d5c7ddcf3ee : SUCCESS in 2h 22m 22s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/015acbdb52144a008dc238a5647ff837 : SUCCESS in 1h 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad2ad1b066bc13cb791bc3d244e3ed9b7ce7bd44","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-20 22:17:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c47d03b061ef203d39f3deb114567ac21cfb79d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-20 23:47:57.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677539/1 : FAILURE in 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"40c93e6ecedce14d100611b220f016cbe87ee0e8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-21 03:24:08.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"359b8405e2bf4b76d39cae98a8618a322697006e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-25 07:33:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7305168bddfcd1d14a90dbf1ed7e19c04d34f3d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 07:33:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3833d1e2bb95564442c3712e1c0195648379ded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 09:24:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e687bbb7ea104c99acfa725249a9b585 : SUCCESS in 11m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b7c2bd01cc1477ea53fa1dbbd86ba8f : SUCCESS in 22m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dc3abf393ae747cc8f09d8ee9d45d144 : SUCCESS in 23m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e6a2a80ace8544e49127919565baac62 : SUCCESS in 8m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff146c79406349518161a1f2931de05d : SUCCESS in 1h 48m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3abf7d6427d047d6a9b40af82e19bea8 : SUCCESS in 50m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73e1d22fe024e33bfbd3813116a5a79 : SUCCESS in 21m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/339cf46117ae437792252e0ca9a23707 : SUCCESS in 29m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e4351b8590a94c569ab04ab8dfade762 : SUCCESS in 1h 34m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9bad008280fc4ed3b89d884076b08d3a : FAILURE in 1h 17m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b16a78040b374854b4ade92c6a3c55d6 : SUCCESS in 1h 06m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/836f9a55be7f4443a60700f466580c91 : SUCCESS in 1h 02m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/530ff3030ef74aa9bb6f830db1114efa : SUCCESS in 26m 44s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/30e164489f4f4fc98c3de8952a33876b : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c29528f5eaa32be7319bdc5a18e5a01824a7635f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-26 20:43:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee5eab947851466a682994211be41e838f87a45c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-26 21:58:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677539/1 : FAILURE in 1h 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6b52cd5d42c0f71a1c5d8ea24ab6d8c5d8889e91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:10:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a366d4758864770bb696914348b3bc3 : SUCCESS in 11m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a897587754f34534829c2ebfd0f7c116 : SUCCESS in 26m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a1273dcc9ce4a2b99af8dd631b13ae0 : SUCCESS in 26m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bf10cd2d48e64bd3923a5aca5be06021 : SUCCESS in 15m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/38f41ed1f06d4f398c180198f5db9dc9 : SUCCESS in 1h 43m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc40948516ab409cb75c5f08cf1c1cee : SUCCESS in 1h 04m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddb33124ed2b44a980f6ee6ce0dda19a : SUCCESS in 23m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5aa8b644141740859e831626555adf7f : SUCCESS in 37m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3d78858cfc394f92ad1d2ee2024f4866 : FAILURE in 53m 23s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/e16641af43d7482eb5271bfd4b31bbb6 : SUCCESS in 1h 59m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e7246e28e7de4dedb7fccd4aaedf5b38 : SUCCESS in 1h 42m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aebf61c6677c447bad5bade4ca2578f6 : SUCCESS in 1h 33m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4872e631c6f44587b49a1e75662d61a6 : SUCCESS in 59m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7bbd6dd6d194c86b2ea5cb52db2ec23 : SUCCESS in 1h 04m 52s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bfa04cd587ea4cbe9a778edda50e41a2 : SUCCESS in 1h 46m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/73116fec4c2b41e287032472af1db41b : SUCCESS in 1h 35m 00s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/6015f25053e24187b5797b24a9530170 : SUCCESS in 1h 28m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0edb92c1d8a343769ee8c2f613eb777f : SUCCESS in 40m 47s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/992b80762d774edbae6d6ddaf3fa530a : SUCCESS in 49m 39s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a05e30217b2c493b850db19f257686d1 : SUCCESS in 2h 20m 38s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e466b384fcef40ab897235023d7e4e4b : SUCCESS in 1h 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5ea57d57027eb9a4e1b180a0cd626d261d7fb5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:11:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"618ffccc71365ec06b04a6ecfc3b6c687f151916","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:30:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df94143263294543a40722c33c02c5fd : SUCCESS in 13m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c47018c949c34729a01a9ec690a7cd2c : SUCCESS in 22m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/09445330497940beb534e1c8e60c4c10 : SUCCESS in 24m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b2fcea2503d049a8b126ca19d8b68286 : SUCCESS in 7m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b534087086274d73a91f744c9dd136ca : SUCCESS in 1h 37m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a9e6c077544046b2b5e9e7e8f4f2ec73 : SUCCESS in 53m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/152fbbca70014d13be6280a7064aefb7 : SUCCESS in 21m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6e28a1e1eb1045d990803de296e53424 : SUCCESS in 38m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dd4a985ca5e94f59b0f27f09fc92cc25 : SUCCESS in 1h 41m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/81c4adb5f57449e79676533cc474e94a : FAILURE in 25m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdd6c4fcd5fd4cf491f18c82c28243a0 : SUCCESS in 1h 05m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8d983fb427384480bfb0315c19274db0 : SUCCESS in 1h 13m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/942a5fdcced742dd9c83defcd979c403 : SUCCESS in 29m 30s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/186591be0c2f4df99e0c1a09eded72a5 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4c272a24d723f11d3712e5fc90ac9a041bf7cbb8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-27 07:20:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b94e1ef52defc10d96d463ff0d3b8faca903138","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-08-27 08:30:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/677539/1 : FAILURE in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dee65b0a2ea242e5838e4c62cefe9927dd6c9455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 10:10:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d414d128f6743ec94a2cc818a29febf : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba62d83db349412ba4bf76e26218fb5e : SUCCESS in 22m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e461e5f069c348f4a8b4d1f80bc783fb : SUCCESS in 25m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2e130d61b85c4aeb9751c45494521dc4 : SUCCESS in 6m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4444b517091f4f3baa4f1ee2c603bf05 : SUCCESS in 1h 56m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2d4eb85601794c2c96b222a24cbcb921 : SUCCESS in 45m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4a3acc178934a26b05792494046c15e : SUCCESS in 22m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/abb3004bd3a54a4baebcb869383b5328 : SUCCESS in 30m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/05c5cf92eb064eafb4c8019d0dcffa45 : FAILURE in 43m 41s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/1251d10995544f13bca42b69b0886ddc : SUCCESS in 1h 46m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/25dbe7c914bb40bab1a5267195d629f0 : SUCCESS in 1h 15m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/12297593f63e4f70b2e608f54ff58e34 : SUCCESS in 1h 24m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35aee1a2df354d768e451b5eb82063a8 : SUCCESS in 51m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/722df853811f4649a33d5118ad795bc0 : SUCCESS in 1h 07m 16s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/788b744a27dd4b52b84999642351830e : SUCCESS in 1h 57m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2fdb6d59332640db96d8f0a83d7bff69 : SUCCESS in 1h 47m 29s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/63242da31a594bd687a9449f163553c4 : SUCCESS in 1h 47m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1333cd723e3343eeb0e6411261e80300 : SUCCESS in 34m 48s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/82c2e320d6b84683b30ab365da75aaf0 : SUCCESS in 48m 11s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5f5459298f964cb9ad853197ebb9e4f7 : SUCCESS in 2h 42m 20s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0224ed0e79a640f4a24fe6f52dac9be5 : SUCCESS in 1h 53m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31b84f9fa857dd52c9d4f00e26e1d0b903b889ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 10:11:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5663edef947710c3b13011cd8ae3a57a3bad0b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:26:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012945bf27a14734b5ba2ef63ec56644 : SUCCESS in 11m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03c34ecf71b74cafafabda198da88d0a : SUCCESS in 26m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/edc2f436f71e4d4ba380d2c0936104cb : SUCCESS in 25m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3849dae0d0f44cdd8b459a257e1ab25d : SUCCESS in 8m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0152ae518672458e8ae697012b58226a : SUCCESS in 2h 12m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ad95fe5f0e3b48edbedfb7fdbf36e18c : SUCCESS in 47m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20d8a0ac5bcb4259be3893d6ca05e3df : SUCCESS in 21m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/35d64ec5d82a47c28f8f605b146259ff : SUCCESS in 29m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/91d6d6d1fa3040e7bebc9b62b0dbbb2e : SUCCESS in 1h 23m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e3e0e06ec23c4d45b4ba9bddad6d0046 : SUCCESS in 1h 23m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f75efb96b6b462bb074f7e6ab32c81e : SUCCESS in 56m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/30ae81f68dd3498280b26394571170d5 : SUCCESS in 1h 13m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d60537cadc4b2dba0c35960671c47b : SUCCESS in 28m 28s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/4b7c8ccc63224bc59c5dc23f4a45cb64 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7eccfe55e6b25e986f4d380624647bbcc436308d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:26:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e37e714b0816a0b95902fdd384cf884643d260d7","revisions":{"e37e714b0816a0b95902fdd384cf884643d260d7":{"kind":"REWORK","_number":1,"created":"2019-08-20 18:35:49.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/39/677539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/677539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/677539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/677539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/677539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/677539/1"}}},"commit":{"parents":[{"commit":"a9445eec6f7e77ed85cbdbf041bd7995b589fd1e","subject":"Merge \"Clear skb mark on encapsulating packets\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9445eec6f7e77ed85cbdbf041bd7995b589fd1e"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-03-25 14:29:23.000000000","tz":-240},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-08-20 18:35:49.000000000","tz":0},"subject":"Allow first address in an IPv6 subnet as valid unicast","message":"Allow first address in an IPv6 subnet as valid unicast\n\nWhen looking at the RFC [1], there\u0027s no mention that this can\u0027t be the\ngateway address.  Permit it.\n\n[1] https://tools.ietf.org/html/rfc4291#section-2.6.1\n\nChange-Id: I3f2905c2c4fca02406dfa3c801c166c14389ba41\nFixes-Bug: #1682094\n(cherry picked from commit 1916bc5c0627ca24e312ca55ea62c8df7eb8f158)\n(cherry picked from commit 0cd42257aaced0b6da0f43bc5da660b6a664141f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e37e714b0816a0b95902fdd384cf884643d260d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e37e714b0816a0b95902fdd384cf884643d260d7"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
