)]}'
{"id":"openstack%2Fneutron~680657","triplet_id":"openstack%2Fneutron~stable%2Fstein~Icf9a8802585c908fd4a70d0508139a81d5ac90ee","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1799555","hashtags":[],"change_id":"Icf9a8802585c908fd4a70d0508139a81d5ac90ee","subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","status":"MERGED","created":"2019-09-06 11:00:25.000000000","updated":"2019-09-10 10:17:41.000000000","submitted":"2019-09-10 10:15:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680657-1568110557205-456ce0b0","meta_rev_id":"e14111390e32cf806fbcc591ffdc717b47ab483d","_number":680657,"virtual_id_number":680657,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-10 10:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-09 23:41:19.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-09-06 12:39:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-09-09 22:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-06 13:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-09-09 22:00:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 12:39:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-09-06 13:14:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-09 22:00:05.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-09 23:41:19.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-09-10 10:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c5260ebdc51a5d467f4e0b98a72201471517271","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aba17cb6b8594f6bc7be4032a36a5fe8bf38f6c2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:01:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a70a7c41b7c38f5b5207bca77492fced4c5b70c4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:02:22.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 68077a4cc9af638256a272cda42b44346d90589d","accounts_in_message":[],"_revision_number":2},{"id":"c32b3f9eb299113b81b2f324e58d51a888e93308","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-06 12:39:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0cdc15359bee40baf338871342a7b1c52685d8a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 13:14:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fae0ae37784664e8ebf78ca6bc813991530ed80a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-06 13:53:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/680657/2 : SUCCESS in 1h 45m 32s","accounts_in_message":[],"_revision_number":2},{"id":"81d347c404090761adb9afedc5b05b7dc341bae8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:36:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/ee96abfe3dde468eaec1bc9ee27d82f7 : SUCCESS in 54m 35s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/cbddd967159b4861a476c19ef7546aaa : SUCCESS in 35m 13s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/6f1f935ba0c64c46a560c02d31a5da60 : SUCCESS in 1h 39m 49s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/679f245e5c84416399c06853729e80cd : SUCCESS in 1h 11m 30s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/087e0194df4542e0bc8d7b8862a547f6 : SUCCESS in 1h 05m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/0c1a2da0648845e4af249811a1fd90a8 : SUCCESS in 57m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a610eb629bec4091a2b626357c4395ed : SUCCESS in 30m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6083a076dda14cd5b95ae163ad9bcbe6 : SUCCESS in 22m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4111045eb0d4bfa85a2652bce33730c : SUCCESS in 11m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6255825dbad748028cc78e03841f0fd5 : SUCCESS in 24m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2cd5a18f0a6e472d86b63f51c460c97a : SUCCESS in 22m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54d32eead93c45e6a8170e953835e487 : SUCCESS in 28m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9897cebca6764eea89d083b7951b095a : SUCCESS in 30m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ba37f44d2345739829817270911a2e : SUCCESS in 8m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/08a73c414150486692994d23b8977903 : SUCCESS in 1h 29m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91fe1a1e6b1f4f6e95706c747d719ab7 : SUCCESS in 1h 08m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b10538b7914c463d9b2e659ddccfac7b : SUCCESS in 1h 05m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b4417ace72944209bce373a29ddfed8 : SUCCESS in 1h 24m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f82951410c24c4abe1e445217df9ce1 : FAILURE in 41m 11s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/19cca9fbe8384f3c95fb2b6212365bf0 : FAILURE in 40m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2daf0ae1c0694d62b5a30da8580b3a5f : SUCCESS in 1h 05m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c27a74826dd549bca399f715e171fcd2 : SUCCESS in 1h 40m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/535e59717552485da06b98daf2f82cfb : SUCCESS in 1h 23m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f46441676ec244c4ad080e2a7625dcf0 : SUCCESS in 1h 13m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bba15b5971124bfca1ab3a47b9cec968 : SUCCESS in 1h 28m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9576a4cfa06e4b0393fcd551e36bb6f0 : FAILURE in 1h 22m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cdb8c7b82a614cb2a197e665250e7b89 : SUCCESS in 2h 02m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c42abf8d57434d988459b58ec8309a60 : SUCCESS in 1h 29m 17s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/5cff9f9bae314f3d888c6d0b0f77c52c : SUCCESS in 1h 37m 53s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/01bc9e80015c4dd3994d7e453dfe07ce : SUCCESS in 52m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/63a663e4fe3b4551b797052bdca89bb0 : SUCCESS in 2h 24m 37s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ee8bac452490480286e350060e461551 : SUCCESS in 2h 01m 07s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/88f11d4ed6e6405e98ee3c064522039c : FAILURE in 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17713f189097c3585424b7d208941ab20044a564","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-09 22:00:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck functional should be fixed","accounts_in_message":[],"_revision_number":2},{"id":"eb115c477131d058d0fac69944f0ec5e7b491ba2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-09 23:41:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/680657/2 : SUCCESS in 1h 40m 48s","accounts_in_message":[],"_revision_number":2},{"id":"6acce2cc0f322541075f55b6853ed07c6faa72b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:07:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/31bfb79199014a1dbb28703979fd5bfd : SUCCESS in 56m 11s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/768c31a0e6634b4c9f545ab6685b995a : SUCCESS in 34m 13s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/0c56fe5bd38847cf953552e5d165308e : SUCCESS in 1h 38m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/fac671df39e443de8a0637dae8843664 : SUCCESS in 1h 09m 16s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/8e2553013f664093a69c4f31e31ebc29 : SUCCESS in 1h 04m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/9d6f2c6645494c22a9cedacc219f3c24 : SUCCESS in 1h 00m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fb6dba2264344e0b1dacd225029901e : SUCCESS in 27m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/724bc5d4329f49f582b09c80f1abafa6 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9795d41b790c4410a52aca79dab8ceae : SUCCESS in 11m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7b7bdb4ba2d42658579955631393b61 : SUCCESS in 24m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b702c3c268c84db29611c7e97835e7b4 : SUCCESS in 26m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de77f6d2f18a43c892ef2c777946dece : SUCCESS in 27m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a1037ec32c74b22b174ebe38868a37c : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9398e7e9ea1b4a0b8ff93174097dbac1 : SUCCESS in 5m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/248d65ed71374faabaa889efc6c1a874 : SUCCESS in 1h 45m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a864f6e21d05457d8744bc11ee5a15d0 : SUCCESS in 1h 09m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aba3b26c24434a0b80be5945310f51c0 : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ea001b7353442d298f619540b1e4cbb : SUCCESS in 1h 24m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ced999f90c594b00aa6e18ca08ac7dcf : SUCCESS in 50m 44s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/fcdc989998da40579ab8b183257f4b31 : SUCCESS in 49m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9307d2d935f346cda0f2945654e40471 : SUCCESS in 1h 07m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/39f138330d814bf4a4b9c8a86fcf8c7d : SUCCESS in 1h 43m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6b882225c2da4fac80b674ada46683a4 : SUCCESS in 1h 32m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/195c05a0da624cba947a400d6cec83f3 : SUCCESS in 1h 09m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9b7fe65edcb24978b5ca512425d2512f : SUCCESS in 1h 17m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c52f1d21e3fc4f0e899138b46b50ba7d : SUCCESS in 1h 34m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1f7309f686a439a814cb100ff88bc41 : SUCCESS in 2h 06m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b70729dda0664a929c00e8f7d77db5d0 : SUCCESS in 1h 29m 16s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/2ebb6fbc9ed4480eb3ce0bb0829d42c1 : RETRY_LIMIT in 3m 50s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a93d6984397448e0a89e2ee4274db645 : SUCCESS in 49m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab18798f7d48446684029384ad265c87 : SUCCESS in 2h 15m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/eeff0cfadb5b4bdcbf4715f1263a5f87 : SUCCESS in 1h 59m 04s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b23a0bd6fabf4738ba329ff159c656e4 : FAILURE in 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a5ea0b833d35570372350bc8cbdc57e6a1ff485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 02:08:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d82d8a1c5b5014ce0dc296218836db42a2ba151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 10:15:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/6d1921910133493ab0b9f7625857e6ba : SUCCESS in 58m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9d014f9e6e4357ab4ca20290b55e3e : SUCCESS in 24m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e953ca1d5544fa8a854627144a82d96 : SUCCESS in 15m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17a95529b5ee47eca5fc39ba0ad43be4 : SUCCESS in 25m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c0db1972a8a406c9a269396dde4686c : SUCCESS in 26m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e38b6bd5dccc4a6cab4ef63cf9d90abd : SUCCESS in 23m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6fcc38c3b9048118128bd9ee05b3e90 : SUCCESS in 23m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62332e4580444fe8b58718182e23aef : SUCCESS in 7m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e64051c63d0044e2877934c77e032e3d : SUCCESS in 1h 35m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7bcda330d9e742f7b1c5b9642c19a57a : SUCCESS in 56m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/82f71435e57a4a08946f41510c7cd9ca : SUCCESS in 50m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c9b88cd46b145cd914d1cfac5dd45e0 : SUCCESS in 1h 20m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8d6e7902e83648bda27e40e4430b54d0 : SUCCESS in 45m 09s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/80dc540f40194e05a08e65358227d1c1 : SUCCESS in 44m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b6d69d3e7efd4c0888189cc7f9ce2459 : SUCCESS in 1h 08m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/edee401f69ca47cc8c9f7401c0c87df5 : SUCCESS in 1h 27m 59s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6f60b6bd123404e9d9478ff7351b02f : SUCCESS in 1h 22m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23aaefa19e3e423288ae056c8118b1a8 : SUCCESS in 1h 16m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9217f6ec33c3473fb9023bec2846c948 : SUCCESS in 1h 20m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/133f869a2bc045fb973d5684d94fe094 : SUCCESS in 2h 03m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8c6d5c022531bfcc426aa8f2a8696a59cc46bfa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 10:15:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1b03fb99effec293742071ce82d4150c8609b6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 10:17:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5fba4c22641434b87dddcf91fc78907 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b07cb21e3d8abe15c6322616b0b8976b54aef8d","revisions":{"cc30d8cbbcc4d88ae63132be8775e0ec556a2dc6":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:00:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/57/680657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/680657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/680657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/680657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/680657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/680657/1"}}},"commit":{"parents":[{"commit":"c09e8295806a5f6932b7a2a3795aac008d6e3480","subject":"Increase number of retries in _process_trunk_subport_bindings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c09e8295806a5f6932b7a2a3795aac008d6e3480"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-14 15:50:39.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-06 10:59:47.000000000","tz":120},"subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","message":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds\n\nCurrently there is a delay (around 20 seconds) between the the agent\nupdate call and the server reply, due to the testing servers load. This\ntime should be higher than the agent-server communication delay but\nstill short enough to detect when the DHCP agent is dead during the\nactive wait during the DHCP agent network rescheduling.\n\n\"log_agent_heartbeats\" is activated to add information about when the\nserver has processed the agent report status call. This log will allow\nto check the different between the server updating time and the previous\nagent heartbeat timestamp.\n\nConflicts:\n    neutron/tests/fullstack/resources/config.py\n\nChange-Id: Icf9a8802585c908fd4a70d0508139a81d5ac90ee\nRelated-Bug: #1799555\n(cherry picked from commit d7c5ae8a0378fa50853e91241426b6470ffe7492)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc30d8cbbcc4d88ae63132be8775e0ec556a2dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc30d8cbbcc4d88ae63132be8775e0ec556a2dc6"}]},"branch":"refs/heads/stable/stein"},"5b07cb21e3d8abe15c6322616b0b8976b54aef8d":{"kind":"REWORK","_number":2,"created":"2019-09-06 11:01:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/57/680657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/57/680657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/680657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/680657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/57/680657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/57/680657/2"}}},"commit":{"parents":[{"commit":"c09e8295806a5f6932b7a2a3795aac008d6e3480","subject":"Increase number of retries in _process_trunk_subport_bindings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c09e8295806a5f6932b7a2a3795aac008d6e3480"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-14 15:50:39.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-06 11:01:46.000000000","tz":120},"subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","message":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds\n\nCurrently there is a delay (around 20 seconds) between the the agent\nupdate call and the server reply, due to the testing servers load. This\ntime should be higher than the agent-server communication delay but\nstill short enough to detect when the DHCP agent is dead during the\nactive wait during the DHCP agent network rescheduling.\n\n\"log_agent_heartbeats\" is activated to add information about when the\nserver has processed the agent report status call. This log will allow\nto check the different between the server updating time and the previous\nagent heartbeat timestamp.\n\nConflicts:\n    neutron/tests/fullstack/resources/config.py\n\nChange-Id: Icf9a8802585c908fd4a70d0508139a81d5ac90ee\nRelated-Bug: #1799555\n(cherry picked from commit d7c5ae8a0378fa50853e91241426b6470ffe7492)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b07cb21e3d8abe15c6322616b0b8976b54aef8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b07cb21e3d8abe15c6322616b0b8976b54aef8d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
