)]}'
{"id":"openstack%2Fneutron~680658","triplet_id":"openstack%2Fneutron~stable%2Frocky~Icf9a8802585c908fd4a70d0508139a81d5ac90ee","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1799555-stable/rocky","hashtags":[],"change_id":"Icf9a8802585c908fd4a70d0508139a81d5ac90ee","subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","status":"MERGED","created":"2019-09-06 11:02:22.000000000","updated":"2019-09-10 09:50:07.000000000","submitted":"2019-09-10 09:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680658-1568109007446-1724c904","meta_rev_id":"2a5e152e8bc7bb1cfd4647794326e39252ce0ab6","_number":680658,"virtual_id_number":680658,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-10 09:50:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-09 23:33:01.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-09-06 12:39:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-09-09 22:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-06 13:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-09-09 22:00:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 12:39:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-09-06 13:15:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-09 22:00:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-09 23:33:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-09-10 09:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4b593e149b0dd05a6b3990874395031d5464e60","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:02:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"efdaa74ea52ceb9666c44aeb56dbfa01382f615b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:02:40.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 57725fc7d4513c37de597a4b86f8d6878e2be034","accounts_in_message":[],"_revision_number":1},{"id":"2be17e391ce76542e1a989bffe8401a15fd4ed69","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-06 12:39:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"808a09f5e657e8648b640bdd74518abc039a77ce","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 13:15:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b20d6feb0b3828b05afecb464e967359e0acad3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-06 13:45:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/680658/1 : SUCCESS in 1h 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"884f820c998b3ee882f36b53a44f55feef648fb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:01:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/2909e15fd5c24e269425e50f635e0a3c : SUCCESS in 49m 03s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/4b0864ae845b46f48191a4eb1be53995 : SUCCESS in 34m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/929fb78efee0461d86843790410592db : SUCCESS in 1h 37m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/f2912cef422c4766a4a4f0088035ab19 : SUCCESS in 59m 38s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/a486811ef17b43ddb1c35552f1980b3d : SUCCESS in 58m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/59062032be4c446d9e9a8dcb934c2738 : SUCCESS in 51m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a25a3b2149947fa9b48cc6ae1a1c127 : SUCCESS in 12m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a01806c5c5541d08af8e131973620b6 : SUCCESS in 29m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61cea6fec57542b5999b883f24af0f4e : SUCCESS in 36m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b09c378c6d534448a6150cb76bdd8e8c : SUCCESS in 10m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/52655adda31b487e8a60c55284764605 : FAILURE in 1h 34m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/59ff70e12526481892c8f4219f43a469 : SUCCESS in 1h 03m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9005e5b2fb6547358219a9b65289c46d : SUCCESS in 1h 29m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/919b6baea716448fa3e38d78b1613bbf : SUCCESS in 45m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f2464b4ef98f4029b4a412e8f83c8689 : SUCCESS in 1h 04m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/28fa46c4ffbd4dc7a64b4af0b3d9932e : SUCCESS in 1h 42m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0a0040114c55408ba98ff2351356b214 : SUCCESS in 1h 41m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b1b448178794e37a06f1bbdee47e6ba : SUCCESS in 1h 19m 13s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/51d63f00b26740a4ad23fe0778042cff : SUCCESS in 1h 15m 12s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9cae4aff5b3a4ebd8d04c0dfab0d2387 : SUCCESS in 1h 43m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/504ac68eb1b84971baa63bcc652dff44 : SUCCESS in 1h 31m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8ba5d65ca73445cb0324ddb825da67d : SUCCESS in 1h 39m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/381c388cdbc2408b8defbffe578168db : SUCCESS in 34m 57s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/999bcd6caf094bdeb3ed4fc26ec2759f : SUCCESS in 1h 06m 55s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/458de0759b8842828ad264270c69e312 : SUCCESS in 52m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/033db15695f44ef99fb3e99a37273644 : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6a647ca9919186e0e0ca7453558ae82a3d684a7e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-09 22:00:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"48d837701d53a777f36087540db3627d7bdfb1d3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-09 23:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/680658/1 : SUCCESS in 1h 32m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9c78bc716238a9343cded5a769ebecb8f2d210a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 01:57:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/1ce481dbdaef474ab0a68870c3e943b7 : SUCCESS in 49m 05s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/eb861f689dd04810b52f4cea8e772707 : SUCCESS in 35m 40s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/c2cd43e4e59f4b60bed5917ea3466ef3 : FAILURE in 1h 30m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/6a1e4b0603de402abcefc6588eda39cf : SUCCESS in 1h 05m 17s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/60c331421d614ebd9a53ae639635a208 : SUCCESS in 56m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/dc3f8a0f49b4477f97667e4e56372058 : SUCCESS in 1h 06m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee46b407e8be4ed2898c31d897359e3c : SUCCESS in 11m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be98bc5cfba845bb9a68490b75dcee44 : SUCCESS in 28m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eee9b33838bc4bb09caf0b4039ce2068 : SUCCESS in 26m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9dd48b02300a4601b9d1852903573569 : SUCCESS in 6m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1e378d75f1e4412bb71ee2e45458163f : SUCCESS in 1h 31m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/48b233b76d164a1caee1fa5c79f6bbf5 : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae491c67a0ae48679984692f017bee8d : SUCCESS in 1h 14m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21690de7e2764b8b904369b665d1ca52 : SUCCESS in 41m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1ae6e13bd9774e4683bef35caaf9ea43 : SUCCESS in 56m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9f4dd53052394687ab57428d0d60bf00 : SUCCESS in 1h 50m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/896f077d9e3a4b92a992249ac739ce09 : SUCCESS in 1h 39m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c851b0cb0ac4b1c893ba34918be5e85 : SUCCESS in 1h 15m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/12edb6f951a7485bae8283af21fe8d9c : SUCCESS in 1h 01m 12s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7ad89bd6005b47e19c68442a5b008d9f : SUCCESS in 1h 37m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4707749624a549a48b3d57c239fc37d0 : SUCCESS in 1h 59m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/189c0fea887f489fbc8f710a3c105264 : SUCCESS in 1h 32m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19c56db4ed054af380b28539cea1fb88 : SUCCESS in 25m 41s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/f51ff020ada24f3aa1f5f7be50b38a00 : SUCCESS in 1h 03m 02s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/79af89214f404f5e8a93eb64b9fea736 : SUCCESS in 51m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8855d64327d1417ab1302d975d9f7343 : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6cf81910aafd5a3b9c260439bbe44f4d05111802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 01:58:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9589ab277a7c84ce19ff6231612fb7124a5a3cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 09:50:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f577858c04bed4bf8f24c6a2fc7f36184f74ab32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 09:50:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/79093d6433e940ce97420df4c9963ba6 : SUCCESS in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b494453a124df58f022e6855137d35 : SUCCESS in 12m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4dab1e1b460642eaa1e80b60e977da3b : SUCCESS in 26m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/21f8fed099594a0ba49267abb2b96216 : SUCCESS in 24m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1c04d6ebaec545888295d84b396015b6 : SUCCESS in 8m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d9755ef2d6847808fb2ce17b3e70119 : SUCCESS in 1h 33m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ae91f202c9342569b368e153841b62a : SUCCESS in 1h 00m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb78919a72564a35acd683f7e0abaeaa : SUCCESS in 1h 11m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/946a3d3384474f79989729229ec7b0d8 : SUCCESS in 50m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9a2c0cf9396d4b25b43bbe9c438a0773 : SUCCESS in 58m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8503ac9abae54360b8c6d9f078f88b17 : SUCCESS in 1h 35m 37s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7fa4274296b43038190bb2c23762218 : SUCCESS in 1h 29m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bc5d846bc484a28831d4406c3773d75 : SUCCESS in 1h 07m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/630cc31bb23b44259a28e4d7c085df55 : SUCCESS in 1h 29m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3abfae70a284148a092c93b9106cbff : SUCCESS in 25m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1812fe861cd44ba2a941e99489a3d1fe : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68077a4cc9af638256a272cda42b44346d90589d","revisions":{"68077a4cc9af638256a272cda42b44346d90589d":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:02:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/58/680658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/58/680658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/680658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/680658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/58/680658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/58/680658/1"}}},"commit":{"parents":[{"commit":"1199207e7da977e1cd3c12f8a4242aafe6330d68","subject":"Increase number of retries in _process_trunk_subport_bindings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1199207e7da977e1cd3c12f8a4242aafe6330d68"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-14 15:50:39.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-06 11:02:22.000000000","tz":0},"subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","message":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds\n\nCurrently there is a delay (around 20 seconds) between the the agent\nupdate call and the server reply, due to the testing servers load. This\ntime should be higher than the agent-server communication delay but\nstill short enough to detect when the DHCP agent is dead during the\nactive wait during the DHCP agent network rescheduling.\n\n\"log_agent_heartbeats\" is activated to add information about when the\nserver has processed the agent report status call. This log will allow\nto check the different between the server updating time and the previous\nagent heartbeat timestamp.\n\nConflicts:\n    neutron/tests/fullstack/resources/config.py\n\nChange-Id: Icf9a8802585c908fd4a70d0508139a81d5ac90ee\nRelated-Bug: #1799555\n(cherry picked from commit d7c5ae8a0378fa50853e91241426b6470ffe7492)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68077a4cc9af638256a272cda42b44346d90589d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68077a4cc9af638256a272cda42b44346d90589d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
