)]}'
{"id":"openstack%2Fneutron~680659","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Icf9a8802585c908fd4a70d0508139a81d5ac90ee","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1799555-stable/rocky-stable/queens","hashtags":[],"change_id":"Icf9a8802585c908fd4a70d0508139a81d5ac90ee","subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","status":"MERGED","created":"2019-09-06 11:02:40.000000000","updated":"2019-09-09 23:50:16.000000000","submitted":"2019-09-09 23:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680659-1568073016913-9b48d48f","meta_rev_id":"d0c62239f7877f274d38874979fa798cee0871b2","_number":680659,"virtual_id_number":680659,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-09 23:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-06 13:23:39.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-09-06 12:39:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-09-09 22:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-06 13:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-09-09 22:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 12:39:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-09-06 13:15:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-06 13:23:39.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-09-09 22:00:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-09 23:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a44cc25e1bd4ee8e7c0ce45b99be01fa2a1c470","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:02:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"fbf5af77b725950ba74d072c918bc47b7894ee99","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-06 12:39:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d50c597bd6cbe27690fc81fabc8e95155bb967b0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 13:15:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6df4d32207ce66fa04828ad903c8102c0766b72","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-06 13:23:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/680659/1 : FAILURE in 1h 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cf5bcde8c2c90614887275935dd3ec076d741b9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/36b980ed9f444dcb89a9c8cdd5aac2d1 : SUCCESS in 50m 08s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/160de43dbb4249e4b22fa6115af96e59 : SUCCESS in 30m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/d256ff0c60fb401296731e523ad90fd4 : FAILURE in 1h 25m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f9d3df9b353d4b5aa22108f57bb355e0 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/96f0fbe5e1074300a043762d405cc55d : SUCCESS in 1h 06m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95c83e60c9449efa60196994322a634 : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e75ccd294f443d6b75369613f98d8c7 : SUCCESS in 24m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/855ef442095244ffa6adc2f4dca37b08 : SUCCESS in 25m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/479b5949922f4d2e8e8a98b6b91077d0 : SUCCESS in 5m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/000f2d70eebf48ec87ca92be4fb5b5a1 : SUCCESS in 1h 44m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c454ed63852e4aaa80aa0d3656d4157f : SUCCESS in 1h 05m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f0a1d42064492686c13a360e5b1791 : SUCCESS in 1h 27m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/052be7719c804dbca81c959b64c0a52d : SUCCESS in 39m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5f1c52ed8eaf43feaab152169d0b177a : FAILURE in 50m 08s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/4a065778bc3e4aa3b31456593167cf36 : SUCCESS in 1h 38m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a9013568187f4b478e7cf560188cd5e3 : SUCCESS in 1h 35m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/544f25b9952c43278069ab1caf47f7cd : SUCCESS in 1h 02m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8752ed7bccd8401a9b3ceb4784f0f995 : SUCCESS in 1h 21m 35s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c19bd57d5ab9436d864889980ba92f7a : SUCCESS in 1h 37m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/860236c26c8d4cfe8722d02ace8a77cb : SUCCESS in 1h 38m 23s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/b39ee74cb4b84a369dfd069a8f553822 : SUCCESS in 1h 27m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23ea633fa1244b1ca054780a1ffcc3c3 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c1ee252ed56b23451cbf7778b59555f6db06e725","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-09 22:00:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f30bf471e7863901d2fc9d059e60a13b5bc9902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 22:01:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5318acdf44237a3f3f5815bf8edda4619e581afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:50:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ac3f65c207ab2118caf96469e0474d75097aa7ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:50:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/265702a969ac4f50bc08ec819b1fb2c1 : SUCCESS in 46m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778e14cb28ad4028aff92ab2ae5f5237 : SUCCESS in 11m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f106315f7a04249b4068517eb5d4c2a : SUCCESS in 25m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/05fe5502de314479aeeee9b8e969a7e4 : SUCCESS in 26m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bf25112238d54676b4c5e56a5bfa6752 : SUCCESS in 6m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/19464d8b13af45119d8d39bbf9cfc76a : SUCCESS in 1h 21m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6e13098ba414930b20add018b90ba4b : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dcf5a4c8d8f40b385d7700211e9f9da : SUCCESS in 1h 42m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/697c7e78ae234742b7ded722a1cfa5db : SUCCESS in 48m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0bea205ffc5c43aeb243a6a8dae8af93 : SUCCESS in 1h 24m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/586fa5a6758743a68c02432cb700d95b : SUCCESS in 1h 06m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6f5f460f4ad240778e8edbf653f56c13 : SUCCESS in 1h 06m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42972c8387ac4a1ea52fe990ae85e507 : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57725fc7d4513c37de597a4b86f8d6878e2be034","revisions":{"57725fc7d4513c37de597a4b86f8d6878e2be034":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:02:40.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/59/680659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/680659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/680659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/680659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/680659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/680659/1"}}},"commit":{"parents":[{"commit":"1482a8e44cfd31aaac948c6133e85f051b220ab5","subject":"Merge \"Filter placement API endpoint by type too\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1482a8e44cfd31aaac948c6133e85f051b220ab5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-14 15:50:39.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-06 11:02:40.000000000","tz":0},"subject":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds","message":"Increase TestDhcpAgentHA.agent_down_time to 30 seconds\n\nCurrently there is a delay (around 20 seconds) between the the agent\nupdate call and the server reply, due to the testing servers load. This\ntime should be higher than the agent-server communication delay but\nstill short enough to detect when the DHCP agent is dead during the\nactive wait during the DHCP agent network rescheduling.\n\n\"log_agent_heartbeats\" is activated to add information about when the\nserver has processed the agent report status call. This log will allow\nto check the different between the server updating time and the previous\nagent heartbeat timestamp.\n\nConflicts:\n    neutron/tests/fullstack/resources/config.py\n\nChange-Id: Icf9a8802585c908fd4a70d0508139a81d5ac90ee\nRelated-Bug: #1799555\n(cherry picked from commit d7c5ae8a0378fa50853e91241426b6470ffe7492)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57725fc7d4513c37de597a4b86f8d6878e2be034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57725fc7d4513c37de597a4b86f8d6878e2be034"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
