)]}'
{"id":"openstack%2Fneutron~680663","triplet_id":"openstack%2Fneutron~stable%2Frocky~I399f100fe30b6a03248cef5e6026204d4d1ffb2e","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1828865-stable/rocky","hashtags":[],"change_id":"I399f100fe30b6a03248cef5e6026204d4d1ffb2e","subject":"Use created subnet in port generator in \"test_port_ip_update_revises\"","status":"MERGED","created":"2019-09-06 11:13:02.000000000","updated":"2019-09-09 23:54:49.000000000","submitted":"2019-09-09 23:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680663-1568073289578-5163b24b","meta_rev_id":"84e3904bd92a6870cd3150916613389d6961a39e","_number":680663,"virtual_id_number":680663,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-09 23:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-06 13:57:56.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-09-06 12:41:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-09-09 22:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-06 13:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-09-09 22:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 12:41:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-09-06 13:16:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-06 13:57:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-09-09 22:01:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-09 23:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44f3d699a1a6f1d335558e84093a81cf59474725","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:13:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"3b84afd65dc6c1972bdbc8c650d3032b5548985d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:13:09.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 05a07662bba4cab3d5089f8c64a0da7cd087f14c","accounts_in_message":[],"_revision_number":1},{"id":"63b9e02f7942e0140d835493f5f9fd71ea4269a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-09-06 12:41:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bcfef04be62879163e83b5c02303bf0616b7d4b1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 13:16:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1537974a175a43462c6229211b61016b9813d273","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-06 13:57:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/680663/1 : SUCCESS in 1h 49m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0aadcc831e1bf1f0f382d691a3058e147adeb55f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 21:47:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68f30e1d6b7f4823a9aa7c8a279246f4 : SUCCESS in 11m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/155a6b9768b04a25a3a9f691356dd8c9 : SUCCESS in 24m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/86f6df2e90494023b0d5c9fb18e7e2b3 : SUCCESS in 28m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a0f1b1a8ae424831867e21ae5c935a3b : SUCCESS in 7m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80679e32dc074084b23b789ea7a5664e : SUCCESS in 1h 28m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a968e700acb04515a430a51ece95e39a : SUCCESS in 41m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6e58355da67d4056bf8868f8cd3aae74 : SUCCESS in 1h 00m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cfda583d18d4119a00ef6cf317b5a2b : SUCCESS in 24m 47s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/c3dfa04244c84fec8d335b55b3054993 : SUCCESS in 1h 09m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/581106acaf53430e9a6fea199bb746c2 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8c1b5f7cd990c17a16259ef4d991b97f31b8cf54","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-09 22:01:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b67b0a04cd67bffb5b04b00b333936b42a41afab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 22:02:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"885b3ba440672c5ff2857d8aae9eac6408e4e0a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:54:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ba09000fb24e6097fc784b8957a2b2 : SUCCESS in 16m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40e0710e8975497c8c55bab434d4952d : SUCCESS in 36m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9d411e5fa8a1404a88559625a83f14a6 : SUCCESS in 35m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3b8bc5f9d1644100b74f644f468c61a7 : SUCCESS in 9m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c95833e4b044c67b9527c2dde5a12ab : SUCCESS in 1h 43m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a0e8529e043455ba8fdc934e3692390 : SUCCESS in 1h 27m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e90bdbda188d42fb843eecf1df26520e : SUCCESS in 42m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/11089c95ecf942d69ef8b61eaa2ff5b7 : SUCCESS in 1h 06m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3869e7a0782b4c71b70193561f56c19f : SUCCESS in 31m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a667c0a31ba1451db3759b4cce2aae8a : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ca72383651cc32a53af0fe393524d301bb8562ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 23:54:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d249a80160693fccaca56cb193c160d9487548c","revisions":{"3d249a80160693fccaca56cb193c160d9487548c":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:13:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/63/680663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/680663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/680663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/680663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/680663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/680663/1"}}},"commit":{"parents":[{"commit":"1199207e7da977e1cd3c12f8a4242aafe6330d68","subject":"Increase number of retries in _process_trunk_subport_bindings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1199207e7da977e1cd3c12f8a4242aafe6330d68"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-13 17:06:44.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-06 11:13:02.000000000","tz":0},"subject":"Use created subnet in port generator in \"test_port_ip_update_revises\"","message":"Use created subnet in port generator in \"test_port_ip_update_revises\"\n\nWe are hitting sometimes a problem in \"test_port_ip_update_revises\" [1].\nThis happens because the port created doesn\u0027t belong to the previously\ncreated subnet. We need to enforce that the port is created in the\nsubnet specifically created in this test.\n\n[1]http://logs.openstack.org/69/650269/12/check/openstack-tox-lower-constraints/7adf36e/testr_results.html.gz\n\nConflicts:\n    neutron/tests/unit/services/revisions/test_revision_plugin.py\n\nChange-Id: I399f100fe30b6a03248cef5e6026204d4d1ffb2e\nCloses-Bug: #1828865\n(cherry picked from commit 872dd7f48400f0ee059876b01d4034aa75f97787)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d249a80160693fccaca56cb193c160d9487548c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d249a80160693fccaca56cb193c160d9487548c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
