)]}'
{"id":"openstack%2Fneutron~680666","triplet_id":"openstack%2Fneutron~stable%2Frocky~I9a54ae99d2d5dfda63cb0061bcf3d727ed7cc992","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1827363-stable/stein-stable/rocky","hashtags":[],"change_id":"I9a54ae99d2d5dfda63cb0061bcf3d727ed7cc992","subject":"ML2 plugin: extract and postpone limit in port query","status":"MERGED","created":"2019-09-06 11:18:27.000000000","updated":"2019-09-10 12:23:39.000000000","submitted":"2019-09-10 12:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680666-1568118219773-50878ba6","meta_rev_id":"65574008e77a37b5da76e71cebbf054c9aeec737","_number":680666,"virtual_id_number":680666,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-10 12:23:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-09-09 21:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-06 13:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"date":"2019-09-06 11:18:27.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2019-09-10 08:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 11:18:27.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-09-06 13:18:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-10 08:44:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-10 12:23:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d687ad101ee3efe97a8995e6a2ca35326d17039e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-06 11:18:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"95e97dfe9223a61f737ad27260cdda252cbbdca3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-06 13:18:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11c40c2bc484a288536549f33701463104b85534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:50:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e5f32c0d5e9e489886d9ca1f2e54ab2b : SUCCESS in 48m 02s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/e60ea1006172440f8781f8155e7ff990 : SUCCESS in 35m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/71135edb09d34223946eabfa1bd36790 : SUCCESS in 1h 34m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/b8485d90796c4f05b1a7d8c29d0663c8 : SUCCESS in 1h 13m 01s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/6d878bdd9e0c4d4b86a77cfef984d975 : SUCCESS in 55m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/8633be976abe40baa61d8664651bd0ae : SUCCESS in 49m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df573a6af3f46cf8c8a52889368c970 : SUCCESS in 12m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/520aa9dc04c54e98933b95215731a0a2 : SUCCESS in 25m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bbba1514bf4343149a3b9fa2be876754 : SUCCESS in 25m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7135a1044854284ac1558dc25eec706 : SUCCESS in 5m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04871c88c3e840b78da7d64ba2830b39 : SUCCESS in 1h 17m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/60d98c177ff04ee4b6bdb8b2e76b6406 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17e5265081814b5c97ee3c578510819e : SUCCESS in 1h 24m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f65834d5e87d44d5b4411cc518f0102c : SUCCESS in 41m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6fbc054a4bef4402a9167af35212e284 : SUCCESS in 52m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3e111173a47a4fe9824f726472b8dad3 : SUCCESS in 1h 37m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6953e12d7ac44ebb9c5d95e3d8a044c1 : SUCCESS in 1h 27m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/751bbfda3d484b93b814b0655f7b7db7 : SUCCESS in 1h 39m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef65ceca2d824838b741580c106afd17 : SUCCESS in 1h 14m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a459c695660467db6f235d39c6b7611 : SUCCESS in 1h 08m 17s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d010eca92b764098a9ec508e2fbb4c03 : FAILURE in 1h 39m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e601c529d294496a948659a483944995 : SUCCESS in 1h 49m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/401e03d5dccb428dbfeb8fd90b84f1e8 : SUCCESS in 1h 28m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/337c6916db484f7191aaf12710349ca4 : SUCCESS in 22m 45s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/ec70f853d07849bebc84e8a1adefd851 : SUCCESS in 1h 05m 59s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f4d2022c752742768099e78cb823088b : SUCCESS in 46m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b6e26b99f6b4c3b9e934014b0abcf4c : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2f146fab5fafba1863587af1a2b6c4e829ae6195","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-09 21:58:53.000000000","message":"Patch Set 1: Code-Review+2\n\nWaiting fro Stein W+1","accounts_in_message":[],"_revision_number":1},{"id":"ac46b2dbefb484b05eb7b56e508c287349aa8802","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-10 08:44:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d352a9ff2d97c2a9ea2ea02f394cdf5a90e9d846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 08:44:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc4c8e9c019df2bfab94d5c422d4d0665e776c04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:23:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da713ff6ba5428adde6488389a109108238a693b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:23:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/523aaa391a104ba1b8b7c71ad5e687c5 : SUCCESS in 53m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd3661a2cdf486e8824941228bfe581 : SUCCESS in 12m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5b7efa2fdf94bc1b8ed1f46b30841ec : SUCCESS in 27m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3271241bbf4849cba8fafa737ccd6e29 : SUCCESS in 24m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/86b6e2e0e62f4ce5844cd4d7967462ce : SUCCESS in 5m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9987b97d97684900ae48224b737f95ae : SUCCESS in 1h 37m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/10017005e04c4723ab297b262d21bb05 : SUCCESS in 1h 08m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89b3071c900f4f67bd57bfc37676521e : SUCCESS in 1h 24m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d3d730fc5724d5b9bedd94d34ebfe33 : SUCCESS in 44m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4632576bc1c3415f9d4be83c08d2b6ee : SUCCESS in 57m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cedf94cd15a34b74b2fe1344dc8be993 : SUCCESS in 1h 30m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5f25de2456ca48d08f7246809c83a93c : SUCCESS in 1h 23m 37s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35437c061fb9470eb899ea43ecec9d90 : SUCCESS in 1h 40m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae7c3166c3884960a526e2b9eddd44e8 : SUCCESS in 1h 39m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d0e3e9af19a84e098d51da8cfee68c20 : SUCCESS in 1h 21m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66f073657f9b40a688905c6dfb5d7c24 : SUCCESS in 30m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/134316c34cea43a18376955d9dbc193f : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3133d8ae9d5fdd94da024343ba8edde7cb4b63c","revisions":{"e3133d8ae9d5fdd94da024343ba8edde7cb4b63c":{"kind":"REWORK","_number":1,"created":"2019-09-06 11:18:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/680666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/680666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/680666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/680666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/680666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/680666/1"}}},"commit":{"parents":[{"commit":"1199207e7da977e1cd3c12f8a4242aafe6330d68","subject":"Increase number of retries in _process_trunk_subport_bindings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1199207e7da977e1cd3c12f8a4242aafe6330d68"}]}],"author":{"name":"Gabriele Cerami","email":"gcerami@redhat.com","date":"2019-05-13 09:38:50.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-06 11:18:27.000000000","tz":0},"subject":"ML2 plugin: extract and postpone limit in port query","message":"ML2 plugin: extract and postpone limit in port query\n\nas with https://review.opendev.org/#/c/656066/ if limit is applied in\nany place other than at the end of the filters, sql alchemy will return\nan error, and possibily we could return less result than intended.\n\nChange-Id: I9a54ae99d2d5dfda63cb0061bcf3d727ed7cc992\nCloses-Bug: #1827363\n(cherry picked from commit 94bc403078b1e5d2cddc697fc59b7f1d2d8a6045)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3133d8ae9d5fdd94da024343ba8edde7cb4b63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3133d8ae9d5fdd94da024343ba8edde7cb4b63c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
