)]}'
{"id":"openstack%2Fneutron~682391","triplet_id":"openstack%2Fneutron~master~I4a0e97845a40888f7f2c7267d9adb7be663ee380","project":"openstack/neutron","branch":"master","topic":"bug/1843290","hashtags":[],"change_id":"I4a0e97845a40888f7f2c7267d9adb7be663ee380","subject":"Add policy config for get_flavor_service_profile operation","status":"MERGED","created":"2019-09-16 15:35:11.000000000","updated":"2019-10-12 08:20:50.000000000","submitted":"2019-10-11 18:17:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682391-1570817839386-47151255","meta_rev_id":"33308c12c43c93d84ee160859b8f518106e24da5","_number":682391,"virtual_id_number":682391,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-10-11 04:24:31.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-11 18:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-11 05:31:05.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2019-10-11 14:03:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-10-11 06:55:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-11 07:11:15.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-10-11 15:36:29.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-10-11 15:36:29.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"value":0,"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 16:32:13.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-09-16 17:08:37.000000000","updated_by":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-18 18:07:36.000000000","updated_by":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"reviewer":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"state":"REVIEWER"},{"updated":"2019-09-20 11:05:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-10 14:34:22.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-11 04:24:31.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-11 05:31:05.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-11 07:11:15.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-11 14:03:45.000000000","updated_by":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"reviewer":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"state":"REVIEWER"},{"updated":"2019-10-11 15:36:29.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-10-11 18:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72e5e934c4507c59631222f7be3c77d8876991e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-16 15:35:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a25b5f04b19f6b64d4382c53926a99b05324fc64","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-16 16:16:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-09-16/682391/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/0384495 : FAILURE in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"468b212e120760ecc28b2ea81f7aba85aac59ab4","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-09-16 16:31:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c084ede15e93cf7b82bdeea56693b18d9d450f92","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-09-16 16:32:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9055a2f97f721df7b63a83dbf6bcfd6d0062ec55","author":{"_account_id":15752,"name":"Intel SR-IOV CI","email":"sriov-ci@intel.com","username":"sriovci","tags":["SERVICE_USER"]},"date":"2019-09-16 17:08:37.000000000","message":"Patch Set 1:\n\n* sriov-test http://52.27.155.124/sriov/682391/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"334686fbcb10747e8f67ff73fa217f44f9acc3f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-16 17:37:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/682391/1 : SUCCESS in 2h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"66269894502a1c1f704b00f6f673bb04086dd68b","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2019-09-16 19:09:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8dc64c7239a9481634a1a22c13ecbcdb0d449df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:43:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4725e8ecb1ea49049f16d02479d12dd7 : SUCCESS in 52m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5b1931567d1946659e4834ab8a161c0d : SUCCESS in 36m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/88898e26d14246efbe62273551e629de : FAILURE in 1h 29m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0594f76d308e4b58aff233a41c83578f : SUCCESS in 1h 01m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6fb3258062774ba29c6754038d45ad0b : SUCCESS in 1h 02m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7702329cbf784808b0ea95d9a1079992 : SUCCESS in 59m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/611fd09cc0bc42e18f4eff3bd87689ce : SUCCESS in 31m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54c40b1970ea440b838a0ece4efbfccd : SUCCESS in 22m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d956db7d201b41e9b5da78158316b2b9 : SUCCESS in 15m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5beda412b1e34106b6523979eebe3373 : SUCCESS in 32m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1479870ad0e14ee6ada86fc840187066 : SUCCESS in 23m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ca737598dad4de9949a0c76a326751d : SUCCESS in 24m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646f626563904ddcb9959090b19503b5 : SUCCESS in 9m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f046e46c61e8439f886eb85ec1e7b9ec : SUCCESS in 1h 07m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/14e33fefc8904a0693015788b08fe06a : SUCCESS in 1h 17m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9ad7ff37044a436ab06b3c7b342a0e8b : SUCCESS in 52m 22s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/eff0cabc111f4864840ba16c008d7efa : FAILURE in 53m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/56c5c744c0dc4ba681b90c227b45d8d8 : SUCCESS in 1h 03m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/672b17dee1fc4f98a8d002efb2a5094c : SUCCESS in 2h 02m 37s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f980cbe2ff940beb6c21646d68daa63 : SUCCESS in 1h 11m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ade934446ca14e778408afd7c371d7cf : SUCCESS in 1h 27m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/25909aca79864312a9d2b6c080c1ba63 : SUCCESS in 1h 14m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/429717fae3524d9e944bf48117c0313f : SUCCESS in 1h 17m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/34004716ec0c4c4a9baadfb9ac045043 : SUCCESS in 1h 48m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c4809d743df4b0eb0bfba9334e19698 : SUCCESS in 1h 55m 50s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5af18e068c444383891a2d6751ecf297 : SUCCESS in 1h 40m 14s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/28c890226a024f4ab24f9b432ca28eb4 : FAILURE in 1h 32m 38s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d9441f7361a24ea6b7b389b920588bda : SUCCESS in 1h 09m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3b87c5a97a9f4600aa20a90dbfd45754 : FAILURE in 52m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85a663e6bf5a49ecbd52b138c09f9278 : SUCCESS in 2h 21m 48s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/15f949bed37a488fad1aafd835ef59d4 : SUCCESS in 2h 02m 44s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3290455287dd4597807d6b9d0fc4f3e1 : FAILURE in 1h 06m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e333d3245814fd7b4c2183f76952d08 : SUCCESS in 1h 00m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e901ed9538984c66bfdf1bee8981420e : FAILURE in 37m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/23596e1fdaf44d0b91ee2174e45a1d79 : SUCCESS in 48m 01s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df24b36e62f54c1b95e96467bae8fd6e : SUCCESS in 58m 10s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c81513e630b344c69da488b8b8198515 : SUCCESS in 1h 33m 48s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c4c7702a1799443c82569c78a00fd486 : SUCCESS in 2h 48m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e53b6b8dc2c274bfc4aa4fb989bf139f5e316acb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-17 07:08:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"135e1c864c1ce698761697921b26d1b7f894e265","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-17 07:09:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"93fec4b788e8bf73a134119fcd6c4fc72bc76448","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-17 07:41:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-09-17/682391/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ee2c650 : FAILURE in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05dbdff457966a231a09e14fc76a904f85fbe359","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-17 09:04:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/682391/2 : SUCCESS in 1h 46m 28s","accounts_in_message":[],"_revision_number":2},{"id":"56eb2d605608e1527c3c5032510975274ece5e3f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-17 12:47:21.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/91/682391/2/check-mlnx/Neutron-ML2-Sriov/3a7d076 : SUCCESS in 1h 21m 50s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/91/682391/2/check-mlnx/Neutron-MACVTAP-ML2-Sriov/6aedf62 : FAILURE in 1h 26m 24s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"a55922149ca701e81181073793ec1565f360c94c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 16:57:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a1338e9a2c804ce3a83b48976c0e4200 : SUCCESS in 52m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/698d3edc224749d1bd8e119f7fc16fb4 : SUCCESS in 36m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cc625a6c8c6247acb1a5cd505b95ea0f : SUCCESS in 1h 29m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a9885965a92949f7ae2d8e82fe4af2b7 : SUCCESS in 1h 06m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ff1581dced114c2992998509cf80c016 : SUCCESS in 58m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48c6f098f2214bf5b22a39677402a9cc : SUCCESS in 1h 03m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/251462e2f4604747b689b123f4415b89 : SUCCESS in 32m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64d1426b3f11489091806f004b299a10 : SUCCESS in 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e49f4c88298444d0895fa467ff856c5c : SUCCESS in 12m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f88cc133950940b386693f76d6b31355 : SUCCESS in 24m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f91aaaa2900341b2bded83d475b778f9 : SUCCESS in 22m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7675bb55a2f34c598d1e048d4539dd45 : SUCCESS in 22m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce8b6a632c84b0b9428aa2d3560ccf0 : SUCCESS in 9m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1398b3cfd597433482b89b58c51d475e : SUCCESS in 56m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/673296da92b2472a9f0ba3a9a115c589 : SUCCESS in 1h 30m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/71fc450354034c38adad8591037acdfa : SUCCESS in 48m 52s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9a6533879510427f9de86d435f473c47 : SUCCESS in 46m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/102051cb618f4d53a48dde3d3f30c639 : SUCCESS in 1h 07m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6fb1bf8953a441208b6264a5e878f76d : SUCCESS in 1h 37m 45s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cdf6e2e6a68b42f4a92a6895cb0c42e7 : SUCCESS in 1h 17m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b89e3e1548fa4d0ebb99db09f44c6f36 : SUCCESS in 1h 11m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb8233c7c7dc46f49637a556287f6467 : SUCCESS in 1h 17m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a6796fc848a4db79076d5ecd3838650 : SUCCESS in 1h 03m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/954876cad32f49648cd61412f2b29c07 : SUCCESS in 1h 48m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae1261d746114d35847310019245cd9d : SUCCESS in 2h 02m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a4757ebd43042c1ac323903858c5707 : SUCCESS in 1h 24m 21s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/19175669df0444c780246726b5766ea0 : FAILURE in 1h 37m 12s (non-voting)\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6fb750948d074ca7a368ea2446c339f8 : SUCCESS in 1h 04m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6f327eb7a51e4d45b93a35403f68f035 : SUCCESS in 54m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b10197edbb7845199f9503c7282d1670 : SUCCESS in 2h 37m 33s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a14c13654c5c42f4a422ef6862095e8f : SUCCESS in 2h 19m 44s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/56d80abaf2b24e42891a793f442447ba : FAILURE in 1h 02m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3610a568ef4ab8a4980d1f59ea3747 : SUCCESS in 1h 09m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f7c15ea3fbb54edca61275a53e9bfb6c : SUCCESS in 1h 31m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03675f8149aa4ebab0075b48478a7d83 : SUCCESS in 54m 52s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8ba96f44cc146ee8ef1ce5e50e720ee : TIMED_OUT in 2h 17m 27s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/37dbc8e97e6c486e95b7596aebf43441 : SUCCESS in 1h 47m 54s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2056f0902d7a44358551629b68767ab1 : SUCCESS in 2h 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4618ce39bfdb08729726d70c0e85ab5596e0b83","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2019-09-17 19:52:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6ece708e65a522775842cf3811379c34662fb455","author":{"_account_id":17896,"name":"Rick Bartra","email":"rickbartra@microsoft.com","username":"rb560u"},"date":"2019-09-18 18:07:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"77a3a4ee139d03d814c8b7b9a34d99167d1d7942","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-20 11:01:06.000000000","message":"Patch Set 2:\n\nThe policies \u0027create_flavor_service_profile\u0027 and \u0027delete_flavor_service_profile\u0027 in neutron/conf/policies/flavor.py correspond to API operations \"Associate flavor with a service profile\" and \"Disassociate a flavor\" in the API reference [1].\nThe proposed policy \"get_flavor_service_profile\" would correspond to an API operation \"GET /v2.0/flavors/{flavor_id}/service_profiles\",\nbut such API operation is not defined now. I confirmed there is no such API in my local env [2].\nThe associated service profile information can be retrieved by \"Show flavor details\" operation.\n\nEach policy defined in the code should correspond to each API operation,\nso I am confused with adding a policy which corresponds to no API operation.\n\n\"Disassociate a flavor with a service profile\" operation also works [3].\n\nIt seems I need to investigate what happens more.\n\n[1] https://docs.openstack.org/api-ref/network/v2/index.html?expanded\u003d#networking-flavors-framework-v2-0-current-flavor-service-profile\n[2] http://paste.openstack.org/show/777965/\n[3] http://paste.openstack.org/show/777966/","accounts_in_message":[],"_revision_number":2},{"id":"c0f6890dfe340e41647ecdc726a244b9e2841256","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-07 08:03:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nAs I commented in the bug [1], we need to define \"get_flavor_service_profile\" policy explicitly.\n\nMy inline comment is for better documentation.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1843290/comments/10","accounts_in_message":[],"_revision_number":2},{"id":"1d68efece87b252cd9e2e81b7766514f8d189530","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-10 10:58:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d585524fdadec335b389a86a114765687313744","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-10 10:59:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"411289cecbca5099372989270a982adfc8d2acb2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-10 11:48:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-10/682391/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ead872a : FAILURE in 48m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7dd594dc57e322d26a8539c3a7af0e4b3f75d159","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-10 13:03:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/682391/3 : SUCCESS in 2h 04m 23s","accounts_in_message":[],"_revision_number":3},{"id":"8f43ab4b217e44a081998a613d323814530a31c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:23:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4257599ad56f42dd8e6841e968a2197c : SUCCESS in 54m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/987c0cbc4aa44afdbadd383d85eb7966 : SUCCESS in 29m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7a791cfe51644b5da1d5bae76fcb2ee3 : SUCCESS in 1h 29m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c0863b1b98744e11a6a62d3544d8ebb6 : SUCCESS in 1h 01m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a26ef9b6772e430b9902753c0b035d5c : SUCCESS in 1h 00m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8be2e09d0eb4f7eb22e829c7fbfb9e7 : SUCCESS in 58m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/894e31f94ce9450b9f02446d896b905b : SUCCESS in 27m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/246d7984f99740ebac263fcb31626dc6 : SUCCESS in 23m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02b7d3f83d24f829394e82b3bded670 : SUCCESS in 12m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ffb690cb71eb45e6b1f8c02b106397fd : SUCCESS in 23m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34283cfcd96b47caabcc1e6960ca7b34 : SUCCESS in 22m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0919cabb22a64417955718b7fd9e75d6 : SUCCESS in 23m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a856a51ec64a2a9c63e743b5bda475 : SUCCESS in 8m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/48d4cb8fa68c418b8d79772ad0aa4953 : SUCCESS in 1h 02m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c4df0b11520b4299a441d7afe632caa7 : SUCCESS in 1h 11m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d6d9ff6ce5b451da18cbac44f5e8e6f : SUCCESS in 49m 29s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3c81ca3a7d1445b4b18fd861dbfbb044 : SUCCESS in 47m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ffdcfaaa97164b80bd4c71e4d2ab246b : SUCCESS in 1h 04m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/04ad748b9d5147d78c814389fc869a08 : SUCCESS in 2h 05m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f9c63979d75e41219ec4f1609de26695 : SUCCESS in 1h 34m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b8789b7a373640b4b01cfa49f6110f3f : SUCCESS in 1h 18m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/adcdf16eff5147869355da68d0700391 : SUCCESS in 1h 08m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/048bda613548459db180e65691e6b0a6 : SUCCESS in 1h 23m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/88e902b1c5604c8ab950ffffb5f3753e : SUCCESS in 1h 36m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3bd413aba5b14a3cab099c4a4dab1c8e : SUCCESS in 1h 51m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c89d51ed9bc459195e879dd1479e615 : SUCCESS in 1h 11m 22s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/49759675e12f40438a32d9534d51c296 : SUCCESS in 1h 54m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0a31b6deba934757b42cfb9477ac5dd2 : SUCCESS in 1h 00m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/25612bfb19834d5b9906a1ee8aa48841 : SUCCESS in 54m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40bb97d47b62465d8590dc6e7f941b35 : SUCCESS in 2h 07m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/599fadac834143269322ef846e60c9c7 : SUCCESS in 2h 19m 04s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/120c8649cd404a10b75f8b24ddb90d70 : FAILURE in 1h 26m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a71c479ad11d4758b60205feaabc5e52 : SUCCESS in 1h 07m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3e6d8863079c49ef9329eb7a5f0c3d17 : SUCCESS in 1h 23m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/30d91be30cce4a8080981614198af899 : SUCCESS in 49m 23s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c2ed536fdd74cb9bfac3016090f683a : SUCCESS in 1h 04m 11s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/36d2d42b78ec44f59c0a03a37afc63ec : SUCCESS in 1h 23m 15s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/06da0d29b0c44019bb56e3f268317289 : FAILURE in 2h 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"131b2b9c9c76853d56dfedc6fd47b007b15a92a2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-10 14:34:22.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/91/682391/3/check-mlnx/Neutron-ML2-Sriov/e778e0e : SUCCESS in 1h 21m 53s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/91/682391/3/check-mlnx/Neutron-MACVTAP-ML2-Sriov/27f0e78 : FAILURE in 1h 17m 28s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"77c1c1a9b34e7b7f98f8524bbc7854d1148e608b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-10 15:32:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22399c7eb8c98bb5d272ff01efb1eff61cd115d1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-10 17:00:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f2412655beb6ec0b8f3af70b57990d128a367170","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-10 17:38:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-10/682391/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/cf92e27 : FAILURE in 37m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"677b00274ebd3fe27e889f52252e0bff186081c0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-10 18:43:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/682391/4 : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":4},{"id":"4a621cb65b8c917cb868604799daa46dcd4dc0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:36:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5d307933bf54ad2bc45d773ce52e188 : SUCCESS in 52m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cebd817f1f1a45ad83771b85242a1e87 : SUCCESS in 36m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/73cbf13ca12a4f559a60955bd96de651 : SUCCESS in 1h 44m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a7e01d905be49e7a13de1af5238a75f : FAILURE in 1h 23m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a30b8dac29bd4183b0fa6bbbc9131889 : SUCCESS in 1h 00m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d53855c0edf14a38adb8efda1cb5ab5c : SUCCESS in 1h 16m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffb834315de6487ca2d2f0d2f6f8659b : SUCCESS in 27m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/339d708f9e91425ba78180bddb72365e : SUCCESS in 22m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f32a90d3ec1415cb5d2b6230dc0af15 : SUCCESS in 12m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae17dab15fd043d9b580e08d274797a4 : SUCCESS in 29m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6baca8635fe443e1b452d99ee8d7e1b1 : SUCCESS in 20m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11842abd6704417f8b1a6acbdfc05902 : SUCCESS in 26m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b167c375494f480f94835f64058c11c1 : SUCCESS in 8m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f8ac05bd52e447dbe8f9e61b7097f0a : SUCCESS in 56m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ef41670dea9b4c8bbb739f365640690e : SUCCESS in 1h 32m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cbb3451b47554dfa8c2038e3549b6f76 : SUCCESS in 45m 15s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/33b62669074f429a90509def74a7a2bf : SUCCESS in 47m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3b581875892c4c918939759cf719e514 : SUCCESS in 1h 09m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/daf47b9e621a4a53bbd89e561cdb89ed : SUCCESS in 2h 00m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a2218d20d9fe43bcb560978b80626e67 : SUCCESS in 1h 35m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/56aed7f71297463fb6b189ab77798731 : SUCCESS in 1h 05m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3cc62d727ef435186c32d80ecaf3021 : SUCCESS in 1h 12m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ffb5e9e8f0f14c7d8e01568faa431f13 : SUCCESS in 1h 15m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee6d72c72290475da77bce68f5b4d25d : SUCCESS in 1h 45m 45s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/896f4df1b5b24c22bded108dbdfd6751 : SUCCESS in 2h 18m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7734a7cd1b64c09af016eda27c2ec32 : SUCCESS in 1h 38m 04s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/d6e3d1ee79b94729b8a3679cd8b9d70c : SUCCESS in 1h 40m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9685a082247c4102a1ec1a1744e1cee0 : SUCCESS in 53m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f63b585a2ba54bafb7c26e16ff366ae8 : SUCCESS in 53m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da213cc7db88438aa1bd781b9a6fbde4 : SUCCESS in 2h 24m 48s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ed3954d500394b94b84cdd9b87f2405a : SUCCESS in 2h 27m 58s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/93cc8ee126cc45128fc19e2f7fddc68c : FAILURE in 2h 01m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f6b4d0631ce4cc392640210f5226a47 : SUCCESS in 1h 00m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/02304a5f9f0c4cd3a1a5fa38d91deab9 : SUCCESS in 1h 22m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4ff705f493e4e58bd72f493faaf08f0 : SUCCESS in 51m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/294de4d61371477699393719aac7754a : SUCCESS in 1h 10m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/936535f82ae44e32a40f335fc94dedc2 : SUCCESS in 1h 20m 30s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/65489c55077344d5b7f59345e7eda29f : SUCCESS in 2h 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e1125f007fa536b8dbc97191ec4b3b7453d0a88e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-11 03:53:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2320233347778dc765d8cb76c1fab548ec15d5d5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 04:24:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-11/682391/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a09a040 : FAILURE in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1aa604bd6e4643f6c215abfd91356d5e71b87b25","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-11 05:31:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/682391/4 : SUCCESS in 1h 37m 35s","accounts_in_message":[],"_revision_number":4},{"id":"83e33ee4c4cf3c941ccb41321c00fbe7fe112c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 06:23:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/451ca2021f7e40e7a7b7b1b56f73421e : SUCCESS in 42m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/125b19090d4a47978a692c6d1bf769d0 : SUCCESS in 30m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/dde2a2198a734ebdbdf79b70538591d5 : FAILURE in 1h 14m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bed9f0930cd04fde997c20f3f7eab803 : SUCCESS in 1h 06m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/84fc29fae1694a508caa96ddd3276c6f : SUCCESS in 50m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e82b86ae73b8460b936307c3c1b91d4a : SUCCESS in 53m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09243b2e519a407cbf200971aef73f49 : SUCCESS in 26m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ceecdb791d04a08ad262a50d9f49635 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ef8abaa3384165af3b1ba3088a1da0 : SUCCESS in 11m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/219f1b67bf3d4c27aa9ab5d7a08b4f81 : SUCCESS in 25m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa599a9fbb604338a814276555b1c130 : SUCCESS in 27m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2bb2e437b61454e83f7b635e86d7b3f : SUCCESS in 25m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5823cb6123634d35a82dd7142e42a932 : SUCCESS in 7m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/773d84ca79b740f8ab391fa752690f23 : SUCCESS in 50m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/525c2836416043508b6e6bcc300bd107 : SUCCESS in 1h 08m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/00a7e065d1184376b0dcee7640188b93 : SUCCESS in 37m 13s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/af9018e9c7b344748fe683970ded42ed : SUCCESS in 46m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5a37c631cfa24f42bc5a41e3c06ca499 : SUCCESS in 1h 03m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f0b847c88b3c465db1112970bc8f8b4f : SUCCESS in 1h 57m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2081a1f9332a488da996f7e302895d90 : SUCCESS in 1h 18m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/78c135ba7b764ee6830508e3be42e28b : SUCCESS in 1h 13m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bbff530ee3f417e992c057ff1da6c77 : SUCCESS in 1h 05m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e7fc6d533bf47da8a41886b456cda49 : SUCCESS in 1h 09m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/12757bb1391346f0a48ffbf5bfef5c5a : SUCCESS in 1h 47m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d814c277e0c14a9092af235fcb41177d : SUCCESS in 1h 48m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b2313af59384fb5b9ebf370974abda5 : SUCCESS in 1h 22m 08s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/e28219adf2c24bd886295ddb64bd246b : SUCCESS in 1h 31m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/258b456e47ef474c8c7597008561bbcb : SUCCESS in 1h 00m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b5c4b2964fcc46cfbc09fb09fa3c07fb : SUCCESS in 50m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85d449ea61de427983a439c78e7f44ef : SUCCESS in 2h 20m 43s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e226cf4589594152959f6168f0273be9 : SUCCESS in 2h 24m 54s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/bdbc3a598510450ebb3846962865358b : FAILURE in 1h 28m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/310e507fa35a439a9a9e783e6a187de2 : SUCCESS in 1h 34m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0bd5115249974462b380ed0efa15112c : SUCCESS in 1h 22m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4dba97d8ad64af187372d00de9badbd : SUCCESS in 47m 13s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/96ba5b84971b4511a5c978643d81ee66 : SUCCESS in 1h 11m 22s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/53c2ce72720f4f34bc3882b52b390818 : SUCCESS in 1h 15m 29s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b28273b0d6244d0dbd568ced6fd49809 : FAILURE in 2h 22m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"633bc60a02c4b3e5d82a6dafced6473dcadade96","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-11 06:55:10.000000000","message":"Patch Set 4: Code-Review+1\n\nThx Akihiro for the update :)","accounts_in_message":[],"_revision_number":4},{"id":"0d444b3fd3e82fd0e23bbc77dfd43bd062021f1e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-11 07:11:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"241f8f25aa419eda2e71a27c20ce91b814077bf9","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2019-10-11 14:03:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"13e4d770554901b201599740f48698431f1a0f88","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-11 15:36:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"2f6f28bb17d03ee73a0a61d56756a91e42b3693b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:36:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"000443b449ac6720c89c6f261f42fdfe5f83ad64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 18:17:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9234673a3f444d599c07783b8cee0e16 : SUCCESS in 1h 04m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ce27e7af609d47998d43dd6f43afd249 : SUCCESS in 1h 12m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79de7855320849e89c2571ee36433a80 : SUCCESS in 1h 04m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1483a9255a442b6a22d06cf3b356e7a : SUCCESS in 54m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7102c44296784fa099e58b23aa9b6377 : SUCCESS in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0875a852425948628d5fb9eb9afd4f5f : SUCCESS in 11m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1976703075a648ee83fd85dd32036e31 : SUCCESS in 25m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0299ac9f9fc943caa04ff82a3c877114 : SUCCESS in 20m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b614ff55e724bc1a8b7be94d412b1b5 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883e4639af9c40a5bc2eb5a3ba063af9 : SUCCESS in 9m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0dd02ae3af6545258e2454055acc8261 : SUCCESS in 1h 04m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b0b3c137d594074a666f1ef2122b984 : SUCCESS in 1h 23m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ecd51e7c85104820980988f48e696971 : SUCCESS in 56m 52s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/70686a1ad9854c728b268d4de79a7b93 : SUCCESS in 41m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/336ec9217a6c47a7b9853beb7937435b : SUCCESS in 56m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/85dad0cf112e44f9a947ac70a4fd11ec : SUCCESS in 1h 35m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d573d429b5ad473eabb61ecb8cb568a7 : SUCCESS in 1h 33m 45s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e477936fa65e4f3db809726e9f35f57d : SUCCESS in 1h 12m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eae6ddf3aaca432eaea469d671964a11 : SUCCESS in 1h 18m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9d680a4bba334448b91de99a69e6396e : SUCCESS in 1h 19m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5245574215f6415994a49e3bcf5ee627 : SUCCESS in 1h 02m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e592d0e2e0524e3ab5857f08de928d95 : SUCCESS in 2h 33m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efde4378d21946cf8cf34dfbe4cc3778 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":4},{"id":"afb1de6fd9defae0209f457b700ef279d2e55d03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 18:17:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"26da0f76189f420a66b8f3190133408eec401164","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 18:19:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e7d1a84ce24bc48ede0d7375777b8e : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4},{"id":"571ca402a8ada80424a5d21504544cd0129e6db4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-12 08:20:50.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit aa0fe1dda846cd83d13cf467648e5d27ffd8d257","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ac0a426017e92d5d73e97c26ef18a888e02da104","revisions":{"02adea047be5fc48af96211306464234520f4133":{"kind":"REWORK","_number":1,"created":"2019-09-16 15:35:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/91/682391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/682391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/682391/1"}}},"commit":{"parents":[{"commit":"0809f4e224c2576060f563e9eb1796456fd14cfb","subject":"Merge \"Remove unused global_refresh_firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0809f4e224c2576060f563e9eb1796456fd14cfb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-16 15:33:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-16 15:35:05.000000000","tz":120},"subject":"Add policy config for get_flavor_service_profile operation","message":"Add policy config for get_flavor_service_profile operation\n\nPolicy for this API call was commented out but it seems that\nit is needed to fix Parole test failure, see related bug description\nfor details.\n\nChange-Id: I4a0e97845a40888f7f2c7267d9adb7be663ee380\nCloses-Bug: #1843290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02adea047be5fc48af96211306464234520f4133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/02adea047be5fc48af96211306464234520f4133"}]},"branch":"refs/heads/master"},"b149eb7e03f7b5e56969e79728c18e6a23431fec":{"kind":"REWORK","_number":2,"created":"2019-09-17 07:08:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/91/682391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/682391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/682391/2"}}},"commit":{"parents":[{"commit":"0809f4e224c2576060f563e9eb1796456fd14cfb","subject":"Merge \"Remove unused global_refresh_firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0809f4e224c2576060f563e9eb1796456fd14cfb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-16 15:33:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-17 07:08:41.000000000","tz":0},"subject":"Add policy config for get_flavor_service_profile operation","message":"Add policy config for get_flavor_service_profile operation\n\nPolicy for this API call was commented out but it seems that\nit is needed to fix Patrole test failure, see related bug description\nfor details.\n\nChange-Id: I4a0e97845a40888f7f2c7267d9adb7be663ee380\nCloses-Bug: #1843290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b149eb7e03f7b5e56969e79728c18e6a23431fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b149eb7e03f7b5e56969e79728c18e6a23431fec"}]},"branch":"refs/heads/master"},"ca4d7103642e65bc1c414a479f5013ab6b7a59c9":{"kind":"REWORK","_number":3,"created":"2019-10-10 10:58:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/91/682391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/682391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/682391/3"}}},"commit":{"parents":[{"commit":"0809f4e224c2576060f563e9eb1796456fd14cfb","subject":"Merge \"Remove unused global_refresh_firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0809f4e224c2576060f563e9eb1796456fd14cfb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-16 15:33:03.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-10 10:56:37.000000000","tz":120},"subject":"Add policy config for get_flavor_service_profile operation","message":"Add policy config for get_flavor_service_profile operation\n\nPolicy for this API call was commented out but it is needed\nfor DELETE flavor_service_profile operation as Neutron\u0027s policy\nmodule looks for ability for SHOW command when DELETE is rejected\nto ensure what error message should be returned to the user.\n\nThis patch will fix Patrole test failure, see related bug description\nfor details.\n\nChange-Id: I4a0e97845a40888f7f2c7267d9adb7be663ee380\nCloses-Bug: #1843290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca4d7103642e65bc1c414a479f5013ab6b7a59c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca4d7103642e65bc1c414a479f5013ab6b7a59c9"}]},"branch":"refs/heads/master"},"ac0a426017e92d5d73e97c26ef18a888e02da104":{"kind":"REWORK","_number":4,"created":"2019-10-10 17:00:28.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/91/682391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/682391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/682391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/682391/4"}}},"commit":{"parents":[{"commit":"0809f4e224c2576060f563e9eb1796456fd14cfb","subject":"Merge \"Remove unused global_refresh_firewall\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0809f4e224c2576060f563e9eb1796456fd14cfb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-09-16 15:33:03.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-10 15:36:30.000000000","tz":540},"subject":"Add policy config for get_flavor_service_profile operation","message":"Add policy config for get_flavor_service_profile operation\n\nPolicy for this API call was commented out but it is needed\nfor DELETE flavor_service_profile operation as Neutron\u0027s policy\nmodule looks for ability for SHOW command when DELETE is rejected\nto ensure what error message should be returned to the user.\n\nThis patch will fix Patrole test failure, see related bug description\nfor details.\n\nChange-Id: I4a0e97845a40888f7f2c7267d9adb7be663ee380\nCloses-Bug: #1843290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0a426017e92d5d73e97c26ef18a888e02da104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0a426017e92d5d73e97c26ef18a888e02da104"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
