)]}'
{"id":"openstack%2Fneutron~684552","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I6482223ed2a479de4f5ef4cef056c311c0281408","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1841967","hashtags":[],"change_id":"I6482223ed2a479de4f5ef4cef056c311c0281408","subject":"Create _mech_context before delete to avoid race [refactored for Queens]","status":"ABANDONED","created":"2019-09-25 07:57:49.000000000","updated":"2019-11-11 15:11:36.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56add5945dc7b844b77b0e7d9159880f6b437221","_number":684552,"virtual_id_number":684552,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2019-10-02 08:05:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2019-10-18 07:46:47.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-02 09:34:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-02 08:48:50.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2019-10-02 14:41:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2019-10-14 14:54:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 07:58:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-25 07:59:17.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-09-25 08:03:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-02 08:05:18.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 08:48:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-02 09:34:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 14:41:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-14 14:54:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"04f0837376ef37ca3459d1cf3a8ccfc4ba22258f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-09-25 07:57:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"6a07ffca40ab449b263c2814ecb822358ff708fd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-09-25 07:58:03.000000000","message":"Topic changed from bug/1841967-stable/queens to bug/1841967","accounts_in_message":[],"_revision_number":1},{"id":"706099ca3b15a0335e8c94ec84b86d4c7fbcf3a3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-25 08:03:48.000000000","message":"Patch Set 1:\n\nI didn\u0027t backport to queens as I checked and didn\u0027t see support for priorities in queens. I might be wrong though.","accounts_in_message":[],"_revision_number":1},{"id":"b6b0c242ac49a85935437aed9e70a93595679fbc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-09-25 09:00:37.000000000","message":"Patch Set 1:\n\nthanks I check it","accounts_in_message":[],"_revision_number":1},{"id":"d1566257cc41a1b252dc40b82629acf6edb955f1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-25 09:13:59.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/684552/1 : FAILURE in 1h 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"092a5e004f46e61630d2e2a7653c9dd7cd91968e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 12:12:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/2c8de7af4e9a4d44a8be1b9091635fbf : FAILURE in 31m 46s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/e56a18fdf5314066988c96a8740a37da : FAILURE in 31m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/95c34e3182a44db3b6d9f88f7095dc4d : FAILURE in 38m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/e0ac5b09883f42ce9fed44c2bcd7a052 : FAILURE in 27m 52s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/5bb600c560ce434ead53d3ad76eb8e0e : FAILURE in 26m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f7e5876f8a1472dae14bf2ea131b309 : FAILURE in 10m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/139b671f6b674dfa9c35b43c135eb6c6 : FAILURE in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e4ee476900942159860656ccd15b2a1 : FAILURE in 4m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ae012d181bdd48e1b179fa5ed187274a : SUCCESS in 7m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c4957783de3c4e96b5e01e9826874b96 : FAILURE in 25m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b281919dd8414def9bbf7c50c9c1e030 : FAILURE in 43m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0151d033d34d405497ebb9a4e8f0eb8a : FAILURE in 26m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc107cfcc0c748c690260a8fd2df3a96 : SUCCESS in 22m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/499ae84de066442399350d8d06330c9f : FAILURE in 24m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a8a7aaafbc4544f3a7a78bb89995289f : FAILURE in 35m 45s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/f893c8b5ffea43f9a2b601ded5e1dd83 : FAILURE in 29m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8ee93c697814460c8b8bffbc07da15b3 : FAILURE in 32m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5405691dc68c44be9ca20726d17d9e56 : FAILURE in 27m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9fbfac681de4b9a9ea4e3273ea8ee96 : FAILURE in 1h 02m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf5eb8c29a3f4587af05a1fba442a630 : FAILURE in 56m 47s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/91ff52b0e11f470b98fb960ed3c85dea : FAILURE in 32m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8651157cedb433abae1232e9f754955 : FAILURE in 32m 23s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/03c5f4b3c7354a4bac419d51b66aa04b : FAILURE in 27m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/902bf0898a354171ab47d420d4c17edd : POST_FAILURE in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a3a4f9ffcd23b45bbd52cf07e8272cc299f764f6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-25 16:14:32.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/52/684552/1/check-mlnx/Neutron-ML2-Sriov/be6167f : FAILURE in 26m 53s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/52/684552/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/7afedda : FAILURE in 37m 41s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"b9ee5242d7d7aab41d4457ca3e4969b3f3c340da","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-09-30 10:14:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"59ce71fce994cf086f1ff6fcce59e883930c1c1c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-01 16:16:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1df00a486c21df4adf24c0e4403f3b2426a72428","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-01 17:30:58.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/684552/2 : FAILURE in 1h 14m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a160d9a7f8151ac05e77ba86f2cf338535925995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:27:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/5a8710a945a54e1c916c5ddad928b0c7 : SUCCESS in 46m 33s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/3d41f9bdf04d47a39825ce18af60e091 : SUCCESS in 47m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/39d00cadac514f5f859b69308375d3b9 : FAILURE in 1h 22m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/47b534cb48574586b1e6fca881dc5d6a : SUCCESS in 1h 02m 45s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/5c15a50501fa4c779663899838b084ba : SUCCESS in 56m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2789c8a7110748148759804b221ac458 : FAILURE in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c61ff75c2c84fa49232ba3a2e42d55f : SUCCESS in 26m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d9010eafa5864e909d0fb353fca5ec82 : SUCCESS in 24m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/52e4d9554334479b95108a94382a0988 : SUCCESS in 6m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/afe82bf8955940838fa2d11edcc0bb07 : SUCCESS in 1h 22m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b37e5cf099614410a2b77a169f72e2f1 : SUCCESS in 1h 11m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dd1311adc7d4d558dee4a53a9fb81aa : SUCCESS in 1h 35m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c88339fcd0e4d93a5419a47d3e40e02 : SUCCESS in 26m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8fb005d1f07749e2a8b6555985bdd6ba : SUCCESS in 42m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8b6a085b6e9c431b977f7d3c5b91d835 : SUCCESS in 39m 58s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/a8fe4f148b3149d39bc92b56f790c987 : SUCCESS in 1h 41m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fc05d2f091b04882b7c27d96358461d8 : SUCCESS in 1h 26m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a2dfce04173040b1ac3ebf653bf467c8 : SUCCESS in 1h 30m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6fb6255e88240eaae0d9ac3a7a0299f : SUCCESS in 1h 04m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/07a4a3ce566549b593078eb59c231e1b : SUCCESS in 1h 09m 41s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9742582d5b0b4745becc2858ff8cf1e3 : SUCCESS in 1h 43m 09s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/809932a30d154e6f9f3d19295e6f297f : SUCCESS in 1h 38m 48s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/63bca5f9b0e047c3a57c582604cc3dff : SUCCESS in 1h 32m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0431424b3744475ad7a539aed2530a5 : SUCCESS in 31m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4f32e43fef22c25fd1a88759b7108cb4009d4906","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-02 07:27:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"81a25dc3dcc11cbea152bfeb69ba3f5b6d963e9e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-02 08:05:18.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/52/684552/3/check-mlnx/Neutron-ML2-Sriov/fdd3920 : FAILURE in 33m 36s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/52/684552/3/check-mlnx/Neutron-MACVTAP-ML2-Sriov/cb6fdd0 : FAILURE in 35m 09s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"42b40fc43d50e3d7ffcc72520e356e49bea6815a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-02 08:48:50.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/684552/3 : FAILURE in 1h 20m 42s","accounts_in_message":[],"_revision_number":3},{"id":"cb18fecb51d8510fa57c93a3fc596e64e454e764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:34:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d8d6b83e666e490b9e7d176c99ac8ba3 : SUCCESS in 58m 26s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/047a8efdbaae40738b9bd12808e6f2f9 : SUCCESS in 47m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/42c77befa76f407db9ad2a318754eb5d : FAILURE in 1h 55m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f75e7b25e5ad4478b72bc768547dc68f : SUCCESS in 1h 12m 06s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/770ea6d40636473b88e02ba01dca7ccd : SUCCESS in 1h 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32629c02f59141bb8dfa04381d55d8b2 : SUCCESS in 11m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3706047c3dea45c0a7fc63f87c97451a : SUCCESS in 20m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37c501c5bb0041c497b40a9d4a0c4186 : SUCCESS in 28m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d43dd5f3407d4bbcb0395054723290ef : SUCCESS in 7m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4be0a761edbf49e89b3ceb8e388fa6e3 : SUCCESS in 1h 37m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54e46f0be1ca449588eb4fdee115c70f : SUCCESS in 1h 30m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16123605716b4dc5b734cfcaef978244 : SUCCESS in 1h 42m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c977673b5dbf4377ae4cce2c218e96c0 : SUCCESS in 24m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9a1c954d5ad44cfae5a7cc50dc205ea : SUCCESS in 36m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2c923e990f4546309b2557a04b545080 : FAILURE in 1h 17m 38s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/6c55417fce7d4de3bf2ca36c221f86f1 : SUCCESS in 1h 51m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/db48922a64f6465685706534d2d33abd : SUCCESS in 1h 59m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d560a19847d2499086988c1b484f24de : SUCCESS in 1h 31m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f88d2d0d3bf644eab1d02faf8d60339a : SUCCESS in 1h 41m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac33e27e008f4c59afad7e79a395382b : SUCCESS in 1h 12m 00s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2df7e2dc07954d1ba6a31ccc73abaf15 : SUCCESS in 1h 43m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/97d18d8017ed47c49ecebd430b29a3e9 : SUCCESS in 1h 31m 52s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/f7ad09efbec24050a4f093211c5f1650 : SUCCESS in 1h 31m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a24e3556230c41daa44901aa0cc33422 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":3},{"id":"6e0deb8366302b64ef97fb21a609e0eaf84a5ff9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-02 14:41:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b1f11f176df69d362cde1eacd3f10c612fe8ac04","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-03 11:19:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5203a6177f522cd790bacb2023fd315a44324c51","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-03 12:28:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"edb5077ea5a9c79ae96cc26f6463b96b12843f9c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-14 11:59:40.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks Brian for checking this.\n\nAnother weird option would be to backport the patch that introduced priority for callbacks. That was introduced in rocky, https://review.opendev.org/541766, what do you think about that?\n\nI don\u0027t want to be too offensive with this, just trying to help out our downstream teams, sorry for making trouble.","accounts_in_message":[],"_revision_number":3},{"id":"37e00609992212b0faf38e91bec24b97743d8562","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-14 14:55:11.000000000","message":"Patch Set 3:\n\nLajos - I don\u0027t think a backport of that neutron-lib change can be done.  I\u0027ve added Rodolfo to get his opinion since he commented on the earlier version of the patch in master.","accounts_in_message":[],"_revision_number":3},{"id":"7aa261a56b56f036584e6ae86cca5aefc5726164","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-18 07:46:47.000000000","message":"Patch Set 3:\n\nBrian: yeah, my thinking was the same, I added here to be sure that all options are listed, and weighted :-)\nBut let\u0027s wait for Rodolfo\u0027s opinion as well","accounts_in_message":[],"_revision_number":3},{"id":"56add5945dc7b844b77b0e7d9159880f6b437221","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-11 15:11:36.000000000","message":"Abandoned\n\nI think it is better to abandon this patch","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f90b643650df18bd9e5b31a57a84fd4620069393","revisions":{"f933b704536273a85ab16e0f1bf0014634277f04":{"kind":"REWORK","_number":1,"created":"2019-09-25 07:57:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/684552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/684552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/684552/1"}}},"commit":{"parents":[{"commit":"b3d3d6d64358f6e8340bf0dbdff716968bf0d92c","subject":"Merge \"Add info log about ready DHCP config for ports\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3d3d6d64358f6e8340bf0dbdff716968bf0d92c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 15:58:34.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2019-09-25 07:57:49.000000000","tz":0},"subject":"Create _mech_context before delete to avoid race","message":"Create _mech_context before delete to avoid race\n\nWhen a network is deleted, precommit handlers are notified prior to the\ndeletion of the network from the database. One handler exists in the ML2\nplugin - _network_delete_precommit_handler. This handler queries the\ndatabase for the current state of the network and uses it to create a\nNetworkContext which it saves under context._mech_context. When the\npostcommit handler _network_delete_after_delete_handler is triggered\nlater, it passess the saved context._mech_context to mechanism drivers.\n\nA problem can occur with provider networks since the segments service\nalso registers a precommit handler - _delete_segments_for_network. Both\nprecommit handlers use the default priority, so the order in which they\nare called is random, and determined by dict ordering. If the segment\nprecommit handler executes first, it will delete the segments associated\nwith the network. When the ML2 plugin precommit handler runs it then\nsees no segments for the network and sets the provider attributes of the\nnetwork in the NetworkContext to None.\n\nA mechanism driver that is passed a NetworkContext without provider\nattributes in its delete_network_postcommit method will not have the\ninformation to perform the necessary actions.  In the case of the\nnetworking-generic-switch mechanism driver where this was observed, this\nresulted in the driver ignoring the event, because the network did not\nlook like a VLAN.\n\nThis change uses a priority of zero for ML2 network delete precommit\nhandler, to ensure they query the network and store the NetworkContext\nbefore the segments service has a chance to delete segments.\n\nA similar change has been made for subnets, both to keep the pattern\nconsistent and avoid any similar issues.\n\nChange-Id: I6482223ed2a479de4f5ef4cef056c311c0281408\nCloses-Bug: #1841967\nDepends-On: https://review.opendev.org/680001\n(cherry picked from commit fea2d9091f71a2ec88318121ed9a22180e1ae96f)\n(cherry picked from commit f913394d70dede8d40d6e96485bd5333dfb6f43f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f933b704536273a85ab16e0f1bf0014634277f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f933b704536273a85ab16e0f1bf0014634277f04"}]},"branch":"refs/heads/stable/queens"},"fce19d002c023cbbf5dc144ac1713ddae702246c":{"kind":"REWORK","_number":2,"created":"2019-10-01 16:16:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/684552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/684552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/684552/2"}}},"commit":{"parents":[{"commit":"b3d3d6d64358f6e8340bf0dbdff716968bf0d92c","subject":"Merge \"Add info log about ready DHCP config for ports\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3d3d6d64358f6e8340bf0dbdff716968bf0d92c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 15:58:34.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-10-01 16:02:04.000000000","tz":120},"subject":"Create _mech_context before delete to avoid race [refactored for Queens]","message":"Create _mech_context before delete to avoid race [refactored for Queens]\n\nWhen a network is deleted, precommit handlers are notified prior to the\ndeletion of the network from the database. One handler exists in the ML2\nplugin - _network_delete_precommit_handler. This handler queries the\ndatabase for the current state of the network and uses it to create a\nNetworkContext which it saves under context._mech_context. When the\npostcommit handler _network_delete_after_delete_handler is triggered\nlater, it passess the saved context._mech_context to mechanism drivers.\n\nA problem can occur with provider networks since the segments service\nalso registers a precommit handler - _delete_segments_for_network. Both\nprecommit handlers use the default priority, so the order in which they\nare called is random, and determined by dict ordering. If the segment\nprecommit handler executes first, it will delete the segments associated\nwith the network. When the ML2 plugin precommit handler runs it then\nsees no segments for the network and sets the provider attributes of the\nnetwork in the NetworkContext to None.\n\nA mechanism driver that is passed a NetworkContext without provider\nattributes in its delete_network_postcommit method will not have the\ninformation to perform the necessary actions.  In the case of the\nnetworking-generic-switch mechanism driver where this was observed, this\nresulted in the driver ignoring the event, because the network did not\nlook like a VLAN.\n\nThis change uses a priority of zero for ML2 network delete precommit\nhandler, to ensure they query the network and store the NetworkContext\nbefore the segments service has a chance to delete segments.\n\nA similar change has been made for subnets, both to keep the pattern\nconsistent and avoid any similar issues.\n\nNote: This is not clean backport. The original fix on master used the\nregistry.receives decorator\u0027s priority parameter (see [0]) which was\nintroduced in Rocky (see: [1]). To fix the problem on Queens this patch\nbackports ps2 from the original solution (see: [3])\nThe differenc from the original, merged patch:\n- On Queens there is no priority parameter for registry.receives\ndecorator, the workaround is (as in ps2) to register for BEFORE_DELETE\nNETWORK callback and get a copy of the network before segments can touch\nprovider information.\nThe differences from PS2 of the original patch:\n- In db_base_plugin_v2.py use registry.notify instead of publish.\n- In plugin.py change the decorated callback method signatures as it is\nsuitable for notify.\n\n[0]: https://opendev.org/openstack/neutron-lib/src/branch/master/neutron_lib/callbacks/registry.py#L67\n[1]: https://review.opendev.org/541766\n[3]: https://review.opendev.org/#/c/679483/2\n\nChange-Id: I6482223ed2a479de4f5ef4cef056c311c0281408\nCloses-Bug: #1841967\nDepends-On: https://review.opendev.org/680001\n(cherry picked from commit fea2d9091f71a2ec88318121ed9a22180e1ae96f)\n(cherry picked from commit f913394d70dede8d40d6e96485bd5333dfb6f43f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fce19d002c023cbbf5dc144ac1713ddae702246c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fce19d002c023cbbf5dc144ac1713ddae702246c"}]},"branch":"refs/heads/stable/queens"},"f90b643650df18bd9e5b31a57a84fd4620069393":{"kind":"REWORK","_number":3,"created":"2019-10-02 07:27:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/684552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/684552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/684552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/684552/3"}}},"commit":{"parents":[{"commit":"b3d3d6d64358f6e8340bf0dbdff716968bf0d92c","subject":"Merge \"Add info log about ready DHCP config for ports\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3d3d6d64358f6e8340bf0dbdff716968bf0d92c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-30 15:58:34.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2019-10-02 07:27:25.000000000","tz":0},"subject":"Create _mech_context before delete to avoid race [refactored for Queens]","message":"Create _mech_context before delete to avoid race [refactored for Queens]\n\nWhen a network is deleted, precommit handlers are notified prior to the\ndeletion of the network from the database. One handler exists in the ML2\nplugin - _network_delete_precommit_handler. This handler queries the\ndatabase for the current state of the network and uses it to create a\nNetworkContext which it saves under context._mech_context. When the\npostcommit handler _network_delete_after_delete_handler is triggered\nlater, it passess the saved context._mech_context to mechanism drivers.\n\nA problem can occur with provider networks since the segments service\nalso registers a precommit handler - _delete_segments_for_network. Both\nprecommit handlers use the default priority, so the order in which they\nare called is random, and determined by dict ordering. If the segment\nprecommit handler executes first, it will delete the segments associated\nwith the network. When the ML2 plugin precommit handler runs it then\nsees no segments for the network and sets the provider attributes of the\nnetwork in the NetworkContext to None.\n\nA mechanism driver that is passed a NetworkContext without provider\nattributes in its delete_network_postcommit method will not have the\ninformation to perform the necessary actions.  In the case of the\nnetworking-generic-switch mechanism driver where this was observed, this\nresulted in the driver ignoring the event, because the network did not\nlook like a VLAN.\n\nThis change uses a priority of zero for ML2 network delete precommit\nhandler, to ensure they query the network and store the NetworkContext\nbefore the segments service has a chance to delete segments.\n\nA similar change has been made for subnets, both to keep the pattern\nconsistent and avoid any similar issues.\n\nNote: This is not clean backport. The original fix on master used the\nregistry.receives decorator\u0027s priority parameter (see [0]) which was\nintroduced in Rocky (see: [1]). To fix the problem on Queens this patch\nbackports ps2 from the original solution (see: [3])\nThe differenc from the original, merged patch:\n- On Queens there is no priority parameter for registry.receives\ndecorator, the workaround is (as in ps2) to register for BEFORE_DELETE\nNETWORK callback and get a copy of the network before segments can touch\nprovider information.\nThe differences from PS2 of the original patch:\n- In db_base_plugin_v2.py use registry.notify instead of publish.\n- In plugin.py change the decorated callback method signatures as it is\nsuitable for notify.\n\n[0]: https://opendev.org/openstack/neutron-lib/src/branch/master/neutron_lib/callbacks/registry.py#L67\n[1]: https://review.opendev.org/541766\n[3]: https://review.opendev.org/#/c/679483/2\n\nChange-Id: I6482223ed2a479de4f5ef4cef056c311c0281408\nCloses-Bug: #1841967\nDepends-On: https://review.opendev.org/680001\n(cherry picked from commit fea2d9091f71a2ec88318121ed9a22180e1ae96f)\n(cherry picked from commit f913394d70dede8d40d6e96485bd5333dfb6f43f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f90b643650df18bd9e5b31a57a84fd4620069393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f90b643650df18bd9e5b31a57a84fd4620069393"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
