)]}'
{"id":"openstack%2Fneutron~685174","triplet_id":"openstack%2Fneutron~stable%2Frocky~If139790eb336ff13b07b094151946af30322ad3e","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1718345-stable/rocky","hashtags":[],"change_id":"If139790eb336ff13b07b094151946af30322ad3e","subject":"DVR: Cleanup ml2 dvr portbindings on migration","status":"MERGED","created":"2019-09-26 20:24:01.000000000","updated":"2019-09-27 15:28:48.000000000","submitted":"2019-09-27 15:28:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685174-1569598128500-d890bd4f","meta_rev_id":"5b6603eaa875e8ed338ed3b04f22dd2c0cb1ad78","_number":685174,"virtual_id_number":685174,"owner":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-27 02:14:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 15:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-26 21:57:56.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-27 12:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-09-27 12:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-27 12:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 21:57:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-09-27 02:14:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-27 12:39:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-09-27 12:52:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-09-27 15:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0dca175f2181953469b7169d6012804f06b535ca","author":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"date":"2019-09-26 20:24:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a69506de64d8db89886eff0fd93d0a8f71713d25","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-09-26 21:57:56.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/685174/1 : FAILURE in 1h 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"031c0860a3428373da48dccad5b0eff4cb4ed463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 23:45:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/8027c31fa7724d4fbe1be836a1e7f916 : SUCCESS in 57m 03s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/0c18e6a7ce334ffe848e49b79bc18da0 : SUCCESS in 45m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/47d8b85e43f9445aa484be761e44483f : FAILURE in 1h 28m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/6b07f06765d54df2b99a3337be1951ad : SUCCESS in 1h 07m 09s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/8670ef728c0f41fdb773646c581467ff : SUCCESS in 1h 17m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/35e7fa864f864fbe89eaf6e8281f7798 : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa71906b6a14dc5a221cfdf6350612f : SUCCESS in 15m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cc2cdd8f186473eaeab8bcb3f3f13e0 : SUCCESS in 21m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ff67fa75ab947a690b59b215f145827 : SUCCESS in 32m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/79f69722b06a47b59d7b28a9776f2096 : SUCCESS in 6m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7a4379ad7604e5b84937b7253a36116 : SUCCESS in 1h 51m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8abb38a4eef34c43b1e81db148bac62f : SUCCESS in 1h 06m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/868103fb42f14744a035caa8b610295c : SUCCESS in 1h 28m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb983489ccc2423dbd7bfd96eb4a9943 : SUCCESS in 37m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3d5601abc4b04f7eba2ad6d07640dcee : SUCCESS in 1h 11m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0dedf6f5d8d44cd19ce6560cd90c52ec : SUCCESS in 1h 38m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cda1d340315a4997994bfc3388991736 : SUCCESS in 1h 41m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/35c196c7d8a742279e9189d297d204b2 : SUCCESS in 1h 37m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaf8b1dcdab148eabb7c8ecd7855bfff : SUCCESS in 1h 13m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cff722618cc345d4979b9d46adc7c2d6 : SUCCESS in 1h 33m 29s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/25a851e5894d4f778b0623fb0b1064a2 : SUCCESS in 1h 46m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e27ce7ed3d54232939a8be2b198b14b : SUCCESS in 1h 59m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ea45ed0380845f4a27186facaee36e8 : SUCCESS in 1h 22m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aea1f28c986e4fa4800a1d396f0d04e7 : SUCCESS in 29m 25s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/5370e93dd405402daa8ff82d14a7e01d : SUCCESS in 1h 01m 06s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/887bf4a6ebea49b283f8cd451135827d : SUCCESS in 52m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c8263cdedc408288c31f420b6f02cb : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c808f76e49c0d02b88d14e2cc58ebdb3269b3bc4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-27 02:14:26.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/74/685174/1/check-mlnx/Neutron-ML2-Sriov/f26da6a : FAILURE in 26m 16s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/74/685174/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/f725330 : FAILURE in 28m 19s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"c3aa36304b6ac36a850c76584e3028912758bc1c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-27 12:39:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22d953baaf17075e4ba1c7e5c4d584b4dae96e08","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-09-27 12:52:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5e6dce2509d9821550fbad9efb9dae15adbf4a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 12:52:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bdbebe91461d045cec5755c497ff02d87c6c1e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:28:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/5333331276b947a2936e452a9db4cb85 : SUCCESS in 46m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bda7b3a4e5b45058b8026fa72630bc8 : SUCCESS in 12m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf45b89a9a6c41c384799334b87d2a61 : SUCCESS in 25m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b746408d936a41dd8f29c6271905f1e2 : SUCCESS in 26m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6f552fff4cbc4770849eeab1886efad6 : SUCCESS in 10m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b665a1dc0c44e3cb0c74e362b0774d9 : SUCCESS in 1h 39m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3fc2548ebbd54cbb90c202a596e03879 : SUCCESS in 47m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2fb38c010ad4393b4572ef6442c38bc : SUCCESS in 1h 13m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d5e5f48eb24436cbe3f0729c9b43297 : SUCCESS in 41m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2794cf3b541b43e6adb2f60ab68359b8 : SUCCESS in 58m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/512e63cb47934148be5541bb66241a42 : SUCCESS in 1h 24m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d5fe0b454d074d1387dd8e3f20a4723f : SUCCESS in 1h 20m 50s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d2d1e6c60ed4330828b065b06aff5b4 : SUCCESS in 1h 19m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bf6a39a3d884f768d6225a273e2e940 : SUCCESS in 1h 11m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c4e4d74eabf742f4b75203eeb654b5e3 : SUCCESS in 1h 17m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22656210f4614be59e4bca9bed6a1dce : SUCCESS in 32m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4c998d1c4b04f0b8318d6873f8c5384 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3e47a164e80825c11373b12d9b084ed1645a9e21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 15:28:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bda3c2ed33906ab2fbacc23f5775b4e0f4b76b0","revisions":{"8bda3c2ed33906ab2fbacc23f5775b4e0f4b76b0":{"kind":"REWORK","_number":1,"created":"2019-09-26 20:24:01.000000000","uploader":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"ref":"refs/changes/74/685174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/685174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/685174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/685174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/685174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/685174/1"}}},"commit":{"parents":[{"commit":"58f290c81ec6e023501366702225ded8fad5f178","subject":"Avoid unnecessary operation of ovsdb and flows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58f290c81ec6e023501366702225ded8fad5f178"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2017-10-02 06:27:29.000000000","tz":-420},"committer":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-09-26 20:24:01.000000000","tz":0},"subject":"DVR: Cleanup ml2 dvr portbindings on migration","message":"DVR: Cleanup ml2 dvr portbindings on migration\n\nWhen a DVR router is migrated from distributed to\ncentralized, we are unbinding the router from the\nagents, but the ml2 distributed portbindings for\nthe router port still remains intact.\n\nThis patch will fix the issue by deleting the\nbinding entry for multiple hosts.\n\nCloses-Bug: #1718345\nChange-Id: If139790eb336ff13b07b094151946af30322ad3e\n(cherry picked from commit 32bfc3edeca935fc9c16b3019aa0af3091cd19e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bda3c2ed33906ab2fbacc23f5775b4e0f4b76b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bda3c2ed33906ab2fbacc23f5775b4e0f4b76b0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
