)]}'
{"id":"openstack%2Fneutron~686345","triplet_id":"openstack%2Fneutron~stable%2Frocky~I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1734320-stable/rocky","hashtags":[],"change_id":"I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1","subject":"raise priority of dead vlan drop","status":"MERGED","created":"2019-10-03 08:02:52.000000000","updated":"2019-10-04 15:39:38.000000000","submitted":"2019-10-04 15:39:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686345-1570203578502-b42f53c0","meta_rev_id":"eaad4d1c9561ff301bdf8f707f30fa88ab6372c4","_number":686345,"virtual_id_number":686345,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2019-10-03 09:40:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 15:39:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-03 09:33:35.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-04 13:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-03 12:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"date":"2019-10-03 08:02:52.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-04 13:13:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 08:02:52.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-10-03 09:33:35.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-03 09:40:00.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-03 12:42:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-04 13:13:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-04 15:39:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2dfea765eb68f6026b8da6116165f23a7a3cfe7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-03 08:02:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"708bb08b58901b333405609d48eb0f3727b342c3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-03 09:33:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686345/1 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"faecbb01f6d4c0b667193871f865c157e1dc3818","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-03 09:40:00.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/45/686345/1/check-mlnx/Neutron-ML2-Sriov/ca9d098 : FAILURE in 29m 33s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/45/686345/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/7421aa3 : FAILURE in 33m 49s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"f46b9d40bda7b077748d7d5117a2b535b670bfd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:56:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/91d09b4b50d44545b1bf4cdca2b70785 : SUCCESS in 46m 46s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/f25b8e35567d4bb2bdb7d97dc2bad3de : SUCCESS in 32m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/cc137e953d494300bec02c9fd074f09b : SUCCESS in 1h 38m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/978434f887fe4c2ab4ee3792911252d4 : SUCCESS in 57m 38s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/d03f5d9e3e2a4dcbaa5ed4d5f3e03197 : SUCCESS in 1h 00m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/28ba0b5a7c084b69a9e9364361dfc30c : SUCCESS in 1h 00m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e079ce9d4f46738ddc86fe2e69a117 : SUCCESS in 12m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b0862c7ff0d4df2addb44c0834c6846 : SUCCESS in 25m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb1d07ffc9fb4481b467769d93e59d3e : SUCCESS in 27m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9ade942e5549488fae18c540a01dbab4 : SUCCESS in 7m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/adfee40f3b9d41c8a551f20a18c9ff8b : SUCCESS in 1h 42m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3d51b619176346dab6ec4e994fcf0505 : SUCCESS in 1h 00m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42a34f21fc4145bf8890eccdc8981bb6 : SUCCESS in 1h 14m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/13b62d7993794f8bbf4abb2cce2d4691 : SUCCESS in 43m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0423df8093a847a3b5f005e7c30eb19d : SUCCESS in 1h 00m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/56ce3beef6e8463ba233ed3f6c4c5dfc : SUCCESS in 1h 41m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/71abd14d72594c479f8d89ab1cd0b038 : SUCCESS in 1h 34m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9c153059937e43139055aa9f084ef140 : SUCCESS in 1h 26m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1bb0343c19b9498780d095b7ad29dff3 : SUCCESS in 1h 11m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/953a949d153a4e33a270bee805976eaf : SUCCESS in 1h 09m 30s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3bf39192c50845e1b8bdb3379b5f6c5c : SUCCESS in 1h 39m 23s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3bb8daa4c04d4fe38b23da86df6efba3 : SUCCESS in 1h 47m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64bb832f43da46e0b63fad2da460ae1a : SUCCESS in 1h 26m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d045c5c3ea844368fa08dfc4845463a : SUCCESS in 28m 27s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/dd1b8a6793724150b40a0583458c1636 : SUCCESS in 1h 02m 26s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e30948fb20584ccda6d742d73a58fdf0 : SUCCESS in 52m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2349098f5faf41fbb823de8be1705d5d : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":1},{"id":"089de30af10b934344400cd984ccc9b35fe54fe7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-03 12:42:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0c29e83720e67095ee0b8ccc88521aa94c00dfb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-04 13:13:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport and security fix","accounts_in_message":[],"_revision_number":1},{"id":"484923a77bf76e08c1c54c1afb1a3f023b5f0a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:13:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26b1874616c12e71be03b4803af5279d3764efe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:39:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/95251cc816f741ed9dd222b185656346 : SUCCESS in 43m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5320b51b0e40b1bb83354a26549260 : SUCCESS in 12m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78b5a540dc6a4baeaf5c87f35d327a17 : SUCCESS in 25m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba6a8359ea894e6f9e158bf59b339dc4 : SUCCESS in 22m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/75a8b40018d644a89c2c40ae0454f667 : SUCCESS in 7m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4da3840371654657b01ec73afb4fca43 : SUCCESS in 1h 22m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f8fe630134d048f7b8d4258ea5d46a45 : SUCCESS in 59m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32df967aa870457e806f0cca60afd1eb : SUCCESS in 1h 09m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/697ef5d068204165a0c57aba364f6033 : SUCCESS in 45m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d14e288c165741da9fa5e4a5cefdc012 : SUCCESS in 1h 01m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7e13a5f578014dc2aae4f7ec38d6167b : SUCCESS in 1h 31m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e9fe007d0a884d97b586a177c1d6ad53 : SUCCESS in 1h 28m 46s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3cce192d5a304029859e787fd2baaa5f : SUCCESS in 1h 26m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1af5dce6cd04a6dbb586e60ff29bccb : SUCCESS in 1h 12m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4cee91e44d724a689c3570b53c50d4c8 : SUCCESS in 1h 10m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5204a18a5ba4212861a462d8fa21b0d : SUCCESS in 34m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b88391e225c24fcfbf7e9e22e975facf : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6a9d77a2c6b32f2b9f6dc530ba6da3e07cb13a60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:39:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffee956d44da34e38e62f82a3bc676cdd6179f48","revisions":{"ffee956d44da34e38e62f82a3bc676cdd6179f48":{"kind":"REWORK","_number":1,"created":"2019-10-03 08:02:52.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/45/686345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/686345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/686345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/686345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/686345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/686345/1"}}},"commit":{"parents":[{"commit":"8bda3c2ed33906ab2fbacc23f5775b4e0f4b76b0","subject":"DVR: Cleanup ml2 dvr portbindings on migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bda3c2ed33906ab2fbacc23f5775b4e0f4b76b0"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-03 08:02:52.000000000","tz":0},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffee956d44da34e38e62f82a3bc676cdd6179f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffee956d44da34e38e62f82a3bc676cdd6179f48"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
