)]}'
{"id":"openstack%2Fneutron~686346","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1734320","hashtags":[],"change_id":"I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1","subject":"raise priority of dead vlan drop","status":"MERGED","created":"2019-10-03 08:15:00.000000000","updated":"2019-10-18 14:33:48.000000000","submitted":"2019-10-18 14:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686346-1571409228686-e873c77d","meta_rev_id":"4f5b8f0033f61dad53cd363dd68eef356169f194","_number":686346,"virtual_id_number":686346,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-18 11:22:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 14:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"date":"2019-10-18 07:22:12.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-10-18 07:20:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-18 07:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-10-18 07:41:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-10-18 12:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2019-10-18 12:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 08:15:00.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-10-18 07:20:52.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-18 07:22:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-18 07:23:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-18 07:41:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-10-18 11:22:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-18 12:27:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-18 14:33:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de63cc704f7159550b5d715d2db8b67d236ba7e1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-03 08:15:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"722541478e0eb6558351a65bbc059e5c2724ec90","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-03 09:23:42.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686346/1 : FAILURE in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f4ac2851c2d94fd4c50a556b74346a1127128880","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-03 09:43:23.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/46/686346/1/check-mlnx/Neutron-ML2-Sriov/93f1765 : FAILURE in 31m 05s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/46/686346/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/d215c7b : FAILURE in 32m 29s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"3502ddd7ea5e06cc1a166eb17f0d978d3f9bf4cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:32:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/182f97b6d0aa4389bc56d87322f0e37e : SUCCESS in 50m 04s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/1d7adb3692884ae69a1fce65204f8d7b : SUCCESS in 32m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/13d85366b5a84ff69c57794e8ff7d93c : FAILURE in 1h 15m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/d5b3d7f53ead493c8212156544c10e58 : FAILURE in 2h 10m 59s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/89c4357929674b65af79684e90c10f0e : SUCCESS in 59m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef316a0f8a4b4a10bd6c6de57dd5b30a : SUCCESS in 10m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbef9bd14a814132a7a5fa3ea5f58db0 : FAILURE in 23m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3914f9719d2a4cc9b26d4eb141a8dc39 : TIMED_OUT in 40m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/06fa80af822e417c8b468ad1e1250abc : SUCCESS in 5m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f5921cdcf874315a6bc16d62a5c1edc : SUCCESS in 1h 21m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e2c229f25dd741bda65073f782e4cfca : SUCCESS in 1h 04m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5347263315274951979bae15ff03b520 : SUCCESS in 1h 16m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1e1faa12d900484e9051ae38425720a1 : SUCCESS in 40m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/918bdf9b9fb14157b7df44502ae82810 : FAILURE in 44m 28s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/0dc7d60be0e64324b14e5e3f511f765a : SUCCESS in 1h 37m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/50d8462c12b1403ea944f2fd55c9d162 : SUCCESS in 1h 21m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1a275a4702624a53b73cf21820208c2b : SUCCESS in 1h 28m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89d80175e4f04333a9ff885fc3c759f9 : SUCCESS in 1h 05m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/be1a4305ff984ca5b2e95f9d3b05f96f : SUCCESS in 1h 21m 03s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42dc0326797142189ebca6aa84e8a45d : SUCCESS in 1h 42m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2b6af8093e1444fd97e447bf906e7ac2 : SUCCESS in 1h 40m 48s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/41ac2caaed434f6f956aacabcb950819 : SUCCESS in 1h 25m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f284426d4dc4477924b2f6630d89c58 : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0981ef39eb8c462de82307f955658afe52aa6fef","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-03 12:58:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0211af66913ea648fb92ae45294300afa215ac0e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-03 13:39:40.000000000","message":"Patch Set 2:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/46/686346/2/check-mlnx/Neutron-ML2-Sriov/b6fde8e : FAILURE in 29m 03s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/46/686346/2/check-mlnx/Neutron-MACVTAP-ML2-Sriov/7d4aa35 : FAILURE in 30m 42s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"e4593183aacfae748b1263977123989f91395aff","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-03 14:11:50.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686346/2 : FAILURE in 1h 12m 22s","accounts_in_message":[],"_revision_number":2},{"id":"cde99f1079991a37a8fbf1980f413f326b3ac2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:08:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/ed7546c13c434232a5db10f4699eeb2c : SUCCESS in 53m 11s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/589f3112093346998be66e283f56e43b : SUCCESS in 33m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/cfc01e073ec641e5a8bd2459c4bd025c : FAILURE in 1h 32m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a537e4cedada4ad6b3740d12dcdedbd2 : SUCCESS in 59m 18s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ea204ba2b24a432d8886d338135b8daa : SUCCESS in 54m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cda01d0150d4c3aad68360aec9b1853 : SUCCESS in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/682f9cf7e41d417c9bb0608164ea8b3b : SUCCESS in 22m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab4abd825ce54bac811139839a96bf50 : SUCCESS in 24m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d5192c81d19d494880d8903624325652 : SUCCESS in 7m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/558b8276fa9c4d3d81b23ab7f4f8e2a5 : SUCCESS in 1h 16m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ab110445ce1f4f47b188ed7c97803275 : SUCCESS in 53m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af4c49d0e6204504b5d588323a4f612e : SUCCESS in 1h 15m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ea3fd90e75e24bab932caf783a2ec2dc : SUCCESS in 39m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6b119ece471c469395c571cf7aeb5bf2 : FAILURE in 52m 39s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/dd57eee49eec487da35323f0e91b944f : SUCCESS in 1h 57m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e8abf7c5792640cb9ee9dc9c7ff8270a : SUCCESS in 1h 31m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6983257f0bd9467bb5571ecc949e148a : SUCCESS in 1h 29m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/adde6d0354354ab58df073c7276f5e08 : SUCCESS in 1h 14m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/95511f1f0d9d465eb894495d16d64cac : SUCCESS in 1h 12m 04s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ef980ae598a843cc884e1e404e6f2562 : SUCCESS in 1h 37m 44s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c94185c2d802430b85e9d45447eac5ca : SUCCESS in 1h 28m 52s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/11174f37ef0c4bb28a10bf2f014056d8 : SUCCESS in 1h 16m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00d582662954483b9b8fdfea2c3b7ddf : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":2},{"id":"66b595b01d3e221f6be6869a847e341d0c0802bf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-04 13:14:17.000000000","message":"Patch Set 2: Code-Review-1\n\nMissing conflict line, else looks good to go in","accounts_in_message":[],"_revision_number":2},{"id":"b0dc9504ca2d2d7131c87a843b8b919f7d2928da","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-04 13:29:52.000000000","message":"Patch Set 2:\n\n@Bernard, which format would you like to see conflict line?","accounts_in_message":[],"_revision_number":2},{"id":"190ef4870c9e2bfcbdd536a0745f1e1b3d612eae","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-04 23:36:14.000000000","message":"Patch Set 2:\n\nIt is the \"Conflicts:\" block you get when committing a cherry-pick, just that git keeps it in comment section (line starting with #), so you can juste move it in main commit message and remove the \"#\"\nSo here it would be:\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/openflow/native/test_br_int.py\n\n(I think, just quickly checking it is the only one that had conflict?)\n\nRecent example with multiple conflicting files: https://opendev.org/openstack/neutron/commit/261880aea5dbc6ffb51a7756e3ba940ea37ea3e8\n\n(I suppose pike backport needs same fix)\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"dc933045d75234fe5bd2117913ba560da54f2975","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-18 07:20:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d41ea3c87fe4ab87d3399636b844542372154f8f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-18 07:20:52.000000000","message":"Patch Set 3: Code-Review+1\n\nAdded the Conflicts line, the only chang is active_bundle parameter not present in queens (and tests passed)","accounts_in_message":[],"_revision_number":3},{"id":"3c875e578c41a2cfba72bd74ac9176e444fd1db7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-18 07:22:12.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686346/3 : FAILURE in 1m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b1d55c01456182696ff0688e23f8119d8d28f265","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-18 07:23:38.000000000","message":"Patch Set 3: Code-Review+2\n\nThx Bernard for update commit message :)","accounts_in_message":[],"_revision_number":3},{"id":"07e4784b14e6314f77c0e4101e95c6918df2d968","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-18 07:41:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3ee16257e7e10aa1f866c6142fd823c4e7756f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 09:25:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/3ea0b8a99efa4743a83f0c0e3db90bc7 : SUCCESS in 45m 03s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/86af408ccfdd4edfb8a3bf24118022f0 : SUCCESS in 40m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/7388f1dfb7ea4f8994584ae3f68cd610 : FAILURE in 1h 26m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f354e8e96feb438c8a95c02c25fad9c0 : SUCCESS in 55m 44s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/7931a7c5886642d490be72e9ce9a852e : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d23032e821aa463aab5a3e5a804e01fd : SUCCESS in 10m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/308f1507238145878d08e94f27320988 : SUCCESS in 30m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19aa36664a074ad1ad176fac8d9019df : SUCCESS in 29m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/728607bbb3a54053a89c069d4e148bde : SUCCESS in 5m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db24d04fb10a47ed9dc62a4940ef115e : SUCCESS in 1h 33m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a7a95edf43f64617aa855829f97981d3 : SUCCESS in 55m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af9ef3682a2e420ab0aa71a50c8ba889 : SUCCESS in 1h 23m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/240b1343b7754cdc964133894dacb225 : SUCCESS in 36m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0a1c3122dbf540199325005552fb84c8 : FAILURE in 49m 52s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/0fe8dc8c6ec14e7cb1386a4b02a1ec16 : SUCCESS in 1h 40m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b26c85f1551f4d80be5994271c9424d1 : SUCCESS in 1h 29m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c1ec769ef0664b0b9c5d0fc736c6a1ea : SUCCESS in 1h 15m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8164f6f9ff14ab9a23a610331c42695 : SUCCESS in 1h 04m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3346ae2a54454a189856d95bcb3cdc0e : SUCCESS in 1h 01m 07s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/18b0a5bc16a54e2a863ec5da3fdebd91 : SUCCESS in 1h 39m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d40eeb4f14d43ba802a500546845e92 : SUCCESS in 1h 40m 38s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/1eed418fea92442d9557a89fe7962fb7 : FAILURE in 1h 32m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5987f7c63f2e4ea7aee01d1ca7a3023e : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a1c49539f022ef27c034bbcbace3ffc637608c7c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-18 11:22:09.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/46/686346/3/check-mlnx/Neutron-ML2-Sriov/90bead8 : FAILURE in 12m 19s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/46/686346/3/check-mlnx/Neutron-MACVTAP-ML2-Sriov/c419906 : FAILURE in 10m 16s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"11be4a2d8c3da0e67fad0fb7dab3e553814a6cfc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-18 12:27:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"10d7c6103f763dc384802571925786d78f848f03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 12:27:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"257fafc0bb3965bfab5ea7bf9e42b495fffdd644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 14:33:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/13b29e3b5d3b43fd95ca7d9b86ef6c75 : SUCCESS in 50m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7352ead323494316a27ef932ad45dae0 : SUCCESS in 10m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96186858375347e2b8c95bd8f3aa2eff : SUCCESS in 23m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/827fabb9a6a4477cb66be2414f78b825 : SUCCESS in 30m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4bd9b17424e149698eae4080bdaf2773 : SUCCESS in 5m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8bed9494d03c4406887544d3af71321b : SUCCESS in 1h 40m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/929ad6a11acf4e9a8d4720d3e72a10dd : SUCCESS in 1h 01m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/755a9de259564e9c84271596e8e147a2 : SUCCESS in 1h 23m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/29fbf4590a6d41acb74948b63339f575 : SUCCESS in 36m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a5ba895e63a64279b46e13ccbebdc631 : SUCCESS in 1h 30m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1cb9e4ab08b1475a9e7d4084d6486883 : SUCCESS in 1h 53m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d756b124fd54e06ac64d356ec28494c : SUCCESS in 1h 17m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ad0ff566deee47708e6ffe8dff6b26f8 : SUCCESS in 1h 16m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e21f825dbc214684b0dbe06963e7e9ed : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":3},{"id":"df1056dfc2d47ca3a1c41c4a7f18d63e7d3a0571","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 14:33:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b0919e64809729f4c0186e9be63b1082510bbb1","revisions":{"c5e7b29c03489bbf34a27c7cfdab821b43e96db4":{"kind":"REWORK","_number":1,"created":"2019-10-03 08:15:00.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/46/686346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/686346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/686346/1"}}},"commit":{"parents":[{"commit":"b3d3d6d64358f6e8340bf0dbdff716968bf0d92c","subject":"Merge \"Add info log about ready DHCP config for ports\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3d3d6d64358f6e8340bf0dbdff716968bf0d92c"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-03 08:13:05.000000000","tz":240},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e7b29c03489bbf34a27c7cfdab821b43e96db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5e7b29c03489bbf34a27c7cfdab821b43e96db4"}]},"branch":"refs/heads/stable/queens"},"91c829ff586460424cace0c81861cecf5f1640a8":{"kind":"REWORK","_number":2,"created":"2019-10-03 12:58:49.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/46/686346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/686346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/686346/2"}}},"commit":{"parents":[{"commit":"b3d3d6d64358f6e8340bf0dbdff716968bf0d92c","subject":"Merge \"Add info log about ready DHCP config for ports\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3d3d6d64358f6e8340bf0dbdff716968bf0d92c"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-03 12:58:46.000000000","tz":0},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91c829ff586460424cace0c81861cecf5f1640a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/91c829ff586460424cace0c81861cecf5f1640a8"}]},"branch":"refs/heads/stable/queens"},"9b0919e64809729f4c0186e9be63b1082510bbb1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-18 07:20:27.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/46/686346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/686346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/686346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/686346/3"}}},"commit":{"parents":[{"commit":"b3d3d6d64358f6e8340bf0dbdff716968bf0d92c","subject":"Merge \"Add info log about ready DHCP config for ports\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3d3d6d64358f6e8340bf0dbdff716968bf0d92c"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-10-18 07:19:48.000000000","tz":0},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/openflow/native/test_br_int.py\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b0919e64809729f4c0186e9be63b1082510bbb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b0919e64809729f4c0186e9be63b1082510bbb1"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
