)]}'
{"id":"openstack%2Fneutron~686347","triplet_id":"openstack%2Fneutron~stable%2Fpike~I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1","project":"openstack/neutron","branch":"stable/pike","topic":"bug/1734320","hashtags":[],"change_id":"I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1","subject":"raise priority of dead vlan drop","status":"MERGED","created":"2019-10-03 08:20:02.000000000","updated":"2019-10-23 01:39:02.000000000","submitted":"2019-10-23 01:39:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686347-1571794742150-f62c7b6a","meta_rev_id":"7e57ea503d50c23b976099bbfa1e83416057c3ac","_number":686347,"virtual_id_number":686347,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2019-10-19 08:08:23.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-23 01:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-19 09:16:49.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-22 22:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-10-21 13:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-10-22 22:39:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 08:20:02.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-10-19 08:08:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-19 09:16:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-21 13:25:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-22 22:39:27.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-10-23 01:39:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ce085dee8e4ea24e5aa79282893777414226ba2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-03 08:20:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7abb5b8aa53ccbf310f58b552fc62b169765a80","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-03 09:27:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/1 : FAILURE in 1h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"92e525286e4a1f2363607174dd504db657228f8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:46:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7200948e0a8a4740863ed1f1bb43fe63 : SUCCESS in 10m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c4bc407caa342f5b25508714eb70466 : FAILURE in 20m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0979a0e4e45b4778a2bb145310d85541 : FAILURE in 21m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d8c1859a1a3047d391cecd84e5bec6af : SUCCESS in 6m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/03d7488c352347b7b2312714266136e1 : SUCCESS in 1h 31m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/41e5183986474f62969242e12c975b6e : SUCCESS in 55m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/17045209112b40f8a8b5390c1e277acc : SUCCESS in 36m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9a5558d810474c3daee44280cb944667 : FAILURE in 46m 15s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/39d77818e1714e6aa576d992b9e2c842 : SUCCESS in 1h 40m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8fbd3446c913483ea11ed56df2d7c547 : SUCCESS in 1h 30m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3404c1681b0641d480eb452f6b6687d9 : FAILURE in 1h 07m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccb5c635e39041b48a81ea5431be0eee : SUCCESS in 59m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b396b41663e64364aadf487e2ce616ca : SUCCESS in 1h 07m 04s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/45ed1225efa7428081760fd08016550a : SUCCESS in 1h 36m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50de3c1f9acb4435866cac1fc00f423c : SUCCESS in 1h 35m 51s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/f371da2f28ff4cd7879d5d29ad072046 : FAILURE in 1h 08m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de25fdec2cb14673a9ef6654364b8d5b : FAILURE in 29m 15s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/d890de15543043aea37c112a48c9ea2e : SUCCESS in 44m 07s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3bfd24c695784bee8730e1fc49574fc8 : SUCCESS in 2h 17m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/56f1ab59d2da4b5295942076138541b2 : SUCCESS in 1h 58m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e1fc73b3385fa38fd395540187ce2ea781f1fc7","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-03 13:00:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a5e229a6b61e5136f9e29beb264a45732366e63b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-03 14:09:46.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/2 : FAILURE in 1h 08m 42s","accounts_in_message":[],"_revision_number":2},{"id":"37cc35236ebea535b3df382d2aecdf6a33037e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 16:04:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2bfbd48143e4cc09d292f199cd295d1 : SUCCESS in 12m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8a804604fde44729185b16cc35fcf92 : SUCCESS in 19m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3f0380ccb9724754882542db70a82472 : TIMED_OUT in 41m 56s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8c92e941f3444890a1ba8dfbd052c6a5 : SUCCESS in 6m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc1674a9720f4cb78fb861daa4e029d9 : SUCCESS in 1h 46m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e327a4337667431eb1e34de83fc10cb3 : SUCCESS in 52m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/465d4c300438454ca17e3d88d575c1a1 : SUCCESS in 33m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1b710d95b4c14f2593315d98340669d7 : FAILURE in 50m 00s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/5353eaa3fe774ea0b43267d540525b27 : SUCCESS in 1h 32m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/811f3e7b3abc4f7daef8d981058e6068 : SUCCESS in 1h 30m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/967bb2fbe6ab4ceab222e1f7919bfe2a : FAILURE in 1h 12m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/038141a70d7947198af76f45cffa03d1 : SUCCESS in 1h 04m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58d652692e074a10981d3e1fa65fec00 : SUCCESS in 1h 01m 12s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/505108045af2435583535345433d684c : SUCCESS in 1h 33m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5785050e5bb48a7936c71b16572f3f5 : FAILURE in 2h 53m 27s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/4c526b1a34c44bc2a925c5b9953d6056 : FAILURE in 1h 25m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24c3db78d6a647c787a51a6d3cce122c : SUCCESS in 35m 02s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/aafc1a6be24e4270bf99fe8807be5997 : SUCCESS in 46m 04s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/dfd483bdea224807bcd001557e6d10f6 : SUCCESS in 2h 09m 03s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c36d70bb0e0746ad8780a43bcb6f4638 : SUCCESS in 1h 45m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77fb5ad33f5a51cf2cabdcaea739bf4b2050e0b4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-03 18:34:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dce5b4759233e993c20bcd713fda34859de366f4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-03 18:44:43.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/2 : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cd50b972f7bff34dc422aa9f927522f32abeb323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 20:57:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555b646c93aa435a83cb79404e2003c6 : SUCCESS in 10m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ace01f2e3c834aafa1160e35a36ef7a5 : SUCCESS in 21m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2bcc1e85e9d94a3d81f9a03af31966a4 : SUCCESS in 23m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/db5d1e58e33544189dddb6485f765902 : SUCCESS in 5m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aff614dc0ab048b7ba5c24aefd8b1d0c : SUCCESS in 1h 21m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/395e1bfa414b4e61b1b6c27d5f78cbb2 : SUCCESS in 50m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c4b8b44b95ed447c87b6df7c233faee1 : SUCCESS in 36m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6a3a509795f64018b5461cd9df57b71d : FAILURE in 49m 45s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/e1ae56bc4f3741b8a34ef107d369d4f4 : SUCCESS in 1h 28m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/50a706a0659e41ba86d75dca143fcb62 : FAILURE in 1h 18m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/29e3d57b924f45bea0935756a150d09a : SUCCESS in 1h 21m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64acb6d26ab8427eb6680e4c8f1b6bb3 : SUCCESS in 1h 07m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1708f6092c3461dac140c15b8dbfe16 : FAILURE in 52m 43s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/534cdef167d54f6e8dc402426ac81c7f : SUCCESS in 1h 36m 01s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8474f5718ee4eaaace78f2e8ead24e0 : SUCCESS in 1h 41m 03s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/d893cd971d174d8baee8b5b3a871c557 : SUCCESS in 1h 23m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60161d3729ae405589bce5327ab16041 : SUCCESS in 34m 09s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/3240b29bc2b04b41a823edf8c2cb39d4 : SUCCESS in 46m 21s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/086f9eb315334e99830daa34737b60cd : SUCCESS in 2h 09m 33s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/06b31f1f51c74d308cc03a1b7656ec3f : SUCCESS in 1h 25m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9ff565b7c4ef13bb4949cdb04355ac87a3fffe1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-04 05:24:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e3b79e57fe2b5bbea9f24024109ef6e2c90bf129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 07:47:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb514bc931264164945c062015cbb809 : SUCCESS in 15m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/278460849d7a4512bfba1f5926444307 : SUCCESS in 24m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9bee33dc1185405ea3b5cbc182c80151 : SUCCESS in 24m 55s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/88357e6522654b81a9e067b8635f11a0 : SUCCESS in 9m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/370d5192693f464493ae4eb2fc3da4c7 : FAILURE in 1h 55m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7c37aa98aae4459c96125f804db4423e : SUCCESS in 1h 18m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f3156faf6eb49a5a2e651a4ffc6ca51 : FAILURE in 54m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/08ef284927064dc3a7238f4912e61c26 : FAILURE in 1h 15m 00s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/cf4ed1389be84fa093db125aba3afffd : SUCCESS in 2h 07m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e18f30c13c384764b6277f4a09c4b6bd : FAILURE in 1h 56m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/019e8c29ebb74149acf99d7d1ff544e2 : FAILURE in 1h 55m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f080ddea12a3432682787d8d2922384e : SUCCESS in 1h 05m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/557551b2beff4fdd9f788025b44760ce : SUCCESS in 1h 18m 29s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1d908581762345fa8a7948b4f12ba346 : SUCCESS in 1h 30m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d9b911fe85564081b7287fb304348472 : FAILURE in 1h 54m 00s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/39318fbd25a84df98eaa1029b9e8ffb9 : SUCCESS in 1h 35m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fca80b1bec346e380e4c228862bc9b1 : SUCCESS in 27m 31s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/09fc8e664ab44c8a940ec198fc3d79d4 : SUCCESS in 45m 13s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/73c266de0c44487497b5121504f28995 : SUCCESS in 1h 54m 26s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/feb9e04715c8465da3350ac901207936 : SUCCESS in 1h 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ec75bbd9eae60f385dc50919206ebb0ae48f943","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-04 08:03:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/2 : FAILURE in 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e58f615c7c29888dddb9564280b0d29daae1a37b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-04 08:21:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"785d7af4154c53950c85556f766318b5c02132b9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-04 08:37:21.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/2 : FAILURE in 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"27cd57a62fbf5d923d72eeb574b423e64a1d91cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 11:18:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b80c953417d241e8bf37aeeaf76958f0 : SUCCESS in 10m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db31781b54de4608ace6163a7b7a3b55 : SUCCESS in 21m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/30dd7767ba7546479f81e2545d321a7b : SUCCESS in 22m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3209d26d4a4e4b9c809d806718d9d152 : SUCCESS in 8m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e73ddc85da30401d8032b7185bbd98ba : FAILURE in 30m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2311d440c0dd4f669f70662ff4d269ef : SUCCESS in 52m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c6b2f13d3b24e6b8fbd1dc507668b40 : SUCCESS in 33m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d2a630dd909646f09530e0c397d9425e : FAILURE in 47m 07s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/3075aaa4d2c345e1b1fd667ba887fb4a : SUCCESS in 1h 32m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ae60c233e5684dbc9f83cdf134456e36 : SUCCESS in 1h 27m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ada0aa3cd6ec4b3db1cdd9a95e1aeece : FAILURE in 1h 18m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed2cd98b49424444a24228509f707593 : SUCCESS in 1h 00m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d8207df8b6d346f3ac9bb9e3e3ee7003 : SUCCESS in 1h 09m 06s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6b59247f3094df395599fd6beadce63 : SUCCESS in 1h 41m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1943140cc00043279ae424440ac6ecc4 : SUCCESS in 1h 40m 03s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/228d48d7060e4078b5cdf55cac778fc2 : SUCCESS in 1h 31m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/648cf2143913451980c783b70ed5b3d0 : SUCCESS in 36m 57s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/fff9f0c8f0804916ab0c2e806f94b1de : SUCCESS in 44m 33s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3dacaf2d668543caabc45de81b0ed2da : SUCCESS in 2h 47m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3672949e3cad46faa0182d51fbd8cb2b : SUCCESS in 1h 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1c7f6fbbd5a45796401ad9fa8905145868bd2b4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-04 12:34:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d1fff4ab69ee1f9fb58631fb151f88ed249e473","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-04 13:15:12.000000000","message":"Patch Set 2:\n\nSame comment as queens, can you add conflict line? Hopefully, this can go in without too many rechecks","accounts_in_message":[],"_revision_number":2},{"id":"cd735521828b2d04948068c1f0a81b31f6a93021","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-04 13:47:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/2 : FAILURE in 1h 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c95224367ce18f709f89352e479bcc0e5123a3fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:53:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aef1cd7a5d744a6958bf92fbf900ce9 : SUCCESS in 10m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9906ce82c4c9432da6a48890ffa8c694 : SUCCESS in 30m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e3cf6136b5a46d9b03596a8a3581705 : SUCCESS in 25m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9e3281f33c8243f2b6ab4d6f045b74b0 : SUCCESS in 8m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a1df700f16c4742b980ddb724a6d787 : SUCCESS in 1h 41m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ef0d7d78e1d4341b77e7b6b043b5ddb : SUCCESS in 52m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/912e20405d0a4688ac4f4aee35ec7247 : SUCCESS in 34m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4eddf079d2104795917602ed77635233 : FAILURE in 47m 47s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/5e12095b6c8b4f758d2f8f825f7589f9 : SUCCESS in 1h 30m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/59af7bf1888c4e6d91fcd8552b62ebe5 : SUCCESS in 1h 30m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1343a99d593e498b8080eb9461c8a751 : SUCCESS in 1h 34m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2de20e3ff09c4d0e8454ffdf587443c3 : SUCCESS in 1h 06m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7b7d0c4ba3c641d9a3c3b1b58af0ec92 : SUCCESS in 1h 09m 13s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a128921b7404ae3b9a19ef9dc260216 : SUCCESS in 1h 41m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c2fe190ff5034b5a94186f110fe32e5a : SUCCESS in 1h 51m 51s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/7cb8232f61d84126abfb582929a2947f : FAILURE in 1h 28m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f0192cb54294be782db68919bdd45c0 : SUCCESS in 36m 34s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/0a86fcfe84544582bf292ce57e3471ff : SUCCESS in 48m 24s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9841c7eccada496c8a057c20a10c0f3f : SUCCESS in 2h 13m 22s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0dcc4be560404bd885fceef0d1499e69 : SUCCESS in 1h 47m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2f149a7aa6bb58987b427ec8ca5a5e6cd06c2ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-18 12:28:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3dbeaa2d4a8766e760c0c8e35204aa2baf9b3499","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-18 13:09:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5f8a5fabb8624ee7efb92123c67d20f1b3221908","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-18 14:18:50.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/3 : FAILURE in 1h 08m 20s","accounts_in_message":[],"_revision_number":3},{"id":"cdb7ea8fbc32cca71bd64436afd71395cb9a0f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:48:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34bec78e7c394d72b85b7bbb394dd845 : SUCCESS in 12m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6e1c898bf6a4f54bf1e6b7777c0f578 : SUCCESS in 22m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c68600b7189140a99bea3315ac456e2f : SUCCESS in 25m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/60a0cad11e5946878cf76165263206d9 : SUCCESS in 6m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/91e0956cf3fb4183b333ed1d2ea5bfc0 : FAILURE in 1h 21m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/68229184ed824980b32944bc1b276549 : SUCCESS in 57m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/580adebe8ea742818789c3aa0859cdd5 : SUCCESS in 36m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/147ef66b235b43afbbf83d74b44df1b6 : FAILURE in 57m 00s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/611e5e0e509845c7b8b4aa193435a009 : SUCCESS in 1h 38m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b3fe3573935145f5bea74637b45fbd56 : SUCCESS in 1h 22m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cc28ff8222324727a4dbc122a3c7e05c : FAILURE in 1h 26m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4728bcb9d6cc42628cd9fff7ec10604d : SUCCESS in 1h 05m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6cca3621ec46416cb2ab0e4868de1f5c : SUCCESS in 1h 04m 56s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/906bce39d4e84378be3b4fa862eb4359 : SUCCESS in 1h 53m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a6229416234741ac9c46a4ad5fc535b8 : SUCCESS in 1h 34m 51s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/bae040bd0921431ebc8a9344888ddaef : SUCCESS in 1h 23m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca70799f51914dd28331cb0c8ef2d27d : SUCCESS in 28m 57s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/4b89839f62a9410996aaaca2ec63c8ec : SUCCESS in 49m 35s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/19e884a95ac2460ba9548a4bd5feb2e1 : SUCCESS in 2h 31m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/672dfc6e85614543a4d1b706a328b76d : SUCCESS in 1h 52m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"127228999d55165d59b7a2eccc6d8d086dd8e676","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-18 15:56:53.000000000","message":"Patch Set 3:\n\nrecheck I only added a line to commit message","accounts_in_message":[],"_revision_number":3},{"id":"912d9ee8d633a30223acbcbe29820966f2e95919","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-18 17:06:23.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/3 : FAILURE in 1h 09m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4f2071b5497bef1c8cd91fba12e8b81ed12e658b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:57:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffb9437d2d44d54b12d366e652ab7bc : SUCCESS in 10m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62d0c316d8ac4cbd9f32ac8c6ef3f78f : SUCCESS in 22m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72b95dfb13f04727a628e0b253ac73fc : SUCCESS in 25m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/cb812f2d01d34eb5b2473d01a6990f79 : SUCCESS in 5m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0e1efa5581774461b2efa3d3dcad8d6a : SUCCESS in 1h 45m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9bfaf45f126d41d9a0aebf26863b7fe8 : SUCCESS in 47m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f9b630f409274a6fb24ce4f9cd2f9bcc : SUCCESS in 32m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/88374fc18a2d4be1b88c11b33840f611 : FAILURE in 57m 33s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/03fe5363cae14850b579c7a57c8982e5 : SUCCESS in 1h 35m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/05fd73e6d62d4c88a213405b3608d90f : SUCCESS in 1h 37m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/49e095a2e1644a2b81aa4b212cf7ff46 : FAILURE in 1h 12m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a530182921c41d58f2754021b5a82a2 : SUCCESS in 55m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32e65d500548458f886a2796406929de : SUCCESS in 1h 05m 32s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f22fdeaae6d1476cbeddb211b5ace5c6 : SUCCESS in 1h 48m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0eadf9459f234946bc066cabfaf431f0 : SUCCESS in 1h 37m 39s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/9cd616b20d7345eebd3c9f7e9f0d8b3e : SUCCESS in 1h 26m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe696463ab394053b75eb2a6547612b2 : SUCCESS in 33m 47s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/1c213640e4644d92ae4dad0f16be87e6 : SUCCESS in 43m 28s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e67cdb1bfc894ddab0d267620b37b22b : SUCCESS in 2h 16m 55s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/796939bae1ce47d8970ea35ae6e90504 : SUCCESS in 1h 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"856adf057186b52d908907b20dc8351a10cd8121","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-18 20:38:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a4424d92df3ab9692af0d020ac39f82dcba18505","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-18 21:47:23.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/3 : FAILURE in 1h 08m 30s","accounts_in_message":[],"_revision_number":3},{"id":"a998ba6cf7a0f72cd163e283a26b25acf4b554dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:13:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cdbf77b38dd42858fbdf49ba3b0a3d1 : SUCCESS in 11m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75f8484fef14404fb90ff19424d4a790 : SUCCESS in 26m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e74d8f902b4d4b43999eb14bdee49645 : SUCCESS in 27m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ca380c484b5949fc830b50ed63980623 : SUCCESS in 5m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f61e53cc28014364ba6623bcf95f37da : SUCCESS in 2h 07m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a5c0080c2be4317b9bd60685b293abc : SUCCESS in 53m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ea6d65c608994fdcbcfcc67d6827d769 : SUCCESS in 32m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9118400466d147118cadc8c8c417d7e3 : FAILURE in 50m 29s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/3cbf49b6a2414e99a6dca1d6867d91fe : SUCCESS in 1h 41m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/80ba529c517847c1818f3484f06f6b37 : FAILURE in 1h 07m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c04ee3b82d7b4ea4bb2ace87ea94f9f5 : FAILURE in 1h 54m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06fd99443e1a445aac5c6a4c4f49e3ec : SUCCESS in 48m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6bc6b830cbbf4c70bbc876b16d31c152 : SUCCESS in 1h 05m 48s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2e57bdd9980f427082bd2f3ebf5e8d29 : SUCCESS in 1h 34m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e641935e5ecb4fba89ad7abfb9c46a76 : FAILURE in 1h 22m 01s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/ffdcc9f2d6a04b9c9dc33cc86a7f4462 : FAILURE in 1h 17m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/576246c3fa2147e7a308bb427291f3e8 : SUCCESS in 32m 46s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/7fc82b9642014224916b636eaa124051 : SUCCESS in 45m 25s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b787b329abd94efdae442d6f6cc7fc6c : SUCCESS in 2h 28m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b223d41bc73409d970197e1dbe0058e : SUCCESS in 1h 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c77453aa1a1885e6e45ffc9b739c7f976d192f98","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-19 08:08:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4e1390d3c8a7a36e4e538238af242d79277f285b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-19 09:16:49.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686347/3 : FAILURE in 1h 08m 00s","accounts_in_message":[],"_revision_number":3},{"id":"825ea802d0f27aab70d9a92c01a52f69131e5b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 10:29:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb00d3d73b1e4d6eb168814fda8178c0 : SUCCESS in 10m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/956a74e414ee4193b49e54061940a4db : SUCCESS in 23m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/54515dee21714d738934a863a72d5d8f : SUCCESS in 28m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8434a52cd75b483aa805428d7a7b45fe : SUCCESS in 6m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e81ffe731b348638c66cf4e37178b57 : SUCCESS in 1h 31m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/971aab1204fb45038dcfab7a4efcaecc : SUCCESS in 1h 10m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0e15e6d50b7c46c3859c2490af2af7c8 : SUCCESS in 34m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f499df51b6614b10b08c639c60371c1c : FAILURE in 1h 00m 53s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/fb08074e53a4435fab58257cdd360cc8 : SUCCESS in 1h 36m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/73c48836581b48779d7a91089b3bcf18 : SUCCESS in 1h 27m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d50a62bf25a64851ac5e8820eeeb4192 : SUCCESS in 1h 24m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e67a2a41e24241028369f2469dab2c24 : SUCCESS in 1h 17m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/226d42ebdd844873a96930015fcd5acd : SUCCESS in 58m 43s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2ab275af0c3743359ba6d880baa595ad : SUCCESS in 1h 41m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f0008a4ef9834ffeb331ee49f069e41a : SUCCESS in 1h 47m 12s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/c542ea9fa373493eba5aeb9e21aa2cd2 : FAILURE in 1h 31m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d7bc3ad3244f46802b6a48439efbbf : SUCCESS in 25m 20s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/12ae767054b64455b60388857983e8e9 : SUCCESS in 43m 47s\n- legacy-tempest-dsvm-neutron-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/fb88fad3ada54f64a9c85a90e642e661 : SUCCESS in 2h 18m 22s (non-voting)\n- legacy-tempest-dsvm-neutron-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e48e4ea170e44c92bc187003d8b28997 : SUCCESS in 1h 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d40275aa2f03e2d4215761561503de3c8cc1796b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-21 13:25:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9e2ec5f61322ce2886a4ba1bb8fd2f3c2e054554","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-10-22 22:39:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d05c4d5584c5901b689f5e3010c92ea648ee5e38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 22:39:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"62c82d4a74b62ce5de7436d0e3d9f22da1e8b993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:39:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad424b1adedd4ef280a9823f49def48e : SUCCESS in 13m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/adef8455f46a4c07b1f236115ef81919 : SUCCESS in 21m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/715fbf1ec55f439cbffc743c81f8a7bc : SUCCESS in 28m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f786f1979cf845c7bb86fd74af7a8edf : SUCCESS in 5m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/98f7e18e6cad479490dd8c0bb1c2d740 : SUCCESS in 1h 35m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6342e7a355d6475c8667fb8bbb07af24 : SUCCESS in 1h 05m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/657167993e5f424fbec876654220c827 : SUCCESS in 42m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/38e9b98f52754d87aa6950c6fe5e94f5 : SUCCESS in 1h 36m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d25bced543fc4ff29246e1d86d1b306f : SUCCESS in 1h 39m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e2b91b225ab49ab9d00574affbb12a3 : SUCCESS in 1h 10m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8533009e127949ce991ca92156135c89 : SUCCESS in 1h 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e2e67bc225f4f079ef4dc7853ed1796 : SUCCESS in 34m 23s\n- legacy-neutron-dsvm-api https://zuul.opendev.org/t/openstack/build/5383f8215f3045908328f471851877a9 : SUCCESS in 45m 02s","accounts_in_message":[],"_revision_number":3},{"id":"c6a0547b654be24ca264123b204409a47262dc28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 01:39:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"797a379c51a05fca356d2f2575bfc386b0287afd","revisions":{"59a846012053abf5d6cad6461be9f6dfd6a0aa25":{"kind":"REWORK","_number":1,"created":"2019-10-03 08:20:02.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/47/686347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/686347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/686347/1"}}},"commit":{"parents":[{"commit":"0908dfcd28dc5498ec5854f3e51aba09fa999d5a","subject":"Merge \"Check for agent restarted after checking for DVR port\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0908dfcd28dc5498ec5854f3e51aba09fa999d5a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-03 08:19:17.000000000","tz":240},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59a846012053abf5d6cad6461be9f6dfd6a0aa25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59a846012053abf5d6cad6461be9f6dfd6a0aa25"}]},"branch":"refs/heads/stable/pike"},"ae160aad875cd6671b2c3406d4d044bee7d7e2ea":{"kind":"REWORK","_number":2,"created":"2019-10-03 13:00:22.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/47/686347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/686347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/686347/2"}}},"commit":{"parents":[{"commit":"0908dfcd28dc5498ec5854f3e51aba09fa999d5a","subject":"Merge \"Check for agent restarted after checking for DVR port\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0908dfcd28dc5498ec5854f3e51aba09fa999d5a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-03 13:00:19.000000000","tz":0},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae160aad875cd6671b2c3406d4d044bee7d7e2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae160aad875cd6671b2c3406d4d044bee7d7e2ea"}]},"branch":"refs/heads/stable/pike"},"797a379c51a05fca356d2f2575bfc386b0287afd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-18 13:09:53.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/686347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/686347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/686347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/686347/3"}}},"commit":{"parents":[{"commit":"0908dfcd28dc5498ec5854f3e51aba09fa999d5a","subject":"Merge \"Check for agent restarted after checking for DVR port\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0908dfcd28dc5498ec5854f3e51aba09fa999d5a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2018-11-08 16:07:55.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-10-18 13:09:31.000000000","tz":0},"subject":"raise priority of dead vlan drop","message":"raise priority of dead vlan drop\n\n- This change adds a max priority flow to drop\n  all traffic that is associated with the\n  DEAD VLAN 4095.\n- This change is part of a partial mitigation of\n  bug 1734320. Without this change vlan 4095 traffic\n  will be dropped via a low priority flow after being\n  processed by part/all of the openflow pipeline.\n  By raising the priorty and droping in table 0\n  we drop invalid packets as soon as they enter\n  the pipeline.\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/openflow/native/test_br_int.py\n\nChange-Id: I3482c7c4f00942828cc9396cd2f3d646c9e8c9d1\nPartial-Bug: #1734320\n(cherry picked from commit e3dc447b908f57e9acc0378111b8e09cbd88ddc5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/797a379c51a05fca356d2f2575bfc386b0287afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/797a379c51a05fca356d2f2575bfc386b0287afd"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
