)]}'
{"id":"openstack%2Fneutron~686833","triplet_id":"openstack%2Fneutron~master~I1c171ab906a3b4c66558163ad26947ebf710a276","project":"openstack/neutron","branch":"master","topic":"remove-py2","hashtags":[],"change_id":"I1c171ab906a3b4c66558163ad26947ebf710a276","subject":"Stop testing python 2","status":"MERGED","created":"2019-10-04 20:19:05.000000000","updated":"2019-10-28 00:54:16.000000000","submitted":"2019-10-28 00:52:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686833-1572223936452-0fd08e9d","meta_rev_id":"c101f2490e2d1f2d8aeb5c14b62878586c3f656a","_number":686833,"virtual_id_number":686833,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 00:52:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-10-25 19:23:28.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"date":"2019-10-27 21:20:40.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-27 19:39:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-27 10:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-26 00:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-10-27 10:09:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 21:33:15.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-10 18:28:46.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-12 12:03:40.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-10-17 21:34:31.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-10-25 11:18:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-25 16:44:45.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-25 19:23:28.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-26 00:21:31.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-10-27 10:09:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-27 19:39:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-27 21:20:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-28 00:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"613a6a05dbd57bdb591eca8166cde478c0ab50a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-04 20:19:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0da16d3d1a0a8af59d23988f33f8d0c6133ac47","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-04 20:20:15.000000000","message":"Patch Set 1: Workflow-1\n\nWIP, need to get feedback from other Neutrinos.","accounts_in_message":[],"_revision_number":1},{"id":"ed7e2f3707508c55fe1d13bd9e56e3aa6570ec39","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-10-04 21:33:15.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/October-2019/Arista_Tempest_Testrefs/changes/33/686833/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"285686de1f6bc0b86c95e2b8fe5259be8526eb22","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-05 02:04:25.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686833/1 : FAILURE in 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"925f1d1c36b0a75b6a055f9d3581bbd0c92dc466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 03:35:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6344e73585e947caaf387502eda97a89 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a08834a7851f485088eaa71b438dabea : SUCCESS in 56m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e4589abafe1143498c44d7ac1e7760fd : SUCCESS in 38m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3c68a0384bcd46dea48533f5263037e0 : SUCCESS in 1h 47m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/448e775890da43e7bfde53530e5dfe61 : SUCCESS in 1h 05m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/21c73dcd8e194cd5b7a223c4c439a28e : SUCCESS in 54m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d025810e53a344d0adcfc0d20afdcd10 : SUCCESS in 1h 05m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96680357e0db4566bc82dc18d07ffebb : FAILURE in 30m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3534cbdb04ec4cb89a96bb77371b56a2 : FAILURE in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d878aa94535466e8db8340000707b0a : FAILURE in 15m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/601bceac10c94bad9dbe9070730bd7a6 : FAILURE in 22m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c624e5b5aaa148e6b6af9490783714c9 : SUCCESS in 23m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c9e08c6afb14d4e8a7c6a0aa6164b53 : SUCCESS in 22m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ad52e89cc04f87b9fe9d635ef3d0cf : FAILURE in 5m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73e6a5ffddc841edb7f14c4f149b122a : SUCCESS in 59m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c19695730d943ccbae9628ca7eee2d2 : SUCCESS in 1h 16m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/871175152a044596a4b547f50cec731e : SUCCESS in 26m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4840cd802b324714b7b29f364811be38 : SUCCESS in 46m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c1ec8875bca3400bb32a20f3b2f17525 : SUCCESS in 1h 04m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/04319efc598445dc99cce6cf8268920a : SUCCESS in 1h 52m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c95cdbb9c0df465d86b64e23870b61c9 : POST_FAILURE in 1h 41m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e4536ae83728435388795606fcb6169a : SUCCESS in 1h 18m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ef1e1d36d2d4f2aa3c4026438c35de4 : SUCCESS in 1h 13m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32e29376967c4aeaae58ba061c69ab2e : SUCCESS in 1h 06m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ad314519529e47cb8f4902d22ecd1549 : SUCCESS in 1h 37m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/094906b46d8e4efc9eb94ed77f2ded91 : SUCCESS in 1h 54m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/97f4666444164d65a94afc237e59c87d : SUCCESS in 1h 03m 47s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/9dc02e32e2d640dfa66fba8c40632ad0 : SUCCESS in 1h 30m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bc8a06061fce4fcc8fe7b8362afe4e88 : SUCCESS in 58m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f676f1ccc2ce40f8a478eed7621d0e12 : SUCCESS in 54m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b19821ccccf49f1999beca8ee477096 : SUCCESS in 2h 23m 47s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/47936351515944a587895a38207093b3 : SUCCESS in 1h 59m 45s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3a0b40b5edbc4c4dbe0a7f512277f50e : FAILURE in 1h 29m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf5359b9d5bd4fd8958ac1d00ea02eb9 : SUCCESS in 1h 02m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/35bd0e4e5777415ab2961fba3e46943e : SUCCESS in 1h 18m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d30bb17fcdfc416ca8cbd0a8e49a1e7f : SUCCESS in 46m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5ddd633b9b24fb6b81fdd4d942e680d : SUCCESS in 1h 00m 31s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3acce0a3502e4060b4bf9021cb6af90f : SUCCESS in 1h 23m 14s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/05adf64d3c784b789d020885e1b101ed : SUCCESS in 2h 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06e025ec20798eb082919bfc75e1a4a59b09a430","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-07 12:49:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b5ac8be753d5d40330fb22227886f4353050359","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-07 14:37:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d49a15a00653cc6e72eea13e445ba5e2a6477b0b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-07 16:27:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686833/2 : FAILURE in 1h 49m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3ff45a42efe6b6a8f940b2bcdc9abc16e93c42cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 17:37:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31c680c224d94a0185a985c19e2dc0a5 : SUCCESS in 2m 59s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/41f2991197af4f85979e1e6d56ad4309 : SUCCESS in 1h 00m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b79a92a1f0d406981de675c01a8dd7c : SUCCESS in 38m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2dc8dd9519b04621959f720286257b74 : SUCCESS in 1h 34m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/02b04fc625b24e088b61fadbeef14fd8 : FAILURE in 1h 17m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9f959ac06a8247279112d6368f895153 : SUCCESS in 55m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/25cc3ab33b7c40bfb3bc9daf67776b7a : SUCCESS in 59m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c98ebc2fbda4e16a610459f12462ee5 : SUCCESS in 36m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/995c63b9515a4bedaa68b5b404c07f26 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c6e1af0e714f57b287641629216264 : SUCCESS in 16m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1eaad9b6a038457a84b0c031f1dcb767 : SUCCESS in 21m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59cf35e0fa194ec692a61d50cd67b243 : SUCCESS in 26m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eabc4104064849af87ac2098922a34a6 : SUCCESS in 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e99f38aa134322b9018cb5751b79f4 : SUCCESS in 10m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/23d5c644120940f9af42b2ffe9a81b08 : SUCCESS in 1h 14m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/39613f402971453d8984245d07b110ab : SUCCESS in 1h 31m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75389ffba8e242e3935feab4f3a314b4 : SUCCESS in 24m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e52fd6e990174c48bf894d192e2f2974 : SUCCESS in 51m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5ffd01a4108b40b5a362f3ec98bf06ae : SUCCESS in 1h 02m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dc05e2538a2d41a891f03690c632ff83 : SUCCESS in 1h 53m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e76ef53b3abc46c2aa95b3ce5d7f363c : SUCCESS in 1h 13m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f013be183f7d4ac091f90f6a1d7cb7df : SUCCESS in 1h 20m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68272e052b12439ea1ed13c358e828d0 : SUCCESS in 1h 11m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/03dcc3d5ca1d43a7956d037506fbf5b3 : SUCCESS in 1h 28m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cfb56463079e4fada92a1d81b916b8d4 : SUCCESS in 1h 52m 00s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dd254ecda1b8438dabafb19d7daba7e5 : SUCCESS in 1h 50m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d79c2b8b53874557b3c60d833ce15624 : SUCCESS in 1h 12m 38s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/586da94d0d4d4e05853d49ba23bc0725 : SUCCESS in 1h 25m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cd70c15da3f34f92ad8e29e297982c0f : SUCCESS in 1h 03m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d35c3cd668694793ad1ea648d4a6ae41 : SUCCESS in 1h 02m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cb342be3b524f418525ec428ac9c950 : SUCCESS in 2h 26m 21s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cfef9201b95c4966a409aa4c01f4d54a : SUCCESS in 2h 32m 01s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/00a3d578ee41426683efa74fcc3d0e03 : FAILURE in 1h 40m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/341cc63bde044e7083ab2370c9130eed : SUCCESS in 1h 20m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c878c9c0e494447a9d5de7bce8505880 : SUCCESS in 1h 22m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ee4bb31dc91453299c56d470e8b3ea8 : SUCCESS in 38m 15s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8835fe59e4c448ab865d52d33739dc3b : SUCCESS in 1h 32m 35s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bd1a694eeaf4514b3bba0d3e5cd408b : SUCCESS in 1h 27m 58s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e1cfffc2a7f449149e85256ccc701d63 : FAILURE in 2h 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0674a39fb9a7afec6780b3de0202a01485579cea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-07 18:44:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15a59accc69ddcf588ee2d780945830015246527","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-07 20:31:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/1 : SUCCESS in 1h 46m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3b24c27d8016b6f22f7183d8153f950d4d5123e6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-08 12:29:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6303f8eb8151e5f47985cdcfa23e51d045d116ba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-08 12:31:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6c599cc2b12e861f184b627d8ec1a184b17c769","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-08 14:11:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/3 : SUCCESS in 1h 41m 00s","accounts_in_message":[],"_revision_number":3},{"id":"74b5405c349608feba8fd11e55d74386887f1db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:15:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a7e7e3ff0d1e4073b7baa8d3367cec41 : SUCCESS in 10m 28s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4af631bca02c41c49bdac121dfb408cb : SUCCESS in 1h 11m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8575c8d2ef094f6a921a3bb0d4778787 : SUCCESS in 38m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/28174f7be9b245aebba0a9b2f1b5005e : SUCCESS in 1h 36m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/35b34fb0cb424f5788cf1a214469af97 : SUCCESS in 1h 12m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3e3582626401476aa8bdd25430c16440 : SUCCESS in 1h 04m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/be4f00c96e684926ad73606ebf06c54a : SUCCESS in 55m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fff365f772c413a8a0d8950aaa45b89 : SUCCESS in 33m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c7dea4fd8464ed6aa47d0d6f436d408 : SUCCESS in 23m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc21a02a00de427584477dffb54eac17 : SUCCESS in 12m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ab78af1ee74714a524666d28c92d18 : SUCCESS in 25m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea5348ccdab462fb651aff28fe97461 : SUCCESS in 20m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88be9744c21342c4a6ae0b427194d40c : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d89986e5c98046758335be598d4cdbb2 : SUCCESS in 8m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/44bc66a3aac6482c85bebc3c82d3deaa : SUCCESS in 1h 04m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b392fa1d69a142f28ceed574d2314140 : SUCCESS in 1h 23m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1df0b09078bf4848a95eaf4e5707eb89 : SUCCESS in 22m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1b1bb90cf1b149dfadbe8053177c4016 : SUCCESS in 1h 01m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2740757bad8f4e24af82881f448a0351 : SUCCESS in 1h 08m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aba9ce986fa342dab97d7a1a0a3972d2 : SUCCESS in 2h 13m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e093ed86dac44650bdaee197d2e4598c : SUCCESS in 1h 20m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ac2eb2c11118459caadc5b329ce6fa13 : SUCCESS in 1h 32m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0623fdb97cc344868c273f11c2dc9fd4 : SUCCESS in 1h 34m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f426997706a04182ae3f7b92b2b61108 : SUCCESS in 1h 34m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d3fc7af42c7040c6bc1842383de48700 : SUCCESS in 2h 04m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ca60b1f32ec34010965b3eb75ff83af3 : SUCCESS in 2h 11m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09de162a3f8242cf9bf9b140ee964935 : SUCCESS in 1h 20m 16s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/0f92d600d7dd4b4d96604693f0b781f0 : SUCCESS in 1h 35m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7553a444d85e4ffd88f04299ac23a026 : SUCCESS in 1h 01m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f7273e5e19e148bcbe7d2fc8a118ed74 : SUCCESS in 1h 09m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d9b870f076346f68bfe197134d24c50 : SUCCESS in 2h 26m 14s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e0f72301a72f4a67ae1d74175996d935 : SUCCESS in 2h 39m 06s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/a988290734374aa0a78b7d01bac4e416 : FAILURE in 1h 30m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/195b7a67d88d4fc8bf02c3f1edf1e528 : SUCCESS in 1h 07m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2a8d184620fa4f1ebb015ae7f5c713fb : SUCCESS in 1h 23m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bec550c28244c44bf0329501c84132a : SUCCESS in 49m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/523e8605f55b48758001acd5471aa28d : SUCCESS in 1h 04m 06s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/203993a08f164a44a8a45d793ff0b1d0 : SUCCESS in 1h 19m 54s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/594ef8925c704ee0a91f579caf1a052d : SUCCESS in 2h 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e170624b93fc19b78eaf363bb20b98bd7f7b79c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-08 17:44:13.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/33/686833/3/check-mlnx/Neutron-ML2-Sriov/fd67a5a : SUCCESS in 1h 23m 01s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/33/686833/3/check-mlnx/Neutron-MACVTAP-ML2-Sriov/4848731 : FAILURE in 1h 24m 21s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"7bbf4dcfe3f7c111755fa3a0b8080e9149cdcbbc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 18:38:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-08/686833/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c55a660 : FAILURE in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e2d4d11b3d53497b9af60db280786cd6b287a9e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 18:40:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-08/686833/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a4fa2f4 : FAILURE in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9b03794058b2662ade84ddf5686fae6c068a31a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-09 06:54:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1214f6ede18ae1566266e6e4b7acb7bcccefd5f5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-09 12:21:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"929205bd54df51e3b01dbd24f86b17961e7bf854","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-09 14:24:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1922742de1bc1c40448eb8ee461d9fc53a00c68","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-09 18:14:12.000000000","message":"Patch Set 3:\n\nI guest we need to modify project-config to remove \"openstack-tox-py27\". But I don\u0027t find how. Anyway, this change does not depend on this patch.","accounts_in_message":[],"_revision_number":3},{"id":"58f158e02a155ea828a91c9d3be0526effc039cb","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-09 18:29:24.000000000","message":"Patch Set 3:\n\n\u003e I guest we need to modify project-config to remove\n \u003e \"openstack-tox-py27\". But I don\u0027t find how. Anyway, this change\n \u003e does not depend on this patch.\n\nWait for coordinated changes to come from gmann and the QA team.","accounts_in_message":[],"_revision_number":3},{"id":"d28b130fabcc0a5df5fcd6d94a015b942a750055","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-10 10:02:40.000000000","message":"Patch Set 3:\n\n\u003e \u003e I guest we need to modify project-config to remove\n \u003e \u003e \"openstack-tox-py27\". But I don\u0027t find how. Anyway, this change\n \u003e \u003e does not depend on this patch.\n \u003e \n \u003e Wait for coordinated changes to come from gmann and the QA team.\n\nThanks!!","accounts_in_message":[],"_revision_number":3},{"id":"cd9d24b438fa2f08fc3548fe80356caca6ea07cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-10 14:05:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1cec89becc0071508c5f3e5bd39727b06f68b7cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-10 14:06:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9f9ac989c9772996ef5c07d13b45c30db1de29f3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-10 14:49:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-10/686833/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/499f765 : FAILURE in 42m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edd0a03510c5a0ee6450521c490e69dab1414591","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-10 15:59:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/4 : SUCCESS in 1h 52m 47s","accounts_in_message":[],"_revision_number":4},{"id":"3790966c3d25f3b2fd1474ef184595ac70f45090","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:49:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d9d957388c414577850c3429f8af2e6b : SUCCESS in 3m 29s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7085091a510e406dac5bd711ad1c68a5 : SUCCESS in 50m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc7130f76999402b9a1f340560befaaa : SUCCESS in 34m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7c6d3d27abfa4ddeb0c59231b60d9f79 : FAILURE in 1h 30m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/56444fa7163e4a5cb6c4dff95ea1a918 : SUCCESS in 1h 01m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7910893a2b4149689173571cafe9e6b4 : SUCCESS in 1h 04m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b1dc3dc5a414e188c812c57d97ffa9a : SUCCESS in 56m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8aec67cf43d432f9406ebcdb03d7b22 : SUCCESS in 26m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38232d279d684ea99f11c85337947ed3 : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a206b2f0ea6a4144a9efa9acda9ddbaa : SUCCESS in 11m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7edc1e89fe704fbfb5f8985ac77d36a8 : SUCCESS in 21m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e38d563e0ae14b48936fef4c915b763b : SUCCESS in 25m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1099d0cb26d6486594771cd87e8bc76b : SUCCESS in 24m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67dde107704d499fb3cc656a546ba66a : SUCCESS in 8m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f2270f96f62349768d7ca6f1522f78f9 : SUCCESS in 1h 01m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2e68f706bea24acebdb0c2a3f2dd7125 : SUCCESS in 1h 23m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3ceea85f8bf4f6680241056923a501c : SUCCESS in 24m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/27073434f6304853bc3f9e583a6739c5 : SUCCESS in 53m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7eb051c1274046a59c9f88751182e582 : SUCCESS in 1h 00m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4fd09e0ec46b43639d2a47765b5994d3 : SUCCESS in 1h 57m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e4da9feb6b894449afe53edbce083458 : SUCCESS in 1h 12m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1a64c9444ded461188b3125922e06c20 : SUCCESS in 1h 30m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/755df4d5a13946648b3db152badabb45 : SUCCESS in 1h 23m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d0b35d39713344459383faf8c77b2c73 : SUCCESS in 1h 07m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1dee18413fa74d4a93cd5cfafe361eb1 : SUCCESS in 1h 45m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a6ec0b2e3da74a00ac8720219fa5ef32 : SUCCESS in 1h 56m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93bf94c6dc714c86b0bceb496d362adb : SUCCESS in 1h 35m 01s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/2f581454e6674750a161154790a18d88 : SUCCESS in 1h 31m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eb774cb1c35a40f69a03991be19c6905 : SUCCESS in 1h 07m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de09ac12963241e3b5ec1ae74aba2b8c : SUCCESS in 52m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd37c90ac0574090a5386ab5286bf19d : SUCCESS in 2h 22m 20s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2ecb32ace81a45429cb228a657a2d4f0 : SUCCESS in 2h 16m 06s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2edd3caaa7fc42b48cfa7eb494bd07b8 : FAILURE in 1h 48m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/295399ee3e6e42edb6a8e1f2193efb28 : SUCCESS in 1h 10m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/80a0fbc55df643408a18b6d1d7f77b72 : SUCCESS in 1h 46m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7927d263b33243df9be00734d0ddcaf2 : SUCCESS in 52m 32s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0a2f7bd93c447bcb92454d4a6ec846e : SUCCESS in 1h 15m 18s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/26d71ed0812147f780252194982d49e9 : SUCCESS in 1h 24m 05s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7a2dedd9dabd4242812316d42d7449d4 : FAILURE in 2h 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"687abdd653032dc2872af4cc5464db48a5522eca","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-10 17:07:07.000000000","message":"Patch Set 4:\n\nCan we mark this -W until the official train release so that if we have any last minute backports we don\u0027t run into issues?  We discussed this in today\u0027s TC office hours and there is going to be a communication from gmann to start this process coming soon.","accounts_in_message":[],"_revision_number":4},{"id":"2a1c7a128d19d199c54b51ddcb9b505cde2d3253","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-10 17:34:50.000000000","message":"Patch Set 4: Workflow-1\n\nMarking -W until Train leaves the station, feel free to +2 otherwise :)","accounts_in_message":[],"_revision_number":4},{"id":"257091c8c871f95583b9c542c85ac1e84f7d786c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-10 18:28:46.000000000","message":"Patch Set 4:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/33/686833/4/check-mlnx/Neutron-ML2-Sriov/2dc03d7 : FAILURE in 24m 14s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/33/686833/4/check-mlnx/Neutron-MACVTAP-ML2-Sriov/539e930 : FAILURE in 18m 57s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"c7cf61e1c5b7e3dadb267e5fc0f7b329f317e940","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-10 19:52:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a11f6acebab0efd54f6b4726429902f26d5fe600","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-11 11:59:55.000000000","message":"Patch Set 4: Code-Review+1\n\npython 2 testing is ready to leave with the train!","accounts_in_message":[],"_revision_number":4},{"id":"fd4023252a2116ae93dd0c9483458b0e0f84cbc8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 17:38:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"994bdf4507fb9d76a3cbd9dd82041f654df4249b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 19:51:55.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"638190e22b28e6b90396bdcf884baa810e79334a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 19:52:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f7c69edd75f0ef8effc2feca0460ed402373c276","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 19:53:04.000000000","message":"Patch Set 5: Workflow-1\n\n-W until Train done","accounts_in_message":[],"_revision_number":5},{"id":"4eeb40eb25d34f5f5b25e459028109cc8b9d3fa5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 20:24:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-11/686833/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/1aa347e : FAILURE in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d02e1f215b04a0203891cda24c0f7c8df8e12ffe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 21:19:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b0076a6bf61d22d8f13b01cca616fa121303edc9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-11 21:20:09.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"93adda30437384e580d14daa8133ad1936b6615f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 21:56:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-11/686833/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/f7eafd6 : FAILURE in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c4d162672c2cf250c1beac8b27956393df9dd0a1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-11 23:16:34.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686833/6 : FAILURE in 1h 56m 39s","accounts_in_message":[],"_revision_number":6},{"id":"c2ff8029e7d92980c2c327f645bc3375d31c7440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 01:12:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b852fef1b15a45bf9071371324c81eee : SUCCESS in 3m 34s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd60e9710640428b92f91c5ecf07a699 : SUCCESS in 1h 32m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f57a71330bb540f6bf463c2a9eb1c832 : TIMED_OUT in 1h 03m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4551c131af744eabb216ce2da82c715a : FAILURE in 2h 39m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9df6bbcee38c483496003090ebc71c7f : SUCCESS in 1h 34m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/59245a9d8d76420fa0121223fba9d36a : SUCCESS in 1h 46m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7593d862d8e4cb882f2fe47d56ab34d : SUCCESS in 1h 40m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/939110426a77471ea250fad43e7f1c4a : SUCCESS in 27m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18a7b6feae6b4aeaaed48fcfed178efe : SUCCESS in 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d354ca6782449ac8c7eca0e5b5cc397 : SUCCESS in 15m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52765597d35042c1896760f68346fed6 : SUCCESS in 24m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17256fc4996249c2b89607d5dcb1d0ae : SUCCESS in 20m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c34d3acfafdc4a6ba62e10a1f1eca310 : SUCCESS in 22m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5456637e414d6495c235b92b1ca8ec : SUCCESS in 9m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f83fba3f2064217afd0ccc99b8aab38 : SUCCESS in 4m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6ee3f4dceb004da3a2fab574037a3506 : SUCCESS in 1h 21m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/70e5634bda7f409e84edb72aa80861d5 : SUCCESS in 1h 28m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2754ab0053424ae68f4be13bc1db2675 : SUCCESS in 22m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c40c2c6ced1d41f38f4affc275f7a2ac : SUCCESS in 46m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/935810ee4ea1456993867e86aad68626 : SUCCESS in 1h 05m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8e3b39190b6e475d95cad3dfb2699e3c : SUCCESS in 2h 05m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cadd9de02c1b46c28921c3e1b363a2c3 : SUCCESS in 1h 13m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/65f5d2c83de8403eac11bb671130a063 : SUCCESS in 1h 18m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/210bba5d41fb491fb03de0b5caae56c4 : SUCCESS in 1h 19m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f29bc7f28edb4f1f8cb20acba3890b81 : SUCCESS in 1h 36m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/891c2998c2c244618bea68ea6001fbca : SUCCESS in 2h 19m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c80bffafe904605819e694341cf1a38 : SUCCESS in 2h 28m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8bc00afb24a4492a9edff5c12d6a93d1 : SUCCESS in 1h 20m 36s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/0af29576176e4e4f9821175119820639 : SUCCESS in 1h 41m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/24fec11b92a645ea89a364861cdd61ec : SUCCESS in 1h 04m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7c3f742bb3904bd08af32eefba947943 : SUCCESS in 1h 02m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7b79faf260e47b99b75ad1ee0cbfce0 : SUCCESS in 2h 17m 49s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c51b61cf81cc453fb41b60384d4efa68 : SUCCESS in 2h 10m 25s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ad5f87d2cf1342429e060ef2f74b888d : FAILURE in 1h 42m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8e2234135224841a0d31d834b52e6c0 : SUCCESS in 1h 06m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f36638ccfc5e42d6aae612de52827aac : SUCCESS in 1h 25m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d7dad7052054714969b1fd9c4af99ec : SUCCESS in 42m 22s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e852cc8164ca40c98f59e20fa08164ca : SUCCESS in 1h 08m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c678f4d388e14c03a045950681df7cf3 : SUCCESS in 1h 18m 19s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/201b19ed1cf24d0fb0f7011f7737c1bd : TIMED_OUT in 3h 48m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"beb16e5ed1d0eaebf88aa0b4507930a43b59c45f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-14 19:47:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7cb3b51c42387b0184fc0b64e5199575d66a1dc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 23:45:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/68bc4912db934706a780da8e1b634a62 : SUCCESS in 4m 53s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d42d728e40b64dada91b96e50fe6e2cb : SUCCESS in 1h 01m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc9bb637300547da8cd25f5c0092a798 : SUCCESS in 35m 15s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2f88abbab5824b73ad4ddfc7ccfb8877 : FAILURE in 1h 41m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/69d7ac44d53f4b6eb917d9148fb716b0 : SUCCESS in 1h 14m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/86904c52810944ba92358cfe953de157 : SUCCESS in 1h 00m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3cac7a32d41e419ba29054d1f3925c1b : SUCCESS in 1h 08m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbaed8af2cda414c9b4d73cb78b9c311 : FAILURE in 30m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d175ee52fe0a47cd95f5de6bbe031ec7 : SUCCESS in 21m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b66fd24b4cd49b68bea3edde9522120 : SUCCESS in 13m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3299fa531f574a95aff245a25c86c341 : SUCCESS in 24m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a021f90a3134d8f9c3378ecea21d588 : SUCCESS in 23m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1d8e420c3d14fc69545a3144ae93e43 : SUCCESS in 28m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3876b22a192a4904afadb41a5fd343ff : SUCCESS in 8m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/201af0fa06194f9695753e6bd98b5ad6 : SUCCESS in 2m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65ca0e19a52a4dfdac608c183918f56a : SUCCESS in 1h 03m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/45684e84d8f946698f6b96735b20f8b9 : SUCCESS in 1h 21m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a311d1964e43dca1b5878a87168409 : SUCCESS in 21m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ca65f0babc904d09ad2eb6c492781deb : SUCCESS in 50m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a797a8aa3d8f4159a8080bf7d193be95 : SUCCESS in 59m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/734e595600c94a70bb566d54045472d9 : SUCCESS in 1h 55m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0c0d79b9564342fe82d2e2499e4bd451 : SUCCESS in 1h 14m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/20b3065f70a74e1ea835ad605753f3ab : SUCCESS in 1h 20m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54dec1be99cf48888703cce8e4455bdb : SUCCESS in 1h 19m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d99804e64b34e7bac8094389f59cb94 : SUCCESS in 1h 10m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae84824ecf4241ecb967df454bae2797 : SUCCESS in 1h 37m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/135b13ccae6d4a14a1caeb133b4010c4 : SUCCESS in 1h 46m 54s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7d8fe696b78743e1aedfb45539fc9a29 : SUCCESS in 1h 17m 54s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/32f22d4f6b75479eaaddb9c61bf984b4 : SUCCESS in 1h 37m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1c3e2a67647843dda9e68b5f96e45472 : SUCCESS in 1h 02m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d7bf5f00fbc2487c82a4eba0054cc91e : SUCCESS in 49m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7b3cbe094384cd4ae4b5b30f917fb09 : SUCCESS in 2h 19m 50s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e7467b74d1a44c6990650fe1ec9855f1 : SUCCESS in 2h 20m 09s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/fd432a2b986f4c8d8cc9e65329f87285 : FAILURE in 1h 30m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1d7370bab1245988599c487dce5852a : SUCCESS in 1h 02m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f8612bb3ef3646438be8d378c36c6bc2 : SUCCESS in 1h 27m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17221f18e7574373a35e0cb25f120571 : SUCCESS in 45m 40s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0381db8efb8d47a099a2be6a4b3bf502 : SUCCESS in 1h 11m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/17168f94597047dc8244cbd40ec7ebfd : SUCCESS in 1h 24m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fb329faffe97434eb2f323023a971208 : SUCCESS in 2h 06m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e2835c14162559f562bec7c8e408e57490c3c816","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 03:03:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-15/686833/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/b3cf496 : FAILURE in 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d0112b3f8f76716495db57a60c014952054ad34f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-15 14:34:38.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/7 : SUCCESS in 2h 00m 15s","accounts_in_message":[],"_revision_number":7},{"id":"fca8cb8dacc2228fb4da3d75824702fe62a55e09","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-15 18:17:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8e03bd46a9c93e3b52d4a2e4649b59ec2009ffb4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 18:49:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-15/686833/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c3a4de6 : FAILURE in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7bb3b3ad192121b7cdc558a6d1163522c9cb0455","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-15 20:24:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/8 : SUCCESS in 2h 06m 51s","accounts_in_message":[],"_revision_number":8},{"id":"1b1fda02e2c6bcff47ac373c13eee0b8cf31e930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 21:31:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/665d26eec13b478182ceba72f4a714bc : SUCCESS in 3m 01s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b8b954953844c108177c28d9da4d1d5 : SUCCESS in 1h 05m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/050aff891c9045b59d4f5029484891f1 : SUCCESS in 42m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/01cc3996e55047a28870f14be72b6449 : FAILURE in 1h 55m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1957b2583f894a4f98f878e60917f5d6 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/55a122dd7b1840549ee267a59fa4ed88 : SUCCESS in 1h 35m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6a0fe79d2d44bb1b1a0917e93f7a330 : SUCCESS in 1h 08m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2fe500a5bd14319801e75d03977c75a : SUCCESS in 33m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d773f74a1baa4b809e398acaa00a0b5c : SUCCESS in 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726a5c92ae9e4265ba7393e3105eaf7a : SUCCESS in 12m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/725eea4d8a0a4539abf5d0cf4fd9dc4d : SUCCESS in 22m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9187557087964dc988fb667324563696 : SUCCESS in 23m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7d3937b129f4128ad2e74bae144ddcd : SUCCESS in 29m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9caa328e78a47118d437a5f0f01178c : SUCCESS in 8m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cb91709b77c4511ac0320e05069db9b : SUCCESS in 2m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/597c91c612e340fdbc53021e60a47c3d : SUCCESS in 1h 03m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4fa8a556e1524b76b8f2d6280df21c95 : SUCCESS in 1h 23m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b60a4b631a8c4d5892ca7f10c8a07a10 : SUCCESS in 23m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/622122cd869645e08f8176f07004f801 : SUCCESS in 48m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ef9f882f432248babf420c9461aa1494 : SUCCESS in 1h 10m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4884ee4db80249b2aa4d42785982f20e : SUCCESS in 2h 08m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/50c805a061ce4b64b5c84da6788c2dcc : SUCCESS in 1h 24m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4b0de12f2e3e4989bf00cbba2e230f4b : SUCCESS in 1h 19m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a622ca78ae0465a84d609c278bf97ff : SUCCESS in 1h 13m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a16ee35e1cd4e07809eff6ec328192c : SUCCESS in 1h 30m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e47d9158a04a43718fdd684bbbbc9810 : SUCCESS in 1h 54m 35s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8485132274404cd483f0f1477df5665b : SUCCESS in 1h 55m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac3995d86fdd441e86bd68d1a531f69e : SUCCESS in 1h 21m 30s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/c52634e94613442e8c934ce4f3d13fdc : SUCCESS in 1h 38m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f90012bff3f3496aaf43a0b9dca70659 : SUCCESS in 1h 05m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6f144a725b02449b9ff648d571cf7c99 : SUCCESS in 55m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19e16aede7fa48928fb4b94b8bf50a4c : TIMED_OUT in 3h 04m 16s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5596b824f6e240128c08995afb57cc4b : SUCCESS in 2h 24m 41s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b3e10a95a5664f95a07fc3e618c5ef70 : FAILURE in 1h 51m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30b7319eeca643a0a53fd66db797899b : SUCCESS in 1h 08m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3f8026846ba745cea92e634f38de559b : SUCCESS in 1h 21m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f966f08d963b414c83f2a5df8ec59a79 : SUCCESS in 53m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2598a286bae470b8cc514587cdd3c85 : SUCCESS in 1h 09m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb7cab5095f74db88a96bb94975a4fc8 : SUCCESS in 1h 12m 08s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/285ffcf7b3834a839d96c866c6d11600 : FAILURE in 2h 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"095e94873183b019587524d49c334e4729eaf2ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 14:10:23.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"05bf5c9042e87b8bb80e4e8fa337f7802e809ac4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-16 14:48:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-16/686833/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/f004cea : FAILURE in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1e14a27e76ffb27e37f0b860c4cc8fe163f1b6db","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-16 16:10:25.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/8 : SUCCESS in 1h 59m 23s","accounts_in_message":[],"_revision_number":8},{"id":"09005f1f22e8b031990eebc2e20ea1b74194c95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 18:05:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eea298e905b64dc1912132f22f51644b : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5f5799ab1bcd459b957b9e30e7f8f619 : SUCCESS in 49m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f4f1857a5b54c92a81b87339a46b423 : SUCCESS in 37m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bb222decaea54fb395dc4bfe9401efce : FAILURE in 1h 44m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/67b4eab8a9b1486cbb1ce251bc8a4841 : SUCCESS in 1h 12m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8fdd2ea1661547138443b36e9c66ff22 : SUCCESS in 1h 00m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/50fd2e1ff3f749eabe03a7e279eabc0c : SUCCESS in 1h 00m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f2cacba621140ac8eb1616409856f7c : SUCCESS in 29m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7d948626ddc4181900c0e82bca8365a : SUCCESS in 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1517993c774094ab8d892f69e6569c : SUCCESS in 13m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dcd0bad16e8547be8e5d6d1e87aa1d84 : SUCCESS in 30m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f919b5a4a3a4f9e95c844f8dffb02e0 : SUCCESS in 30m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6fc3600642e1424cbcc73e3f73cf3434 : SUCCESS in 32m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8149861a3cb449929b7175e01e629487 : SUCCESS in 12m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c666c67203341248957d04a2416243e : SUCCESS in 2m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/02b26f44b4354088acd1322c60c22817 : SUCCESS in 52m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/216aaa32450544eebe1325d64dd9b154 : SUCCESS in 1h 17m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e019a4c79fe46db938803afae8d3fe8 : SUCCESS in 25m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/805f2ae8acf54a5b93c876f698185701 : SUCCESS in 48m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a3ae1316cfd54f2893ad18944d9a6338 : SUCCESS in 1h 05m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/859ae38a9ef34c49b2a32d10588c7099 : SUCCESS in 2h 20m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/977a68c0560f41fbb40841aed20a49ab : SUCCESS in 1h 24m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/364461155f6e4e6587b047c5d1338d8e : SUCCESS in 1h 29m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86b14c3965ab4de68d2d6683781e938d : SUCCESS in 1h 21m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c90b5773a5de4e57997854cbb62a321f : SUCCESS in 1h 19m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1922c64ab324bc88ee9aa1f3ebd59c2 : SUCCESS in 1h 56m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fdf365b88a4743ddbcb751dd9a7a8be0 : SUCCESS in 1h 55m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3c3fc0623117447699929a8703705199 : SUCCESS in 1h 12m 39s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/26fcd8b98085447ca211803b56b1fb83 : SUCCESS in 1h 23m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ecf1ab2ff7624f39a6839accae715492 : SUCCESS in 1h 08m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c7ea2e444d14f2cad7a041582a0b192 : SUCCESS in 50m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1718d5e4f47142c3986735270d0c793b : SUCCESS in 2h 11m 30s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c937906b1bfc47beb3534a59c0f14ebe : SUCCESS in 2h 13m 56s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2fbc68f56b744a35a778f4976b721486 : FAILURE in 1h 43m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6209ffda742a4bb3a0a8b3f6bcfdb43f : SUCCESS in 1h 08m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5b7e04c6e11143a480cd30d65a304908 : SUCCESS in 1h 32m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/08d0b17e2de44b5ba207d33eaca59bdf : SUCCESS in 41m 50s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/831a44ecc3764b0590cff3e8b15f8294 : SUCCESS in 1h 07m 27s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/72b3aee34a86409ca305fe6b65bb7067 : SUCCESS in 1h 10m 34s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5d0a646726164fadb3700852aebd835f : FAILURE in 2h 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"72883db77094228946c59325355fccc1eb284dc3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 18:42:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0c96767eee958b6eec8d10185de56440780036b2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-16 18:42:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d3ff9329fc507900d3495f2facc67166b8fcee06","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-16 19:15:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-16/686833/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/57722c8 : FAILURE in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"66a7a3c56b9efb132a431f7b5bec752b5293980f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-16 20:23:03.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/686833/9 : FAILURE in 1h 39m 58s","accounts_in_message":[],"_revision_number":9},{"id":"18407fc39fbaac1fd9be85da38988c73f6bbbee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:31:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7afd74dca4694809be43eab93c94410d : SUCCESS in 4m 05s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2981f32e21374df6b4e50b03e660b696 : SUCCESS in 1h 00m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd56ffc62d85432587937b76e11ab6c4 : SUCCESS in 39m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f4f8746c7b1d447bba7daffd80151524 : FAILURE in 1h 42m 35s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3edd4d2946e74584a22152fb7538938a : SUCCESS in 1h 07m 37s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f79fbba6ea474f8f82481a0dda42b7f7 : SUCCESS in 59m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c55dbb22bea40b29bf55a0827dce182 : SUCCESS in 1h 17m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61dc721f068c4503a9cec24bb5930683 : SUCCESS in 28m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e509cfbb9094485a5f2cdb869b9ce46 : SUCCESS in 32m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5eb5f7a2d3e42adb0da09be4ee3f267 : SUCCESS in 11m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae270f7e57584a1b86acc1aebeadb367 : SUCCESS in 25m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea5f2bb98874e3daabc07b91fe678c2 : SUCCESS in 22m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca8326a98b7d4d08958377ffe6b84f17 : SUCCESS in 21m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0df6e6e40040cbbdee6390996fb548 : SUCCESS in 10m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2f05a7ba2ea4292a07cf4c563a842e1 : SUCCESS in 3m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee5e702fcdf94075ad24b245afb7847d : SUCCESS in 56m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/046109ae6ea84ce48b98ec484a467572 : SUCCESS in 1h 22m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/994a7de8533c4728b8590171ba69c7df : SUCCESS in 23m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4adc40c95e2448e2a1d083b91842f3bc : SUCCESS in 43m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/30b8d8738efb4b768b82bb6adc58f76b : SUCCESS in 1h 13m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/69f81d706c0542a1b8c631dbb340ef3a : SUCCESS in 1h 57m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dc3cd631a91b4910bd218b3fa519ae5b : SUCCESS in 1h 22m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3ad1eaaa5af240bf9f0b685a1a8aa5b4 : SUCCESS in 1h 18m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d23e4d594d9e4beaaea353573f06ec65 : SUCCESS in 1h 15m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53fd6d96dc4347b689255c7e3f3b2da6 : SUCCESS in 1h 25m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80ad0f6c229247b0bf1c006f73b9c26d : SUCCESS in 1h 46m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/acc9918370c2414ba11fa9ca40014d8f : SUCCESS in 1h 59m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3141ae9f58994ab7a50b223da64b69c7 : SUCCESS in 1h 32m 01s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/770c7a4d192e4e0b9d66fa297208a731 : SUCCESS in 1h 35m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/232fb554e47c467f88377411646907c3 : SUCCESS in 1h 03m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5d3e5fda9f4344a5be8741a07f479097 : SUCCESS in 59m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e142539497241718b3838c0e2061f56 : SUCCESS in 2h 27m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ce92fec37d754cecb1643bbc1841019a : SUCCESS in 2h 13m 28s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9c0f06bead9c4a4f84f80c602bd45f3e : FAILURE in 1h 33m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ed5e57933804356ac03fd6bde5a2468 : SUCCESS in 1h 11m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/655b233193674f59a2362aa5f835412d : SUCCESS in 1h 35m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/489dee41b1304e8c9044e81452575af3 : SUCCESS in 44m 44s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f59f038b79d148f2a17623831815b2f4 : SUCCESS in 1h 04m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4563c640c6e24c89986a5de2ff6aa227 : SUCCESS in 1h 35m 05s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4690ed9d53554908aef2c8e3961a49fa : SUCCESS in 2h 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"102fe2bd90df3a48018cc3eea55940f4d4ae1890","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-17 21:01:12.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"b7d32e713bc0a0a3627da98cb652c03be225d836","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-17 21:34:31.000000000","message":"Patch Set 9: Code-Review+1 Workflow-1\n\nMarking -W until the TC office hours discussion on 10/24 where the python 2 plan will be discussed, as announced.","accounts_in_message":[],"_revision_number":9},{"id":"118529320fdafa5087a46fe789c33547c6ce3c30","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-25 06:57:10.000000000","message":"Patch Set 9:\n\nI think that according to http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010356.html we can now rebase this patch and go with it in neutron.\nWe can probably also start dropping python2 jobs (and support) from other stadium projects, except neutron-lib which should happend after M-1.","accounts_in_message":[],"_revision_number":9},{"id":"28a542920341c7b39dc3b38897fb37334320d907","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-25 11:17:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"533e92eb141f725dd2e8d2d46da576c361c309d9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-25 11:18:26.000000000","message":"Patch Set 10: Code-Review+2 Workflow-1\n\nSetting again Nate\u0027s -W","accounts_in_message":[],"_revision_number":10},{"id":"0ed8ca92b05c71b6ca4f7104fee6ac4843634187","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-25 11:50:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-25/686833/10/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/732eb3b : FAILURE in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cdbedc9a567b5224dec6e7010646106c77eb65e9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-25 12:57:02.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/10 : SUCCESS in 1h 38m 49s","accounts_in_message":[],"_revision_number":10},{"id":"8dae32ec6b300888e95dca4cbb4970189cfc1e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:04:21.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5584fe8f526e4eacb697a1e5ecb9f37d : SUCCESS in 2m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e712b978588f4832965dc0124d9a0033 : SUCCESS in 48m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/33ff9f8dda9048f984777368ffab556b : FAILURE in 29m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9dbf2fbf55de419ea868939f09e8a50f : FAILURE in 1h 29m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d4579739cdd9432d8addc9ca8dc892f2 : FAILURE in 32m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bad3b68980164c538fca52e2c54b3c0d : FAILURE in 29m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/52ddefc0737341ea97663948264e9c01 : SUCCESS in 57m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/956715939a844389847d18a8a42b35e0 : SUCCESS in 25m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6461aed6289c4c0eb5ffc66007f17c6d : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0200417bb942a581a56227a3595055 : SUCCESS in 10m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0d7c4ec98c2427588b6dba58747db94 : SUCCESS in 19m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57d4ae3964b4461297e18f3e2b90b302 : SUCCESS in 19m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/576143a0373f4cdf91e63f2eedc944e5 : SUCCESS in 19m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d05da23b2fe24a7ab837214087fdd709 : SUCCESS in 10m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7bc8cedfd4e04997b32090b98f390cb4 : SUCCESS in 3m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/85d0fa7870fa463ab737b5332526e56a : SUCCESS in 1h 08m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c68b6aea76948a9a5461bf89ec7c9c7 : SUCCESS in 1h 06m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67e67021225d48f9a92a198d57dad2da : SUCCESS in 21m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c82cee333a1240f898645177804f2b30 : SUCCESS in 45m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fe597300727245b69ea4c5b4e5561a63 : FAILURE in 1h 03m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/eae163cfc2224a0b8cfb034335a0e65d : SUCCESS in 2h 02m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/762400b04c964af497216922987e9254 : SUCCESS in 1h 27m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d4684fdec24942429478454d3c36d9e4 : FAILURE in 57m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68795d0d516c49e98bde06aab36c82c5 : SUCCESS in 1h 16m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd02f0e8beb84048bef49021040e2c5f : SUCCESS in 1h 07m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/07b66bc0551c4bcaaac542f767098b03 : SUCCESS in 1h 38m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/010b01f767904c81927d5914f826bee6 : SUCCESS in 1h 44m 41s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb28b82123cc464690b71fa6be4fb448 : SUCCESS in 1h 21m 00s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/72d54198e2824618bdf79566e9de9d11 : SUCCESS in 1h 24m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/73012948d3c74c5e9f458bdcc462a141 : SUCCESS in 1h 11m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a228c42efe064754a22221d703bfb22b : SUCCESS in 53m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d772f16d43b84c2a871354a14bafe955 : SUCCESS in 2h 16m 09s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/afd3499308dc43edaa816c64dbf08ca1 : SUCCESS in 2h 38m 29s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b214aa0c8780478a87c348b1306a8c5c : FAILURE in 2h 06m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2165ed71edb4970814a93cdd8e37fec : SUCCESS in 1h 02m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1874b701d63348f88cbb557d96eb7122 : SUCCESS in 1h 33m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/222f94715a564d6f93370ee3c410cd15 : SUCCESS in 49m 33s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc27aad847a04583a2314d203da82fa4 : SUCCESS in 1h 10m 06s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbc532fa94fe4b54981ff6ac1477c881 : SUCCESS in 1h 23m 24s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/83c894180a544916800d000aeb550380 : SUCCESS in 2h 09m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"66b1cf508a5c09abdc4ea5e5ec0e8d2905c35a12","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-25 14:57:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b2324803ff9aa81cd0c1c6292b71b910f60eef38","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-25 14:57:25.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2060383f9bac0c48659adf6593ff281d24ca2933","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-25 15:29:29.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-25/686833/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/df539d2 : FAILURE in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"68c13bcc05338ebd9ededa386432e3109f4fd823","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-25 16:38:49.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/11 : SUCCESS in 1h 40m 49s","accounts_in_message":[],"_revision_number":11},{"id":"e689740efb4368d5fea8f56f640db666d99ac1cc","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-25 16:44:45.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nCan we clean up a comment on python 2.7.4 in setup.py [1]?\n\n[1] https://github.com/openstack/neutron/blob/master/setup.py#L19-L21","accounts_in_message":[],"_revision_number":11},{"id":"4dcacd6a4f98cfbc1bbc97bcb321f3ff606e8f14","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-25 17:13:59.000000000","message":"Patch Set 11:\n\nAkihiro, setup.py has this comment:\n\n# THIS FILE IS MANAGED BY THE GLOBAL REQUIREMENTS REPO - DO NOT EDIT\n\nThe file is the same way in the openstack-requirements repo, so maybe we wait?","accounts_in_message":[],"_revision_number":11},{"id":"f08b7857e2ed9cad324daebc0220abd696c3d930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:51:37.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa1fdb5e68b6457ab6f11156d40ca7a8 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/43a797a59cf54616bf56a8fabb8f2185 : SUCCESS in 1h 01m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0da0dc9b225a46b9a22a741397073053 : SUCCESS in 49m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1e2add030f494827a0b341f832fc55c3 : FAILURE in 1h 40m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1b71885479eb419e9b9f736c1ed64af4 : FAILURE in 32m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a04b52bd81724ea3a711e5bca1d15f82 : FAILURE in 1h 31m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fc811e2b1b541e9870ee4a3fd522cf7 : SUCCESS in 1h 17m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac68787765464ecdb10b8ddf5bebce21 : SUCCESS in 36m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e13f192951364ce1b9a260f02890a9a9 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486e2ec4e0f44f16aca3a527bef5ec54 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b33360d9f54f00a57f2bd4f182d114 : SUCCESS in 19m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/690f8e8eb7474e60b03ce33ad6bf4664 : SUCCESS in 20m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21801ff9d17e48be96894f6efd176199 : SUCCESS in 10m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7e5a6455ca31416cb5c31461c7bc305a : SUCCESS in 5m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/502224dd4a2d48d791d556d1140c567e : FAILURE in 1h 14m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb6bb5a4d9664de4b778295ed965b2cf : SUCCESS in 1h 31m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0499855ff914182a4fbd85713b84c93 : SUCCESS in 20m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1d6cfc3ecd9749a78b6760f07d6ee057 : SUCCESS in 45m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/48ea7708886948f2aa752d5cad2d73a1 : SUCCESS in 1h 02m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6f6a9fc33c0e4cf3aaef1e4cca9785c1 : SUCCESS in 2h 20m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cbadfeb4c3934119851e87f51b0dc079 : SUCCESS in 1h 34m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ea452c1e15a04c1185a3965612e6574e : SUCCESS in 1h 33m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3c3dfe391f74d80810b025dc679ef25 : SUCCESS in 1h 09m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e873945ccfa4046a42508398b4662c8 : SUCCESS in 1h 30m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f19856190c604cbfb9c0ef68206df072 : SUCCESS in 1h 50m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88074fc403034b868c31e4e17d93da30 : SUCCESS in 1h 54m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cd9894c34f614f80ad841dc5e50c9941 : SUCCESS in 1h 36m 29s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/11ff564e6f3448d5a3a7819419d32df2 : SUCCESS in 1h 45m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/67e7e8661a6e4bfb9552151f4ff98794 : SUCCESS in 1h 28m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/45d20f6781f34b32a955d62e47ca51c7 : SUCCESS in 1h 18m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b483d2c1e8b48d1b5e4f4f1ad2d58bf : SUCCESS in 2h 39m 37s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/10d54b643dc04bcb97d63c11a169a123 : SUCCESS in 2h 06m 39s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9357ef4b4ee44e6d84d8ecf846ab1def : FAILURE in 1h 17m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95a133ad82e48cfbc8422d7588e358a : SUCCESS in 1h 03m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5cecbe1e771242feae683dcefe040132 : SUCCESS in 1h 20m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a9720bda95f4df29d5ede5572afc192 : SUCCESS in 42m 53s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e0dadc0521444299168fd18e8481183 : SUCCESS in 1h 01m 41s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fb16ccb227f54105a75bd645d951cd1c : SUCCESS in 1h 16m 11s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1788e8550500497985f8a80b30dd4400 : SUCCESS in 1h 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"19e32e249f59c4042b229f65edfa9d8b20dbcefa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-25 18:50:08.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"f8ae7794c76169216747b5ea4600e66c8989058d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-25 19:23:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-25/686833/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/486c8bb : FAILURE in 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"48a44e70f08d149d94af06f6c10481d2cf6afe3b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-25 20:32:13.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/12 : SUCCESS in 1h 41m 34s","accounts_in_message":[],"_revision_number":12},{"id":"ca7d547dce3379788888f09d6d80e1515f69578a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 21:10:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e90d6752e17a4ea7801ee6c8b74c3cdf : SUCCESS in 3m 31s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c172de5fe87848688a96c7ee233a9a1d : SUCCESS in 52m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49df4ee63d7548f7bf215f1c277b6025 : SUCCESS in 32m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ca2886784a4b42cfbdf581b196445c69 : FAILURE in 1h 41m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5b7eb2d5616540509476c44850a52950 : SUCCESS in 52m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c65a545d840e4db3bb1377813ac653a5 : SUCCESS in 51m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cfb96f1f20124ca1b8e85618bf207c7f : SUCCESS in 53m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/143a0f54108043f88ea3400dcdac5436 : SUCCESS in 42m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c529f66da18640919d878ed39660feae : SUCCESS in 26m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f19604987af45bfb8b0ce1bc662c63c : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/637610a837554b138e28c81dd70bae95 : SUCCESS in 25m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9f8d396ab3a424786c97f829ad4ec11 : SUCCESS in 22m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9481128bdbe4d7992e022f523e80ccc : SUCCESS in 8m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e4581b9bc7fb48828967c6cc7436869c : SUCCESS in 2m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/266df1a83f2c4113a2ab1682a10d1572 : SUCCESS in 52m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ddd66118d0b40e38c47c1ff7e831290 : SUCCESS in 1h 25m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bde4a3b16a7349e1b4fd70f639ff153a : SUCCESS in 22m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2873d26ad3784a2f83e35a03fe644df5 : SUCCESS in 53m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ddbca8094ed84184b0f1c60dd3f19ca6 : SUCCESS in 1h 06m 53s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0746eabac9d74c6cb181284ded088a79 : SUCCESS in 1h 42m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51681520d3c9481ca7046657bf8d462f : SUCCESS in 1h 15m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8040a9368fce4cd995c148fa8f540a76 : SUCCESS in 1h 20m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b73eb124d5434aa5b7efebcc6ebce69b : SUCCESS in 1h 09m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4036c4b05d74e8aa23e96c0a585b0e2 : SUCCESS in 1h 19m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/13bc245859ea4943adc36af2c78b12e3 : SUCCESS in 1h 42m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98ae36ec6b704b67a1f5e2ad8ca5a6a7 : SUCCESS in 2h 03m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27ff4a2fa83044e2b9113cc4826f317e : SUCCESS in 1h 07m 06s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/4cb56eabc7b74c27998cf636c6d74779 : SUCCESS in 1h 22m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc3bc9f1e7b24dd2b9dce251d241a15a : SUCCESS in 53m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/084697de9d054deb90ae391fcc18dd6c : SUCCESS in 51m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a7bd233b96524b629305ce7c7cf4983f : SUCCESS in 2h 18m 10s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7c70ce9a74cb44e58dafa2e87d13dab2 : SUCCESS in 1h 57m 11s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3e200a10346d4e0dae457ae8831d404b : FAILURE in 1h 16m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94a117b8c4de45bf8841018e26b4e4e8 : SUCCESS in 57m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/55ff5624b96141c6ac4c03c751fbef57 : SUCCESS in 1h 15m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2936532a472545b09818b319645798e6 : SUCCESS in 42m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c3639025da847949fa2f7dbd54a0898 : SUCCESS in 1h 07m 46s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4be4797e254a4f6ebcda84e48c20140e : SUCCESS in 1h 08m 21s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7d358f20efe64debb169a22ced9728ac : SUCCESS in 1h 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5354efbfef9af1018254ef5ad8ae8b6efa001862","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-10-26 00:21:31.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"14ac4da9f49bb1ca85ebd5c402fef2ee02ea991d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-27 10:09:12.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nI think we are good to go with this finally :)","accounts_in_message":[],"_revision_number":12},{"id":"4f80510c6fdf9c348acb7c4b48f9da16b564abbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 10:09:30.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"302b0a3411b631fe6e00cb0740a09e8df7be4556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 12:45:51.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3e83c29ac87a41a5b5c6b77eb4c5f8ff : SUCCESS in 2m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5eb9e094c80547ad89167b3ee7f041fe : SUCCESS in 41m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/226fffb002274030a547dd5ede6c12fd : FAILURE in 1h 24m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7085ed57dfe14626b338cf92af18538b : SUCCESS in 1h 26m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5665f280e796406ea74ff4195753069e : SUCCESS in 1h 23m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96781701c0394c72bd226a940a02e5b1 : SUCCESS in 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63370ad95d1549e4ba9657aa9d85eaa6 : SUCCESS in 10m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc0399d073194cd8ac382031990bfc93 : SUCCESS in 19m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1cef27f007c4de481735c0734db73ac : SUCCESS in 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785ae2855e634f5aa81fb1605a34d9c1 : SUCCESS in 7m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bed0a7799a0b4c238c2a19566b576359 : SUCCESS in 2m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/60504377df204e33b6911fec37c89d5c : SUCCESS in 1h 01m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/509f746be6854714a5f77dd1f173d013 : SUCCESS in 1h 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d8a6e90a81a491885c14533dc194756 : SUCCESS in 15m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c5bdd0f5a69464ead4886971ec43f1b : SUCCESS in 51m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/928d2c1dabb6451b9e06f17ddbf59827 : SUCCESS in 1h 04m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7475570fdfd644af8a6b3cd0cd7e7d18 : SUCCESS in 1h 34m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e598e1ed0296424fbcd2b1bccb45b852 : SUCCESS in 1h 43m 33s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d410005138004fd29558255fed5f06e0 : SUCCESS in 1h 17m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df34a04e82c04d518d63831343a4a7d7 : FAILURE in 1h 28m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6b973aefd1904edd857f5588893a5428 : SUCCESS in 1h 23m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9f23a7ea94504f939b691bf3f3f88464 : SUCCESS in 1h 18m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bbda594b5e345aa97b6a4be02752360 : SUCCESS in 2h 30m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00b89e09ee4440fcb553607a97ef7969 : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":12},{"id":"20ced7703c4dc3f77d09f6311e0bfd1916491ce8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-27 19:39:41.000000000","message":"Patch Set 12: Code-Review+1\n\nrecheck I want this in :)","accounts_in_message":[],"_revision_number":12},{"id":"adc6a81c89adf741df5696d029946756586dddfa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-27 21:20:40.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/686833/12 : SUCCESS in 1h 40m 32s","accounts_in_message":[],"_revision_number":12},{"id":"93973daa01f7bce200e861654fce8d5e22dffbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 22:26:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78b9f1733c5845c8b1ecd01ad9aadd71 : SUCCESS in 3m 06s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/dfa221344a3442f1b19cec70c84b3be2 : SUCCESS in 47m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e7a0769c69b432e846e15cd09e7bfb7 : SUCCESS in 31m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5e52a7a4f97d4eb792c6b60c4907d69e : FAILURE in 1h 36m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d5bc8ce628bc4c679f71e0ee2361ae28 : SUCCESS in 1h 09m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6d5740c2463f4cc89ac7e764077cb562 : SUCCESS in 53m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6586fb67c4d847b5876d153ffe73c6fa : SUCCESS in 50m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0fd01ad6a354d2fa57e7fc9ceaca03b : SUCCESS in 28m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f720ca9ac94421fa7e51d50cc12376c : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f19444a3b274bd6adb096182e13e238 : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6267209955146348e198eed2a5702fd : SUCCESS in 22m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d73793e31614a42836e4022d0c1776b : SUCCESS in 24m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db628d16a9ba45738d3c426e60d4d573 : SUCCESS in 7m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7ceb870accc84cada383bccc2359146f : SUCCESS in 3m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/54f20ba91c88410c813b0f4a25c36834 : SUCCESS in 52m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a1878196bdfc413da208f22e56df3131 : SUCCESS in 1h 34m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bd1c56821964b50856294913f4ce5cc : SUCCESS in 17m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e20914de84de415bafd3e3178dd3f8fa : SUCCESS in 59m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1801a2f94dbc4ba6bcf0978e0102975b : SUCCESS in 1h 21m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3e4f3cc8e8c34cf18cf9d8bb902c3b64 : SUCCESS in 2h 15m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5b36d4567a3452098517c674577c556 : SUCCESS in 1h 37m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/318b1f7c6d9b4961b720555b7ff3a203 : SUCCESS in 1h 35m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/693849ae4c294ad6afae1054b53c50f4 : SUCCESS in 1h 11m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dab99ebcbee5410ba640e0feff8229a9 : SUCCESS in 1h 15m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6e287fe8712e4715a2797c63ff0a3397 : SUCCESS in 1h 36m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e61fbe68b184d97ae99c665e997da9a : SUCCESS in 2h 18m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b51ce4429a1e4595a1ae5fb5a0100513 : SUCCESS in 1h 19m 45s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/0fa504be8a244435a6052c6ea89d0d26 : SUCCESS in 1h 13m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c44c84fcb541492dbc2be9173fe9174a : SUCCESS in 48m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c4d927581aa941d58c96548fbf812468 : SUCCESS in 49m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/305bbec88cb74651989a9d9d3b7029d3 : SUCCESS in 2h 43m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ed748b1752df4438a30f05dde4e7942a : SUCCESS in 1h 57m 21s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b2cdbeb15e4e4ef598aba3646bde2885 : FAILURE in 1h 37m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46e56ce578854dab93f34acebdad0403 : SUCCESS in 57m 01s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/edb611c601c84bad9a1671ade6dce649 : SUCCESS in 1h 17m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9155347abc8439f8324e395749a874b : SUCCESS in 59m 14s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe16e11160a141bb8ed830cc9b076531 : SUCCESS in 58m 04s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0ecee1efe7d45b39316207e9ad94b76 : SUCCESS in 1h 14m 33s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4d66174cb776483ba9be8b353b2ca076 : SUCCESS in 1h 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b056902a1252c36b7679c9400533fe2d5303e4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 22:26:57.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"5202c2154fa1e32ad7d87468c87b2473fdd2a049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 00:52:16.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/19b43b74b191432c9005bbe5dbb2ee07 : SUCCESS in 2m 38s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/06cd52a991f74173afdf2a8e1542890a : SUCCESS in 46m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ddb257dd294c4472aaaeb651695b1ad9 : SUCCESS in 1h 04m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/883313b016324cadb2bfccbad4be6b27 : SUCCESS in 1h 17m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/215bb85e11b0468983caa7d679d2135b : SUCCESS in 49m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56ebd2baba2343a3b8ea683bff1f6cb8 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f40cf090a74105870e85045dd1b7b2 : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5300a2b13c44f12902a160a307b8737 : SUCCESS in 19m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a255a34b284143b09a7d5c9233b15ff3 : SUCCESS in 19m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28f73141c504916b3a5ef56b4e81204 : SUCCESS in 8m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/91ac9597952f4645bb3545528bb62140 : SUCCESS in 2m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f9400f2cdf45486398816c5b03ab98e7 : SUCCESS in 56m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a9dc5be585654590a8993721314d9455 : SUCCESS in 1h 29m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bea1083fe6f4750b509ae4f58da73d6 : SUCCESS in 17m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5224a8554ce240b895d4d740acc1124a : SUCCESS in 52m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9f6681f052c744ebba926af513b8d409 : SUCCESS in 1h 13m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/df201b5954034da1bcf7a8217d884f36 : SUCCESS in 1h 21m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f275ed37abc44e86b687bb67b87c0384 : SUCCESS in 1h 32m 15s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4fa8335192fe4444a571eee2a64757a1 : SUCCESS in 1h 02m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2eebdd378ef43ea95b872e37027e148 : SUCCESS in 1h 19m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f8edfa3bf3884b99bcff4ef264c90f3b : SUCCESS in 1h 28m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/38069b71af75431d902c8eb1abf75642 : SUCCESS in 58m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7bc8fcaed246471d88b20196f36f47b3 : SUCCESS in 2h 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e054051855864006b230f933e921dfbf : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":12},{"id":"21d482e8eec688c6afa5837e3155f69bcea856c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 00:52:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"e22786182b55ab730015bf06b0727c372d3495dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 00:54:16.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d032862ea9dc4b7db3d52ce18d6dfebf : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4e390892b27445288a7a6ecdcf6eb5e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"684246526092d780cb662910a018f9ececba9452","revisions":{"ade2f535b42f27431dd90fd6ea61898669495db4":{"kind":"REWORK","_number":1,"created":"2019-10-04 20:19:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/1"}}},"commit":{"parents":[{"commit":"4c41e65415e9d20861e6652c9a4252cd4ef0796d","subject":"Merge \"Fix designate driver not being fully region aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c41e65415e9d20861e6652c9a4252cd4ef0796d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ade2f535b42f27431dd90fd6ea61898669495db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ade2f535b42f27431dd90fd6ea61898669495db4"}]},"branch":"refs/heads/master"},"d6a5e25ade30bc7bb123957573613cf9cb889da1":{"kind":"REWORK","_number":2,"created":"2019-10-07 14:37:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/2"}}},"commit":{"parents":[{"commit":"4c41e65415e9d20861e6652c9a4252cd4ef0796d","subject":"Merge \"Fix designate driver not being fully region aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c41e65415e9d20861e6652c9a4252cd4ef0796d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-07 14:37:19.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6a5e25ade30bc7bb123957573613cf9cb889da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6a5e25ade30bc7bb123957573613cf9cb889da1"}]},"branch":"refs/heads/master"},"fa4205114ddb22c36e86edf1975337e1d0c5295d":{"kind":"REWORK","_number":3,"created":"2019-10-08 12:29:59.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/3"}}},"commit":{"parents":[{"commit":"4c41e65415e9d20861e6652c9a4252cd4ef0796d","subject":"Merge \"Fix designate driver not being fully region aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c41e65415e9d20861e6652c9a4252cd4ef0796d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-08 12:29:17.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\npy27 unittest tox target is kept because the job is defined\nglobally for all projects. We can remove the py27 tox target\nonce the job is removed.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa4205114ddb22c36e86edf1975337e1d0c5295d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa4205114ddb22c36e86edf1975337e1d0c5295d"}]},"branch":"refs/heads/master"},"e0e863294fb86223cdfd8397fa1869132b657020":{"kind":"REWORK","_number":4,"created":"2019-10-10 14:06:06.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/4"}}},"commit":{"parents":[{"commit":"4c41e65415e9d20861e6652c9a4252cd4ef0796d","subject":"Merge \"Fix designate driver not being fully region aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c41e65415e9d20861e6652c9a4252cd4ef0796d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-10 14:05:35.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\npy27 unittest tox target is kept because the job is defined\nglobally for all projects. We can remove the py27 tox target\nonce the job is removed.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0e863294fb86223cdfd8397fa1869132b657020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0e863294fb86223cdfd8397fa1869132b657020"}]},"branch":"refs/heads/master"},"080f14ff8c1372b5061f1f40fa68dadccea5e07d":{"kind":"REWORK","_number":5,"created":"2019-10-11 19:52:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/5"}}},"commit":{"parents":[{"commit":"4c41e65415e9d20861e6652c9a4252cd4ef0796d","subject":"Merge \"Fix designate driver not being fully region aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c41e65415e9d20861e6652c9a4252cd4ef0796d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 19:51:05.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/080f14ff8c1372b5061f1f40fa68dadccea5e07d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/080f14ff8c1372b5061f1f40fa68dadccea5e07d"}]},"branch":"refs/heads/master"},"051119ac398a8d156a5f5eff1e04e70da85d7f3d":{"kind":"REWORK","_number":6,"created":"2019-10-11 21:19:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/6"}}},"commit":{"parents":[{"commit":"4c41e65415e9d20861e6652c9a4252cd4ef0796d","subject":"Merge \"Fix designate driver not being fully region aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c41e65415e9d20861e6652c9a4252cd4ef0796d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-11 21:19:13.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/051119ac398a8d156a5f5eff1e04e70da85d7f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/051119ac398a8d156a5f5eff1e04e70da85d7f3d"}]},"branch":"refs/heads/master"},"5d7b7cf55b4ee27145bf1eb28e1072567605c230":{"kind":"REWORK","_number":7,"created":"2019-10-14 19:47:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/7"}}},"commit":{"parents":[{"commit":"715001491de4297d2ab3fd0a89f48999df15b6c9","subject":"Merge \"Randomize BaseFullStackTestCase._find_available_ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/715001491de4297d2ab3fd0a89f48999df15b6c9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-14 19:46:55.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d7b7cf55b4ee27145bf1eb28e1072567605c230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d7b7cf55b4ee27145bf1eb28e1072567605c230"}]},"branch":"refs/heads/master"},"809c372a8b21a0f4c2c8b04fa558c08bd57a22b9":{"kind":"REWORK","_number":8,"created":"2019-10-15 18:17:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/8"}}},"commit":{"parents":[{"commit":"715001491de4297d2ab3fd0a89f48999df15b6c9","subject":"Merge \"Randomize BaseFullStackTestCase._find_available_ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/715001491de4297d2ab3fd0a89f48999df15b6c9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-15 18:16:52.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/809c372a8b21a0f4c2c8b04fa558c08bd57a22b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/809c372a8b21a0f4c2c8b04fa558c08bd57a22b9"}]},"branch":"refs/heads/master"},"72b12666dd2cc026049e5f90d9c1d2b5044d28cd":{"kind":"REWORK","_number":9,"created":"2019-10-16 18:42:33.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/9"}}},"commit":{"parents":[{"commit":"715001491de4297d2ab3fd0a89f48999df15b6c9","subject":"Merge \"Randomize BaseFullStackTestCase._find_available_ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/715001491de4297d2ab3fd0a89f48999df15b6c9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-16 18:42:17.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72b12666dd2cc026049e5f90d9c1d2b5044d28cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72b12666dd2cc026049e5f90d9c1d2b5044d28cd"}]},"branch":"refs/heads/master"},"b704d0197d64897c37e01062daa7eaf9a8297504":{"kind":"REWORK","_number":10,"created":"2019-10-25 11:17:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/686833/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/10"}}},"commit":{"parents":[{"commit":"f5a827c2be06f24a1f8025f120f16c12eb1b1f55","subject":"Merge \"Temporary disable CI job neutron-functional-python27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5a827c2be06f24a1f8025f120f16c12eb1b1f55"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-25 11:17:19.000000000","tz":0},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b704d0197d64897c37e01062daa7eaf9a8297504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b704d0197d64897c37e01062daa7eaf9a8297504"}]},"branch":"refs/heads/master"},"0ed8764b863eb863026c9b3aebe7a2d3cbd37083":{"kind":"REWORK","_number":11,"created":"2019-10-25 14:57:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/11"}}},"commit":{"parents":[{"commit":"f5a827c2be06f24a1f8025f120f16c12eb1b1f55","subject":"Merge \"Temporary disable CI job neutron-functional-python27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5a827c2be06f24a1f8025f120f16c12eb1b1f55"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-25 14:57:11.000000000","tz":-240},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ed8764b863eb863026c9b3aebe7a2d3cbd37083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ed8764b863eb863026c9b3aebe7a2d3cbd37083"}]},"branch":"refs/heads/master"},"684246526092d780cb662910a018f9ececba9452":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-10-25 18:50:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/33/686833/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/686833/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/686833/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/686833/12"}}},"commit":{"parents":[{"commit":"8acf9ea32baad3b4fecfb973c17f81e5a3b7b5d9","subject":"Merge \"Reset timeout exception in DietTestCase when retrying\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8acf9ea32baad3b4fecfb973c17f81e5a3b7b5d9"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-10-04 20:14:37.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2019-10-25 18:50:08.000000000","tz":0},"subject":"Stop testing python 2","message":"Stop testing python 2\n\nSince it\u0027s no longer supported past Train, lets stop\nrunning the tests.\n\nUpdated docs and made some pep8 code tweaks as well.\n\nChange-Id: I1c171ab906a3b4c66558163ad26947ebf710a276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/684246526092d780cb662910a018f9ececba9452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/684246526092d780cb662910a018f9ececba9452"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
