)]}'
{"id":"openstack%2Fneutron~687516","triplet_id":"openstack%2Fneutron~stable%2Fstein~I2e690ea0cf5fa0614e39be2b0e83afad3daa7f48","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1844124","hashtags":[],"change_id":"I2e690ea0cf5fa0614e39be2b0e83afad3daa7f48","subject":"fixed_configured\u003dTrue when Add/Remove port IPs","status":"MERGED","created":"2019-10-09 09:36:45.000000000","updated":"2019-10-09 16:37:16.000000000","submitted":"2019-10-09 16:35:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687516-1570638914646-94a06386","meta_rev_id":"47c1444d2f21d77342648473fcc9df08c9982db5","_number":687516,"virtual_id_number":687516,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2019-10-09 11:25:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 16:35:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-09 11:11:29.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-10-09 09:37:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-09 13:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-09 12:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-09 13:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 09:37:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-09 11:11:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-09 11:25:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 12:28:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-09 13:11:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-09 16:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69ed69ecc80d41b2396f9d6dcbe7e4f4efa4b2c5","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-10-09 09:36:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"6de0ce498430b92414e2a6aeac30bf4a13f4765f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-10-09 09:36:57.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 885351c825e32d574d2b817961f6765bbdba673f","accounts_in_message":[],"_revision_number":1},{"id":"691ccda9e5e794876ff37bb3f87a60132157997e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-09 09:37:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"34f81bcc3975b0e0453b630556e4cb33ddb37a2a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-10-09 09:38:19.000000000","message":"Topic changed from bug/1844124-stable/train-stable/stein to bug/1844124","accounts_in_message":[],"_revision_number":1},{"id":"6efc8d507c9d7dbf9a371c69f13c45dd0e8545c4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-09 11:11:29.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687516/1 : FAILURE in 1h 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5b8b77b42e34a9109069675601ec3e0c58ca4f7f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-09 11:25:15.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/16/687516/1/check-mlnx/Neutron-ML2-Sriov/95be2f1 : FAILURE in 10m 45s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/16/687516/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/0186326 : FAILURE in 9m 26s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"cf2727b342c90d15c743685f25db85273b85c1c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:56:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/dfa2ddac07494b26a3f3d67e8782d1ca : SUCCESS in 52m 39s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/e29d8126da5749279ec42c4f6c33d951 : SUCCESS in 43m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/0c3a45c8d61542238e79432975b12149 : FAILURE in 1h 31m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/410f6f28760e45e38357026e521c0451 : SUCCESS in 1h 12m 12s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/33f669ee22d54b169cad611edc401878 : SUCCESS in 1h 12m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/f5778e40e1a94f128af3e7995837d8b7 : SUCCESS in 1h 02m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/624e69428c81445c887af4d9254ffac3 : SUCCESS in 28m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a898fcbc87d45688a8f8be4a3da0fea : SUCCESS in 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76fd20e40ebe4538ae73f949c910adfd : SUCCESS in 13m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f76d9841604f4ed98930ff4eef84d260 : SUCCESS in 22m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/27b149a7e7c54371aaed5e0ab0c40709 : SUCCESS in 30m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6695f5729ba74da6b1dcb74bbaf88e79 : SUCCESS in 21m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9567dff40b84ec1a327110adc9e682e : SUCCESS in 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322824e174d841b48c8e3bf3553c8943 : SUCCESS in 7m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b0c0fcee8cf4473fbfc31248baccb15a : SUCCESS in 1h 27m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/96aae99421be45d9aa8a7b1cc1edbfcf : SUCCESS in 55m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0ede43eb64e04961bb020b3b2bfa9a0f : SUCCESS in 56m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce9712191f0c42f985a66f93eeffc896 : SUCCESS in 1h 45m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1f7bc99403e47b4b43410bdb10145b8 : SUCCESS in 24m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a6a4881cc5e34556a70bea79f9593b0d : SUCCESS in 46m 17s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3c097588e5a543569521e4b20fc83cb6 : SUCCESS in 48m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2d70fdce5acc475893ff4e2d2816c1c2 : SUCCESS in 1h 06m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/273b630d4ca94a7c92f6005001ae4180 : SUCCESS in 1h 38m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/26bdb6c4b12349d18b47a1e63e1c1db3 : SUCCESS in 1h 34m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d7d05a919e1c47c59206d77b43fdbf63 : SUCCESS in 1h 22m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fadea70877246b08e916eca74512fd7 : SUCCESS in 1h 09m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f38a20425ef243e299b225a57c12f287 : SUCCESS in 1h 14m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/59e4bf9b986e461db3e6254467e3e0e8 : SUCCESS in 1h 36m 47s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2aeaa340b0564b6481d4c9519dc3aab2 : SUCCESS in 1h 48m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05d744f1e3a444418952dab1ee3a95b5 : SUCCESS in 1h 20m 22s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/70130e7898e847d5816d7b41286d7112 : FAILURE in 1h 38m 17s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d7702e771457412da549c252ec7f848b : SUCCESS in 50m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad5d90f2a7bc49539f907675d856e2ae : SUCCESS in 2h 13m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d8f431f75f5f43aab2f166f6dd46472e : SUCCESS in 1h 52m 57s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6421143f3e1641a49a1ee8df20d736cc : FAILURE in 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4deedebac0c31b837c770964edec08219459366","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-09 12:28:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4e34c453a45788951f97e1b314797daacb41a4a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-09 13:11:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27e93f812f9e390a603828891710ebbe481f81b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 13:11:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eea7dba2b20ad5277ccc3ca7927580793db1b935","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:35:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5050db02b942c9455d264825882e4273fac83cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:35:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/8c1ad1610f654664a1b1fa1d9bf70c35 : SUCCESS in 50m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e9e257200344403bf2f5985dd833467 : SUCCESS in 25m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012fb5bbb8f84ed6b9ce2935e857dede : SUCCESS in 11m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c1d260221404cd3a3ae94c1ae84aa71 : SUCCESS in 19m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bdc1f885c5b7446884fe3e35e028636c : SUCCESS in 27m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67668d76b7134c6891d05ffec36070f1 : SUCCESS in 25m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/826bbfbd700144e8b1a19b9de3cbe234 : SUCCESS in 18m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee76c1055514bd58aa0f1f5baf22f40 : SUCCESS in 5m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/90c3a9a68aaf4b99940faaa8415c21c0 : SUCCESS in 1h 41m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/72cf74b3dbfb41c1a44b309b18fe1796 : SUCCESS in 53m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7191b3b16a08467aa4042ac5b59f857d : SUCCESS in 1h 02m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f7bdb53c21e44ac899f5fca0499c967 : SUCCESS in 1h 14m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6d7cb5e61a43689fe94b3ed6429edb : SUCCESS in 21m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a31c3f928f0a4fd89ec2b9968925aa74 : SUCCESS in 43m 06s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/470f9fa6e4fa45e3b14717eafa39dcf6 : SUCCESS in 43m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bec5df369210420b91c3ebd22e64996f : SUCCESS in 1h 03m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/10ded3c5d63648b0b91313414b4c35f3 : SUCCESS in 1h 19m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c2531c2a8bc84e8eae960963f9f3e568 : SUCCESS in 1h 20m 37s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82de96df58e2499d8f4bddb4c07f927c : SUCCESS in 1h 22m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47b4b60cbc79434cba648e328ecffd91 : SUCCESS in 1h 11m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a442ce739e974902a44d51882ac3ff96 : SUCCESS in 1h 11m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26e71b4524bd470f9272eef662bae057 : SUCCESS in 2h 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"47c1444d2f21d77342648473fcc9df08c9982db5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 16:37:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a47fb27369451fb16c1e8f68dea5a2 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73cb2337eb9b455a955ad8497147b48c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"074e1832aae335d261ea938126256f1b672c1572","revisions":{"074e1832aae335d261ea938126256f1b672c1572":{"kind":"REWORK","_number":1,"created":"2019-10-09 09:36:45.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/16/687516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/687516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/687516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/687516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/687516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/687516/1"}}},"commit":{"parents":[{"commit":"9d04e38e8679c3f87c364bc6c9e39b43dfb959b9","subject":"OVS flows for custom ethertypes must be on EGRESS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d04e38e8679c3f87c364bc6c9e39b43dfb959b9"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-16 19:52:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-10-09 09:36:45.000000000","tz":0},"subject":"fixed_configured\u003dTrue when Add/Remove port IPs","message":"fixed_configured\u003dTrue when Add/Remove port IPs\n\nWhen updating a port with the fixed_ips request the\nfixed_configured argument should be set to true when\ncalling _ipam_get_subnets() so that all subnets are\nreturned if host is not set.\n\nOtherwise the ip allocation will be deffered and an\nempty list of possible subnets for the port is\nreturned. Which in turn led to raising an error that\nthe network requires subnets to allocate an IP\naddress.\n\nCloses-Bug: #1844124\nChange-Id: I2e690ea0cf5fa0614e39be2b0e83afad3daa7f48\n(cherry picked from commit def8e95aad1e4588c369d537ee66234245eefdf6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/074e1832aae335d261ea938126256f1b672c1572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/074e1832aae335d261ea938126256f1b672c1572"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
