)]}'
{"id":"openstack%2Fneutron~687922","triplet_id":"openstack%2Fneutron~master~Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8","project":"openstack/neutron","branch":"master","topic":"bug/1845176","hashtags":[],"change_id":"Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8","subject":"Add OVS QoS driver cache for minimum bandwidth rules","status":"MERGED","created":"2019-10-10 14:44:19.000000000","updated":"2020-01-22 05:39:01.000000000","submitted":"2020-01-22 05:36:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687922-1579671384334-9ea6cd4b","meta_rev_id":"522594902d0e1e690fdf647f8436bb2c8eb8b14d","_number":687922,"virtual_id_number":687922,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-21 21:48:36.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-07 18:02:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 05:36:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"date":"2020-01-21 22:17:16.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-20 01:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-01-08 09:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-01-21 07:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-20 01:30:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 17:52:11.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-10-22 15:10:56.000000000","updated_by":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"reviewer":{"_account_id":18051,"name":"David Shaughnessy","email":"david.shaughnessy@intel.com","username":"davidsha"},"state":"REVIEWER"},{"updated":"2019-10-30 09:46:51.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-06 10:48:23.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-07 18:02:09.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-08 09:17:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-20 01:30:31.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-21 07:55:17.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-01-21 22:17:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-22 05:36:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba8c71cbde3fc8dca71e654a99e7fffc948ab9b9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-10 14:44:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fe6b2ae2f41d32ef6f76a9bfc4be201434729b6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-10 14:44:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7e267b854b68944eb2ab388462f7fbc25d2dd604","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-10 15:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-10/687922/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/d41ff30 : FAILURE in 53m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6a2b2134b58c7a33aeeb892b8d5dfca6ff3649f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-10 16:44:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/1 : SUCCESS in 1h 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5e4d98e658453c2deacdb0e4f71c2b8714b498dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 18:47:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f20c6faea1424fe4954e8a6eb49e5734 : SUCCESS in 53m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77a7c419703b4ae08ca1b50978ff3ca9 : SUCCESS in 38m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/231bf14b5e2f43d39423ce423815d8a4 : FAILURE in 1h 27m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/559dc780b8794129ae470afa6ebaf4c7 : SUCCESS in 1h 00m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f43b75bc9f334ba980cf1dd64d25cdc6 : SUCCESS in 1h 06m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7992d63822e74f72bbd0141a5bf30271 : SUCCESS in 1h 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/546c623a34154647b2a1cd55a9fae2a2 : SUCCESS in 31m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f11501603cb433e8a2581e3f1695b3b : SUCCESS in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d309e139984d6fa29d96a6ef5f9c41 : FAILURE in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19ef4d3fc0bd45e7b0a6879ee713879d : SUCCESS in 27m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7e74e25fc9f451b8f1bee5de4d49b38 : SUCCESS in 19m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f785366b35d4c7e807eafdc236a4283 : SUCCESS in 28m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcab2bd2ad0d4a0296cbf256c00890c7 : SUCCESS in 7m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bffd52bf00254a32b92126149dada5ff : SUCCESS in 1h 07m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01eb8acfbc204c89a72f599b86c45e2a : SUCCESS in 1h 22m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/46f408e01e97497ca3f2377de61c0323 : FAILURE in 44m 06s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/de10ba767a1b4519a2f06d8e7141db65 : FAILURE in 54m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/37a1ac49f23b4c6bb185138a90f5a933 : SUCCESS in 1h 13m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/41515f74e9954ebbb08e3d96810330f0 : SUCCESS in 2h 08m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6443c221f04f49e788670c3153331774 : SUCCESS in 1h 37m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1db0441fd9064467b90d6a3b0d584b58 : SUCCESS in 1h 19m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ece9b3c93e154f4ca4b901677ad5c420 : SUCCESS in 1h 16m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b536aea6c7df46088c4313c7f56e6ca3 : SUCCESS in 1h 57m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/617c58d0ee4642ccbeb9d3c3378800e0 : FAILURE in 1h 37m 00s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d133400265d548b1824438277495139f : SUCCESS in 1h 24m 05s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/08e161ba22ac4b9aa263a2f3f6d177b7 : SUCCESS in 1h 21m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b6872c1bb46e4ff0b06d9faf0159bd0c : FAILURE in 1h 11m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7e62ddf24b064ba7a53a66c6ca44d3dd : SUCCESS in 53m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0e1e8b3ac494e49bc6242cad69efec4 : SUCCESS in 2h 24m 00s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d83881bf4c9c4e9eab627580e3b21308 : SUCCESS in 2h 26m 27s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/25944126307b49898ff623073549e511 : FAILURE in 1h 51m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a06a888b54049aeb30960470cfe044b : SUCCESS in 1h 05m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ef8eaf54c6664c99b4f33311e2b5d2a4 : SUCCESS in 1h 32m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1ecb98ef6364a139aa89f513827711e : FAILURE in 47m 57s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28f3837cbe224199977f8192333610df : SUCCESS in 57m 56s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/254421e48efb49edb0b2894e0a7423ae : SUCCESS in 1h 22m 09s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b79d096c3a16416e8e63e6fe9277011b : SUCCESS in 2h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85255a70955d61667cd40dc080be213567d2e498","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-11 01:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/1/check-mlnx/Neutron-ML2-Sriov/ece2b44 : SUCCESS in 1h 21m 12s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/e6ed02f : SUCCESS in 1h 02m 35s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"83d54c13bb646509a0b9ce31c03e33385dd5e2c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-11 11:11:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bef903c80f3da8f6963cae3e5b35612f52d4574","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 11:45:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-11/687922/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ea4d131 : FAILURE in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be27389d82c2ee75c31aad8e74f15fdd1e1dd763","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-11 12:50:27.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/2 : FAILURE in 1h 38m 26s","accounts_in_message":[],"_revision_number":2},{"id":"80119e84548cc5a377d3b4f852e79efb44ddcb42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 13:47:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ce5ce3ad8e4f47d4bc4a430fa02101bf : SUCCESS in 48m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f5bc6516da4147f6b211bc202de75d41 : SUCCESS in 27m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7fcdcc6b1d5243e2a36c42268d37b71b : SUCCESS in 1h 28m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4d2705e3eb0e460aaea92585367c0bb2 : SUCCESS in 1h 07m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4c802cba6dfe450e9d751b5ffb8f4148 : SUCCESS in 46m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00dd801e57d240a9aaea47b2fb30004f : SUCCESS in 56m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1945d85a7549459546b906d60d5786 : SUCCESS in 35m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ded8094c696244d988a3460d50c869d7 : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a77a50e5a04b24a2e92d4e0edba96a : FAILURE in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69ed18c844144124b2034a3398a8cd63 : SUCCESS in 21m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab190c36b6084450bd133f29932c930b : SUCCESS in 30m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5252915a277541eebd64835541f2fcf5 : SUCCESS in 30m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780d8d85fec746b5a5956d2ea4d2bed3 : SUCCESS in 9m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0e616b0a4d324fa1b04d246ff9261ee2 : SUCCESS in 1h 14m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b363db5efff45669f128423a19c219c : SUCCESS in 1h 26m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/81bee0ca12d041279e18dbfc45325364 : SUCCESS in 48m 05s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9f173cd659f54a7c9d271e404e576a05 : SUCCESS in 52m 00s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/86780e8c466a4a44acebc51b09f6b52a : SUCCESS in 58m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0584b43945a64a7b8debe103f0e3b3f8 : SUCCESS in 2h 28m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d3006867349c4c5a9fda3349a340b998 : SUCCESS in 1h 26m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3ed87348d4940008eebebb85cd675ed : SUCCESS in 1h 22m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6ffd3675867424e9445cc8e314ae174 : SUCCESS in 1h 26m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e14a1df9a6044cb5abfbf743478defe2 : SUCCESS in 1h 43m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dbbdd65005644d99939526490fd4e0e2 : SUCCESS in 1h 49m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/75086341a5c2470a8a52ac7c57e8ac0e : SUCCESS in 1h 18m 10s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/8ca5a1ff72b943089661aa7208cf0947 : SUCCESS in 1h 39m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f2a7316d1c0a4238960f3e55b44de8f8 : SUCCESS in 59m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3949169e8ac74386ae0beff2d97c0958 : SUCCESS in 51m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/407dcfb579714b2ca371889beb06b04a : SUCCESS in 2h 24m 30s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3ea6e23aa4b84490acebdddb6a183868 : SUCCESS in 2h 14m 57s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2604f4c1879b4c3f8dbeaec01237c8e3 : FAILURE in 1h 34m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6db0d4dc50e43d2864c52d60530775e : SUCCESS in 1h 00m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ee2ad0ec3a2742e2aaea8b1d87eb9792 : SUCCESS in 1h 19m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b0e19b6d2dd4c1bb04ac67e992d3066 : SUCCESS in 49m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/223145a038d24a28a2feda1b702e6533 : SUCCESS in 1h 05m 11s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cb36859037746fc8ff4d467aaee87e8 : SUCCESS in 1h 11m 40s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a5012ccd138647b5ab26a5995674db70 : FAILURE in 2h 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9dc553f9506ca87b6685ed0691c882e1d3a89310","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-11 15:34:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"47418b9560d56ee8d3b46ecb200b3d8bcbd44d5e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-11 16:09:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-11/687922/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/596daef : FAILURE in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"966f6b520d55b9ff9b96b8f71c5a048f2121595b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-11 17:15:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/3 : SUCCESS in 1h 39m 57s","accounts_in_message":[],"_revision_number":3},{"id":"acc63e4559b8c2fae2973a8a7c8fdd6d891cb85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 18:17:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f8e4b733ba44dff8f865aacd402c259 : SUCCESS in 54m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2bb54e9200f44ff1b95468b32c243212 : SUCCESS in 35m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/20287c0a77a44bd7abfbd6401159d3ea : FAILURE in 1h 50m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2915dde80887468fbe0774c588fdadf1 : SUCCESS in 1h 01m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7f1794ea120e4e8689f7002638ee5e7f : FAILURE in 1h 16m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36a44ae80d9a49698093b174e5559741 : SUCCESS in 1h 28m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcb21465c3a041fc8c6c1cba157c2dbe : SUCCESS in 26m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/943c3f89519c44a58e0b5c86ad4efff8 : SUCCESS in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43d296633c047beb3cd58dce1152610 : SUCCESS in 13m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/479d64aa066544f09e06bd9a3596fe1a : SUCCESS in 27m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f21e8d9787e41a78db613fb488eabe1 : SUCCESS in 19m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97c0f318b6994c09a189318b822214c7 : SUCCESS in 20m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c735d4488c1942289e8ebb45db40fee8 : SUCCESS in 10m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2e58559dedfb454d91713bda9ae986d6 : FAILURE in 46m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/94e023edced04ac4ad6175c4edd28b0c : SUCCESS in 1h 22m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/78070a516776444f94d2ca3794a341d1 : SUCCESS in 49m 50s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e21b8b5bce1d47c280c2b5b2a24222c0 : SUCCESS in 47m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/006f3b2692e644e58016eba8ea4d9ab7 : SUCCESS in 1h 04m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1909b71b5d864d7bb5d77236ef598567 : SUCCESS in 2h 19m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8b957d0f3ddb4c3b8833cafe6f271704 : SUCCESS in 1h 16m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fbcabde77dc54f34b3edf5e1dc581796 : SUCCESS in 1h 32m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/885f29af862c440d8576b244fcaa2b37 : SUCCESS in 1h 09m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/54907ab00f01479a9eb5870153c582cc : SUCCESS in 1h 09m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5cfdbdc212804b8a8c467ec383cd7f0e : SUCCESS in 1h 58m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e9b8656377cc43b39768465f3c1e13f3 : SUCCESS in 2h 18m 15s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/000f9f6de23a47fa961bab7e244f0103 : SUCCESS in 1h 34m 44s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/e1bc9d02534549528f83476e5e7f54a7 : SUCCESS in 1h 31m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/424a8d5a88464b7a9f878e4ddc8c7746 : SUCCESS in 1h 15m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cebea69f2d7241a7bbb36bbbb5dd0a7d : SUCCESS in 1h 02m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b93f34b045fc4d35a10e01fbaf4be506 : SUCCESS in 2h 19m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b4fcd77f1d2d4176aa7457b8848ba4ce : SUCCESS in 2h 04m 55s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/04629def429e4969b630e1ef43e839e9 : FAILURE in 1h 42m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c34d3706048e4900a8a99fb4d0135c2f : SUCCESS in 1h 09m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e5afd260b1f147eb8620081c1a217cbc : SUCCESS in 1h 28m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85a856339da445888811b9664b8679b4 : SUCCESS in 45m 00s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8aa41586ab3c443592674ebb3fd1acce : SUCCESS in 1h 02m 45s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/56ef8951851745418399d678d2eaf332 : SUCCESS in 1h 17m 05s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a7d01faaba0e42b4a6dc65dd9b58e91e : SUCCESS in 2h 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58a7d48386d7782f84f09cae74b9b7d39812701b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-12 00:37:18.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/3/check-mlnx/Neutron-ML2-Sriov/f85b257 : SUCCESS in 1h 20m 52s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/3/check-mlnx/Neutron-MACVTAP-ML2-Sriov/eed4066 : FAILURE in 1h 30m 06s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"278203ffbcb367f3e03fa6be46ea39a478bcfff6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-12 11:26:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"34c24bde6649486363ecc210fa56f91c034df348","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-12 12:13:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-12/687922/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/37d21e8 : FAILURE in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc645c06b0592c2ab495c34ffd4430d262dfdd9d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-12 13:03:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/3 : SUCCESS in 1h 36m 15s","accounts_in_message":[],"_revision_number":3},{"id":"589dd9377a4e81aa05d7ef265ca051d80c11a0b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-12 14:54:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2b5ae821fbc34a989ec18e6b8201eeeb : SUCCESS in 47m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c118a7b679004172b23f5ab83180ade5 : SUCCESS in 34m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/732ce5f528534f5aae4cb6467165035b : FAILURE in 1h 22m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/420f2added504f84b91ec761c4d42815 : SUCCESS in 1h 03m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/654ee548c1034f5cb0bab31302bd859a : SUCCESS in 54m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30394ef08f3448e5a74512479847590e : SUCCESS in 1h 17m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddcef22977b14fd985112102f674a98e : SUCCESS in 29m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ec0a9788d794861b1e0ed6580f07c1d : SUCCESS in 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d95eed7ed947c789423b0baf4d8cdf : SUCCESS in 9m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c558d8bf8a384a28b2039e539d2fdcad : SUCCESS in 25m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f7c12817f7c4c6eaec34b48bfb9fa50 : SUCCESS in 34m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aedb82b9fd0940c7ba459c45ed881e80 : SUCCESS in 33m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705d57ccd4fb4413b1f8051fd2b88065 : SUCCESS in 7m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1c5dfd5788514f97821eaec8d4321df3 : SUCCESS in 1h 06m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ffa8526591f4aaf9f76aa85c58cab75 : SUCCESS in 1h 11m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e19b887d75f4f85a5cf81f6c8da2c3a : SUCCESS in 38m 03s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/6089b83c9f524a95bb990103e0497abd : SUCCESS in 42m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aac91d896adb41879c7272032545f3a4 : SUCCESS in 1h 03m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/53861966b89c4d94b76ad228f6466da3 : FAILURE in 44m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/02e332d848b14a97a668be95766deaf4 : SUCCESS in 1h 30m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d06a3e6dda246ec88094e7b1de46243 : SUCCESS in 1h 33m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3594d4467af4d8fa372ea0eaeeebea2 : SUCCESS in 58m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/edb159e3e11745c88914d41b5ecdc4e4 : SUCCESS in 1h 35m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/12f0b27af49944d78490ab23b493cb68 : SUCCESS in 1h 30m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a0ebbc779ced4fd3abf4358863fcfdff : FAILURE in 45m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f4c83673b7d646a8971c5fcc8dbe3ecd : SUCCESS in 1h 32m 55s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/a448803a88924e9cb3333d82d14e3e4c : FAILURE in 7m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eefd46ce919641609a13460b030b60e6 : FAILURE in 42m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3e2760dd8d234a23bbea3f84c1859115 : SUCCESS in 1h 15m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac63f0a35acc4387b38ddb0b99060d9a : SUCCESS in 2h 36m 01s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fef8243519b147c7a841ae14cf5206ee : SUCCESS in 2h 51m 23s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/15fd8ac9f856458b876e431326e28098 : FAILURE in 2h 03m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/027d91b488cd4b2a85a3ece36559befe : SUCCESS in 1h 28m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cb1f480a76694e4dbf560afc94fc18f7 : SUCCESS in 1h 31m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4036cc1058da48298994da337df31ae5 : SUCCESS in 56m 22s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0eb16f9416674a18bc8ba11cad315911 : SUCCESS in 1h 07m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6540690d25ee4b2598886a5a888fa82d : SUCCESS in 1h 51m 58s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/df4b806262af40fbbc28770162b34e79 : FAILURE in 2h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3909a85258cbedb472a108110319eb8c9ee6ab84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-13 08:50:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ad5d7fe9058c285208d48c60ea2a7a7b6c27e131","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-13 09:23:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-13/687922/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/75c6c88 : FAILURE in 31m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d9af412f9ecd059a763feae5e2ba2e44dbf1590","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-13 10:28:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/3 : SUCCESS in 1h 37m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f0433217a4c4540a2520290283e4387cd3c91d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-13 11:17:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e74517cbcd674469b1472d39c2ac61ff : SUCCESS in 52m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f67ce7d67f394d8ea6b5c80731132aa3 : SUCCESS in 33m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5434a1883aa44c459a85d66a2ca821d7 : SUCCESS in 1h 21m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/59aea2637a664b22aae24e82d41f0876 : SUCCESS in 1h 02m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c08b1bab86334518a805b76d5977ba53 : SUCCESS in 1h 07m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69e81bbb14cb4d4d91dd597ed722b332 : SUCCESS in 54m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd2bb0f9222448e0bf0956f33b3a715f : SUCCESS in 28m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a87e8dc43f224aaeba0e0baed520cc4b : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2dfeb30d3e349a689c3f6ae5859e9a3 : SUCCESS in 12m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62b4877573f145a89e460e8f30b55aa0 : SUCCESS in 21m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d84cfd685d4bf990e7f585199903e1 : SUCCESS in 21m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aaea3f5265224c788c3b8543fe34a701 : SUCCESS in 21m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a4b63d334c4b7095f6a9abe0fb5879 : SUCCESS in 7m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5d015085c3674c859128604ff1f603bc : SUCCESS in 1h 08m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34259f8b487d4677a05400b1aa7943e7 : SUCCESS in 1h 10m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f315801ea6b149ba8664aebbd086efff : SUCCESS in 48m 03s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/54e75f4be9d94feaabb5e850f506f8f7 : SUCCESS in 50m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cf5af7a259b8418aa85c4e7f2ea2d543 : SUCCESS in 57m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a0eb34d99c24584b74630c4b6f39e0a : SUCCESS in 1h 53m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/06034ddb4e2c43cbbdf38ca887193b71 : SUCCESS in 1h 08m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/68c6c115aeb1417fb340c753e44b929c : SUCCESS in 1h 28m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df41416f8d724492a0347cdd9d553c32 : SUCCESS in 1h 05m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8ccb41c8bca341f3aa2ae6728a86f326 : SUCCESS in 1h 06m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f429e68d9e304716904ed4eaf68763dd : SUCCESS in 1h 37m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f624b4332f884fdda81ab26a7cb975cd : SUCCESS in 1h 42m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9286a58a496e4edda800e80d0b0ece94 : SUCCESS in 1h 10m 03s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/2efe20b0bfb040e5afe439550f59bbab : SUCCESS in 1h 36m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/84e522156595467cad3c34333b627622 : SUCCESS in 56m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f46f8846101e4cd4909a134167bc6988 : SUCCESS in 1h 02m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/972734793346401da198eede229e329b : SUCCESS in 2h 10m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f3c6293e99f64143b987e4797435a12c : SUCCESS in 2h 20m 39s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/a9061f9f049642f5b14e93d49baf9c8d : FAILURE in 1h 25m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38f1758ef4d34bfc91c7da86b73f3f50 : SUCCESS in 57m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/697f6509764542388a1f77c4302295ae : SUCCESS in 1h 16m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c50283cc857c4bcc8edef96731717941 : SUCCESS in 45m 08s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0212ccee5b3647088356da6c57c57f3c : SUCCESS in 1h 00m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/79040ef5271049a2ac09186de04aeb4e : SUCCESS in 1h 06m 40s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/edfcaaccc37e4f75af05b9f16b2d5377 : SUCCESS in 2h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9f5f23365d63ac5369d4b68ba7ac501f2663d8c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-14 08:54:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1a266173f2126213a93054826c101887e7cc8376","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-14 18:01:30.000000000","message":"Patch Set 3: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3c1f33211ef350a806047af2aeb8ef215cfc80ec","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-15 17:11:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bfb16495bcd90248da44e3537905623244625fd9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 17:43:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-15/687922/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/7fdce48 : FAILURE in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"11f8b6f77438c7e244a58a8955642768cecf624e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-15 17:52:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nmedium-nit, still trying to digest the rest of the actual patch/issue.","accounts_in_message":[],"_revision_number":4},{"id":"ac85f42fa268483ad5787d4d4d8e50a0099ca3d2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-15 18:57:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/4 : SUCCESS in 1h 45m 59s","accounts_in_message":[],"_revision_number":4},{"id":"485e7a6227aed85bddf82486c355b2870bd79bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:58:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/570d09c372da487ea9bc8790552218ae : SUCCESS in 55m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f42c9b4a239b4747ace4211c89a1810b : SUCCESS in 36m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/663ee33893f54560b5d1c3e730b80dca : SUCCESS in 1h 23m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d584c42db8744cf2bd3470b6540fd057 : SUCCESS in 1h 25m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cc3b21a9d6d64e6fbf4a693ac6f30245 : SUCCESS in 1h 09m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e192740232194210ab49a9bb83b869b0 : SUCCESS in 57m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d10ef96396e04b0bbb00629989b076c8 : FAILURE in 28m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f79763a5cf54c9a9c0177fd3223dbf6 : FAILURE in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37195429158c4cee9d3de2ec4ba109b7 : SUCCESS in 11m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6e2dc13375a4fa9bfb125717a4ecdcc : FAILURE in 33m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b446e5b78484bb6824b2ce33d7cd8da : FAILURE in 21m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35003ea4645d4a14994c4131adf4826f : FAILURE in 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b03c71f95a46448cb5988a6323134a : SUCCESS in 10m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/47f355e7559b41578cc02a967061d411 : SUCCESS in 1h 07m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/23fef77bd572414f8813e12d18b7fbdf : SUCCESS in 1h 15m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c31c88c4ac0144ea8d479635ef899e9d : SUCCESS in 46m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/04941b2197264dec9c069f57d26439b5 : SUCCESS in 45m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/93cfbb1c7f50412e97afae579bf6ae17 : FAILURE in 59m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b176855951be487883800844ac6ace8e : SUCCESS in 1h 45m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9811bb29d6e14afab850106b835193c4 : SUCCESS in 1h 20m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4b9fb344d50d46609ccb8028946f14ff : SUCCESS in 1h 21m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df3fb3300c5d45f7a60f1bcd15cc4851 : SUCCESS in 1h 09m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c891585b72d64a52bec9110f9110309d : SUCCESS in 1h 15m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/10c8d73360324eeb899c807102ff1a6e : SUCCESS in 1h 40m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ba009000dba146828cf0bc0704398cf4 : SUCCESS in 1h 57m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e15eaa70f0f14cedb9d06ba5f45252e9 : SUCCESS in 1h 25m 01s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/48f8eeb7aed44dd2ac0ce90252b0ae60 : SUCCESS in 1h 48m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ac0b40c16f08482a884c1e661fbbb31f : SUCCESS in 1h 07m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0294e5fb5c00419ba68f2acb4313d2c6 : SUCCESS in 1h 03m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/09f7ac26672f4b1f887138bca2ef0766 : SUCCESS in 2h 22m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d477b308b2554372b50c390590daa3c5 : SUCCESS in 2h 01m 37s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/128bb68bed5948ab913f04f20af02732 : FAILURE in 1h 36m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56ab0050082f46ec914e75eb400d55c6 : SUCCESS in 59m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf63219ab822494dbc6a976ba7b6c27c : SUCCESS in 1h 42m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5db1184c42042b9a6e4e296192e94a3 : SUCCESS in 45m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64b525ebaca74ec89525a1d35facf0be : FAILURE in 1h 01m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e756324e0c040b9a8b53b79a215d5ac : SUCCESS in 1h 42m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/804ab976e21a41629c3bd272aadff316 : SUCCESS in 2h 47m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9fed2beaeb548b0b6214293cf28040a95666b0d1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-16 03:58:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/4/check-mlnx/Neutron-ML2-Sriov/9c8288c : SUCCESS in 1h 20m 52s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/4/check-mlnx/Neutron-MACVTAP-ML2-Sriov/d524cf7 : SUCCESS in 1h 15m 35s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"fca2fd0ec34025b14411d1fa2a391266cbf9b1f7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-16 06:52:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6debf1fbced65142be76842074e6deda774accbe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-16 07:44:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"48c4939994ed44cc5609c7a4a81af13c9b53a356","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-16 08:16:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-16/687922/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/0c8327d : FAILURE in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c6bfee75a9a6729f3c640c493c88951fd678a2ba","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-16 10:29:17.000000000","message":"Patch Set 5:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/5/check-mlnx/Neutron-ML2-Sriov/c8a4777 : SUCCESS in 1h 12m 26s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/5/check-mlnx/Neutron-MACVTAP-ML2-Sriov/8b41f30 : FAILURE in 1h 32m 49s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"4843722c4bec5d47a9047ec137a0e72f89c7f655","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-16 10:30:16.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/5 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":5},{"id":"1c030bd0cb2f6f1c4facf3870c6e7eacea7ec770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:37:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6cb0c9f279de48a6af4344f9a938523f : SUCCESS in 1h 01m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c35e2506eccd40e7aa1d1a7abf0fc484 : SUCCESS in 49m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c45cf802e2364e4fbfee57c4b2a59a2a : FAILURE in 1h 37m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/49b73208ce8b45efbb95a247693b2529 : SUCCESS in 1h 06m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b92c66ed181142b3b7f605bf57b0aefe : SUCCESS in 57m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a08b28d8d3540d8b5cb3e899e1e29bb : SUCCESS in 1h 00m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b043412a50ad49619a0bead72c6cfcdd : FAILURE in 31m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f07f2680706412d8efc287ac82f4b0f : FAILURE in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada221f76da04aa78c567aa468965380 : SUCCESS in 11m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e53ebc8c8e2417abe2ec6e59ceeb42b : FAILURE in 25m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83578806b28c439bbda383c51570fccc : FAILURE in 26m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3fd75c2aecb44d559b4170629fa608e7 : FAILURE in 21m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e198fb84794e28aca77cf4b533645a : SUCCESS in 7m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/adbf319b3318494eb2f13e8ce62c7ecf : SUCCESS in 57m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/538cb95024484caab801524fe78ef885 : SUCCESS in 1h 21m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2ed773d73a594035acde703b69892e2b : SUCCESS in 46m 24s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7954295c5c9c4f3db49bee22d66ec0b8 : TIMED_OUT in 2h 19m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/03c48888cd124ef6a3dc8dd464b23e1a : FAILURE in 1h 07m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1a98560e863244498db152676aa474a0 : SUCCESS in 2h 05m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9adfa51c374041d69fe56e8144061c89 : SUCCESS in 1h 15m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4d3149d3268e4d0bbfb76d6cee36155d : SUCCESS in 1h 20m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/639893829af84257ba7698a277d881e9 : SUCCESS in 1h 11m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7999202144854cb2b8f42b8f850b3311 : SUCCESS in 1h 21m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b0954c837d8c4455a5cf63e39dff69f9 : SUCCESS in 2h 05m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d708dac8b175494fa46c886307f0d42e : SUCCESS in 1h 52m 54s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/563e001aa2874edfb0af9ab872c8ea9a : SUCCESS in 1h 18m 59s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/99f91578b1144bc4a57c87a2bedcf732 : SUCCESS in 1h 51m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/961eab582012437ebd599e97a06f3e74 : SUCCESS in 1h 17m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0e0e3f6230da4dfbb16d2dacfc00f235 : SUCCESS in 1h 15m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1250ea8477a44ec9e39d40ab2fba5f9 : SUCCESS in 2h 20m 45s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4953a9f2cdb14ed6a42fafbac3e85d49 : SUCCESS in 2h 06m 09s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/48e3f59d4b7e4503bdea5c6e0eb27239 : FAILURE in 1h 42m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/042723735d324daea78e07dce80384ae : SUCCESS in 1h 10m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6c1a870bf11c4361bd7c34dec64b04d1 : SUCCESS in 1h 43m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d3a32b7faec4fb7acd02b6ccfa3e48f : SUCCESS in 49m 50s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0ac185de4ee45628c4d846eb42694a7 : FAILURE in 57m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcb934aa9a6c47d6814157870aa3fb52 : SUCCESS in 1h 29m 31s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f94ab3db70524ef09efe6aed45ed4f33 : SUCCESS in 2h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4609ac529fb47f04df45a1b945e117c7f571a054","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-17 13:51:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"42086ae189219c6a20316e88a88995e522109909","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-18 08:56:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0923b0e3330c91122cbf0f690f30d40bcb616c7d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-18 08:57:30.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/6 : FAILURE in 41s","accounts_in_message":[],"_revision_number":6},{"id":"cdb0dfedb42fdf590b612eca4634babab6e6eb27","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-18 09:37:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-18/687922/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/8301ef6 : FAILURE in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1fdc03d06131c6aff2406a7ac247e5abdffa7ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 12:01:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4687df570fe849d1b2841093caab8b93 : SUCCESS in 52m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ba48d108e67478aa3516d8e807658e1 : SUCCESS in 37m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1126c3ebc3a74e0a8b4d71a36e931ebf : SUCCESS in 1h 20m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c53572d182cc4cae85f81d6891063f95 : SUCCESS in 1h 10m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/513610d6cec24fa8a8291e9e3bc0dca8 : SUCCESS in 1h 09m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e010be253a9d44e2843e31565c2a1cfa : SUCCESS in 1h 08m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d4fc1c8b9914968b61b5c2dc9cf14dd : FAILURE in 27m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d769ad05990b448182d47f9ff756ddec : FAILURE in 27m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa54397ae144f338cf074d2db0f2043 : SUCCESS in 11m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/277e4a06e984431b81122701f139c3a8 : FAILURE in 22m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7cba0170dbd493fbd001cf07ebedc4d : FAILURE in 22m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d3084cfffad4b1fbec44c381a0e48df : FAILURE in 22m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ac8a62e6e7436b80462bc300426ce9 : SUCCESS in 7m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ec726e42478940cf9ce9593158dbbfbd : SUCCESS in 1h 00m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ebebd875e4d14f1496e2e25a2fbf0680 : SUCCESS in 1h 18m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b87b36e70e3f44688deeb6bf0604f8cf : RETRY_LIMIT in 10m 08s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/f45bd244bfad4de9843b310c2d4ab370 : RETRY_LIMIT in 11m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a4d581c6689f4a45b35754aa55201f7e : RETRY_LIMIT in 10m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/32b5479b6d25445e8da3fdae0633e0ad : SUCCESS in 2h 08m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1cb87a0c343f49e98d51492f7187f6d2 : SUCCESS in 1h 24m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e22a715a3ef349f5b07775e2ce404c32 : SUCCESS in 1h 22m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/782d29a5517f49a5868a1d485ce0dd76 : SUCCESS in 1h 15m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2135174b1dfe410faa333829c26f32a7 : SUCCESS in 1h 13m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8f3651abf2cb46dfb2bcf3e713dfaeb4 : SUCCESS in 1h 46m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1dd25829b2b4529a62865957614318b : FAILURE in 1h 41m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69daf685cc4345ba895a89865c3a6e03 : SUCCESS in 1h 13m 52s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/52df2ba417954083a1ae4fea4c53385e : SUCCESS in 1h 35m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1cb748492cd04605b3fb738786ef8661 : SUCCESS in 1h 05m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4cd26c7621c1428ca7815fbf0ebb0dc9 : SUCCESS in 1h 07m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a77053c6f3c49ab908571fb41dff7e1 : SUCCESS in 2h 24m 33s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/07824eaf40df4073aeb915156ed4bad5 : SUCCESS in 2h 27m 08s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/466daf4e109d4393aa0924545c684c1d : FAILURE in 1h 32m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d8d454628c0419fa951f7d6a9b7c719 : SUCCESS in 1h 18m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/007a25904194424b97f4771dc7b7bfdb : SUCCESS in 1h 24m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d79cdc868c55476b969c4543e377a7bd : RETRY_LIMIT in 11m 24s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c8a0cb2cf2540798984a8bda18e4378 : RETRY_LIMIT in 6m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2618123a45a46eca00f1e5205b5b135 : SUCCESS in 1h 24m 39s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/87ea7e83b0004576bb1fea79a3837ec1 : SUCCESS in 2h 05m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f170d8b976579f084b05343cea4e5ec7a4ba665b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-19 00:47:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/6/check-mlnx/Neutron-ML2-Sriov/9e4846f : SUCCESS in 1h 18m 08s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/6/check-mlnx/Neutron-MACVTAP-ML2-Sriov/8942bc4 : SUCCESS in 1h 02m 15s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"eb4933e1f73fe1382912a3a35cf13f38c22bcbf3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-21 08:28:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c8569ec2b97472a14c8923591c83f11de4659bc9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-21 09:10:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-21/687922/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/39e5379 : FAILURE in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0a2ae696b440e4ad029af6823919be8ea1f96040","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-21 10:10:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/7 : SUCCESS in 1h 40m 51s","accounts_in_message":[],"_revision_number":7},{"id":"c553d9be39d966fdf82d211867b0eb4ec276d1f2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-21 10:47:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/7/check-mlnx/Neutron-ML2-Sriov/392fc8d : SUCCESS in 1h 23m 49s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/7/check-mlnx/Neutron-MACVTAP-ML2-Sriov/1c8a34f : SUCCESS in 1h 04m 33s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"10a2d4cdc196a71becbf7b79eac939797674a3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 11:00:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/13e15612ca8b45418561f0cb30a26703 : SUCCESS in 59m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/869804a9030f4a008e2495f32bf6312a : SUCCESS in 37m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5cc3020e1eb5444d84a53fbe1542d63d : FAILURE in 1h 35m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cb372a2f2ec34769ac791ce6d435c2a4 : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7ac8a2222f8f4a37af039315b2f15923 : SUCCESS in 1h 05m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/454e927c459849e0ab08899d7466b7e8 : SUCCESS in 1h 11m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b66399b145344b649b124ef55f10119b : SUCCESS in 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c61066c9db1e4f52b67dc4f3150ac0e5 : SUCCESS in 21m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a1b6b4d9f94c85951ab18a2841de10 : SUCCESS in 12m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2395864b0d854b6da955b93a34d8e936 : SUCCESS in 24m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05a28e38cd7c4569a113aea455d8c29b : SUCCESS in 25m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/564f211da7a54693b38bd923813b2315 : SUCCESS in 20m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2bb50934d5f48aeb20f87912097045d : SUCCESS in 7m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5eba57a9aa774282928f5da2ef784b2c : SUCCESS in 59m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/17f5241152714e1a89565a983494cf76 : SUCCESS in 1h 12m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8aeee67b149f41f0b25f2caabd920c9e : SUCCESS in 46m 19s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/fdcdabf788514182b19311540574bbfa : SUCCESS in 1h 08m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/29ecc2ce43504990a92d08003c0c1c18 : FAILURE in 1h 00m 27s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b71a4a28b5f744f1b9fa1d8ef067d11f : SUCCESS in 2h 15m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/59955abc9b3f4f5b96e0a3020ef310df : SUCCESS in 1h 35m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/663c423543c4495dacfc3439ab99fe08 : SUCCESS in 1h 15m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/64a05b52f92940e68d4e1e5f150b1e86 : SUCCESS in 1h 20m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf1cb935e5b14dc1bc26a0b80d719ebd : SUCCESS in 1h 16m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4bf08e14a86f428fbad470ce5a53674d : FAILURE in 1h 33m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f919850ae8f14e75a3df9e8007217923 : SUCCESS in 2h 06m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/77aba7b0780e4141bd700d196cba7628 : SUCCESS in 1h 27m 30s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/0ba2d80c20da411099138595455b1aba : SUCCESS in 1h 34m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b0d0708c161484284b5f67546d120d4 : SUCCESS in 1h 03m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cfe41b7ef6df41bea0dda3cb8ecb3fa1 : SUCCESS in 55m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0533e99dee7e43bcb739fa49b9f1dded : SUCCESS in 2h 25m 18s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/15566c4d56bc4162bce7012395c5f792 : SUCCESS in 2h 18m 22s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3c8a32b3b5cc4b538ce891ba8c5cb629 : FAILURE in 1h 42m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bd58a6b978041ac8586a33890b94c0e : SUCCESS in 1h 11m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fa06e537984346a090274b46ddecd31f : SUCCESS in 1h 30m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/72e52801aa3a474c8a5d9fab1b6d346b : SUCCESS in 54m 06s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/889be4c5cb884b79ab6176a23c60a6a1 : FAILURE in 1h 04m 50s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bfb7cc45ec8498abd21bc7195dcc18a : SUCCESS in 1h 17m 02s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/48232661861d48f4b64f8a5d415ae84b : SUCCESS in 2h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5345a573a4107e47b2b8f3b1c8750fc83d267b1c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-21 20:26:02.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"703b6e65ba355e9660709a72b79880e9bdd0b9f2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-21 22:23:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-21/687922/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/11ab16f : FAILURE in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fa04056fbcaf0dd147f393ee3c04205447f53bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 00:12:41.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2953a3c2f15e4bc9bfcda46d14f5c4e1 : SUCCESS in 51m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/556b1aadfb754b4280acb57f4a10ea21 : SUCCESS in 31m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/88684d28dd9945f489a6c0a98a2c10be : FAILURE in 1h 31m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7a192c3c1240498e9574a27f1f420203 : SUCCESS in 1h 09m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64043450c86a4219895173f5299f4803 : SUCCESS in 1h 11m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/006c2b2afc824235be1641977a7a9a95 : SUCCESS in 57m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0ac8aea92ab4ee18e76bb4a8623cdde : SUCCESS in 31m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47f0022809a64f47afe2eeae4904f759 : SUCCESS in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef667e8ad99c4087a336d4ef5dff5b12 : SUCCESS in 13m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e3e7e8cd4844ac4a71c21e4f06b0a9e : SUCCESS in 28m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47037201d9ce4a018e50c60a95686826 : SUCCESS in 21m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2abd63e55a0947e4abaf400cc9d5d11c : SUCCESS in 23m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db7b69e225643d7b579ccd66204633d : SUCCESS in 7m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d97da77beb4d4fb4b41ffbdfd7a18455 : SUCCESS in 1h 03m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7dff1f12ed1341e8b12561271b02d884 : SUCCESS in 1h 25m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a672fc67a7af486cac0004fb1b140b8a : SUCCESS in 46m 01s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/4ad220efe9e44a02901429bd7b1432cd : SUCCESS in 59m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f251315086884c3785cded8ae9fc9f21 : FAILURE in 1h 00m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cc4ef416f3d2496fa5c926b9b53babdc : SUCCESS in 1h 54m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4c64d306f6724a9e9d8cbc6670ccdbb6 : SUCCESS in 1h 31m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8035d6b07e3946dc944e5fac04259c43 : SUCCESS in 1h 44m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/969076b0b08243ef88909350a52caf18 : FAILURE in 1h 23m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b082d1135004b61a0e953fe33662d60 : SUCCESS in 1h 10m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b1e309155abc4a4fb754e07e5de70d84 : SUCCESS in 1h 41m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f76e2466c2554b3cace37f3608d100d4 : FAILURE in 1h 50m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe062aed4d7041018992a21db9e0348e : SUCCESS in 1h 33m 40s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/4699036d980b4df88f14f73f56965d4e : SUCCESS in 1h 29m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/38f6acbcca3c440a8408d2d564d16358 : SUCCESS in 1h 00m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b0b79c4cb9e64e3c8aa925ddf1d47564 : SUCCESS in 51m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a108480302f84d028c19e8f91df1900e : SUCCESS in 2h 20m 27s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c011400861a14085915911521f7b7cb3 : SUCCESS in 2h 08m 14s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b29de75eebe24b9185bf596fd81b76b4 : FAILURE in 1h 23m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0945b3d4e494a549748324ac8235b83 : SUCCESS in 1h 00m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f0ca08d42acf4544bbad7dada25160a8 : FAILURE in 1h 23m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5b1de3f29434c86b095ca182d0b7632 : SUCCESS in 52m 23s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae8d34cbf7474e4b9133fd50a31d7285 : FAILURE in 1h 06m 50s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce2897bcbd1d407788c5f7231dd09da2 : FAILURE in 1h 02m 03s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/14ced19e9ceb4784b3f5152405d9d51c : SUCCESS in 1h 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2d1b5d96d1a30e280f9fbfa832c3ae59688662f8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-22 04:53:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/7 : SUCCESS in 2h 01m 22s","accounts_in_message":[],"_revision_number":7},{"id":"11eb7e8db612539e73a2caca77afccdea8d1db0b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-26 19:20:16.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"2fe0062321a848215de7c004d397864a020247a7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-26 19:52:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-26/687922/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/10e5892 : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"85961861ec14b91c14d16034098499e5b5bd21cf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-26 19:53:21.000000000","message":"Patch Set 8:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/8/check-mlnx/Neutron-ML2-Sriov/4e19271 : FAILURE in 4m 32s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/8/check-mlnx/Neutron-MACVTAP-ML2-Sriov/234c950 : FAILURE in 7m 46s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"f110f93efcd8107e0c4a636d9e4546210b2dc273","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-26 20:57:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/8 : SUCCESS in 1h 36m 55s","accounts_in_message":[],"_revision_number":8},{"id":"a543158de4b6bd7452019a849e780f3e102a6d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 21:46:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/212889426c8943c0b3e3b064c7d26b8a : SUCCESS in 55m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4b9381610b3e4f909bcbba4244811870 : SUCCESS in 38m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/6042cc95dbd3428595fb8cee7631bbbb : SUCCESS in 1h 21m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ee6b58aad10e45a2be05dd571133fbd3 : SUCCESS in 1h 09m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f8d22befcfee48cc949e535d06c83dd7 : SUCCESS in 1h 06m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fd9b87259bdc42af82e3b75e2bec5723 : SUCCESS in 1h 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/036113a54a494b10a99f5a7090d88387 : SUCCESS in 28m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53065b3befe04e3b8bba70a304c02d36 : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c5059d579b4906969baa6ef53ff518 : SUCCESS in 11m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca91a2a65d714028860c89c5702a6f3d : SUCCESS in 24m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38305d9004dc464b8e44ac4bd30520b5 : SUCCESS in 23m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf65c3ace0ef4ac3aec3f888450d3291 : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df52809feca43a19cf646784f59eb81 : SUCCESS in 8m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e863b914c8c342058714efce93b466c6 : SUCCESS in 1h 00m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8bab35e89a644b4090c16f4e690dfe27 : SUCCESS in 1h 18m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b6c9dd6054a0482e9c10b64c2e43b91a : SUCCESS in 45m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/97ec0fb63b9f491080f09c2876ce0a8e : FAILURE in 1h 00m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e797fcaa97fd435e81cbe4ef056ec6cd : SUCCESS in 1h 48m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4710455e621041f4814914a24a74fd0f : SUCCESS in 1h 15m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ecab5da8342c4c269c5e3adad90fcec1 : SUCCESS in 1h 18m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4053814e2e37456aa56f03cd5e2f02f6 : SUCCESS in 1h 05m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6838495546a4e7f87b06f41ec35b7fb : SUCCESS in 1h 11m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7963f317f6b841708cc08fc3de20f4f2 : SUCCESS in 1h 45m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a61cdde96f34120915d90edf26a5cc1 : SUCCESS in 1h 49m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3919b30d192475c9e7f68dc7e44664f : SUCCESS in 1h 15m 49s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/13301c1e44274ba7b2437f73383557fd : SUCCESS in 1h 23m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1708342a4b7e4429a675c4bf18ccac68 : SUCCESS in 59m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d7800242257a479e947422c1ab4e9098 : SUCCESS in 48m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0578c0566726433786d4c09bcce31280 : SUCCESS in 2h 24m 47s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/29e7bdcd3590462da30ea622a20de691 : SUCCESS in 2h 04m 38s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/fb4beb5e6a5c48f88917319091fa0b22 : FAILURE in 28m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff30186c3844b8391cafc056d038386 : SUCCESS in 56m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/55212f6ff4aa4197a5e087b63fae29f2 : SUCCESS in 1h 51m 17s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/adf8492c2c97423baf94e551d5837afe : SUCCESS in 40m 13s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ff83a5c386d4f10a97ff6ca92ca8478 : FAILURE in 1h 03m 12s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b18e19860a64108863fb3073d6449fc : SUCCESS in 1h 17m 18s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/138c6d810efd4ded99436a1674c3fe76 : SUCCESS in 1h 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4a56fc750152fdb66a6a53715f30deca1d85d97c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-27 11:23:31.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"6de1368530dcace2e5c235162981f8f5a83c3563","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-27 11:58:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-27/687922/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/0d37250 : FAILURE in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b222c8c1aa54373f279dd6c8f8ead8d5922c29c2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-27 13:04:23.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/8 : SUCCESS in 1h 40m 20s","accounts_in_message":[],"_revision_number":8},{"id":"51dae3087af911083a9710688cc56b4d03bd9212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 14:21:03.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6f1fe76a515c429da95515e98412a3ee : SUCCESS in 42m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac1eb9e8b61049c7868e7f1b10d6be4b : SUCCESS in 31m 23s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f9d732c18b0a458590073367bdff5c2a : FAILURE in 1h 45m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ff6ddbb704944565b9d2de075ac04e06 : SUCCESS in 1h 02m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7b60feaaee76408b8faefdeb29e53ee6 : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1082051cee2b4db6a5f0cba665575b26 : SUCCESS in 55m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930d969dfbbf463387b69159882c3d0f : SUCCESS in 38m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04fca975b9cd4178aa58b6af0489c4a8 : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89163b218924a569592c699e4c9e29b : SUCCESS in 10m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9559fa6b4f6145fe80e3470b82fc7343 : SUCCESS in 32m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/995dc956ae7c48b9aa9b1520beeb0087 : SUCCESS in 19m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c194b3103dc449be8b700272125f8c02 : SUCCESS in 31m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b09f2ed0384e9aa19165a3797d9574 : SUCCESS in 10m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/621734c5a8634849b4b41c7b48ccd9bf : FAILURE in 1h 08m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/69aa08a9bf23409f8347e767a74e0bc7 : FAILURE in 1h 20m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9498f9dd1114e1e980ca74f526842e1 : SUCCESS in 54m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d17f7bec7140462fae60de89f00ffe67 : FAILURE in 1h 28m 52s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3d47e3589d2f4a1bab2118d799cc2c2b : SUCCESS in 2h 08m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d0736b13321f468d8934bee01b034a07 : SUCCESS in 1h 37m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4229dd96d72d412d8c5ea83904b4aeb8 : SUCCESS in 1h 21m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddc0d6002a0b4716824935f56e202209 : SUCCESS in 1h 41m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/35f2b65977a04591b5691869fe6048a2 : SUCCESS in 1h 24m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/40be7dd97e6a48758c8504004d867b05 : SUCCESS in 2h 03m 42s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90cf5c238d5f4ebda6d31d3aa3ff6008 : SUCCESS in 2h 23m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3784abbbf0bb42bb84bd32963e544182 : SUCCESS in 1h 15m 13s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/0fde444a839246728a012931c8b97a44 : SUCCESS in 1h 32m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ef7b377b420040b5be47f22358c68327 : SUCCESS in 1h 03m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/102c7e3f913845eb8702886234f6ab02 : SUCCESS in 1h 05m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd75b6d6f697462aa595d49c8c542d56 : SUCCESS in 2h 48m 06s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4140dba4164f41c8b6b8ee670911b58e : SUCCESS in 2h 32m 13s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/274a9f8bbb7a4f4e9cce3681060f0e15 : FAILURE in 41m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9de71511bb554f60a303e68f50e6ff46 : SUCCESS in 1h 25m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/533aca0d98d743ce80fffdb3ee63efb2 : SUCCESS in 1h 59m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab7a330bb30141f98f8db4d52df0fe9f : SUCCESS in 46m 58s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e639945582348fea56e00d57ff60523 : FAILURE in 1h 02m 09s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0afabccd9f74aaa8001410feca53746 : SUCCESS in 1h 17m 20s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a7c9afd9b6574877b0167804492d9a1e : SUCCESS in 2h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"25843744f861327388023df61a792ab7c841bf7c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-29 13:12:54.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"9d60ccb91bc2d89834bfa9f22112ca0b4a2a66b6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-29 13:45:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-29/687922/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5e0a7f9 : FAILURE in 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f55aed1c9bcc12daab4857ede0b3ddf9ab7a2220","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-29 15:14:07.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/9 : SUCCESS in 1h 42m 52s","accounts_in_message":[],"_revision_number":9},{"id":"f5df81c654fe9b596ad9a24212f88c47e27ec830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 17:51:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb6acf14d64b414ba150fc8dd531db46 : SUCCESS in 45m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b784738b81ad4d789beb0154c779970c : SUCCESS in 41m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/7ce256419a4c4e069365718a8928804f : FAILURE in 1h 58m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e9ed1d7d61c046629d156e878ab6edb0 : FAILURE in 1h 03m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dbdd0a91e1ca48bba8c76ed9913f5f3a : SUCCESS in 1h 10m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b03e3e860eb4f1994767877c9e1fe17 : SUCCESS in 54m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/204b02b87d4e4f6f901a72a12426d195 : SUCCESS in 33m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54a957c06bdd494aab1bcee33390ba4d : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749e0b45c4d84883aa36a884448fb121 : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bda306d9c744feab7ab0edd6855c914 : SUCCESS in 22m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b6ea9c7d3324501b0a537cd7598845f : SUCCESS in 28m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a2ba4d18f84771b495af8f816e5b9e : SUCCESS in 8m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/31d37606c7c34524ac1ca5aefa0eedf1 : SUCCESS in 1h 02m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b52d9fff3bd4af6a408d02918a485dc : SUCCESS in 1h 29m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1e9b215c9bc44fc298a0787476fc7203 : SUCCESS in 52m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0a318247e08240d495f745a3d81a9c81 : FAILURE in 1h 10m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d254e11bf7bb4d6487fe2599659213b5 : SUCCESS in 1h 40m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/efd3bbf074614df7a785f39123de809a : SUCCESS in 1h 35m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/08c3a1b716e2442c86d00aa30a857568 : SUCCESS in 1h 28m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/960d8736dfc746a584e02957fe7f0161 : SUCCESS in 1h 15m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/14378fe986b1475e814d10ca58e12a6a : SUCCESS in 1h 05m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5f099ffc1b704fada01b00484e96d81b : SUCCESS in 1h 34m 45s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3252c114bfa4c2bacaec9743120d495 : SUCCESS in 1h 58m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1671897fac8b4b028b07346bfb92ebcc : SUCCESS in 1h 15m 40s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/683aa0c82fb14d48a719146c7e56febf : SUCCESS in 1h 47m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/29e91c27b4564c7f9d92ee0041079d2d : SUCCESS in 56m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/80f5624dcfcb41daa3dfd2d5e2b987bd : SUCCESS in 48m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/99865201b7384dd0baad3a5bbf6654c9 : SUCCESS in 2h 34m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4193529be48045479b7c9e41d7c63faf : FAILURE in 2h 22m 11s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/bb618671c3104c41a51fc8a2963b004c : FAILURE in 1h 57m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8a6e59a0d4d43d4a2fe67487943bb36 : SUCCESS in 58m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/791de7fbcb87435fafc851321425c7a6 : SUCCESS in 1h 21m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c06fd5a9597a4a4ca706251f5f52dc74 : SUCCESS in 41m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d0549789bc440c48a26c164aa1cdf68 : FAILURE in 1h 00m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9090bb3ab6834547a2016cc90fbbe252 : SUCCESS in 1h 34m 46s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8350ef5bf4f946e8a0a77bd53c5dc2d0 : SUCCESS in 2h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8e460f60f9d4d6a77e9d2092bda9f2f326bcf332","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-30 04:59:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- Neutron-ML2-Sriov http://13.74.249.42/22/687922/9/check-mlnx/Neutron-ML2-Sriov/83f8923 : SUCCESS in 1h 15m 26s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/22/687922/9/check-mlnx/Neutron-MACVTAP-ML2-Sriov/13a4609 : SUCCESS in 1h 00m 25s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"b3af2425dc50512a3d5b891c17a6f25f5e7fd492","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-30 09:13:09.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c671a3ea0d1105d9a5b6785d5c06ab869b9978d0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-30 09:46:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-30/687922/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/03a0fa4 : FAILURE in 33m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b9219e9ae8c9f05288862a29136b3edef2c6f425","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-30 10:52:39.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/9 : SUCCESS in 1h 38m 57s","accounts_in_message":[],"_revision_number":9},{"id":"7ef2f070b1932599bd4a234b7bb31b839ec409d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:17:31.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/57b481853334406d819813f89cb42e4c : SUCCESS in 59m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a390de71cdf4ac8b40689291be73c23 : SUCCESS in 40m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/59ef18392842408987f0917cccb29237 : FAILURE in 1h 43m 46s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/17fbd341336f401fa1274421b68f41e0 : SUCCESS in 1h 15m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3e0050a3b7f842bba6591db729060cb1 : FAILURE in 1h 05m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1e6d29f261549c8824b24dbd29f2a60 : SUCCESS in 58m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a096af5bc9cd4c9684688a3427dfed74 : SUCCESS in 26m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8038c9f476d4c379f5416a295ec2d23 : SUCCESS in 24m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96e62d88339463cadacc4fec3d047ab : SUCCESS in 14m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b45dc0bd67cf4078badcb3fc210bc3c1 : SUCCESS in 19m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae3e5d80ff9146b4b44e6628ae6e3e09 : SUCCESS in 18m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b06b0e9dfb242d9b727b2921ab63a13 : SUCCESS in 8m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ef07d2956d243afaeacbb0d77a3bf78 : SUCCESS in 1h 02m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8975aab74ee04adaabfe45e3927b9d24 : SUCCESS in 1h 35m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d6d145c2e7064ef08cc0eb9c0121eb1b : FAILURE in 52m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/acf81f540b1e4f7dba6ee325bb083726 : FAILURE in 1h 11m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bd217cb928774856b021f0dc4cc869f5 : SUCCESS in 1h 54m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ca1a7a9d0fd6426ba7e705e401ae355f : SUCCESS in 1h 08m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2c5f161cb995486db120c7393e630941 : SUCCESS in 1h 08m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c410eb2c149e4558924d36b0875a6abd : FAILURE in 51m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a6dbfd007d24ba2b8ae388a9298e873 : FAILURE in 51m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9d49ce561fc04283a55b0df5065606e7 : SUCCESS in 1h 31m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e455beff6fb47299b0bcd7c62c75a8d : SUCCESS in 2h 12m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bbed741079f24c3f957488439271b643 : SUCCESS in 1h 22m 34s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/d5a0c0f61afe46f799b0f08314a4a47d : SUCCESS in 1h 28m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a88cd81cced04cce91ceac5f6d937b63 : SUCCESS in 1h 04m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/166d580b4e3a4b35833d33a84b51b285 : SUCCESS in 57m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46246118dc4841fb870a03c3321ba6d4 : SUCCESS in 2h 38m 15s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cebd3208968645e9bb1b9969939fe967 : SUCCESS in 1h 51m 28s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/5c4e0dcdef3d44efa1d6860a11c7cfb1 : FAILURE in 1h 45m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a6296c016fa49bd87867617c6aaebee : SUCCESS in 1h 14m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/54c22b0393a7462680a1e0508a1b28a7 : SUCCESS in 1h 36m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cdfdb1e107964890818769ec8518ae69 : SUCCESS in 56m 08s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f878967f8574682b08e9f1e6bb33ea9 : FAILURE in 1h 31m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a1d60fb71d54bc5924b70eef01ae267 : SUCCESS in 1h 17m 13s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/443c299caea04d0c877f2162d179c6bc : SUCCESS in 1h 45m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d3f9a71fb9d86edd7f27f752895b15cdbaf0094d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-29 18:13:10.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"92bd3938b9c45fdd616bf955e50592061ff3f5c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-29 18:13:40.000000000","message":"Patch Set 10: Workflow-1\n\nThose new cache rules are affecting fullstack tests, I need to review it.","accounts_in_message":[],"_revision_number":10},{"id":"c107cb19c8a80a1090ea12763f950e6002c26c68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-29 20:32:24.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/10/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/10/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"1cb53df33f6f7d12b6bd20b504e14c334b6c87ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 20:59:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ec56514e7ec64181be172bb859c0f219 : SUCCESS in 58m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d390e2f617c641278892e227d709dcb0 : SUCCESS in 38m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/029408888ab94cad8190e77599b81e61 : FAILURE in 1h 33m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79017e992da145b394851a781a27a3d0 : SUCCESS in 1h 01m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7f8e5ec456bc49e39c48159c263d82ca : SUCCESS in 58m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f264396d19e496ab5b5cc0f7cc2f3ac : SUCCESS in 1h 05m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0b6506bc0424322a0b2ab5deffe85a1 : SUCCESS in 33m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/839a59b295094ba6b4444457beae6074 : SUCCESS in 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06523a449194ac58243ca86911bba9b : SUCCESS in 12m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7948fdd97a6247359b802cfc0ccf7ab4 : SUCCESS in 23m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/558309a5e7734be59f4d7244e87843f3 : SUCCESS in 24m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d59217c49549d8bcbd775a091b49e9 : SUCCESS in 7m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8d9853955d1040c69454f04c26d8ac56 : SUCCESS in 1h 02m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/04933673533f475b9ff7a8ba6d877b06 : SUCCESS in 1h 08m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/736ed85350664d1bb13cb470efe797fe : SUCCESS in 43m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4ec232567e0e447b890bb9fc601fac56 : FAILURE in 1h 07m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8be1fb0f66f94d6fbbdbf048afe65432 : SUCCESS in 1h 52m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/eabc989b75c2401fbd7c667eb05c486c : SUCCESS in 1h 22m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a94096b5e6b04456bca317e00b30e35a : SUCCESS in 1h 23m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2dc13f5b17341c6a93996107c1a7491 : SUCCESS in 1h 07m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8fa0a02f4d2840c08704c9280c535f8c : SUCCESS in 1h 11m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/594edfaf7c1f43bd8db3e244eb5a110f : SUCCESS in 1h 41m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/14e5bc015dd44481a25ace125debebce : SUCCESS in 2h 06m 22s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00d0836ca585405db0bdbfb811580e55 : SUCCESS in 1h 08m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f2425db6bfc43f9acbcf077f08a56c2 : SUCCESS in 58m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f66bc0e8b194c9d8b4786176da38a35 : SUCCESS in 2h 11m 50s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/305e64deae9a49a69ae91803ad801a23 : FAILURE in 18m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d60e6edba203429b9b8f488e1b249f28 : SUCCESS in 1h 05m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cf2b38586e34423084adaf4a047260ab : SUCCESS in 1h 15m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0187b5f3e0314d4fa0e74627ad01876d : SUCCESS in 42m 31s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b624ad95b7944ef78f35083e06936d77 : FAILURE in 1h 06m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/61b39ac8a963435eab64fdfdb86ce8a5 : SUCCESS in 1h 15m 15s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c8b4d635dd74edeb05c0df13f3ef088 : FAILURE in 1h 48m 31s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/358cf16e9a5f4b8fb4226f9e9ce0480a : FAILURE in 2h 40m 21s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a6bcdfcf943b4dc0b0e5ba149c968c10 : FAILURE in 2h 06m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"167cdda13c6d6d4946fc0ddd1d8c3a667fe4ff14","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-30 00:39:14.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/10 : SUCCESS in 2h 06m 45s","accounts_in_message":[],"_revision_number":10},{"id":"4e7b85bf6bad9b49ca329e409e4475ca9a2867db","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-30 21:58:42.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/22/687922/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"3ee38d78bba8b43dba7a05adfadba648aa8256dc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-11 13:48:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2bbd5954a45bca8a67601eb934258f1e0f8dd7c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-11 14:56:53.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/22/687922/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"8faad4d98efa885f13fb666b21a549d864d66e79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-11 15:22:08.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/11/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/11/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"4a204935708ce1e244277aac3e052476206ce132","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-11 15:42:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f55007a48d84beb87502e33fd16dc22d195cecac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-11 16:48:44.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/22/687922/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"bdfc6f582be95e6a3b25a6483e6839763eff9301","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-11 17:53:56.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/12 : SUCCESS in 2h 06m 35s","accounts_in_message":[],"_revision_number":12},{"id":"f69eb55f5bab5021a45ab2b86b208909b373245b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 18:24:39.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bca1f980a64940da90f63d1a3165db4f : SUCCESS in 53m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9836d46a80ce40a6b88c8a70dac6d60d : SUCCESS in 37m 36s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/13bffc792ca2409b87e6d9c765bcd5c0 : FAILURE in 1h 24m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6511cb6a72794dd6ab9e1b6f5ea39a2e : SUCCESS in 1h 11m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/821414e18e794ca38ebdd58fd17fc42c : SUCCESS in 1h 07m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e8247ef346d48a484f3c807ca5fbd1f : SUCCESS in 1h 05m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94c50415159547a5a01a6512a5a4e468 : SUCCESS in 27m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66beeb93405f4b54921f91a0e4c6aa93 : SUCCESS in 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c1b0c29ec704f98949b4146854d41dd : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2710836f81614af4a8551361a67bb55b : SUCCESS in 28m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ce7757f991c4560b9ffe29fa9111e12 : SUCCESS in 24m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c0adbf22894b379dca932644f35473 : SUCCESS in 9m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/faf2b74d6d004edda40a49ea21293c13 : SUCCESS in 45m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f0910a90da084e57949cd795128218c0 : SUCCESS in 1h 31m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/521508dd576c44d897a13d9aac896143 : SUCCESS in 1h 59m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b32128e54d194bf0b2d368ba21669a79 : SUCCESS in 1h 16m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/173c72b2fd314b2fb857cacb46b5e537 : SUCCESS in 1h 34m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8cef5652aba4846bc29b0b62ebb7572 : SUCCESS in 1h 16m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/18e93ec708234faa9a57c9dd41f9f832 : FAILURE in 1h 18m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e6fb5f46f0b1441aa8a309f1106d0f15 : SUCCESS in 1h 09m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8082fe316a894d21b88001d27d779b1a : SUCCESS in 1h 48m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1ab83e49355c4993a5ed045b3b07fceb : SUCCESS in 2h 24m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a518bf01f664c87b69ee9736bdcbd59 : SUCCESS in 1h 11m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/edb663d8e350438faaec8297ff22cd45 : SUCCESS in 56m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca80b057bc03400c9411165676ec7fea : SUCCESS in 2h 16m 32s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2fb0e97092244584a627dcfe3d4bb605 : FAILURE in 1h 47m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21a475d0ed5142af864b1d60029a5dc0 : SUCCESS in 1h 08m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7b1de95a97394b528b5c182c73270f78 : SUCCESS in 1h 00m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54d612c5b37046a8af6ac223c346f1af : SUCCESS in 42m 13s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a2ea3cb4c3a84f3fac42df59001420d5 : SUCCESS in 1h 05m 27s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8486fbd188a94958a248aad422b55ea1 : SUCCESS in 1h 20m 12s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/89da09c0f37b49638e0007a6de031922 : FAILURE in 1h 37m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f4ad6d1445e74cb98ce944ee5dfdf436 : FAILURE in 2h 34m 44s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d02e203e5df44e25a26c0d6461616b45 : FAILURE in 2h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"faef7f7bfc50281ac47f7df81af55e524b7edb98","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-11 18:33:38.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"277b693de55727c0cd89d34edc52e82c869cd8b1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-11 20:25:11.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/12 : SUCCESS in 1h 51m 02s","accounts_in_message":[],"_revision_number":12},{"id":"b12905153544fd481577491f4a68e06898cf66cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 21:53:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/61ef25afd1cf459a98342dcc9500b918 : SUCCESS in 54m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0cd169720e8a47598e39a94afb781eb4 : SUCCESS in 41m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4def61b4df5b4b7e911969497dc8b70f : SUCCESS in 1h 24m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ac08d9a1882e4a4596c2b6b0c326731b : SUCCESS in 1h 11m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5dd0ad8575134f949ccc2676c24d8aac : SUCCESS in 1h 08m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d938de7e8054f139b8b2cd72c361262 : SUCCESS in 1h 01m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c0c0f6524ea40068615654fc5172efb : SUCCESS in 28m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db46905c0b2e488fb212fdb5f2f2c2ed : SUCCESS in 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f05b7c39ed44c1498f3fef9e791e822 : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbb9c81eee5a438aa376298ba6895fde : SUCCESS in 25m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c428acd0ea73426792ed3772a6931533 : SUCCESS in 28m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0cf78aae084038bc97a2d992a64931 : SUCCESS in 10m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/08d506c2708649018de9ea79ddcac5d5 : SUCCESS in 56m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c1544f81567b4a37bd59f828f121769a : SUCCESS in 1h 04m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6933806678af4d57ae887aa0e6aca729 : SUCCESS in 1h 57m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/88783e4409d34efa81f6e1fcb1e8d33f : SUCCESS in 1h 27m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0291644e47f43218e70646589c50061 : SUCCESS in 1h 21m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/987041f40e3e4b59ad1a19849e256c67 : SUCCESS in 1h 22m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c3b6e7be4b4a463cb3363a69fe2d93bb : SUCCESS in 1h 13m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fab073abf14747b084a3b04a2ae68063 : SUCCESS in 1h 20m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/052cd08b5bd3466996563c7d4cf63c97 : SUCCESS in 3h 05m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e651eb3d0401469cbcfdc377b1bc1b4c : SUCCESS in 1h 45m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d5fd1337d4b4254ab9b375d7a4606ea : SUCCESS in 1h 20m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c518d0dea1734519a97e06246ce58335 : SUCCESS in 54m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0188786b66f49d981a19b656e87d98e : SUCCESS in 2h 16m 05s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/25753c8ce60e431cb44d3aad0cc50403 : SUCCESS in 1h 56m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/290c751678ae4cd385d01305ab4bd8ac : SUCCESS in 1h 02m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/895b47b9fc2b4503b89adafe49d3e887 : SUCCESS in 1h 15m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9baba3ff670043d9bb39174c164a41e3 : SUCCESS in 43m 33s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1d70ed3e271477e8ac7461bb5c71d04 : SUCCESS in 1h 17m 46s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a04e8348c7ec499bbc9a449bd0b729a8 : SUCCESS in 1h 13m 41s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ec6703e09274e8d8fdbf643b579caee : FAILURE in 1h 36m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/91f30286da5d46f29ade6678cf25839d : TIMED_OUT in 3h 04m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4f65df6fab234bec97885e7cdab4b92c : FAILURE in 2h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f713b82a21f31406e311e6d900965a3b45d14e91","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 09:18:24.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"91f3a0a92f403d9165dbadc13ba31d50682d2841","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-12 10:10:18.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/13/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/13/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"65e771d705ea66d7855eca57d1c212374a27ff4b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-12 10:38:39.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/22/687922/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"079f85b5541b213fae6634639734b74697b9a0fc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 11:20:09.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/13 : FAILURE in 2h 01m 14s","accounts_in_message":[],"_revision_number":13},{"id":"eee8bcdf43d2f10ab12b52783459beba06d8554a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:13:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f21df0b6b1a8491987f5e4c78f2239f6 : SUCCESS in 1h 02m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7e6889706ce4a3fadd8108530def5c6 : SUCCESS in 1h 02m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/468479b72d1c40c78f02e0ff416e5be4 : FAILURE in 2h 09m 44s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8320d043d8c94301bad1b8d2cbe81fc3 : SUCCESS in 1h 40m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/80561b69e6c84ba0aaafc10770cb9de5 : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/77c6fa134bd6466ca90a96bbed350012 : SUCCESS in 1h 37m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc2d64c43b0149dcb1936f1c72a5d1ce : SUCCESS in 42m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a93c45243184ebfabf196c6aca44b68 : SUCCESS in 35m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74558daf7634b9b960d961a249b40e7 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae965ae9322a455bb230cdcc978d7338 : SUCCESS in 29m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b8c61b5e53c490cb97033fceb5a329c : SUCCESS in 38m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d35e07aa63e456c93c12dd7f8a2b20d : SUCCESS in 10m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/52ff34fc896e45b8aee1e53da2023b4d : SUCCESS in 44m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7ce80f54a5b74e59a9b90a6acb8004a2 : SUCCESS in 1h 08m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/51fa8d3d42594b96b9663db4083a789e : SUCCESS in 1h 48m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/49bab29e79f84595a09fe7bbcd11a482 : SUCCESS in 1h 15m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ed35cc2ce3d740f480788076c805f838 : SUCCESS in 1h 11m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c644ebe16bf146c6a6ee08e21fec1565 : SUCCESS in 1h 03m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/73112ebbb2054570b136c0d38e0325c3 : SUCCESS in 1h 08m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c2c381435f34372aa5d4a7dcea93f66 : SUCCESS in 1h 18m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1442d0c5d3b046f49f0d8af07027a809 : SUCCESS in 1h 39m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d825b8b3a294865ba0e3297a791bc89 : SUCCESS in 1h 53m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b83297a12fdd48c1aa6e871c649b376d : SUCCESS in 1h 18m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0df987e9f2194a02abb9dd74b1f5815f : SUCCESS in 57m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3341be1056cf4686a86d6d39b01e8eec : FAILURE in 2h 24m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/859cd5ce76fe448b8941e9b8f8cd65b0 : SUCCESS in 1h 44m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ae05365415943409fe49cdb990d6563 : SUCCESS in 58m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/50fad561de23495197f524a2ede2d08b : SUCCESS in 1h 12m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7523dd8f12a44afb0b207022aef969d : SUCCESS in 51m 52s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2ae5275aa784355815da6d3d5ffa297 : SUCCESS in 1h 25m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c395e4908a7045ad8e492e8fc55b9fc0 : SUCCESS in 1h 22m 42s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7acdd41c4ae648c2a4364e82c8a5a50a : FAILURE in 1h 54m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5060a60a2aa04cd884d627aed5613a2c : FAILURE in 2h 51m 58s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/97383c2fa53a4bbc8aa9e04532a7a201 : FAILURE in 2h 06m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"19a581f349ed34a95625efc3684eeb008f6a602c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 12:54:17.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"9ba436bde60e7aba6e9062a90670d998fea22bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:45:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/075a9ca18a8146dfaa111f2f8cc92539 : SUCCESS in 51m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/25f7a50bb0604a119fede55fa95a69df : SUCCESS in 33m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/6f2ff4eedcfb43bbb070e89962fb4213 : FAILURE in 1h 28m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7d811452acef4d9494df493402999f3b : SUCCESS in 1h 10m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/463ef7e8fe964f1187a3379108009980 : SUCCESS in 1h 07m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1c58c6b45a724c4287a3ad33c3704117 : SUCCESS in 55m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32feca6fa6904895880097e97618835e : SUCCESS in 26m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bec88449525f462ebba0587ca6191cbc : SUCCESS in 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f36b0f56424f06a3aad5b85965bef0 : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c263c4b2728347f495f6b766ab6118cb : SUCCESS in 23m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4736d93b70804ed1a19ca8885378946d : SUCCESS in 29m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596cfdd431fe4809b602251ad5b27254 : SUCCESS in 8m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/11f01bb6eb594ba5aeadc875457814d8 : SUCCESS in 51m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8cc50e758ee74a028af07b9ee3da6842 : SUCCESS in 1h 10m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/21080cf30f6d4ad08b46c32e0287481e : SUCCESS in 1h 49m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c29fc73eeb99406a9d24c917278fa1b6 : SUCCESS in 1h 19m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d40b072c761b4faaa4e18c15ec0f53b5 : SUCCESS in 1h 17m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04a5285f2f854de9a7a8257369b9b317 : SUCCESS in 1h 10m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a7dc4bca3ed47f0a5ae31fe19e67767 : SUCCESS in 1h 08m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/468f86cf4418414bb8ea1c4a63b5aec0 : SUCCESS in 1h 37m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/46842a86a6b04945b3b361e99ca0886c : SUCCESS in 2h 11m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ffcdb462b0474b12a9b8f86f1136676a : FAILURE in 1h 27m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bba4b1e9706c4ee69276db6ab937b347 : SUCCESS in 1h 17m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9341b0831d364d18ab3d6d7bee07e1ed : SUCCESS in 1h 05m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/311f2367d41f4821879b985ed0b93c4d : SUCCESS in 2h 26m 23s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9787a0fd13d04d4a9913791a3517f678 : SUCCESS in 1h 36m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbe064b1c86448ca8344b22958a5017d : SUCCESS in 59m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d5a0f56bcec448da926dca76311abc49 : SUCCESS in 1h 12m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a73b2071577145288ee10d67102694a1 : SUCCESS in 47m 55s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38e742ff5c784d348a0f382ea2852e6b : SUCCESS in 1h 06m 54s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb0cf3c10dbe4e7592af1f17b2ebb354 : FAILURE in 1h 07m 36s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9f71599e7984a33bc96c9c9da841062 : FAILURE in 1h 57m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/36e1ef5c44184883807c0e87f1dc1e73 : FAILURE in 2h 45m 26s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/98c68afd67fb49e3a4627ab30167f954 : FAILURE in 2h 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cebff9aea76af39b1beab33b0a65d2bea581a962","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 17:04:43.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/13 : SUCCESS in 2h 03m 12s","accounts_in_message":[],"_revision_number":13},{"id":"6fa6abbce13abed45cc02a4d4f7b549569fe7869","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-06 09:54:15.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"706d16d1d7e9951ea018158f135e23add8793877","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-06 10:48:23.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* Neutron-ML2-MLNX http://13.74.249.42/refs/changes/22/687922/14/Neutron-ML2-MLNX : FAILURE\n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/14/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/22/687922/14/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"e55837b357e1dec71f66745c36580747762c1708","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-06 11:27:25.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/22/687922/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"406a75e9e2c9cd59b53295e4cd2264e5f10bae45","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-06 11:42:38.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/14 : SUCCESS in 1h 47m 50s","accounts_in_message":[],"_revision_number":14},{"id":"803fc98a5eb95713dbc4e4ccd23a37a11c2d2f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 13:18:51.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8a267397a354b05a16d16a3b225004e : SUCCESS in 58m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee20abd563514f468721156e59ec8ff0 : SUCCESS in 43m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ab080c84c44b40a793f2ecc438799ed1 : FAILURE in 1h 53m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e1334b8c9d6948e9901d02500236c750 : FAILURE in 1h 14m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6be8619802ff4a57883759a87eabedd3 : FAILURE in 56m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f755d1437aad4eb9b4ab2ff3f0eaff06 : FAILURE in 57m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cffef74c9e944069a9952936771494a : SUCCESS in 42m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a144f327f624e7eb054bfda84c3ac35 : SUCCESS in 22m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613f664eb769453f8ec350de312acb51 : SUCCESS in 12m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a9c1c7356084e4f8d40497afd70958c : SUCCESS in 34m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f022e89a0d894f1e8924ad86712bffd6 : SUCCESS in 28m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b84259b75b477e870fe86f16102c4d : SUCCESS in 14m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6b6a13d5832243699897fb408ee6e918 : SUCCESS in 51m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3113870e47a543e2901c8bf6e4913253 : SUCCESS in 1h 27m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9277c840114c482ea8bcea4aa5b67489 : SUCCESS in 1h 54m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/057bb6bf295c46a2817e780a25b53d47 : SUCCESS in 1h 39m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/77621b0fc4d740bd9297d29b523f1326 : SUCCESS in 1h 22m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b06a047c510544569ac9abccc3c796dd : SUCCESS in 1h 20m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/46d509e7daae4012a7605e6bd6d32ccd : SUCCESS in 1h 47m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cfd4f8f8d5054b299dcbb2e763a29373 : SUCCESS in 1h 13m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0735f0625e1427195e999b164fd39e0 : SUCCESS in 1h 48m 01s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ae1efba374b411e9b3cc6828e6629f9 : SUCCESS in 2h 00m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f7903509187e44c3872868a7f3cf15fe : SUCCESS in 1h 40m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1ffad160de594f3a9bacfc9f138e5a4d : SUCCESS in 1h 03m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0db51ba75f44fe98ab5b6160b15a36a : SUCCESS in 2h 24m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/962fef2bc16c4da1888c73276c5a5317 : FAILURE in 30m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a643c69f6ecc4679a494e31269c9f8f0 : SUCCESS in 1h 02m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8ee681f0f3a143d396c57a7058d39845 : SUCCESS in 1h 17m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/259d8a4377b040e595c8ab7a1ef902b9 : SUCCESS in 1h 13m 34s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b74be81c2ae0475f8d745ceb4b662c69 : SUCCESS in 1h 18m 03s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7531c7171f424c90aebb28874b3f6ed8 : SUCCESS in 1h 17m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/87ba7320e25343c6a08b64e8d8be97d1 : FAILURE in 1h 28m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8ed9bf4fd3e2409b87825bdbcf9c27e8 : FAILURE in 2h 38m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/860672308f2746e98cbfdea4703bd263 : FAILURE in 2h 03m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"65ba7ad1cc07bc5adc318379f3d4df0889cc7506","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-07 12:58:01.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"1da6a0c1458ae4c13280577cad7052a78ab13288","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-07 14:39:35.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"b53b665eb63cc27df1ed5ad15fed1131c5aeb0c1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-07 14:40:49.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"c305c0eea353a49669265c2b1ebe72a8800a6ad1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-07 16:43:39.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/15 : FAILURE in 2h 02m 11s","accounts_in_message":[],"_revision_number":15},{"id":"eeefc66987d55e47a52982570d39710998e1fbb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:26:55.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e0dcb6d02adc4aed8b3aad6420ad71cf : SUCCESS in 1h 00m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55bdeabfa5264ae69d7ea8fa5788a5a8 : SUCCESS in 40m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ea2396b37d2c4f29b3eea70d44bf406e : FAILURE in 1h 37m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/489a137d6f8a492baa3b98c40654d56f : FAILURE in 1h 08m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b390ebd621a7422e92b5af4a2c29f5ad : FAILURE in 58m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e6c3d23b3b446c99480eed442d7f069 : FAILURE in 59m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d26cf63207b94d93b91b61784729ec04 : SUCCESS in 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/025ca686581249d8a4124bace21a4f3b : SUCCESS in 26m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642e97f4de5049558597d34dce438621 : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b18cdce96ee14d34b448a95be927c3f4 : SUCCESS in 22m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63e373f39e3f4257b692d2460a630bb1 : SUCCESS in 24m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1720ae95a849e3bc973fc7d1fed1a8 : SUCCESS in 11m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c328f376b114a139f84850a1885c957 : SUCCESS in 44m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/af7c38a45a3b4940b42bbb09bf6a195a : SUCCESS in 1h 02m 02s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/09b338c596b74f50a5ba0f7663569125 : SUCCESS in 2h 05m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ccdf0bc1954141e487ffa51dbfae11f9 : SUCCESS in 1h 28m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0196526406f1489c88583c89e8644751 : SUCCESS in 1h 20m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5df1e194a278488fb86bee8b74e53846 : SUCCESS in 1h 10m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f03f8828fbde4cc0959985a3965c8aab : SUCCESS in 1h 17m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/eaf554a9a3d6422ebfbcdd667beb1949 : FAILURE in 58m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/00be751895cf44e38124ab765531e40e : SUCCESS in 1h 45m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/51c0fcc272ed4033b0f894f2659a999a : SUCCESS in 1h 53m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/73fd7e2fdc6140dfb3f43db4a768ed8a : SUCCESS in 1h 13m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/681a9df02d1142eb80dcee087a267c99 : SUCCESS in 59m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c0d68c9d929343f6b898c2fdcae73987 : SUCCESS in 2h 14m 14s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/14bbfdb3522a498d9d6b1c93cb3e8fb0 : FAILURE in 26m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/521b6af5420a4eae8181811ffc550822 : SUCCESS in 1h 05m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/759fd94074b04b99a44a058690144ed3 : SUCCESS in 1h 11m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f625b34a6154603801afe29593bfb25 : SUCCESS in 40m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d89b38c509a400aadf1c43458e181ab : SUCCESS in 1h 02m 03s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb471ac364fd4cc48a6dc446cf7d3327 : SUCCESS in 1h 14m 32s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2379f0c553794fe6a0358dba0639f72b : FAILURE in 1h 26m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5b4b41ccbe6144d08a061bfd3d5e4d86 : FAILURE in 2h 42m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fba795e129b048c6b5d0346b2c034b1e : FAILURE in 1h 58m 41s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c785b6e27d0fc03a5fafce48a3aa46f771d7e7df","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-07 18:02:09.000000000","message":"Patch Set 15:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/22/687922/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"7c5dee3adf0ddbce1edd9624620b5545e9fa8461","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-08 09:17:22.000000000","message":"Patch Set 15: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"007d4eaf8fa85b3a3deb24bfcff0648e91f38fb8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-08 11:02:51.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/15 : SUCCESS in 1h 45m 00s","accounts_in_message":[],"_revision_number":15},{"id":"ec41f77a22ec030b04d3f4a458150f4c583de3c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 12:12:25.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/66861668d87844418c7fb50fb3673464 : SUCCESS in 54m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3a1198458eab42a1bfd3fda42d8ceec6 : SUCCESS in 34m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a1ccc8d3cb8a4ab382abcc2a7cc8e349 : FAILURE in 2h 53m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/81f6a562a40b47eabe4fc1f0e57599e7 : SUCCESS in 1h 44m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/50fe98a437f34ed4b073eeb5fb982128 : SUCCESS in 1h 30m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b91c41f3350a46758746e3f9a2687f2c : SUCCESS in 1h 18m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3cfff859fe24720b886f436014aba1f : SUCCESS in 46m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec92058b56b7407dbcc88036ab2f8700 : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b03d5d50af4c62a56581fd683edf74 : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8063b897ace4eac8a763a8a850fc40b : SUCCESS in 22m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f09659f4e014421687138eefb9b781ba : SUCCESS in 36m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e576ff11974513aab1502ee914ac15 : SUCCESS in 9m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/928d45e32a054342a2cf621a272ecbea : TIMED_OUT in 2h 12m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7e9cb6e22cac4985b1ac3c52ed415b44 : SUCCESS in 1h 04m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d3c0372b51314cbbb1c3ef738a2a4b14 : SUCCESS in 2h 15m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/31984cb150d847809dd3cfebc361c64d : SUCCESS in 1h 38m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0fd8ebfd564440cb9098b12eb4b035e7 : SUCCESS in 1h 37m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2a35290750142fab3bfe14711fb900e : SUCCESS in 1h 27m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/31ecf0a6afa941c48e2d2c388b9c32ce : SUCCESS in 1h 07m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6767665bd17c49c49234c745335eab83 : SUCCESS in 1h 19m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3303c2d15e0a429daf8144d44200c632 : SUCCESS in 1h 33m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5d13147964e4803bbcab04d973caadb : SUCCESS in 2h 15m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c96d79e25fba485eac451208e0147da9 : SUCCESS in 1h 33m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a67f80797b0a4c1aad6d45c4575ee4c4 : SUCCESS in 59m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b886a20aa26b49bbaf410e38fc20b829 : SUCCESS in 2h 13m 08s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1855b3096eb5475dbf17e754b50bbb16 : FAILURE in 25m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/451b6f2af2d24e48a773436f904d7248 : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cb309ce35dd94023850b06a4903e8679 : SUCCESS in 1h 16m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fec79dae2a84cf0a3bcbe8a416f3e85 : SUCCESS in 38m 22s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd12eaa1ac6b4548881ee727197cb4a6 : SUCCESS in 1h 07m 44s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/228141b4908a4ea98262169dcb75713a : SUCCESS in 1h 08m 51s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/59f6881e4f774aada0fe69709743b57d : FAILURE in 1h 13m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f39ecdbfae147bbbfc93bd3fe03965e : FAILURE in 2h 31m 00s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6056c465c8c4420db97d7d6c09810f8e : FAILURE in 2h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8e93bb537298a36d1b003eea9ce4c4a8e1a5c9e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-10 14:16:19.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"710a1161cf94384d2d5aca93a1697126f51e989d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-10 16:16:00.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/15 : SUCCESS in 1h 59m 12s","accounts_in_message":[],"_revision_number":15},{"id":"309730f5b34ed46d793c30e5a2b4455a25145c79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 17:45:39.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/837f281b25ea4f1dbe230c754992bcef : SUCCESS in 1h 06m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f0c22322883f4f34bf3e0f6411ec24df : SUCCESS in 43m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3bc88459f15b4fae89e2f118bf75220e : FAILURE in 1h 50m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/96c68badff8e4a14b6ea7ca6b23c35c2 : SUCCESS in 1h 16m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6ef5e9398bcc484994d71a834577a215 : SUCCESS in 1h 35m 47s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c0ccd939209e434aa8fc87e6ef43b016 : SUCCESS in 1h 12m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5728f86fc755467698519cb797475b1e : SUCCESS in 38m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cf349c07a5d45f28cb48426be6505f1 : SUCCESS in 36m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2f10f74bae4a7aa689e58959171d30 : SUCCESS in 16m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/946de71b939840e0b8a8647334cdd4f1 : SUCCESS in 28m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59fb6cb20ead47968886a575c229ed88 : SUCCESS in 32m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad408024981b432786d02bf8c751ea45 : SUCCESS in 8m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/473dc64f52c441259b19e116bc48da7d : SUCCESS in 50m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2a38bd20ca0f4b9db2c22a123d8eea32 : SUCCESS in 1h 22m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/78a9b0fe568e4f02bfa6b761369963b6 : SUCCESS in 2h 10m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1b1f845f10374c8ab5c3986d9773fb39 : SUCCESS in 1h 22m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/88c082a86a5a492eafa749e4a5b94d9b : SUCCESS in 1h 15m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c587f951f0214a91b5f6eb3726906296 : SUCCESS in 1h 05m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41bf73ac31694d3984fcd01c8462418b : SUCCESS in 1h 08m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/65a750cff7b54f83b720e2d15f594ca7 : SUCCESS in 1h 09m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d9b74d077588494bbc4cee5824fa1f49 : SUCCESS in 2h 08m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f42905bf6d104e809be9a603f7d7c501 : SUCCESS in 1h 57m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3215e0972cea442f953f9c10fb88014e : SUCCESS in 1h 13m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6cfb7120a41c4ce08317db4ff9b7db14 : SUCCESS in 49m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aae24cee611d4476ab23b6555770c3e0 : SUCCESS in 2h 15m 05s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1b648e59f0c142309a4f5b15f9a30da4 : FAILURE in 19m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76a96621d84c4dffa8d61fa6ed383fdd : SUCCESS in 57m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/35e842be5c72408db541c9b3f83f39b0 : SUCCESS in 1h 13m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d64f344cdb8c439797c288042f4977db : SUCCESS in 43m 00s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8c710818cc04d298ffe59c902b1c1ef : SUCCESS in 1h 02m 29s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8268528943ce451b849eab8bcf334d26 : SUCCESS in 1h 11m 41s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c00148e819224df69b177097ff113eaf : FAILURE in 1h 20m 30s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/48ae0e02029741ccb803aa9200368794 : FAILURE in 2h 41m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/20ff4dfa2c544baf809b20b5c56e1463 : FAILURE in 2h 03m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"96b43eded2753387215567cf39492e3a71c290d9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-20 01:30:31.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"30b702b210f643de8b6cddeb9e098aa4c8da8bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 01:31:17.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"b40a11d2d0242108dd7353f949ce5480713c90f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 09:00:29.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8ee01dc1e9884acb820ba2f5aaf07012 : SUCCESS in 1h 02m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c883651560034cc5a419beeef863d6f3 : SUCCESS in 1h 21m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a63a9888247e4befaa9b4d836f019334 : SUCCESS in 1h 18m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/23cf9a1923084675a0df49d918e51dbc : SUCCESS in 1h 25m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efc2d2a9b8bc45d3aa7e5399780d1ca3 : SUCCESS in 36m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51de88da21e74e10a75fa2d640cbd251 : SUCCESS in 21m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/380b9587403b430eaea66b21ff7568d3 : SUCCESS in 32m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3330b303afd445094f0b862c2f7aa32 : SUCCESS in 30m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373946a946e84868a1909930d8915ef5 : SUCCESS in 17m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b6769d62a48c4baba0a3151014381a2f : FAILURE in 56m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/74d9ad99c216450b8c869d8418a46bb5 : SUCCESS in 1h 12m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/96e32c5c69004156a62539b4b64266ea : SUCCESS in 1h 43m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fa607035b07e445083cc83129169836d : SUCCESS in 1h 21m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b160059af1ea401389d7c65c20a3d5c6 : SUCCESS in 1h 17m 57s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b2b468b9739e4cf1b438f95877a04f1d : SUCCESS in 1h 22m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4337a4841b9741c298ef179576ffb956 : SUCCESS in 1h 13m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/84447c52c0e84ee98921b6730942bb51 : SUCCESS in 1h 09m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ee74dc0c5f5416ca238743bf42cbad5 : SUCCESS in 2h 28m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9092f01b3b3548c4be92561dd41fe4ae : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":15},{"id":"c03b2cc5bdad8edb7a0a1202029cda9287088770","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-21 07:55:17.000000000","message":"Patch Set 15: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"f276a2164945509078ee13783a19c5c0ec321763","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 09:40:55.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/15 : FAILURE in 1h 45m 07s","accounts_in_message":[],"_revision_number":15},{"id":"1c0cda943b33bf1fcddb2222950f3638522f257b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 10:36:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c13a458582884037ab09a3a148e73437 : SUCCESS in 46m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10ff1a4f2b264b1e8f9a996b5e0239e9 : SUCCESS in 36m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8dff8133dc5e4115a2359e3bf26d1f8e : FAILURE in 1h 32m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/eeed4f40bdf64dd3b2a81b2912609c92 : SUCCESS in 1h 08m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d3b15177c5384dabafd5f83370cc4d13 : SUCCESS in 1h 03m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04ca49b6246443a4b335e8e7ba6eb46f : SUCCESS in 57m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0db06b16f3b64744abc51b5c85c3291a : SUCCESS in 42m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1169cb0f72ad4492a0ab130123ebe197 : SUCCESS in 33m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dbbe8ea0e504e8daf61cd10683a76c8 : SUCCESS in 13m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d5211abf64f43aab72449ed8bb48dc2 : SUCCESS in 26m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9e36a4fa3f7490a8fc3b6b227f43652 : SUCCESS in 28m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198d7315cd204bf0a3ef73098b4264bb : SUCCESS in 8m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/250d4fc0748b4b5497f57a3ebf434c72 : SUCCESS in 41m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/580c5a7c742a4c8eb0ae60e635999a52 : SUCCESS in 1h 19m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/91a45d4a07634a268ca435afdf0b4969 : SUCCESS in 2h 05m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f925a3c47024910adedcecad5bb5459 : SUCCESS in 1h 15m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4c23807bf3c3466ba00cbbdbc7e9725e : SUCCESS in 1h 07m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f7b0e501d314831a4242d3977c6645e : SUCCESS in 1h 18m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bc9f5f72d3d451a9ccd28978bc73438 : SUCCESS in 1h 12m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aecd442ac60649afbb3d3272d0b06183 : SUCCESS in 1h 07m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c357332d42474936b5a6bca35f24567c : SUCCESS in 1h 52m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5aa11a90fb54711b71e79a1dea3bfca : SUCCESS in 1h 56m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ef645066c3c241f09c580d456b74aed8 : SUCCESS in 1h 07m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cc22a513a4054f52beb99f03de27042e : SUCCESS in 51m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/401cd0521e0d49b2b396eb4a4f047de5 : SUCCESS in 2h 16m 10s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/145d11c6566b4bf1ab5e849498db5945 : FAILURE in 17m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5912ee4fb0a5498ca1ab020ca2bed0d3 : SUCCESS in 58m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3311cba5a8c34255967d37490d6e6a1f : SUCCESS in 1h 08m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7ecee7d45d446b7a366c40c71fbc574 : SUCCESS in 44m 43s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/19241d3f94804e0b9fd8a1a876eeab0c : SUCCESS in 1h 04m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5ef3edf146c4d66a79fcb37a9634c8a : SUCCESS in 1h 35m 38s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/800b85102a684f3f8d2538b9a9c81a52 : FAILURE in 22m 37s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c02113be9cf24641b51c24728f9e6422 : FAILURE in 2h 38m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4269e90a99d347389801610ada1b6d73 : FAILURE in 1h 48m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"faa9322f9dd5e71ead9df4b7939049d78d14787a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 10:36:35.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"1a3417170543a6a09d2368811efa39b6087f9c0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 12:36:11.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01a0daafc31147969c309dc93f1d5842 : RETRY_LIMIT in 5m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a1821e3e44014b65afde5c21ad9554d8 : RETRY_LIMIT in 6m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ea5084157f3d46f5a38e289ab1903223 : RETRY_LIMIT in 5m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a491f96160994b758a50827fe1bbdd1d : RETRY_LIMIT in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de278fedb73f4ff4bc04ed3653bf69ec : SUCCESS in 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e0d7c0f80b4f8cb4fae97119598b6a : SUCCESS in 15m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958765d3e1d54c65bb21f54114233eef : SUCCESS in 29m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/821d8c2f40324dc9a2738d1a87e3c819 : SUCCESS in 30m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c86d717b6f64dba9a50f700e82128c6 : SUCCESS in 9m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c175c17798b743a19fdc3a2f06b04504 : RETRY_LIMIT in 5m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e0af1d47f9014dc2a9bb249bb9fdb9df : RETRY_LIMIT in 6m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/243427a40d724e889748fd585e920545 : RETRY_LIMIT in 4m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5503b96c37c345eb83c50bb9eff76bf9 : RETRY_LIMIT in 5m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/603e15e617d0497fabb97ce7bc86389a : RETRY_LIMIT in 4m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06e429fa8c3944648e2ef1ef8a815280 : RETRY_LIMIT in 5m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb417cc9e0a6499c8e984022224eed47 : RETRY_LIMIT in 7m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/67c67341acd94a19bc4ab32751f7c0af : RETRY_LIMIT in 6m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4b75e923ce34875b73c52c61415606f : RETRY_LIMIT in 11m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b164faac770454f95ae46eed5e32df3 : RETRY_LIMIT in 5m 58s","accounts_in_message":[],"_revision_number":15},{"id":"893ded63c1b0a92a864c711c27790c39ccf63a88","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 16:43:11.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"8eda6e275270d9cd286e98fe8b43eeee8179e921","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 18:31:27.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/687922/15 : SUCCESS in 1h 47m 26s","accounts_in_message":[],"_revision_number":15},{"id":"467f3464f6b18ba8bc04841a0d0f54aa2c7e3d00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:19:27.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/81b776a287394488a506c4555fd7f404 : SUCCESS in 1h 15m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8771f8690a4046779a3041b5c32e56f2 : SUCCESS in 35m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b2b4823438cf4d43b04e4cefa51b03f2 : FAILURE in 1h 33m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/79810a59a541472cb7b6075c59831df1 : SUCCESS in 1h 04m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/91f1693ec72f4a15b18ac567082582c9 : SUCCESS in 1h 17m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dcfacbff78c84bbca4c3d0aaa96145af : SUCCESS in 1h 01m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34913e6f0b5643afafaa819fbe62b16c : SUCCESS in 36m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eea268e3eed741efb74e9aebe9c31225 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb8ab0553144fc0ac3351f324bb16cb : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d037fd6027421d9df235cea6f3929d : SUCCESS in 24m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c834309ab2646f6b042a17eb7294455 : SUCCESS in 27m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2512e2f5ba9d4f6da03012c113c438bb : SUCCESS in 10m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/262012898b2f4c34ae44790088af7a23 : SUCCESS in 41m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/472cd031c28f4f0e99738893e49a26f6 : SUCCESS in 1h 13m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b8419654c50d4b5db5d8f6bc766289bb : SUCCESS in 1h 52m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f14dbe0864cd4d44bd85a115a7bef81a : SUCCESS in 1h 25m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3e6bb7cab4464d218bd18f27a248c609 : SUCCESS in 1h 14m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f468c23790b84f2aa8d5ab3c6865acfb : SUCCESS in 1h 10m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c3549d4547440fca6e5ade5d21a99b5 : SUCCESS in 1h 10m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e62b7a5bb0624e189e7274d77a3c7a2c : FAILURE in 1h 02m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eb85d435c07b44e4bf6ef358f0f1e948 : SUCCESS in 1h 39m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f6661805cd3e435b955c40b56c25d78d : SUCCESS in 1h 57m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7979c86230e049d7897888600506face : SUCCESS in 1h 16m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6aac6c3583d94008bea39922874dfa6a : SUCCESS in 54m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/657be708e46f4088affb38ccfc06cd8f : SUCCESS in 2h 20m 36s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4e75459e9f92428f85771779a00d75ed : FAILURE in 16m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9fe92030f1b403997e3a95947bfd5ca : SUCCESS in 1h 04m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/12bf80b7eabc408abe670878b40d7e68 : SUCCESS in 1h 38m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d3b7c5f9daa4626ba760ee9e36cd283 : SUCCESS in 44m 53s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac395320971d4edfb264c68c37b5c350 : SUCCESS in 1h 04m 31s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfa6c721f7ef41009f49407a5ab0d8bd : SUCCESS in 1h 21m 10s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a40fa84c6071451b8ec2407b5712b18d : FAILURE in 29m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/656522736b3348b68f1da743e0bcf07f : FAILURE in 2h 33m 06s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/38dd6e0cb4a54f86843af73556573a26 : FAILURE in 1h 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e9bc596d07b1a435c56889704e0267c4a504243c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-21 21:48:36.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"163778a36df8576dcf4733cd40ec71088e79c30c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 22:17:16.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/687922/15 : FAILURE in 28m 11s","accounts_in_message":[],"_revision_number":15},{"id":"0cca6f163d9ab7c6feb19e08798bb446f8817357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 00:36:16.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9d8d459dfc4843b9a19c36a5cc212619 : SUCCESS in 52m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b75cd17e9c8d48bc88f28655e08440e6 : SUCCESS in 35m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bce7643517034ce3bd4ac2fcd0297f2e : FAILURE in 1h 58m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/33a195b816d54523a150b0ffcfc8f77b : SUCCESS in 1h 09m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/dec6449a40b74c56b4353419d52e5e26 : SUCCESS in 59m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/774da761682b491792056ddca369c537 : SUCCESS in 1h 00m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6eb490e8a334585a029b1fd07ce8416 : SUCCESS in 42m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f15c457da2ae4e558ad16250807bfd1e : SUCCESS in 29m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339b0eced208431e9adc15d61343b54e : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e774a59063024e2dbbc2d4c21549b75f : SUCCESS in 24m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4d9457ac67c46439becdf3714cbc574 : SUCCESS in 25m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7014657cc6ad4f849d321931fcca47e5 : SUCCESS in 7m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4c315489e0d346ab8442ae17c3f14461 : SUCCESS in 45m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b5c2a694b57643a5a2c51521677f544d : SUCCESS in 1h 00m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7fead5fda51a43c795d70c4b7a2be70c : SUCCESS in 1h 57m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/877872741b38452497438358c203f6d3 : SUCCESS in 1h 11m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e383d532f46a45e2b6537eb2ead4e7c9 : SUCCESS in 1h 12m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be6411514ab744159fe0b2b5c41caf98 : SUCCESS in 1h 10m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/22fec3fcb75e462eb16f2ef625d648e1 : SUCCESS in 1h 11m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf38b0b036a3481a81608947ebac7adb : SUCCESS in 1h 04m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/23066e55d11c42cd8d9ac064430ce4b1 : SUCCESS in 1h 38m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/11323352cb9f442f96c42bf78be5fd73 : SUCCESS in 2h 35m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/56f23c748d074c3288188196dcad2135 : SUCCESS in 1h 08m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3d4bf2b5274b4176968fa28bdcad83c9 : SUCCESS in 1h 02m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0802adf333a14c1bba8871511aec07ff : SUCCESS in 2h 16m 44s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2f6c5ecf4fd04beda7913a059eb22746 : FAILURE in 17m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ab8ab644ff4411abbcc8572b28870b0 : SUCCESS in 59m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4fb1a182b3a64795b85c851a57f02d3c : SUCCESS in 1h 18m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0baf96d624f3421a9108fcabbad96c7b : SUCCESS in 42m 20s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdd38f6632534c50838bbc131ebbb319 : SUCCESS in 1h 01m 31s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cd75f93960f435c812b6ccc6dff308e : FAILURE in 1h 01m 14s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/066c31c67c454686b1726762f92c0eb3 : FAILURE in 30m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9df586d8a67841128066fcb8058cd4b7 : FAILURE in 2h 18m 48s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8d5a74416d3b4bc7aef74ca7b17e8bff : FAILURE in 1h 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"727aeab663a0bf0dae949d600d30319c893fe000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 00:36:38.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"c3d7d636bab00ce25a2ad15153e59d01fe6a7997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:36:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"d7a9757d5e224401357c21d7e58f59dfe7e0c1e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:36:24.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/73c59f9827f54a04a05729f1647c201b : SUCCESS in 1h 05m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0b9c5763b579479da9fc393d08571f55 : SUCCESS in 1h 19m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/03a1b0cb349d4671990d81c73d2646a7 : SUCCESS in 1h 12m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f58ecfc26316428982fd587961e97b78 : SUCCESS in 1h 04m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49781b80d27946e4987aad448f64d2a4 : SUCCESS in 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3d2357335e4a6b82bc315c8fedeb5d : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa31688dd8a548298668a7fe8aa82fe9 : SUCCESS in 33m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09b988ff34dc4277983c14a73d08ae24 : SUCCESS in 41m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa87ec487e243999b374ba02ea41772 : SUCCESS in 11m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/090e0a09fa524a9fa578765c27eb4efc : SUCCESS in 1h 06m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2d179755654e49319699fedb51f90c9a : SUCCESS in 1h 08m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e7066e4a344047c68154c15c3f432ae2 : SUCCESS in 1h 28m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/06ed2e72abc3489b94de892b8a64d76a : SUCCESS in 1h 23m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d8d869b5c7f486495f98350f9b90c44 : SUCCESS in 1h 25m 45s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e13cd45244f74b4f988419515475eb47 : SUCCESS in 1h 47m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba79c6db4a944e2c8a754be2145b43e5 : SUCCESS in 1h 52m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b921b3d4f8c5495f9bcb31397b24a011 : SUCCESS in 1h 13m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a556f8ab8e6443e0bee2dc53719a46a3 : SUCCESS in 2h 11m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e802d06c2dd42749da0710ed68cb854 : SUCCESS in 1h 00m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/000405296917439198785aca2b472dda : SUCCESS in 2h 54m 29s","accounts_in_message":[],"_revision_number":15},{"id":"4cf74ca9f84861c294dadbfe4b554a3f0f9a8b5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:39:01.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd95a74900b4476e960b5517cd3b6091 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"83d7eb961dc8a5d4e815bc3f110f8523a532caeb","revisions":{"6240b20e9ae1a1c8179360b88e416a20b4a8b68b":{"kind":"REWORK","_number":1,"created":"2019-10-10 14:44:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/1"}}},"commit":{"parents":[{"commit":"2ffaa40b43ceb19f1584afd3d8a9c18efeca7f6a","subject":"Merge \"ovsdb monitor: handle modified ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ffaa40b43ceb19f1584afd3d8a9c18efeca7f6a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-10 14:43:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-10 14:43:35.000000000","tz":0},"subject":"[DNM] Test remove Queue used in Qos","message":"[DNM] Test remove Queue used in Qos\n\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6240b20e9ae1a1c8179360b88e416a20b4a8b68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6240b20e9ae1a1c8179360b88e416a20b4a8b68b"}]},"branch":"refs/heads/master"},"f4dca1333442d9e2bf6286491ef31ab18cc40317":{"kind":"REWORK","_number":2,"created":"2019-10-11 11:11:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/2"}}},"commit":{"parents":[{"commit":"2ffaa40b43ceb19f1584afd3d8a9c18efeca7f6a","subject":"Merge \"ovsdb monitor: handle modified ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ffaa40b43ceb19f1584afd3d8a9c18efeca7f6a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-10 14:43:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-11 11:10:45.000000000","tz":0},"subject":"[DNM] Test remove Queue used in Qos","message":"[DNM] Test remove Queue used in Qos\n\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4dca1333442d9e2bf6286491ef31ab18cc40317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4dca1333442d9e2bf6286491ef31ab18cc40317"}]},"branch":"refs/heads/master"},"eb7060b0e9fb22f6987fde325d2974fa2da85e20":{"kind":"REWORK","_number":3,"created":"2019-10-11 15:34:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/3"}}},"commit":{"parents":[{"commit":"2ffaa40b43ceb19f1584afd3d8a9c18efeca7f6a","subject":"Merge \"ovsdb monitor: handle modified ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ffaa40b43ceb19f1584afd3d8a9c18efeca7f6a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-10 14:43:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-11 15:26:00.000000000","tz":0},"subject":"Check referential violation when deleting Queue from OVSDB","message":"Check referential violation when deleting Queue from OVSDB\n\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error should\nbe solved: the Queue is removed from the QoS registers and then deleted.\n\nThis problem is happening now in testing, where several QoS rules are\ncreated at the same time in the OVSDB. This should not happen in a\nproduction environment, where only one OVS agent must be running.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb7060b0e9fb22f6987fde325d2974fa2da85e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb7060b0e9fb22f6987fde325d2974fa2da85e20"}]},"branch":"refs/heads/master"},"af54d133a98a7f1c98a0805fb409fd48f434a98b":{"kind":"REWORK","_number":4,"created":"2019-10-15 17:11:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/4"}}},"commit":{"parents":[{"commit":"42c076004103103c1e4d8df128ae0ea0a9dfbe69","subject":"TestMinBwQoSOvs must call the correct methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42c076004103103c1e4d8df128ae0ea0a9dfbe69"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 17:10:17.000000000","tz":0},"subject":"[WIP] Check referential violation when deleting Queue from OVSDB","message":"[WIP] Check referential violation when deleting Queue from OVSDB\n\nI NEED TO CHANGE THIS DESCRIPTION!!\n\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error should\nbe solved: the Queue is removed from the QoS registers and then deleted.\n\nThis problem is happening now in testing, where several QoS rules are\ncreated at the same time in the OVSDB. This should not happen in a\nproduction environment, where only one OVS agent must be running.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af54d133a98a7f1c98a0805fb409fd48f434a98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af54d133a98a7f1c98a0805fb409fd48f434a98b"}]},"branch":"refs/heads/master"},"87790b09755084b92111f16e9b23941b82bd7aa3":{"kind":"REWORK","_number":5,"created":"2019-10-16 07:44:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/5"}}},"commit":{"parents":[{"commit":"ab5d318c89080bfd77250b7fc47e79a4ec853eed","subject":"TestMinBwQoSOvs must call the correct methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab5d318c89080bfd77250b7fc47e79a4ec853eed"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-16 07:33:31.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87790b09755084b92111f16e9b23941b82bd7aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87790b09755084b92111f16e9b23941b82bd7aa3"}]},"branch":"refs/heads/master"},"52e66b428080873f8fda0589e8b84f7c2c6cf65c":{"kind":"REWORK","_number":6,"created":"2019-10-18 08:56:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/6"}}},"commit":{"parents":[{"commit":"42872ca2cb38294c33800195395f1b572995cf62","subject":"TestMinBwQoSOvs must call the correct methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42872ca2cb38294c33800195395f1b572995cf62"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-18 08:55:21.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52e66b428080873f8fda0589e8b84f7c2c6cf65c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52e66b428080873f8fda0589e8b84f7c2c6cf65c"}]},"branch":"refs/heads/master"},"1d5058c5abd8ac04908bcffcf32395abfd1d4d8b":{"kind":"REWORK","_number":7,"created":"2019-10-21 08:28:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/7"}}},"commit":{"parents":[{"commit":"49c9393c3bc2a30262592c5806e6670a23f506f5","subject":"TestMinBwQoSOvs must call the correct methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49c9393c3bc2a30262592c5806e6670a23f506f5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-21 08:28:17.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d5058c5abd8ac04908bcffcf32395abfd1d4d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d5058c5abd8ac04908bcffcf32395abfd1d4d8b"}]},"branch":"refs/heads/master"},"f9174fb0f4ad98ec3733f3838954a3ec0dbb9d50":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-10-26 19:20:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/8"}}},"commit":{"parents":[{"commit":"c898c24168836492dd43b109609e6babbce67091","subject":"TestMinBwQoSOvs must call the correct methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c898c24168836492dd43b109609e6babbce67091"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-26 19:20:16.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9174fb0f4ad98ec3733f3838954a3ec0dbb9d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9174fb0f4ad98ec3733f3838954a3ec0dbb9d50"}]},"branch":"refs/heads/master"},"9206d4b3620175495be7da819049c4a4f63ab8dc":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-10-29 13:12:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/9"}}},"commit":{"parents":[{"commit":"80e160d72b1c47aba1634dbbae40e19089f3592c","subject":"TestMinBwQoSOvs must call the correct methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e160d72b1c47aba1634dbbae40e19089f3592c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-29 13:12:35.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9206d4b3620175495be7da819049c4a4f63ab8dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9206d4b3620175495be7da819049c4a4f63ab8dc"}]},"branch":"refs/heads/master"},"4339ce995afb7aa0a32ba5109f6a28b110b0bb8b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-11-29 18:13:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/10"}}},"commit":{"parents":[{"commit":"8375f4cc1c0001d271faa39361882599daa873fe","subject":"Merge \"[OVN] Disable ovn jobs temporary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8375f4cc1c0001d271faa39361882599daa873fe"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-29 18:13:10.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4339ce995afb7aa0a32ba5109f6a28b110b0bb8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4339ce995afb7aa0a32ba5109f6a28b110b0bb8b"}]},"branch":"refs/heads/master"},"d53e146cb41597303e05c03021b6617122dd654e":{"kind":"REWORK","_number":11,"created":"2019-12-11 13:48:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/11"}}},"commit":{"parents":[{"commit":"3ad5ad02c94ac0359956db9aa9c3314ebd9651d3","subject":"Merge \"[OVN] Move OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ad5ad02c94ac0359956db9aa9c3314ebd9651d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-11 13:48:11.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d53e146cb41597303e05c03021b6617122dd654e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d53e146cb41597303e05c03021b6617122dd654e"}]},"branch":"refs/heads/master"},"d4f5a34ea9defc3a3793fea74dd1ed1d80ffe298":{"kind":"REWORK","_number":12,"created":"2019-12-11 15:42:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/12"}}},"commit":{"parents":[{"commit":"3ad5ad02c94ac0359956db9aa9c3314ebd9651d3","subject":"Merge \"[OVN] Move OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ad5ad02c94ac0359956db9aa9c3314ebd9651d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-11 15:39:32.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4f5a34ea9defc3a3793fea74dd1ed1d80ffe298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4f5a34ea9defc3a3793fea74dd1ed1d80ffe298"}]},"branch":"refs/heads/master"},"fabb0582547da533264c9f93dcb64ec12dd7e32b":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-12-12 09:18:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/13"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 09:18:24.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fabb0582547da533264c9f93dcb64ec12dd7e32b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fabb0582547da533264c9f93dcb64ec12dd7e32b"}]},"branch":"refs/heads/master"},"768c52c73ce28e9ca1d4df62c448d56dadabcef8":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-01-06 09:54:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/14"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-06 09:54:15.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/768c52c73ce28e9ca1d4df62c448d56dadabcef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/768c52c73ce28e9ca1d4df62c448d56dadabcef8"}]},"branch":"refs/heads/master"},"83d7eb961dc8a5d4e815bc3f110f8523a532caeb":{"kind":"REWORK","_number":15,"created":"2020-01-07 14:40:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/22/687922/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/687922/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/687922/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/687922/15"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-10-15 15:51:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-07 14:40:28.000000000","tz":0},"subject":"Add OVS QoS driver cache for minimum bandwidth rules","message":"Add OVS QoS driver cache for minimum bandwidth rules\n\n1. Add OVS QoS driver cache\nAdded minimum bandwidth rules cache in the OVS QoS driver. If a new\nport is detected in the integration bridge, managed by the OVS agent\ndriver, the QoS extension is applied (if configured).\n\nIf this new port does not have a QoS rule, by default the QoS\nconfiguration is reset in this port. Because this port has no Queue\nrelated, the call \"QosOVSAgentDriver.delete_minimum_bandwidth\" should\nnot execute the rule cleanup.\n\nThis QoS rule cache implementation per porti is currently used in the\nOVS QoS driver for the DSCP and the bandwitdh limit rules.\n\n2. If a Queue cannot be deleted, log the QoS registers using it\nIf a Queue is still used in a QoS register in the OVSDB, it cannot\nbe deleted. With the current design, only one QoS rule is created in\nthe database and several Queues are assigned to it.\n\nIf something external to the OVS agent is handling the Queues and by\nmistake a Queue is assigned to other QoS registers, this error is now\nlogged before raising the exception.\n\nCloses-Bug: #1845176\nChange-Id: Ia9077fc20e4ca360819a2e368c8c1f9250e5a6d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83d7eb961dc8a5d4e815bc3f110f8523a532caeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83d7eb961dc8a5d4e815bc3f110f8523a532caeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
