)]}'
{"id":"openstack%2Fneutron~688656","triplet_id":"openstack%2Fneutron~master~Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7","subject":"Check mtu on network update","status":"MERGED","created":"2019-10-15 09:03:08.000000000","updated":"2020-01-07 14:31:31.000000000","submitted":"2019-12-22 00:28:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688656-1576974529467-7a9322a4","meta_rev_id":"b867d657a827c4496dd136b60473ab8ae040b554","_number":688656,"virtual_id_number":688656,"owner":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-16 15:55:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":1,"date":"2019-12-16 16:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-19 08:22:30.000000000","_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-22 00:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2019-12-16 15:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-21 22:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-21 22:06:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 09:10:02.000000000","updated_by":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"reviewer":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"state":"REVIEWER"},{"updated":"2019-10-15 09:10:29.000000000","updated_by":{"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},"reviewer":{"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},"state":"REVIEWER"},{"updated":"2019-10-15 09:35:11.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-12 14:20:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-13 16:37:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-16 15:55:36.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-16 16:45:49.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-21 22:06:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-22 00:28:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0040054c093cd85ddb320b81c83544120cbba5ec","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-10-15 09:03:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75af4ac5cc43b26c5c6588dae041832c4e2be897","author":{"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},"date":"2019-10-15 09:09:50.000000000","message":"Patch Set 1:\n\nSeems like this issue is there at least since Queens release already.","accounts_in_message":[],"_revision_number":1},{"id":"5006e918386b89772f935163eab060760e57cd45","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-10-15 09:10:02.000000000","message":"Patch Set 1:\n\nIt will be removed, so I think it is unnecessary to fix it.","accounts_in_message":[],"_revision_number":1},{"id":"65a19b2ed154dae53827825b27094f78500a43c4","author":{"_account_id":30081,"name":"Glenn Van de Water","email":"glenn.van_de_water@nuagenetworks.net","username":"glennvdw"},"date":"2019-10-15 09:10:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e50d3a514e32c67c9bd0d5683bd13415c16802ce","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-10-15 09:12:16.000000000","message":"Patch Set 1:\n\nWhat will be removed? Is there a patchset for this?","accounts_in_message":[],"_revision_number":1},{"id":"ccb7053cb4604f122ec9e31d6d070284f8ac0dfe","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-15 09:32:18.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/56/688656/1/check-mlnx/Neutron-ML2-Sriov/ef4280e : FAILURE in 20m 27s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/56/688656/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/294ba94 : FAILURE in 23m 17s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"305362eb9555f9f43abcdb522f888f2e1540ca03","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 09:35:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-15/688656/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/8cc3314 : FAILURE in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0f36e22310a873a2d45e32f8dbdcf51fedbc921","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-15 10:52:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc2389e5f9db6df531e1303fc2f94438b9de3e4e","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-10-15 11:19:45.000000000","message":"Patch Set 1:\n\nI would of course be glad if another patchset would solve this problem, but the patchset you are referencing does not fix the issue at all?\nI think we have a misunderstanding as to what the problem is I am trying to fix.\nwhen doing network create test --mtu 1500 you get (In my deployment physical mtu is 1500, so 50 needed for encapsulation): Invalid input for operation: Requested MTU is too big, maximum is 1450.\nWhen doing network set test --mtu 1500 you do not get any error. \nthis patchset adds the validation.","accounts_in_message":[],"_revision_number":1},{"id":"fede37cae5e720c6140141f03ade26f96d3f1927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 11:30:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/903f3d460b334a5786c98020fe9a2760 : SUCCESS in 48m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c922d3eced1440469ecdb29a30a1ca00 : SUCCESS in 35m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/09f98799f7cb471383e15d92c44cbbbf : FAILURE in 1h 38m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/652abb848f4f4b1c84ca124704ad4eb7 : SUCCESS in 1h 15m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1cba4dc8ca4a4d0c9334eefd4a32c47f : SUCCESS in 51m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8ed31854ba7948efa5c2b848555fb76a : SUCCESS in 1h 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9110942b6aec4687a13c5b4ff718084d : SUCCESS in 27m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f2bffea33174835b04470e451d131be : SUCCESS in 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4dafae64784a83917bbc1b126554ea : SUCCESS in 11m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2f318db4b78432d9591f06137fbcdb8 : SUCCESS in 21m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/532a03713594467c9ad2a95912ac61f6 : SUCCESS in 20m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe6a36b032b64203ab85a936b2bd8a0f : SUCCESS in 22m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db177c8f5c8d4fd78a0dc61e0cbccbbc : SUCCESS in 6m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/304c82b52afc40bc80506183588f3df9 : SUCCESS in 1h 01m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fc3fe55f93754474b2d1c6556cf0537e : SUCCESS in 1h 12m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bba44aa42a9f43fc879ef7a163e13c24 : SUCCESS in 38m 04s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/947cc059888a4239a6a2fdd14d712756 : SUCCESS in 45m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8d6efdaa174e47b2aa4fef4e31ebbfb1 : FAILURE in 1h 06m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0de88b2274cf433b9a88bd6b7cc31568 : SUCCESS in 1h 56m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d580d1bf96d946189893933b853b9b51 : SUCCESS in 1h 21m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/87642a8ff8d944c4ab9d8e9cd8ffb3f3 : SUCCESS in 1h 32m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0795f6c15f6441f0bb282fd00c4a23b1 : SUCCESS in 1h 13m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/48caf669df7e46e2bbb75d48fd9314f9 : SUCCESS in 1h 12m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/387cdfba77ba4e08af94534fa4f75094 : SUCCESS in 1h 40m 10s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2433412bc5024e3289dde180d77432b2 : SUCCESS in 1h 48m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/80b23233fb0e46ba98afbd4cf289118e : SUCCESS in 1h 15m 29s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/890348a28cc7427f93f6dc1cbcb254b9 : FAILURE in 1h 16m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/db295d32399540a1852e93e0b8d8af55 : SUCCESS in 58m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4839a9d6d504cbeb9c0e87414f0078b : SUCCESS in 52m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ec56246d39642b4b897a1290ecb8b7c : SUCCESS in 2h 17m 23s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0a3154ff06f2448ba9b343f794700ea3 : SUCCESS in 2h 08m 00s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3fcd419f169e4a1588238fe41926ebca : FAILURE in 1h 25m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c92ccff648640e09c27c7dc152f8afe : SUCCESS in 1h 15m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/72315a85b9e146579624848f18d975fa : SUCCESS in 1h 27m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d847e0b06ad40269491f11251e72b04 : SUCCESS in 41m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31b75657029e4b3b9c88c1a0bc9ea98b : FAILURE in 1h 13m 09s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3deaaebd8dda4054b46f3c729177b87d : SUCCESS in 1h 11m 44s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/79612f10cd714f8da6dae212d784516f : SUCCESS in 2h 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8563a04e911f5ce390c1d4a1f1dd2e8198425201","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-15 14:02:58.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab48804c344f9e4541cdeb261c6873917e8bec5a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-15 15:21:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/688656/1 : SUCCESS in 1h 56m 19s","accounts_in_message":[],"_revision_number":1},{"id":"da6fc1376468ef5cf21cf5521823217d8ea6d69a","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-10-16 07:47:43.000000000","message":"Patch Set 1:\n\nI will work on that if I get we get to the consensus the intent of the fix is good :)","accounts_in_message":[],"_revision_number":1},{"id":"1f84bda68d1e2b0a0444dbb88bbea2e8dad5beb9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:37:15.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"427041fafd2eb7c8c0a8e4fc98dfa95901faab27","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-11 10:26:04.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9ddad0b45753744ddf31c19c6f0d190e1d5953b5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-11 10:26:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc04ed68fb122828d2651e1175d0be8bb9827ed4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-11 10:26:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"23fa932f3c2306675e7f41ab43b89b108ac6a24e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-11 11:31:25.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/56/688656/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6b9f093f15a028fa91244c211ce2f8dfc134366f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-11 11:57:57.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/56/688656/2/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/56/688656/2/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"44ce6cf848e9b9d324e99d116b9d1a5fa4c12091","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-11 12:15:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/688656/2 : SUCCESS in 1h 48m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9c22c3b070a8e32079ee13bb08ff1c5e61478100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 13:08:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/887175f950e54def992b37bb338ab128 : SUCCESS in 44m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/85f40325db254a88b7521cbc78bd6452 : SUCCESS in 32m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8d1db36a46104e0fb740cb390a9069ba : FAILURE in 1h 28m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/195f0cd9eb494e9fb476b2d394dae9ad : SUCCESS in 51m 38s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a026fc249ab94055bd31a8a6e1746679 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8d27c1908194da59e9f9c0657b3b004 : SUCCESS in 52m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11710adcb29640658042a3b759998dd5 : SUCCESS in 30m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/615046b27df24d449662020f84545096 : SUCCESS in 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91e8645c3d3496a906e8764df3b0c38 : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6c2e16b4c5640f888d744cd946d7389 : SUCCESS in 21m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4b958efe1024b9f918230294cd72129 : SUCCESS in 26m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c262e942da7461bb81a1907aada203c : SUCCESS in 8m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95bfb81b278f43899498c07b0dbac800 : SUCCESS in 48m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/34f29f68761e456c9310c66b3b306da6 : FAILURE in 1h 04m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fbfead81c03d4446b757ec4e019a4df6 : SUCCESS in 2h 04m 25s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0b3509cb35814e65878a28e8aecf6507 : SUCCESS in 1h 20m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/11025fda6c164ad5ae4d23d97b226ffd : SUCCESS in 1h 19m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/39fcaa8c8f56430e970f8166a4199f7c : SUCCESS in 1h 13m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9ea17022ff4447fa8eb9da01cf8d92c : SUCCESS in 1h 07m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a72fbeb0fb0341498cf75a09ac0d21b6 : SUCCESS in 1h 16m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a97f66a9eb4849beaa9fccd6475055d2 : SUCCESS in 1h 44m 34s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0662751317394590af24bdbd15b269cb : SUCCESS in 1h 59m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/afe7bc6d81854737937fce9003c0ca2f : SUCCESS in 1h 35m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8dec054ca086413c83461faf20bdd1a5 : SUCCESS in 57m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c3fbfe2020e43e7824000f85c2fd17a : SUCCESS in 2h 17m 19s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/0117d3f04dad49ae98aafeeef9dd8fc2 : SUCCESS in 1h 34m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffb8d12082f94022aff10720a0596093 : SUCCESS in 1h 02m 59s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b4f9645abb8c4b58a32b404ab45fa46b : SUCCESS in 1h 15m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77ae50fec605453893e670cdcad55eb5 : SUCCESS in 53m 57s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e4fd4ac28b649feb37f257991ad7e2d : FAILURE in 1h 07m 42s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d11a1f0bcfd74c5183e1947848faa3c7 : SUCCESS in 1h 32m 47s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d8ed6d24a9f40b38b8fd6910b787be2 : FAILURE in 1h 48m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d65fa9770b764734aa2653b92368dc0a : FAILURE in 2h 35m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9834fed63f804957a39fcdcab8a4e98e : FAILURE in 1h 54m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3dab57f8c44cfa56154b70b58dd908d5d5741a1d","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-12 13:54:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"487028163f77c1034b464fe72c5c007d437f8a3e","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-12 13:56:01.000000000","message":"Patch Set 3:\n\nThis should do the trick :)\nLet me know if there is anything else","accounts_in_message":[],"_revision_number":3},{"id":"ec08c4731473a5902754320006b2aaa3b112f162","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 14:00:01.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0a485ad8d479bf0c8a1d57405851f46783992027","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-12 14:01:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"406bf81f886fc7865b4ee3356f8f51e531418572","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-12 14:17:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fd3ea9604b68fd91f9c3afc7a52d90a5a400f210","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 14:19:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"19682a1806fcb5eb89a7b79b33b6d85991aa7b60","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 14:20:32.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e (1 comment)\n\nI didn\u0027t see PS4.","accounts_in_message":[],"_revision_number":4},{"id":"d31f2ded68ef412cfa788e4a990c49875d01c24e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-12 15:49:24.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/56/688656/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"dd768494c81c432675a44521bee86fb066b50e43","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-12 17:04:13.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/56/688656/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"4bbc9e0db12cd536daaf6e97ff4a8361cf514eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:14:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e8546fed0d3b4d5d9d5e251fb64393d6 : SUCCESS in 54m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07e2f096385c47b883025cfcfdb94bc8 : SUCCESS in 35m 42s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/21c6c78f2dea442ba90f634b3607f8ef : FAILURE in 1h 39m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/01ba627d006043e0b7b9c583a2ab4679 : SUCCESS in 1h 15m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b183e43b915457a964ffe0a5665de14 : SUCCESS in 1h 14m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5988fc516ba54e9fa11ed51bae259ea2 : SUCCESS in 1h 16m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aef9383622d4e45872d65ad3ef52c2b : SUCCESS in 31m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e726f97a60c24898805af568dd4b2a32 : SUCCESS in 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e609d9a82924334ae78ca66d0aa30f0 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da7e5bd83db04b53bc694125cdf3f617 : SUCCESS in 23m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ab1043e22094c33aced4998f640eb7e : SUCCESS in 23m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefd2e989ef64e50847d38f7100907f0 : SUCCESS in 10m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/543c456ef2814cb2911a9cba7fc9ec95 : SUCCESS in 55m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/900f4343c32d460f865e09310876cde3 : FAILURE in 1h 09m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c3677be81a9b42d79a1bcb03515b5c38 : SUCCESS in 2h 03m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5e5d6ffb934f45e1a21e4cffebd3fb47 : SUCCESS in 1h 16m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3f803ddf78874acf8fda00d7c75afdf6 : SUCCESS in 1h 29m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab241b8f96864cc299ef905a703c44ab : SUCCESS in 1h 14m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/589e8a2e21254ceaafcfd2dff588a285 : SUCCESS in 1h 15m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3271ac47ebab48729548840ec152c2bc : SUCCESS in 1h 17m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/053d40589b784c2c84461cc25d7beeb9 : SUCCESS in 1h 41m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d1d73804ea834bbba0c726cb01ce6c70 : FAILURE in 1h 32m 03s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3bfa6c5f23af4a3b8f67650b7abfadef : SUCCESS in 1h 45m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3b63148ddb7e42978bba80be944b4b60 : SUCCESS in 1h 05m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2626c0645cc84b999ba23b815dff73d7 : SUCCESS in 2h 27m 39s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/42b64fdb0e3a4cfebd6462db345fa0df : SUCCESS in 1h 29m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/176734eec5eb465fbef176545875c3f6 : SUCCESS in 1h 16m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/62a9c55ae8d64514b6efa72adec10901 : SUCCESS in 1h 14m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5985bc069b3145b2a45b101471a22a7c : SUCCESS in 52m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/13fa545b58184515961ea27859930a94 : SUCCESS in 1h 21m 49s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b78a732528fa46ba98c9415504a24a59 : SUCCESS in 1h 27m 03s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/81da7a7f2cf349a9b56a563e24802d46 : FAILURE in 2h 24m 02s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/47f9386ce16a490bb273eb1da8ee956b : FAILURE in 2h 50m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4f43d60407504820bd6dff83820b026d : FAILURE in 1h 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e8f1300fea39c6d61436ef3fa1a12c2bd0fd800","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-12 17:31:54.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/56/688656/4/Neutron-ML2-Sriov : FAILURE\n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/56/688656/4/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7de8e293648bb1862380e976678acb62cc6780be","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 18:47:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/688656/4 : SUCCESS in 2h 02m 12s","accounts_in_message":[],"_revision_number":4},{"id":"fe2f567a29a69dd1a6df76f98ad0749afdc9ebe7","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-13 10:22:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6a461d6c0ac81e3f263d47ed21379678b66b17d3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-13 12:21:33.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/688656/4 : FAILURE in 1h 55m 45s","accounts_in_message":[],"_revision_number":4},{"id":"317908c4f59872b9015cb9f3eb461f96934345e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:03:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cfec42569bac4b52bee6cd1f92756db0 : SUCCESS in 51m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5870731f5cec43e7b3ba124f11e13df0 : SUCCESS in 36m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/58ed8370369f459d904572de222a4fe0 : FAILURE in 1h 43m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/342d4736bf164ad0b1620ec106d256bf : SUCCESS in 1h 07m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c72f6df652e64b4599beb76c0a7c6b82 : SUCCESS in 57m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/514fb5aa189c4fcca62e88e497186193 : SUCCESS in 1h 02m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671a53b006824e60913b607b25f1b063 : SUCCESS in 29m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40317ea1bf5944b69bf56f6f5d51d66c : SUCCESS in 25m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51296576a08b4be0b0876c3910a936e6 : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6badad53e74046fca2c9c9f536665f0c : SUCCESS in 23m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc0b71c185fa407697f201c9d3d0f8da : SUCCESS in 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50852c33d69406dbd43628e88015a74 : SUCCESS in 9m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9d4dde4694c549ddb3154c6b062db043 : SUCCESS in 45m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0399822fcc4d45fbacf87aea74989e14 : SUCCESS in 1h 03m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1eb35389d1f54c78b603f250d71e47bb : SUCCESS in 1h 45m 00s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6b442174c7e0433f9f91a189c1105c57 : SUCCESS in 1h 13m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b5dd6bef12904484bf8e63d37abc03b6 : FAILURE in 1h 10m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51808cfd2d7049e98d60ed0d532a3c99 : SUCCESS in 1h 24m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5b4d38fbb3f447ee916fef3f612b8ee4 : SUCCESS in 1h 08m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0c0dc585cb5f42a18c7110128eb0b45a : SUCCESS in 1h 09m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a601cb9b4b6b402484de63a804788862 : SUCCESS in 1h 36m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5b35830f6984e9dab59009071aeee46 : SUCCESS in 1h 41m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9bbeb5ca49c54b6ca98376ba1067995b : SUCCESS in 1h 08m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bd084f1664874a2191c375f83ebc98c5 : SUCCESS in 59m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5fae373bea3441184c8ae321e2f13dd : SUCCESS in 2h 19m 03s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c4775b7c113a4af092531b18b714f198 : SUCCESS in 1h 15m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/429b7c09e25747b88c8dffeb46c1e2c4 : SUCCESS in 1h 14m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fbfbee47bf2849368c7931a1554a3c41 : SUCCESS in 1h 14m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27d8e0ce714c4c9a9d8ffcfd27582898 : SUCCESS in 45m 12s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ab736567fb649aaa21158b4122ff85e : SUCCESS in 1h 19m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5259ac3f994a4d5aa7f5157524c9566f : SUCCESS in 1h 15m 39s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/77f614354e3a44b4945e247230e8b7d9 : FAILURE in 1h 27m 20s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5028a075d1d44c798ac1f90cd4dcc290 : FAILURE in 2h 35m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2585b4518e754d5ebb27f3f5b3904e14 : FAILURE in 2h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f09b955430fe2846254a06fc1d52b8bd88d42e2c","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-13 14:47:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9217c341fa8820f09fffbf182b397411946af7af","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-13 16:37:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/688656/4 : SUCCESS in 1h 49m 30s","accounts_in_message":[],"_revision_number":4},{"id":"bd55e5e6bfc245aeb40542bb865c0f27875b8ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:50:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/057b140c4cbe4a7c91c23e949600e93c : SUCCESS in 1h 16m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d44b626dca454095a5a58df2d5e53dd1 : SUCCESS in 55m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9fd32ed25f5647a385083b56b1be86e7 : FAILURE in 2h 20m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8db2fc4d082c4f76ac364b6680df9809 : SUCCESS in 1h 33m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/df635f33b98949a3ad8ed6607475afb6 : FAILURE in 1h 31m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f0608d2e43444a3f9a684caf41beb403 : SUCCESS in 1h 22m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38e74993862a4848917e6b5d110b1289 : SUCCESS in 25m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62a38922b96f424cac3575ea64f918ae : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76210e7cd51748039b74e1eea11b1d0d : SUCCESS in 13m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d039af7b3e742eca7778178597ea799 : SUCCESS in 25m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77737865e5104f23a091ecd5ae51d25f : SUCCESS in 27m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac85cc54b2a418893eab51b4108bd90 : SUCCESS in 8m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9a4ff65779b7457fb5720055200d80bf : SUCCESS in 55m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/25fd53835c3d4b4fb9d8c85b6f256910 : SUCCESS in 1h 12m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/54bc4da7939e42a3af2a44a392809813 : SUCCESS in 2h 23m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c8d69826ff244ebabda12ab6e73bf90 : SUCCESS in 1h 30m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4fb9900f5dd344d382b2143cb2c72cb7 : SUCCESS in 1h 25m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31254712ce2349208d9b4863cc582ba5 : SUCCESS in 1h 10m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d2420bfac81476e86cf62f8d70383f6 : SUCCESS in 1h 25m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/29413ee9822b41b58ca3148569d20efb : SUCCESS in 1h 23m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3970abf95b14416386db00f95ec2b28c : SUCCESS in 1h 42m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a7ff236bbda40cbb9d009c09fc16272 : SUCCESS in 2h 02m 02s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09f52b99dd674eff94760e6c6dac6cad : SUCCESS in 1h 40m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6a3bbff50fa245a4b68216db8621a438 : SUCCESS in 1h 00m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5450fe8d2fb244a18868742f0ae114bb : SUCCESS in 2h 36m 40s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9a1ef7c6bab0490c91822e1b012823c6 : FAILURE in 2h 05m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3244f7c6203243df8846dd908010f8b5 : SUCCESS in 1h 14m 03s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d9c98c5149154a55a7672007912ebe28 : SUCCESS in 1h 29m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e50bc8b75ad4c309441570f8ed9b734 : SUCCESS in 46m 31s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82accdf6832140faadee1b77d0358cfd : FAILURE in 1h 08m 41s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/14ddc3b47cb5446c8ff77431d706b594 : SUCCESS in 1h 40m 16s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/87fd4e2e5e424bfe85343ca9d8279a78 : FAILURE in 2h 10m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/473137233ad74296bdb000bf1f906778 : FAILURE in 2h 54m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5b546c46dca2461ca88c9fb5de1a158f : FAILURE in 2h 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eba061245f52f12f43ff09c8458d2ac587905565","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-16 08:29:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"851c122a9cdf1992d5b753c846bf618440a0ee14","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-16 08:52:59.000000000","message":"Patch Set 4:\n\nIs this type of CI failure normal? I seem to be hitting all kinds of unstable CI issues?","accounts_in_message":[],"_revision_number":4},{"id":"bc49d2a1b8f70d4631f34e071146f17525516a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:48:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/043274f9bc754de0b077a5fbed3bbf25 : SUCCESS in 57m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f82d364c3b714f859082afadf510c486 : SUCCESS in 41m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1eccaaf53840476ca8eec73ce8b9bb62 : FAILURE in 1h 44m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8fdd6231db0c40f8aa2c5463be12801c : SUCCESS in 1h 31m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9375dc5e86f74a308de7a45bde97f6a3 : SUCCESS in 1h 19m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9be373be567e4a2eb8d6032ba5ac4de7 : SUCCESS in 1h 09m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd15884b677d4b9687743904458c90da : FAILURE in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00987254869149cb85ee6257449a487a : SUCCESS in 24m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a046c169ef49ea83e4a4d2c8578eb0 : FAILURE in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14a6170854e44033945935a3c6608100 : FAILURE in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30721cb8ca3949579e614101e0a4f018 : FAILURE in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df55b0104eb46ad88994d688c78bd55 : SUCCESS in 11m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/26006fe062aa42f8aeaa2f0e2c21b9af : SUCCESS in 50m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f864e070e73d4911bf7877c8b0bd6e32 : SUCCESS in 1h 33m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/29f421ba04f648c19a5f05b702eb65a2 : SUCCESS in 2h 10m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/73e128f2aa2c43bdb16d050b115607f5 : SUCCESS in 1h 16m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5fe01ed60fe44b0fa045054b0456c861 : SUCCESS in 1h 25m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c40ec39092e439c93b9e7eda4362a16 : SUCCESS in 1h 40m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7d829e8c7904a10b1686576c1369d17 : SUCCESS in 1h 21m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e0bba81182c4c3c90bc5e89621ae845 : SUCCESS in 1h 35m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/90920c8345704b91b6cb89031ecd2523 : SUCCESS in 1h 43m 28s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4b110456943d4ec28f572b097c0b1912 : SUCCESS in 2h 05m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/944a6974d8454bf4838ee871d7e90c66 : SUCCESS in 1h 26m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/94201ad6a6c547069178b59a9f3012d0 : SUCCESS in 58m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9869470c8031457e9e640c27b7d71fec : SUCCESS in 2h 36m 31s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4011ae9289ca4ede9e6d19b4fbcd3335 : FAILURE in 24m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff1a9564feca40278d205f5d1b91edbe : SUCCESS in 1h 05m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/42be1321036b48e5a95b71c6a4e570ca : SUCCESS in 1h 15m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/870efba499394a9788a5731787ce3b3d : SUCCESS in 51m 08s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a4ec94d5ff14eb39caed7cfe86725d0 : SUCCESS in 1h 11m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b461b5d9d6142209c998c2dc28baf4c : SUCCESS in 1h 15m 30s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b94d86da967a4671b2ec80915efde894 : FAILURE in 2h 04m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ee8e3a5d385b4094baced9ba36f3ae6c : FAILURE in 3h 07m 37s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ff125dc9908241c4a54b5859cbb0e248 : FAILURE in 2h 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b613c04fc57f4be267443453c8841c090a1acafb","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-16 15:04:47.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3ab693cc1b513c2f87c1bb6ec8321ccd96d2f683","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-16 15:11:27.000000000","message":"Patch Set 5:\n\nWe need https://review.opendev.org/#/c/699021/ merged to solve the CI issues.","accounts_in_message":[],"_revision_number":5},{"id":"130a358d5fdc902a75fb712e63e6d30a23cfcca8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-16 15:55:36.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/56/688656/5/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/56/688656/5/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c3992d393b739a2644d15d47b9ca560502e47b13","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-16 16:45:49.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/56/688656/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"b0046842c39a09e964be9d9821007bb2bb0f4cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 18:01:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cb5b74e3ce684e64be7d9e5ff556eb8d : SUCCESS in 47m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ebc1db998d042e9ab7c111a029fec19 : SUCCESS in 35m 32s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f088cf4fd1974e5290a023e09da69d3f : SUCCESS in 1h 31m 42s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1015b863172f470e921f6fa43ea8e107 : SUCCESS in 1h 04m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/443b486bad5c4bb8b13972fc77c80b26 : SUCCESS in 1h 05m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a3702d6f9264ba79f47c71c1e1949fc : SUCCESS in 47m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52a79ac9b9a4435298dd0cd98b4c7993 : FAILURE in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2adbe742fe9445a385e4b13ab448990c : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad500b53d9f24ebd84190354d02e7ce5 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/955ee6853b83437f93d2f3345e59496a : FAILURE in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fb4e8b889dc4c0ab29aa14fbdd6fd49 : FAILURE in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3605b42c3a4f3885cc4687106d002e : SUCCESS in 11m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1ec0a5dd04374857bb156ba9b424e7cb : SUCCESS in 52m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/48f05b99628f458d9a4ce9fee4ca9d53 : POST_FAILURE in 1h 22m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/41afd0f884134121855828f66b8c91de : SUCCESS in 2h 06m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/aeaaadf5c925414ea8f627cd7fe6c5d0 : SUCCESS in 1h 31m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5319e6d2839847ea901b2831a6c645ff : SUCCESS in 1h 07m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b15eb9f30a646b3b2219b552af4940e : SUCCESS in 1h 37m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0c783881e08d4f458818cc3d0863d9b6 : SUCCESS in 1h 15m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d59117724eda45e8b6fa2e91dc9eea87 : SUCCESS in 1h 18m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5a666aa126bc4aebb5701186ae66de52 : SUCCESS in 2h 09m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/72df03bd8c0b484b94f90d4599bc7206 : FAILURE in 2h 21m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15c703d5b219466998788b5ecf69fccf : SUCCESS in 1h 47m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/396cda69fcf0471da8ea8ca6a7ffdcc0 : SUCCESS in 1h 04m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/220af8054bb345978b9407ad609faa88 : SUCCESS in 2h 49m 39s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/561c26bb727d49b282aabfad5ac8fc20 : FAILURE in 34m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f523b2aaffa498aa4535514560a28e3 : SUCCESS in 1h 05m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/02efd1ce214c4efeb905675d9dcaaddb : SUCCESS in 1h 39m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cc03a8b352f140ab8d7518eeb94c6251 : SUCCESS in 50m 42s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/773beaa1bebf462b87c35df16e4014e6 : SUCCESS in 1h 09m 21s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b5fe829db494cad8bfecbd793344c2b : SUCCESS in 1h 22m 47s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f3f2978c2194d66969a11f233a42771 : FAILURE in 1h 26m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7f57f618d8f14760a7b84ba083ef72ad : FAILURE in 2h 47m 09s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c94227c58e974dbfb297bbc89abc3989 : FAILURE in 1h 47m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9cc85c789484817a1f5759dd22d6ebf656b551a0","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2019-12-19 08:22:30.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"120fafac76efe49bf66fd69b47464a7153a4b664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 11:00:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9de69b99a10841fcb448772432a7b615 : SUCCESS in 53m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/55ed2e4387f14e2587a0d4e7e1f0ebc6 : SUCCESS in 34m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/fa8ff7f7965647028b60803b4cac48fb : FAILURE in 1h 33m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0286822a97d2449ebfb2870fbb9e8dbc : SUCCESS in 1h 12m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6318981ded98413eb9e411b00ce6655a : SUCCESS in 1h 10m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53caaede5c9a4033bbf5c0c66c2a336b : SUCCESS in 1h 08m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dce178f4835441ba10f84da9de9dfd3 : SUCCESS in 34m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddf307ff5c7641d3bd7395a61139d35d : SUCCESS in 26m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb5d946e6bb47889956b58d73e2b738 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab09ab33eaca464ab50ae7dcd26338bd : SUCCESS in 24m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/367f97758631410da9dd6e39e0cb24c7 : SUCCESS in 25m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f4f6063977426d9a595c2ab46936ca : SUCCESS in 8m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fcb6d1d193c340a7aa6267424e8d94a7 : SUCCESS in 52m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/696101b901d54c3e9e0c320abb905c2f : SUCCESS in 1h 04m 47s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/54a9a2625aa6427db340e994c7439d6e : SUCCESS in 2h 01m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/806863d16d9c4655b43541b62f73fce8 : SUCCESS in 1h 20m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/69ae3003300a46ada919d617dde61719 : SUCCESS in 1h 32m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a102bf6d6cfc464795ca64e0fc8b0bcb : SUCCESS in 1h 17m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dc9a20c2c1ea40ca8ee2eb2a3fe4f3c3 : SUCCESS in 1h 12m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/41bac485f6dd42b0beca85e2ad0c5c31 : SUCCESS in 1h 21m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ac8476a6cfb04e20be90341b2e131f4f : SUCCESS in 1h 45m 00s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3ad8e8823b84ad68228abf6cba7a007 : FAILURE in 2h 14m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e90f03f6207d4df8854f082755eb1eb0 : SUCCESS in 1h 17m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ac0c9cc4cc8c4956b34d78ac46273296 : SUCCESS in 1h 03m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bd4f47bf0e94d0788997c38ae726856 : SUCCESS in 2h 20m 55s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/6e3313bf9e3146adbb10f0e974448c76 : FAILURE in 22m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/346d55dd8f824e5ab4d5b38729c036c8 : SUCCESS in 1h 15m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ded4a4f80d6e43009b7932d9ebb99a62 : SUCCESS in 1h 14m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee7c6cd6f6fd4e5ea1f0b867a194cee0 : SUCCESS in 46m 09s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9bb221db59e4b28b52570fe018dbd90 : SUCCESS in 1h 14m 46s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/23f83b56954f4b7eba9177a5e1aeeda3 : SUCCESS in 1h 30m 16s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6358a1eccab54ed5959606c4f09f6c7b : FAILURE in 1h 33m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d965cec7943b4d9f8375705cbbb9d177 : FAILURE in 2h 36m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/926dbddba7284d0caf6e472a1cafb429 : FAILURE in 1h 57m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f3d0ac49640b63be68fc680fc034e1166f3aef0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-21 22:06:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"04279f3fd720953b2ae1822bf7155102ccc7b584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 22:06:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a643e1f845ce2e20b919409d53b9243af45eac4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:28:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a57200ccdd644775a96c1e329b13e51a : SUCCESS in 43m 19s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cd6b23ebde3a4ecaa520c2c31b55a881 : SUCCESS in 1h 10m 20s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9b6a71a9003946be976b0443f0e676e0 : SUCCESS in 1h 01m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b580a9e2622446f89496e0f20adf7051 : SUCCESS in 51m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87d0e872d61e4250a515c0210575cd25 : SUCCESS in 23m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70250f8de8945afb43beb9260706a51 : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46ec01b22a5f47db8f5137f58559e533 : SUCCESS in 22m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10e83bc087a744b89666e6c2b5655389 : SUCCESS in 26m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec602226fcf64d5ea0ab6bea522dc7c0 : SUCCESS in 7m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aa6e038e284849ce8fdb7a7ac14e9e4d : SUCCESS in 46m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5f7cf7a66fd64d4cb0a15b67e7c6e381 : SUCCESS in 1h 10m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4725e8c316ca4778a513d4d39d1b6833 : SUCCESS in 1h 13m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c50b29d792c74ad986b0c39c370eb057 : SUCCESS in 1h 10m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e7ca5b0b0a464c66aa12a85025a1fa0e : SUCCESS in 1h 19m 32s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e0245b6a32af4051bd97b3c8cb4bfd73 : SUCCESS in 1h 22m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c33596281964bf4b4668d10414bef99 : SUCCESS in 1h 12m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/72309f95c673456496735eb24bb80dc3 : SUCCESS in 1h 18m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b51e1b533f4b42c6b021e48e0c82fcfd : SUCCESS in 2h 16m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/507637488514443e8c86589f2f2e96ec : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":5},{"id":"fedaf18471492e4d1027016e108800b9778792a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:28:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d74db0bcfbc86ccbea7d32ec41dcf991abebeb98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:30:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fe7408e8e14a7a987ab29349609ef7 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":5},{"id":"c8f530c0169559ebd2f236957ec9c84b132aa5fd","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-12-31 13:13:52.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8680863ae91e5be0a33039a0607802678b674836","accounts_in_message":[],"_revision_number":5},{"id":"b867d657a827c4496dd136b60473ab8ae040b554","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-07 14:31:31.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7ecd0fa1a10dc87df248095888948fcec27c1c0d","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f711090ed3d9b6c449a4fbc8ffe2eac7df6da193","revisions":{"398c0fcc24e0e6608299881ef57a653280922277":{"kind":"REWORK","_number":1,"created":"2019-10-15 09:03:08.000000000","uploader":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"ref":"refs/changes/56/688656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/688656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/688656/1"}}},"commit":{"parents":[{"commit":"6b52e361f0eae01e67b1d9be47faee36b0627efe","subject":"Fix misspell word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b52e361f0eae01e67b1d9be47faee36b0627efe"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-10-15 08:58:46.000000000","tz":120},"committer":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-10-15 09:02:59.000000000","tz":120},"subject":"Check mtu on network update","message":"Check mtu on network update\n\nWhen creating a network the mtu of the network is\nchecked against the maximum mtu value.\nThis should also be done on network update.\n\nChange-Id: Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7\nCloses-Bug: #1848152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/398c0fcc24e0e6608299881ef57a653280922277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/398c0fcc24e0e6608299881ef57a653280922277"}]},"branch":"refs/heads/master"},"833fade6233f7ccbc883a6ddb91d07ab9fdd8fd4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-11 10:26:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/56/688656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/688656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/688656/2"}}},"commit":{"parents":[{"commit":"3ad5ad02c94ac0359956db9aa9c3314ebd9651d3","subject":"Merge \"[OVN] Move OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ad5ad02c94ac0359956db9aa9c3314ebd9651d3"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-10-15 08:58:46.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-11 10:26:43.000000000","tz":0},"subject":"Check mtu on network update","message":"Check mtu on network update\n\nWhen creating a network the mtu of the network is\nchecked against the maximum mtu value.\nThis should also be done on network update.\n\nChange-Id: Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7\nCloses-Bug: #1848152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/833fade6233f7ccbc883a6ddb91d07ab9fdd8fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/833fade6233f7ccbc883a6ddb91d07ab9fdd8fd4"}]},"branch":"refs/heads/master"},"a11fe783129a59d6426042ef81cf8dba6cdffab1":{"kind":"REWORK","_number":3,"created":"2019-12-12 13:54:58.000000000","uploader":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"ref":"refs/changes/56/688656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/688656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/688656/3"}}},"commit":{"parents":[{"commit":"3ad5ad02c94ac0359956db9aa9c3314ebd9651d3","subject":"Merge \"[OVN] Move OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ad5ad02c94ac0359956db9aa9c3314ebd9651d3"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-10-15 08:58:46.000000000","tz":120},"committer":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-12-12 13:49:28.000000000","tz":60},"subject":"Check mtu on network update","message":"Check mtu on network update\n\nWhen creating a network the mtu of the network is\nchecked against the maximum mtu value.\nThis should also be done on network update.\n\nChange-Id: Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7\nCloses-Bug: #1848152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a11fe783129a59d6426042ef81cf8dba6cdffab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a11fe783129a59d6426042ef81cf8dba6cdffab1"}]},"branch":"refs/heads/master"},"1ab3e19fbd146d0b46831d322e6890895197dba3":{"kind":"REWORK","_number":4,"created":"2019-12-12 14:17:29.000000000","uploader":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"ref":"refs/changes/56/688656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/688656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/688656/4"}}},"commit":{"parents":[{"commit":"3ad5ad02c94ac0359956db9aa9c3314ebd9651d3","subject":"Merge \"[OVN] Move OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ad5ad02c94ac0359956db9aa9c3314ebd9651d3"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-10-15 08:58:46.000000000","tz":120},"committer":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-12-12 14:17:04.000000000","tz":60},"subject":"Check mtu on network update","message":"Check mtu on network update\n\nWhen creating a network the mtu of the network is\nchecked against the maximum mtu value.\nThis should also be done on network update.\n\nChange-Id: Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7\nCloses-Bug: #1848152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ab3e19fbd146d0b46831d322e6890895197dba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ab3e19fbd146d0b46831d322e6890895197dba3"}]},"branch":"refs/heads/master"},"f711090ed3d9b6c449a4fbc8ffe2eac7df6da193":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-16 15:04:47.000000000","uploader":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"ref":"refs/changes/56/688656/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/688656/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/688656/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/688656/5"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-10-15 08:58:46.000000000","tz":120},"committer":{"name":"Tom Stappaerts","email":"tom.stappaerts@nuagenetworks.net","date":"2019-12-16 15:04:47.000000000","tz":0},"subject":"Check mtu on network update","message":"Check mtu on network update\n\nWhen creating a network the mtu of the network is\nchecked against the maximum mtu value.\nThis should also be done on network update.\n\nChange-Id: Ie7ea79cb753f360bb0bca7df8484710b5d32ccf7\nCloses-Bug: #1848152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f711090ed3d9b6c449a4fbc8ffe2eac7df6da193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f711090ed3d9b6c449a4fbc8ffe2eac7df6da193"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
