)]}'
{"id":"openstack%2Fneutron~690582","triplet_id":"openstack%2Fneutron~stable%2Ftrain~Idaea33d62fa31edd7149ec916ec314438375724a","project":"openstack/neutron","branch":"stable/train","topic":"bug/1848738-stable/train","hashtags":[],"change_id":"Idaea33d62fa31edd7149ec916ec314438375724a","subject":"Fix mismatch of tags in dnsmasq options","status":"MERGED","created":"2019-10-23 14:19:07.000000000","updated":"2019-10-24 05:01:02.000000000","submitted":"2019-10-24 04:59:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690582-1571893155476-34f7ffe6","meta_rev_id":"beb5ecda55555779ba78a7b6215ff44a89733269","_number":690582,"virtual_id_number":690582,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-23 18:23:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 04:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-23 16:01:21.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-23 19:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-24 02:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-10-24 02:41:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 16:01:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-23 18:23:33.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-23 19:30:57.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-24 02:41:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-24 04:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3c62879e1becf1d2d784de94b74e442f822216c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-23 14:19:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e12b0063e68e2840fdad1deb5ce701583acc1769","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-23 16:01:21.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/690582/1 : FAILURE in 1h 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"77416587266bb99d2920c2d8b7e516ce14199598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 17:15:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/609779a8cfd14a2d8550a2d5f4e79af1 : SUCCESS in 1h 02m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a81176ad03d437bbab81dbbe78910a2 : SUCCESS in 39m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5c97236668364e789cfc919f6becd0cb : FAILURE in 1h 42m 10s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/88b4360ecef94780a78b15f1f6f85be2 : SUCCESS in 1h 05m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b52fe4a639f249aa90791e59a9eff4ec : SUCCESS in 1h 15m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5a55aaeaa2941f4ad2a94e02ed36bd8 : SUCCESS in 53m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aec36328d57348a3a9a24bdbb83b26cb : SUCCESS in 29m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/899a549426c64b2d96cbd5420963a122 : SUCCESS in 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5de603aae39d446284733e59a8c8e35f : SUCCESS in 12m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fee81008af1e4778b5f6ae62db43d0ae : SUCCESS in 23m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99c92392931b4ce8853e460ee9dca354 : SUCCESS in 21m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c26e59ec07f41bc886a08d130d06310 : SUCCESS in 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9c498b58e74719855ccf71edd37a95 : SUCCESS in 12m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8a23f81c50004b40919765a20dd8728f : SUCCESS in 1h 04m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7080844cffa74659b65ae20cdc998711 : SUCCESS in 1h 20m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6a541facc12646e5b0989fd05ecaccb7 : SUCCESS in 51m 29s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/914f826110b840ccb78f72a7c4f2e407 : SUCCESS in 50m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4afe50215bac4224bedc2de18be84790 : SUCCESS in 1h 02m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0a1584ed42c34e7e96a24394c9e6a6d6 : SUCCESS in 1h 53m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5f324c75d62d4b56b917220bc0b56f74 : SUCCESS in 1h 21m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5bcc02a85104431088f138962e2e219b : SUCCESS in 1h 22m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5135408401c8459689e39feb90b798e9 : SUCCESS in 1h 11m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/773e09ed76de4b06abccb8c3ae9cee4b : SUCCESS in 1h 25m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8dffe6a59fa44cdea2cd0527bb0d42ad : SUCCESS in 1h 54m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/91435633f7a74c3c846a59a86567067b : SUCCESS in 2h 04m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/812212761fea42d2a494f8f5c7027d5a : SUCCESS in 1h 24m 42s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/3b3e337867914fedab22d54126a06154 : SUCCESS in 1h 28m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fba541647e8e44c791bf7f539e6a0648 : SUCCESS in 59m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f2c2b1d5690449c29d0147c3e9c97edd : SUCCESS in 55m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d1e187537a1f4984a4e6cadc60d15f12 : SUCCESS in 2h 36m 04s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7d91093c146947839d1b2e49b579449e : SUCCESS in 2h 11m 37s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/00b905cb991a4a6ebc8f1542c035da0e : FAILURE in 1h 31m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ceca130ccc904db68220ee30b3262243 : SUCCESS in 1h 01m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4814740eadfc46b6bc0b1a9d029db077 : SUCCESS in 1h 19m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/027397ae090347c4bf9d84edcade3e04 : SUCCESS in 48m 07s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee96820c6d7d4d9e9ffcedd789cd7687 : SUCCESS in 1h 09m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/20cb6a0795be45e1a55f858410686008 : FAILURE in 1h 15m 48s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/733cca02544e40adb407e14ac0cce213 : FAILURE in 2h 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71fb3b40ca022b88c5b97fd488a8cdb503365843","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-23 18:23:33.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/82/690582/1/check-mlnx/Neutron-ML2-Sriov/b84e297 : SUCCESS in 1h 20m 23s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/82/690582/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/2e8593a : FAILURE in 1h 54m 45s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"cab5c1fbf80fbfeb3f4bdf66f425958776256ccf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-23 19:30:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afdec9cff2b0052821c3d7b9779b1b7c2177dc88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-24 02:41:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"125f8a88837cec8c27fb36969872bb66ca38bcab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 02:42:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ce0974d8d902310206ac0d2dc014ea24d3b7122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 04:59:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aabae89e6b5aaeb1bd87e5a07dafa2c93de2d3e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 04:59:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aea6f9a075ec48e2b044db46ee37d4cf : SUCCESS in 53m 26s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a619b9c79ce04d44a54101dc882a737f : SUCCESS in 1h 08m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/353f2e14d5374740a56c9f56d1dde95a : SUCCESS in 1h 01m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e258f0b692354673ae0eec400efeb660 : SUCCESS in 1h 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04f028a2acda4c44b7c155028e39fb49 : SUCCESS in 23m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbda001cf924560942e751c0926703a : SUCCESS in 12m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7882e4f898e4818995be2a985e34d62 : SUCCESS in 25m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45cbff2946644da39d26176d55412650 : SUCCESS in 21m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44254e1a16994e3dbacb96372a252064 : SUCCESS in 20m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df20acce8a1b4a8e988c32feca4aa61b : SUCCESS in 7m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f3b26a7a52aa4de0b30aa5aba5c74cfe : SUCCESS in 56m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ea8a50677b3d476282487b66e21121e9 : SUCCESS in 1h 11m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/047e25b792084077b031bd7478048820 : SUCCESS in 42m 11s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/cb2377146ff34401a24cc9495b56fb05 : SUCCESS in 48m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ba8022604f1845cf87f42a375b1174e7 : SUCCESS in 1h 00m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f90136a0e6cb47f288054cf7498ad32c : SUCCESS in 1h 25m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/598194ef6b0a404ca672801f7699637d : SUCCESS in 1h 16m 09s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c012a4af8c1c4a52abfbb852faf165dc : SUCCESS in 1h 10m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/566e35aa534b4c4f98a6c4b42fde8399 : SUCCESS in 1h 12m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ce7484abe714efeaef2170386bc4559 : SUCCESS in 1h 11m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5b451da58c6a44cd847381d0f5a21924 : SUCCESS in 1h 06m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bfa9f4c01ea4a6ab1da41c15e598e52 : SUCCESS in 2h 11m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44e2cbc808ea430bb0ff9d4a1f4ac2b9 : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"25d2c31f5d0ab509588a7d3647f208dc0a1e4c3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 05:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34d7d2d1e1f47c7bb4be3973cc5c232 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88f2073526c2add7e8aa07656caec4c35c8b8459","revisions":{"88f2073526c2add7e8aa07656caec4c35c8b8459":{"kind":"REWORK","_number":1,"created":"2019-10-23 14:19:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/82/690582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/690582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/690582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/690582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/690582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/690582/1"}}},"commit":{"parents":[{"commit":"bd04d1e0a597135e797440af0f24ba40028950fe","subject":"[nova][train] Docs: Remove deprecated RetryFilter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd04d1e0a597135e797440af0f24ba40028950fe"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-18 05:04:43.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-23 14:19:07.000000000","tz":0},"subject":"Fix mismatch of tags in dnsmasq options","message":"Fix mismatch of tags in dnsmasq options\n\nIn some cases (I don\u0027t know exactly how) it may happend that\nwhen new subnet, e.g. IPv6 is added to the network, subnets\ncan change their order based on uuid.\nAs before this patch we were using in dnsmasq options tags like\n\"tagN\" for subnets (where N was just number based on position of\nthe subnet in the sorted list) it could happend sometimes that\ndnsmasq ended up with mismatch of tags configured in \"dhcp-range\"\ncmd option and set in \"opts\" file. That caused problem with serving\nproper DHCP options to the vms.\n\nThis patch fixes this issue by using tags with format:\n\"subnet-\u003cuuid\u003e\" where uuid is id of the subnet. That was it\u0027s not\nbased on order of subnets in the list and will always match with tag\nconfigured in opts file for specific subnet.\n\nAs we was currently using port id as tag for \"per port\" DHCP options,\nthis patch changes that to use tags like \"port-\u003cuuid\u003e\" to make it\nconsistent with options configured \"per subnet\" and to make it easier\nto debug from where each option comes.\n\nChange-Id: Idaea33d62fa31edd7149ec916ec314438375724a\nPartial-Bug: #1848738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88f2073526c2add7e8aa07656caec4c35c8b8459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88f2073526c2add7e8aa07656caec4c35c8b8459"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
