)]}'
{"id":"openstack%2Fneutron~692285","triplet_id":"openstack%2Fneutron~master~I72082226e6b3930efb537cab9d979e5770211f72","project":"openstack/neutron","branch":"master","topic":"bug/1849976","hashtags":[],"change_id":"I72082226e6b3930efb537cab9d979e5770211f72","subject":"Move migration to proper dir and fix objects tests","status":"MERGED","created":"2019-10-31 07:39:07.000000000","updated":"2019-11-05 11:00:16.000000000","submitted":"2019-11-05 10:58:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692285-1572951512561-df6c426c","meta_rev_id":"dd75d90410f12cc28b311364078afbcd56dc5bed","_number":692285,"virtual_id_number":692285,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2019-11-05 03:41:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-11-05 04:17:23.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"date":"2019-11-04 22:54:22.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2019-11-05 03:45:32.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-05 10:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 08:01:41.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2019-11-04 17:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-04 17:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 07:09:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-11-05 08:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-11-05 08:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 09:21:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-11-02 12:26:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-04 14:26:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-11-04 22:54:22.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-05 03:41:25.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 04:17:23.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 07:09:24.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-11-05 08:01:41.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-11-05 08:43:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-05 10:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7bafed8b985447564f922163de5f76f04f6a44c","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-31 07:39:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bea1edaf00e68f8285562a562409aef2546ac3e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-31 08:10:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-31/692285/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/2f9b150 : FAILURE in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f0546d9ca2f5a8f2a88f94d0b0e11fbeddb697b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-31 09:19:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/1 : SUCCESS in 1h 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4f4f1f02c469059d7b4e75a6542a3f0578485661","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-31 09:21:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"943a43322bc409caa850dd3474ffb7e9e008979a","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-31 09:25:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"907707bdbb42ae6587d6c82415b694710778f5ad","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-31 09:26:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f70ab1395410ed155860ad583a026dd804c30955","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-31 09:27:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"342f70063ad3479b576acd94bfb56c34a8054fbb","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-31 09:59:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-10-31/692285/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/2874b1e : FAILURE in 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"32ebd3214860327631019f737eda0ca357c2a82c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-31 10:13:47.000000000","message":"Patch Set 3: Code-Review-1\n\nEven with this patch, the problem in the code persists.\n\nIn an environment with a network assigned to two agents, the constrain \"uniq_network_dhcp_agent_binding0network_id0binding_index0\" will fail.","accounts_in_message":[],"_revision_number":3},{"id":"f605bb6f2f71fac459f51f0edd87f1958c4c47a9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-10-31 10:29:24.000000000","message":"Patch Set 3: Code-Review-1\n\nThis does not match the merge result of revert patch https://review.opendev.org/#/c/691710/ and re-add patch https://review.opendev.org/#/c/691852/.","accounts_in_message":[],"_revision_number":3},{"id":"0f6749f13cf4e4ab614725ea410328094c4266b7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-10-31 11:12:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/3 : SUCCESS in 1h 44m 43s","accounts_in_message":[],"_revision_number":3},{"id":"595315ccea4d57ce72f32a74206cbe8182afd472","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-31 11:16:16.000000000","message":"Patch Set 3:\n\n@Rodolfo, binding index assignment is handled in dhcp_agent_scheduler: https://review.opendev.org/#/c/288271/28/neutron/scheduler/dhcp_agent_scheduler.py - autoincrement was only needed for transition.","accounts_in_message":[],"_revision_number":3},{"id":"3a2f12d303b1641f6eb2b07188391662a5893f27","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-10-31 11:18:48.000000000","message":"Patch Set 3:\n\n@LIU Yulong re-add patch https://review.opendev.org/#/c/691852/ will likely be abandoned (if revert patch is abandoned), why you think this patch should match some merge? Please comment issues you see wjth this patch.","accounts_in_message":[],"_revision_number":3},{"id":"de484a645f131915a02b5edc386f6697a467ca6a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-31 16:31:23.000000000","message":"Patch Set 3:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/85/692285/3/check-mlnx/Neutron-ML2-Sriov/5d44c83 : FAILURE in 1h 04m 03s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/85/692285/3/check-mlnx/Neutron-MACVTAP-ML2-Sriov/82b928d : FAILURE in 1h 11m 10s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"01062c9210d4a7ca6742811ce2bd8e2f815fbad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 18:44:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5e40e434b5af4206a6b8131469637d61 : SUCCESS in 1h 02m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b367ce17ea194aa29d5bc2a007cdf144 : SUCCESS in 49m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9d03a37604404b83b94b2bbd2b2fef49 : FAILURE in 1h 30m 33s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bbc8898b62c44e9b8f4e4d4142d6b641 : SUCCESS in 1h 19m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2e171cee34574a8b92ed63c4f63c4bee : SUCCESS in 1h 08m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b1205c68f1eb416e8ff22a4b9cc2f564 : SUCCESS in 1h 08m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bcef4bbb5df4ee9be459cf58b19f31f : SUCCESS in 27m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6129016989f9432ebe415cf713649f0a : SUCCESS in 21m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ead3f07d164709bdd796d2fd4391fb : SUCCESS in 11m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0d10faba871403ca86b8316fed6b999 : SUCCESS in 25m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/211b5eff05bd44d1b66dfca7d079722b : SUCCESS in 32m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1e365d19ab4e8fa1b3d6b342c9e0f9 : SUCCESS in 12m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/68d084420a9845acb28acee6fde39123 : SUCCESS in 51m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4f2ef488c3764b94b6fc0d075112ec11 : SUCCESS in 1h 07m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba2936351ca840caab22a6f2e57432ea : SUCCESS in 43m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/afa2e034794d41fd8c16f5df18875dcd : SUCCESS in 1h 19m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bcab2fadd7874d11bf6ba40d65ebff93 : SUCCESS in 1h 45m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/97123c91621f431a94c366180f4093dd : FAILURE in 58m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d042e998706743218b0c47167175850f : SUCCESS in 1h 24m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5de602eae5784f4fa76fd058b58ef412 : SUCCESS in 1h 23m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e71f296251f4ffc8bd086c2ecd7bc58 : SUCCESS in 1h 05m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dc1889eb191f4b078c0a7915c18d723f : SUCCESS in 1h 38m 04s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6859f77ca82f4c1db1bef9ae6935f3f9 : SUCCESS in 2h 06m 12s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/998ff9aede884967889baeaca782ce95 : SUCCESS in 1h 05m 43s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/98903f984fb84f2ba53822364f685df8 : SUCCESS in 1h 19m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c2913e6ef1e84b6aadb4038195dabaa0 : SUCCESS in 1h 12m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ef0adbe529564f6e96c53cea56bd9ba5 : SUCCESS in 1h 03m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/361155c7a59346d682438e3cd60981c3 : SUCCESS in 2h 42m 14s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0eda6b206ecc47bea4f7cadb802efe56 : SUCCESS in 2h 12m 37s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/53b8713956c8456ea7838e98df7f0c6a : FAILURE in 1h 48m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e20688511d32473a85e40271419392f5 : SUCCESS in 1h 11m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1e6c0b06f7e8428488f581c93ceb0c12 : SUCCESS in 1h 44m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c2979e583bf419989980b17e80ae60c : SUCCESS in 55m 11s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/42e80f96c7834e85ac26a0d5652fac2d : SUCCESS in 1h 04m 16s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/47621601b150418ba0a14424aa9e1b50 : SUCCESS in 1h 23m 19s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dc4371c621af4b08a5b5c6b64d536477 : SUCCESS in 1h 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f13743466e308be63860dbd1e45d7e11d7181988","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-11-01 06:29:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ae27fa885f82076eaea378fa3ee8ce56afb64665","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-01 07:15:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-01/692285/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/80d5070 : FAILURE in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62d96f5cfc183d3c122f0d64b46e3e9b54f92ca1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-01 08:20:19.000000000","message":"Patch Set 4:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/85/692285/4/check-mlnx/Neutron-ML2-Sriov/2552c74 : FAILURE in 49m 15s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/85/692285/4/check-mlnx/Neutron-MACVTAP-ML2-Sriov/b038e69 : FAILURE in 1h 00m 40s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"89d4fece4e2552542e6df63d30431de2b3b8eebb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-01 08:53:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/4 : SUCCESS in 1h 40m 31s","accounts_in_message":[],"_revision_number":4},{"id":"21983dd984db3bdf0c96c572f1f901acfd432c3d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-01 10:43:24.000000000","message":"Patch Set 4:\n\nOK, reverting the original patch is going to be more complex than jut fixing the \"binding_index\" column and fixing the tests.\n\nBTW, you should not move the migration or the alembic heads, just removing the \"autoincrement\" parameter.","accounts_in_message":[],"_revision_number":4},{"id":"901f1584b6fa15e76aab82303de7d73711e070ca","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-11-01 10:57:53.000000000","message":"Patch Set 4:\n\nBut the original patch was merged during ussuri timeframe, so we need to move migration to proper dir, are you ok with that?","accounts_in_message":[],"_revision_number":4},{"id":"c377e4482060bc5072c218a262ee2b367a6aff75","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-01 11:14:15.000000000","message":"Patch Set 4:\n\n\u003e But the original patch was merged during ussuri timeframe, so we\n \u003e need to move migration to proper dir, are you ok with that?\n\nNo, that\u0027s not correct. The original patch was incorrectly merged in train but we should not modify this now. Just the DB change and the CI\n\nNext time we\u0027ll be more careful...","accounts_in_message":[],"_revision_number":4},{"id":"d07c25d0a00169b4716213dd046040ffb294dae8","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-11-01 11:36:58.000000000","message":"Patch Set 4:\n\nCan you please clarify why it should stay in \u0027train\u0027?","accounts_in_message":[],"_revision_number":4},{"id":"4774ba955b900922469b658f5447b1fa26e29e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 11:44:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1f521cc912074d5b8c1dff45f3aa2e5a : SUCCESS in 1h 03m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8cb343d5b45b459da55c8769e09029a4 : SUCCESS in 48m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/4471bb5473334354bf707538107ae432 : FAILURE in 1h 27m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/d31367d905fe4b28b05d400e49777473 : SUCCESS in 1h 02m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a581785fbc5479e82d11177d89d0a9c : SUCCESS in 1h 01m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/54f724ceb8974d0d86783edacea61f2c : SUCCESS in 52m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b22ba16cc27a461f992bdfdc1ff51c4f : SUCCESS in 27m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c40ea857658478f96c2032c8bcffb5a : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35ed78488e64b6c92e72fc92fe77fba : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eddb0b0ddcd546b3bbae537c543a6cc4 : SUCCESS in 25m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f65a21be54b842db90b71027cf5c29e7 : SUCCESS in 24m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fece245ef2943f49f1a4b81d03027f3 : SUCCESS in 13m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/798efcd7b77842e7b8f3eb190bb5c733 : SUCCESS in 1h 11m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b9ee0a83c6974a9c8ffd874f643fa02a : SUCCESS in 1h 06m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f4965441440a43579cdc3fa0bd6911eb : SUCCESS in 47m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a28e7c8dbea7422991b3fd71cbc4860d : SUCCESS in 1h 05m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b95284d0d8c94ce1bdf669afa481839c : SUCCESS in 1h 43m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/303bde2beb9d4e3299c29bbe0f041ab1 : SUCCESS in 1h 33m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1498e475bcaf403cb65301901597e0a2 : SUCCESS in 1h 16m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c443ab68dbeb4047ab7ead4e170e14ef : SUCCESS in 1h 12m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b5a299ef87347749aef2e1d7d1569c9 : SUCCESS in 1h 13m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/266941325afb4367a9375dc5019ef603 : SUCCESS in 1h 58m 43s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/08f419d39cad489d81112841f523ad5c : SUCCESS in 1h 47m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b6732b669ef41f48dabb8255407bec1 : FAILURE in 1h 26m 08s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/8b9ee04355c34fec95aa24958463be43 : SUCCESS in 1h 27m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c1f09cf821564c40aca0be9997a2f551 : FAILURE in 1h 14m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3bccbf32047442f0ba54ae1b3f86ad75 : SUCCESS in 50m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f862a623bf944c6f925c4d4ab23cff5f : SUCCESS in 2h 33m 44s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a04d839b36af4f13ab5d746e40836abd : SUCCESS in 1h 51m 47s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3a96fce3bb77490b98dfda5c17cfbfad : FAILURE in 1h 29m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2102127d1a03471fa83a6bc881887be7 : SUCCESS in 1h 01m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5b480a5842ff48808bdc4eaae05dc3a9 : SUCCESS in 1h 32m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfe9e31907e640f7b7adb3af89a767cd : SUCCESS in 50m 19s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5161788d7bb4004a92dc646e60ccd2d : SUCCESS in 1h 03m 47s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8097ed4ded344427b6896eff048ff316 : SUCCESS in 1h 10m 24s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/37ab2051c1c14a47b8a1a90b2cd619d4 : SUCCESS in 2h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"094a75063f0f4f94e11a54d0566bed302bd62dfd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-01 12:01:42.000000000","message":"Patch Set 4: Code-Review+1\n\n\u003e Can you please clarify why it should stay in \u0027train\u0027?\n\nTo stop playing with the migrations. But at this point I don\u0027t see any reason to keep it in train. Actually because this migration is not in [1], we should move it to U.\n\nAs I said, next time we should be more careful.\n\n[1] https://github.com/openstack/neutron/tree/stable/train/neutron/db/migration/alembic_migrations/versions/train/expand","accounts_in_message":[],"_revision_number":4},{"id":"4ac4ba83aea874bc1d327e2ac99f9138760ec381","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-01 12:02:23.000000000","message":"Patch Set 4:\n\nrecheck\n\nJust to check again that the CI is fixed with this patch.","accounts_in_message":[],"_revision_number":4},{"id":"97278a2d893a034e2bf88349dc00d0937d1e3b08","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-01 13:42:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/4 : SUCCESS in 1h 39m 47s","accounts_in_message":[],"_revision_number":4},{"id":"c02dadb68ffe752b13ebe9801bbf54da3b35ae5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 14:32:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8d8e47214a804dd79ff766fcc29efbbc : SUCCESS in 46m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d9c2ddb8266497c99f85375f33ed958 : SUCCESS in 34m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0c015cd9c0b24f20b0a5014538cb704d : FAILURE in 1h 37m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3a704b631a4e403cbe0437819aa1cb0a : SUCCESS in 1h 40m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68d64a7b191b43b69a39fdbc1a00e021 : SUCCESS in 57m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d63796e4f5e2481e8e93908fb7006570 : SUCCESS in 56m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7555f17570b4d14b6a2ac218647f6bb : SUCCESS in 29m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1211ecc662d54c6e952fe7de38c006a8 : SUCCESS in 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee5bef9ae33422b9ce303a70af27858 : SUCCESS in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/670b7596c5e24eee99d2fcab82906cbb : SUCCESS in 23m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0891d5a527a14281bdf5b7367cdd84dc : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442a5e80c7be4558b9e1d63b9bf7e622 : SUCCESS in 11m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aaaab2849448478c9648d0546b24b3d7 : SUCCESS in 59m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1ae8362895a487494ef094cb6daacf1 : SUCCESS in 1h 16m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b8c375b53e1f47198ff947dcd133a10f : SUCCESS in 1h 01m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/998cb4bb41664d7d96ebd5cf141c5ef9 : SUCCESS in 1h 11m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7b0403e3b4c74130b346d99dfc4c713b : SUCCESS in 1h 42m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a0b48c4c1d234080b9d15549a544aef2 : SUCCESS in 1h 20m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d6c1d7eb3b964f85a75da686586427a4 : SUCCESS in 1h 27m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3aefe05abff64c16972b7b443ed013c9 : SUCCESS in 1h 11m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/122aa6c3f9844371956f18536c52827e : FAILURE in 1h 20m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/85e29b6e22dd428795a972720c18c08e : SUCCESS in 1h 25m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2d3482745f44d5eb73ce3709f0e686c : SUCCESS in 1h 55m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/44e5e371cf624186b6e09d8762f1fd64 : SUCCESS in 1h 22m 49s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/909394e8019f4c98a043f1d94826a123 : SUCCESS in 1h 33m 39s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f775ef307c004ebc9aeb7016380c018c : SUCCESS in 1h 06m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8b129d13d2c940cb8d6a47c851673b94 : SUCCESS in 1h 14m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b51e95ffa454f178033519101c7fa28 : SUCCESS in 2h 11m 53s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5e71a2438ec04b98b75a15e9dd62a5c0 : SUCCESS in 2h 23m 38s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b14c8216a05447769c4a25af17616b23 : FAILURE in 1h 36m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c24dd9d6a7e84981a7eaa123ac62d7ff : SUCCESS in 52m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/061c826078ba4353bcc4615eca10f440 : SUCCESS in 1h 22m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6a38c0083c241459bf3a2beed864e6f : SUCCESS in 38m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/95512ffd118344c59e3cb32c7b251d97 : SUCCESS in 1h 00m 30s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1bebce08f3fd40578ea89c0e8e764648 : SUCCESS in 1h 05m 06s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e57780bcc6424fb8a9838602d922a1f1 : SUCCESS in 1h 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed2522d6c86ad34eff228154f19c4284b45bce43","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-11-01 14:48:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"301ac16a37c4b391b4a0c682e6ba87ba89cb2302","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-01 15:20:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-01/692285/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/b9f117f : FAILURE in 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4c06d6144457f8855bd8fd2b6720052b6d3674d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-01 16:29:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/4 : SUCCESS in 1h 40m 16s","accounts_in_message":[],"_revision_number":4},{"id":"3b0cb8bf643a0c11c8171c1b734440c8b8a9c437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 17:31:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/12879c8112c34847a6859eefeb1c9ccd : SUCCESS in 54m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cca5b76a2b994abe9460a655996155b6 : SUCCESS in 39m 04s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/66f0d4e86f8a4ec085ac7729146fcbfe : FAILURE in 2h 02m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e136a16d37dc4211863a623b6e5658e8 : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/570063493bfc49d69d8de43ab8a71b15 : SUCCESS in 58m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/52e575b58d694611a1e24abf37d572e8 : SUCCESS in 1h 01m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d71869ef4c84eaa894716513706d38d : SUCCESS in 34m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/062c6b92376246b38a2382255dad1c0a : SUCCESS in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedcc6ec02274db89a9bcfe0e66044ce : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/893505d196024973ac824e08d28fe800 : SUCCESS in 21m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15867d0a48364f98aa7aa70f0eebb675 : SUCCESS in 21m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610f499effb148c4af927249f9b210df : SUCCESS in 9m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fa93936dbbbc46fbabcb91cc2fad719b : SUCCESS in 58m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/532f8b3f6e804f9bb14e5d2625623c11 : SUCCESS in 1h 24m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/67299e737d3d45099803833261f470e1 : SUCCESS in 44m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9ccde85df5e74a50a9a521deecfa82b4 : SUCCESS in 1h 12m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/214298623a084eb9aeba8b2d610faa44 : SUCCESS in 1h 50m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0ab8b39eda54447881fd9f88e62de8c9 : SUCCESS in 1h 15m 02s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f8edb08deda4496f89dd318ef9006d79 : SUCCESS in 1h 26m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08aa9d7a672c4ec68dd0bcb7a0e4450b : SUCCESS in 1h 17m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fc3cbc06aa2443e38a1685fb544ea360 : SUCCESS in 1h 09m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/348317a643424800bcf8bcc09127f449 : SUCCESS in 1h 50m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/42a517f3bc8143dfa9c299f7e10fe29e : FAILURE in 1h 25m 32s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cc34e990298f4ab1afe525e4d929c69d : SUCCESS in 1h 16m 56s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/0c94c1f8db6c406595ce3aace4f64013 : SUCCESS in 1h 36m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fcba9a35ebeb4babac043e894179435a : SUCCESS in 55m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/af842bb13fcd4116a2f1d778937d10b2 : SUCCESS in 1h 05m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b3238fe38e2494c83e920ab0b149fed : SUCCESS in 2h 19m 05s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/37773dcf817d4822adb8503a5941215e : SUCCESS in 2h 34m 59s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1f6dfe00a4524caeace091de7a88fc66 : FAILURE in 1h 25m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfa314e2a761472283bf59f51ea08a8d : SUCCESS in 1h 03m 50s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f5bc9d10d455408abc5abdecd5baf64e : SUCCESS in 1h 23m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e31bb78bb7ee4900a4c811a5b3c441a7 : SUCCESS in 46m 18s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e36eee621234087bdc27208611b4365 : SUCCESS in 1h 03m 57s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/91014c5c86bc4644bf99fdc21cbc2411 : SUCCESS in 1h 24m 12s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b7b6a36fe0c042a5b7174835798cae66 : SUCCESS in 1h 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e07a59f964b8330ecd1245edc935f53644256945","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-02 12:26:35.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e \u003e Can you please clarify why it should stay in \u0027train\u0027?\n \u003e \n \u003e To stop playing with the migrations. But at this point I don\u0027t see\n \u003e any reason to keep it in train. Actually because this migration is\n \u003e not in [1], we should move it to U.\n \u003e \n \u003e As I said, next time we should be more careful.\n \u003e \n \u003e [1] https://github.com/openstack/neutron/tree/stable/train/neutron/db/migration/alembic_migrations/versions/train/expand\n\nI agree we should be more careful next time but now lets go with it and fix UT in the gate","accounts_in_message":[],"_revision_number":4},{"id":"c809f85bc88c50ef3da1faf5bdabd661a88a4fc7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-04 14:26:55.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"89d08232e49b0695493d3af5450d4d0933b4ab9d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-04 15:23:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-04/692285/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/948b6fa : FAILURE in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ee804bba7cac2a626f12eaff67747401de6e5ca5","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-04 15:23:21.000000000","message":"Patch Set 4:\n\nThe neutron-grenade job already failed with a bunch of \u0027No valid host was found. There are not enough hosts available.\u0027 errors.  Will watch and recheck.","accounts_in_message":[],"_revision_number":4},{"id":"6862639b92acec04978b89b5bd220bf6c81228b6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-04 16:05:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/4 : SUCCESS in 1h 38m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7686fcad3b2c39dcd03f2dc8cfa3455dce825ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:55:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b63f1cdacf7d43368a91b3e6064a6fb8 : SUCCESS in 52m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cfe6bc61d66347029ad95307a4d80056 : SUCCESS in 38m 22s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bcdaa2d69aa34c9ab039af9b12d3193d : FAILURE in 1h 33m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9afd72e0daca4dfa9bd1110eaeb64d7d : SUCCESS in 1h 14m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fa50d8fa7067470b9fb0a9e2f33d45f7 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e922597a3b054315b86269081cd30d79 : SUCCESS in 1h 00m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f756df157124d79bca9ba920eff3473 : SUCCESS in 30m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b390d3b283741ed949d2748c915bd60 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261457a572b14b42a840e1c4251647f9 : SUCCESS in 13m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6624061cdbe940dfbdbf09510edeaa37 : SUCCESS in 26m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/849713b11621460caa06ff10303960e1 : SUCCESS in 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9947a2af7e4de58b6ae844b679e833 : SUCCESS in 8m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/afa078dd70f74c06814bb7b33e10f4e9 : SUCCESS in 1h 01m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c273300790d4174b2447746cb1ee25f : SUCCESS in 1h 34m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f82e4e6510a646b4bb8333b2c66277ba : SUCCESS in 1h 00m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e0ef465234514a3685a86a20d02d47b2 : SUCCESS in 1h 11m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/82844e020f5b437fbf80cd2acb124998 : SUCCESS in 2h 10m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/260fc0981f8445198015736b1ba0073f : SUCCESS in 1h 26m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/81a54a7d13734d7ca9e63190a4868df7 : SUCCESS in 1h 33m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed8b5abb581a4ffb9ef9bf6c076a2800 : FAILURE in 1h 26m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/db5968163809401e999655093a964a11 : SUCCESS in 1h 36m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a65fd39cd08e4649aa123677c3707845 : SUCCESS in 1h 47m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eafbe3ecf0ca4b739de5b7e07466ed11 : SUCCESS in 2h 17m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb79648ee4874be0b36fdd706228b1fd : SUCCESS in 1h 27m 29s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/42e4ccdef44042bc98caf2d7f93b582a : SUCCESS in 1h 33m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bb231d28858e42379728a60e26e05d9b : FAILURE in 44m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b1c4029352a846378492a29f900893ba : SUCCESS in 1h 08m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/74bc7683ad8f4d7eb46ae574859f01fc : SUCCESS in 2h 08m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/739dec6807914c50a7d542ae8ebbf5e0 : SUCCESS in 2h 23m 19s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/31288e739f0f499aa51e08229b6971d5 : FAILURE in 1h 30m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be7fe1b3734b4aeba9c981c5a7b9669e : SUCCESS in 1h 05m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f49767e15d1f4abfb53026ee7ca7bd50 : SUCCESS in 1h 27m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1091a3b3cd645b0b11293968c05132a : SUCCESS in 53m 10s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/12da3a00c14b4d3b88adb140770ca714 : SUCCESS in 1h 08m 50s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cac590f3cc745ef93995aef15148746 : SUCCESS in 1h 14m 49s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/228e83fc398c4ff8a06a2753efc37025 : FAILURE in 2h 04m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4edf7dd95edc20e7fd3d4abcf7a2db11b88af7dd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-04 17:44:44.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d607822c210d3f467ba3ffb8b87ff9a3b7e3b89e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-04 18:28:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-04/692285/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/23154be : FAILURE in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"054fe34f2a2a7593b255bc2744f81f6737a024c9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-04 19:33:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/5 : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":5},{"id":"7b1e59c8d0b852911665535652bf883afc99b1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 20:19:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65c67af60258473a88b44bac213cab8e : FAILURE in 1h 03m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a74b00a155ec426db21b029c236c373f : SUCCESS in 33m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/deca3448503d414983268266a2b5d0a2 : FAILURE in 1h 53m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/28e0e004d89f47bcb9a0a2fcb56d6cd4 : SUCCESS in 1h 10m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/60922e8ab85a4060af222c087462a45e : SUCCESS in 1h 09m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/79c67857216b4d4e989739131a11da19 : SUCCESS in 55m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74dc5b1c93874a80a3aedab0ea20e5e4 : SUCCESS in 25m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b0dd8f0e444cdcbb85da4f0269bef3 : SUCCESS in 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c742608de2d64fb3b889b05ea2f120ad : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a6784f758874b458ce2013266ed5e9f : SUCCESS in 22m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca7c47599d564618ba552fadfb2434ea : SUCCESS in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f9cd60c9684ec78f9ac6f48b0fef42 : SUCCESS in 8m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/015bb009fcad47218efd75e7e352366e : SUCCESS in 56m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/82eb3d5d0822494e85cce33efb67676c : SUCCESS in 1h 10m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd86491949f047fba7c97b568bfcf9d9 : SUCCESS in 45m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2bd539a4dc124c9383eb401edbef0b4d : FAILURE in 1h 05m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/47024af06f914b1bb8ecf5949364c6ff : SUCCESS in 2h 03m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/69c87fe804c941ddbcd5e1f55dd33a58 : SUCCESS in 1h 08m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/73f6dfe7229e47deaf11805f5a668cdd : SUCCESS in 1h 14m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d2d7db0760a41b79abac1e9861e8131 : SUCCESS in 1h 09m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9943d55f887d43e9b43d2aad732d0e58 : SUCCESS in 1h 04m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bda365e21a5d4de4b9183bb182285032 : SUCCESS in 1h 50m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c652dad86e5e4724b68e6b191f47a7e3 : SUCCESS in 1h 53m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d13c440896c14c60a6c806fd9fb118ff : SUCCESS in 1h 11m 07s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/ed4c8088dd5748ae9cbc2c8229f4fb17 : SUCCESS in 1h 22m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0f2f105e974048088006c2a13fcdb5bd : SUCCESS in 1h 09m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b9ae8239347d45dda7051588a5bf89d2 : SUCCESS in 55m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df5e35b0e515496ba007a9db61e9ea3a : SUCCESS in 2h 08m 41s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/202b27c28f70444ab2d5cfe692a51a27 : SUCCESS in 2h 27m 31s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/25bf3d1f69a24948b6fe8ae54ae31387 : FAILURE in 1h 17m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eec3ea79abc4129b45b5bc420c8823c : SUCCESS in 1h 03m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b65ffb48e5a0408b9c6a868697228206 : SUCCESS in 1h 23m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8080f07d572542f183741dbe9cb43bf7 : SUCCESS in 42m 40s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31aff964c40845d0a9c73bb3bce6eb29 : FAILURE in 1h 09m 04s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/46db0bc29d534774b26ac8dadf0db10e : SUCCESS in 1h 22m 19s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e9f351f5e1d74d8aab85011fcfa9be64 : SUCCESS in 1h 38m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff273d8107c80951abcd103632d4e5473e8c4bdb","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-04 22:54:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"354bfb06c27c9ac67172358dc9346940f52d401e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-05 00:27:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-04/692285/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/bef3390 : FAILURE in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"40e551591b72d733bb1a24ef31a5cab1acdf3aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 01:48:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4d7fcca927554ca499eb1f00099fecd3 : SUCCESS in 42m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad1f8f6986e049c98d8f44d031ecc36d : SUCCESS in 27m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f8ec6f77b22a490baacf88febcf9fcef : FAILURE in 1h 33m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/480da4fed35d437aacbd57a2ba6aa604 : SUCCESS in 59m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/39a88e21dfcd4d618933d6fbe0e1df41 : SUCCESS in 53m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1f4c2912aec848a39962e2322f40e587 : SUCCESS in 55m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ddda80863541ca916fdb35a313c355 : SUCCESS in 26m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b81353822624dd19fa82cad4c48f106 : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd77ed83eff0447697aab1c66c88a2f7 : SUCCESS in 10m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb88cf2752a4391b3a2b81969056596 : SUCCESS in 20m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f41227cf6d74be7a9c4aad9d0d1df34 : SUCCESS in 18m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d803f35ef68429b8f43320f0e463b75 : SUCCESS in 10m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4fda899544ee4e34b5a17039a4acfbdf : SUCCESS in 1h 09m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8faed6b1eae4c268cf30edbcf15cc1d : FAILURE in 1h 32m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f175b6160fd473badc816a08d5229bf : SUCCESS in 52m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/832af800b3dc4533a77d8991ffd6dd75 : SUCCESS in 56m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/319c7e9f7e454281a0c8caa0776df9df : SUCCESS in 2h 20m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c4234d75ca4b4c75883a763c133bf2ca : FAILURE in 1h 38m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/92a2059bca544c1897074b1ed7fa4aa7 : SUCCESS in 1h 33m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb1b685200f64549a0ba6be7230d9cec : SUCCESS in 1h 00m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c4c7659cb70e409689f522153041c734 : SUCCESS in 1h 00m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82e87b4fe0ff461ca5a1fe9984d6b648 : FAILURE in 1h 58m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d851fe639144484befc4a3af9dbb147 : SUCCESS in 1h 46m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e90adc607f64308a438a6b792100746 : SUCCESS in 1h 05m 30s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/ee7c5d09796940f481425fd4170cc057 : SUCCESS in 1h 46m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83c2deb705254271b2c2f6a1e9a4ae32 : SUCCESS in 1h 01m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e5bd81d2e86f4e8d8262a609113414f1 : SUCCESS in 53m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92d177cb218e4568ad81109803361a2e : SUCCESS in 2h 07m 03s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/64cce695856d4ce4ba787ddccdf65db4 : SUCCESS in 2h 43m 48s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/93fe172fb2424a71a4fdb7b6ed3054f4 : FAILURE in 1h 30m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe0948bc092c4863957ed68a11cadabc : SUCCESS in 1h 20m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/89e2cb498b00425b813128c4e50e17c2 : SUCCESS in 1h 52m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b953f3092294dc19d02f0468095b308 : SUCCESS in 46m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c9bb0256041419da9eab64879590331 : FAILURE in 1h 18m 55s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b96343a463424e058d359c4073f70fbb : SUCCESS in 1h 11m 11s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/00ed48770d2647eea1cb5b1dd089e46a : FAILURE in 2h 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6b0d6ae7695e9cfbf64cd41e78f3b97fcbdfa2b1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-05 03:41:25.000000000","message":"Patch Set 5:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/85/692285/5/check-mlnx/Neutron-ML2-Sriov/9496bf0 : FAILURE in 46m 48s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/85/692285/5/check-mlnx/Neutron-MACVTAP-ML2-Sriov/53799b9 : FAILURE in 58m 15s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"429a4e8edb7f3cd512d653a3413f2da2cb24f529","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-11-05 03:45:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"08ff9ae02752c6237128f26663a189264f935893","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-05 04:17:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-05/692285/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/256ec1d : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6233a835a63c1bc7608562699395fef81e616e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 06:06:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c58aea93e89d4aaa95bfac4db233250f : SUCCESS in 48m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/878443deb9a04e70904076ef5c06b604 : SUCCESS in 33m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/38d565d116c54943a2ec0edc1258c3d8 : FAILURE in 1h 28m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/02ac9eaa61254d6fb506519c19302a34 : SUCCESS in 56m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4c14175e4be84522909cf9721123afbb : SUCCESS in 1h 04m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05392ac4040a4a609bc8845b59c6eee3 : SUCCESS in 49m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff914ecacebf48a197841bdeab546420 : SUCCESS in 24m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc76ff594ace4c1d8378c3b9e0f101ac : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7673307c51e740129ef97d8814af6d83 : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eddc7b18466f42bda2f2b0e12cca2d50 : SUCCESS in 18m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a8861aa976c47849db5b59b94311961 : SUCCESS in 19m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b0fc73fd814e98941910ffc32df140 : SUCCESS in 8m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d38450a6f7a844e4bf25d98a6a824df4 : SUCCESS in 54m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/151f1b23ea634716a111b9143c5d8328 : SUCCESS in 1h 06m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/85bd17cbcf424c728364d8ecb42b89f1 : SUCCESS in 43m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aba38c3dca594b9fb7855e514da4ffcd : SUCCESS in 1h 03m 01s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5745e776bc7a4b8c87ec276f4a55c564 : SUCCESS in 1h 53m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5e52171421e647cea20a90166fd3ed09 : SUCCESS in 1h 17m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d5371d0edd9545cfb059de87baed050f : SUCCESS in 1h 10m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ccc59c05b2e4f3d8288c5c03b5ffed4 : SUCCESS in 1h 05m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f9ca10c91b4e41e39f832b3fd7223c7f : SUCCESS in 1h 05m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3a20d83942234f7a8d7086fc5364dace : SUCCESS in 1h 31m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a43fc8d598b47d798ebb4cf440a6f74 : SUCCESS in 1h 38m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c56416454382421fbb5c0089e9307b4d : SUCCESS in 1h 09m 30s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/13e66364d6be4473a16fc7312295f157 : SUCCESS in 1h 29m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bc90935b9b904d5a856771c40b9d6288 : SUCCESS in 57m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/903e81d04c814c4ebebe6d3d3a664169 : SUCCESS in 49m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be1b3b6f515f4545a323807c263edb6b : SUCCESS in 2h 15m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f7e48fadb7ab46b1bca5135b56ce9ecf : SUCCESS in 1h 55m 51s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/358111d6ba1c4986b3af6a782977bf05 : FAILURE in 1h 16m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/042e3bd0cc2f4cc38457c544d6166e8e : SUCCESS in 54m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fb5cdd46a15e416790895fffcae262b1 : SUCCESS in 1h 20m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/74ec37ab199c4fdaab43d9ad8e220d7d : SUCCESS in 41m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8e8b2f67f894bdeafa5b0f73e640d7f : SUCCESS in 59m 51s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7feff6a6408b46beae993d2ef5fdb278 : SUCCESS in 1h 08m 31s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c8d8b80f444f4059b25851431678ccb1 : SUCCESS in 1h 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2d5539f2c1540d45c261f38da9cbeaf11fdddbdd","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-11-05 07:09:24.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d346ab3baabe193e1c8b829701d750b6b6708c92","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-05 08:01:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692285/5 : SUCCESS in 1h 59m 59s","accounts_in_message":[],"_revision_number":5},{"id":"f4ed3741759f79e79e0ec5b7fb0dcd0303985625","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-05 08:43:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nPushing...","accounts_in_message":[],"_revision_number":5},{"id":"b66a5d960971e1f5247f6e2a39b0fc518e8c9cb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 08:44:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ce86570f76bf9ffd9fc769717e6635f626043c29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 10:58:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad6ce8c4207445f3a78f1291d77746a7 : SUCCESS in 46m 11s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0c6275821c68436abe65b2c34e2b785f : SUCCESS in 1h 07m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/03fe15ccbbc548c6a09efae83b4d96fc : SUCCESS in 55m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/574d55f843bb467fa30f66f0e5c84190 : SUCCESS in 55m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69b27e999dcd496fba0b6e5b820ebc2b : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a863863dd9a242c99673769e526d6621 : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08c90efbd59c4c2eacbf24576701323a : SUCCESS in 23m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c352eff45b944d9a89dddc9560a9b134 : SUCCESS in 24m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e46f7f0e604f16ae195287af59a7c6 : SUCCESS in 7m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b84c84075ccb47b69fc383585b8d1206 : SUCCESS in 1h 05m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e068792a63746b1b4490d60d29ae5c9 : SUCCESS in 1h 17m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/45cefc01269d45fca718b625fc4e6749 : SUCCESS in 47m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cb31c945dff545b3bc049832efbc64e4 : SUCCESS in 1h 03m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e16e7240cb524a1fade698b48a1b6d16 : SUCCESS in 1h 37m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/aaae01b009cf4ceb8abab079656c0f1d : SUCCESS in 1h 14m 18s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/057dbf37166348fd8dae3e6c076a7511 : SUCCESS in 1h 14m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e20be7be010941cbb2ad47a311303d30 : SUCCESS in 1h 03m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/478180569a034ee3832f6d27c29aaca8 : SUCCESS in 1h 15m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d224c99636554ee98cc297f54983562a : SUCCESS in 53m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36322cb5e90f4129b6b0bbaf36cddfe0 : SUCCESS in 2h 12m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcab5e717def4599b2eebbb592f4a32c : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":5},{"id":"b12dac5db3b377472185d026b4abddb207536025","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 10:58:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2d46506f62b990bd6ae35a6e821373e7308e1320","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 11:00:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e00baac9a045bc90c592e9b2f58a88 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0a3c53e411769141adc44b14af5efb854359af1a","revisions":{"73fd0f9aaca369b860986a86d296e452fb6eb0d6":{"kind":"REWORK","_number":1,"created":"2019-10-31 07:39:07.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/85/692285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/692285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/692285/1"}}},"commit":{"parents":[{"commit":"843b5ffd9a8ee3f4d9d8830f43aa3d517cc11e07","subject":"Merge \"Make the MTU attribute not nullable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/843b5ffd9a8ee3f4d9d8830f43aa3d517cc11e07"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 07:36:59.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 07:36:59.000000000","tz":240},"subject":"Move migration to proper dir and fix objects tests","message":"Move migration to proper dir and fix objects tests\n\nThis is an alternative to revert patch:\nhttps://review.opendev.org/#/c/691710/\n\nChange-Id: I72082226e6b3930efb537cab9d979e5770211f72\nCloses-Bug: #1849976\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73fd0f9aaca369b860986a86d296e452fb6eb0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73fd0f9aaca369b860986a86d296e452fb6eb0d6"}]},"branch":"refs/heads/master"},"6c82b1ddddadba718be776dcb47facd164a6bb4b":{"kind":"REWORK","_number":2,"created":"2019-10-31 09:26:35.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/85/692285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/692285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/692285/2"}}},"commit":{"parents":[{"commit":"843b5ffd9a8ee3f4d9d8830f43aa3d517cc11e07","subject":"Merge \"Make the MTU attribute not nullable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/843b5ffd9a8ee3f4d9d8830f43aa3d517cc11e07"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 07:36:59.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 09:26:15.000000000","tz":240},"subject":"Move migration to proper dir and fix objects tests","message":"Move migration to proper dir and fix objects tests\n\nThis is an alternative to revert patch:\nhttps://review.opendev.org/#/c/691710/\n\nChange-Id: I72082226e6b3930efb537cab9d979e5770211f72\nCloses-Bug: #1849976\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c82b1ddddadba718be776dcb47facd164a6bb4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c82b1ddddadba718be776dcb47facd164a6bb4b"}]},"branch":"refs/heads/master"},"133ba85795096e3753189a13eb340d878373294d":{"kind":"REWORK","_number":3,"created":"2019-10-31 09:27:23.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/85/692285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/692285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/692285/3"}}},"commit":{"parents":[{"commit":"843b5ffd9a8ee3f4d9d8830f43aa3d517cc11e07","subject":"Merge \"Make the MTU attribute not nullable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/843b5ffd9a8ee3f4d9d8830f43aa3d517cc11e07"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 07:36:59.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 09:27:10.000000000","tz":240},"subject":"Move migration to proper dir and fix objects tests","message":"Move migration to proper dir and fix objects tests\n\nThis is an alternative to revert patch:\nhttps://review.opendev.org/#/c/691710/\n\nChange-Id: I72082226e6b3930efb537cab9d979e5770211f72\nCloses-Bug: #1849976\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/133ba85795096e3753189a13eb340d878373294d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/133ba85795096e3753189a13eb340d878373294d"}]},"branch":"refs/heads/master"},"eb5a58d59ea415efa1eef7df036db6a08be8e174":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-01 06:29:26.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/85/692285/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/692285/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/692285/4"}}},"commit":{"parents":[{"commit":"cc3886fd784a91cb42f0add60efb8da9f96328d8","subject":"PSExtDriverTestCase: initialize variable in case of exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc3886fd784a91cb42f0add60efb8da9f96328d8"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 07:36:59.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-11-01 06:29:26.000000000","tz":0},"subject":"Move migration to proper dir and fix objects tests","message":"Move migration to proper dir and fix objects tests\n\nThis is an alternative to revert patch:\nhttps://review.opendev.org/#/c/691710/\n\nChange-Id: I72082226e6b3930efb537cab9d979e5770211f72\nCloses-Bug: #1849976\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb5a58d59ea415efa1eef7df036db6a08be8e174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb5a58d59ea415efa1eef7df036db6a08be8e174"}]},"branch":"refs/heads/master"},"0a3c53e411769141adc44b14af5efb854359af1a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-04 17:44:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/692285/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/692285/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/692285/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/692285/5"}}},"commit":{"parents":[{"commit":"38c06ff3d043f9df561670cce2dff549cd33d63f","subject":"Merge \"Add Source links to readme and  update Documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38c06ff3d043f9df561670cce2dff549cd33d63f"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-10-31 07:36:59.000000000","tz":240},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-04 17:44:44.000000000","tz":0},"subject":"Move migration to proper dir and fix objects tests","message":"Move migration to proper dir and fix objects tests\n\nThis is an alternative to revert patch:\nhttps://review.opendev.org/#/c/691710/\n\nChange-Id: I72082226e6b3930efb537cab9d979e5770211f72\nCloses-Bug: #1849976\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a3c53e411769141adc44b14af5efb854359af1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a3c53e411769141adc44b14af5efb854359af1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
