)]}'
{"id":"openstack%2Fneutron~692998","triplet_id":"openstack%2Fneutron~stable%2Fstein~Ifffdee0d4a3226d4871cfabd0bdbf13d7058a83e","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1824334-stable/stein","hashtags":[],"change_id":"Ifffdee0d4a3226d4871cfabd0bdbf13d7058a83e","subject":"Fix py3 compatibility","status":"MERGED","created":"2019-11-05 12:48:13.000000000","updated":"2020-01-16 08:43:21.000000000","submitted":"2019-11-06 11:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692998-1573039115553-9a54f4b4","meta_rev_id":"415da6006ccb74711f9b6fbbc4f0496f0051c7a9","_number":692998,"virtual_id_number":692998,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2019-11-05 13:30:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 11:18:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 14:21:47.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-11-05 13:09:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-06 06:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2019-11-06 01:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-06 06:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-11-05 12:48:13.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 12:48:13.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2019-11-05 13:09:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-11-05 13:30:25.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 14:21:47.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-11-06 01:41:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-06 06:10:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-11-06 11:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecab5fd43dbe32f5b4f12192bab2fbac421720c9","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-05 12:48:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c722181e91f46132503a9aa0050645214f69fca0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-05 13:09:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(already in Train)","accounts_in_message":[],"_revision_number":1},{"id":"887366cb6721e8b44715ceaba716f4a827a96df1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-05 13:30:25.000000000","message":"Patch Set 1:\n\nBuild failed. To re-run the job please post a comment with \u0027recheck neutron-mlnx\u0027.\n\n- Neutron-ML2-Sriov http://13.74.249.42/98/692998/1/check-mlnx/Neutron-ML2-Sriov/8e7ac8f : FAILURE in 11m 38s\n- Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/98/692998/1/check-mlnx/Neutron-MACVTAP-ML2-Sriov/39adaab : FAILURE in 4m 32s\n\nTo re-run the job post \u0027recheck neutron-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"63c7642683e6f4e7ca6b22f1f5862fc9eaf95959","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-05 14:21:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/692998/1 : SUCCESS in 1h 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e10ba1d0f8092ce7b2996d41c59f631a955b3cae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 15:10:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/a4774abc5149463885c3f0f59058c786 : SUCCESS in 52m 36s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/e13cb7b317d24b46a58d2ba99f66dd4c : SUCCESS in 39m 18s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/2ecb1e5157ca4fd5983024c8d91f9080 : FAILURE in 1h 54m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/177429e01e294764b92fbce30d89dcbc : SUCCESS in 51m 17s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/d1056953679e45e3af8e79b3f13afcde : SUCCESS in 1h 10m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/37a620287e3e4c6697273af1c5937105 : SUCCESS in 51m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/905941e6982544b7b644fa90275fec7b : SUCCESS in 26m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b9026487ef144959e9b906cb85c5c1a : SUCCESS in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b78f7c05ee14d24afd54a53f3b73d79 : SUCCESS in 12m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8c33e53997348538ebc3c00b001f868 : SUCCESS in 29m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32711b250d3548ecae58ed0e5e971129 : SUCCESS in 25m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72fda165a6eb4cb4981bdab38502558a : SUCCESS in 26m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2fa92a136a24983b9a4b55aabe1b47a : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f0b7675ee04067a9344c14ff20355e : SUCCESS in 5m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f462689d65840b4a40d3383bc89fbe2 : SUCCESS in 1h 26m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/66b5068e88f146d4b124afafd47ac798 : SUCCESS in 55m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f1655e0f5bf64a61bb03013f5f9c35af : SUCCESS in 54m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e6a3b36b1944811af9d8a13a65c9a40 : SUCCESS in 1h 20m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d3fb719984d2435eb3ac7aa93e92f8cd : SUCCESS in 44m 56s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/0ac0d478ad924e93905156bda329cc99 : SUCCESS in 42m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/60810fbeda244c77804ad86f1d7ecce3 : SUCCESS in 1h 04m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/712f7f7f3257460aaf0c4ea7ba716cdb : SUCCESS in 1h 42m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c9e83c6387214181a236bf4f7b1e8e14 : SUCCESS in 1h 27m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a38b575dfa2540758521a1d271280afe : SUCCESS in 1h 15m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f63a66a9451d4710a3fb77b7453b40be : SUCCESS in 1h 03m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6cb4c818d7b44232bb2c107d0261d9f1 : SUCCESS in 1h 06m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/99198bd61d674b9f93a4ba56424719c6 : SUCCESS in 1h 43m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e94d8db217c482a87abcc207dca65a2 : SUCCESS in 1h 51m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9ca0960e19044f3fbe7a65e5c20339bb : SUCCESS in 1h 18m 30s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/1cca7df8480440d9a1c446b4421a7a89 : FAILURE in 1h 30m 10s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d1bab8e164f406f934d28ab1d71adf2 : SUCCESS in 54m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/14561e538bd04bd692ca1d64bb260f01 : SUCCESS in 2h 20m 05s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/731f5cd736d5462dbb40b64df14fa939 : SUCCESS in 2h 00m 24s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5a6178e16ea84a1f95c33c20a0aec976 : FAILURE in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"925a81ac6d11f6bb351b3bf4fc7afef69acd84c9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-06 01:41:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"497fd44f166359c544af4bd9dfd7b37f00ecf46d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-06 06:10:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6dcd32984df62a778b449583997e7e4a7ccb001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 06:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90a529ce3a629523e17b70c4971089a318a964c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 11:18:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/67235d79e1dc40c0b38277c48259195b : SUCCESS in 49m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/192313aa1fd6421d9288be7577b9611a : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d214a634824b6ca1b2f08330d7da10 : SUCCESS in 11m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19d5028dce40451086a93aee6e776db9 : SUCCESS in 26m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7bd6bdfd8de4eeab494f4cf9ab05dea : SUCCESS in 20m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb0a922a23074288a8788ebec9b4e715 : SUCCESS in 24m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee6b5fd078f24886a7866397db89a0eb : SUCCESS in 23m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd913a72e5694fde8422f7b67b272e4b : SUCCESS in 6m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/85b1e6dda93540789c895d9b97d28899 : SUCCESS in 1h 42m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/837db0e348ac4978a6223c4f005e9286 : SUCCESS in 57m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fed87c59f6f04d8697f07794820e3844 : SUCCESS in 58m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50ab17fef55d4f47b1f273161d749597 : SUCCESS in 1h 32m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c534bbd73bb54ae1819967511522ced0 : SUCCESS in 39m 57s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7ec39e011bb04c5dade1b92005a7cb1c : SUCCESS in 46m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1f7a46b966c44444a89cb1332deb1be2 : SUCCESS in 1h 05m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/84c6018b1cae4d5487b6b6563dbaba80 : SUCCESS in 1h 18m 59s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/28ab1802a8064d5cba92e7f541d29114 : SUCCESS in 1h 11m 42s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6a85a2bc2124569b8b0139f79ef7d0d : SUCCESS in 1h 24m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/baa619786d6745468853b5b08f343723 : SUCCESS in 1h 07m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/154bbcd7391147d4af926738716724a8 : SUCCESS in 1h 03m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a628334b87c64754a565c8b1168e4be7 : SUCCESS in 2h 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3309ff410803d84ca52859a9ef850bde193d9b42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 11:18:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"214c6ee6e8e922dee1df5324ff44c5435463bbf9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 11:20:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84272f387e74ea98735786aba3b871d : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2818b8d24f4429c3e1f55103b18b7e6e78ed694f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-01-16 08:43:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 9361af0ff66fa8f4970158f2d8dc852d8d0444d1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70391139906a673bde88210757656b35cb4c45bc","revisions":{"70391139906a673bde88210757656b35cb4c45bc":{"kind":"REWORK","_number":1,"created":"2019-11-05 12:48:13.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/98/692998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/692998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/692998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/692998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/692998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/692998/1"}}},"commit":{"parents":[{"commit":"0fb7f8ddac0d55dfc64c0c0c2931b3c17cb00d37","subject":"Handle OVSFWPortNotFound and OVSFWTagNotFound in ovs firewall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fb7f8ddac0d55dfc64c0c0c2931b3c17cb00d37"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2019-04-11 15:14:48.000000000","tz":120},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-05 12:48:13.000000000","tz":0},"subject":"Fix py3 compatibility","message":"Fix py3 compatibility\n\nIn fetch_and_sync_all_routers method is used python\u0027s range function.\nRange function accepts integers.\n\nThis patch is fixing divide behaviour in py3 where result number is float,\nby retyping float to int as it is represented in py2.\n\nChange-Id: Ifffdee0d4a3226d4871cfabd0bdbf13d7058a83e\nCloses-Bug: #1824334\n(cherry picked from commit 49a66dba31f23d54972e962beef66c39307904df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70391139906a673bde88210757656b35cb4c45bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70391139906a673bde88210757656b35cb4c45bc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
