)]}'
{"id":"openstack%2Fneutron~693645","triplet_id":"openstack%2Fneutron~master~If1a6a2a174d0f737415efa2abce518722316a77b","project":"openstack/neutron","branch":"master","topic":"bug/1647421","hashtags":[],"change_id":"If1a6a2a174d0f737415efa2abce518722316a77b","subject":"Dont schedule Network, respecting network_auto_schedule config","status":"MERGED","created":"2019-11-11 02:52:04.000000000","updated":"2019-11-27 06:07:07.000000000","submitted":"2019-11-27 06:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693645-1574834713908-ffe4ac41","meta_rev_id":"478bcf682bd5763fce5029e4bb286f456e100c7d","_number":693645,"virtual_id_number":693645,"owner":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-27 01:29:20.000000000","_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-11-26 02:39:30.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-26 03:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 06:05:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-26 04:07:49.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-26 09:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":2,"date":"2019-11-26 15:52:09.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":1,"date":"2019-11-26 15:52:09.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 03:17:45.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-11-11 03:56:53.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-12 16:34:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-25 15:06:43.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2019-11-26 02:39:30.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-26 03:01:45.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-26 04:07:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-11-26 09:07:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-26 15:52:09.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-27 06:05:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f12ad5ecbd89482572791bf17d8023764c337331","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-11 02:52:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c00fd4f9cd056c212bb2dce8025eb904118400ad","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-11 03:08:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c78ca48c4cd4534f249d935807a63f5dd122524","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-11-11 03:17:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a7a7e982be2bd3e726186a6c44ba59a65c06f948","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-11 03:56:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2019-11-11/693645/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/85512a4 : FAILURE in 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78f92708054492c1a889cb4742ffe996fc0e6f60","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-11 04:46:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/693645/2 : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4bb7252e21d30d135cc8c371af94f711dd357084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 05:36:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/da9b54965ab342bc9823b6b1838f88ae : SUCCESS in 44m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c929e7d5d3e945afb72f9a78ce560ea2 : SUCCESS in 28m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2febd34158df4c7dae865cf901ab50d3 : FAILURE in 1h 29m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0c80196acb5d4d6f940e9aec81d3c912 : SUCCESS in 1h 11m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/906e7e9d2dba416eb3d78d4464505434 : SUCCESS in 54m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e6ccc74aa4e4b6eb9b4a20aa450ceac : SUCCESS in 1h 05m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eac0f9ba3d1f47b3bd451652728fdca5 : FAILURE in 23m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ea37c39c161482688e10ca15a9ee8ba : FAILURE in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486700493a8044f4bbf6e52ee32e9c4d : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd984558a874a00bf085a72d5e63380 : FAILURE in 20m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/947e32b021bf421f827464f1b76b993b : FAILURE in 19m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1449c2548df74df68575a9fee258bf77 : SUCCESS in 7m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d303448006164ac583d23915ae6eee8a : SUCCESS in 48m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/98b51cd8e9f14121ae74794dd826abbc : SUCCESS in 1h 07m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dbb0643ee2c40019802daa8a4856cc6 : FAILURE in 18m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/44d517e5b5ec4dca8126ea01ee465cce : SUCCESS in 36m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0e4a3aa0b5824281a2d8f8e1cc2947b3 : SUCCESS in 1h 29m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b744c724250446868fc948f4ff9a8a1f : SUCCESS in 1h 46m 30s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/63d6c63bc9944b33bce93a724cb38041 : SUCCESS in 1h 12m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/acd289a6039c4cda88dcb8153cc64f98 : SUCCESS in 1h 03m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0f3c66900774287a824f17589b85cb2 : SUCCESS in 1h 02m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82512cc6120d4f3293fffa3aa6538e0b : SUCCESS in 1h 16m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9a5d745a31c46ec9c486869930c086f : SUCCESS in 1h 30m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/20c7161969794a18bbf63bd837bfe314 : SUCCESS in 1h 53m 44s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48dc36f32fe9444bbe3e6744b00aed00 : SUCCESS in 1h 13m 08s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/dc016dbf8fda4cc994f3275d4374587f : SUCCESS in 1h 42m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a4e6dfcbcb3e44e59d0481680ca51b38 : SUCCESS in 58m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a21e0015159a4792adf1d42eea929d23 : SUCCESS in 55m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e58b84e7a0340418555fe463239a30a : FAILURE in 2h 22m 30s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/98c057be82c142e3946d415de203f083 : SUCCESS in 2h 01m 22s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/724b00c40acb4ca4a94e5935c3bb39e2 : FAILURE in 17m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d222d9a775584327bf3d8bad80b7c50c : SUCCESS in 1h 02m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4d8080a17e394b518b2a907a0e8f7428 : SUCCESS in 1h 19m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a0d7e5da70541bd8bf063ba616501db : SUCCESS in 47m 31s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f21af385a354704aef40c51ef8e1f72 : SUCCESS in 55m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6a52999297a4b92976b89ebd5cef1b1 : SUCCESS in 1h 18m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/250e7ae3e2cc45748d42f022a09bc1bd : SUCCESS in 2h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50ca41f13c99e4e467edaec9c158c6ca0c5a77d6","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-11 06:11:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b9680f4f1593d6a35271697bde0a873f9f4c3648","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-11-11 11:20:29.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nNeed to fix the testing.","accounts_in_message":[],"_revision_number":1},{"id":"a26a2dbd462e4de029ac81f5b555dc1b53fb16ee","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-19 03:40:58.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThis issue is particularly problematic when we have BE which dont support DHCP agent provisioning ( like OVN ).","accounts_in_message":[],"_revision_number":2},{"id":"05f1b838c227705ac127eb7d8832eef36bbf3d14","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-19 03:54:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"280a371e84bd6d08b4d8771822c582d9418c1c34","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-19 03:58:59.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/45/693645/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f6b02208b45e4c4408808d31889193d6fbe2edc5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-19 04:19:54.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/693645/3 : FAILURE in 24m 41s","accounts_in_message":[],"_revision_number":3},{"id":"39a7ad0e6e2ef5382a547abaf2f6fc4a657815be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 06:33:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8787c6019ecb4002a95966c016114a7c : SUCCESS in 52m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/856ee260d96d4b7bb071228f6205a979 : SUCCESS in 34m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e30f7fe5d6a44a32aaeaf48147570994 : FAILURE in 1h 21m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3ef6775aafab47d9b764703362728bf7 : SUCCESS in 1h 11m 11s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fee2e9b22ce845f69cf17a7a3dc8a55e : SUCCESS in 1h 14m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f9c5a81f5a847848196badbd92c995d : SUCCESS in 58m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a50db1f5db94ff7a1790d398c312cf2 : SUCCESS in 28m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d956a02617e451fa8bb591bc40905e6 : SUCCESS in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310c820e21f54d94a5fcc6e17a1d781d : SUCCESS in 11m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b490128fba84eea8c5cea7d11366389 : SUCCESS in 21m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c37671a52cd44e25b72ec5f83bede6fe : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055cd85b5cf64c16a4a6b63b580d845d : SUCCESS in 7m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5d0ccb9d92e84bbe8fa2d6c62d028a1c : SUCCESS in 1h 00m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bbeb873111bd486e8ccdba4d3b4a0346 : SUCCESS in 1h 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33deb82a6f734b7e89e07646fc9b4bab : FAILURE in 19m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e64c744dbc3046eda17e83ff15169421 : SUCCESS in 46m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ff45e2ab3a544a6bb7a5463c7694486e : SUCCESS in 1h 09m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/22df341c968b48a6b2133d83f5a0c361 : SUCCESS in 2h 26m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/725d35399371495286ef21ae7af54c62 : SUCCESS in 1h 13m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2a8549e59cc04fe88f165652b232acfd : SUCCESS in 1h 20m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47d931f14f8e445ea7728e4c25b16d17 : SUCCESS in 1h 24m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/254d9afbd6df4b8b855c617ec457b2c3 : SUCCESS in 1h 16m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4360b1976814d808b78e9820249e66b : FAILURE in 1h 44m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/16e42f2fea90406bb92bab4eaa82c3de : SUCCESS in 1h 58m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/722c177db07845fcae73538952b2e1fa : SUCCESS in 1h 18m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/988692d333de47859a765767587dadb2 : FAILURE in 53m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8bab206b0c6846238965add1c338fd5d : SUCCESS in 53m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f5613a553de41c391ccc5ef0a34f535 : FAILURE in 2h 36m 59s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/911edc05311745b298b61bbdcd4b22de : SUCCESS in 2h 16m 41s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1632fab0a3ff4c289fd4569b7465c53e : FAILURE in 19m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6736d894b476442089080805bfa7f153 : SUCCESS in 1h 04m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4c2c333e28b04f15ab7e1f7d976f948a : FAILURE in 23m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c71697148cb24b669f1dc4d981807644 : SUCCESS in 48m 14s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f425f324f05540819c9bde6d5f38a73a : SUCCESS in 1h 03m 07s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/04a5ec3b666c4f7581b0faada366df59 : SUCCESS in 1h 14m 19s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac14273683764059b3b0b611b1a569f8 : FAILURE in 1h 56m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"17d9fe5d98c8d5c978850cc8d1d06c994f50ce68","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-25 05:23:52.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1c4697f64a2a80becd14fdce96480bbda5c86040","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-25 06:19:28.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/45/693645/4/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"b6e43c835c0911a9a915bdda95c2a8d636d8e42a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-25 06:40:48.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/45/693645/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a00c619c368b7ca181eae5005a636a2e286ba659","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-25 07:12:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/693645/4 : SUCCESS in 1h 48m 17s","accounts_in_message":[],"_revision_number":4},{"id":"511eb6a89b6094d4ebab5b4894cd83a8c99b2d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 08:10:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/26e50afd7766414280ca7272143f91fc : SUCCESS in 1h 01m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/daa7b010156941a6bc09248f67c6a543 : SUCCESS in 36m 48s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c462aef5532546aba9f2f1875b4c9bc3 : FAILURE in 1h 53m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6db306b6c38c45759156441a6babf249 : SUCCESS in 1h 17m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/38a59c63083549648dee724d70406691 : SUCCESS in 1h 02m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9c842ac30f56450e9dd3fd84882db66c : SUCCESS in 1h 11m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79cd67593f3b42298e6cdb33115227f1 : SUCCESS in 29m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d40af00dc4684ad4ab71e1666ec6adcc : SUCCESS in 22m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eebd1a7d88c42d0bd1de1082d14eb55 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00ed1a561422402bafe2e441ffc0e815 : SUCCESS in 22m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c534e74a82d24e9ca23e922c3ec673dc : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe0a26c6f8d4cdbb5a8fb005b8e6253 : SUCCESS in 9m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f4a116291b0945d9abf342069626d31d : SUCCESS in 1h 30m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5f3b66309d02451897f3ab24f0ac9c50 : SUCCESS in 1h 50m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d574502221664b298af811054cefdfbf : SUCCESS in 16m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/39b31e3a0ca6470192802e4d749daac9 : SUCCESS in 1h 11m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4371d2b51be54428a5a35e96b46e0bb7 : SUCCESS in 1h 15m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2baa15384bed45dfa29946bc2efa135f : SUCCESS in 2h 26m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d112deddd8d54f008cb101768cba62b8 : SUCCESS in 1h 47m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6994565f38d9483e95c96831bbfa71a6 : SUCCESS in 1h 26m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35af2b379ece4f6a9260a75dc118341b : SUCCESS in 1h 34m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f23bde20a544333b4e8a10142f588f7 : SUCCESS in 1h 17m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ad5a4ee083c44df1b877508c60e8fcdd : SUCCESS in 1h 40m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/088d4f53b0174ef58176b9fbb5e4ae45 : SUCCESS in 2h 03m 12s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/55029d936fff419d9bfc4c593366d178 : SUCCESS in 1h 09m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eb75432a21264b32a5196e0866930177 : SUCCESS in 1h 08m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a4f8eee75e5844f58cb8526c6f35da75 : SUCCESS in 1h 19m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1937e806004346f3ab655e20ef760b80 : SUCCESS in 2h 30m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/da9007c402e240c7a42540e1998218e2 : SUCCESS in 2h 14m 52s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b8f0b608fbbe4ba3913173fae1d8fa39 : FAILURE in 9m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a2ba625e6a246d784b570634b82fe2b : SUCCESS in 1h 14m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/77e0cc08540d4f84ba9315e7e34bd64d : SUCCESS in 1h 40m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77030e65551a420b94d8321f42ab5f7a : SUCCESS in 50m 38s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/191e2314c27949fd893539f8d3bb7d90 : SUCCESS in 1h 32m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a22a70398536424496ec42b484d9704b : SUCCESS in 1h 22m 11s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3258aa51248440068e83d74198ef77b6 : FAILURE in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"29a19519c59a2acea92a9efd0fdc4617d0186297","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-11-25 15:06:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bf8e0c4f15ac699a3b89ce96d28c7e51ae9d2637","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-26 01:55:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"89f6d6e9ef1b3aad265179e277670f7d450a1f7b","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-26 01:55:09.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9b0e684000e82f69338dccfe7a13c868040201b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-26 02:39:30.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/45/693645/5/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e657fd2083e49cf3b1a83a58441174e49949a937","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-26 03:01:45.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/45/693645/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"8f1cbe7ead8c79405d496764d02fce6ccd60c460","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-26 04:07:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/693645/5 : SUCCESS in 2h 12m 06s","accounts_in_message":[],"_revision_number":5},{"id":"994275b917dd2edd03f258900384fd634fda1b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 04:28:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/805e4fa91d2246df8a2593157ff8c506 : SUCCESS in 51m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d075f49703d04f0f9b5ff7f11050f5d7 : SUCCESS in 38m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/21315c25d0704b91bf8e4c12c3a12712 : FAILURE in 1h 32m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fe1a3901aa4043179d8927088ab7cb29 : SUCCESS in 1h 11m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e219f7f079e348f6897466d2bfbe0129 : SUCCESS in 1h 16m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c3c896305624a5f9bb05baa437da6f3 : SUCCESS in 1h 02m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5116af04e5324ba48490ed61b455e456 : SUCCESS in 34m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46a96d36c7b940c1be6a0358b6c992eb : SUCCESS in 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b95ca90660045199ecec8f103c86af3 : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8fd3f878df14ba882e9f0461e5f9147 : SUCCESS in 24m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90f2b7185ab7477c8bfb1cae99725fb0 : SUCCESS in 21m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5284d016e54edb82ada08036883809 : SUCCESS in 10m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b97efea93e4942cf92561982da2e08c0 : SUCCESS in 1h 00m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e39903fa39fc41e491f71d8660c18ff7 : SUCCESS in 1h 22m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47d5c5642a4b4ce18c2d59a8654721f0 : SUCCESS in 21m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a10568f845624d5892049955e5916e5a : SUCCESS in 46m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e01a300b572846bbb373d49d2c6b9957 : SUCCESS in 59m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8a0880379b884aaba0eff5fc5d0417b0 : SUCCESS in 1h 55m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51688babf70d494384df34ae5034e212 : SUCCESS in 1h 17m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4f99c0d559b34a9eb0bafc79af1e0617 : SUCCESS in 1h 10m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57a9dd8cdd314016b8996ae41ae510d9 : SUCCESS in 1h 09m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e21a9154b6d543699b40c07739100108 : SUCCESS in 1h 05m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e2617d0a18a741dea87f192d0835555b : SUCCESS in 1h 38m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa806c47035b413a9888c098bdb33246 : SUCCESS in 1h 54m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d4073f386e454792a01216398a74b786 : SUCCESS in 1h 14m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bb1ee944379c44449f74be4e1cd67271 : SUCCESS in 51m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92b85e69574146cfb308e709a0fb07ed : SUCCESS in 2h 28m 46s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0108d40c9e8a4429b389fdaa928d6abd : SUCCESS in 2h 10m 11s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3ab2d5b53b8e4ec186d411bc5d339c19 : FAILURE in 13m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49dd965cf1ee48bfa9496af04eb982f7 : SUCCESS in 1h 26m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ac0bfa2976864c60b909658801bf0c25 : SUCCESS in 1h 09m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fd39b1521fe48c3aa4186f541a2e8ce : SUCCESS in 43m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/09034227b55345559243a3d39b88122d : SUCCESS in 1h 18m 40s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a38350f942724271babfc4b3a5c34804 : SUCCESS in 1h 12m 14s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dec6f85c4fa34230aaafc97462232247 : FAILURE in 1h 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd5ad44606b8799b67c4d4cbe0ad56e175b27f3b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-26 09:07:36.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3cd4e1008db9f5fd5af7eec231ccd012d4f76535","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-26 15:52:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"293df81974b8cbda4d9cfb66f6f3ce1440c68686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:52:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9be4d6c6a960814d37de8aca17e63662d3330486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 18:23:45.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/033913ecea994042af784052ca35b4af : SUCCESS in 51m 41s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4773450f9aaf4fc189498db09e99c7c2 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f4159721cc4b44668152be62d031d36e : SUCCESS in 1h 12m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6df4fefcb30945b097be7d8b2481f2b7 : FAILURE in 58m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4c59c11984b41239a2a9f3042ee0233 : SUCCESS in 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c81fafb940b4ceea64c789ffd750f9d : SUCCESS in 12m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd890ef974804acda5eb800424f48f44 : SUCCESS in 22m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae7e0f068e694d94b5d8bde5580e22ad : SUCCESS in 24m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77327325c24e44d1a4676b5a218651ab : SUCCESS in 7m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b4860b2b1e0437191de8d45b314d863 : FAILURE in 9m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d115583d4df34e7d9196f12a8c650cf7 : SUCCESS in 1h 16m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b174dbad2b2346718c654599d0792241 : SUCCESS in 23m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6aa3b5b765874fc298655e8b5cfc7bb3 : SUCCESS in 47m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/43d268dc10f04b15bf73b0db62241039 : SUCCESS in 1h 03m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3a736421518644cbb4f6d3552689442f : SUCCESS in 1h 24m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/97829d195204471c997a772c4a8140e8 : FAILURE in 1h 03m 09s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/31e9df27781a4b9b903e1158b66206f9 : SUCCESS in 1h 20m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40401a3b1482430f9d476fca45b09a3c : FAILURE in 11m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3092c41da0214c5c820e9b6b5e30f60c : SUCCESS in 1h 14m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54db610f2feb47448959839f7b31b19d : SUCCESS in 2h 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8d10522e4944a59bcdc41ec9c9650f3 : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":5},{"id":"977669ca819eac935285774ef934842d5e6bc1f2","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-27 01:29:16.000000000","message":"Patch Set 5:\n\nGrenade failed since it cannot find IP address\n\n2019-11-26 16:05:35.005 | Err:1 http://mirror.mtl01.inap.opendev.org/ubuntu bionic/main amd64 libharfbuzz0b amd64 1.7.2-1ubuntu1\n2019-11-26 16:05:35.005 |   404  Not Found [IP: 198.72.125.4 80]","accounts_in_message":[],"_revision_number":5},{"id":"cf0fe6e93e2ca01e5e7847b5a37f851bc979adcb","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-11-27 01:29:20.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a2556f923229626b841ac9ec0716e44b15bf060c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 03:46:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2f560fe60aa544c287a03f54519b55a1 : SUCCESS in 46m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a339c683c8a04aabbf65a0a13c2a1d36 : SUCCESS in 35m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b262c8e69c4d4c63952c76e88898913b : FAILURE in 1h 38m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1c016007413c45508771f41d32acf0b7 : SUCCESS in 1h 13m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3181cb982d134378965ac0b744dfb343 : SUCCESS in 1h 05m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08db62f7caf84ab198d0950c500f0aeb : SUCCESS in 58m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c1df6a165f84ce18b529438904520d1 : SUCCESS in 35m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7569e90da2fe4ee69b54158241210bea : SUCCESS in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf506414f21d44d39f3888b7991b33de : SUCCESS in 20m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51cd1faef9e446f48575816e680962e5 : SUCCESS in 26m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13210ae5021d4458b0961a16f4ac8fe5 : SUCCESS in 26m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a28100a3dc94a6187443a10169f574a : SUCCESS in 7m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bb522bbff1e14e4697be045dd7f44c06 : SUCCESS in 59m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6bf8b1247ec74046b60b1c44c995d8cc : SUCCESS in 1h 22m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/393b5f9f43734b138d24dab37f3efd15 : SUCCESS in 22m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/56ef67eb9a354513a1be32eeb635abe1 : SUCCESS in 46m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/361c1905fe724dcf90d09ceea7fdc209 : SUCCESS in 1h 17m 11s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/977e75bd1a134a7fa200953d3047d074 : SUCCESS in 1h 59m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c3ea8e50688848a988bb48654b714488 : SUCCESS in 1h 19m 11s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b7b091839fd9468e8ef6ff8027df898b : SUCCESS in 1h 20m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca1b17ca6b674d52a98866f4b4a56be3 : SUCCESS in 1h 16m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7554ab3525d4f1bbc7662bf1a2829d6 : SUCCESS in 1h 13m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a142188b3d14369ba5599a0f1fb52e6 : SUCCESS in 1h 51m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bbfa240c77d14209b02a48a4bbfcb04a : SUCCESS in 1h 58m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe3dd06c2bd64b3ca54d12f84f629a63 : SUCCESS in 1h 16m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dd5eb94064a942a4b7e2f4ffc352974d : SUCCESS in 59m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df1de0c0d0f9440c889fdd122ce94504 : SUCCESS in 2h 10m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/98d3d80aed87456b93f9fd75dc21d433 : SUCCESS in 1h 53m 50s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/5c117bd76b564361992bd12357e96cd4 : FAILURE in 18m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/800351ed26e24ae7adf82bafbf85a73c : SUCCESS in 55m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2651e7fd04594ebb8e0edef81cc5150d : SUCCESS in 1h 13m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94b1d5233fd84f97bef2b969d4964e38 : SUCCESS in 45m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61f836aeac784f51af432c1e709fe64a : SUCCESS in 1h 08m 04s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0797deedb1464a9ebe5dede969b8552d : SUCCESS in 1h 08m 47s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee0d4d68cb274825aa68ffaf6bc6d560 : FAILURE in 1h 58m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c036194a9702875dc5919164515565949d20532f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 03:46:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a8d878cb3d4c0530d2945b0edb88f5190b6ff438","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 06:05:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"136af71cfa52dcf41673418989b3af0de1c2942d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 06:05:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/16306c7fde454accaf5ad2133f29dd5e : SUCCESS in 45m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/89841c221f5d4085aa657dc10e6040dd : SUCCESS in 1h 01m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f9b2da216db446bb8b6955cd294e4d3d : SUCCESS in 55m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0ddc03522724720889433a24a0f261a : SUCCESS in 53m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86f3bbab86a1474290b952a9b83c2240 : SUCCESS in 23m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f02fd5015a8493ea348e12818bb547c : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/822fb29309df40e0afb24e30b1eae073 : SUCCESS in 19m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02fbe13cbefa4e15a73a7ec481dc9357 : SUCCESS in 22m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb97c72626ea4e92aa3a4b15dc40febf : SUCCESS in 7m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d5ba69aa1c5d43b6bb382db142949d8f : SUCCESS in 59m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/74739c9e85de48d197cd73a1b2f2d94a : SUCCESS in 1h 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e74af39394e44d2b97f309db2b218bb : SUCCESS in 20m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/36f119afc6bd4e31a795d21db0a6acbb : SUCCESS in 43m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fd19585b7a9f40bd8bba93a274acdaad : SUCCESS in 1h 05m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6e9c60b10e36482cae5e13718b5ba989 : SUCCESS in 1h 17m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4322471600da4194a6b56fca4cf6f464 : SUCCESS in 1h 13m 39s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8664824378804e948e0c3c6b21edc044 : SUCCESS in 1h 13m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69b74d8f7ce94b23a97cb7bf86c2d024 : SUCCESS in 1h 02m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f374cd3ea5f046a2aecb488879c6c5ef : SUCCESS in 1h 06m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23b77c19dcd64fd88593b7e9941895f0 : SUCCESS in 2h 12m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9d2e92b823045f8adcabbdd33cbbad2 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":5},{"id":"49ec3a5d0f2ad9ab754a35a82046d43b81f54e1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 06:07:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875d2f4f310946ceac1a41e7fb0448cd : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa3fa62d4c6445efb5acbb41ed53df7f : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"139b496ef957364caee6861fd16676cdfb76a38b","revisions":{"893cc1b26c1adf310cb33ac628191780713244cd":{"kind":"REWORK","_number":1,"created":"2019-11-11 02:52:04.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/45/693645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/693645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/693645/1"}}},"commit":{"parents":[{"commit":"d15ad2e481b9b888079d55ba46d73ca5b7253285","subject":"Increase waiting time for network rescheduling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d15ad2e481b9b888079d55ba46d73ca5b7253285"}]}],"author":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 01:54:42.000000000","tz":540},"committer":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 02:51:52.000000000","tz":540},"subject":"Dont schedule Network, respecting network_auto_schedule config","message":"Dont schedule Network, respecting network_auto_schedule config\n\nCurrently, if dhcp mapping from a network is removed, it is reassigned\nto the network. This is because of the Network Scheduler\u0027s schedule\nfunction, which considers balancing the networks with the agents, whether\nenable_dhcp is set on its subnets or not. It does not take into account\nthe network_auto_schedule config option. This is particularly disturbing\nwhen considering backends which have their provide their own DHCP.\n\nWith this patch, if network_auto_schedule is set to False, networks wont\nbe automatically scheduled to DHCP Agents. If DHCP is to be mapped to a\nnetwork, it can be mapped using the CLI itself.\n\nWhile it may seem that this change is breaking what is already working,\nbut as mentioned earlier, if there are network backends which provide DHCP\nsupport themselves, they wont need the automatic mapping, which the term\n\"network_auto_schedule\" actually stands for.\n\nCloses-Bug: #1647421\nChange-Id: If1a6a2a174d0f737415efa2abce518722316a77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/893cc1b26c1adf310cb33ac628191780713244cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/893cc1b26c1adf310cb33ac628191780713244cd"}]},"branch":"refs/heads/master"},"cbe41c28285d9b0e506930b3723b34b51861f395":{"kind":"REWORK","_number":2,"created":"2019-11-11 03:08:21.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/45/693645/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/693645/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/693645/2"}}},"commit":{"parents":[{"commit":"d15ad2e481b9b888079d55ba46d73ca5b7253285","subject":"Increase waiting time for network rescheduling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d15ad2e481b9b888079d55ba46d73ca5b7253285"}]}],"author":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 01:54:42.000000000","tz":540},"committer":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 03:08:03.000000000","tz":540},"subject":"Dont schedule Network, respecting network_auto_schedule config","message":"Dont schedule Network, respecting network_auto_schedule config\n\nCurrently, if dhcp mapping from a network is removed, it is reassigned\nto the network. This is because of the Network Scheduler\u0027s schedule\nfunction, which considers balancing the networks with the agents, whether\nenable_dhcp is set on its subnets or not. It does not take into account\nthe network_auto_schedule config option. This is particularly disturbing\nwhen considering backends which have their provide their own DHCP.\n\nWith this patch, if network_auto_schedule is set to False, networks wont\nbe automatically scheduled to DHCP Agents. If DHCP is to be mapped to a\nnetwork, it can be mapped using the CLI itself.\n\nWhile it may seem that this change is breaking what is already working,\nbut as mentioned earlier, if there are network backends which provide DHCP\nsupport themselves, they wont need the automatic mapping, which the term\n\"network_auto_schedule\" actually stands for.\n\nCloses-Bug: #1647421\nChange-Id: If1a6a2a174d0f737415efa2abce518722316a77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbe41c28285d9b0e506930b3723b34b51861f395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbe41c28285d9b0e506930b3723b34b51861f395"}]},"branch":"refs/heads/master"},"366fb343d9d92f0a0e98523325ecf574d8101ad9":{"kind":"REWORK","_number":3,"created":"2019-11-19 03:54:44.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/45/693645/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/693645/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/693645/3"}}},"commit":{"parents":[{"commit":"d15ad2e481b9b888079d55ba46d73ca5b7253285","subject":"Increase waiting time for network rescheduling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d15ad2e481b9b888079d55ba46d73ca5b7253285"}]}],"author":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 01:54:42.000000000","tz":540},"committer":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-19 03:54:40.000000000","tz":0},"subject":"Dont schedule Network, respecting network_auto_schedule config","message":"Dont schedule Network, respecting network_auto_schedule config\n\nCurrently, if dhcp mapping from a network is removed, it is reassigned\nto the network. This is because of the Network Scheduler\u0027s schedule\nfunction, which considers balancing the networks with the agents, whether\nenable_dhcp is set on its subnets or not. It does not take into account\nthe network_auto_schedule config option. This is particularly disturbing\nwhen considering backends which have their provide their own DHCP.\n\nWith this patch, if network_auto_schedule is set to False, networks wont\nbe automatically scheduled to DHCP Agents. If DHCP is to be mapped to a\nnetwork, it can be mapped using the CLI itself.\n\nWhile it may seem that this change is breaking what is already working,\nbut as mentioned earlier, if there are network backends which provide DHCP\nsupport themselves, they wont need the automatic mapping, which the term\n\"network_auto_schedule\" actually stands for.\n\nCloses-Bug: #1647421\nChange-Id: If1a6a2a174d0f737415efa2abce518722316a77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/366fb343d9d92f0a0e98523325ecf574d8101ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/366fb343d9d92f0a0e98523325ecf574d8101ad9"}]},"branch":"refs/heads/master"},"cad1ff23fe843c0d96d41f8eb1c1cc7885e5b882":{"kind":"REWORK","_number":4,"created":"2019-11-25 05:23:52.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/45/693645/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/693645/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/693645/4"}}},"commit":{"parents":[{"commit":"d15ad2e481b9b888079d55ba46d73ca5b7253285","subject":"Increase waiting time for network rescheduling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d15ad2e481b9b888079d55ba46d73ca5b7253285"}]}],"author":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 01:54:42.000000000","tz":540},"committer":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-25 05:23:46.000000000","tz":0},"subject":"Dont schedule Network, respecting network_auto_schedule config","message":"Dont schedule Network, respecting network_auto_schedule config\n\nCurrently, if dhcp mapping from a network is removed, it is reassigned\nto the network. This is because of the Network Scheduler\u0027s schedule\nfunction, which considers balancing the networks with the agents, whether\nenable_dhcp is set on its subnets or not. It does not take into account\nthe network_auto_schedule config option. This is particularly disturbing\nwhen considering backends which have their provide their own DHCP.\n\nWith this patch, if network_auto_schedule is set to False, networks wont\nbe automatically scheduled to DHCP Agents. If DHCP is to be mapped to a\nnetwork, it can be mapped using the CLI itself.\n\nWhile it may seem that this change is breaking what is already working,\nbut as mentioned earlier, if there are network backends which provide DHCP\nsupport themselves, they wont need the automatic mapping, which the term\n\"network_auto_schedule\" actually stands for.\n\nCloses-Bug: #1647421\nChange-Id: If1a6a2a174d0f737415efa2abce518722316a77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cad1ff23fe843c0d96d41f8eb1c1cc7885e5b882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cad1ff23fe843c0d96d41f8eb1c1cc7885e5b882"}]},"branch":"refs/heads/master"},"139b496ef957364caee6861fd16676cdfb76a38b":{"kind":"REWORK","_number":5,"created":"2019-11-26 01:55:06.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/45/693645/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/693645/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/693645/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/693645/5"}}},"commit":{"parents":[{"commit":"d15ad2e481b9b888079d55ba46d73ca5b7253285","subject":"Increase waiting time for network rescheduling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d15ad2e481b9b888079d55ba46d73ca5b7253285"}]}],"author":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-11 01:54:42.000000000","tz":540},"committer":{"name":"Reedip","email":"reedip.banerjee@gmail.com","date":"2019-11-26 01:55:03.000000000","tz":0},"subject":"Dont schedule Network, respecting network_auto_schedule config","message":"Dont schedule Network, respecting network_auto_schedule config\n\nCurrently, if dhcp mapping from a network is removed, it is reassigned\nto the network. This is because of the Network Scheduler\u0027s schedule\nfunction, which considers balancing the networks with the agents, whether\nenable_dhcp is set on its subnets or not. It does not take into account\nthe network_auto_schedule config option. This is particularly disturbing\nwhen considering backends which have their provide their own DHCP.\n\nWith this patch, if network_auto_schedule is set to False, networks wont\nbe automatically scheduled to DHCP Agents. If DHCP is to be mapped to a\nnetwork, it can be mapped using the CLI itself.\n\nWhile it may seem that this change is breaking what is already working,\nbut as mentioned earlier, if there are network backends which provide DHCP\nsupport themselves, they wont need the automatic mapping, which the term\n\"network_auto_schedule\" actually stands for.\n\nCloses-Bug: #1647421\nChange-Id: If1a6a2a174d0f737415efa2abce518722316a77b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/139b496ef957364caee6861fd16676cdfb76a38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/139b496ef957364caee6861fd16676cdfb76a38b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
