)]}'
{"id":"openstack%2Fneutron~694094","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ia1f7807ea695ca216e19e4a563367609efe7c53e","project":"openstack/neutron","branch":"stable/queens","topic":"freeze-neutron-tempest-plugin-version","hashtags":[],"change_id":"Ia1f7807ea695ca216e19e4a563367609efe7c53e","subject":"DNM It\u0027s just a test of neutron-tempest-plugin jobs","status":"ABANDONED","created":"2019-11-13 15:26:22.000000000","updated":"2020-01-30 11:09:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4064fdff1d807a1057b91aa5bfd2b9e0aa4e2eb6","_number":694094,"virtual_id_number":694094,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-13 19:53:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-13 22:55:09.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 19:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-13 22:55:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"}],"messages":[{"id":"3037abfa950313df44b37bb023084576e95d39a0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-13 15:26:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"739cfd282960494b4a5f8738b17eed9545647a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 19:53:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/ef426c2516eb4670be76d350e7186b55 : SUCCESS in 54m 16s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/3ffb647323d444b6ae990e5c515ec9d8 : SUCCESS in 40m 33s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/3e044e99422644cf9bfd8e7c913f7b07 : FAILURE in 1h 28m 17s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/4dbf0ece331c459fa5580fd2663ff4d9 : SUCCESS in 55m 34s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/c697b0134152436da6e05e93a9ec8e6b : SUCCESS in 1h 00m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4585d8364ea457fadc89a9a48691470 : SUCCESS in 11m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9127ad062be5480da6b670c6dbd2c4ba : SUCCESS in 20m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c48bd1d983848c38b28d112a376dcc5 : SUCCESS in 32m 10s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/314deff94bc249b593f18c3c0bbe03c7 : SUCCESS in 7m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/87b93f30de1a41c0813ad13fe9db8d1b : SUCCESS in 1h 21m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5424f91640d949c1900b5bfcebc928b5 : SUCCESS in 54m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db6518e2e454404eb37f81734b772168 : SUCCESS in 1h 30m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/904f5ea15d374f8a985d3c6d37224b41 : SUCCESS in 40m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/58d691bb53c24fde865dd1c91d9a7e8d : FAILURE in 29m 42s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/90dea1ccf87a405289d28cb8917000a4 : SUCCESS in 1h 41m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf60cdfa6a54454f81fce35f00a709d8 : SUCCESS in 1h 53m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/613b3462b20e43ef8d12c4ff9913af89 : SUCCESS in 1h 17m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/619abe989b044955a765dd664c5ef3a3 : SUCCESS in 1h 18m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/51b4c7a0d07f47719d7860abd25d096d : SUCCESS in 57m 31s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f9d7961864447f698ce47555060144d : SUCCESS in 1h 34m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/340fdbc96da64869931cd67693104001 : SUCCESS in 1h 30m 43s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/e14474bcf2a54b5e9b2bc24bae924dfe : SUCCESS in 1h 18m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a2cb22b3b6e4d3da6ead07d3b916605 : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3a2eaabf84849d60249af06a20e97bba4504a9de","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-13 22:55:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694094/1 : FAILURE in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5434467a3ca539fd47cee77b80c0aeca9e96da21","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-30 11:09:05.000000000","message":"Abandoned\n\nCleaning old DNM test run","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0045b12f8d0df315ad5e767a37f3a1d7cc7beed6","revisions":{"0045b12f8d0df315ad5e767a37f3a1d7cc7beed6":{"kind":"REWORK","_number":1,"created":"2019-11-13 15:26:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/94/694094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/694094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/694094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/694094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/694094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/694094/1"}}},"commit":{"parents":[{"commit":"fe57823263b01cf378b2493870b0cc2e657c3b77","subject":"Log OVS firewall conjunction creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe57823263b01cf378b2493870b0cc2e657c3b77"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-13 15:25:37.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-13 15:25:51.000000000","tz":60},"subject":"DNM It\u0027s just a test of neutron-tempest-plugin jobs","message":"DNM It\u0027s just a test of neutron-tempest-plugin jobs\n\nDepends-On: https://review.opendev.org/694049/\n\nChange-Id: Ia1f7807ea695ca216e19e4a563367609efe7c53e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0045b12f8d0df315ad5e767a37f3a1d7cc7beed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0045b12f8d0df315ad5e767a37f3a1d7cc7beed6"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
