)]}'
{"id":"openstack%2Fneutron~694559","triplet_id":"openstack%2Fneutron~stable%2Fstein~Ie119693854aa283b863a1eac2bdae3330c2b6a9d","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1850864","hashtags":[],"change_id":"Ie119693854aa283b863a1eac2bdae3330c2b6a9d","subject":"Switch to use cast method in dhcp_ready_on_ports method","status":"MERGED","created":"2019-11-15 14:52:17.000000000","updated":"2019-11-23 03:12:58.000000000","submitted":"2019-11-23 03:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694559-1574478612251-2b47506d","meta_rev_id":"278f4a02880c201c3e29e897e66c65d5e587638d","_number":694559,"virtual_id_number":694559,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2019-11-18 13:58:41.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-23 03:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-15 16:41:14.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-11-18 22:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-11-20 20:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2019-11-22 22:42:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-15 16:41:14.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-11-18 13:58:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-11-20 20:16:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-11-22 22:42:47.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-11-23 03:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc7aa8bfdfa84a315a8ce73bcc8c0257c6244b30","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 14:52:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb2d87ad0898a8f4e66636256e31d1eede68cbe2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-15 16:41:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694559/1 : SUCCESS in 1h 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f5f5e72330164f0dd1c5de7496138632bf516582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 17:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/1f8a4d57a75a44ecbba4445779ff37c2 : SUCCESS in 44m 11s\n- neutron-tempest-plugin-designate-scenario-stein https://zuul.opendev.org/t/openstack/build/672682f9191a43dabef858f70f12adf1 : SUCCESS in 39m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario-stein https://zuul.opendev.org/t/openstack/build/f0e1d86f28374902bbce5a10fcc40b5b : FAILURE in 1h 43m 00s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-stein https://zuul.opendev.org/t/openstack/build/d47f0289acbb491abc276c938e4ce748 : SUCCESS in 1h 05m 39s\n- neutron-tempest-plugin-scenario-openvswitch-stein https://zuul.opendev.org/t/openstack/build/c712233f2a914e0497320ee7d685794f : SUCCESS in 1h 01m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-stein https://zuul.opendev.org/t/openstack/build/6091b81e029f4cd48e70c7fb84ca290e : SUCCESS in 57m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aec93dd4879749abb1709987c4044adc : SUCCESS in 27m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cca46e8a13e846ca9908c5c40a0055ed : SUCCESS in 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/377ade705e4345459e15b5e0cd16e632 : SUCCESS in 12m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9e0669759474492b2fa0122a6aab5f3 : SUCCESS in 25m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab6b691b3c4244b88e0caab42d86f051 : SUCCESS in 23m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61573afd483349438af84eb8b6b30ff4 : SUCCESS in 21m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca46a68467ef4d1f99f724015ac793b2 : SUCCESS in 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5351da2d977a4d83a3e2b70a1fb494ae : SUCCESS in 5m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/80595b4328044b8691a409f4f54cfbb7 : SUCCESS in 1h 44m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1c96836f755643ebb9d0974e4690aa5d : SUCCESS in 1h 05m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7da33ea57c8c4d29ac274281e3c0a308 : SUCCESS in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/117aac86cf864db38d7fbac8bbc6b734 : SUCCESS in 1h 16m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/da2bf20ef7b148369a0204d3058b29a2 : SUCCESS in 48m 55s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/9a503701ce064cb187445e960cdb5ca1 : SUCCESS in 49m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e0e7c7b263a040ba8b3b8d536539c664 : SUCCESS in 1h 20m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/54ecaa773b464e39a92ac210b2523843 : SUCCESS in 2h 16m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/91f5067ea6fa44588a2226794f1822ff : SUCCESS in 1h 17m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/329a6e988b1544619d8549c78068feca : SUCCESS in 1h 32m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0469159c6e91492faac61190168d3142 : SUCCESS in 1h 12m 38s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c994ad67a60b49b2bf126a441bd2b20c : SUCCESS in 1h 18m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/485c8dc247ef4d82b3fd309b57ad4084 : SUCCESS in 1h 38m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f3d6178f726445c8bcd6f12bd8224ce9 : FAILURE in 1h 59m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b36c8533710c4102928f1752d25a71f0 : SUCCESS in 1h 51m 33s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/fc5bfd034fd64bf08aee4cffd2fbb998 : FAILURE in 1h 35m 02s (non-voting)\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3efff11fcb5641efbf55a3e6e08c7dc5 : SUCCESS in 49m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ee7b9e5d1634ba5a45ba1d21d30cb30 : SUCCESS in 2h 35m 06s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/983c8fdd7aef4009b3fe5bb67fb1e822 : SUCCESS in 1h 56m 29s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/88db6e9041aa4f1aa203268d7250828d : FAILURE in 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2d5735ee675c27cbfa216a2b7b3e973a97f23e6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-18 13:32:25.000000000","message":"Patch Set 1:\n\nDon\u0027t you need a train backport first?","accounts_in_message":[],"_revision_number":1},{"id":"819e78e337a3a1a2d335f7e8946b6eb1104b6854","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-18 13:58:41.000000000","message":"Patch Set 1:\n\nI found it...","accounts_in_message":[],"_revision_number":1},{"id":"3af435ba7cf57f0c2ec6fa63906a8652b20579ca","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-11-18 22:22:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9002c972ddea54cd33b5d09c4ab4c81b284a9df1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-20 20:16:37.000000000","message":"Patch Set 1: Code-Review+2\n\nWaiting on train https://review.opendev.org/#/c/694558/1 before W+1","accounts_in_message":[],"_revision_number":1},{"id":"0b4f72549c6f387e267d552f3bf620c3831f0081","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-11-22 22:42:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30199623dbc1ead7bf45a522939d0e14999f409d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 22:43:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23a2b650e1c69944fb5480ae24363255f45cea4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:10:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-stein https://zuul.opendev.org/t/openstack/build/0f96576855234ba28af87cce77b6cd77 : SUCCESS in 57m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7487b0e4f3d44538976775c002d02e3 : SUCCESS in 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4807e8ebff1b4e78a87ebc4d44cc0148 : SUCCESS in 11m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/093376f1966b494490e6b7b973b3b2e7 : SUCCESS in 22m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/06144f9d25064b56882984b92ecf4242 : SUCCESS in 23m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/892abe79615243bb8ae8b15a17469f1a : SUCCESS in 29m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f35263048ea643cf9d3f3d21642ad3be : SUCCESS in 22m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4cb7c262c246498d003199bca76b06 : SUCCESS in 5m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a5c2595058c44c55b9b0ad3e83ccf8ee : SUCCESS in 1h 21m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e4882516ad8a466a8d0cabee9dfafdc3 : SUCCESS in 1h 07m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5866916629124f9eb98f3acee4785447 : SUCCESS in 55m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c15a437d3c3f4d279e364c61dff9cf34 : SUCCESS in 1h 17m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dc2cb0d32a8e4a479f6fd45e2513ed9a : SUCCESS in 38m 57s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/5c9f460f9e0845d59c1aeb9fa08cbcda : SUCCESS in 43m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/19625bb207064855bd2539098742806e : SUCCESS in 1h 08m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ca450721e4244214b33fc3f15813537b : SUCCESS in 1h 14m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/36fa714d6e294a8499bd1a4388be1cc7 : SUCCESS in 1h 09m 10s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ec068850a184f0d8309a4657215a0b2 : SUCCESS in 1h 18m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f32ad6128db496a8c07e513fac7ff1b : SUCCESS in 1h 00m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f61f8991bdb48f3b1d72a13336eaf4b : SUCCESS in 1h 08m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13a9d63159f043dba3b2e29d2154bfff : SUCCESS in 2h 09m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c9f7423d9448c38d35581fe2445946bf12f09c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:10:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f1fc1e1d614d65af196e68ab4036950c92c3a568","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 03:12:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e1e879cfc34b95a36f69fdf389688a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36977d5d0ce337bf0bcdeaee49ce6e9723a3810c","revisions":{"36977d5d0ce337bf0bcdeaee49ce6e9723a3810c":{"kind":"REWORK","_number":1,"created":"2019-11-15 14:52:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/59/694559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/59/694559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/694559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/694559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/59/694559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/59/694559/1"}}},"commit":{"parents":[{"commit":"7ca6c28f70ac271349b81728ce7511ae4f05aea6","subject":"Merge \"Set rootwrap daemon timeout for fullstack and functional tests\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ca6c28f70ac271349b81728ce7511ae4f05aea6"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-31 22:12:03.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-11-15 14:51:49.000000000","tz":60},"subject":"Switch to use cast method in dhcp_ready_on_ports method","message":"Switch to use cast method in dhcp_ready_on_ports method\n\nWhen DHCP agent reports to the neutron-server which ports are\nready, it was using call() method from rpc client.\nThat caused blocking dhcp agent\u0027s dhcp_ready_ports_loop thread which\nblocks to send info about other ready ports if there are any.\n\nMethod call() should be used when RPC caller returns a value to the\ncaller but that\u0027s not the case here. On neutron server side this\nRPC method is only calling provisioning_complete() method to\nfinish provisioning of ports. And is not returning anything.\n\nSo to make sending dhcp ready ports to neutron-server much faster\nthis patch switch to use cast() method from rpc client.\nThis method don\u0027t block to wait for return value from RPC caller.\n\nChange-Id: Ie119693854aa283b863a1eac2bdae3330c2b6a9d\nCloses-Bug: #1850864\n(cherry picked from commit 1a686fb401eca1843b81292fc88c13a2e6fd274d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36977d5d0ce337bf0bcdeaee49ce6e9723a3810c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36977d5d0ce337bf0bcdeaee49ce6e9723a3810c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
