)]}'
{"id":"openstack%2Fneutron~694568","triplet_id":"openstack%2Fneutron~master~I5e190c3db8bed0a264b911cdf425aa4c9b51f768","project":"openstack/neutron","branch":"master","topic":"stop_configuring_install_command","hashtags":[],"change_id":"I5e190c3db8bed0a264b911cdf425aa4c9b51f768","subject":"Stop configuring install_command in tox.","status":"MERGED","created":"2019-11-15 15:07:36.000000000","updated":"2020-02-02 15:52:52.000000000","submitted":"2020-02-02 15:51:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694568-1580658676225-9ea6cd4b","meta_rev_id":"ec3014abc650f11245303b2648d6ec6d23a7c855","_number":694568,"virtual_id_number":694568,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-02 08:50:14.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-02 15:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-31 17:06:25.000000000","_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"date":"2020-02-02 10:39:35.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-01-28 10:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-01-28 10:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-01-28 11:09:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-01-28 10:07:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-01-28 10:07:53.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-28 10:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2020-01-28 10:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-01-28 11:09:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 13:13:18.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-01-22 08:53:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-22 20:36:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-22 22:39:46.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-23 12:43:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-23 18:46:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-25 13:15:58.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-28 11:09:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-02-02 08:50:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-02-02 10:39:35.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-02 15:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cffa5d32e5b21207a6cadf8f285e9cffe03f56d4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-15 15:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2120f4ff306fd434ab47ece022edc2d77319dbc4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-15 15:09:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d047ee51be7cad9842d3df5a3bb0bba9892afd87","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-15 15:10:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"68afb149d3df82e604e1da7af238091a30d1c5c5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-15 16:14:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0850864939f7aecd8568635d5f9b1117a394dbce","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-15 16:52:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/2 : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6cd95ee22538219be7b26dd57e376b32837612e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 17:00:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2309f34eb364456894649526c59b9123 : SUCCESS in 28m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f96bfa6844654f68b44a1359e531578b : FAILURE in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a3468e61ef4b72816c07bf33aa0866 : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6819f1a1cecd4cb4bbf2a67225e5582f : SUCCESS in 23m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea161e26b023439293f8b7180e36c506 : SUCCESS in 26m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfdd9ba6775849869a10983f73425c17 : SUCCESS in 10m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c66b4bf69aae4d88a16e5eba4f3f291e : SUCCESS in 1h 24m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/455cc85256b745d182f2d2d7e6fd6eb4 : SUCCESS in 26m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/951792503a014b7ea6309acaa9cfd41e : SUCCESS in 51m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/633bf68073904635907e2591487df3f9 : SUCCESS in 1h 05m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/96e3643ffe3b4a73a6344b648d701898 : SUCCESS in 1h 24m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/473e0b256f604cce8a908fb35ab6f859 : SUCCESS in 48m 20s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31c8a74b6f144989a54db0bc50af90a8 : SUCCESS in 1h 26m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2c0e3d40f9333df1a0efe1512882f5f3e063bc6","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-18 09:21:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5790b3b101e5bc8f4ab8d36bb9a0c273a726cd74","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-18 10:10:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fb80f73bd0f99119502b8e8b537abd46339f1bab","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-18 11:02:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/3 : SUCCESS in 1h 40m 58s","accounts_in_message":[],"_revision_number":3},{"id":"87e09caa77f56ed769d630a19611b7da95a2d4ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:11:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b9b9dfab8742ca95e0320c05f3b9c0 : SUCCESS in 31m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/963ce1dbc9a84ec1bab99b3209304092 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a04abc8e3ca442187db293dd5b39d24 : SUCCESS in 15m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7b962a055014c74984f626e5535b91f : SUCCESS in 27m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e98665f42cc840cfb2de1576bfd7af65 : SUCCESS in 28m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b20c26ba334675b180be88bc1950e3 : SUCCESS in 9m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8b971847c6874d6aa87d5b85ffb8ff50 : SUCCESS in 1h 06m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/004b989f3f9241df928ae890bd3d41ba : SUCCESS in 20m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e982f16fc6f148ddb32e87f33506ac5f : SUCCESS in 48m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/38773662184e4c889c62958f6ff39ae1 : SUCCESS in 1h 04m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/51424dee59354018a2371e791d3a9b91 : SUCCESS in 1h 43m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8c44382969948858e480d69ad123385 : SUCCESS in 50m 37s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a335daf925af4b07af634451d4c59e24 : SUCCESS in 1h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f5eac1a2b28666b205d7e78b361c63189d78514","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-18 11:54:49.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2d99138ee5d63c77d87428b328ca2bf0d06c924b","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-18 13:13:18.000000000","message":"Patch Set 3:\n\nWill need to recheck once the results come in - the lower constraints job failed because of ovsdbapp package being too old (has 0.9.1, needs 0.12.1+).  I think the RAX mirror might have some really old files.\n\npip._vendor.pkg_resources.ContextualVersionConflict: (ovsdbapp 0.9.1 (/home/zuul/src/opendev.org/openstack/neutron/.tox/lower-constraints/lib/python3.6/site-packages), Requirement.parse(\u0027ovsdbapp\u003e\u003d0.12.1\u0027), {\u0027os-vif\u0027})","accounts_in_message":[],"_revision_number":3},{"id":"64ea72cbc3ec9b44822410f346dac74ecbdce600","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-18 13:32:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/3 : SUCCESS in 1h 37m 39s","accounts_in_message":[],"_revision_number":3},{"id":"fc4ebb42cc5fd926e901c8466ed6b4832aafc833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:16:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdd0caaa73144a21b3b91f0b50ebab65 : SUCCESS in 27m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbd9f620b7974ddc8e2898a24e9711c1 : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af0b0d48a9a46b7bc8d3c1bf2912b30 : SUCCESS in 13m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ecf633ffc54beb96f0f19e82ceb7c0 : SUCCESS in 25m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7ea2712e4434b2da9a0e8318c63afd4 : SUCCESS in 22m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbaa970c66344e08818198dfc52764ac : SUCCESS in 8m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/31014f22d7ff4fe897dad5346a4f622f : SUCCESS in 1h 20m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e17603cdcbf4c9d910b41857cad171c : SUCCESS in 21m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e6bbf14fc14247caa3627553b62688fa : FAILURE in 52m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e799455072d64453ad0f7f4a93dc7c8a : SUCCESS in 1h 02m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e10e9625a29944f0b50b4b0d43e0e515 : SUCCESS in 1h 29m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee771fc87ae14ed2887914ac3a73c7c9 : SUCCESS in 50m 03s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d78c4b37b4ad41729bc0bcf2db379397 : TIMED_OUT in 2h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2948a0c59f125ca9de0adb31ecf77ba875d402be","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-19 14:30:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2e81448749b80a047a39fe30865027c8598a1773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:04:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e05458bd44b84a3ea3f5cc1fcaf0b232 : SUCCESS in 26m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3fa7c33a97141b2b4795a5e47a06361 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb72c45d8b74755987f71b29c8de7d6 : SUCCESS in 15m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d645a1ac6bf243d1aa34bcc9c03c3b89 : SUCCESS in 27m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca6aa1ed351642e08939fe7b3e921019 : SUCCESS in 19m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09719347f1544d78020fed0e08fd431 : SUCCESS in 12m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/23d2eefa4029456287b2e77cde6e1497 : SUCCESS in 1h 24m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85707b00030e4d1b8ffb92d3201c6c39 : SUCCESS in 23m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/42e505cf55b049fabfc1b97f6370c2e6 : SUCCESS in 46m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e23481c861f94758a16cafc02a9481c0 : SUCCESS in 1h 14m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/654d6a43179b4aad8df0a49e5005e02d : FAILURE in 19m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e84c60843cfb499f98dc5c2783da69d0 : SUCCESS in 43m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f9ada79e63d43eb8e802d6fa2838d9e : SUCCESS in 1h 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73f76c5d2516667d0b4108b7f71778a949e8a8e1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-19 16:14:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/3 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":3},{"id":"46a85b705b975d4cddead7ddedfbeef7920bee01","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-20 10:00:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4994f3abcd6f63b2723bc6cb817fc81c6daae536","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-20 14:13:45.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3cbddcfc12a15f9ee2adcbe1c84ae37086f1e003","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-20 15:33:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0df91b89911c657fd8e723063e2ea7c36d677248","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-20 16:01:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/3 : SUCCESS in 1h 46m 48s","accounts_in_message":[],"_revision_number":3},{"id":"fdaf4fe4df205e447d9173fd887822e75650a6bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 16:25:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2123c21581254693af1b98daa036222f : SUCCESS in 30m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d7fe818c24e48bd9b6987497661350d : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d40c41d9dde74deb841d2e89aff94cc1 : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/070ff34d21824df3bac5c6da3b4d7b4b : SUCCESS in 24m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aece47c329d4626b2bbc5fcf865dbfc : SUCCESS in 21m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5ac0db93ce413093855a71ab555a5f : SUCCESS in 9m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e19feca6d24d4597a79db842f19d4850 : SUCCESS in 1h 22m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b1506a70552401ba405b102c755a1b1 : SUCCESS in 28m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/90dd80caba034691ab695ed201dd8038 : SUCCESS in 53m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3fe5ba458ab347ba882deb13ef211ab0 : SUCCESS in 1h 12m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/53c0097e5c804c7bb8e5f01a7ae7b5dc : SUCCESS in 1h 31m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/357b24c74a364252b34ab16466ba3c55 : SUCCESS in 54m 55s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/91f9dafcff8d40a9abd81b2198370c3a : SUCCESS in 1h 08m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ccb0a21be0ef2b73beacc141d9b9c002a0592a5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-20 17:44:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"88c150d1fe26bb08296c21dc1324b1a539da9aeb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-20 19:34:05.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/3 : FAILURE in 1h 49m 16s","accounts_in_message":[],"_revision_number":3},{"id":"744e451bf20582924b8909d9bff8315e9839a948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 21:13:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a73db1029033475d8f2cf9a5177b6974 : SUCCESS in 31m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26634fe7ca3b469b8e9fc6c377f35ec5 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6614b602e67d4ce785da817befd0d087 : SUCCESS in 12m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1137df5c6b2d40cba137033778bddd91 : SUCCESS in 25m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6497c918dce452a92794cd624075e90 : SUCCESS in 24m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3fc580c61e49f091edf38634cb5d1a : SUCCESS in 8m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/860633bc5d9b49439dc7e1133af69c75 : SUCCESS in 1h 24m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0144956f23c4cd3ae23b05303003a80 : SUCCESS in 23m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a681062529b34fe6bf8847aa7ec97e9c : FAILURE in 57m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6178d65564de40679c7110509caf73a6 : SUCCESS in 1h 18m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/967e065686a54eb1a9dd7a6d90697c58 : SUCCESS in 1h 24m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da23b7d051cc420b8ca77f49e78041dd : SUCCESS in 48m 10s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a62cff5274d3491e8de6331d06a2eecb : SUCCESS in 1h 08m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b99da9e378afd20216538a33aabe36fd43ef7f1c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-21 12:41:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a479738ce9c693baf41b71c2eb3d0352763fcd2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 14:13:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88d1708f114747fcb71d22190765b95b : SUCCESS in 23m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a2708d499b24fca9d3633624c3dd168 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db190e8f184b47f0ba2edb474dad0891 : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9482d8f89904a2696b8823e9ef64e2c : SUCCESS in 25m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31b94372bc2c4845aa904aa613dc304b : SUCCESS in 27m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c958ad0d3cdc47f0ac7bd42dd9dfc6a3 : SUCCESS in 8m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/842038e117014b2689f4c65f14c0c5bf : SUCCESS in 1h 09m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d249b1220c4e47efac6123b2f42d0d6e : SUCCESS in 21m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/54c77ae764e647c59b70303ffc6c2a4e : SUCCESS in 43m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/81bce64f512b4364ae57481a7efdb794 : SUCCESS in 1h 09m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/15edea52fda6458d9ed13be4ed2b8834 : SUCCESS in 1h 17m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de34eb4f0c5f45b5a8cf5717fd2c8650 : SUCCESS in 43m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38708a0e734342f49394465292aea3c9 : SUCCESS in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9c77e5cbafe8e02a9dbbb8d8159f0c4051a6fbc6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-21 14:41:43.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/3 : FAILURE in 1h 59m 16s","accounts_in_message":[],"_revision_number":3},{"id":"95556c585c6c715af6e4f63eb035bf8bf9d6146b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-22 09:18:24.000000000","message":"Patch Set 3:\n\nAs I see lower-constraints job fails always with the same issue:\n2019-11-21 12:54:18.780688 | ubuntu-bionic | pip._vendor.pkg_resources.ContextualVersionConflict: (debtcollector 1.2.0 (/home/zuul/src/opendev.org/openstack/neutron/.tox/lower-constraints/lib/python3.6/site-packages), Requirement.parse(\u0027debtcollector\u003e\u003d1.19.0\u0027), {\u0027os-vif\u0027})\n\nBut have no idea how this change can be root of this....","accounts_in_message":[],"_revision_number":3},{"id":"ad08b7749cb9b2f1b54577a8ec4cce58ae20db4d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-22 11:18:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"095c4f047686f48899705a7d9f9f39768b35f050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 12:48:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b3d2d2c8e48475fb9d09380d061c6fe : SUCCESS in 39m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc7052860e04490bb65bbe900b50fa90 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a4d8ca35764d82a7532f74f5cdc2f3 : SUCCESS in 18m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb2230e4dd543b38db5780151268780 : SUCCESS in 26m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf67e841b2e242e08902b07e8ab594d3 : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a28b11eba44391b413217629675f6b : SUCCESS in 9m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/268d79452194437ca57530cd5c64a0f2 : SUCCESS in 1h 29m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2117cafe7ce43baa1963d6a4725d7f3 : SUCCESS in 17m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d6642d1157444ed8bb0fd75f702dca8 : SUCCESS in 50m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bb40deefe5bd47b79a761c9d36875845 : SUCCESS in 1h 07m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c092cbade59247ba8d3b9382bb2196bc : SUCCESS in 1h 14m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc4cac449e7846868d750462d1bc6b0c : SUCCESS in 47m 01s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ad8cad37297419fa668bc775d36247b : SUCCESS in 1h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c7d9d866bca41fa3d99a9b5ee09fda00ccf4ff8c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-22 13:07:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/4 : SUCCESS in 1h 48m 56s","accounts_in_message":[],"_revision_number":4},{"id":"4868b1a0867892e6b955107f7938f0ad6b956d2c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-26 12:02:03.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a089f0939003756eb82934524d88cefb3589e94f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:16:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47cb4643b77e4e0aa5e3dada225e0deb : SUCCESS in 27m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ecd2220492449c1a1c570d25a87fa99 : FAILURE in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c193bf846ee47c48b6746e3b996c92c : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3986ea2febe148d595dd7afc538deed9 : SUCCESS in 20m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29c4ddf97ef54c97afca01e0656e2497 : SUCCESS in 23m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d190f7c5a4d54c4cb6852d6a14238aa7 : SUCCESS in 7m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c615732c8c98455e8f64e8b78c0b2863 : SUCCESS in 1h 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/012253670f0a4ef8bd0d6b4c922081d3 : SUCCESS in 22m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba1b1255371d4085bd9605a37ab4b04c : SUCCESS in 45m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/adff7e87975f400294633c18a9b38ead : SUCCESS in 1h 13m 58s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/23a3d7b14d374563aec96ddd7c88db69 : SUCCESS in 1h 10m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b2cb1d4107c4e1c905f14e5e88c73fd : SUCCESS in 51m 35s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4907368397614687ba12f4548f426dc7 : SUCCESS in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b20346fa5cfb336615b25b8326f383a85dc70e97","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-26 13:58:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/4 : SUCCESS in 1h 56m 03s","accounts_in_message":[],"_revision_number":4},{"id":"b01e846df39305223b1d549c7d0c25b7e7e80206","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-26 15:08:57.000000000","message":"Patch Set 4:\n\nThe lower-constraints failure is reproducible with this change:\nhttps://zuul.opendev.org/t/openstack/build/9ecd2220492449c1a1c570d25a87fa99/console\n\nThough I have no idea why it causes it","accounts_in_message":[],"_revision_number":4},{"id":"957c1cdc867cef684d482eb32f00d539d3cce29e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-11 18:20:55.000000000","message":"Patch Set 4:\n\nIt looks like this is maybe having issues due to the changes to lower-constraints.txt from the OVN merge, although it should be automatically re-based.  I\u0027ll rebase to master just to verify, but it\u0027s running fine locally for me.\n\nAnd since I see there are a number of other changes with this topic, were you going through all the other repos?  I do work on Octavia and see it has the same lines.","accounts_in_message":[],"_revision_number":4},{"id":"1a48b9780d26fc902877d93a15e6b5a78f35bd69","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-11 18:21:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"996622ded93b26f9f2e9facce006cc40823a7024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:33:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9e269905a1d4e29ab2b4f78dec25211 : SUCCESS in 40m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4267eea9e3be40349b15a1e05eea7bd5 : FAILURE in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd149f626fb49ff8fd4f068994e451e : SUCCESS in 14m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be374860f46445448c9898d2de064abd : SUCCESS in 25m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2267ace4e0aa4937ae7630984c7acb73 : SUCCESS in 33m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0852a3061934a40a01f4603e70b9592 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f937035cd72402596e47cd0bb4d7e43 : SUCCESS in 21m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcf62ef2494c4a289bdae42fb48c2acc : SUCCESS in 49m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f264c06be5cc40cdab1b69867ec2445f : SUCCESS in 1h 02m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/40cdb8223c6c4f28bead73aae91e37a1 : SUCCESS in 1h 08m 07s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fbddf00a9fc4416bbe79b2b2b8c7b5d : SUCCESS in 48m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad07e8c0b08740789cc27201288eb35f : SUCCESS in 1h 03m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5276ba5a177f796357d68d36c08cf035b275ce2a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-11 19:51:19.000000000","message":"Patch Set 5:\n\nThe failure changed, but it\u0027s still complaining about os-vif dependencies not being new enough, this time it was debtcollector.  I just don\u0027t know why it started with this change, but maybe as a background task we should look at moving forward some requirements anyways.","accounts_in_message":[],"_revision_number":5},{"id":"1cff0c83324f5832a6b31bc44daefce73f4f48cc","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-11 20:08:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/5 : SUCCESS in 1h 47m 10s","accounts_in_message":[],"_revision_number":5},{"id":"bd71e2669e561cecacf5516022fe76dbc27d3216","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-12-30 13:04:17.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f1c32f474bfee0fd9c0287ddd730b8a74d507f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 14:19:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ad9e2f04174444d9593308212d38c95 : SUCCESS in 25m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbb3589ee5ed4cccadb444dcca890e97 : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e419481d6544802a086f60813b5944d : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c7361119bbe4c209db4e788183aad51 : SUCCESS in 22m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a607c0dd41543029ec7fb24ffa4bed9 : SUCCESS in 28m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6665ab616d14ab3b171a307980185b1 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/977afd17a95d46c0b2aec9a64c723516 : SUCCESS in 13m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2ab1fd1d7b814b2a9e2c6a9ac84d0669 : SUCCESS in 43m 13s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b67c7133056f40869ed8bfa760a2e77c : SUCCESS in 1h 01m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dfc54c3156fc48978c15c15ed348f1da : SUCCESS in 1h 13m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7623341b80de4349a46105998b09e444 : SUCCESS in 53m 53s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ec204bcb0784d93a24e113ade20e4e4 : SUCCESS in 1h 02m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ecee36cbd2f834d09dccb440caa5e4aef1e11e52","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-30 14:50:19.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/6 : FAILURE in 1h 45m 32s","accounts_in_message":[],"_revision_number":6},{"id":"59ac32b6dea4c65d1a20ec072267ad6fa5cfd58a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-07 11:38:10.000000000","message":"Patch Set 6: Code-Review-1\n\nthis failure in openstack-tox-lower-constraints job seems to be consistent. Can You check if this isn\u0027t related to this patch?","accounts_in_message":[],"_revision_number":6},{"id":"211b1a24d32d71be0bd55c49670eda0b4ef65c16","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-09 13:19:46.000000000","message":"Patch Set 6:\n\nI can\u0027t reproduce it locally, instead I have something like this:\n\n$ rm -rf .tox/; tox --notest -elower-constraints\nlower-constraints create: /home/lajoskatona/github/neutron/.tox/lower-constraints\nlower-constraints installdeps: -c/home/lajoskatona/github/neutron/lower-constraints.txt, -r/home/lajoskatona/github/neutron/test-requirements.txt, -r/home/lajoskatona/github/neutron/requirements.txt\nERROR: invocation failed (exit code 1), logfile: /home/lajoskatona/github/neutron/.tox/lower-constraints/log/lower-constraints-1.log\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d log start \u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n_____________________________________________________________________________________________________ summary _____________________________________________________________________________________________________\n  lower-constraints: commands succeeded\n  congratulations :)\nTraceback (most recent call last):\n  File \"/usr/local/bin/tox\", line 10, in \u003cmodule\u003e\n    sys.exit(cmdline())\n  File \"/usr/local/lib/python2.7/dist-packages/tox/session/__init__.py\", line 44, in cmdline\n    main(args)\n...\n...\n\n...\n  File \"/usr/local/lib/python2.7/dist-packages/tox/reporter.py\", line 74, in logline_if\n    message \u003d str(msg) if key is None else \"{}{}\".format(key, msg)\nUnicodeEncodeError: \u0027ascii\u0027 codec can\u0027t encode character u\u0027\\u2018\u0027 in position 53556: ordinal not in range(128)","accounts_in_message":[],"_revision_number":6},{"id":"cb599b0abcf9f1fd3b93919f58a8865f61cca33d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-09 13:21:55.000000000","message":"Patch Set 6:\n\nI got the same strange output for neutron master...","accounts_in_message":[],"_revision_number":6},{"id":"3cc399d4a38cb2a1b5ca8e0acc8c9c304269b8d6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-09 14:20:19.000000000","message":"Patch Set 6:\n\nOk, it seems it was that I had tox installed by pip2, so after reinstalling tox with pip3 I was able to run lower-constraints.","accounts_in_message":[],"_revision_number":6},{"id":"f68369250115d3e90ecbab51b84fdcea29d37097","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-14 10:17:19.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"eb37a2afa5a1f9136b3aac0c4e7fa5e111bed6d2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-14 10:18:58.000000000","message":"Removed Code-Review+2 by Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":7},{"id":"b21748bbc26a1089b396cbd1dc0fb3620b4f59e1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 12:55:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/7 : SUCCESS in 1h 56m 42s","accounts_in_message":[],"_revision_number":7},{"id":"990bad1eeb8cbbbebed12908c2814982c2a362d6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-14 14:21:47.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"04c3e80ae3ed765cb50d5efac41ff6e79b09d5de","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-14 14:23:29.000000000","message":"Patch Set 7: Code-Review-1\n\nDaniel: could you please check my comment regarding changing debtcollector to \u003d\u003d1.19.0 in lower-constraints.txt as it is in os-vif\u0027s lower-constraints.txt?","accounts_in_message":[],"_revision_number":7},{"id":"03e6fb517676a3f22bffaedb7da4562af589501e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-14 14:38:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4993a9b07e8766f4319b80747b79f6bbaf5e1e69","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-14 15:18:28.000000000","message":"Patch Set 8:\n\ndebtcollector bump seems to make sense, not sure why it gets finally detected with this change but +1 on the change\n\nI\u0027d like this to be mentioned in commit message though, to avoid later \"why did a pip cleanup change requirements in this commit?\" head scratching later","accounts_in_message":[],"_revision_number":8},{"id":"ca4e3f5bd0fbce0fe8900911b2dc09618444ec57","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-14 16:08:05.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"ebd74c9f7c323c24a4495424a05d994b6ddf8299","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-14 16:18:38.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"3fae0b0ecfa25c86aa3789f0c6e1f482e1326540","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 17:58:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/9 : SUCCESS in 1h 49m 23s","accounts_in_message":[],"_revision_number":9},{"id":"5d7f9c10a4dbcb0d1ee63d309ae053bee96cba8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 02:33:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/211a479a9b9e41fb9ab25bbcda877859 : SUCCESS in 52m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48ce2afd5fe948e587f48542939075f8 : SUCCESS in 36m 26s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/60f79a79afc14063bbe1e97c32459473 : SUCCESS in 1h 42m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7cd3268addad4eecabce95bd20fe1749 : SUCCESS in 1h 12m 57s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e099fa82359e4853a55e90abaf22fa55 : SUCCESS in 1h 08m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3d517a1dd1304c88931e7bbe82cfb97a : SUCCESS in 1h 01m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8a319e8a7244029ba59c60b04b29901 : SUCCESS in 37m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e5a277089f94daa919625fd3a7f3911 : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda0172c0b0946318907de9c0af7b839 : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbda90dab38e42e9a10efdc0b8fd6f2d : SUCCESS in 26m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c8ac058aade421daa75fdb3f0109056 : SUCCESS in 29m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f695840b0249b78b901606a7873fd9 : SUCCESS in 9m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e92b5817b6924a8ea6939ad421de718b : FAILURE in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3fe419c1d004311a2f23668b829b0fe : SUCCESS in 15m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7e819cd536814b60961a960e1b7dc442 : SUCCESS in 47m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ae849c33d94f46dcb747f71bf9ffb396 : SUCCESS in 1h 14m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/65d283312ad04463802461c7cae2f110 : SUCCESS in 2h 42m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3fdbde2d56ed41568e23a19cac3a24e0 : SUCCESS in 1h 16m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f03214006026490eb3e60baeaf8ee0e3 : SUCCESS in 1h 17m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d283a11e0f6d4f4d99fd15a68e55f077 : SUCCESS in 1h 15m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4b625412d0745e0833c7d67c47daeb6 : SUCCESS in 1h 16m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dbff4724a9dd474fb206fb317cf66bdf : SUCCESS in 1h 16m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/de179af56e3847708a0e830eb496ad5e : SUCCESS in 1h 42m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5b4c8ebd501d45889f260c674444aa82 : FAILURE in 1h 35m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d3168900e85a4ae4b6ef9a1d3bd1dcea : SUCCESS in 1h 15m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c22c6bb5d48a4825a9fabe6f4af1ee10 : SUCCESS in 57m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70fbf675c995426693563edd905feb23 : SUCCESS in 2h 24m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b34b3b04c48948c69c036f071844384c : FAILURE in 23m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/079c46683ac6405f88db897e87e33e5b : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/512ed04f7e1f4083ada2b89839218231 : SUCCESS in 1h 11m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee325b35e3344351a42df8d01d323fcd : SUCCESS in 43m 14s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b822375da7554ea1b2c770a1eb793be3 : SUCCESS in 1h 03m 22s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/142cb1b968744de68d59c2928ab157ec : FAILURE in 1h 39m 50s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/418429e1f78940e9b4bc71872679d880 : FAILURE in 1h 34m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1fd3892f18f44c9c86698d3ba0b42d92 : FAILURE in 1h 04m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d734ab841a95415cab433bce5c344e8f : FAILURE in 2h 01m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1d66d4085d45686173cb56ade5bce44bb36a4c8d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-15 12:41:58.000000000","message":"Patch Set 9:\n\nIt seems to be a terrible iterative check, with zuul as (at least for me) this new lower-constraints failure with keystoneauth1 is not visible in my local environment....\n\nbut with tox -erequirements the other failure is visible at least locally:-)","accounts_in_message":[],"_revision_number":9},{"id":"fefe6fa0ed6e22e21318b0534992bfec575d0fbe","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-16 09:57:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4dc6a9326fecc7e41b226fa3eb65b202f78fc651","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-16 12:04:49.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/10 : FAILURE in 2h 06m 33s","accounts_in_message":[],"_revision_number":10},{"id":"5070d0926cba8209fb04daffb734dbe38b1bda33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 13:13:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df1389e06c744903ba101f5608607820 : SUCCESS in 1h 04m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/221c945b92af4d10b6364e05d2d96f4c : SUCCESS in 35m 35s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f81fe313bd9e481b98124b2267321364 : FAILURE in 2h 27m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/5b7b98734cfa4112ac66316143a88c85 : SUCCESS in 1h 16m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f14847c6083446a39fe03c7ff458d9f6 : SUCCESS in 1h 05m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58d79eabc13f4a4ca4a6fdcaa8050b05 : SUCCESS in 1h 05m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34aeb853ff7d4c4591430698ecf82d44 : SUCCESS in 33m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de4abb596d3240dfb37bfc631564da82 : FAILURE in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38e9a2ea79c74737b7302ddd787fdbec : SUCCESS in 13m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e7d71cf4f6488b9b7dd309c9b37a2d : SUCCESS in 26m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4080402e57ce40169aef1b7e237612c8 : SUCCESS in 30m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61bd9640652c42dfb8dc81bb1ac1a16e : SUCCESS in 11m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/967ce28ddfd34be485fb64d7ac3f7da9 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d1dd541726b4cf68bd3a6f4b435e4a7 : SUCCESS in 15m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db1354610dbf4bf191318d80ea4acdd0 : SUCCESS in 54m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3e8d1953499a43e1976d3f80f0976031 : SUCCESS in 58m 55s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ea2e884b8c5a481d97520c258b71fb75 : SUCCESS in 2h 02m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9c7e55abeb194363825d68cfb79cfab3 : SUCCESS in 1h 11m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bc46efbb696840d3849188a0637b3110 : SUCCESS in 1h 14m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db64ba37f7dd4a7a87dbc8fc9e9d30fe : SUCCESS in 1h 07m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/602b623dac8d442097d1ce2cfb98f3d6 : SUCCESS in 1h 09m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b69a15a3ea7142e2a355b180ab4d559b : SUCCESS in 1h 11m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c02da2e9735748d190b3a37aee2879d4 : SUCCESS in 1h 57m 50s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a35b69abe1b746d09d065ecbe03e3b4f : FAILURE in 1h 40m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d44777a1755844d3952dbf25451ddd78 : SUCCESS in 1h 12m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/54edae2e491f48938455e42e4bc71fe9 : SUCCESS in 59m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b88bca6ce6814482b825d18e0418036d : SUCCESS in 2h 58m 24s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ab0bd0220ba74d7e882c282f7cde9f41 : FAILURE in 24m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2b9bba05cff489ba8800ce120c2936c : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/745a6a7e963e4f8a86192cd948e4a9b9 : SUCCESS in 1h 05m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9927e2f46bdf48ef9c2c76a6eca8de75 : SUCCESS in 43m 22s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3c7aa7cb61747edb18d906854617994 : SUCCESS in 1h 02m 18s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c46f9957bdb74142b1cf9226f80888b4 : SUCCESS in 1h 07m 46s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/94c1f96ddd5f4823bfac4452ea226f86 : FAILURE in 22m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/61b5fa19086744b796ff33be359e2f2f : FAILURE in 2h 54m 26s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/13254459ffff4fccb8915894593dbd33 : FAILURE in 1h 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9599d61dcaf7b5a2be9f55ef00c33af13dd05355","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-16 14:31:48.000000000","message":"Patch Set 10:\n\nLooking at a recent zuul run on another review, it looks like we somehow missed that failure even though it was reported:\nhttps://zuul.opendev.org/t/openstack/build/1b3a66ec673948f0b2bc19be9cf71cdf/log/job-output.txt#663\nand the job showed as \"SUCCESS\"\n\nAll in all this change is a good thing then, because we now need to have proper lower-constraints!","accounts_in_message":[],"_revision_number":10},{"id":"87df6e26992bb4cbe1e4697c4a6d33312990385d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-16 15:31:49.000000000","message":"Patch Set 10:\n\nNow I can reproduce the issue on local, I will fix it.","accounts_in_message":[],"_revision_number":10},{"id":"2044beab55b77e476a958b3024da4906def9a8c5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-16 15:53:29.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1f3c0505936d1d08b29d53cadc2bcdfad90780d4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-16 15:54:12.000000000","message":"Patch Set 11:\n\nI will cut the patch, but I want to see if now it\u0027s green.","accounts_in_message":[],"_revision_number":11},{"id":"8056ee1582c7c73644978fc7851ffbc799a5bf57","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-16 18:01:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/11 : SUCCESS in 2h 06m 54s","accounts_in_message":[],"_revision_number":11},{"id":"852335ae3b7a55d48169d754b3265b35bf6513bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:36:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a10bec29449146738d23247f80af4f7e : SUCCESS in 50m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1c87afac33740a489e1e7060aa4681b : SUCCESS in 34m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/16d6985b5f284930bfbeb0c762df7e36 : FAILURE in 1h 48m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6c0e444872e240a8a722a081553b0e46 : SUCCESS in 1h 16m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/050436c1df924d59a882833309ccc3d7 : SUCCESS in 1h 14m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab5563a5e1754ca8b0dad9a4325744d8 : SUCCESS in 59m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf43e760052e4887a46e14690b35d37a : SUCCESS in 41m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d857d4e8db8458684275a9a28257910 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6af66bc77b4c89b628a2251789f72a : SUCCESS in 18m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3134503e54e44008bd097490fa519cf : SUCCESS in 28m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39bcec06186a40a29bff2bb911d948cb : SUCCESS in 30m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7cc027cf3ac4a2288a26dd3aa909646 : SUCCESS in 9m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3831c0f1158469a9c04ca91f4c1eb20 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fee829a5ef114d71ba7a86cafcc9ed6f : SUCCESS in 17m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e1163002eb474a33a1c7654644111f38 : SUCCESS in 42m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7cd29ce8bcf2457c8a0bb516d48ebb38 : SUCCESS in 1h 03m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/91adf0f9106a40468523c0bf97f3b1bf : SUCCESS in 1h 54m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cf711660e31e466da058b3edc0053528 : SUCCESS in 1h 19m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c124d52c4cb64e5c8ea1389e88ef0212 : SUCCESS in 1h 18m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8db7b5adca9648118aeeecd3713c10ee : SUCCESS in 1h 41m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/805af5cfff374091b099e66d52ed1c4c : SUCCESS in 1h 12m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7487e09a8cdd4c26b0a24eb84724b33b : SUCCESS in 1h 07m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e518adfe47b14b849e20015b8f4476fa : SUCCESS in 1h 42m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/db3b64ef4cc1439e91d5767784176092 : SUCCESS in 1h 51m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62a262b54a1b44a6902c2fbc9f3961eb : SUCCESS in 1h 18m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c044e99e2f1346108441e566213d6ff8 : SUCCESS in 1h 07m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38bc822e45134b74b07caaa4f9c04a5d : SUCCESS in 2h 56m 54s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/72862e479c0b4c9e9f2551625ed527b2 : FAILURE in 22m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39f31b349721495e96ac19092c1747fa : SUCCESS in 1h 17m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b662ccb7e275484f8430954e23028af1 : SUCCESS in 1h 31m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/231204af90f847079f4c14f59693fbad : FAILURE in 54m 21s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/03263ce322c441fa92386e9b68f395b1 : SUCCESS in 1h 08m 44s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0e2550479c54a99b8f9edc805c12e7d : FAILURE in 1h 05m 08s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2d48135d1a9b4133b7f24b0c41c3c013 : FAILURE in 39m 53s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/503bbe615c844d64b4b94e8579b439d7 : FAILURE in 2h 20m 29s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8ed5518340134e49a61cb6803656bf66 : FAILURE in 2h 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dd59ff04d16d11c2e33e018b3dcf6cec0fb724aa","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-17 08:00:32.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d82f86966908c290a22734d2fef6d20a7cb4166e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-17 09:45:58.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/12 : FAILURE in 1h 44m 52s","accounts_in_message":[],"_revision_number":12},{"id":"44a4d8ce1dde51742f47cd6a0c7f2826ed15adfd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 09:58:52.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"6fe47a43748de33c2df9d7e8131a879bcfde4baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 13:54:30.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ad9196a24bdb46a1873d8083520c7c85 : SUCCESS in 59m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a21e598f6f854516a01a4493eba434ad : SUCCESS in 39m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/59677d421a594b15810e9e1c6ed83061 : FAILURE in 1h 55m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4de8b1a92d5e4412a7fc3e6da48c7856 : SUCCESS in 1h 11m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/889040c8c10d4dd1a68035b329d32a0c : SUCCESS in 1h 11m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c3bb001cafa4ea29a1d4804a452450e : SUCCESS in 1h 05m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/122b103ccf0146ee9ddf575981f9ed5f : SUCCESS in 36m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/392427316b23443ebc6f0438c9e68265 : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54277cd9458d456caa217c6fa3b4b9bd : SUCCESS in 14m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a16cce5164142bb9adc0cd2a49bf9ef : SUCCESS in 27m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2370229ca8b94694924cb7fa66e68ac6 : SUCCESS in 31m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740be91a71c84cdfaa8bb3c4897fc6e0 : SUCCESS in 11m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f48a2c3731b438984da52e4aa90bb77 : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/276e98a717e74b9db33df087d63b6359 : SUCCESS in 17m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e85af06357d9483eae61db562d1bcbb2 : SUCCESS in 47m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0be2bc6b9acd42fc935d71a341de2ac6 : SUCCESS in 1h 04m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e45999bd05cc48f3b6035a30fefef966 : SUCCESS in 2h 07m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c14111e6720a484ebf36c01b1cb5bb8e : SUCCESS in 1h 24m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0d6cb4c3fd340d48c9ee0040f34c767 : SUCCESS in 1h 18m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c35a4b92b0e94d51b3fe8d03f8553099 : SUCCESS in 1h 19m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5e758880e20b4de18d5f38b0075b3c6f : SUCCESS in 1h 21m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05e9c7698e894592881e1219fda4f49a : SUCCESS in 1h 22m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f2ea657dfc054952905ce881d77a7ae5 : SUCCESS in 1h 50m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/768d274ae09f4bb2b2c91103a2b7ccab : SUCCESS in 2h 03m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04ef7523948547cc99d5b4dcdeb615f8 : SUCCESS in 1h 20m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/28f9225915cc407995ef83c24e216d66 : SUCCESS in 58m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9750c7e1e91438eb879a11e3690a920 : SUCCESS in 2h 33m 40s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/6723744200154961aa179fda489ee47a : FAILURE in 23m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbeaed70182a493f880370e785dccb28 : SUCCESS in 1h 17m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f4e2ea0efb2d428abd715b4ecdc7cf2a : SUCCESS in 1h 40m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0223f314d13e4c589c80e26705b7182b : SUCCESS in 46m 58s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/77ae773c55d745029f0022ec6b4fece4 : SUCCESS in 1h 24m 16s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/02411631ff3c4859b56a5032d136674b : SUCCESS in 1h 41m 21s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f4da0d3f38eb4420a17f441689405b86 : FAILURE in 41m 21s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f926c31e793a4d4d976fe6fd65b0f3d8 : TIMED_OUT in 3h 04m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/86d9ac0fffa046cd8dbe5db82734fa34 : FAILURE in 2h 40m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0a16986dfa681020a2437f1cfcb8ababc4475a84","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-17 14:28:16.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"420f508b11673d3cba6e98d3c7c6b142c5d56172","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-17 16:20:47.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/13 : FAILURE in 1h 51m 56s","accounts_in_message":[],"_revision_number":13},{"id":"99b4367d6f0e0bdb743c3fb01f10827e0ea968b0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 17:35:51.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"f9a1195b9ef57809eb493fd90402303527154185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 23:32:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f6431401cd654d12a8cde201c9c66c31 : SUCCESS in 1h 10m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ea4e658215a4937a9d27c1de2bb00ed : SUCCESS in 52m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/772109c3b39b45faa83ada1ec349f764 : FAILURE in 1h 46m 48s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/197ffa42648f4691bd9f1753a9b8f8bb : SUCCESS in 1h 09m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/18fe2bcf513e497a9adada99d9dd6ef1 : SUCCESS in 1h 13m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d49b6523a2564914a2adc65728db3e29 : SUCCESS in 1h 22m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d68da621c58e4d31bcf280968a062a63 : SUCCESS in 39m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35de4b42644348b6b3cd2e6813271646 : SUCCESS in 33m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9053fb3eacfe4871af2dfb97986b3814 : SUCCESS in 15m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96b116c9ab104ca78fce526b5f180bfe : SUCCESS in 41m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/042937d55fcd4e5281bf9e27b1af3199 : SUCCESS in 33m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b68349643a48c1bb4801609221ccc3 : SUCCESS in 9m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/913d457573014707858b01c245a9478e : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/497bdf5537864c4f8134594504693f69 : SUCCESS in 18m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fc1a6bc871ef42ecbfb02d2505f78e9c : SUCCESS in 52m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6c71a84f5fbc4625a63ffc6fd66058d5 : SUCCESS in 1h 08m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/45c52b0173bf4f1b8751bd69383e13ef : SUCCESS in 2h 18m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2dc94accc57b4b9ebd66111979c06f0f : SUCCESS in 1h 07m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2f2d98e61b4743b69ad7969d6da63d08 : SUCCESS in 1h 42m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8d642c3678a4c9b8b3b2bc9c0bf5baf : SUCCESS in 1h 48m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c97c08b7ce74caea7c82a91a4152a91 : SUCCESS in 1h 19m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7a84e9419a2c43169fe43dd8bdb270e9 : SUCCESS in 1h 45m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cac2e7379726412da08d0e3d3cfcbad2 : SUCCESS in 2h 04m 02s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d0562c6fe0944fb9e838eabd743823a : SUCCESS in 2h 30m 50s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08e61373b556452ea01e015a45ab1da7 : SUCCESS in 1h 10m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e0a6da8ca99e43e8b55aef7d951052ee : SUCCESS in 1h 21m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b37a182d19cc44d5b315971bfabad862 : SUCCESS in 2h 23m 11s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4d618fe4a5674bb9b30c4c11e7cf4ced : FAILURE in 19m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b445b3a739a8471b9fb86cd39c865e2d : SUCCESS in 1h 09m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/54fb644d27384c46b2aec18975b771bc : SUCCESS in 1h 34m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4b155b9e3e1404399f77f041e55130d : SUCCESS in 54m 47s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/43fbaa141ee24cd199163e676b7c016c : SUCCESS in 1h 20m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1536c9c16c024dfe8b52c98c564dd169 : SUCCESS in 1h 39m 04s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/924c2ff6098944f1b900f3875dd13989 : FAILURE in 34m 35s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bb63203c8dcc4468aa64359d2fe7609f : TIMED_OUT in 3h 06m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/73d185b816d14158adf4761c45dd161e : FAILURE in 2h 00m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"379b8ebc5191f62ddd98841c972a51946f708b35","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-20 07:00:44.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"a7ed22a0453d1129db168949459a99d43ec37c37","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-20 08:52:30.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/13 : SUCCESS in 1h 51m 12s","accounts_in_message":[],"_revision_number":13},{"id":"735e99045844dad37dc794038d3cf0184d668027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:09:05.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/83bb0ef506be4d5e98dacd29af3f52cd : SUCCESS in 1h 08m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/01ed4a2558704877bb2cddcf7adb9fbf : SUCCESS in 56m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2044cc95d96147948e2bef6ded755262 : FAILURE in 1h 58m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0de6ee082be545588b4118ce787efe5f : SUCCESS in 1h 17m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9112e2ce88034e77b757633d1ae938d2 : SUCCESS in 1h 17m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e0877feef72749969d41b33bd6777e73 : SUCCESS in 1h 27m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9ae011cc5f24e478092e64ef4751690 : SUCCESS in 35m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/486471a0473a41b58b3407ab148990bf : SUCCESS in 31m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530dc4bda4f345b180b79500f882fadc : SUCCESS in 16m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db5fa66e02ba46fd9ba2315e8c55272d : SUCCESS in 31m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/028491bd30654a18a9fea9d223caf66d : SUCCESS in 36m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2578dc72cf5489586c9a7b81d5153ee : SUCCESS in 14m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc6c7532506e40698194210cce4aedac : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ff9117a40d145ca91544e812707087d : SUCCESS in 12m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b5d2047a8bba4e17bb81d6ab0cc1e8d8 : FAILURE in 20m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0e1b8749633841cb83aef44bc93a93dd : SUCCESS in 1h 33m 41s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/8cd5196d83b04717be78dd6ec0b84d26 : SUCCESS in 2h 17m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/df65bdc2d3d74bcdac4d185c4cc1e9fb : SUCCESS in 1h 10m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/870b6ece7fe54cd284a489bce8ea1991 : SUCCESS in 1h 56m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6244a6fd1a94f1fb998747c34f3be81 : FAILURE in 1h 23m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e24f57d923c4262879b36e157cdca27 : FAILURE in 1h 45m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c5326434f33f4f5389fc773141e5a960 : SUCCESS in 1h 49m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eba74f72dbb54b7594be07a3d71689c1 : SUCCESS in 1h 35m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/88697963ede246e598b3016a71b596f9 : SUCCESS in 2h 07m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c45a09b41b0e4e319dd2655d8c598348 : SUCCESS in 1h 43m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/037f57b553494d9c8ecabee76942b979 : SUCCESS in 56m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9514d36fc1b41a38bad19bcf57c93c8 : SUCCESS in 3h 05m 38s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/830732d2d38a4104890416c5435ed3cd : FAILURE in 22m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3c205081e8b48f893975903b8f3227a : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2bda55dee8ab49f39493bd217415bceb : SUCCESS in 1h 11m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8cbf376d425469fa487e5f971250985 : SUCCESS in 55m 53s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8292f6e317c04f0e9b49ef1e9db50127 : SUCCESS in 1h 18m 46s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2779b0a187e2425393fad59fdd2003f7 : SUCCESS in 1h 41m 59s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/603f7eeaba6f44c1bdb4ac55d2a85843 : FAILURE in 33m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8446eac450634145a899d07ac758f974 : FAILURE in 2h 43m 06s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6299f03f07b343519612408f9f234fa5 : FAILURE in 2h 08m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ee8871b6ff6da5087e273b69b45a5ceb35f5130b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-20 12:02:21.000000000","message":"Patch Set 13: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"fe2a1c0aba5beb7c5a2a91e534659b382a41398a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-20 14:27:37.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/13 : SUCCESS in 1h 59m 11s","accounts_in_message":[],"_revision_number":13},{"id":"aeb249c59fb70dd3c5a9f5dde51e81431dfd4827","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-20 14:54:40.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f414063c5e1aaaf2a91636cd0f04134eaf81f3a3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-20 15:08:34.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/14 : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":14},{"id":"9eb9676782d0c092ad0c5bd0d8752a9eb560c367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 19:47:52.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/213b4cec78334ff4b82c4577eeac2746 : SUCCESS in 49m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0ee29554de8f4838b35e7642a0113895 : RETRY_LIMIT in 5m 12s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9abee0b63ecb41ccb84f5c90c595101b : FAILURE in 1h 46m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/722a10396a6847bba9a46a8e5a52aead : SUCCESS in 1h 05m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4ce8b71bbbd349d4ab0261984ff55fd1 : SUCCESS in 1h 12m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58f187b0db65474ab4bf3014d86a02a0 : SUCCESS in 58m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50bb05e72a794e67b1341d33dd51b828 : SUCCESS in 34m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0166dd82c5374a6fab4ba737d98530fe : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425ed8f706d74c4780560f90c60e9c9b : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e55e066beaad4f69b0632af910c22120 : SUCCESS in 26m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4428aaa24a2d4aa1ad4b679795f0e5c2 : SUCCESS in 31m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f6606f263e54ff7981c777723c81429 : SUCCESS in 12m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1e59c203bb84ac38beacde1911562b3 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edb305350e62498c9160cca1b3c6358d : SUCCESS in 16m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/011a86cf8fa44e3eb551d1fd855c6c61 : SUCCESS in 49m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/120550bf32074f0eb0cf08cd919128e2 : SUCCESS in 1h 10m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9715b37654144bf7ac8d5ae6ec092dd7 : SUCCESS in 2h 27m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8df6dc0a20764c5e90ef840a12068997 : SUCCESS in 1h 10m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/22ade552fde64209823bef687b1ae778 : SUCCESS in 1h 08m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e4b77f42ad749e88a630fc3113071f9 : SUCCESS in 1h 07m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5055b06f56ae445297ce8654877120fd : SUCCESS in 1h 08m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3361207ed89a46cd979a95167346abc1 : SUCCESS in 1h 39m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5e29b42d121040aaa7a036630bb3e3bc : SUCCESS in 1h 46m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1ff57d888fe465ea87a6f9c8d372432 : SUCCESS in 1h 54m 48s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/46af088bf6794e82a765f45aa1f5188b : SUCCESS in 1h 11m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a273ba523df84cf481f0e3cc68753c58 : SUCCESS in 51m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b7fc9d2aabf4e1f86681b3fee9a0c74 : SUCCESS in 2h 20m 10s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/09b7576c966641e2ba8fb9512b50d5ec : FAILURE in 17m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c77578d984840d39511513c2290520c : SUCCESS in 1h 03m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/84c002da606744349f53a329b7abcd03 : SUCCESS in 1h 21m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/afc897c805cb4b49a5dc60adb850d982 : SUCCESS in 45m 07s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af69cd6667f845deb397902c605266d6 : SUCCESS in 1h 10m 05s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9812fe6f8ec2445cbaf563d83575c05e : SUCCESS in 1h 19m 33s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e8818c6eef68476491d21e922fe60d2f : FAILURE in 26m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7233a1bafced468594ad68cdbb111554 : TIMED_OUT in 3h 09m 02s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/622bda1ccd80486e9c1eca60b0c8b367 : FAILURE in 2h 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"eb42ca3145886e9601dca0273ee0eded88d68b88","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-21 08:53:07.000000000","message":"Patch Set 14: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"499f25ca3f2ff2ea70fa79cb3e55ba635328972b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-21 09:40:51.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)\n\nLGTM but I made some suggests inline","accounts_in_message":[],"_revision_number":14},{"id":"e947727f87eb8fd00b473f7c71a3f3c35bade23a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-21 09:50:00.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"f49dbdc8e71ff838940eb400ec1aca8ac2e4d6fb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-21 09:50:04.000000000","message":"Removed Code-Review+1 by Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":14},{"id":"934a095346f577722b5a4d1062195bec5e2d797f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-21 09:50:11.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"84bdfc4fc303976fc184ab1f3a38e4d0387804dc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 10:42:06.000000000","message":"Patch Set 14:\n\nI agree with Herve here, please add info about bump of requirements in commit message too.","accounts_in_message":[],"_revision_number":14},{"id":"ed952dbad78e19086d41e21b73f5dff26cce6863","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 10:42:11.000000000","message":"Patch Set 14: Code-Review-1","accounts_in_message":[],"_revision_number":14},{"id":"5e6ecb1beaba120f6976b59600381bf8ed326e75","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 10:48:23.000000000","message":"Patch Set 14: Code-Review-1\n\nI totally agree with Herve and Slawek.\n\nThe issue in lower-constraints is shown in [1] for example.\nIt looks better to clean up all of them.\n\n[1] https://zuul.opendev.org/t/openstack/build/241c79207e7841b5a3c4a2ff46e450ef/log/tox/lower-constraints-1.log#413-417","accounts_in_message":[],"_revision_number":14},{"id":"79d49e173e3b7045219d3b6fc82fa122b0143863","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 10:57:48.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/14 : FAILURE in 2h 04m 13s","accounts_in_message":[],"_revision_number":14},{"id":"c902855c7d2de5b93f013638c7b72461b1bd3084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 11:51:37.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/43072cb2038e4507a10a8642c4df66f8 : SUCCESS in 48m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00b8465ef21c46d4a317319a88ef97ef : SUCCESS in 33m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bbbce7e5acab4a1ea18b2df8227c616c : FAILURE in 1h 43m 06s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6b93d3d38ff14237a7a28b2454c473ac : SUCCESS in 1h 15m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b540c14502de4b44a8a5fd4afe07b5b2 : SUCCESS in 1h 02m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8defeac0e9b04e06b49f94bf30df74ec : SUCCESS in 53m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8a45eca97e7431fae8bb0ff270b2ae1 : SUCCESS in 31m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2604ea9e8f1a486cb809137ac25d5063 : SUCCESS in 25m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86019aa2c24b4523be3b66b474620a6a : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/479e7c5765cb4c919ab8408a7a6e7875 : SUCCESS in 27m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9765cbed7cf542c58804ecb4f376fb00 : SUCCESS in 32m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bb56d0f7d6647848314c23cec951a96 : SUCCESS in 10m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a568cd01800d4691a8ea938a34f635f0 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e5f670e5b444c3f8dbd8b47b26b7d57 : SUCCESS in 15m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c0c7721879fe4d3ba0d5bae8bc340d1e : SUCCESS in 46m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/abd376aead464c3594cd6593c2a774ed : SUCCESS in 1h 04m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/966ace66d6b445d58781970a16b17f23 : SUCCESS in 2h 21m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/154b4452fbe44ab9901be357f19f3c3c : SUCCESS in 1h 13m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bd0958cc5da34503bd62ca9a55a35ea4 : SUCCESS in 1h 13m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5aef1444c08e4a94a5017a8d4f46eec5 : SUCCESS in 1h 18m 58s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5e74d532b91b4aeda111a69f1d5c9275 : SUCCESS in 1h 16m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b4f29994e2604ae8a23dc2b7ce6a7199 : SUCCESS in 1h 13m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3438a355f5564036a538340a104d5c35 : SUCCESS in 1h 39m 08s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c0c358d507dd46b7945bf165f1ab248d : FAILURE in 1h 38m 18s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a59b5a61429425b8e98d8441f4aae3d : SUCCESS in 1h 09m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/913183d54c8c40f18e7c7c237b930e72 : SUCCESS in 52m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1ffebfab61b4f6ba2c16974d9d0e019 : SUCCESS in 2h 19m 41s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ba458506e1a243dd93b7472de9045021 : FAILURE in 18m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de0e54a727464f63bba0218134490ffe : SUCCESS in 57m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/829ab7bce1a24fafb2e403027c72aa43 : SUCCESS in 1h 08m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d231ccd8c7f44bf38f95dd72ff204825 : SUCCESS in 41m 39s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/41a32fba6bf846d1b80ed8a10ebb75d2 : SUCCESS in 1h 05m 35s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/35c0c9c162654a7eaeac512a3cc8cc6c : FAILURE in 57m 43s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/31c6829fe17047468cc63ea1661c0367 : FAILURE in 34m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/74d92e82b7d4403ca389615ffd908673 : POST_FAILURE in 2h 52m 56s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bbd6566976c842439da1b9bf70876f46 : FAILURE in 1h 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"97777130a0b1b6aaac7efc168380a389a26407d9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-22 08:11:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d9feeac33a8f5800b9687b5fb57c28024b7064be","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-22 08:20:07.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"3cc25d75269c1164ff11a8ae19b365ff95d23690","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-22 08:45:34.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/15 : FAILURE in 24m 52s","accounts_in_message":[],"_revision_number":15},{"id":"44cea47b11155964dc4c51a61b4703995d07f722","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-22 08:53:40.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"e90ce8edac6d9683db6db5b7e04a20d03f87c04f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 11:26:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/abf412fde2884791aa65dd1d44b23d95 : SUCCESS in 1h 04m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8a4b6598b61449cfad3e315059947d80 : SUCCESS in 43m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bad01121cb784ebb83042193c2225713 : SUCCESS in 1h 26m 41s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a466c4d99ae401cab771d752e7c62b9 : SUCCESS in 1h 29m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ab768d7f1be04d8ba442fd3ece7b6793 : SUCCESS in 1h 11m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c61ee8bc6f8448a5aea33abb581be8ab : SUCCESS in 1h 07m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc4083daa1364435b6c32b6ae13d2a2c : SUCCESS in 39m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ac2dd4338324fefb64146b834200e80 : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c3cec200ef4634a522b89386fd88c5 : SUCCESS in 13m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8409ed1b79d04e49b6809aaf6a88d81f : SUCCESS in 27m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d562c83c0544f47b330cfd7947adde4 : SUCCESS in 31m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2de6b45a465446e9f3e3d2112918e13 : SUCCESS in 9m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61d0eb70d3c64dd0942dc34a236eeb81 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d24a12aa19144494a3d6b6a7329d1528 : SUCCESS in 15m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/502f926cd71b4687931bd36c61544f10 : SUCCESS in 45m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9163618bbed2420ebe6512989b866a39 : SUCCESS in 1h 00m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f78d929e176c4087991b566658c1ea5c : SUCCESS in 2h 23m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/656f61d61c5f496c85d55a29d6cabe3c : SUCCESS in 1h 26m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b140145651a04791bddfe575fb139ddf : SUCCESS in 1h 17m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56231d9711f7469d89f5a7dd870c77aa : SUCCESS in 1h 18m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/684d8bb9eb004abca4fecff0e075bb41 : SUCCESS in 1h 15m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ce3ab41643be4e8f8c3946fdc62ac33f : SUCCESS in 1h 15m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ff113339f9db4ceb92ba9aabce842748 : SUCCESS in 1h 47m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d892655c17eb4ca7aa929e2c0c968a20 : SUCCESS in 2h 17m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ace0b3a1c7c043f4bcb42b0c1b0f87b3 : SUCCESS in 1h 15m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a501a127128e4e9c9f750079f19b39cf : SUCCESS in 1h 03m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/18935722f5b74e7ba60c9ec17a31673b : SUCCESS in 2h 12m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/883c09b3d5e14cf38bad7ce1cbae3411 : SUCCESS in 2h 31m 46s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/91222ac2ea3e424bb049c93f8d362897 : FAILURE in 17m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a805e4acb294cb4be41544806063de4 : SUCCESS in 1h 03m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8289b9825e324ac1ab81ce86fc37c59b : SUCCESS in 1h 10m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8027f3996e4e4940b147f872961da50c : SUCCESS in 46m 13s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a11faa89b75f4cb085440ccdc0d64883 : SUCCESS in 1h 05m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9bd9499c0184e9fb05f66822948c668 : SUCCESS in 1h 16m 28s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6d066bd4ec3244f59b1405bf3cb1d35d : FAILURE in 2h 59m 35s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/60207f983906451293c8ea301cdbb6e3 : FAILURE in 2h 06m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"79533140af7ed9bbda87d14e506c6bea9c170159","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-22 20:36:33.000000000","message":"Patch Set 15: Code-Review+1\n\nhttps://zuul.opendev.org/t/openstack/build/61d0eb70d3c64dd0942dc34a236eeb81/log/job-output.txt#488 looks good for requirements\nhttps://zuul.opendev.org/t/openstack/build/6ac2dd4338324fefb64146b834200e80/log/tox/lower-constraints-1.log#415 shows one remaining error for lower-constraints:\nflake8 2.6.2 has requirement pycodestyle\u003c2.1,\u003e\u003d2.0, but you\u0027ll have pycodestyle 2.4.0 which is incompatible.\n\nBut this one is probably OK as we do not have control on flake8 version?","accounts_in_message":[],"_revision_number":15},{"id":"bfcd9eb02fd7218c8b7a6f3ce03fe044dd1a566b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-22 22:39:46.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"463c7c42c75bc05d8e3b336458ad85356c8d3685","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-23 12:43:00.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"0967b183636bf4a45f316e1aaa92039c3674fb78","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 18:46:51.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"fe2003b3063ada8d990097c019e4eec07674f56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 18:47:12.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ec0f0ad17075d8bffdefb76ee18aa5117d428280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 01:40:03.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b1f84fecffb147ee8849fba40399e361 : SUCCESS in 52m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3bc8b3fc7afc4061b37e5a960de905d2 : FAILURE in 55m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d70df69bdacd4c7da78beef8a3462ab7 : SUCCESS in 1h 16m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a9dc05afa8844ff2bd13e56383e11c62 : SUCCESS in 50m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57ec007a2b864d30888fc4c9e06fcad1 : SUCCESS in 23m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497826973be5410f95d468c294679467 : SUCCESS in 11m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db2733ba6d64c11abf9b9b3e8fa2a29 : SUCCESS in 23m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a76cc3477ef43a384ee69412aba6a97 : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f808237579e4529bf45defe299cd2f8 : SUCCESS in 8m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71d6dd16e1c144c5b2c00be74f0f0eb7 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ecfae1f12904b98af220d8bd43734fe : SUCCESS in 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4af05b8358014631976f2a5f6996b224 : SUCCESS in 39m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/34ff6fa45037486aa743aa4c69e7df99 : SUCCESS in 1h 19m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/14767c00375644db84db5c4efd8e10d1 : SUCCESS in 1h 21m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ff3ee9841ff2416e92f7b552189f6c80 : POST_FAILURE in 1h 19m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/542b842011b5466cb0578fa2b9e1b987 : SUCCESS in 1h 19m 08s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/adbe80badf3b4adb9f1ae58cd872f44f : SUCCESS in 1h 20m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ed32672bf85420480a0031a44ff15de : SUCCESS in 1h 13m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6340f007eb3f46fd999e415c536a21b6 : SUCCESS in 1h 12m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/961ad69067c8406cb88cdbae1c13522a : SUCCESS in 2h 14m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5d36db44def428d938e91dce5f52efd : SUCCESS in 1h 03m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a8df4462967c41fab49e72a6da4f47aa : SUCCESS in 2h 24m 59s","accounts_in_message":[],"_revision_number":15},{"id":"65f9105cb763a05006adfbe901000dbeb107a04a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-24 09:32:19.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"3940c5d7a232e5447b86795d10873d0ca4d48b62","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-24 11:16:03.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/15 : SUCCESS in 1h 43m 15s","accounts_in_message":[],"_revision_number":15},{"id":"75cfc7615b0431f7b7fa979a2b4bc358af39bf51","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-24 11:41:26.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"3d18ee8fa2b7ceff0dc8d1019b4a00337d083fa2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-24 12:39:54.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"f4f147855c8457cc58514ee592fc4dc8bad56135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 12:56:05.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3ddd25c03c2f4845b4ad1477b8bb0060 : SUCCESS in 46m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7894e6dc83a54b9f88d65ccc9f675dec : SUCCESS in 56m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c92283c0c3c04602bd3b0aa126bb4341 : FAILURE in 1h 23m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ada01a034d5c4c15a2029adf208111e2 : SUCCESS in 59m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3b33a4c2f8fd4bc5961f0499be9197a0 : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a425a275c29430dafdb83c347e66d72 : SUCCESS in 52m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b39896e225e645bc95204e5d3cca44b8 : SUCCESS in 33m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a3c7dc3c0204ecc86ebfc6a7b689ca3 : SUCCESS in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da8a713de0e4058b185f25ed821dc5d : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc9f75f9cb77444db4cac6b64a03f94b : SUCCESS in 36m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58333ec58e9146d18df6ee5819e78e24 : SUCCESS in 29m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d23dc11de45460fb4be5447538dc709 : SUCCESS in 10m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a6da38363804e8c94bb6d185108d696 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ea6eed34a804ce29f129c69009a8c2d : SUCCESS in 14m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8d86f7dfccf24f35a44ab6d3d3958a83 : FAILURE in 45m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/94788e79528149da807c8e5ecdb12f91 : SUCCESS in 1h 02m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6e8206b9cca14cd1880aed9b96dcbd57 : SUCCESS in 1h 44m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3366a540d1da4b19a2392d16c1e8e1cc : SUCCESS in 1h 05m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/16328f2322bb4b05b264085562dbdd59 : SUCCESS in 1h 23m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9998b1eb1b7c4df394d2c776c83c6fc3 : SUCCESS in 1h 08m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e134272cd9004fc2bdfcd4b1f74157f7 : SUCCESS in 1h 07m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df48c987649d4482a92b85683548adff : SUCCESS in 1h 10m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f67a5545902441aa8da49800717bbea9 : SUCCESS in 1h 38m 48s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4962c970d91543b987f627ec8e4b36da : SUCCESS in 2h 03m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c002eab3295e49578b9243d91fa676b2 : SUCCESS in 1h 18m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8ed58f2d148a433b92ad217192396242 : SUCCESS in 53m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ceff61b5536e486aa3ee23db412d5050 : SUCCESS in 2h 17m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/86e6a17707464fc58a3053f338a7a745 : SUCCESS in 1h 41m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35a86711f9364d8b98b516552bc4ccf1 : SUCCESS in 1h 04m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bc4b77f794f94b5d89278896c004cc37 : SUCCESS in 1h 18m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dbe8d8920309403d9cab73fa53b186fd : TIMED_OUT in 2h 13m 55s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc2d74c4cb754668bb0744588cf534be : SUCCESS in 1h 06m 51s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/df6a9daa2287409e8ee620093285efff : SUCCESS in 1h 09m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4f5eb8015ca84a62a3ae72e023d7de35 : FAILURE in 2h 43m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6302374db10e4be095d9b2792b7bd32e : FAILURE in 2h 01m 39s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"eee598af939e0556afa3bf06b6588266521bbafe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-24 13:31:54.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/15 : SUCCESS in 1h 49m 59s","accounts_in_message":[],"_revision_number":15},{"id":"fd5be4905892071eb29f16397d798165ea12294b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-25 07:33:42.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"3f03ae123585004ffba8242fd510390a58901d95","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-25 07:34:13.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"8a3009c8bbdba78042bbde046be70e0c2595487f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-25 09:19:16.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/15 : SUCCESS in 1h 45m 06s","accounts_in_message":[],"_revision_number":15},{"id":"928953c0ba1aba253cc30b784f2ed0b71c7478f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 10:42:53.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c2b76dc3f9414ae79c3d467c4e0eae25 : SUCCESS in 45m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/804ec109e14b4c33852f479a514ac0f2 : SUCCESS in 37m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9efc6d989a4f48aba041dc1169c60d7f : SUCCESS in 1h 26m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/56014d185cf748369ea380fa9a9b2d10 : SUCCESS in 1h 08m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2eeb3f7078754c40a8704760729cd4e8 : SUCCESS in 59m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5440f141ecb4b92b59d193395f33b89 : SUCCESS in 1h 25m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e4638b01a9c49f9a30c78356772e4ef : SUCCESS in 37m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a62667eff0d4b9da7b5ec012da7b427 : SUCCESS in 26m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac974a23e4424b158e3cfe8d6856bbc1 : SUCCESS in 15m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19292c7eb1c04517be823cf806e6ab82 : SUCCESS in 33m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10962a850c6b40bfbcce2755052ccec1 : SUCCESS in 33m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860898e64c73473f9150b96de92033f1 : SUCCESS in 29m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df0eb6835ba44a18d12480037135549 : SUCCESS in 10m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7348c6568207426c81f0c5a84ed74337 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce0b3572a474265908d1375c6d07327 : SUCCESS in 14m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/12c96cac2b1c47c7947999b321cd59e4 : SUCCESS in 39m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e877f1b10e994c4295d5baa413ffd2bf : SUCCESS in 1h 15m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6533a68150eb4f9c9c095b03600b2738 : SUCCESS in 2h 02m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a4b19b3a5cae46e68d199550ea89d484 : SUCCESS in 1h 06m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3e971a0fd63f476a89f12586753da175 : SUCCESS in 1h 15m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/404e86b2fa044e9eaf9501114e1a64ba : FAILURE in 1h 20m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89c966eed2114ff99696a0336fb5786d : SUCCESS in 1h 06m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10226fec1d8446dd9eeeacecb36d99dd : SUCCESS in 1h 32m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/780ecf99514e4985bfb1e8df7a0fc3f1 : SUCCESS in 2h 04m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dfda2b5a1e884e7c919dea2972071564 : SUCCESS in 2h 05m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b5d94164fd9453eb7af50c391bb4c46 : SUCCESS in 1h 33m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4fdf5fef4d524a8fb8bfe9b05403e370 : SUCCESS in 1h 04m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28c21195675247348a52b02bb9ff6706 : SUCCESS in 2h 40m 14s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/03ccbd3309314e1783c5d3e73bdd2f22 : SUCCESS in 1h 43m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb07cbc7029a4b09bd001c26fadd50eb : SUCCESS in 1h 12m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/748cbc4ab887476b888ecc64723fac2f : SUCCESS in 1h 35m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95f7418a2b7943e993437dd8e098f658 : SUCCESS in 51m 25s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88d89263fdc84e4da87c57d7db1c4eb8 : SUCCESS in 1h 25m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffa98ba587364ee5b544f4e89176e041 : SUCCESS in 1h 35m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/04893980375a403bbfada21eaa83246a : FAILURE in 2h 58m 30s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2fcb9c5cec4f4dc08261acae91926658 : FAILURE in 2h 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"01a1c56ad9a1c3daf2825dba1ac868529654f7bb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-25 13:15:58.000000000","message":"Patch Set 15: Code-Review+2\n\n\u003e flake8 2.6.2 has requirement pycodestyle\u003c2.1,\u003e\u003d2.0, but you\u0027ll have pycodestyle 2.4.0 which is incompatible.\n\u003e But this one is probably OK as we do not have control on flake8 version?\n\nThis happens as we capped hacking versions. Hopefully we can switch to hacking 2.0 but there is some limitation we need to address first. This is only related to coding style checks, so I think it can be acceptable as long as it works well.","accounts_in_message":[],"_revision_number":15},{"id":"9ac50735483be3a6d68c770837665e9c37c0793e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-25 13:16:04.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"1b14c59d267f8b23a646a2f45e91a420e4cafdd7","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-25 15:02:21.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/15 : SUCCESS in 1h 45m 49s","accounts_in_message":[],"_revision_number":15},{"id":"4a547cfacb55a8dbab79821e024cb7f87200638a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 16:22:08.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/93c2f1503ca145719268a0e226c3768a : SUCCESS in 46m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f667f8a3dc04913945e758106206c15 : SUCCESS in 32m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ee52ec151e334bf19caa65c40cb7257a : FAILURE in 1h 30m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1e74ad14e41149b68c37381153a355ff : SUCCESS in 55m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/946d4a4f15bd426d872537a3853e1764 : SUCCESS in 1h 01m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce9f6e3a89f74496ab1ddfe8f349c442 : SUCCESS in 1h 03m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a3069a6a45407a867ecc8b13b0874c : SUCCESS in 34m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3dfabefa9bb4c7fb4a4bca51ad8407c : SUCCESS in 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e71fcaa2e094bd28607dd741fd4bae2 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53a6a2a5ba0f4ea781d4210d875bed63 : SUCCESS in 22m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95727d69da3845a9957edbf081a0b69b : SUCCESS in 39m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b1a07e15fa544fb80ef9e6465accf95 : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/607dd6c64c7041a38c5739820f9e76af : SUCCESS in 8m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5b835f527fe4df19d0a4a08e0a5633e : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92e7a172e5264cb9af30bc92b446a552 : SUCCESS in 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffe16ebf1f5349fd9d1aa2acc6bda0f6 : FAILURE in 54m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5bd9fc50945b4a67bd91a0703fe0cb58 : SUCCESS in 1h 08m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f66d5418ab244801bc168e6d21e539a1 : SUCCESS in 1h 57m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fad0d524b368414f9043afc9c76ba65d : SUCCESS in 1h 19m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7d544f38682549d4b8826f3a87068cd7 : SUCCESS in 1h 37m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40bba1e015b44a049c5e9800d3cf8068 : SUCCESS in 1h 34m 55s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6905f8c208d5412dbb8fe4845eb0f64e : SUCCESS in 1h 38m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d22dc32a5c3a4a99972386e0ed97a860 : SUCCESS in 1h 50m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/89e1343bcf2c47f380c2a8e56ca9802b : SUCCESS in 1h 47m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c039a5cd21648f98b1aff7a5accd613 : SUCCESS in 2h 19m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a7743ae947d4e5f9d3304614eb6c3c6 : SUCCESS in 1h 33m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6a0eb784de3e4afb80e8e1583e05f30f : SUCCESS in 53m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17d1e5fa59e84c0eabaca0ee2a0f215c : SUCCESS in 2h 13m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b59e1f9c0bed40e59e87183e0ec0cf51 : SUCCESS in 2h 06m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d02ce8be00f04db9b54cbc2522e132a9 : SUCCESS in 1h 23m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8763bb424b154c099a36e09590138750 : SUCCESS in 1h 08m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe7ba9a1826b4ec1bd96294e58965c50 : SUCCESS in 41m 31s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1bab1c347ee476f83bfaeed64acd9a5 : FAILURE in 1h 02m 49s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec834669e7b64c82b41d6eaa91962a86 : SUCCESS in 1h 07m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e22301f31b04441fb74b81284c5e57d1 : TIMED_OUT in 3h 04m 25s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1ec469906f4945bab7858abbeb29a5aa : FAILURE in 2h 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0f57b183320e292ff8568404f4a707cfb0bbc1c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-26 07:07:26.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"f319967cd1fd165866fabfbb0f8ac44c5acc1623","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-26 08:57:54.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/15 : SUCCESS in 1h 50m 00s","accounts_in_message":[],"_revision_number":15},{"id":"a25a60c093960d5d90992875d268c51cb15421f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 10:01:40.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9feef526489f4665800fb554e9322c39 : SUCCESS in 1h 29m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e14fdee248f49ccbb859f66d3c7d1ab : SUCCESS in 58m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/1b16eaed482045f992266a352c13af5a : FAILURE in 2h 26m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fd36eaa53a2e44c4b48becf9f417a193 : SUCCESS in 1h 54m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/377f75a05aa34a198fb989f0de64a641 : SUCCESS in 1h 54m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38a9a05661c8495e87c44c380fb34b63 : SUCCESS in 1h 50m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6621392c89914de3b515507deb2e5a75 : TIMED_OUT in 51m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96bcc912a605475e91e1786c3739baca : SUCCESS in 48m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde070ec43434a8ca29935e22c73d1a8 : SUCCESS in 22m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ee0388460a4231a33e8dd996ce1783 : SUCCESS in 52m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/861cdda76ea246eba2ed2d1c28726194 : SUCCESS in 33m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea52d80d9a324c2392ac753392bf7417 : TIMED_OUT in 41m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb897398d7f4223926cd7f473ece428 : SUCCESS in 25m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6d362386c65405a8a4250669441a483 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3edb3e15e7f3409da552e1526d37b5c2 : SUCCESS in 15m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07c720cfb33c4c3f9f609e7311b275c3 : FAILURE in 1h 20m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/391a4de782724982be69c827160966a9 : SUCCESS in 1h 24m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e72d9e91fc904020accd2a622d61d849 : SUCCESS in 2h 16m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ab7ac9b567af4356a555e9b29cf6bf99 : SUCCESS in 1h 30m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cf3e187cbbc547c899690ffde6e02f42 : SUCCESS in 1h 28m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e6bc1fe65c640e6a3891a9912631623 : SUCCESS in 1h 30m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4f5a70a25e18485c938d1fd8ec9a0fc5 : SUCCESS in 1h 30m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d5c430e647440af9f0cd54a725764c4 : SUCCESS in 1h 46m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c213d84574044166a4a1389c342bed7b : SUCCESS in 2h 18m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5ec67c5373da4cdd9796f8e07e6c990d : SUCCESS in 2h 06m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/200fe398d8b94f6587512d9b2b2f51f1 : SUCCESS in 1h 28m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dd0b4c5d11dd4a8db677f5f7fb7b126f : SUCCESS in 1h 27m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93bfa1323da54513bdeda692d6b11941 : SUCCESS in 2h 44m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c723f6fab99a4caeabb3bd192bfae5b5 : SUCCESS in 2h 09m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6edab354395b4f949caaeb1a2a691bae : SUCCESS in 1h 10m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b5b4c1dd4f6e469cbed82e7b4083615d : SUCCESS in 1h 20m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b27432ac338b46b98bb630333fb96150 : SUCCESS in 44m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57bba38f11954a87992289d7c070fb0c : FAILURE in 1h 25m 58s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/283460ab2b2a40cc93c1260730d5b48e : SUCCESS in 1h 40m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5bb5679a8bdd4988ae8bb1760d2104e0 : FAILURE in 2h 41m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/31998d9226a04daca1512e6327e82685 : FAILURE in 1h 57m 57s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1a66583f8ed34af16c38abf975460a98ed4df976","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-28 10:07:53.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"3cf70981edb8dc90c4805e0a6113b91ce9715e0c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-28 11:09:13.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"30a55a09d63e54bacc2333b2c2949ad3dd309dad","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-28 12:10:52.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 2h 02m 28s","accounts_in_message":[],"_revision_number":16},{"id":"ecd2713cccc6fa05ed7df52aadfbb406cb1d7cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 12:56:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5b1db27c0fee4df58fad650790c4f3ad : SUCCESS in 52m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce20c830d41346c48955bdc72bc5ab99 : SUCCESS in 37m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/23e3764ff0534c4688ccf65b416f8c9d : FAILURE in 1h 59m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/37f924cf4313421388b91004b819eeb2 : SUCCESS in 1h 27m 52s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f7b584d63aaa48d28314ca4a8d717030 : SUCCESS in 1h 30m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e78ae517456142d8889735a38a8a76be : SUCCESS in 1h 18m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf9c42ccbef346e5ac881b98afb3a195 : SUCCESS in 35m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d9b87f51fe41cfb60e61840e215c5b : SUCCESS in 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738dd2a6a3e546e29e525f0bf9e0a53c : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41cc1cdc5997443eab19e0887891b9c0 : SUCCESS in 28m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a35a2792e6714d2492b25f9e8edd6252 : SUCCESS in 30m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a3e9c3ce79e4c41835e2829eac4b9ac : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7683f8f3d0d403e8431f788783904d9 : SUCCESS in 11m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9cd06c399cbd48159e4d298dbc803121 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db7f2e4e4d0b43c1928e6fb76b38c32c : SUCCESS in 15m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5019402f3d6d406e98ba4ee1e174e5da : SUCCESS in 47m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6b4dae94ce8042029c332bad48a85153 : SUCCESS in 1h 10m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9d171a3ead884c76a84cc66ba37ff4f8 : SUCCESS in 1h 44m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8020ca5dd8744d8595bae884826a8cc6 : SUCCESS in 1h 28m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ef137c3420fb4420b151ea835c265816 : SUCCESS in 1h 07m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09b83ca23cad4a59a3814a7f9bba33b2 : SUCCESS in 1h 19m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e091decae67a4d7f972149bb8bbd02a8 : SUCCESS in 1h 13m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/92a30bf1c6d44a61990e20f27eff73b5 : SUCCESS in 1h 19m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a96a044ed846489e9e2ab566eda6feff : SUCCESS in 2h 04m 20s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/20c2bbc0e5f044abb17316f2bd4767b7 : SUCCESS in 1h 49m 47s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2492d576f4bb4bf5968d82dba8c5a53d : SUCCESS in 1h 10m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/141369a6e10748d3800e2bee3bf51841 : SUCCESS in 54m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/241f102f952446b88415dc1507429ff5 : SUCCESS in 2h 13m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cb42888bfede45c4b6d8a99fa3566ccf : SUCCESS in 1h 46m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/491e044bbde54e078581ba94a18a3ccf : FAILURE in 1h 35m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4f8f36d9e9ae492d89eab8fdac80f964 : SUCCESS in 1h 23m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b837f3698a943dabc8a1d8d10ef38a4 : SUCCESS in 41m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/49c8abfa85a244548a4c141df14abc74 : SUCCESS in 1h 19m 07s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/903d52de0c00493799c130350aab4751 : SUCCESS in 1h 15m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/358931206bf046269a9a6b39595d9c3c : FAILURE in 2h 40m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/852a0e4ddafd4910b3256574ae2d039e : FAILURE in 2h 05m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cda88b78860546fc99251d45ec5629d849f661c9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-28 13:46:07.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"6de5616aa54dafee56826c4c6cb52908db511866","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-28 15:34:26.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : FAILURE in 1h 47m 49s","accounts_in_message":[],"_revision_number":16},{"id":"46ee4fca3c18b838286c4ca4b7631af6938fc2b1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-28 16:12:04.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"6c96dbe8fc169e9705d8766704f72143c0d53cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 16:35:51.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b5425216421c495584d8c644ebdf337a : SUCCESS in 1h 19m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f30e3da38324d03a3f59850e8b78aa2 : SUCCESS in 34m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b74a16ea18f342f6953c5da60be6825a : FAILURE in 1h 36m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a946134f00c942d1bf7d125a03b48dd7 : SUCCESS in 1h 04m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/bc0817e1957f4dd2ba68c2ca6d1aea99 : SUCCESS in 56m 55s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/826401b01dd2460bab4a4c109650d94b : SUCCESS in 59m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6f6223d8b164780bd7169fc876ecaf2 : SUCCESS in 30m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/511dcd4127784d459321704d57586719 : SUCCESS in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c8916738c8497fb3725d57c8eb217c : SUCCESS in 12m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd7db3cf676848c089570d1369069c17 : SUCCESS in 29m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9af70f4a3a7a4fdeafc40566e86caea0 : SUCCESS in 31m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc90dca0dfc74a238ee2a163ff8c1bdb : SUCCESS in 22m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63e2ae72de145b0ac4654a0354d5ad5 : SUCCESS in 8m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/29174aa6e8924af9a74e138c23340e3e : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7359343161c94beb9bd7eda981558779 : SUCCESS in 16m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/af296306e2ee4900862c475ea1178f4e : SUCCESS in 42m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/94992c46b606413b904ea65b0c8675be : SUCCESS in 1h 03m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b0c65d05c2d54c83b76b93581cef1b87 : SUCCESS in 1h 53m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/38ce85b6363a49f59508da5f15be7936 : SUCCESS in 1h 15m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2b1e4b4356e3437ab9e9491e5f1235fc : SUCCESS in 1h 12m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19b1dbc453a44460bf5be9d81c983aa0 : SUCCESS in 1h 07m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a9fc726c7dd49c1b5ff79466ec53353 : SUCCESS in 1h 05m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8e6d839590db4906a342fc99e54dae00 : SUCCESS in 1h 04m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ca4713ecaede46c09ca95af437b19360 : SUCCESS in 1h 33m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa071ee0100d4929b060681e827df5e0 : SUCCESS in 2h 30m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6e8f03fc0edf4f0bb81f938446f98f44 : SUCCESS in 1h 08m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/90a6a16dfb91482ca92372a9ae000534 : SUCCESS in 1h 03m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9446e66c7c54a6c9a184aff66d86ffa : SUCCESS in 2h 37m 49s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a25bb4661f644e409dcd5561f871f483 : SUCCESS in 1h 38m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1778964ae364420a95aa9f0ef86c1189 : SUCCESS in 1h 06m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3499c96e05c546679d634a40531c74ec : SUCCESS in 1h 27m 57s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ca02c78b5d04e2188e6177a28696a24 : SUCCESS in 39m 49s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/94c160e48ad945e2af411c52d569936e : SUCCESS in 1h 13m 58s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/bad86ed55e3d4d52bc64d854c793d3b0 : SUCCESS in 1h 17m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6cfaa8ec2ec049b9b8bd735a71ba3d6f : FAILURE in 2h 47m 18s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/51fa54c74d854cdda3c8e170361326b0 : FAILURE in 2h 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"945494c1473f2119f60874049823f2cb715ee1d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 16:36:20.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e310f2f156b0c7e993734626be5e7604da0a65f0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-28 18:26:52.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":16},{"id":"d4fef92732fbd58334650be7d866115734168b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:34:46.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f6e598898ac3450c99a58524d5ee797e : SUCCESS in 51m 27s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/908eca35520349a9a3131c3dabe70d10 : SUCCESS in 1h 13m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/84ed15fb9eac41dfaf446713b8701df8 : SUCCESS in 1h 44m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f784beff7b61472a8a9630701755f447 : SUCCESS in 1h 07m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce459342e43d4e869872eab03cdc18f3 : SUCCESS in 33m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a0d278fe0924ac8bf4c999b3109b4a7 : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1deb288858044970b89a972525fa8295 : SUCCESS in 24m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/059f68dfdfe94f84ba31f43be16d2ddd : SUCCESS in 29m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dca85d8cf704a97882bb2b8664885f1 : SUCCESS in 8m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b162f3bcd8df4beabb354e15089f705e : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d7f9ea05eca4b17bba88cfc799202f1 : SUCCESS in 14m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3cf80f0551fd41d8b342775d14afd753 : SUCCESS in 42m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c13b3aa4d18b4e4c924abb2ea5391b10 : SUCCESS in 1h 38m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b61de584e974180878df853bec0957a : SUCCESS in 1h 40m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/09c1be7cd17b45f4a86c8c8c246b6ab5 : SUCCESS in 1h 23m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a2dfe0b2dc3f4d90b7428931f7d83e0f : SUCCESS in 1h 37m 26s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac0866306d9a40758e5413ebaf562c94 : SUCCESS in 1h 23m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79f8257d23bc4fa18338e20ef9eeb1f8 : FAILURE in 1h 14m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/128ae6601c3c49d6948248a280bd80e3 : SUCCESS in 1h 05m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ee2ab2e96bd4f8783df63060307b1fa : SUCCESS in 2h 15m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53480dc0591f49c1a8a29e4f667309fd : SUCCESS in 1h 24m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b7aadc21bf42418293c03f9e5cfde8c3 : SUCCESS in 2h 16m 20s","accounts_in_message":[],"_revision_number":16},{"id":"d9ad82d79afd458465e7122c4939d142090dbd54","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-29 08:23:12.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"23ad2f773a935fe19ec1c9c67f393deb7e0b6ccd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-29 10:14:42.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 50m 52s","accounts_in_message":[],"_revision_number":16},{"id":"e877ff71878e835241511f7dc5de710d9b02b265","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-29 11:18:48.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"79be45631086317c446c3744e9732ec7b3d80153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:35:46.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c0bbd054bbb9418fb35d4ffcac476097 : SUCCESS in 48m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4fa23ce157db4faea5ad6238f2dd85a9 : SUCCESS in 44m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0854f2fb3c7e46c7913c6ce75f691a3a : FAILURE in 1h 45m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4965d4e80e0141658e6077a9b8491f4e : SUCCESS in 1h 21m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/74a8d88cae8847c2bf8afa6afff061c9 : SUCCESS in 1h 05m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0943df7521c4c38b13def5ac1327403 : SUCCESS in 56m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8ce6c676a914abeb570913fd31ef96a : SUCCESS in 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c5088623af44eb59c4aa3b8bb34668b : SUCCESS in 32m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3031d9a080404530b3bf8620f37692ec : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b614280a9fe4b618c59477d101a71bb : SUCCESS in 25m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6918e0aced5145e0b88510585ea0afd7 : SUCCESS in 35m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/375a094d54474791a98911e82ea13ee1 : SUCCESS in 25m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bb31ffe9e24cd1aafcd3b2ffe32314 : SUCCESS in 10m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/574bdc07724442a88b7b8c9873180805 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3ff04eb93734b649c08642f21da729e : SUCCESS in 15m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4fe1d730858e4e41a289edc8e2f49957 : FAILURE in 1h 06m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a61ed8c5fd324d1ea7818cf8c557612c : SUCCESS in 1h 14m 31s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b71af8b3eb5f44d7b79cddd538c881c2 : SUCCESS in 1h 45m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/aa9212bf801c47a6bc4548707920baa6 : SUCCESS in 1h 14m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fed5731e34ee4764a63f2c742336a918 : SUCCESS in 1h 16m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/96ebe27dd8eb4f6e85d1d381deaeddb7 : SUCCESS in 1h 07m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41725b21711145628230257604258dd2 : SUCCESS in 1h 04m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/684aabc597c14ebd9ce240d4c458cf36 : SUCCESS in 1h 06m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a105531af8d84e0b8f288d63d4af8cfc : SUCCESS in 1h 34m 04s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e3d10b63198a463399661e790b97e223 : SUCCESS in 2h 08m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d6e8cd1c7c5c446bb1f8d915834bedee : SUCCESS in 1h 10m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8341d25942d0412599bdd779c6550191 : SUCCESS in 54m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cd009d15ece45fcb58cd96c96a25350 : SUCCESS in 2h 18m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/007ca2496c044171b32950387fd1f030 : SUCCESS in 1h 49m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ce4efdbd7e9402a8cf82c569b0919f2 : SUCCESS in 1h 04m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/037a7bdbfc9f49858cdbc0d04ebc973a : SUCCESS in 1h 36m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/570e0b7a9742446f9590f46202740231 : SUCCESS in 57m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd052c7fd58e454bb82dea1b850d5f84 : SUCCESS in 1h 24m 23s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a018f8caf354f76906454580a5955da : SUCCESS in 1h 37m 48s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5868d5340ff3452e87ba1aaab4f54a09 : TIMED_OUT in 3h 08m 24s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/fd033a91a337491bb6db3df4d999094a : FAILURE in 2h 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9c711455bbb1e9fcb30e64e3e3737f90884c55e5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-29 11:49:00.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"5d4ad8ae328cd533665d55de865820766d224495","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-29 13:10:59.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 51m 42s","accounts_in_message":[],"_revision_number":16},{"id":"e4ac3ef465bd8a6dfe7a5ba6dfd4ef0e97d4bc86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 14:25:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/27a6a9ba8cee4beeb4918138ab4ee61a : SUCCESS in 51m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/952563c9f2084bcc863134c4938d1488 : SUCCESS in 35m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/eff6931288ff418b95a0c6712a191dbd : FAILURE in 1h 33m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6da96b247d3f47fe9944e41a7f9a8536 : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64173859bb37401bbd7927bb4118f9b1 : SUCCESS in 56m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac33230f64bc458eb94794e6dfbfd39c : SUCCESS in 57m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e27f4fd3f7e43f8b3339c2cf41dbf39 : SUCCESS in 31m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc31d5ae102e4922be109bb9a5bf68ce : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6bdd996ace4487a90e3749eba23242 : SUCCESS in 15m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb6893179e3c4276ab049fa65dbb8c14 : SUCCESS in 31m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25833b16997b4384a524282fc1222280 : SUCCESS in 25m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa11935819054a3481ddaea99cb6e136 : SUCCESS in 29m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92058d44258a408993f016262b981d5a : SUCCESS in 8m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dca926b392f04d3c900b574c8e8e177d : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86801f56bb749fbb0b3687844af03b7 : SUCCESS in 16m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/54c9c3f706dd44259db59fc51dd9f744 : SUCCESS in 38m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/894c0f81ef28467f94f344cbbdc608e0 : SUCCESS in 1h 24m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2c7ccd90e4a04ae9a8a8d7c62625028b : SUCCESS in 2h 04m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0ce92ac76507455fb11e052de22404bd : SUCCESS in 1h 11m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/84de2145472f47f286405b5a2c227303 : SUCCESS in 1h 08m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3cb9504ccc04de49e5bf036cddc9a9d : SUCCESS in 1h 03m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f17de39ead644038873aaa0d0d910ee : SUCCESS in 1h 08m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d5d68e233ee4193b13e3462fdc465a4 : SUCCESS in 1h 30m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/18eb1611a3ff4c3d82729907f8d9a92e : SUCCESS in 2h 11m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5994612badb94d4481c763a5feff395d : SUCCESS in 1h 46m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5b972cc5348b464498afe291aa0e2b7f : SUCCESS in 1h 08m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4e37f8bc0264e7c8ab891ad0c399773 : SUCCESS in 52m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/918c0b073e5b49d4b0cb7c5d26ce94ec : SUCCESS in 2h 23m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f42011a003224b23b81aeac3c7c55dc4 : SUCCESS in 1h 31m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67b8e1e1b9a943019f1d3624250e51a2 : SUCCESS in 58m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fdfadf6866f34a01a904b904d80bc4ad : SUCCESS in 1h 10m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3d25b1cfeb9433b8b6d92ed94245949 : SUCCESS in 38m 10s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a02123070d414a11beeee8627ac75771 : TIMED_OUT in 2h 15m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6068797e1cdd489586e26af9f5f8600d : SUCCESS in 1h 10m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4715d879d22e4eec8fe23cce67ea64f9 : FAILURE in 2h 27m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3bd9e733330f41e08b93d87b3952d053 : FAILURE in 1h 59m 04s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"739b99c7be4958ca75418caa3208c0ea0db96d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 14:25:39.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"cf3cfc1dc3e55c0c5fa2cc9961266cdf99af21ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:02:58.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/729489519301488e97aff914986b25c4 : SUCCESS in 59m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4d6ba4b4a5f04d008a02225fbdeb829b : SUCCESS in 1h 52m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2ec64ddd3f3842238f6e85177201bb24 : SUCCESS in 1h 15m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2e5bbf28443b48f3bdb7d59b989f76a1 : SUCCESS in 1h 53m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b83b7c024644837b43e466542a557dc : SUCCESS in 31m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5089a23c20d94d7ba0990e7094d4e3ae : SUCCESS in 16m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a8cea0b96934cf5a5c513155f108d44 : SUCCESS in 28m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/effc3b516d434ee9826b7a9d53bab775 : SUCCESS in 28m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a76944de1ec4e849adedc87b84a9210 : SUCCESS in 8m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/598c635abef742959229f8c822ff079e : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/829f8cc81f614b63932f0ed4953c296a : SUCCESS in 15m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc04b310d1e84332acfe21d466277954 : FAILURE in 54m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/df235b925afb49c18ee6ecbe65f07f8d : SUCCESS in 1h 08m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9cd26c719ff34b4ba8c4a2aad2aa52cf : SUCCESS in 1h 41m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ac764d638d334b74926d242f62ee135d : SUCCESS in 1h 56m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1b9df20b5d774016864eaa7aea02e2df : SUCCESS in 1h 35m 13s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d727099c0db64670a134ce4c1cfc303f : SUCCESS in 1h 43m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a258770ee894ce4b6fb59603f457190 : SUCCESS in 1h 48m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3edbb07db9e44972be03e3b5bec2ed8e : SUCCESS in 1h 51m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0b0cf9b99cb4748b85f593816c4647e : SUCCESS in 2h 31m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afb6a8c551dd43b398f4d7daf21b2163 : SUCCESS in 1h 02m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ad4910e96e14462e80b16e95bc6e3063 : SUCCESS in 1h 36m 40s","accounts_in_message":[],"_revision_number":16},{"id":"82c2a119a5c32a7093899e0cb9c9d7c4bbd75878","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-30 07:29:48.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"9211915e0c9519aae616aa1df8cead786926ecfb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-30 09:23:47.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 53m 31s","accounts_in_message":[],"_revision_number":16},{"id":"70a60267ac97ed3c2c7110b5e060c7fd46d41505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:35:18.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8528fdcf742b474e815b5708c67fe075 : SUCCESS in 48m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0cb337fce94a43f2b6bc39ee6200ef91 : SUCCESS in 32m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/511d97b63423440281fba8b0a2b23cd3 : FAILURE in 1h 26m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/6fb2a5a11a9244729667efcd7e23e270 : SUCCESS in 1h 17m 01s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4d800ecc1234425ba66406f0bba6c6f5 : SUCCESS in 57m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9c2d48ad265b4765921870182c410b3a : SUCCESS in 1h 00m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73d539822914483b8c2959d2d9b46079 : SUCCESS in 35m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be1c71c1c3ad4271bee091701df82672 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23f4bc914b1415c83571eba9113ea08 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0efdd79a23f486092645b6b0563455a : SUCCESS in 30m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f86703565cd74019b28837e358218fe5 : SUCCESS in 30m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af315830f8fc4aba9f9629d977308ed9 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63dc3e7fcd0d40189c9db2aef592ee1c : SUCCESS in 8m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a822b7e90e964512ad8885687ecb2813 : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edc2f5e91d5a478882d1692d959395b5 : SUCCESS in 15m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/545b81055ca5480ab4660cac2a624146 : TIMED_OUT in 2h 13m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cd5d72d029f749cfbd555d4136fd3734 : SUCCESS in 1h 07m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a909ecf0718641bbbe28c8f710502562 : SUCCESS in 1h 51m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0c3572cb9e72459f873979326f4ea6cd : SUCCESS in 1h 07m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c4052637a95e4f3ea5653ebcd292ec74 : SUCCESS in 1h 22m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec654c2f6e8749ab822ed86069b79fbb : SUCCESS in 1h 04m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28aecc2caa494175a1ff77fccb4c5840 : SUCCESS in 1h 01m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/267cd74ddc1c45b6b62ea207055ea688 : SUCCESS in 1h 20m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/421e1a3dda9143e5be8b927c6a65116c : SUCCESS in 1h 33m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4bb63b0e672472ab6d400b671b8f4cc : SUCCESS in 1h 50m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fc2fd5987c1f44b8a9ca523815a30f1f : SUCCESS in 1h 12m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3718491d61b74359aaf2b550bb7dad7d : SUCCESS in 52m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f7e30941d98a48eabc6c6a95b4aa67f3 : SUCCESS in 2h 24m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9b896d1bcc25406a8991e22e9aca71bf : SUCCESS in 1h 32m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f4020ebb4bd4bd2a6507d3e96f1862e : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a1b85e55b4684f838923132e52bba371 : SUCCESS in 1h 13m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/86064f9a6dea4b5598651b80d1e27394 : SUCCESS in 42m 57s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2763b84d188f479d92b88b83e36639b5 : SUCCESS in 1h 04m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a02372c12f924907a1bab0a23b173c7a : SUCCESS in 1h 16m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9098ce64b20f41fea5d482584fe02faf : FAILURE in 3h 02m 27s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f2566d4f28c2459f9b98770f6069e759 : FAILURE in 1h 51m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0b3fc92d198d0fcde4b46c171a11287b0a5327ec","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-30 14:50:22.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"1fe79dfe621fe710a6239a6a128fc47998ba7823","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-30 16:48:32.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 57m 38s","accounts_in_message":[],"_revision_number":16},{"id":"0231fed91f97e2ef897a29eed9db7421795cb273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 18:19:14.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6c8fab466e4b49d18cd55e3e192adf33 : SUCCESS in 47m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c47c411b4ecb463ba980f18c304030fb : SUCCESS in 38m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/38b602b5240f4bb19f7d11e05220eb5f : FAILURE in 1h 49m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dabd7de7db8d4ae98f65edfb03bbab3e : SUCCESS in 1h 03m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7a24a6537d824420a1297ef90f616a23 : SUCCESS in 1h 07m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cd56c59b655440b685b0fa23eba1defe : SUCCESS in 1h 08m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4992537b605749c9bda3b6a2274b4b26 : SUCCESS in 47m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/663f06e781f543f6ad35b5c10866e55b : SUCCESS in 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c7a51e18934f429fb4c35cc5d9e6b8 : SUCCESS in 16m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/387f67879c7748a1b1956e2a9789c084 : SUCCESS in 29m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64fc4746d49544758cf972714ae66392 : SUCCESS in 36m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abf37877205f45aa884c4546868f99d2 : SUCCESS in 32m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b409a1c001b34cd89bab5cdd74ef438c : SUCCESS in 10m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/16f422966a8c49b58c68bed90e4db47b : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52956e9354c14f0990cb1b5de7611738 : SUCCESS in 13m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6d7113cb48b542f0969021ce88c360c3 : SUCCESS in 41m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a8c2327ada81436fab07e0c1931f86a0 : SUCCESS in 1h 28m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/62d3c374f528467a8cd0ce881ff92a40 : SUCCESS in 2h 12m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4e4bc2b587694f379bbd3fc7ad6cb3eb : SUCCESS in 1h 15m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f448ef0f54d34b5c837ec98ac3e7837e : SUCCESS in 1h 26m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/329b458dcba5440b96df267f5013c5ae : SUCCESS in 1h 13m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9ca2fe179a3429784d6f4d0d753a5b3 : SUCCESS in 1h 08m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e24cffd1ba2449a9154738ec684eb0e : SUCCESS in 1h 14m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b51c86c28ffe45b8b2170d44a5b5ad87 : SUCCESS in 1h 53m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e5faef47c9a46ca9b49f506f0550e98 : SUCCESS in 2h 05m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9f1c04e4275e42b88ae19c50dc776c40 : FAILURE in 1h 14m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed873e53d1784a9993a6efaac290eeb0 : FAILURE in 1h 05m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b5b78509acc04605a6d17a282ba0d39b : SUCCESS in 2h 32m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9ebcda10ec43451b8c2db714fdc7a7af : SUCCESS in 1h 55m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fd65f9971564dfc86617efaff967a67 : SUCCESS in 1h 08m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/495e1961b53d467aa1e847f2dbf4a48b : SUCCESS in 1h 14m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/daf991cb2be948c7ba476e33dcc3408d : SUCCESS in 50m 14s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c71a964949c54315a7877390848cd958 : SUCCESS in 1h 05m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/57c6c29aa8114a1c860e706f2dd78ea4 : SUCCESS in 1h 46m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d798554ccfd548ff9ef20e115ee72495 : FAILURE in 2h 36m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a6af94d8a1c04664b1b7acb02578ead7 : FAILURE in 2h 00m 44s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cd732f7e4bdefbff65228f0dfaafdf782eac0494","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-30 22:39:36.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"c61fcd2896258513993470443e441be989531d70","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 00:25:11.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : FAILURE in 1h 45m 07s","accounts_in_message":[],"_revision_number":16},{"id":"79793e0334d3a57bc10b7922c3f0ce57d28e1e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 01:30:26.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/db0a3bbe7ef34f9e9e8cfa98da7d1ad8 : SUCCESS in 51m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c9511a2b07043f9b1c0ebc747dc60d5 : SUCCESS in 31m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b0b59faf8fe445688c1183c9e599e121 : FAILURE in 1h 41m 52s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fad78e8d518f41938119d349fedf218e : SUCCESS in 55m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7d9423fd982744d1ade27f8f60f3c25f : SUCCESS in 52m 46s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/335ccbde5a364f6690cdc54cc9086a38 : SUCCESS in 58m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27a470c313f44dbbb1d7df4102e712ef : SUCCESS in 38m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/212a7918aae24e9fa51b589ac3fcf166 : SUCCESS in 32m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0380ae57e1b74df2b6481a6d69323670 : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86ef53f586754f33a2fb36d0c564a219 : SUCCESS in 24m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69b6130eb6d64e36a9e25fdffef90296 : SUCCESS in 24m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82397b9fd7fd4b3eafafbe5d67a3bf14 : SUCCESS in 24m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8572cd27cc2b410b9057dfb65143f7ee : SUCCESS in 9m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b322d81bac734fd683ae8685be7626d2 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54ef029d19b64fb8afab0c96280bdda8 : SUCCESS in 12m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fe6b8423a01c408c8555630cf7cc37aa : SUCCESS in 43m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3bf6c555a9734de1bde9bca104359315 : SUCCESS in 1h 11m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/60b9a61e39eb4b56951b7087945563d4 : SUCCESS in 1h 59m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c6b49761ec9642788501107c47a325a8 : FAILURE in 59m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/66c16e1f3b1f42bebd66f509aaabe04d : FAILURE in 1h 02m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/835bb342d2cc4f988551f9c154818228 : SUCCESS in 1h 15m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28207d8b2b934645b05ccb2a9b471add : SUCCESS in 1h 13m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d3cdf5d477d4e378d2c867c61c697be : SUCCESS in 1h 19m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/867fb1a3be0a40d681966a5032935900 : SUCCESS in 1h 39m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/efe6120d83204a15964c837db038fcf8 : SUCCESS in 2h 18m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20e0a7cbe6b24266ade23c4b5d1cb380 : SUCCESS in 1h 22m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/306e77a3424e44f99e83e8bfc577cbb7 : SUCCESS in 59m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3f2744d543b441dac27343e7c533c5f : SUCCESS in 2h 38m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/38df4f7fb02c4e1cb92243cf7e7a351a : SUCCESS in 1h 54m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1d54d8ee99c4072b11ec30b4a2c6679 : SUCCESS in 1h 00m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6be86b86cd3f401b8e1a567f62e88802 : SUCCESS in 1h 08m 43s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1727bfc704d5484ea8d1adeca84ba35e : SUCCESS in 41m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5378c7d7fb2a4cb4809f4c34dfdb0461 : SUCCESS in 1h 03m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/64f5397eabd64452b1f9023ab175244b : SUCCESS in 1h 13m 55s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a0a14e07b54e4f34b2829828a4f4a820 : FAILURE in 2h 35m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/51344779bff648c1a318ea7169a56230 : FAILURE in 2h 05m 57s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e7b8503e233fce518741d61e4619da439eacaf55","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-31 09:03:53.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"1901a3883740b7173ae03202ff9ae8fc1e280d60","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 11:13:04.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 2h 01m 31s","accounts_in_message":[],"_revision_number":16},{"id":"e69a311deecd57e7dc53ab46a83a20b6d5ab5291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 11:29:44.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eb08b250151a4f91a4712d6bd49917c0 : SUCCESS in 1h 22m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac5959b2e98947f19c2647bd9663660e : SUCCESS in 37m 57s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/af144503cc3b4faa9e67db4cef27a4f4 : FAILURE in 1h 32m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dcae1e1f2f7a41acbaeb0571ede21505 : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/935a171e2bb54eb4ae704b0a485b82bb : SUCCESS in 1h 06m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fed299a4bc7d44089a1d8390d537d0f2 : SUCCESS in 1h 13m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/047202d4dccc4c9bb25b8c4db22015ed : SUCCESS in 43m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28e49cc7d79a40ea8780a67035aa40ea : SUCCESS in 26m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4943f1f7cce4f959d5c4a3a83d7c7fd : SUCCESS in 14m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6400757fbdc74d99b10f2147e8994aa9 : SUCCESS in 34m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b39ab17f0b434a739e3e61610559b739 : SUCCESS in 35m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc47f9e2a4b34f048921cc770d2ee47f : SUCCESS in 30m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ad2289b0ab451699ea04d4dccd5eba : SUCCESS in 9m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fee514ab7db94b8a962d2497d43b619d : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d3e85b3744946ef90fdf8e20ce22acc : SUCCESS in 17m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a8b3d67cfb574230853ff87089d0809c : SUCCESS in 48m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2965fbc3aca44cf38e0a31dbba35242a : SUCCESS in 1h 09m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c45062ce720c47a3b5cbd59c617709b9 : SUCCESS in 2h 05m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/292183567bda406b9cd44279de0a5754 : SUCCESS in 1h 21m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/124a645355584b06be4e22070e696189 : SUCCESS in 1h 31m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47415e97528943b9ad4b09aa8143b9d7 : SUCCESS in 1h 01m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/767fef05d2f244e9adaa158176b2ba84 : SUCCESS in 1h 19m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8887911bb53c4cbaa8f320a9fff99417 : SUCCESS in 1h 22m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2db9a7baf8644d5da118a5352fa8db89 : SUCCESS in 1h 49m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4664cf53b225416c99374f7541489536 : SUCCESS in 1h 45m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a7c1fd374fa45a6a53e9f476f7835f1 : SUCCESS in 1h 08m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dce5d8ebb79c4700b99073ff391ad30e : SUCCESS in 55m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3553df43abb2488c8f0c2a25a12a66ed : SUCCESS in 2h 12m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1fd783914ca74f338641805f7178521a : SUCCESS in 1h 37m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c9dab463c3541a3b4a779c5106bf02f : SUCCESS in 58m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5b9848b9b9b6400489717fdeac162aeb : SUCCESS in 1h 08m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d87ef9971f1046f7a78a0d12f400d9e9 : SUCCESS in 40m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8229a77fbf5746d98b415d60486152ad : SUCCESS in 1h 04m 32s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa0b185305994cd1bffc1302d67df725 : SUCCESS in 1h 08m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f62a808f097c417dad4b804fbbbd02b3 : FAILURE in 2h 22m 16s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/87148dc1082243aba1f95211b2acc227 : FAILURE in 1h 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"93363d15b0b8f14969cb00bcdadf4709c488ed4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 11:30:06.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"92cad9e7d4c4897236bdd2f7b1d2b594e9f236cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:49:49.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6664cf1639fb4d669fb233fb7ed65e6c : SUCCESS in 52m 20s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e0120b9a020d4b1e969b781b06aa66c8 : FAILURE in 1h 18m 22s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/17b19ca7fa0047a3a444e9fa6425a007 : SUCCESS in 1h 08m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2f385b47a8443c284c1cc81fa35ec15 : SUCCESS in 1h 23m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e5cc55e0ddb458a873340820ca67da5 : SUCCESS in 34m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a24c442c2024046b640f7130f267fca : SUCCESS in 15m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df4034dde7794967a7e966613f970918 : SUCCESS in 37m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f540225a4d91429ebc76b488fa2a6dc7 : SUCCESS in 25m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6925a1cecb4f75ad7bddd59d4a900a : SUCCESS in 9m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afaf62e664ea420f8210b1b0daf9c442 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fb352acaf2c414593df3d7449a8fdbb : SUCCESS in 12m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffc1858bea0b4145870aee6adc2761e9 : SUCCESS in 50m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/73c6ddad4d504b9ea10b93bfe61e456d : SUCCESS in 1h 05m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3e047a3873d441da2b074eee1c8fcb2 : SUCCESS in 1h 18m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7a1cd621461141569dcecb15f78a67bc : SUCCESS in 1h 33m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a6c867eb65cd4fbda3b9c15436c13d8d : SUCCESS in 1h 09m 11s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27ff146b2e79493e99d85f9ee1bf40be : SUCCESS in 1h 22m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a619b01259741fcbf33fe20573bd0cd : SUCCESS in 1h 17m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/86bf133c499a457c845921295a1473c6 : SUCCESS in 1h 13m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7edde6dcc90148138176fd9381bcfa58 : SUCCESS in 2h 09m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4618dd1fa8d04a8c8cc2d2966da0b333 : SUCCESS in 56m 51s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2927c22b09044cd3a93fc4d51f4087db : SUCCESS in 1h 34m 26s","accounts_in_message":[],"_revision_number":16},{"id":"4cd1f2cce3b877c8e0eb3b804463e954720be837","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-31 15:24:55.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"d2c89a51101fd4a86a0990d64a712ee6f4ff01b6","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-31 17:06:25.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"1cd5f67ec64277569f4e91d14dbbdbee668d95eb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 17:12:57.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 47m 31s","accounts_in_message":[],"_revision_number":16},{"id":"ff8d57d27013cdbc25f32e86807591b64f00d2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 18:16:31.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/aba9b3a39e2e46b384be9df1b7d92e50 : SUCCESS in 58m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa67663c5cfe435e830417f88a793450 : SUCCESS in 36m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3c4f35f7468646b2890a16d3b6fa98a5 : FAILURE in 2h 06m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9a290309cd9c4807aae65d3dd339a715 : SUCCESS in 1h 25m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3d44485d006e4dab88c932841eab659b : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/529105ee57964c7ab77886a14ca83c9d : SUCCESS in 1h 11m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11411b025c9247fc871465538fb25dbc : SUCCESS in 39m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e7eae1322eb496fbc02bb04e1942d6c : SUCCESS in 35m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9cc8a79fc746558f45c61abb766ff8 : SUCCESS in 14m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4101c2fb8cb14c5db971c3dfa94fc2c4 : SUCCESS in 30m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4c59b70b3b7490b84f56915b5e20e94 : SUCCESS in 29m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0881c3740b144e19a0f80e417920c11 : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43d955465462495282dae3fc948f1d1b : SUCCESS in 11m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/694aae6e08f84e179ef72695e4ec7804 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce2b2d3f300b4c89806e24d4a11871a9 : SUCCESS in 15m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f0aaeef685d044aaa2f7d8209314e25f : SUCCESS in 50m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/440b29d59f944978b851f6a1c4c06664 : SUCCESS in 1h 09m 46s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4fd58c64bb3d4de28d15aec3b44cacbd : SUCCESS in 2h 49m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/64b1250810444253b90ecdeff8d1a18e : SUCCESS in 1h 30m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d393dff241ce437ead28c592c8b4bb58 : SUCCESS in 1h 16m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8f30fbbdd634446ad7f007c9e44bafc : SUCCESS in 1h 15m 18s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/33ca781230d34e859a206365efcb1112 : FAILURE in 1h 51m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ffd11e562e3b4f86b67209fe8e571581 : SUCCESS in 1h 28m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6472635dc9ac4308ac19d684eb6ef146 : SUCCESS in 1h 48m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a92cdf51f3f4cb18a51e852eb882804 : SUCCESS in 2h 39m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b3babd881564b4ca9c245d7b96d9461 : SUCCESS in 1h 52m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4b21a1cef141493487e3a2c3de0fe265 : SUCCESS in 1h 06m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf0517371a80481b95e3e20ec412dbae : SUCCESS in 2h 45m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/110816f36ee04e47bd6822a090a75d5e : SUCCESS in 2h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/317c485f68d9441bb7ac97720751b8f9 : SUCCESS in 1h 35m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4fb29d8612b04e11879f5bcce0e7fc06 : SUCCESS in 1h 33m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a163d39217104f9abe8eda19a27dc201 : SUCCESS in 42m 59s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2eccffe1a6f468291941bb6e436e43f : SUCCESS in 1h 07m 58s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/98aefa3fd90145bc961fbf90c41b5be0 : SUCCESS in 1h 40m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0b6ae1f6c98c4afdb7ff6b600664eed4 : FAILURE in 2h 44m 15s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/63f2b8d8a2544e8c90cff064d9400f90 : FAILURE in 1h 51m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1c38c01be97b2fdc1743d0b68281fd1efeae19f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-31 20:52:59.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"fa6d561b0b585dee206d7b0a4ae6f62ba068b818","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 22:56:15.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 2h 02m 47s","accounts_in_message":[],"_revision_number":16},{"id":"8482c1eb9376070dd5d6407cc404d309aad9de8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 23:28:48.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eab87dc1e8464ca8bd5d265ac5b3214d : SUCCESS in 46m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d6fc2f005654ac9a9c8340ed7c2001a : SUCCESS in 36m 53s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b065376a880d40c09fd0e877a6926986 : FAILURE in 1h 25m 32s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fff6192076ae436db2aa0d63f4e7d503 : SUCCESS in 1h 20m 49s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9c544a96903a4007bf78946f415b6420 : SUCCESS in 1h 14m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a80bfe709e4c41faa305b72aa11fdc2e : SUCCESS in 1h 10m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8e0885b3c6e4eac91e0318bc3b4f32c : SUCCESS in 38m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e79bedccdd494fdea762eb597f0d828c : SUCCESS in 29m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58fa0fbbf384bfe8e962d600751e792 : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc402b2ce8a441879ff2faa9872f9f9f : SUCCESS in 23m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06782d01c2f04a04a1d6d67f83f5e145 : SUCCESS in 31m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd3e03773234eb5805c775db5f67656 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e039cc32fae444eada56f529c16ff05 : SUCCESS in 9m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/382b823140d04a5fac48df28dc8856db : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/061f0a7a47684e5db3946b82cbf56251 : SUCCESS in 15m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e4bbddfe5c5428a8782712fb18c626e : SUCCESS in 45m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f46bbf29129140b7aedafabff1dc2e53 : SUCCESS in 1h 09m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3c07b3ff94984a0589520b6b23c715e3 : SUCCESS in 1h 46m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/461404e5699e49a3a4f5a5ce7205a4cd : FAILURE in 1h 05m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/210e46f4a54b4449b15cc96cdfaabbc2 : SUCCESS in 1h 08m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bda966c21f1d4e1bab4da131fa78af37 : SUCCESS in 1h 03m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7508b8ded7e347f4bf2774e5944a60f4 : SUCCESS in 1h 09m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f117a8be5b749cfbe499bda3cb27882 : SUCCESS in 1h 07m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/73e7a4c28bb747e1883fae4763a13092 : SUCCESS in 1h 38m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/05a47d41b32b48d69b3f34dd92d926f0 : SUCCESS in 1h 42m 39s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4abec3f068f4f9e810c7be7fdd221eb : SUCCESS in 1h 05m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9c73724e7b56434aa38899f6077f797e : SUCCESS in 54m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/546d9162ea8247a1834fec65b6428635 : SUCCESS in 2h 14m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5e8c67dbde7d4df99b46e71ce8c02532 : SUCCESS in 1h 55m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8e4d0299738470080bfb9021aa3cd40 : SUCCESS in 58m 17s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cd72fafec8414f49ba5ae258b8b03877 : SUCCESS in 1h 12m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/afa26ef3a5cd45f79b8d15dfd7cb1162 : SUCCESS in 39m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d3d2bb6ee204f39895c0fdd1fb32773 : SUCCESS in 1h 03m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d79f07b15564410caab1650e6551b854 : SUCCESS in 1h 13m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bb84b050b60a46df85b69c0fea9dd536 : FAILURE in 2h 30m 11s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3a740e41315b4bea9140229f22d7acc7 : FAILURE in 2h 06m 28s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b3e2b646c13ff8f67e790eed6e2fb4723653b6e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-01 23:49:57.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"3b7f789ddbc25480f88c031812f63da232161221","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-02 01:42:34.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 52m 09s","accounts_in_message":[],"_revision_number":16},{"id":"7d1512249631898f6bbf02293cf42e63bb97dd26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 02:28:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c9f4c234a1444202a2aff7ccbd64e60f : SUCCESS in 51m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c720530491924ad59a9426ac9bbcb07f : SUCCESS in 34m 24s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/40bbd3e5f153465c8dd4c752a1604bce : FAILURE in 1h 52m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/fe0567a093294a3ba7a5bede8491e485 : SUCCESS in 1h 18m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b04faf29e9a8444ab6b527f8a2bb340a : SUCCESS in 1h 08m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/daae245a7ea2449ea0eeaca2519ad053 : SUCCESS in 1h 01m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3b6f5b0d85b4c248b4ee2e853e40b07 : SUCCESS in 37m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce1ce74a6ba5492793d800feae187799 : SUCCESS in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a6675779964572a22345a3260f0b95 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d430558ec8437eb65c0d2524d22a38 : SUCCESS in 27m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea7ff19dccee43e19d27b25a4d189abd : SUCCESS in 30m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7bdf7f705e41d397e93febb788a54c : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10073cdd26e04fecae80027d1dae61bf : SUCCESS in 7m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b09ca28935e4425d8247e304fb8cbe89 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77ef8f2056ef4bef8d1eb09124b7e21d : SUCCESS in 15m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/379abb43cb684ad4bc8c77ceb8c2adec : SUCCESS in 47m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b1fde74229e04792881518cc65d9cf59 : SUCCESS in 1h 03m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3f2bd610e1684ecab8ce0edba8d074ad : SUCCESS in 1h 56m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5b2d8392f5f6406a8481a93f315c25cf : SUCCESS in 1h 08m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/535f86be9134479786b5faa867a8d95f : SUCCESS in 1h 14m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc3049a0d31244228389168146402a36 : SUCCESS in 57m 40s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ad13801b8db141bd800e05b48aad0fda : SUCCESS in 1h 05m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1ffa48dd896740109f9777090cdbc81d : SUCCESS in 1h 15m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d988e6834f73414088acda80e217c227 : SUCCESS in 1h 42m 46s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ba652360de24d3296868735465f7e2d : SUCCESS in 1h 41m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21eb90a7758d458da112f51c74841fb6 : SUCCESS in 1h 07m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6367a23657de430182b91ebe7398a922 : SUCCESS in 53m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa12f12d51544ab1abc344751798cd60 : SUCCESS in 2h 18m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6e6334b716024c5289050e7510d17e33 : SUCCESS in 1h 41m 36s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a1f061e749074c529eb3e3f7e4890f0d : FAILURE in 1h 36m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ea74483e63740ec87b5aa059d62d6a8 : SUCCESS in 58m 59s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/46056e3566f34a82827664be6d14fb17 : SUCCESS in 1h 07m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bed6c927b19b414a94f26ea8cb15a94e : SUCCESS in 40m 34s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c57a18f45dd4a82b47533f4cba12044 : SUCCESS in 59m 21s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/83378cf743594579986ccb640968454d : SUCCESS in 1h 13m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/42942ce1da17476380707bf24b91822f : FAILURE in 2h 32m 38s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0710dbea04d543c7b7a8e3f25b78f830 : FAILURE in 1h 59m 01s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"57da3a8f1fba2e042d45fd9041c06eff0765dcb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 02:28:32.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"b47f873198e1c741fcb4e28e6b7e9fab56fc37a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 05:02:10.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1eeff11489b449ab875e976fe0a5337f : SUCCESS in 47m 50s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a85c24aa232d4b02aca9de21029b0414 : SUCCESS in 1h 06m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e1feaeae2dc943a4ba0f398d06126ce1 : SUCCESS in 1h 18m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8e3ca01eb285423a9e2494ed28940105 : SUCCESS in 51m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f489791bddc4159b5f46a2567afc98f : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2cbfecacbe49dfaf3b528770a15fce : SUCCESS in 12m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ba5781d415746cb93db3ef800e6fb47 : SUCCESS in 22m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9b0b325cfa94205b4bbb76b495aba09 : SUCCESS in 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0d9856a3aa4938884ee20d8e30a1b8 : SUCCESS in 8m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a77a818848948a3977003d2c969cb5b : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf25464662c478c9efbb5e4d959c8c9 : SUCCESS in 12m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/850591c5d2f049a3b93f6d85ba216dd1 : SUCCESS in 39m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d29e724e00f743709e74235fa1a34852 : SUCCESS in 1h 14m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d7f30c271fed44cc9e794add25dc5c2a : SUCCESS in 1h 12m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c642930ce63543ecb3240a6a6ca778fa : SUCCESS in 1h 23m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c148a1a4fd1742d28305d73434d3cbd9 : SUCCESS in 1h 21m 33s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de7acf72e8a74e66a5ada0162b474554 : SUCCESS in 1h 14m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a409174c701e46238097a19095664de2 : FAILURE in 49m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eae2fcdd1bac4ddba5b4c97944e28171 : SUCCESS in 1h 17m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d704f07c2fd84425a4862e4ddc328040 : SUCCESS in 2h 26m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9eab226855942e3ad4220b884fc4738 : SUCCESS in 1h 06m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c44dc8f1a81c4a079a7a58707e34743d : SUCCESS in 1h 35m 28s","accounts_in_message":[],"_revision_number":16},{"id":"356f3d58567a38e5478efef190f2962ff3a2b33d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-02 08:50:14.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"a51abf5c9fc5c74f95d2feab3c1c7e32ad180aee","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-02 10:39:35.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/694568/16 : SUCCESS in 1h 48m 52s","accounts_in_message":[],"_revision_number":16},{"id":"d087b898f61ee2eb9b0e63459ce900571786fe14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 11:27:07.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7e2e40e15ced41bd87a14532f40d134d : SUCCESS in 54m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/491e507e328243d0a6cfa6753447c090 : SUCCESS in 37m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ffcbfe56b78447649d5a3078e97078e3 : FAILURE in 1h 43m 03s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bbfeebfd7b774e7d816adf1a2230e0f9 : SUCCESS in 1h 08m 53s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/49adef70347a46fc80218a644df8ff14 : SUCCESS in 1h 03m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28c7aed811c64be990c78521170d1059 : SUCCESS in 58m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27920bd841d840a58c35212d82b0c168 : SUCCESS in 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc388cf977d547e7ae131f9696bf380c : SUCCESS in 26m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c291bd1f064cefa8930a48c849a162 : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f05dfd334b7c4b6d8a5369bc5faf264f : SUCCESS in 25m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71170cd4c6f948d987783e5a93eb3330 : SUCCESS in 30m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b1f717792994037ac57b7592e6ab40b : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830005e105d449c284cc41c04805768e : SUCCESS in 7m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bd976d2c51a540be80fdcf7afbde65c3 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/220c9d16e15340f0af0951ad6c3fea21 : SUCCESS in 15m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3e6bb913db614458b21aa2d6558f0881 : SUCCESS in 43m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a9b23e0519da487fa926d4c61bd040d9 : SUCCESS in 59m 58s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/061dc3bdd3aa469d92156507b6546dfd : SUCCESS in 1h 56m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8ed2d88193f14bbd8b05927ededabf7e : SUCCESS in 1h 13m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/68f41fcaeefc465ca112407f9bdf1917 : SUCCESS in 1h 11m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/425109a29ecb41028119be43d08d472f : SUCCESS in 1h 04m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/816a751fd1b34634acc85ce41f748a59 : SUCCESS in 1h 03m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/565ff75e220840429ce5c1256f9c24c6 : SUCCESS in 1h 10m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b89d74b17e3b4b03a76e20ae815940f9 : SUCCESS in 2h 00m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/978dc2f633c949b4b2edc41dbd9431ea : SUCCESS in 1h 45m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/27f2bc554ff649d486859167db57794b : SUCCESS in 1h 09m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7ef01224e3374df6a5fbfcacfbaf4bb3 : SUCCESS in 51m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35e24770fd6c46a8b8bc3142f3b3fcac : SUCCESS in 2h 14m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/95fe5c25e7d345008f69f0d6cd3d5840 : SUCCESS in 1h 51m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d5b3e09c5b53427992fe1a14184508e2 : SUCCESS in 1h 40m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dffde7aac364022abdc3748fe5fadd4 : SUCCESS in 1h 09m 59s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e5f3667f3a13459698c1d73cb07b0566 : SUCCESS in 1h 06m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/64e2b33615294641a7b6908a8b88ddd8 : SUCCESS in 45m 24s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f1a8cd939c446dd9f1735bf726b30b5 : SUCCESS in 1h 03m 49s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9225d8ea84e42da8e27cc70f3073c84 : SUCCESS in 1h 16m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d99de5a0f1fb4f6ea47d8df699cfcb42 : FAILURE in 2h 23m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f818e346713e4ddc829af6df9c3b45e2 : FAILURE in 2h 06m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a419ecbc855295ce6488bf06d2407c0540a4753c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 11:27:28.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"eb490d81a38695b65a7dc5f9eb7c7d640e9b62e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 15:51:15.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef71e5bd317d4129af2488f7ae4b4277 : SUCCESS in 48m 59s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/89cc10920519435da3d57236e8780d71 : SUCCESS in 59m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f2aa19b93dcc4eacb04b09da8c3fb8c7 : SUCCESS in 54m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2d0fa98748a348979e815ff9d3f25199 : SUCCESS in 52m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c90b4752a9d4e5d99cdc8e894a7b734 : SUCCESS in 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bcb0ba480c14aa088e587d7a82f97dc : SUCCESS in 11m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09dee8c55cf04b68ba1398bf9ff105b3 : SUCCESS in 22m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a1d4878ad6241a4afbc8bd7001781e6 : SUCCESS in 25m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32b4fd9f9144480be7350c287628786 : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5e2f2f8a7514747bf2638636f13b0bd : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0e61461747b47fb934581d761039632 : SUCCESS in 12m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/22e577583d9640a7aa7b59d4c5cfa44b : SUCCESS in 39m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4beb230eca104d38b0de8ea6c2fa7501 : SUCCESS in 1h 01m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cce9667f350b4eb699717ae161c285f7 : SUCCESS in 1h 04m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2ee61c98d28b4dc8997ec49abc7c56a5 : SUCCESS in 1h 07m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/edc97313fd3a489aa5ae52b4a0c8034e : SUCCESS in 1h 10m 11s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7eb91dae5204ae280fcf1581153f5c9 : SUCCESS in 1h 10m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93e414b2814549ba8ef8a41499af92e7 : SUCCESS in 1h 05m 16s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f834ff99b3ca4182bca2af2f9f1be32b : SUCCESS in 1h 05m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5210c686a7d64880b0dc5692fbf44e98 : SUCCESS in 2h 14m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a7ee59dc66947db8bb016aa8abc5884 : SUCCESS in 1h 00m 00s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ddeb3ea20d534864bde3b7ecf7991b49 : SUCCESS in 1h 43m 47s","accounts_in_message":[],"_revision_number":16},{"id":"83bfcc1ef01cc16f1d18875650f6e0ad26593494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 15:51:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"dc17e52ebde48041bd3b834e4edadd0c972a308f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 15:52:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a7bf9f35774486b132546bbde94731 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/052657f7ca2945fdb6f1a44c6ee8d7ef : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"f8a3b2c589727acc828ac0348c4fffeb818fbe50","revisions":{"14b6bc41ff5e773565e40af92d4a128f942944d5":{"kind":"REWORK","_number":1,"created":"2019-11-15 15:07:36.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/1"}}},"commit":{"parents":[{"commit":"058dd639a9817fcdb638809b8228f20a1417774a","subject":"Merge \"[fullstack] find ip based on allocation_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/058dd639a9817fcdb638809b8228f20a1417774a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14b6bc41ff5e773565e40af92d4a128f942944d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14b6bc41ff5e773565e40af92d4a128f942944d5"}]},"branch":"refs/heads/master"},"69860228e0a6dc0380db320c2b3ab85dfb696592":{"kind":"REWORK","_number":2,"created":"2019-11-15 15:10:40.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/2"}}},"commit":{"parents":[{"commit":"058dd639a9817fcdb638809b8228f20a1417774a","subject":"Merge \"[fullstack] find ip based on allocation_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/058dd639a9817fcdb638809b8228f20a1417774a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:10:27.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69860228e0a6dc0380db320c2b3ab85dfb696592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69860228e0a6dc0380db320c2b3ab85dfb696592"}]},"branch":"refs/heads/master"},"7e3c34381703ca30454049eccff1023a79ddac50":{"kind":"REWORK","_number":3,"created":"2019-11-18 09:21:21.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/3"}}},"commit":{"parents":[{"commit":"058dd639a9817fcdb638809b8228f20a1417774a","subject":"Merge \"[fullstack] find ip based on allocation_pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/058dd639a9817fcdb638809b8228f20a1417774a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-18 09:20:50.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e3c34381703ca30454049eccff1023a79ddac50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e3c34381703ca30454049eccff1023a79ddac50"}]},"branch":"refs/heads/master"},"ca728d42c95fea5be06b1a7c17c719d07b6f75cb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-22 11:18:20.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/4"}}},"commit":{"parents":[{"commit":"67b613b795416406fb4fab143b3ec9ba8657711f","subject":"Merge \"NetcatTester.stop_processes skip \"No such process\" exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67b613b795416406fb4fab143b3ec9ba8657711f"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-22 11:18:05.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca728d42c95fea5be06b1a7c17c719d07b6f75cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca728d42c95fea5be06b1a7c17c719d07b6f75cb"}]},"branch":"refs/heads/master"},"b4598a8f3e4cb3e8178aed13d8aec7a91bef92dc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-11 18:21:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/68/694568/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/5"}}},"commit":{"parents":[{"commit":"c627d0773f9906891edf37207d9b4744c98dbac1","subject":"Merge \"Load the glibc library only once for Pyroute2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c627d0773f9906891edf37207d9b4744c98dbac1"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2019-12-11 18:21:03.000000000","tz":0},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4598a8f3e4cb3e8178aed13d8aec7a91bef92dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4598a8f3e4cb3e8178aed13d8aec7a91bef92dc"}]},"branch":"refs/heads/master"},"199e776a24ff1db20d5321d4a09ba10919749050":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-30 13:04:17.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/6"}}},"commit":{"parents":[{"commit":"68f89c79c803a3c9420e2a8df7a56499e88091ad","subject":"Merge \"Remove locks from privileged ip_lib module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68f89c79c803a3c9420e2a8df7a56499e88091ad"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-12-30 13:04:06.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199e776a24ff1db20d5321d4a09ba10919749050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/199e776a24ff1db20d5321d4a09ba10919749050"}]},"branch":"refs/heads/master"},"4da1677d5ac2ff7f0a04b9fa294708bce284c17e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-01-14 10:17:19.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/7"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-14 10:16:51.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4da1677d5ac2ff7f0a04b9fa294708bce284c17e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4da1677d5ac2ff7f0a04b9fa294708bce284c17e"}]},"branch":"refs/heads/master"},"1f4db61a91be05436d965c2fbb7e28beddda2418":{"kind":"REWORK","_number":8,"created":"2020-01-14 14:38:46.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/8"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-14 14:38:07.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f4db61a91be05436d965c2fbb7e28beddda2418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1f4db61a91be05436d965c2fbb7e28beddda2418"}]},"branch":"refs/heads/master"},"46e657c275514d5eb288a537c59d109979a21cad":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-01-14 16:08:05.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/9"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-14 16:07:23.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46e657c275514d5eb288a537c59d109979a21cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46e657c275514d5eb288a537c59d109979a21cad"}]},"branch":"refs/heads/master"},"e49ff3d632673bf8b80f288b79a4711d4e6f6673":{"kind":"REWORK","_number":10,"created":"2020-01-16 09:57:40.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/10"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-16 09:57:13.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e49ff3d632673bf8b80f288b79a4711d4e6f6673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e49ff3d632673bf8b80f288b79a4711d4e6f6673"}]},"branch":"refs/heads/master"},"fdd466e8b14b3d1e772996a079416b8dcceb2e2a":{"kind":"REWORK","_number":11,"created":"2020-01-16 15:53:29.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/11"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-16 15:53:14.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fdd466e8b14b3d1e772996a079416b8dcceb2e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fdd466e8b14b3d1e772996a079416b8dcceb2e2a"}]},"branch":"refs/heads/master"},"9a1e7f6bd2d16f139285af3137918b602772f174":{"kind":"REWORK","_number":12,"created":"2020-01-17 08:00:32.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/12"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-17 08:00:18.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1e7f6bd2d16f139285af3137918b602772f174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a1e7f6bd2d16f139285af3137918b602772f174"}]},"branch":"refs/heads/master"},"986303e2140809b8cb90bace95468acc69082879":{"kind":"REWORK","_number":13,"created":"2020-01-17 14:28:16.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/13"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-17 14:28:01.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/986303e2140809b8cb90bace95468acc69082879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/986303e2140809b8cb90bace95468acc69082879"}]},"branch":"refs/heads/master"},"6d11ca089ba7d68c4a1ec7cae2dfecef2ce00699":{"kind":"REWORK","_number":14,"created":"2020-01-20 14:54:40.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/14"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-20 14:54:25.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d11ca089ba7d68c4a1ec7cae2dfecef2ce00699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d11ca089ba7d68c4a1ec7cae2dfecef2ce00699"}]},"branch":"refs/heads/master"},"04b8cb2aa905cdb0f6de166a022497090eab918d":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2020-01-22 08:20:07.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/15"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-22 08:17:20.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif. This change\nshowed we needed also a newer version of cffi and keystoneauth1 in lower\nconstraints. Update also the requirements file for debtcollector and\nkeystoneauth1.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04b8cb2aa905cdb0f6de166a022497090eab918d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04b8cb2aa905cdb0f6de166a022497090eab918d"}]},"branch":"refs/heads/master"},"f8a3b2c589727acc828ac0348c4fffeb818fbe50":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-01-28 10:07:53.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/68/694568/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/694568/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/694568/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/694568/16"}}},"commit":{"parents":[{"commit":"bbcaed7f345a64a88b3810b2c4a8465ebf72a30a","subject":"Merge \"Move arp device check out of loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbcaed7f345a64a88b3810b2c4a8465ebf72a30a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2019-11-15 15:06:28.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-28 10:07:40.000000000","tz":60},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\nThis change showed we needed a newer version of debtcollector in lower\nconstraints, aliging with the version required by os-vif. This change\nshowed we needed also a newer version of cffi and keystoneauth1 in lower\nconstraints. Update also the requirements file for debtcollector and\nkeystoneauth1.\n\nChange-Id: I5e190c3db8bed0a264b911cdf425aa4c9b51f768\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8a3b2c589727acc828ac0348c4fffeb818fbe50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8a3b2c589727acc828ac0348c4fffeb818fbe50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
