)]}'
{"id":"openstack%2Fneutron~696296","triplet_id":"openstack%2Fneutron~master~I9fe64f954d227efaab5e96c6150df44f36a2530a","project":"openstack/neutron","branch":"master","topic":"bp/neutron-ovn-merge","hashtags":[],"change_id":"I9fe64f954d227efaab5e96c6150df44f36a2530a","subject":"[OVN] Import ovsdb related code","status":"MERGED","created":"2019-11-27 12:21:55.000000000","updated":"2019-12-08 03:01:54.000000000","submitted":"2019-12-08 03:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696296-1575774014926-a31ee413","meta_rev_id":"ca185bb31d2235d3733877eba61556ca38d6833f","_number":696296,"virtual_id_number":696296,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-07 22:01:36.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-04 14:09:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-05 07:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-12-06 08:44:15.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-08 03:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-07 23:43:49.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-04 13:18:50.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-12-04 13:18:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-12-05 22:16:33.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":1,"date":"2019-12-04 13:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 22:48:46.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-12-04 11:35:58.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-12-04 13:18:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-04 14:09:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 07:35:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 22:16:33.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-12-07 22:01:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-07 23:43:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-08 03:00:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c581ffaf7cef819123273d366a5f17068cec6a41","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-27 12:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86880b7ec1dc97337df78ded03e25932c2dd3f99","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-27 13:05:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ab361a81e40c7b57c1c16797c123008c4610fab4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-27 13:06:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/1/Neutron-MACVTAP-ML2-Sriov : ABORTED\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/1/Neutron-ML2-Sriov : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"b5304e23badf6065aeeb05ef85970a96a2692bdc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-27 13:12:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"584c432dcc75d57dcf7321adb67a20257084aa76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-27 13:58:35.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/2/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/2/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"42ad425958e4f69a57929518f935cb515d854cbd","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-27 14:25:09.000000000","message":"Patch Set 2: Workflow-1\n\nwait","accounts_in_message":[],"_revision_number":2},{"id":"448b7ecb1352b087cb681acaa45b6cc1dbd36950","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-27 14:51:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/2 : SUCCESS in 1h 45m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2828071f3a83c3ea1b2c2778f69c7d7d900b3a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 15:34:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df5cde1459a54ee1a558ef068e7007f4 : SUCCESS in 55m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d04d4716507c44c6a3ecf7465908e1cf : SUCCESS in 33m 05s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0705a4b6652c4b74b885b183fbff7f15 : FAILURE in 1h 32m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9fc70d1c9fd74f03aea103209061de05 : SUCCESS in 1h 00m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c3d55939a9d34fc3a9acb9afd7d75974 : SUCCESS in 1h 04m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c76d9f79943146e997bb0ecea1bdd770 : SUCCESS in 57m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acda8cbb950941318e52c7d7953a376e : SUCCESS in 30m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e33851ed0574409fa038a54300ef7592 : SUCCESS in 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6754be4382b4cf19afcf0ea7ba2ad89 : SUCCESS in 12m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eac1784f134644c4a07a0ff1acf75407 : SUCCESS in 22m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d614b216bcd742a3a557c3229eb00e9f : SUCCESS in 29m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60ce4e07fc644c4a350893e6f04caa4 : SUCCESS in 9m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6df571fb29c498c9c7494762537db76 : SUCCESS in 1h 04m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5b8789898a854a7a896d56ef83517cc1 : SUCCESS in 1h 26m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/829c2779bd8d4f839560254c4426b806 : FAILURE in 44m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f26d6c347db84123be8c181f6844df7f : SUCCESS in 1h 00m 43s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/650ac2b01c7e41328ba5b2516e94033e : SUCCESS in 2h 18m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3d5f83b647e34b26bd1b3b7117f44424 : SUCCESS in 1h 45m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/09026c8dc15846b58381c328b6f2a60e : SUCCESS in 1h 42m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/159dda4478e74072ba777c499bc1b948 : SUCCESS in 1h 12m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ba95a6b7fe9648698cfb8853b0f5aeab : SUCCESS in 1h 19m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/63f69f95e6904482ad7ceb525438344c : SUCCESS in 1h 49m 24s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9bba3f019a747c08cf6d71a910b538f : SUCCESS in 2h 12m 01s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8466cf5575614db7bd1aafa1e34d39e3 : SUCCESS in 1h 17m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/00962d053f3f467abf8b7b7f621d4ef4 : SUCCESS in 55m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d5cdb92d69e4cd49d470f1fe61b416c : SUCCESS in 2h 10m 29s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/468ec0a26bcf43f2b5e8614f81f3b6a7 : SUCCESS in 2h 23m 34s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/6538f3f31b0a4c90bcfcf705c0c599ed : FAILURE in 18m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d50701e23e0f472e921866024f149bfb : SUCCESS in 1h 12m 20s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5a397c4e8f544e888c0c58108a1fe43c : SUCCESS in 1h 04m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c41aa2f70472421c835e44d51f51be0b : SUCCESS in 45m 29s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6246ba823d1a4176b219d5714928c037 : SUCCESS in 1h 02m 07s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e44ad388f934baf9e10676a14dbd73c : SUCCESS in 1h 18m 13s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a91c945dbc9b429798d154e4158f5c6e : SUCCESS in 2h 07m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1f6cc2f977c3d3bef50e6889aa8c6a4adfaf408","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-11-27 22:48:46.000000000","message":"Patch Set 2: Code-Review-1\n\n* i don\u0027t understand why you put these files here instead of neutron/plugins/ml2/drivers/ovn\n\n* i don\u0027t understand why you submit very small patches like this. are these small patches testable independently? otherwise i\u0027d suggest to submit the entire networking-ovn code as a single (or maybe a few) patch.","accounts_in_message":[],"_revision_number":2},{"id":"624cec2d19ea3014998c7b3fbc29234cbe4adc29","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-28 08:41:07.000000000","message":"Patch Set 2:\n\n\u003e * i don\u0027t understand why you put these files here instead of\n \u003e neutron/plugins/ml2/drivers/ovn\n\nWe talked about it basically and we decided to move it here because those files are used not only by mech_driver but also by ovn metadata agent.\n\n \u003e \n \u003e * i don\u0027t understand why you submit very small patches like this.\n \u003e are these small patches testable independently? otherwise i\u0027d\n \u003e suggest to submit the entire networking-ovn code as a single (or\n \u003e maybe a few) patch.\n\nWe decided to move it in more smaller patches, even if this will not be testable easy at once just to make it easier to debug and more logical in git history later.\nThis will not be considered as done until we will push patch with release note which will say that ovn driver is now in-tree finally. And than it will be testable in gate for sure.","accounts_in_message":[],"_revision_number":2},{"id":"5fff961dd85398f4f8047d7456c19dff130395e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-28 13:24:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4783e79914b7d35b56bb65c18926405a8a35ed2a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-28 13:51:05.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"58b2b870edc01733e62e1214221d0ced6a1a049a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-28 13:52:41.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/3/Neutron-MACVTAP-ML2-Sriov : ABORTED\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/3/Neutron-ML2-Sriov : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"d5d3959b79e8316c6b01557853e6ced5a2ea8b17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-28 14:45:07.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/4/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/4/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"55198908b549c49fea5dcf674f5132395dde348c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-28 15:09:06.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"e44b36ebe2cd8abd0b9b040f4d1762c6e4c1eed4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-28 16:16:07.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/96/696296/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2125d72eb3a351e4c1420785a96457db02ce5ce8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-28 16:35:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"ea630d4fe634f76dbebd22ae550336311d878d91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-28 16:37:07.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/5/Neutron-MACVTAP-ML2-Sriov : ABORTED\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/5/Neutron-ML2-Sriov : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"211fbc8c4064775a5b6cc8e06d97caaafed2e502","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-28 17:11:36.000000000","message":"Patch Set 6: Code-Review-1\n\nThis code should go to\n- agent.ovsdb.ovn.commands\n- agent.ovsdb.ovn.worker\n- agent.ovsdb.ovn.api","accounts_in_message":[],"_revision_number":6},{"id":"7c5a44139facde8cae0a8f1a608bbebe51438b18","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-28 17:19:08.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/96/696296/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"7f63a20303e13289860fb00112fc0395db5c5c88","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-28 19:20:43.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/6/Neutron-ML2-Sriov : FAILURE\n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/6/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"4de2624fa5969f520c4ab25de1664e670685d807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 19:54:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/995bc2c0c35e4cc18fade00380584aa0 : SUCCESS in 1h 03m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b3a28f99da54b829e61922d00093481 : SUCCESS in 34m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/aff54f4e321845828f63d1f5617613ec : FAILURE in 2h 11m 25s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/62fd0e4aafe7417a8b79d4f55f96168d : SUCCESS in 1h 06m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a31520bc7ecd4b76b2c8a29ca48f58f7 : SUCCESS in 54m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c71cb00f48b74f14837502085c347dd0 : SUCCESS in 1h 05m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd49ace164054c0286a554593f019bb0 : FAILURE in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b00f62f173984f7883cdab8b4bc2318c : FAILURE in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7ccdf9f5e74b09ba1591be4e1c88ea : SUCCESS in 14m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f144e9078025420f8fcad75d2747bdcd : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c49302200c04f1aae8a2d18e9a86812 : FAILURE in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc4587b20e048d5a417a42492f0e4f9 : SUCCESS in 7m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/219ee4fa73924d5fa04247f73abc2052 : SUCCESS in 57m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/88d260c231de41fab949378d79e9e493 : SUCCESS in 1h 14m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6bfe8f0b75c14fc4a396a3ac6282b843 : SUCCESS in 44m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2ab9bef808b44216a3b7d87768d30cc8 : SUCCESS in 57m 56s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5c7b0e80d7514e3eb850a5d9812abbc7 : SUCCESS in 2h 14m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dfcb2ad676a34762817736dea1003973 : SUCCESS in 1h 19m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9c4fc5bdd1bd4892ac2708ca1ced1ea9 : SUCCESS in 1h 27m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d088a0f1e7da442fb88899eb45d66dc9 : SUCCESS in 1h 09m 52s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c0c2782f1a95408dbc728a747d98efab : SUCCESS in 1h 13m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/69cc106e72554c62bcf4d2748bd67778 : SUCCESS in 1h 36m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b124da10dcc0460385c7743f85886a5b : SUCCESS in 1h 45m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58f844fbc85746d5bc138628d0b70615 : SUCCESS in 1h 06m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f94ee876cb234482ae01227da3dea01e : TIMED_OUT in 1h 33m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5abd19c8cf374d3280639e31ca09fb51 : SUCCESS in 2h 23m 29s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/fcc72c3e553a43b69bf3013b7a7137da : FAILURE in 15m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d4cd715973c4786a4717bc2d381a08f : FAILURE in 1h 32m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ff958b74b9cd4691934194b112b79d65 : SUCCESS in 1h 07m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2466a6bd68e41cbbbcbca969fa70521 : SUCCESS in 46m 17s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5b211ca258f407b91f1f8d77ff57cbc : SUCCESS in 1h 01m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd2895fb1416400a8154146af847ff10 : SUCCESS in 1h 14m 12s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/de21f5bb1f03461c8416a33d89a14be8 : SUCCESS in 1h 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2863555c600cb3c7d3175e56dd99227cbc85e301","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-29 03:59:13.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/96/696296/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"3179e098dedbf6809458c027173f273c9ff63208","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-29 08:39:07.000000000","message":"Patch Set 6:\n\n\u003e This code should go to\n \u003e - agent.ovsdb.ovn.commands\n \u003e - agent.ovsdb.ovn.worker\n \u003e - agent.ovsdb.ovn.api\n\nWhy to agent? There is no agent using it.","accounts_in_message":[],"_revision_number":6},{"id":"766f22a6f561ece79886854337f68a077b174886","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-11-29 08:45:31.000000000","message":"Patch Set 6: Code-Review-1 Workflow-1\n\nRodolfo, I believe we should merge this patchset with https://review.opendev.org/#/c/696621/1 because ovsdb/commands.py tests needs ovsdb/acl.py [1]\n\n\n[1] https://github.com/openstack/networking-ovn/blob/master/networking_ovn/tests/unit/ovsdb/test_commands.py#L19","accounts_in_message":[],"_revision_number":6},{"id":"6cec43cd3097c69010c676ba91ad0d0e0d3b3d2d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-29 10:50:36.000000000","message":"Patch Set 6:\n\n\u003e * i don\u0027t understand why you put these files here instead of\n \u003e neutron/plugins/ml2/drivers/ovn\n \u003e \n \u003e * i don\u0027t understand why you submit very small patches like this.\n \u003e are these small patches testable independently? otherwise i\u0027d\n \u003e suggest to submit the entire networking-ovn code as a single (or\n \u003e maybe a few) patch.\n\nEach patch is going to be testable, that\u0027s why we are pushing small patches. Tests are coming in next PS.\n\nAgree with having this code in neutron/plugins/ml2/drivers/ovn","accounts_in_message":[],"_revision_number":6},{"id":"3bdf02a17c9550c6c8e6d1b3ad0d2fc004af3c3a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-11-29 12:06:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"64cabb34e7b4e41a850d3f067ec67049df349004","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-29 12:38:56.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/96/696296/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"146ae72de39950d16ff4dd45f3b7e5a00fbf25b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-29 13:33:30.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/7/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/7/Neutron-ML2-Sriov : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"b768f89b9af8edf54e2a76edac510a4921de2c95","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-11-29 14:27:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/7 : SUCCESS in 2h 01m 52s","accounts_in_message":[],"_revision_number":7},{"id":"8edaa1abc0cf4c37f6a1257a4acde99f7e2f7f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 14:52:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/af119d71fec8402aa7b6b207da19e51b : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bcd9ad9643e44ad297367ffa86f864b8 : SUCCESS in 33m 49s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0ced2e88c3464bec92eca2a20e0362c7 : FAILURE in 1h 42m 02s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/edbb3247af674f0fa7b1592e98bc465e : SUCCESS in 1h 08m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5debdb0b2353410297f1fe0bab09197d : SUCCESS in 1h 11m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/893588ddff1c4e6a906c39a955e47127 : SUCCESS in 59m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51293ccd6a1d4efdac51d8e751ec1c44 : SUCCESS in 32m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67a2230dc7cd48cfb50bed826f750eb8 : SUCCESS in 23m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248c08cec0c141b78191a8add80b7b49 : FAILURE in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34cd55c76f314d8691b173ed33459ba7 : SUCCESS in 22m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc9e8fe51b854cb48e4f27999a3a711d : SUCCESS in 30m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dc05906aa749e58ab5adc50683fbe5 : SUCCESS in 9m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c6b48cf0deab45bab517e415a1657174 : SUCCESS in 1h 00m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2f2658509ee431297e84c648ef7dd2e : SUCCESS in 1h 16m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/57208b1591f94fc1a480a44794d0e130 : SUCCESS in 45m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3fb067811a534789a7a4eecaad37756c : SUCCESS in 1h 01m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e94ea669cc014add8d6e1fa55f9edcae : SUCCESS in 1h 57m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/517299fae16d4851937ca6bc9aa133a4 : SUCCESS in 1h 09m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/989db9f5e4614fdfa1b22ab16d663793 : SUCCESS in 1h 19m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0c24c1d19ae4278a17507abb5f7bdb5 : SUCCESS in 1h 14m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/80a30ceffd864eafafb81ea211d1efe1 : SUCCESS in 1h 20m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b6a1341c3f045a9b3c90442ff5d6398 : SUCCESS in 1h 42m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e62afc026a1944c992d609cd14ee13f6 : FAILURE in 1h 37m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33f4d0bdd8d44e5c8c477e70813f34ad : SUCCESS in 1h 14m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4434da6d57344728bd10f9fa0826c6fb : SUCCESS in 55m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4072ab454d1f4188a569ee9a49d20b09 : SUCCESS in 2h 38m 10s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/0a6aa32136e742cd97fdc457af201cba : FAILURE in 17m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff09ec3bc9294468b48b0b5ea1acb18e : SUCCESS in 1h 05m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/15dbf318ac0e43a488364bdb68c7ad92 : SUCCESS in 1h 08m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbfdaa96559a42528444dba17a9555f0 : SUCCESS in 48m 35s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccbeece9ee3b43db928e239e23e6a334 : SUCCESS in 1h 05m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/597c4c1e7e1d48b0ba230c82089d4ed7 : SUCCESS in 1h 21m 23s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/28204880fbb84ac8bc85a5d75c21dd3e : FAILURE in 2h 05m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3a36134c1a664d8eb9ac33317a8d6614 : FAILURE in 31m 34s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/25b5fddf70074c709567d3f3a553c29f : FAILURE in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fff323c925b244544bb915e76dfded24d05f26d2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-29 15:37:13.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"660abdff9ff69c47d3566e7b26924c9e177e33dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-29 15:42:41.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"1e0aa4fe64860c228ebc921eb67d5821077b04ea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-11-30 04:37:15.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/November-2019/Arista_Tempest_Testrefs/changes/96/696296/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"c9035a458cbdb134a2bc039fdf8605f2f0186f5a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-03 12:18:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"57b7c4c9954325ca31a8a67760eb06b2d6670949","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-03 12:20:03.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4135993b42a1529b56e4409e2a55962412dbacdd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-03 12:28:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b5a442204cb93da922adbaa0bf6066cae811a6fe","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-03 14:19:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/9 : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":9},{"id":"e7cffa36f8859eb4b62748bb19fa73d0feb22d9a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-03 14:45:50.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/9/Neutron-MACVTAP-ML2-Sriov : ABORTED\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/9/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"15f7db97356dd4ac88f0dc2eb5b866398cb5e7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:09:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1ad6f87b028f4c47bbf105d5c5730088 : SUCCESS in 57m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e8a60113b44a4c268616b41f0dc960a3 : SUCCESS in 34m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/295dd70e6df44fa884b3e9e0b6caa1b0 : FAILURE in 1h 32m 34s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e556e2390ac44cc0b64a9d4368c01e3b : SUCCESS in 1h 18m 18s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/50a905a04dcc4743bdc045811a3d775f : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/338196ab7941478dad9ecb21e6b1006b : SUCCESS in 1h 09m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b1c3317562a46e7a75a513c728392b6 : SUCCESS in 26m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c5dab29272e40e983cdeb368419b1ee : SUCCESS in 25m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b858950051d496fa5729cdd32504472 : SUCCESS in 12m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0377c6e9377c4cd39bac41116715b2cc : SUCCESS in 21m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2104bf773ffe4ee3a674916e9bd09cc7 : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a0875655c34f6db925fca6d300ec92 : SUCCESS in 7m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a615a1dc731e4ee3bdf606091b25e371 : SUCCESS in 1h 00m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dacb321ff7d14e3a9dcb9065a78507f2 : SUCCESS in 1h 18m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c5dee760ff464978a90e24fbb277072d : SUCCESS in 40m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ac08e0d205874ab49e21c75254b727fc : SUCCESS in 1h 05m 12s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/7321249db68a49c1a608f526a06fafb9 : SUCCESS in 2h 09m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/862e3c7c45454bef93816fe1348911a3 : SUCCESS in 1h 13m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7cd4132c0bec4c809de11ff5b4a7f0d7 : SUCCESS in 1h 15m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34cb7ebf850f4c1dae4cdf51cf06803c : SUCCESS in 1h 10m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b41764053fe4bac9af58c838f833d59 : SUCCESS in 1h 24m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/727e496feae84b74a09d905b21b6583f : SUCCESS in 1h 40m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/45d5d1a0144949879561f7544ed89143 : SUCCESS in 2h 05m 11s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d90695b377f74ce78963375a4dd21f59 : SUCCESS in 1h 13m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/848d60fa229c4cb184cb99e56403f270 : SUCCESS in 54m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d49b106ea56f415ea0c988498b98a675 : SUCCESS in 2h 30m 03s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/5c42d6176a7548ef89d881b72725b5a4 : FAILURE in 12m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d2f04e26cee4cc2bcb86e83d1a07044 : SUCCESS in 1h 05m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f4ad9386db4d4e62803c62858733e048 : SUCCESS in 1h 25m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f21d34080a15446c8714af4bd97f302a : SUCCESS in 50m 16s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/917525e33e654dd9ab3dff4d2a41ca56 : SUCCESS in 1h 12m 28s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/400b1a237d4f4355a79ffd857a6c5d24 : SUCCESS in 1h 23m 48s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e335b64f7d48406689214064ea5bd3de : FAILURE in 1h 55m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/de9ae4c4b32541f081abf91e49632808 : FAILURE in 36m 45s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/48c63ce6816c4f43a6ff3f7760efc2d3 : FAILURE in 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"991e671ba5ff446791b7fe5026ca9a095f1b0b95","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-03 19:50:39.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/96/696296/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"a0eb1cc73b8c7f14d59677d767cf6fe411ac298d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-03 22:41:56.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/96/696296/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"bf5a02c4d46aa5f4af789dcc4cf129b549bac311","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 11:19:07.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"266c58110777baec282ea6d77bb84dd6f1f5f314","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-04 11:35:58.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)\n\nFew nits but, LGTM!","accounts_in_message":[],"_revision_number":9},{"id":"ef75a2057b840f11f7dcb8898bf6d3d664d6c0b3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 13:18:50.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"55767a6dbfc805dacefcc1a1f9cafe8ab1e2e39d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 13:18:57.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5878c5871dea1d177236b2f9625f64c1a1c28cfd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-04 14:09:26.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/10/Neutron-MACVTAP-ML2-Sriov : ABORTED\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/96/696296/10/Neutron-ML2-Sriov : ABORTED","accounts_in_message":[],"_revision_number":10},{"id":"8b7f399dad0cf7674904bf947e1079130bb5ad17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 16:07:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b64ca8f37d03468cb79070cf5a925703 : SUCCESS in 1h 05m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/213073bba5944d6f8faed3b6f6197407 : SUCCESS in 36m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b74e7012ab5646a7a70ebf53b6a6322a : FAILURE in 1h 56m 59s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2ec9215391f8444dbe7493e9d3a87dd9 : SUCCESS in 1h 12m 48s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/591759b428ff475a80630a379ab03654 : SUCCESS in 1h 14m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81119b745cf4493ca479d715d30bbf78 : SUCCESS in 1h 18m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8627510b61514c66ba7a736f12d43fe3 : SUCCESS in 45m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3815d633d1ff43adaf390070237fd1a7 : SUCCESS in 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81970dd9425447cbead25c66a40d71a : SUCCESS in 10m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1912605905347ef9c78112e201c1a7e : SUCCESS in 28m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/488c1bc8b5a741d1bc5133d9e8c94fd8 : SUCCESS in 27m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df00870b79241efae957dfe2d2a7b2e : SUCCESS in 7m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2d097d997ff84c2fad377cf5f2e8a9a7 : SUCCESS in 52m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e2f93eec3a4d41afa51d2be4c3f04102 : SUCCESS in 1h 05m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d2caae5f374047b09b91dc556e73d921 : SUCCESS in 42m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/43e850bdc15d417e8b03a68ca5c3a4ca : SUCCESS in 1h 05m 26s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/73aadd3ca9d941f689d4d3ce390733bd : SUCCESS in 1h 44m 05s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/573b68ffec824755a2743bd78d07f46c : SUCCESS in 1h 11m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/777e0ffd66fd49b988e49ef67c78cb87 : SUCCESS in 1h 17m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be53189908f94f3e9c1f584c4670979c : SUCCESS in 1h 07m 30s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b46ebd2a69754c85b8d65532bc90c03f : SUCCESS in 1h 07m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0e1f3f365bcb4b499428f6961ba4736a : SUCCESS in 1h 37m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a61c6e552e2548719359123f20934412 : SUCCESS in 1h 50m 23s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b9aac3033bb4b599f743d82bffff5bd : SUCCESS in 1h 15m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e76bef02c1b246e8b27c92e8f59934fb : SUCCESS in 1h 11m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f99b09dfde24edc8fdf48d05765245a : SUCCESS in 2h 37m 37s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/8833b3f1643440839794aac34696f38a : FAILURE in 18m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbb925d2c5304050a13343066dedb83a : SUCCESS in 1h 10m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a449e1955a1240e4a93b4b7f0d7d28a7 : SUCCESS in 1h 25m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/70b69e291cc443fdb50c517e8c6ff0c8 : SUCCESS in 47m 58s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c59d7eceeae4432e99651b410a228cbf : SUCCESS in 1h 07m 38s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/97106cb2d3404df7b7dbdf83fc80cc70 : SUCCESS in 1h 24m 12s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d62f09b80eda4f4395a2baa7ce677475 : SUCCESS in 1h 56m 01s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4e1f146b47e84ae195f8610bfbdd2c5a : FAILURE in 38m 56s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/eada30cc3dac41959c9be63800266399 : FAILURE in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c133ddb04b92fbb9729c2420af5c7b25d1d5de85","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-04 16:58:53.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":10},{"id":"b9b6503041952a8828c1350c2dcd91a3c270c7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:58:11.000000000","message":"Patch Set 10: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"cfb2620b97c0064664caf6d784a426e8cb0fd64f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-05 07:35:23.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/96/696296/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c2093eb59dd454a40300c1c94c911dabac283526","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-05 22:16:33.000000000","message":"Patch Set 10: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"834ef81874ae34df99d973e28f417e7ad44c3c62","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-06 00:12:40.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : SUCCESS in 1h 55m 22s","accounts_in_message":[],"_revision_number":10},{"id":"82c59e7d714c78f709b82aae21963fd7f1a59609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:27:10.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d90fcc40cc8041148eca47608725086b : SUCCESS in 56m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ddbe452894fc4ec3b34e2df5a515c194 : SUCCESS in 41m 19s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/fd1f5ac13c3240b29d317088f6d7b91e : FAILURE in 1h 59m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/97e6f47dd62646909ed9bbef7b5dd0c2 : SUCCESS in 1h 09m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/61887899640149309ab1de1fb3a2f0a0 : SUCCESS in 1h 20m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93eed038b7bc4deab549bd45bc0f45bc : SUCCESS in 1h 04m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e8b170a203d4752a261f353a4afed16 : TIMED_OUT in 51m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a137fb641854290915504b77efa1bb6 : TIMED_OUT in 41m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ca9bdc30fb4e5c82fa8a359a0d49c2 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e7bb31cc4b345cda369cb4fc3b691f8 : TIMED_OUT in 41m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26705103b5d14801b6e43c3f058b9bce : SUCCESS in 22m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df87f88d935743dd922353c3657d231c : SUCCESS in 12m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b55f5ebd7eda48839c244f0354b34f62 : SUCCESS in 53m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3576e11ab5814636966886b46f0f76c6 : SUCCESS in 1h 23m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/d7cc9a7ac4c94fbcace2739a5779b986 : SUCCESS in 2h 02m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7b898494777e491087f46b815e21a0c2 : SUCCESS in 1h 57m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3ed59bb5ee2a43a8b247ed1d0e6e53c5 : SUCCESS in 1h 10m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89f05ae906d8430284d0f027f0cdad6b : SUCCESS in 1h 17m 07s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7629b9d0d8e14293ab10d58bda2a277e : SUCCESS in 1h 23m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/207cf511abbf4603bc3a57da5eaf6af9 : SUCCESS in 1h 17m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bdd09d7890824d998d8bf35cc1303efb : SUCCESS in 1h 42m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d62d7eeb0d7d4008af2f203ab1d08f11 : SUCCESS in 1h 48m 20s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82b02d88021d4efdbd3ada82e457e2b9 : SUCCESS in 1h 30m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4919ef6044db40c0bcba223dd9c7ed26 : SUCCESS in 59m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f2f0d7e8997445ba77fb000ee4ca34b : SUCCESS in 2h 43m 19s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c0c6fe78b28442f480991f41a3c53212 : FAILURE in 15m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e96eb0a318d46e9bd2542b370918ce4 : SUCCESS in 1h 14m 03s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7286c4a05247457889d13e116b36f75d : SUCCESS in 1h 19m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2df6a92a4d564d70bf43eb6d29b9c8a9 : SUCCESS in 47m 05s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f64299fb84d0487b9d42b03bde6044fb : SUCCESS in 1h 00m 36s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/69c4dc149ec047bf8789f593951fd6ce : SUCCESS in 1h 28m 10s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/010ac410362f4033856362920079e7c4 : FAILURE in 2h 11m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d3bac1286e4144099e38b536398b72f1 : TIMED_OUT in 3h 04m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f9faf6c883634bbe91cc617d21b6c147 : FAILURE in 2h 19m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d1168f343d58cf453e8ef7f943cdac2191cf80af","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-12-06 08:44:15.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"f28f29f79a78bd1ce3154e9d036aec961d2b5177","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-06 09:00:25.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"b68174171e66d66fac6c858fda8e02f95d8050fa","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-06 10:47:00.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : FAILURE in 2h 02m 12s","accounts_in_message":[],"_revision_number":10},{"id":"7406ffeb763d503212ba615dfff7de2c28e9087f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 11:17:20.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/02271534370b4abfbd95a39921f62427 : SUCCESS in 53m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90c76fd492cd485eabab6aa7a120c515 : SUCCESS in 56m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e44971f85cbc45a7a1be3898e29ab471 : FAILURE in 2h 15m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/cab78b494a31426eb9681345c64c3359 : SUCCESS in 1h 06m 19s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e2aca64a62441e0a586f9b9b5765491 : SUCCESS in 1h 06m 31s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29465c48b8e94b00ad0c17f2443c42e1 : SUCCESS in 1h 04m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa32ae6999b4979bf6a43eb23323050 : FAILURE in 26m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91f2cd0f672646d3a64e4b1653cea4c4 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5aaf8ca6b05417087c15405a80dd69b : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02d0160f84284e399dcdf7afd11b3249 : SUCCESS in 19m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dcf3d2003b14a21b4d3b27f5aa7f00b : SUCCESS in 23m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280d780dbc054caa94c9ef78d60f17a7 : SUCCESS in 9m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f9edbc8b12ae4647bf2ca1c7f8c4c0d0 : SUCCESS in 1h 00m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fb504feed60241c1afde940a08a5ca2f : SUCCESS in 1h 04m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4b71c27ca5cb441dac72bd7058d66e41 : SUCCESS in 1h 38m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/eb692fd722044ca9bfe08d2430242c92 : SUCCESS in 1h 41m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1975dbe2987c494585a9c24748770a50 : SUCCESS in 1h 17m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0018f740179b4928b6fb37d26626255c : SUCCESS in 1h 11m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0abfe75281f4d678c7eff0ca2f45d75 : SUCCESS in 1h 25m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/82f9cde15324429dac1511e90812d5b2 : SUCCESS in 1h 55m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/366768bcffc14ccb964cbbb3a7b82db8 : SUCCESS in 1h 35m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/04ee32daff71460ca02807b42ba18a0a : FAILURE in 1h 21m 10s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/157846986f3041fbba26f2751215fd99 : SUCCESS in 1h 09m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9a84f68580ac4634ac47e746d3d6e2d0 : SUCCESS in 52m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e395d7d626fd4859a56469effb10d02c : SUCCESS in 2h 18m 47s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c20baa1f633d4aae978499dc4a4a8dcc : FAILURE in 11m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d54e2a41075e4e16a4768f8d3c249c88 : SUCCESS in 57m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f8e803ee12d64b40be69fb4086881872 : SUCCESS in 1h 07m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c662ad0192e431297024262e6bc8df7 : SUCCESS in 41m 05s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d9652da70ca4599a0ba614a28cb0eb6 : SUCCESS in 1h 00m 47s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/380122a310f442ce98e0f616ea54b684 : SUCCESS in 1h 11m 34s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4ddee070380f47e3bc53d490f4902f77 : FAILURE in 1h 56m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/45e0efd81f344543a90d1a11ee70289a : FAILURE in 2h 25m 58s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/810b7c607d1449deb6ed968109a779b5 : FAILURE in 1h 50m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0ec98f41bb2ceac81f97dee5e424a64255fce5b9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-12-06 14:12:55.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"495fa15725f5247d7135a2b8e0212aaf543d3d39","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-06 15:57:40.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":10},{"id":"bf5c74f2dd4a682a70c9a160894a3d84604a256c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:04:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c3507e00881b48b3b67ee22cd80365fb : SUCCESS in 1h 04m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b16571738e6f49da968803cca0ff64e4 : SUCCESS in 43m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/f115f398566a49649bffa581ddaf1abf : FAILURE in 2h 10m 09s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7e156394a1614aa9ae6dc9ef87c9edb5 : SUCCESS in 57m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/80667260339b4892b02f8fa9ce9fb0f0 : SUCCESS in 1h 04m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/845c93f01108419e988276507cb79f63 : SUCCESS in 56m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c878f3e91ab4e5492baea3209358085 : SUCCESS in 30m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b4e71d4d55b4164848555ed05dbe18c : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0808a1167f049c982851d60e57eed87 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3aac0bbb08a406aaaa903dd49d0b688 : SUCCESS in 21m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f33601fb25b400cbf1151e223d34eb6 : SUCCESS in 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e9e5fbcff14cb7ba3b003a2970c025 : SUCCESS in 8m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d93b681a2df44c1aa9a6678b53417e7f : SUCCESS in 1h 06m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/28e770aa8a0a45a0bccfb4e413f14a28 : SUCCESS in 1h 29m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9c8b4c1ede724d2ab9813054f7a36513 : SUCCESS in 2h 04m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/579beb04a9604a85946679eda957c4db : SUCCESS in 1h 38m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e61d8931c6474c70bec77b1c2cb9ee19 : SUCCESS in 1h 27m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a89671a45f9d431f88109682547c58d9 : SUCCESS in 1h 12m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd0cf54f4e4c463c864f6dd153feeed1 : SUCCESS in 1h 19m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/681940aa23384619bf6986f547122d13 : SUCCESS in 1h 23m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/904a3e82e75e43968ba84269acd3b63f : SUCCESS in 2h 06m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/898803f7b5d041279e8263de19f7b216 : SUCCESS in 2h 23m 25s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/12acefd5cbc44a36b14c4dab347e6278 : SUCCESS in 1h 32m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6826900df1c94e1aa21227cfc67f23ab : SUCCESS in 1h 00m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46b6d47b1e114062ade1a6ec08998bcc : SUCCESS in 2h 18m 52s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1a6d57bdc4484d519ac95d0abbe0fc3f : FAILURE in 17m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6592b11da6a14ef783975f1e4d6b6729 : SUCCESS in 1h 09m 26s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4c8913ade62e4330b10aab4f4b14c795 : SUCCESS in 1h 13m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7daa320ed70f4b7e9d43fb5d45a5bb95 : SUCCESS in 45m 35s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6d619c8204d4d5e86b8afb8082941cf : SUCCESS in 1h 02m 53s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/74c4208ddc0343a38d8e81870cafdbc2 : SUCCESS in 1h 19m 26s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5824c3ff1fd1488c9850cd846dc91531 : FAILURE in 1h 36m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4c38ce63442f4db68c685faabb97bfd7 : FAILURE in 2h 45m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/95ea50245ed64c59b57351bdd17dab59 : FAILURE in 2h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6980127fe2e177591c8e3f85343e66578c7b06eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:45:48.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"c435bbd3372333801a546ffbcb8eb9322953fb34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 00:31:05.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0b7f0839643a458bb2e80f69c451f19f : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0d764961ff354ac2a919f3fa8ec1a88c : SUCCESS in 1h 00m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68dd1d82a49947029699c54d64875f9e : SUCCESS in 1h 06m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0b314e1bd7424be2b6c3ff98cc150839 : SUCCESS in 1h 05m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6492f1b6ed4b404aab01caedba944e57 : SUCCESS in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4019acc9e4284743912e9f9a14c2570b : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7885eca8599476da7421595532e7e7c : SUCCESS in 21m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0182306c628a4d54a22c29b570df4aa2 : SUCCESS in 21m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba18ed1bc6694ce8ad6ef9a8e7ed9ca1 : SUCCESS in 8m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f132924ef3e147f0877ee9d12218edbc : FAILURE in 53m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4c3aac96ae2e4066ba024fef9e715e0c : SUCCESS in 1h 00m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34bd6d6512be491ab2a6f3aca8e44317 : SUCCESS in 1h 06m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/998b64265927407b9a3164423948a618 : SUCCESS in 1h 10m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b5cabff2bfa74e2d83f0fe79cd15917c : SUCCESS in 1h 08m 22s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29e2f034f98a4293a6964cced3a2e0ad : SUCCESS in 1h 07m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d439c8071a43497e871ced35fe19dbe0 : SUCCESS in 1h 04m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/333fc9eb1fb5419ba8fa4d241b6decfa : SUCCESS in 1h 04m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f56fde2604249f0829868b47076b4f2 : SUCCESS in 2h 35m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a8c1f962ca1412dae9b791298e985e6 : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":10},{"id":"3e76459b2a2cef0e0788bf2247e9985efbe8cb99","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-07 09:45:38.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"67a868ccd112fedb58ae37eed875f6ecab4a3bff","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-07 10:29:44.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : FAILURE in 43m 39s","accounts_in_message":[],"_revision_number":10},{"id":"97343fa873220139bd35729feda6c437d27cbaa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 12:40:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/edcc69ea376c479abb507839e416395a : SUCCESS in 50m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f19ecd4af8394039ab8e901541ef2df0 : SUCCESS in 32m 51s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3f0024aede104d58b83b3903f9b6895f : FAILURE in 1h 26m 58s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/98a469b05ed44943a11b117363181f16 : SUCCESS in 54m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/fb098dc3e68c4f128627548c9563847c : SUCCESS in 57m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/608e21441af843b48c66873e5b381bdf : SUCCESS in 51m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b36f8b34160d4ebf93a66e3b619a2e3e : SUCCESS in 24m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fba4dc4515d74f12a4ddb0abe0394543 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8947098f381845ccba70092feabf89f1 : SUCCESS in 10m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44fc46afac8c45b9a9e375076ee9bb96 : SUCCESS in 18m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c69b6a263a044293899297c859ba6a26 : SUCCESS in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52a77c4e3564627b7fc671245875f17 : SUCCESS in 8m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f56a44db62a543e98d8f033f302e768f : SUCCESS in 45m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f2a4724fc66a494795baa3b2e8c8900d : SUCCESS in 1h 00m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a09cd4851f324eb1927925f66f43163d : SUCCESS in 1h 45m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/18876d33a2674a958e23ecef9c3d9249 : SUCCESS in 1h 07m 27s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5e2c8e95984b4337a38754d7ed452d14 : SUCCESS in 1h 09m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5af0bc7bf124b998f8ef26551a14557 : SUCCESS in 1h 02m 35s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a5294908fcd749adb6a376bcfcca5182 : SUCCESS in 1h 05m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7da25ab27f884103b36f3f78250dac25 : SUCCESS in 59m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8e4a26c7072b44c78c02467cf8928248 : SUCCESS in 1h 38m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bcb95e5b8ad34faaa490db9a546426a3 : SUCCESS in 1h 46m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30fb618e59bc4ba28fd7838654010bfa : SUCCESS in 1h 12m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/697c8011805a4dcb94467e6114f9dc3f : FAILURE in 59m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac07fd05744e48ddb4dd5c737e29f396 : SUCCESS in 2h 24m 13s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/71b4663e6f3f48f69534ac6591447757 : FAILURE in 14m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5c163f07816474ca2dd921f41537cad : SUCCESS in 59m 01s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/818f9d928d53445e9ae51d36386633b7 : SUCCESS in 1h 03m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7398f6cb6ea44a828020dce733004487 : SUCCESS in 44m 09s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/926e28a0a9de41a59e0a4436959019cb : SUCCESS in 58m 06s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/ace91ab9594145829f07e740882cccd1 : SUCCESS in 1h 14m 21s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c91ca816f20545298155e056b27b93ea : FAILURE in 2h 18m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/392f3c5d4c914d70b0ff927b942d742e : FAILURE in 2h 48m 19s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/63a15221e10f49b798efd06474ed3861 : FAILURE in 1h 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a5986863a59644d67b67f6cf4b9bef47a8d880e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 12:40:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"af28772a5c6a988042491717347a584333283891","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 16:49:40.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8c7ab3da4099484eac02241bbccb2c4c : SUCCESS in 51m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1fee6118ef814f1cb1224cbd53eefa81 : SUCCESS in 1h 00m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b7cb2adc86284274b7e56df675eb19bc : SUCCESS in 51m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1504dc85ec7640deb864d2dd9f2476a4 : SUCCESS in 57m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71eab03df5c3407cac45aaffb2fe47a7 : SUCCESS in 21m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da3cff094d9d4709b4712d276cbf555c : SUCCESS in 12m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6917c6cb60a64dc49682a8a3d529b985 : SUCCESS in 21m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d110764bae2468a80d70db9c4ea0f66 : SUCCESS in 25m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8587dce4334799877905272d472cec : SUCCESS in 7m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07e299cfbd1f4d6da0972a5c0ef90f07 : SUCCESS in 46m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1b8d2e547c18494fafc8dab25394861e : TIMED_OUT in 2h 15m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b172ef265924419b9e1a1a9fca5e1f3c : SUCCESS in 1h 06m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a051b6497df543b4a2633373e5661dc5 : SUCCESS in 1h 38m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/824d4e34b7c949c09a41bee6d10c8c49 : FAILURE in 1h 25m 23s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78fe8985df8946b08415cead9d4fbf1b : SUCCESS in 1h 37m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/842975e9a8af47ff97473d4cd884d78c : SUCCESS in 1h 03m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b35ecaa2c3d41d29612a575c2b55444 : SUCCESS in 1h 09m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/989adfab8c074f509c95ffa26f2e2b55 : SUCCESS in 2h 42m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd70ac8ab6c34c65907d1cdd0aa0a4b8 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":10},{"id":"de62b5be3b78244aceaf629d7870ca8b302edda6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-07 17:11:16.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"36605346686db601c1f7000dad54a2b836458022","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-07 18:58:45.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : FAILURE in 1h 47m 00s","accounts_in_message":[],"_revision_number":10},{"id":"f122cfb930e9daa71e50b12e56f80c735f9f6d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 20:23:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/140d0be86a1e488e84b91400e04d7010 : SUCCESS in 1h 13m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e79a2ee1668a445ca337025d18ee5512 : SUCCESS in 32m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/55060d0df7ed4724ab6ddcc43575fc4f : FAILURE in 1h 42m 14s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dd721e75c68342e288d768883a95f83c : SUCCESS in 57m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6121bcef2faf44559283fad89c526046 : SUCCESS in 52m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/361c514c2ad1477f84fb9fed8a1f10b1 : SUCCESS in 59m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5f354bdb6b8445894a95ee5263b3756 : SUCCESS in 25m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce1cda068bf445f4aedf6eb94534a9d7 : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd511fd1efe4ee280551b9d2f223de5 : SUCCESS in 11m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/176fcdfd020247ec8b24701e829aee6a : SUCCESS in 21m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/964e3837c966481ca2f865c8e8ac06fe : SUCCESS in 21m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226819433e6a46a49393938aa3d2fe8d : SUCCESS in 8m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ecaf17c612a44935a4d6155f4716727b : TIMED_OUT in 2h 17m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7bd068ca241642ca9944d1ded4615559 : SUCCESS in 1h 23m 00s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4d0bd980947e454bafb0cf8558a46c84 : SUCCESS in 2h 36m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ce7d6aef74cd43a8b905f043197d1788 : SUCCESS in 1h 52m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/0a42ec1549b649e88dfcb6c25c4eb96b : SUCCESS in 1h 41m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa8a92f3f0294495811595c0793d3edf : FAILURE in 41m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9367c0f84254f3f91cd109240a2a951 : SUCCESS in 1h 49m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a99ed3bd401742558cd6ca47ac4f1af5 : SUCCESS in 1h 16m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/36c0d0807f8d4a1f800cf77b791c420f : SUCCESS in 2h 02m 55s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/394fe048096c480a9ac81314fe2c1293 : SUCCESS in 1h 57m 30s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/efe1cf0210144c64818cd8ff106610eb : SUCCESS in 1h 35m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b77f2be531a94c4495c60bb869354725 : FAILURE in 1h 12m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/80cd10a01a7644968c431e3519a15e3e : SUCCESS in 2h 54m 15s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2a99677f5c5443b68ec6840b31f3bdca : FAILURE in 16m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88c492d5c5af491fb8cd1888e52ac92e : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0cefc99758c042e9bf6f00c7aa211a13 : SUCCESS in 1h 17m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20e0927cd1894f6d8b987ae679f79650 : SUCCESS in 50m 06s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bbe2f863f69d428cad0d561d61e4929e : SUCCESS in 1h 01m 48s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/9990420f7b7647e2b31b229a164dd630 : SUCCESS in 1h 51m 31s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d0d932e02b3142af848fc470656efb8f : FAILURE in 1h 58m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b7c93e3aa2cc49e5b67f216dc82f1a1d : FAILURE in 3h 04m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4604ec42f1fd46ddb222df12406018c0 : FAILURE in 2h 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ea4adf39eeb78eb602520a75a41cbbfb2c3f8bd7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-07 22:01:36.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"17478b5d00f111e70f6a068e81314a1f1c9d4188","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-07 23:43:49.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/696296/10 : SUCCESS in 1h 41m 46s","accounts_in_message":[],"_revision_number":10},{"id":"c492ccbbbf04e1821a18014e2704cdcf010e0c8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 00:33:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/30d76cf714eb4b20817f7909b975dee9 : SUCCESS in 52m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90bbb5b60918411e9108b77581071130 : SUCCESS in 33m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/969c2fceeed845679e154da9df34bee3 : FAILURE in 1h 37m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/deac0912f41a40e481701ca8a97114e4 : SUCCESS in 1h 11m 33s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6e0d78d2e5af4eaebfb0b60395340cdf : SUCCESS in 56m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d3e0f126a034ea4b8563abf980ebfb7 : SUCCESS in 1h 01m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ff17762ca9140cc9a55da1ad7dddf7b : SUCCESS in 26m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bea8453c5cc4c1eb7894bff014f3987 : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c65681ef1db6496a8c11a494a6a660b0 : SUCCESS in 12m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ac8801ec934588894cad28ca2dd8ed : SUCCESS in 21m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4aab09e15732449990960396d15c4e7c : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca16272e804b4a4b80f8745c21d47a80 : SUCCESS in 7m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/52e979c53605484882af09a93b5d1828 : SUCCESS in 47m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7d1b5a6c46924b7a84031e1e81033e4e : SUCCESS in 1h 00m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3ec2fab134e24425bf15ffd78b7e0023 : SUCCESS in 2h 04m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5517bd60f0e4192b12388ed883069fc : SUCCESS in 1h 23m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a4c9e6cb31d3414a91333285004a467c : SUCCESS in 1h 16m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a808da99bb65492c994abfad4b77f259 : SUCCESS in 1h 08m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b9b1c4a3692948f9ad863fd3c9c044a7 : SUCCESS in 1h 13m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee8df93fa6654fabb11027a35bb30930 : SUCCESS in 1h 24m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f42eed273b90431a850ceac7501d74db : SUCCESS in 1h 44m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/836c1bc71d474829b7586385129c16cb : SUCCESS in 2h 04m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/256c9d339c4b412087ae3faa0a9fbb17 : SUCCESS in 1h 10m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2bd4ed52384b494083ae63c9a0fbc5c9 : FAILURE in 1h 01m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e70a3d9b81bb418eacd5072907dc5f4f : SUCCESS in 2h 25m 37s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/0d3f1da6b7d9425e95c6688b7a12d9a8 : FAILURE in 14m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8d6c1966b5a4ff2a19e9e64b480ebff : SUCCESS in 1h 15m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f03af6b6e23f4a738bda504b4e38a588 : SUCCESS in 1h 03m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a71c06092fed4ff78ad8279c8b4abe9e : SUCCESS in 41m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af700cebff0140c0bbb784c5857203d8 : SUCCESS in 1h 04m 07s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/87ea27e880f74e4384961cbe5868f7bb : SUCCESS in 1h 15m 07s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2edae6aaf27d4e2e94ddc03501db29dd : FAILURE in 2h 01m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9a9d0edc53e343b99b453dcde9eb4454 : FAILURE in 2h 20m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8203b5634639408984ab4884509ed223 : FAILURE in 1h 43m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8973e8b26055ad27045cc34462b5ea4577a5dc45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 00:33:47.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"5bb7a98224b3caf273b8d9d313d75072a0c235be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 03:00:13.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/759782a34f9a4c5bad82600beccbf309 : SUCCESS in 54m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b7f688b4195d478d80f1f5a2bfd5d35f : SUCCESS in 1h 00m 47s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9ce8a93891f46b397a51da94d9d8f50 : SUCCESS in 1h 01m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58476c6cdac14a1e9d32489f2ff9b160 : SUCCESS in 1h 01m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cff1ed1414b941ea8ae4bfb4ffc8068b : SUCCESS in 34m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f1b4e12aa54ae0b1fe20e19909291d : SUCCESS in 12m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49745d3739ed4e1bb51da3d852e033fe : SUCCESS in 21m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48ebad847c6e4f7fadfb4946b67aaa49 : SUCCESS in 22m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a826447e3d56451cb8a84a84f3a310e1 : SUCCESS in 7m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/814b5d64547c417ab1f1fe6a897b631f : SUCCESS in 1h 02m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/60eb62f9cb3846c4a4d4ca8f2ec08ced : SUCCESS in 1h 09m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/59675b39beec4f8abcf6377885e9d6df : SUCCESS in 1h 12m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3b77f1adc22f4c749387b52499665274 : SUCCESS in 1h 12m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/adf20f6ef10442fcb85ccc40fd0def0a : SUCCESS in 1h 14m 54s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5995a966fa90455a9a5f0395e152ed71 : SUCCESS in 2h 16m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06edfbeacbbc48eb835db6ab9580fef1 : SUCCESS in 1h 02m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a590948386d40d5999aa1c89204cdff : SUCCESS in 1h 10m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39c2bc52e3e04bf49fc6b271a75f3c24 : SUCCESS in 2h 15m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f92e5b7bfd8449b9cff31b3c956feb9 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":10},{"id":"aeead1e67918aecc657d9e032b24a1dbdcdd54dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 03:00:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"1a378b397af1aaaa0e9430f50969eb25bf4b8d0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 03:01:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5e32df038b4312b2535afdcc01bfe3 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4b5cf9e5fbdba90f4218fff9ba558077cfbddb96","revisions":{"3b499f5304c66592102b656092d58e54913c6d41":{"kind":"REWORK","_number":1,"created":"2019-11-27 12:21:55.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/96/696296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/1"}}},"commit":{"parents":[{"commit":"d78ff76ff81b550caa93bd5ab4fb83489caab20a","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d78ff76ff81b550caa93bd5ab4fb83489caab20a"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:10:42.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:21:39.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e ./neutron/ovsdb/ovn/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e ./neutron/ovsdb/ovn/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e ./neutron/ovsdb/ovn/api.py\n\nRelated-Blueprint: neutron-ovn-merge\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b499f5304c66592102b656092d58e54913c6d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b499f5304c66592102b656092d58e54913c6d41"}]},"branch":"refs/heads/master"},"e826e91481ea028bf28ce0013ce6ddc27c38062f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-27 13:05:20.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/96/696296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/2"}}},"commit":{"parents":[{"commit":"d78ff76ff81b550caa93bd5ab4fb83489caab20a","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d78ff76ff81b550caa93bd5ab4fb83489caab20a"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:10:42.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 13:05:17.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e ./neutron/ovsdb/ovn/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e ./neutron/ovsdb/ovn/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e ./neutron/ovsdb/ovn/api.py\n\nTODO(mjozefcz): Those two needs the db code.\n./networking_ovn/ovsdb/ovsdb_monitor.py\n./networking_ovn/ovsdb/impl_idl_ovn.py\n\n\n\nRelated-Blueprint: neutron-ovn-merge\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e826e91481ea028bf28ce0013ce6ddc27c38062f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e826e91481ea028bf28ce0013ce6ddc27c38062f"}]},"branch":"refs/heads/master"},"4661333740a0b32fd79e7830d453b64085a61445":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-28 13:24:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/696296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/3"}}},"commit":{"parents":[{"commit":"866238d1e4c29e411fe77dc4cae4d24812b53230","subject":"[OVN] Add revision number maintenance methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/866238d1e4c29e411fe77dc4cae4d24812b53230"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:10:42.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-28 13:24:23.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e ./neutron/ovsdb/ovn/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e ./neutron/ovsdb/ovn/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e ./neutron/ovsdb/ovn/api.py\n\nTODO(mjozefcz): Those two needs the db code.\n./networking_ovn/ovsdb/ovsdb_monitor.py\n./networking_ovn/ovsdb/impl_idl_ovn.py\n\n\n\nRelated-Blueprint: neutron-ovn-merge\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4661333740a0b32fd79e7830d453b64085a61445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4661333740a0b32fd79e7830d453b64085a61445"}]},"branch":"refs/heads/master"},"2ce03f99fd4314b67ed9eab03ce59265e5b16412":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-11-28 13:51:05.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/96/696296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/4"}}},"commit":{"parents":[{"commit":"d661a58356a4ee5b659bd4700b7050ed523297fe","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d661a58356a4ee5b659bd4700b7050ed523297fe"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:10:42.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-28 13:50:25.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e ./neutron/ovsdb/ovn/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e ./neutron/ovsdb/ovn/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e ./neutron/ovsdb/ovn/api.py\n\nTODO(mjozefcz): Those two needs the db code.\n./networking_ovn/ovsdb/ovsdb_monitor.py\n./networking_ovn/ovsdb/impl_idl_ovn.py\n\n\n\nRelated-Blueprint: neutron-ovn-merge\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ce03f99fd4314b67ed9eab03ce59265e5b16412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ce03f99fd4314b67ed9eab03ce59265e5b16412"}]},"branch":"refs/heads/master"},"987810b784bc3fe88141634fba91a9cfb32e7aa1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-28 15:09:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/696296/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/5"}}},"commit":{"parents":[{"commit":"24d1844a827fe39e03a413dd9f35aca610377cac","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24d1844a827fe39e03a413dd9f35aca610377cac"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:10:42.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-28 15:09:06.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e ./neutron/ovsdb/ovn/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e ./neutron/ovsdb/ovn/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e ./neutron/ovsdb/ovn/api.py\n\nTODO(mjozefcz): Those two needs the db code.\n./networking_ovn/ovsdb/ovsdb_monitor.py\n./networking_ovn/ovsdb/impl_idl_ovn.py\n\n\n\nRelated-Blueprint: neutron-ovn-merge\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/987810b784bc3fe88141634fba91a9cfb32e7aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/987810b784bc3fe88141634fba91a9cfb32e7aa1"}]},"branch":"refs/heads/master"},"4c88b8017af9a90e3bb03f0b3de0fe4a22020a75":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-28 16:35:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/696296/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/6"}}},"commit":{"parents":[{"commit":"d4739fc027e4f403a9294f3d902da533ac38c3b1","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4739fc027e4f403a9294f3d902da533ac38c3b1"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2019-11-27 12:10:42.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-28 16:35:38.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e ./neutron/ovsdb/ovn/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e ./neutron/ovsdb/ovn/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e ./neutron/ovsdb/ovn/api.py\n\nTODO(mjozefcz): Those two needs the db code.\n./networking_ovn/ovsdb/ovsdb_monitor.py\n./networking_ovn/ovsdb/impl_idl_ovn.py\n\n\n\nRelated-Blueprint: neutron-ovn-merge\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c88b8017af9a90e3bb03f0b3de0fe4a22020a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c88b8017af9a90e3bb03f0b3de0fe4a22020a75"}]},"branch":"refs/heads/master"},"a33618a098a4d19a2cb60eb1cae262ff32d14416":{"kind":"REWORK","_number":7,"created":"2019-11-29 12:06:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/696296/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/7"}}},"commit":{"parents":[{"commit":"0d5b38d1577ec26da1d282bb2a3b464151f4302a","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d5b38d1577ec26da1d282bb2a3b464151f4302a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-29 11:11:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-29 12:06:34.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/api.py\n\nACL commands will be implemented when the related code is merged and\ncould be tested properly.\n\nRelated-Blueprint: neutron-ovn-merge\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a33618a098a4d19a2cb60eb1cae262ff32d14416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a33618a098a4d19a2cb60eb1cae262ff32d14416"}]},"branch":"refs/heads/master"},"fc07bbe053a1bac27b136a305a5608c01e9ff6ed":{"kind":"REWORK","_number":8,"created":"2019-12-03 12:18:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/696296/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/8"}}},"commit":{"parents":[{"commit":"0d5b38d1577ec26da1d282bb2a3b464151f4302a","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d5b38d1577ec26da1d282bb2a3b464151f4302a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-29 11:11:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-03 12:18:05.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/api.py\n\nACL commands will be implemented when the related code is merged and\ncould be tested properly.\n\nRelated-Blueprint: neutron-ovn-merge\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc07bbe053a1bac27b136a305a5608c01e9ff6ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc07bbe053a1bac27b136a305a5608c01e9ff6ed"}]},"branch":"refs/heads/master"},"e953bb1fc073fda3f607ebb90c731891b5cdecf7":{"kind":"REWORK","_number":9,"created":"2019-12-03 12:28:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/696296/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/9"}}},"commit":{"parents":[{"commit":"0d5b38d1577ec26da1d282bb2a3b464151f4302a","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d5b38d1577ec26da1d282bb2a3b464151f4302a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-29 11:11:13.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-03 12:21:00.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/api.py\n\nACL commands will be implemented when the related code is merged and\ncould be tested properly.\n\nRelated-Blueprint: neutron-ovn-merge\n\nCo-Authored-By: Reedip \u003crbanerje@redhat.com\u003e\nCo-Authored-By: Numan Siddique \u003cnusiddiq@redhat.com\u003e\nCo-Authored-By: Flavio Fernandes \u003cflaviof@redhat.com\u003e\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\nCo-Authored-By: Changxun Zhou \u003czhoucx@dtdream.com\u003e\nCo-Authored-By: Gal Sagie \u003cgal.sagie@huawei.com\u003e\nCo-Authored-By: Amitabha Biswas \u003cabiswas@us.ibm.com\u003e\nCo-Authored-By: Richard Theis \u003crtheis@us.ibm.com\u003e\nCo-Authored-By: lzklibj \u003clzklibj@cn.ibm.com\u003e\nCo-Authored-By: zhufl \u003czhu.fanglei@zte.com.cn\u003e\nCo-Authored-By: Na \u003cnazhu@cn.ibm.com\u003e\nCo-Authored-By: Chandra S Vejendla \u003ccsvejend@us.ibm.com\u003e\nCo-Authored-By: Gary Kotton \u003cgkotton@vmware.com\u003e\nCo-Authored-By: Aaron Rosen \u003caaronorosen@gmail.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e953bb1fc073fda3f607ebb90c731891b5cdecf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e953bb1fc073fda3f607ebb90c731891b5cdecf7"}]},"branch":"refs/heads/master"},"4b5cf9e5fbdba90f4218fff9ba558077cfbddb96":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-12-04 13:18:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/96/696296/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/696296/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/696296/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/696296/10"}}},"commit":{"parents":[{"commit":"65692127f6e6d6e4980febb2107e867119a6d5c2","subject":"[OVN] Move OVN commons to neutron tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65692127f6e6d6e4980febb2107e867119a6d5c2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-11-29 11:11:13.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-04 13:18:50.000000000","tz":0},"subject":"[OVN] Import ovsdb related code","message":"[OVN] Import ovsdb related code\n\nThis patch imports ovsdb related code from networking_ovn.\n\nPrevious paths in networking-ovn tree:\n./networking_ovn/ovsdb/commands.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/commands.py\n./networking_ovn/ovsdb/worker.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/worker.py\n./networking_ovn/ovsdb/ovn_api.py -\u003e\n  ./neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/api.py\n\nACL commands will be implemented when the related code is merged and\ncould be tested properly.\n\nRelated-Blueprint: neutron-ovn-merge\n\nCo-Authored-By: Reedip \u003crbanerje@redhat.com\u003e\nCo-Authored-By: Numan Siddique \u003cnusiddiq@redhat.com\u003e\nCo-Authored-By: Flavio Fernandes \u003cflaviof@redhat.com\u003e\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\nCo-Authored-By: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\nCo-Authored-By: Changxun Zhou \u003czhoucx@dtdream.com\u003e\nCo-Authored-By: Gal Sagie \u003cgal.sagie@huawei.com\u003e\nCo-Authored-By: Amitabha Biswas \u003cabiswas@us.ibm.com\u003e\nCo-Authored-By: Richard Theis \u003crtheis@us.ibm.com\u003e\nCo-Authored-By: lzklibj \u003clzklibj@cn.ibm.com\u003e\nCo-Authored-By: zhufl \u003czhu.fanglei@zte.com.cn\u003e\nCo-Authored-By: Na \u003cnazhu@cn.ibm.com\u003e\nCo-Authored-By: Chandra S Vejendla \u003ccsvejend@us.ibm.com\u003e\nCo-Authored-By: Gary Kotton \u003cgkotton@vmware.com\u003e\nCo-Authored-By: Aaron Rosen \u003caaronorosen@gmail.com\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nChange-Id: I9fe64f954d227efaab5e96c6150df44f36a2530a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b5cf9e5fbdba90f4218fff9ba558077cfbddb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b5cf9e5fbdba90f4218fff9ba558077cfbddb96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
