)]}'
{"id":"openstack%2Fneutron~697473","triplet_id":"openstack%2Fneutron~master~I94da5d6469111dab743cd3b748e1a2abd1770a60","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I94da5d6469111dab743cd3b748e1a2abd1770a60","subject":"Remove unbound SmartNIC port","status":"ABANDONED","created":"2019-12-05 13:46:41.000000000","updated":"2019-12-11 08:05:34.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ffd615b2ff9b46186fa0a8d3ff5ea3b14cf697b6","_number":697473,"virtual_id_number":697473,"owner":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"date":"2019-12-09 10:29:45.000000000","_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-05 14:35:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-05 13:49:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-08 16:06:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-08 15:10:50.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-08 13:22:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-12-09 17:24:19.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"date":"2019-12-05 18:47:47.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 13:49:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 14:35:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 18:47:47.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2019-12-08 13:22:32.000000000","updated_by":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2019-12-08 15:10:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-08 16:06:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-09 17:24:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"4c86c6fdbaabb1a24ac43982a45c76d3cd7813ec","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-05 13:46:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f2ba52ffd1fee75f486beaa33c12f8028452703","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-05 13:49:33.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/73/697473/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"95f24cf5342b27c4120d1a2b4e192cde47107066","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-05 14:35:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/73/697473/1/Neutron-ML2-Sriov : FAILURE\n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/73/697473/1/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"07057075985a29b6362527e58a945eee8dc6d7c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-05 15:27:20.000000000","message":"Patch Set 1:\n\nHello Hamdy: which patch is the valid one you are going to continue with to close this bug? This one or https://review.opendev.org/#/c/697450/\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"e119e97c48f2200f1308c55d929768718f8aafac","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-05 15:56:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697473/1 : SUCCESS in 1h 52m 44s","accounts_in_message":[],"_revision_number":1},{"id":"37c53b8124157c497512288ead7231ff970f374a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 16:38:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/85e615c18b5744cc91250acfaf16ce73 : SUCCESS in 1h 01m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42043327f9b64ce4aed169ead4d2ec2e : SUCCESS in 31m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2171108a79f64ac5a5c14b1e33fb70f2 : FAILURE in 2h 04m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/93c1b3e58b4e4bf5969ca7655076e204 : SUCCESS in 1h 02m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/83f4ba2428274454a12e94d64f57501c : SUCCESS in 55m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aefb6a71b9284c2bb34e0e16542e702d : SUCCESS in 57m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a15b63de9c224b918bc9041227e4bb39 : SUCCESS in 24m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33127555f65a4c8ea47376de84e235d9 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24147472675c4acb9fe47e4e7c93a5b5 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00ff63ee0d05484889889d3f157ea183 : SUCCESS in 22m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/401de052ff0a4bcf8e4e1fb96452ae02 : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b29747e1eca44278f71a89866a3b4e9 : SUCCESS in 9m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1f60a2e08773408b9bfcd0381c7cb9cd : SUCCESS in 44m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e5f2d0d72c7b4c768029ddfe4b866ce9 : SUCCESS in 1h 02m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/303cc052db2f408f86ffafc1692b9ee4 : SUCCESS in 1h 42m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/47499d103ae34651b14544880dbc1dc5 : SUCCESS in 1h 11m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4f6ac88e6f0044a99a724d4487d10192 : SUCCESS in 1h 28m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e23e57dd9c84b79ac73f09e0484b332 : SUCCESS in 1h 22m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a4b6318766374024ba77ea7e83c218b0 : SUCCESS in 1h 05m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7f7f3823bf7b41f89f1dab613bb4b3fe : SUCCESS in 1h 11m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4921e4740bf54529b8d41ac5a7335f5b : SUCCESS in 1h 36m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5f89b969f53b4ca89a254c0fd10dc7f6 : SUCCESS in 1h 43m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fadb4044da1244a2bf27b29ec1b6925e : SUCCESS in 1h 21m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8f5c6cb6ce9741c1bcf0e75d444d1868 : SUCCESS in 54m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eeb185d13ac04f6d9409698d29440d0f : FAILURE in 2h 21m 02s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2bc35c38c8904d74bb5319feba264890 : FAILURE in 15m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7190de64aa0e468fb4b2a8006780d583 : SUCCESS in 1h 02m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/26d48e949baf474eb585c8c8a1ee24ba : SUCCESS in 1h 13m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cffff9ab67cb489681124d59d0f11e85 : SUCCESS in 45m 30s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/42afa3039e9d4470a997d182a9152687 : SUCCESS in 1h 04m 05s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f90140ff168440b8d8a89eb22416080 : SUCCESS in 1h 14m 19s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9b2bc41fdbb44771a2ee11a4ec19a926 : FAILURE in 2h 11m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1a8cad1b4c5b40a5bbf08d5054add4de : FAILURE in 2h 45m 48s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/09642689d40642c89fd31bd0bcecefd4 : FAILURE in 2h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e18df6b405e278c22eb24fb03acf759b46fcf509","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2019-12-05 18:47:47.000000000","message":"Patch Set 1:\n\n\u003e Hello Hamdy: which patch is the valid one you are going to continue\n \u003e with to close this bug? This one or https://review.opendev.org/#/c/697450/\n \u003e \n \u003e Thanks!\n\nboth of them are valid, this one fix the problem that when we unbind the port with ironic we will remove it from the br-int. the other one is for the sync mechaims so in case we lose port_update for unbind when the agent start we want to able to sync all the ports according to neutron serve db.","accounts_in_message":[],"_revision_number":1},{"id":"13d8e921be8a1a0c12ae232346b275474d3cea21","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-08 13:22:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70f234598cea0735ba5d05ef721bd4047f5eebc0","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2019-12-08 13:22:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nCommit message could use more details.\n\nAn alternative solution to the issue that does not rely on ironic keeping the binding profile of an unbounded port would be: \n\nfor ports in self.updated_smartnic_ports that are unbound, attempt to get the vif name by invoking get_vif_port_by_id() then for every port that still exists in the integration bridge - unplug it.\n\nthe reason for -1 for me is the fact that we rely on ironic behaviour that may not be available.","accounts_in_message":[],"_revision_number":1},{"id":"c945d6f49ca1172026a93afb73faf6ae0c270eb3","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-08 14:29:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"310c23a59c0849d930a4d28446af0a407688049f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-08 15:10:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697473/1 : SUCCESS in 1h 47m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c94ebd5356d8816b574157669a3ea3020297af3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 16:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/63fb3dfa93754734b719cd5d2316f34a : SUCCESS in 48m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2531ac0906bf4a32973233e649d70284 : SUCCESS in 35m 20s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/46790d1cd3074d23b60fd643475eb6a6 : FAILURE in 1h 50m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0db43f63058c4c10b08ed3cab563d200 : SUCCESS in 1h 09m 00s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/30f39b0b761748eb919acc122f0ccd51 : SUCCESS in 56m 45s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/07c478c2038d44329b2d5fd07cb90bb5 : SUCCESS in 55m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb8418f7e1b848d3b044e1c19b6b3865 : SUCCESS in 31m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/615660d681384c2bb4a623ddfe986592 : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae2d4daf79841dca2ee24d6afc7bb7b : SUCCESS in 10m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5329f8101aae42b18e89f0226c9159ea : SUCCESS in 21m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb7fed0c5e20405a9ed6c4323a9ffdb0 : SUCCESS in 23m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ccc301a5194bd09840455294ed09c3 : SUCCESS in 7m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a4dcbe64c21c4b4db6a6c7b056f7fe98 : SUCCESS in 42m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/883d7358d2be4cd1b840fd07ae769286 : SUCCESS in 1h 05m 57s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a7c958e7f671478cb9615ec25c693156 : SUCCESS in 1h 54m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3bcafdb19da54ee0beb9a7197e8d93c5 : SUCCESS in 1h 16m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fc97755557c84dc3a3714af0306b5169 : SUCCESS in 1h 14m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/20280a5cc9d1478f830291ff63981bb7 : SUCCESS in 1h 10m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bdd1a30cb244489e9368bfd6228c93a1 : SUCCESS in 1h 07m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e18a64f7c112489eb97fc03953af417f : SUCCESS in 1h 13m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c3be65b4b6ae4fb5aa8bcc46e4a80dfb : SUCCESS in 1h 47m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f339407801de465fa3810aac9da101c4 : SUCCESS in 1h 45m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/80018ff2fc83472da867e2c1d5bec118 : SUCCESS in 1h 09m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/baeaadd8a31a4e49bace29e092beb4ca : FAILURE in 1h 06m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1655b7bded9347ec8c83a4e2e510521f : SUCCESS in 2h 18m 56s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4574ef48356c4822b4b4baf69fe06ad4 : FAILURE in 15m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/625cff7212ad4046b7d4c484eaf091a7 : SUCCESS in 1h 00m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/576564014f0a49b386072e1aaf73adb8 : SUCCESS in 1h 11m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9870d56eac44060a1f8cb426fe595de : SUCCESS in 42m 10s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a64792aff8f640d1afc153e702afe47c : SUCCESS in 1h 02m 19s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c0269e649aa4ca695e6fe981ab8c4c3 : SUCCESS in 1h 19m 34s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d10feb03d4ef4cd883a009d3fd0da121 : FAILURE in 2h 10m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d23fc08c6211458f818341b349882f28 : FAILURE in 2h 37m 07s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b026110a235b45f3bb7270e88446c2c4 : FAILURE in 2h 08m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45c58c1ed3f0aee41a1c6c076932b450bfd10dac","author":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"date":"2019-12-09 09:37:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f72766edf942a919a011405b8d9168a3a7951be","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-09 10:29:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22f0d5e63695bc8ed33e64c34b94cd3118953f16","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-09 17:24:19.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ffd615b2ff9b46186fa0a8d3ff5ea3b14cf697b6","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-11 08:05:34.000000000","message":"Abandoned\n\nWill review the fix and push again.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c6cee7686fec9d4102c9a7b82f5a274e05682d1","revisions":{"5c6cee7686fec9d4102c9a7b82f5a274e05682d1":{"kind":"REWORK","_number":1,"created":"2019-12-05 13:46:41.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/73/697473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/697473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/697473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/697473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/697473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/697473/1"}}},"commit":{"parents":[{"commit":"7245502abf8564285b7be3ee33ee20031bdfa5b6","subject":"Sync deleted SmartNIC ports on agent restart.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7245502abf8564285b7be3ee33ee20031bdfa5b6"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-12-05 13:48:51.000000000","tz":120},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-12-05 13:48:51.000000000","tz":120},"subject":"Remove unbound SmartNIC port","message":"Remove unbound SmartNIC port\n\nChange-Id: I94da5d6469111dab743cd3b748e1a2abd1770a60\nCloses-Bug: #1855260\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c6cee7686fec9d4102c9a7b82f5a274e05682d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c6cee7686fec9d4102c9a7b82f5a274e05682d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
