)]}'
{"id":"openstack%2Fneutron~697872","triplet_id":"openstack%2Fneutron~master~I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1","project":"openstack/neutron","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-12-08 07:27:16.000000000","updated":"2019-12-22 00:32:18.000000000","submitted":"2019-12-22 00:28:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697872-1576974532633-716aec10","meta_rev_id":"1351177cc73e080e5fc7e1e5582f34e418fc1c5a","_number":697872,"virtual_id_number":697872,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-12-21 08:50:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-22 00:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-21 22:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-21 22:08:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 09:32:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-13 11:00:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-21 08:50:05.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-21 22:08:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-22 00:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4dec737e0ec8d6b32c41a9e287544832b9ceb71","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-08 07:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53a992bf207ac719c7f41433a4c280982965d1c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 07:59:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d41346ecf334e598ee3636edf029b3f : SUCCESS in 26m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13a488c29cc14be99b37c790c3c7979c : SUCCESS in 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bfcd38270734d7fa80afbc4cc185772 : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/199fcd40acb442019bafb9c7b3989907 : SUCCESS in 19m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4b7451d61e04962a32ac83955aaa67b : SUCCESS in 21m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af397411c9f49f6a7f706d7f834e802 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964c4b52d5ab4bd3905be6f60b7a94f7 : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0470ff0cb38d936cd308fefc2ecb187075939e57","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-08 09:10:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697872/1 : SUCCESS in 1h 42m 44s","accounts_in_message":[],"_revision_number":1},{"id":"84ccac9c77f5a9a29bf93aaa926c6af704fd3176","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-08 12:42:05.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b2248c8682d2173bbe90d452c4a3e84dad2428b1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-09 07:38:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a658d497f8d01489935b817966932c9be086840e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-09 07:45:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d26796c9460786ceffe3fecc164f31b26f195874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:14:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5397d399fe043b1a46a6ecd2c02267c : SUCCESS in 32m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f01720f72ec1405186b996bae912bc9f : SUCCESS in 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0856b78ae13c485191a7dd1fd63e60c3 : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dbeb82bbe534c749ce878314448ac30 : SUCCESS in 21m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adf8b554d25a49f39f145d92fda1561a : SUCCESS in 20m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0b3fa42abc4b0bad66d7e7da28b786 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9de424861a434201849bfd9f5782647e : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f7fca473bbb137ef0bae5a6df83d0b9cf442b0eb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-09 08:48:04.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"26c473e377efaa04bbcd68424ab2b9ee1116e0ba","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-09 09:28:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697872/2 : SUCCESS in 1h 49m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e45d7efb2006bb7429cd8565a0101320197eab74","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-10 08:51:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"61c3798d943e30eafa57cfdfe5018d38c45354dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 09:34:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b183314601041fb8d0bf97be128194b : SUCCESS in 34m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eaad2907b6b4e94b46724cd431f1916 : SUCCESS in 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9485177cb347e2aaa1c1954c1076da : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb8bf5018876454caadd58dd45ae4c8b : SUCCESS in 26m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/359e2fdbd576444aa4b36eaa81869c67 : SUCCESS in 27m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b4127fcaa64f0a9467c19518a8de20 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0809ddc3e8dd43eaa460644293a1e2e8 : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":3},{"id":"dd7c9218b503e0b2bb8854c11f20e258003b6c60","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-10 10:01:08.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"15841dfe1e1ae1c8f854b256b9696f2e60890286","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-10 10:38:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697872/3 : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":3},{"id":"3936bd3966f6cf9c565bad5e1b7c03e1e16f6960","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-11 08:32:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"57bc27784b205c5a57fb36b3911e29fd8a8b8460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:58:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc9daf94b4d14094820af37f5b248428 : SUCCESS in 24m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31addea123a649c48d1ff95864ad71be : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d430634c5b642128121d2701fe433fb : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42239747792f4bad802e7f9e6c78de9b : SUCCESS in 17m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/718746d95d5b4c44b941a88e76daa3ff : SUCCESS in 25m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd848fa817a9404b8f1be985b068e198 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/488f1c1b97354283a456f119e5d38de8 : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":4},{"id":"8e5a2b5c2736700f14508c06dcede3d43530e0d8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-11 09:44:04.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"2982ca47054311b0eea398ee6e8c19a4c938c4d9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-11 10:26:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697872/4 : SUCCESS in 1h 53m 13s","accounts_in_message":[],"_revision_number":4},{"id":"99ad68279407cf1490c66c52bea915e66b1279f5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-11 12:29:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"acdf8a8491f6f0b76e2b4748a41841615478895b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 08:42:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d5c3c482f1e066aef1222938dd988ec66ffbdb16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:11:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a844c33225164416b72740117cfed8a8 : SUCCESS in 25m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d074144d1904ca4a637b4231d03683c : SUCCESS in 23m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd9246cee8fa4a6a935d6f38fee81b59 : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5070fd61d4e4433a9383b176fd83d111 : SUCCESS in 23m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca180c99615c4979bdb7b59a7fbbd545 : SUCCESS in 25m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0488a042c524c1992bb5436b1aa548b : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42cc9508c6a34c37bf167424b8333da7 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":5},{"id":"7082e6ba32d68b50b1ad1c1e0b5efb3ac21540a8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-12 10:01:01.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"371c84af7e3f4bc6e5ecc5f8a0ded409fd8c23b4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 10:30:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697872/5 : SUCCESS in 1h 47m 15s","accounts_in_message":[],"_revision_number":5},{"id":"250ab0b08ee98b6fc521c553222b2e086b6a4229","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-12 14:35:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"91aa3c135d5e913e0633f447d6ae2da3084b1635","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-13 09:02:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ca04394a94a9ef70d810c6eb6f9893365843dfd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 09:44:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70caa7c935be4664a2a7d50ce3d675cd : SUCCESS in 35m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c897c20845f4af785b520e44155ded7 : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ecb401f10e46a9a1827d33be896634 : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd3ff1643e40438db06155732d772479 : SUCCESS in 29m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d73324c8378f44e8842047160f8562b3 : SUCCESS in 27m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfce137e0ea4295ab32cc21100f913a : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b580ff21ba3a4eb5be339fd66279eae1 : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":6},{"id":"cedebe9cc10b4b115b2ccae632521188bbdb852f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-13 10:31:24.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"71e71c808c4a604d9740c635182d23eb426ce570","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-13 11:00:38.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/697872/6 : SUCCESS in 1h 54m 56s","accounts_in_message":[],"_revision_number":6},{"id":"515362a755db33bb71ec5ddc1fff0e1778d8d814","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-14 08:21:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eb19377624cbc765a80df7d6186903f02f0347dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 08:44:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e29b582e7784e57a377abee190840c6 : FAILURE in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd7c603bb96949419f94f68c9fba0138 : SUCCESS in 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592e612c54834c77a10c9c7529e01f8c : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f69326d8d1d8475c93c0e2a29c18bd76 : FAILURE in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/625717257e564fa0904cbbdec0fd972d : FAILURE in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a243b0c711d34f85b167c4e1897b2a31 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be317c69b31044688192eff3839a26d8 : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":7},{"id":"8f9ffe02204dd1391984ca4090f239bb6a4a236d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-14 09:48:52.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"2e88746493fad9274b82060363d6c8f3711fd6c5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-18 09:32:17.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2c0ae50e719c7a9b81faf423df63a5e9605323d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 10:29:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89177e8ab61f46c4a21717898342f97e : SUCCESS in 48m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25c95c365c504649909698cdca4ec3f1 : SUCCESS in 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1280054269413ab33439de7aeacd3b : SUCCESS in 12m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7bcac11d7034682952161017939b024 : SUCCESS in 22m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad5c93d57ef04a459fd801c1a68fb3c7 : SUCCESS in 33m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3650d86919142e0a185863a7d162104 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb8ff765dd35473b8cb5abc660b9305e : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":8},{"id":"a868933f6c4525b3404f85b450d528848a69e4af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-18 11:02:13.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"6c25a8e7468cd9a83c6f1568687b83571d33d328","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-19 07:47:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"587aed868275c8aaa7c87fbb7223078d7198b02c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 08:24:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a09af267193f46129cb7a7ae3f13afb7 : SUCCESS in 31m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1012806fc18468e91fd0f0f3fc276f8 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21190859a5b47bc99cb9c33b831caf0 : SUCCESS in 16m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71d67f75f4d34ca2acc25cedc8f79cf5 : SUCCESS in 22m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cee8158158147b392238734ff9d7403 : SUCCESS in 29m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f568b560d5e649b69462b32ffdaf6091 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc3378ae8def42bc835e07a84bf30737 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":9},{"id":"1f3c9e7bad3abd524935785ae4471f879564cfca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-19 09:05:32.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"03b8242401d0f81d1f996fd9195590b21c11bac2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-20 07:42:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cfa8f31c31fd513eeb10380d86ff88fecbb1fd9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 08:21:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e76dd035c114780b4f8a9c993ab7ccd : SUCCESS in 38m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37d56a07b7b848488e5bb62279b0b3da : SUCCESS in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec106009d87c476798be5b6c1e86f088 : SUCCESS in 13m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96ae8b8aa74c4a1cb1307325920748ee : SUCCESS in 24m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1f8d3f5763d4fca95e430dcce69c615 : SUCCESS in 32m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b66c72240ab4abdadaef35a71bf66d8 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ec9c619a8204996925ffcd485afc301 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":10},{"id":"22d41821f0296bf70163d8567ac62cbb64e39606","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-20 08:55:22.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"1904c0e7e4eb50875ce86038f8dd2067e2013db8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-21 07:16:13.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f5cdd8a7349988d7097a33be3a4e3e7e3d5efcd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 07:46:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5e9a9f2471649fd9b63246dc2660d2a : SUCCESS in 28m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a50c899f5b24086930b5eec74003f83 : SUCCESS in 25m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa747f4116a41ce95d0f07661aaff0f : SUCCESS in 12m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f314051f7394ed39569fe6dfefbd60b : SUCCESS in 25m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/605bc1d144ab494b90e844eb8b439f8b : SUCCESS in 24m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bcbc5f168b5402e808fc76d1edfe544 : SUCCESS in 16m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17ad4e65faea4bc2be45843a9f5a9b25 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":11},{"id":"2e2dd6df5bc523a55062bb770b845689e8a48631","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-21 08:50:05.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/697872/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"6cdf86965ac653eec121406aa4e2abf52a34ef67","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-21 22:07:26.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"cab4327d12b6944775ccb1a69891c9d84f511682","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-21 22:08:26.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"530f9b3f2131c31eee882b4250030168c5c44bce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 22:08:41.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0128eb6c547dfea146754b91b4bd42dcd5530138","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:28:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"c47e2d4de24a64097ce9ef38c868056f60168dc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:28:52.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/710ddfa1188b406997cb2986a1467958 : SUCCESS in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2bbb16913e42729ecec7aa9ca7aba3 : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc4d232365a04c5fabd54b4e2eb36b1f : SUCCESS in 23m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b782b459096b420c9b861e682d538fb4 : SUCCESS in 28m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca80190135e434a9b8c83de7e031fde : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d96ec1789214447997285f1483a6d62c : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":11},{"id":"60feca42e04cdf67094665ac07fb9aef95316ac5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-22 00:32:18.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0bf9dc2bdf7473c91b32d4d168df60d : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4fe5e268b8f4831ab7a5374199e35d5 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"2a8b5860a0cc7674bba7da207ec9bc8828eeb829","revisions":{"1e569ccb7f61d77627f5cbdbfe2aef4c4bb76150":{"kind":"REWORK","_number":1,"created":"2019-12-08 07:27:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/1"}}},"commit":{"parents":[{"commit":"9426572b9ea813ed46d8501cad6a22315149bcc1","subject":"Merge \"Add neutron-tempest-mariadb-full periodic job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9426572b9ea813ed46d8501cad6a22315149bcc1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-08 07:27:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-08 07:27:13.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e569ccb7f61d77627f5cbdbfe2aef4c4bb76150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e569ccb7f61d77627f5cbdbfe2aef4c4bb76150"}]},"branch":"refs/heads/master"},"4e0060cfc780a2fc84f11b7d8f445de3b1a8f7b0":{"kind":"REWORK","_number":2,"created":"2019-12-09 07:38:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/2"}}},"commit":{"parents":[{"commit":"b8cf079b765ba988e15a42e30e283b06a70f2578","subject":"Merge \"[OVN] Move OVN common.acl to Neutron tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8cf079b765ba988e15a42e30e283b06a70f2578"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-09 07:37:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-09 07:37:59.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e0060cfc780a2fc84f11b7d8f445de3b1a8f7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e0060cfc780a2fc84f11b7d8f445de3b1a8f7b0"}]},"branch":"refs/heads/master"},"a16a44321643b3c33b466323fa1a398010b9872a":{"kind":"REWORK","_number":3,"created":"2019-12-10 08:51:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/3"}}},"commit":{"parents":[{"commit":"3d256d1b4251f9bfbbb2d81e06a8c4261519aa0b","subject":"Merge \"[OVS] Handle added/removed ports in the same polling iteration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d256d1b4251f9bfbbb2d81e06a8c4261519aa0b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 08:51:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 08:51:26.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a16a44321643b3c33b466323fa1a398010b9872a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a16a44321643b3c33b466323fa1a398010b9872a"}]},"branch":"refs/heads/master"},"78f3af52efedfea8aea2820b0be13ec7887244bc":{"kind":"REWORK","_number":4,"created":"2019-12-11 08:32:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/4"}}},"commit":{"parents":[{"commit":"3ad5ad02c94ac0359956db9aa9c3314ebd9651d3","subject":"Merge \"[OVN] Move OVN metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ad5ad02c94ac0359956db9aa9c3314ebd9651d3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-11 08:32:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-11 08:32:41.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78f3af52efedfea8aea2820b0be13ec7887244bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78f3af52efedfea8aea2820b0be13ec7887244bc"}]},"branch":"refs/heads/master"},"41f9fd0fca48c51b66d1ceb213e7da24a7141bfa":{"kind":"REWORK","_number":5,"created":"2019-12-12 08:42:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/5"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 08:42:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 08:42:29.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41f9fd0fca48c51b66d1ceb213e7da24a7141bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41f9fd0fca48c51b66d1ceb213e7da24a7141bfa"}]},"branch":"refs/heads/master"},"19aebaaa8fa17fc0c038c5ed96541b500d207af9":{"kind":"REWORK","_number":6,"created":"2019-12-13 09:02:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/6"}}},"commit":{"parents":[{"commit":"e04e1d80e465a04b831047ac0ffb566c80cdd029","subject":"Merge \"List SG rules which belongs to tenant\u0027s SG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e04e1d80e465a04b831047ac0ffb566c80cdd029"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-13 09:02:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-13 09:02:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19aebaaa8fa17fc0c038c5ed96541b500d207af9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19aebaaa8fa17fc0c038c5ed96541b500d207af9"}]},"branch":"refs/heads/master"},"20259f2c6b00cb35d6086924c6f96840abfa3da8":{"kind":"REWORK","_number":7,"created":"2019-12-14 08:21:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/7"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-14 08:21:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-14 08:21:19.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20259f2c6b00cb35d6086924c6f96840abfa3da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20259f2c6b00cb35d6086924c6f96840abfa3da8"}]},"branch":"refs/heads/master"},"f196085bce24075fccc586f9bd5db0653c4f00fe":{"kind":"REWORK","_number":8,"created":"2019-12-18 09:32:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/8"}}},"commit":{"parents":[{"commit":"0db38cf8a863131336426c78d7114b7b15c98974","subject":"Merge \"Remove unused arguments in test methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0db38cf8a863131336426c78d7114b7b15c98974"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-18 09:32:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-18 09:32:13.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f196085bce24075fccc586f9bd5db0653c4f00fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f196085bce24075fccc586f9bd5db0653c4f00fe"}]},"branch":"refs/heads/master"},"e3fab3907e1ab4b14677d58a15f869a6eaf08b0a":{"kind":"REWORK","_number":9,"created":"2019-12-19 07:47:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/9"}}},"commit":{"parents":[{"commit":"858d7f33950a80c73501377a4b2cd36b915d0f40","subject":"Merge \"[OVN] Import OVN Client, L3 and QoS related code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/858d7f33950a80c73501377a4b2cd36b915d0f40"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-19 07:46:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-19 07:46:58.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3fab3907e1ab4b14677d58a15f869a6eaf08b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3fab3907e1ab4b14677d58a15f869a6eaf08b0a"}]},"branch":"refs/heads/master"},"df7cd44925758add5fe34aa34bb3eba3c91a43c4":{"kind":"REWORK","_number":10,"created":"2019-12-20 07:42:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/10"}}},"commit":{"parents":[{"commit":"4b7c2dceb47764edd99e3e58748a314b90be0c09","subject":"Merge \"Add Neutron Priorities to dashboards list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b7c2dceb47764edd99e3e58748a314b90be0c09"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-20 07:42:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-20 07:42:49.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df7cd44925758add5fe34aa34bb3eba3c91a43c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df7cd44925758add5fe34aa34bb3eba3c91a43c4"}]},"branch":"refs/heads/master"},"2a8b5860a0cc7674bba7da207ec9bc8828eeb829":{"kind":"REWORK","_number":11,"created":"2019-12-21 07:16:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/697872/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/697872/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/697872/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/697872/11"}}},"commit":{"parents":[{"commit":"0115f71af8784dd7a180e09035c9530124e24bcc","subject":"Merge \"Bump min pyroute2 version to 0.5.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0115f71af8784dd7a180e09035c9530124e24bcc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-21 07:16:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-21 07:16:10.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I78cbd02c7dc0c60b12c9c4e90046336157b8bcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a8b5860a0cc7674bba7da207ec9bc8828eeb829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a8b5860a0cc7674bba7da207ec9bc8828eeb829"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
