)]}'
{"id":"openstack%2Fneutron~698565","triplet_id":"openstack%2Fneutron~stable%2Fstein~Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2","project":"openstack/neutron","branch":"stable/stein","topic":"bug/1856156","hashtags":[],"change_id":"Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2","subject":"Use constraints for docs tox target and cap hacking","status":"MERGED","created":"2019-12-11 18:21:52.000000000","updated":"2019-12-20 12:34:22.000000000","submitted":"2019-12-20 12:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698565-1576845114985-0548d2f5","meta_rev_id":"82668e30f938ecf51e288e4e6c1358592ac8dd4f","_number":698565,"virtual_id_number":698565,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 12:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-12-19 08:10:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-12-19 11:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-12-19 15:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-12-19 15:27:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 18:16:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-12 19:43:31.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-13 02:52:50.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-12-13 11:33:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-12-19 08:10:31.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2019-12-19 11:04:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-12-19 15:27:22.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-12-20 12:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9da44fe6b81dd5ec531bb12e697c0aa1817fec0b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-11 18:21:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2223008cd57ca4acd003b8156ce506904dd88c4c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-11 18:23:31.000000000","message":"Topic set to fix-stable","accounts_in_message":[],"_revision_number":1},{"id":"1fb91e5cdf798a948b573e21d58c4d3e464ec76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:32:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9281b416aa914c96b6916cf1adb78346 : SUCCESS in 34m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/096c32e5b51e4a299042c809ba8b8689 : SUCCESS in 24m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c122eaca854dd7b5878cfbc0583f25 : SUCCESS in 13m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/439bbefa112e437c885b2009191ba1cb : SUCCESS in 21m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5602c34804e142aa9b62a17fa41b48ba : SUCCESS in 28m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2f283f9c35a4d8c956f9c4b0706250d : SUCCESS in 30m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70a276d936e9416ab6ccda8acbc428e8 : SUCCESS in 33m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dff50e3b5884b06bf11055e0da4e99d : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cd6d2cec4c1497098d7e042c0096bcc : SUCCESS in 20m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a6d09916376548828d20deb4ad6d9e59 : SUCCESS in 44m 19s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3a6638bfc7634e42901414b55b90856a : SUCCESS in 41m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a43cb2b4e8824b9e91241f6b9cae06b0 : SUCCESS in 1h 09m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/55919e9038324300a9dd51520fc6d9a6 : FAILURE in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb8d3bdbe1091b8fa53315039d96bd7bf1b79f6c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-11 20:16:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698565/1 : SUCCESS in 1h 54m 32s","accounts_in_message":[],"_revision_number":1},{"id":"40f53d44972dd3cd6629c7772afde41565897e51","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-12 09:05:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41b70ed9ced08fcad96a1585bca1bff7828df9f3","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-12 09:12:54.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t see why this change is stable-only. Ignoring the constraints file in the docs tox target is a bug in my view on all branches, not just on stable/stein. I understand only stein is broken right now, but the same bug can appear on all other branches too, it\u0027s just a question of time. I suggest doing this change on master first and backporting it as usual.","accounts_in_message":[],"_revision_number":1},{"id":"1702c19d135af6e9f51a9f49455ebdc959b18f16","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 09:18:08.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\n\u003e I don\u0027t see why this change is stable-only. Ignoring the constraints file in the docs tox target is a bug in my view on all branches, not just on stable/stein. I understand only stein is broken right now, but the same bug can appear on all other branches too, it\u0027s just a question of time. I suggest doing this change on master first and backporting it as usual.\n\nYou are right Bence, should be fixed on master, too. I\u0027ll -W this and start the same on master.","accounts_in_message":[],"_revision_number":1},{"id":"268f69e7cc6c70bf5a1f07eeaf40387c7377275d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-12 09:19:05.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e (1 comment)\n \u003e \n \u003e \u003e I don\u0027t see why this change is stable-only. Ignoring the\n \u003e constraints file in the docs tox target is a bug in my view on all\n \u003e branches, not just on stable/stein. I understand only stein is\n \u003e broken right now, but the same bug can appear on all other branches\n \u003e too, it\u0027s just a question of time. I suggest doing this change on\n \u003e master first and backporting it as usual.\n \u003e \n \u003e You are right Bence, should be fixed on master, too. I\u0027ll -W this\n \u003e and start the same on master.\n\nI opened a bug report: https://bugs.launchpad.net/neutron/+bug/1856156","accounts_in_message":[],"_revision_number":1},{"id":"f88e4ce6e74e70ee394b1d6aa636e02993a41b16","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 12:11:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16c388ded62c979382142e28b9b505047ac3aaa0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 12:26:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54b5948aa9118b7ea186b281b1f47fa5b1bc5147","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 12:26:46.000000000","message":"Topic changed from fix-stable to bug/1856156","accounts_in_message":[],"_revision_number":2},{"id":"40229518ccf0c4e30d0845d2c9defd03e1aca244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:32:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30f532127f694d12ad6c1b4e45b58a5a : SUCCESS in 27m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a536e1d276ba44cf87efc2e8b46fa74e : SUCCESS in 29m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5380270983492198861bf91347af38 : SUCCESS in 12m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/700a14d4b7c4449ea25e688ae17b13cc : SUCCESS in 24m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3f1693d1a53f405f80ec938406f000e2 : SUCCESS in 29m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd1638fe42a8435e99edffdcb72e6c9d : SUCCESS in 23m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da78b52aaea24d8c94f8c29e31d0d362 : SUCCESS in 24m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38948dc6bb84edfb3de278aafab07df : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e156911fcafd4078a12dd683af64ef3d : SUCCESS in 18m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bfff7dc009364f64955824ec04d235f4 : SUCCESS in 48m 12s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/931d1e169b994f0e8f096d2ca89539c9 : SUCCESS in 47m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/092052c77b68499c8e3ea086beeb5a7a : SUCCESS in 1h 13m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1938cb1894664fa8837f646c458e641b : FAILURE in 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8629fcd2e1db614cb2d29cfc339ff668a35f919","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 15:53:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7a75219fec31e59352338768e075c0fa6ac30c2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 16:42:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698565/2 : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":2},{"id":"34dd1d60d8c6e0dd1b34141eb89706a8dd2eb9f9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-12 17:09:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b8a7f566ee9a94ce91dda89e10d30c4f15c8708","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 17:58:11.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"763c0c8b28decfbd1a69bff7438ef2321cd800c1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 18:03:49.000000000","message":"Patch Set 3:\n\n@Rodolfo @Yamamoto thanks for the reviews! I forgot to add the details to the commit message why that change is needed. Now I\u0027ve updated it.\n\nIn short: tox-docs fails without that addition :)","accounts_in_message":[],"_revision_number":3},{"id":"b2b09c50a62601485d46c224ae3ec3890d056443","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 18:13:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"363437e775dc009d95d932fd8d3eed2b0bee662b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 18:16:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"50322ab855949dfc9da49f26452006d4df2b8586","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 19:43:31.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/698565/3 : FAILURE in 1h 36m 30s","accounts_in_message":[],"_revision_number":3},{"id":"97e2605dc6daefd4ed45aba2703a426cae7bb07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 19:59:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac615ed6a7a4401ba472d6d09b2c6e18 : SUCCESS in 27m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/feb390f218e6407ca609f895f953e8a8 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a1947a32ef42749ee13f52279b316c : SUCCESS in 12m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c1e554b9aad467e9de1022d80e3504e : SUCCESS in 29m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4f1b8ce7b9e4aacb1548cd824fcfb3a : SUCCESS in 27m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9c37be8d72a47aabfcb8d93c735b7fe : SUCCESS in 28m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e099390459324827812697aefba2aeb7 : SUCCESS in 30m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a914448885684ae79e7c35c152d842ae : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d098cf8bf124d9a8e05ad0ee75c89f9 : SUCCESS in 27m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4038cf38f76d45c6bec721e65279d404 : SUCCESS in 50m 10s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/41a31205f3a84b1882872179375fce92 : SUCCESS in 1h 01m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fb7486bd68f24638b71bae685c9b40cc : SUCCESS in 1h 20m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/21d2e830fd994d65b1d7dc2541504387 : FAILURE in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cd2fc7f0662570f8cd3584e0d484900dc4121fd","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-13 02:52:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7d53bdf5826d072b394f801fd4b057ebf3004851","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-13 10:00:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"98cb471f26f4930e58aad07b89580fd3f01e5853","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-13 11:33:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"79d7801ec6ae98f9b90b909e51ef1fc970a7d927","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-13 13:56:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"355a3d4c3cb2a970204d49b08d1e8a57fb662228","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-13 21:39:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ffe2a63cd22acd6ecd2358f268220e1b25af94c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 21:40:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21a2b5ac7c9f23fde06af41cf44092ecce2d81ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 00:27:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08624683aa044ad780ac5d5bcb919fd6 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5271895fba974700818b323ec52bc4a1 : FAILURE in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fab8d3242d3b476d92b34e1712f846de : FAILURE in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af229f76581d42fa9ff39f00381c1e45 : FAILURE in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81fa1fd14eb94a3c9323091886b8125a : FAILURE in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0b5d4fbf97d42f2af18ebd71b468de6 : FAILURE in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74eed2c01b52417990d9becb19691805 : SUCCESS in 5m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f8c40b7747a74629aca5ea46984ad63f : SUCCESS in 1h 23m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91745e81fcad41f79fea9b85dfd685a4 : SUCCESS in 1h 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/933bd738ecf6435083815198d6fc0ed0 : SUCCESS in 16m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8a8b40241912471a82357ee32f515e14 : SUCCESS in 50m 08s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/45efc8af9bf648c99596b9300f8cce66 : FAILURE in 1h 17m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ec9c4415a73f43798d5725ccb6b76d39 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":3},{"id":"37428f013a4945a2dad583c45617ee18f7c98e79","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-15 10:48:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b2794312d3a5a5ea0420e1ad336c9116e66f8b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 11:54:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/718a4a073638434991c28f598c02ef4a : FAILURE in 5m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4bace55b1b7425dbe946096fccec452 : SUCCESS in 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394ba41a2d444640b517dc97c34b3c9f : FAILURE in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fde7758f710a44e6a1dcad41a50ba2ee : FAILURE in 3m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7cbe414b19b248d1940b5e581db4c149 : FAILURE in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8468a697b9c24eda8bc4f4199275f8f9 : FAILURE in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d077293d0fd46db9b78530f0b984e9d : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a779d43ae134bae8b1d9cb5e971e987 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/290121393dbf4323961e64bcaaf1bc5b : SUCCESS in 19m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/64fb3dba2ec8431598a23a14bb1f73b7 : SUCCESS in 43m 45s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c231ac8ea13a43e0acc7bf48b213ef62 : SUCCESS in 42m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6abcdbcb417946e8acc04f979d82f7d0 : SUCCESS in 1h 00m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/33417a75ed424548a52b63b6407428dd : FAILURE in 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"796c462f0b68c940efb0e2a2d2f1360e96cb1b2c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-18 18:34:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2b404924eede5d446f1cd28ef029653aba6254be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 20:37:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac447ceead2f49e88f3d9352c08dcf89 : SUCCESS in 27m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1102a857a37d40aaa49dfee1c069364e : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf609a4d7ee9480ebe505ffaf42daf6c : SUCCESS in 11m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05c53a46460a4466ac6f75c35d5de016 : SUCCESS in 22m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7c0a786a2c55435aa284ec7ae511d056 : SUCCESS in 20m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f4b4029ec914a74b8bfdbfc469aaa65 : SUCCESS in 20m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f00e86be465a412294cb1687a62be599 : SUCCESS in 29m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2faccba9319a4985ad3f0f6317988076 : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/97ffea1773964d539f48c8616e62c04e : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bf5644b81274f80b770ce3024d2bd34 : SUCCESS in 23m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1fa1e4cffe7d44f0a2c85b90e4d239c9 : SUCCESS in 47m 37s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/26de58047e17421da286793ea60acfca : SUCCESS in 52m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/67729b2ce6224ffbb7a3a23efa95e119 : SUCCESS in 1h 10m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a9180f2683fa492c806edede66d39860 : SUCCESS in 1h 55m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/17068f08e88a4dcb9e4b2f2bb488bcf8 : FAILURE in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"69c4a30ba9582a78a45f9c197635571a13503c7a","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-19 08:10:31.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2038a8756fa33a68ff460c3e67a375d88130896","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-19 11:04:36.000000000","message":"Patch Set 4: Code-Review+2\n\nPrevious fix was ready to go in, this revision just adds hacking cap","accounts_in_message":[],"_revision_number":4},{"id":"a5ea2df1ae4011326cd24e57af2b250b77638e10","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-12-19 15:27:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2912383a7596aef40507c9a59a20a2cd23448a64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:27:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eb9bb9d2b09518fbe6a068419ca1f4ed0e97a3f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:01:26.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d97496ee9bc140f28660f0ba34235d52 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d69ba93b8ab4360a784803048079694 : SUCCESS in 12m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8b58da532854cd8bf631a73fc8b928e : SUCCESS in 21m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/201334c7370543f7bd144bfbfa967ca9 : SUCCESS in 26m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5205248ffd644dbd8d904eff326da241 : SUCCESS in 25m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a5f67b413cf42fb837c44086f1e105d : SUCCESS in 22m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924907aa37bf4b12baeb0af49d115532 : SUCCESS in 6m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ca0cfbf674c47709baf165042b7686f : SUCCESS in 4m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b81e5f9b11094b4284247ff79b646d09 : SUCCESS in 1h 24m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/498fa12d21364ab4bb55d4d531ad5fb7 : SUCCESS in 1h 22m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61b4fb20d22c40da9492d64966fbdff8 : SUCCESS in 21m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1922e3f34307417aa9548bf980f703d5 : SUCCESS in 46m 43s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/65bfa54dd5ba46c7ad0645078e9eef25 : SUCCESS in 45m 55s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6af5d9bc6c41431f9daf3c219d5911fb : FAILURE in 1h 11m 30s","accounts_in_message":[],"_revision_number":4},{"id":"e44c21fa8e099e6649745262422974e01426a93f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-19 19:46:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"195acf2ac90b9517d4be9d71236f2d4f7a3b706c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 21:32:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69cdc51370da4f289bc15556ace93567 : SUCCESS in 30m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57aad484d1fa4b9c8caf5692ceaaf5e9 : SUCCESS in 24m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb89fb1a1627465eb774918833a1cee9 : SUCCESS in 11m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0628fff663243529607fa65b14fe29c : SUCCESS in 20m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f64d1e1a5c844648bc2c35474c389e93 : SUCCESS in 20m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10a891a25bae4a5db51541c9e78d21b2 : SUCCESS in 21m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f161dc331774419b94fe291d23118f7 : SUCCESS in 23m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20262f249011459dbf05c73380b3c186 : SUCCESS in 5m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f9883e26fe77409f8bd0d359a5abf3ab : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2c2cab309144dca57bff2c83e1bbb8 : SUCCESS in 23m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f1fab12abddd4451b49af04f07983705 : SUCCESS in 53m 25s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c8889e8402614a71b3c0e4fe5847c983 : SUCCESS in 52m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c268460ef1bc4b9b9e9fe195525eebc9 : FAILURE in 1h 26m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e778b80f3efc48b9974511a181831da8 : SUCCESS in 1h 37m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1b30ab07b41b4ee48012e96b00f42d55 : FAILURE in 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0dfbe4a03348429fc82c0c0a620b047f2561de1f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-20 08:19:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9e3f4536c47a77bc463f0ddff8ad7ee62dbdfba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:06:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7825c0ad9994ad987f09c91795eace8 : SUCCESS in 29m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df5e879b08cb4d479d05af99ca059315 : SUCCESS in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0359021c5c2e4960ab182c92b390d32d : SUCCESS in 13m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11456b02db7c41808461e10881d62b0a : SUCCESS in 22m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6b92b966a855494b931b3e8594904e62 : SUCCESS in 24m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c30934b406e41bf9d85d6838fa8f7b4 : SUCCESS in 22m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40fde316fea94b4a88954acc362b29ca : SUCCESS in 38m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c208d0399cb4a44837660bb76b203da : SUCCESS in 6m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e4017d4c1ac4a9fa3dd89dda859d8ce : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0f714afb9ba4f13a80ef4b8ca5c71f6 : SUCCESS in 23m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b5d81d68f8564e64ab0bf52c0554a56c : SUCCESS in 43m 40s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/1c90aade90354e9a871769ff3d71e9c2 : SUCCESS in 49m 14s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4566d0ddd7964ce8b210d8d01b2d3a33 : SUCCESS in 1h 00m 30s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/59f79247730e424ca4063939827d99fe : SUCCESS in 1h 41m 35s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/78e340dedc464099ac028d2403c10504 : FAILURE in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5ad880502a8b1c7347cd842bf84cb4e59c3777e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:06:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1dafcb283debc1cdbfd84841ce5b5f4df504851c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:31:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63e3ec920a114a95a127e0798c08019e : SUCCESS in 23m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a96eaaa2a14a04900d958b729ae19c : SUCCESS in 12m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/468851af64824818b35cc6bd5d3e4cb6 : SUCCESS in 20m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e5debc877c44e8782910020c8975ac9 : SUCCESS in 26m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53fd9fb1dbda4014b5bff68f0f7a5f1d : SUCCESS in 22m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8778b4ea1cc1440c83d6f007e24c9416 : SUCCESS in 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b7d387497d4509852c535414f5f134 : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/26012fc29c2c4820bc2169fd18a1c64a : SUCCESS in 2m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/274ff394d7b84b86a3e9965eceaacb41 : SUCCESS in 1h 42m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30ce216f5def470bbd6c45c7c86b49e5 : SUCCESS in 1h 14m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d57f1f57d3f742b8bf73c8f669e25b07 : SUCCESS in 23m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f1e5a3ffd2d4bac85af16ca916b2c40 : SUCCESS in 43m 30s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/29a2ee481bb7409386eff92571dcbb3c : SUCCESS in 45m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/357e8eb06c574ca7a7c6b2d951095eda : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6472c81c2a399afe981dafb42347710c2587af6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:31:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"82668e30f938ecf51e288e4e6c1358592ac8dd4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:34:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90633c778548402da02f07831eecd3d9 : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bb1865842ef420f8387c7af2cb14fc4 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"34239888eb661415ed372808304a42339563dad6","revisions":{"df0a1f0adfc9e92573801cb41970f09ea8c1cb2e":{"kind":"REWORK","_number":1,"created":"2019-12-11 18:21:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/698565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/698565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/698565/1"}}},"commit":{"parents":[{"commit":"57edfee5da12ed32f625de6120c344ce69b46619","subject":"Convert Pyroute2 byte results to strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57edfee5da12ed32f625de6120c344ce69b46619"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-11 18:11:00.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-11 18:20:38.000000000","tz":60},"subject":"[stable-only] Use Stein version of neutron-lib for docs build","message":"[stable-only] Use Stein version of neutron-lib for docs build\n\n\u0027tox -e docs\u0027 does not use the upper-constraints.txt, that is why it\ninstalls the latest available neutron-lib version from pypi. With newly\nreleased neutron-lib version 1.30.0 some old constants were removed and\ndocs build fails. To fix this the proper upper contstraints needs to be\nused in the install_command of the docs tox target.\n\nChange-Id: Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df0a1f0adfc9e92573801cb41970f09ea8c1cb2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df0a1f0adfc9e92573801cb41970f09ea8c1cb2e"}]},"branch":"refs/heads/stable/stein"},"bc872e7673dc1c5d7aab0d42f969de0ae0070be1":{"kind":"REWORK","_number":2,"created":"2019-12-12 12:11:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/698565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/698565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/698565/2"}}},"commit":{"parents":[{"commit":"57edfee5da12ed32f625de6120c344ce69b46619","subject":"Convert Pyroute2 byte results to strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57edfee5da12ed32f625de6120c344ce69b46619"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 10:14:10.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 12:10:03.000000000","tz":60},"subject":"Use constraints for docs tox target","message":"Use constraints for docs tox target\n\nIn Stein the docs target started to fail when new release of\nneutron-lib appeared. This is because tox installs neutron and its\nrequirements without any constraints. To fix this both the upper\nconstraints and neutron requirements needs to be added to dependencies\nof docs target.\n\nNote: patch adjusted to stable/stein\n\nCloses-Bug: #1856156\nChange-Id: Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2\n(cherry picked from commit 07be7934359aa121761396ba47128904f19cb2e4)\n(cherry picked from commit d1c4ba581085f4978da82a597b1387d60e79ff0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc872e7673dc1c5d7aab0d42f969de0ae0070be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc872e7673dc1c5d7aab0d42f969de0ae0070be1"}]},"branch":"refs/heads/stable/stein"},"52b589be72d8a86cb9abf328f0a6562c3f32dff0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-12 17:58:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/698565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/698565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/698565/3"}}},"commit":{"parents":[{"commit":"57edfee5da12ed32f625de6120c344ce69b46619","subject":"Convert Pyroute2 byte results to strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57edfee5da12ed32f625de6120c344ce69b46619"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 10:14:10.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 17:57:09.000000000","tz":60},"subject":"Use constraints for docs tox target","message":"Use constraints for docs tox target\n\nIn Stein the docs target started to fail when new release of\nneutron-lib appeared. This is because tox installs neutron and its\nrequirements without any constraints. To fix this both the upper\nconstraints and neutron requirements needs to be added to dependencies\nof docs target.\n\nNote: patch adjusted to stable/stein and by using upper constraints in\nStein another sphinx error comes up:\n\n  ../config-services-agent.rst:70:Error in \"code-block\" directive:\n  1 argument(s) required, 0 supplied.\n\nThis also makes the docs build fail so a \u0027none\u0027 argument is added to\nthe code-block.\n\nCloses-Bug: #1856156\nChange-Id: Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2\n(cherry picked from commit 07be7934359aa121761396ba47128904f19cb2e4)\n(cherry picked from commit d1c4ba581085f4978da82a597b1387d60e79ff0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52b589be72d8a86cb9abf328f0a6562c3f32dff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52b589be72d8a86cb9abf328f0a6562c3f32dff0"}]},"branch":"refs/heads/stable/stein"},"34239888eb661415ed372808304a42339563dad6":{"kind":"REWORK","_number":4,"created":"2019-12-18 18:34:50.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/698565/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/698565/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/698565/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/698565/4"}}},"commit":{"parents":[{"commit":"57edfee5da12ed32f625de6120c344ce69b46619","subject":"Convert Pyroute2 byte results to strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57edfee5da12ed32f625de6120c344ce69b46619"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 10:14:10.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-18 18:29:04.000000000","tz":60},"subject":"Use constraints for docs tox target and cap hacking","message":"Use constraints for docs tox target and cap hacking\n\nThis patch fixes 3 problems to fix gate jobs in Stein.\n\n1. In Stein the docs target started to fail when new release of\n   neutron-lib appeared. This is because tox installs neutron and its\n   requirements without any constraints. To fix this both the upper\n   constraints and neutron requirements needs to be added to\n   dependencies of docs target.\n\n2. Patch adjusted to stable/stein and by using upper constraints in\n   Stein another sphinx error comes up:\n\n   ../config-services-agent.rst:70:Error in \"code-block\" directive:\n   1 argument(s) required, 0 supplied.\n\n   This also makes the docs build fail so a \u0027none\u0027 argument is added to\n   the code-block.\n\n3. Cap hacking in test-requirements.txt\n\n   hacking as a linter is in global requirements blacklist and so is\n   not in constraints. Recent release introduced new rules that\n   required fix on master, on stable branches we should rather cap to\n   the version that was in use during this release development.\n\nCloses-Bug: #1856156\nChange-Id: Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2\n(cherry picked from commit 07be7934359aa121761396ba47128904f19cb2e4)\n(cherry picked from commit d1c4ba581085f4978da82a597b1387d60e79ff0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34239888eb661415ed372808304a42339563dad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34239888eb661415ed372808304a42339563dad6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
