)]}'
{"id":"openstack%2Fneutron~698692","triplet_id":"openstack%2Fneutron~master~Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2","project":"openstack/neutron","branch":"master","topic":"bug/1856156","hashtags":[],"change_id":"Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2","subject":"Use constraints for docs tox target","status":"MERGED","created":"2019-12-12 10:36:44.000000000","updated":"2019-12-12 20:22:16.000000000","submitted":"2019-12-12 20:19:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698692-1576181957471-d94f54cc","meta_rev_id":"aa22d4f7215b0a7e0309ced81385ee82677eb2c5","_number":698692,"virtual_id_number":698692,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 20:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-12 16:45:20.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2019-12-12 14:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-12-12 14:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-12-12 13:50:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2019-12-12 13:11:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-12-12 14:44:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:11:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-12 13:50:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-12-12 14:37:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-12 14:44:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-12 16:45:20.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-12 20:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffc5ff851605a78ac846751d92f1b4fcca3b74d3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 10:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ad42c98299a97c5f89d0941e4e5200599baefd6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-12 12:03:12.000000000","message":"Topic set to bug/1856156","accounts_in_message":[],"_revision_number":1},{"id":"b05edeed3ca5393df0fbd7438d87b7e57c6e8039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:50:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a62324054d6548ceae19eaf114a17f32 : SUCCESS in 39m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6de68206846548dfb0d482db23194434 : SUCCESS in 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8468910252e410e929ba13c02eb8dc3 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2791a1b423404ca59bd4a12235eee0c2 : SUCCESS in 21m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d278eed056e54c06bc1afd843a20f7f8 : SUCCESS in 20m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80096c1e2b24e91aed7ff74c5408ad9 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a62e58f678e8498fa46b28e99333377d : SUCCESS in 24m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/70eafad59f934181a1cd817ff10516d6 : SUCCESS in 47m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/076f1a75eb3b44f18ee6cc63751b87f8 : SUCCESS in 1h 05m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e239169f23724ddeb3493cfaa9b18599 : SUCCESS in 1h 15m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c47cb9d792074aeebf852725532b6357 : SUCCESS in 49m 43s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7130832a042943c38f3c409c423e8867 : SUCCESS in 1h 04m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"343b5f258f48581b0f19d191c024f2b67a7dbc57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-12 13:11:50.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"52cc52163526ba7b0e679dc0ee76abeacb12f840","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-12 13:12:40.000000000","message":"Patch Set 1:\n\nTake care to adapt this url for stable branches if you decide to backport these changes\n\nhttps://releases.openstack.org/constraints/upper/train\nhttps://releases.openstack.org/constraints/upper/stein","accounts_in_message":[],"_revision_number":1},{"id":"c8b34b3bd998d78c497746f83233cd12156722eb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-12 13:50:17.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Elod","accounts_in_message":[],"_revision_number":1},{"id":"b40c849ef1b7caffaee7191d96238cd8cd2b4d9c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-12 14:37:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dba141b8349af7f4736b677c548dde0c02f72016","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 14:44:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5322d07b73f1f5820aa9c66eaf0e197fadc6ae39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 14:45:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91ff2128e5b8aac7ac36558b6745107d7a36df9c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 16:45:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698692/1 : SUCCESS in 1h 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d6c983382f946029fb5f729e35faff84c42b09b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 20:19:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b4cada475d5465c8c86f70cb09ab275 : SUCCESS in 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eebf5d805f04b5a934472b54d8232c4 : SUCCESS in 15m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5913a1d1bb1944d39bfef562ea66c514 : SUCCESS in 25m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6d365b21fe7437096a819975f009444 : SUCCESS in 27m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a422462df84e6a8758ddd2f1c236f6 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fab40effead14bb58eb70dea5211d64c : SUCCESS in 26m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c088464faba142978b4d5e30c052b3ee : SUCCESS in 54m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b94b61f0721c426fab0c685c8a180cce : SUCCESS in 1h 07m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a40c34ca69c9456e895857d0997137c7 : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f6a9200dc0cd94e0b6603b7bc0adce1319a4b455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 20:19:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa22d4f7215b0a7e0309ced81385ee82677eb2c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 20:22:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb5655a8e754a88baf360d2abd28d84 : SUCCESS in 2m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/128b54d957074533b13da28a1789b4a2 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07be7934359aa121761396ba47128904f19cb2e4","revisions":{"07be7934359aa121761396ba47128904f19cb2e4":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:36:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/92/698692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/698692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/698692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/698692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/698692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/698692/1"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 10:14:10.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-12 10:14:10.000000000","tz":60},"subject":"Use constraints for docs tox target","message":"Use constraints for docs tox target\n\nIn Stein the docs target started to fail when new release of\nneutron-lib appeared. This is because tox installs neutron and its\nrequirements without any constraints. To fix this both the upper\nconstraints and neutron requirements needs to be added to dependencies\nof docs target.\n\nCloses-Bug: #1856156\nChange-Id: Iea61238f37fdf24c0264f96d104ee0b3b6aec8e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07be7934359aa121761396ba47128904f19cb2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07be7934359aa121761396ba47128904f19cb2e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
