)]}'
{"id":"openstack%2Fneutron~698737","triplet_id":"openstack%2Fneutron~master~I472d2f07f27e9162089d08d416fb02ae8b5544e7","project":"openstack/neutron","branch":"master","topic":"triplestandalone","hashtags":[],"change_id":"I472d2f07f27e9162089d08d416fb02ae8b5544e7","subject":"Move tripleo standalone job to ostempest","status":"MERGED","created":"2019-12-12 15:19:26.000000000","updated":"2020-01-20 08:03:12.000000000","submitted":"2020-01-20 07:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698737-1579507190157-d94f54cc","meta_rev_id":"983224721d787d8e1e2b330f8056ffb5d1ccb6e7","_number":698737,"virtual_id_number":698737,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 07:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-01-20 01:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2020-01-12 10:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-01-20 01:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 20:11:02.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-16 11:59:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-12 10:38:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-20 01:03:00.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-20 07:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41d39d12962289bccf01d4feecf3a96d471de1b0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-12 15:19:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38da2af50c721850e1e12ea7c4a7ac6391e7cb5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:19:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3adfada5401b4af380ae7c8ec9f7ffa9 : SUCCESS in 59m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b36cc8b74c9d477e8e1804a9e8811afe : SUCCESS in 43m 06s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cfd184e27a4b4151abecfff5fbba855c : FAILURE in 1h 45m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/1597165625a34cb4b0c945b7a6428980 : SUCCESS in 1h 05m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9ef2e8affe8c4c6999043769644fba4f : SUCCESS in 1h 17m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29c3a6b40e204cce8962d823e6584af0 : SUCCESS in 55m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8868654c28654390992aad90003c760e : SUCCESS in 39m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d362c541af74f849fd9d6ec45e4bcb6 : SUCCESS in 26m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9fd2a5e9c64d2c8fa844c48d80bd17 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abd493f2f04f42749493fb27698e0216 : SUCCESS in 25m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c398594191c3406cbecaa1e3c022b762 : SUCCESS in 26m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b209140f084d9c896ecb52b1ac9a00 : SUCCESS in 12m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b1157c9946dd4693a476db1e508b36db : SUCCESS in 47m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a322fdef92aa4c24952edad3bb6a16b6 : SUCCESS in 1h 15m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e79b7b27070842b6b49085f820c2e38a : SUCCESS in 2h 07m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4b4d182b2fda48f0b70e0134e9fa4520 : SUCCESS in 1h 22m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/31891aa050184d39b83f44f2f46c495a : SUCCESS in 1h 18m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d190281efe047598be859d847f2bc02 : SUCCESS in 1h 10m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/26ffd68bf0c34955acc88df413f70656 : SUCCESS in 1h 24m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/06b2018d84b1400c8240028acb6f47e3 : SUCCESS in 1h 28m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/79da097223fc4eca815b238bb00fe563 : SUCCESS in 1h 44m 49s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/803af461b4614ba8ba8a9570dc178a32 : SUCCESS in 2h 30m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d254d9bff4f64b9095feeea2170703c4 : SUCCESS in 1h 14m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eb5f1289befb4288a12144109aca623c : SUCCESS in 57m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/afae91523e5d49f9836e0a381487483b : SUCCESS in 2h 18m 21s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/69c3d5f874e24a72b4f92d65320e1819 : SUCCESS in 1h 27m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96cc60136d514064861f46eb1e6ccbb0 : SUCCESS in 1h 03m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cc96191cb82e47358f06358b485b9fe5 : FAILURE in 1h 19m 15s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62d26ec27be344dfa9f919291b10e8e0 : SUCCESS in 52m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e063a6b7c3b46a59dd4ddb21265072d : SUCCESS in 1h 26m 11s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/51741b1ee71c40d8874da126edbc7788 : SUCCESS in 1h 37m 57s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/55f229d83bd34e93b8efb5b23a91ad9c : FAILURE in 1h 17m 45s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8103969c90ae43e8b63ee32b130e4ca3 : FAILURE in 2h 23m 53s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/ca93567abd2d4030b6a8a88cf969bdf4 : FAILURE in 2h 00m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81cb3a08c1b940ae8fd955b660221054b579ff8f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 18:44:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698737/1 : SUCCESS in 1h 50m 00s","accounts_in_message":[],"_revision_number":1},{"id":"755d17c53d1a9e066e7b9dfe2c233e17ed8f546c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-13 17:47:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5f9ffb7751044d341f5615feb93f70017fdc59e8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-13 20:11:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698737/2 : SUCCESS in 2h 00m 22s","accounts_in_message":[],"_revision_number":2},{"id":"68cdf44158999cae8404b6ea1606321b1256e919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 20:23:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b9bfe14661924fb4a58d4ee4604d15dd : SUCCESS in 56m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/447145af56be42adb198f6198915fe13 : SUCCESS in 40m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a7fc4978097146e1afcad6324bd7ae52 : FAILURE in 1h 42m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/3ba69d4dbae74d2ca8238a9c569b0856 : SUCCESS in 1h 19m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/0018d04674544d03b1dad3e1e8eeb185 : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/466866fa95474ac798e8943a22426e2f : SUCCESS in 1h 18m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671f29b41e4947e68a12849d0d0a8561 : SUCCESS in 27m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e3839057b814334bd4ad48ff1bccd8f : SUCCESS in 22m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b368e12d6f438a94a7c0b437999eca : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba72efd01ac4945a158903aa718351b : SUCCESS in 21m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf534379dbae4101829a0d4b90915573 : SUCCESS in 24m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5453a7ae8a45f48321cd7a41da0651 : SUCCESS in 9m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/12e628e4009a429a8a25f64c78f081e2 : SUCCESS in 52m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0dc453d0fcdb442290d539763b81bc69 : SUCCESS in 1h 12m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/078d8a9754ca4b4d8275f5f3e21af7a6 : SUCCESS in 2h 11m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6893214cc810415e9156c0f54a7f40e9 : SUCCESS in 1h 16m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6afbc7e0c013490cb33ec9399321104f : SUCCESS in 1h 25m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f94fb7e559f42a6a3464b863640da9a : FAILURE in 1h 08m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9c58489cb90643388bec49766e7d8667 : SUCCESS in 1h 14m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6dd975894d34f4f994db95533a6b754 : SUCCESS in 1h 23m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4a85557b05a84fb1b2f7bf7a1a8a220d : SUCCESS in 1h 50m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/61d8b8c14e484b94b6e6ccf132068cdb : SUCCESS in 1h 44m 23s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba04243455c14bd7ba67191b14869f84 : SUCCESS in 1h 15m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e068965f72a34bfd888d5cfc991c44d6 : SUCCESS in 58m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f86557f0fec74d34a57813e5da853fcb : SUCCESS in 2h 25m 59s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3aea40f1fc7042d69da762d5a7efcfac : SUCCESS in 1h 45m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5ef36e270dd4088afd622f896fc5684 : SUCCESS in 1h 05m 03s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0d6e0365182a4978a3d898150784191f : SUCCESS in 1h 27m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/adc439d01c9c4b3b8f74c74308d31abd : TIMED_OUT in 2h 20m 58s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aff1166901884d48aeac0b069c58a9d4 : FAILURE in 1h 38m 39s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/95384e461bbd4c1383460664cbada0ac : SUCCESS in 1h 53m 45s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/206281b3253c4df0be18f303e7f31abe : FAILURE in 1h 11m 10s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/43381c4e0e1f47d2865a86925641032c : FAILURE in 2h 20m 48s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3d4463ec7639446fb0dc05f1006c0df5 : FAILURE in 2h 04m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a370e610e6326e7c1f53c891d757e968709723a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-16 05:30:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1b49fca1fb4caeed74afce9054fb5128a28f75da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 08:45:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/212f757da6574b9a831a9f6fffdb727e : SUCCESS in 50m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/412f4b734ddb4affa027baade6131fad : SUCCESS in 44m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2ee53caf329d45838211196668ccdba5 : POST_FAILURE in 1h 47m 12s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/001817e9c4b34f6982876e8140a7eb14 : SUCCESS in 1h 02m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8426002350454501b0af619a4385f11f : SUCCESS in 59m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/855b4f92b97d47058fd9cbcabd02f5bd : SUCCESS in 53m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9a6888f77524affa5b62d02ae33b41b : FAILURE in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7262d4fd4d2462ab07c584f5c770905 : SUCCESS in 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b76681804b4b7ea6e5a62d3f8e8eba : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c775ee72b984022b45b5c3dde7e6572 : FAILURE in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8a82aa75fb44a9190ef1b3bfd5e7c6f : FAILURE in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1057c9b5047045f9afecc14c26384e1a : SUCCESS in 7m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/72042f20aad949fd99d36a0ad60a7221 : FAILURE in 46m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/413dade5f81e40f28595782b9c2200e6 : SUCCESS in 1h 12m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/0394f6f89cb14fcbaf42995f46d09b4a : SUCCESS in 2h 02m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/21f3eac5b369410eac44c75048bad32e : SUCCESS in 1h 16m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8af9ba209ba346ab9235018e35680b2b : SUCCESS in 1h 16m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/048e25a7772c4fb5be2099ceeec17b37 : SUCCESS in 1h 01m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25353bd7073546d0ada985606c2d223b : SUCCESS in 1h 22m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f88de0a69e7241a98edb9ac3366ceecb : SUCCESS in 1h 16m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0dedd6ea5c864b1499a2afcb201e0b87 : SUCCESS in 1h 51m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9187076178e4c198ab74618b49be608 : SUCCESS in 1h 54m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91e1a4e1dfc2412c8721a27dbabdb0a2 : RETRY_LIMIT in 15m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a6819fbbb07a4731a83427ed84f11b17 : SUCCESS in 49m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7d79fa0daa54f89b3ea15c01d6a7578 : SUCCESS in 2h 22m 10s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/87c21a396ea3469aa8dcf09d71fc2048 : FAILURE in 17m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a4a9697843a4853a9223d41a68b4c61 : SUCCESS in 1h 16m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2d11d81fda704b0fa95a9811ffd3b4dd : SUCCESS in 1h 12m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b49eb01207564699b9a294f2ad122108 : SUCCESS in 53m 20s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68dc0c9992004afd9062067b182c30a6 : SUCCESS in 1h 14m 29s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/088ab660f64342dca8a15b13e0d62db9 : SUCCESS in 1h 22m 06s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f151e2fdebe041ca83c0938acedb069a : FAILURE in 1h 15m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1db87264f3054d91902d021d5319b6d6 : FAILURE in 3h 06m 06s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9dd0ff99a4ff408094d358528946dfdb : FAILURE in 1h 51m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3ce4e6957da4c3d4dc6794e7a9d4473656dc690","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-16 11:40:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d4593601fefb45842cfbe25427c36ee71372c4e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-16 11:59:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50129433931e4f7ec9a6e3063380564de7040a95","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-16 12:24:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ecd634f9a99a9b55732f50f1d54479c18f9bad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 15:12:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8857591baebc42dbbdf9b11b9fbe65fb : SUCCESS in 1h 00m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e2b22b9109b419d9f7dd64f872035cb : SUCCESS in 54m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2cd50eb1185d4f01b34b3bb002cc9d99 : FAILURE in 1h 32m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/96af3d0eebf043289f948efb43e66acd : SUCCESS in 1h 19m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4a3e55389f6548c2a8bca02c8fce55ac : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47e674cbdfa84bbab3ff33b7a3b99c5d : SUCCESS in 1h 08m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c3a475fe0e40e8877f7c242fa74c0f : FAILURE in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6c8ceb676a543c09ace71c65f5892ff : SUCCESS in 23m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ffd74ff95bc4e07a439faab2c207825 : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01ba813c5e8b4c69b95239858dbf4ecd : FAILURE in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60c1674e22a34f06859528683e33f325 : FAILURE in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be657c07d78e4bb88aafe33259d7df3f : SUCCESS in 8m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/78ba1ec5c6af42589625675bbafe7a49 : SUCCESS in 51m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e02b6caa686b4ef8884dec7586bf1166 : SUCCESS in 1h 06m 33s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cd562ec231a541298256afeb6178179d : SUCCESS in 1h 52m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2f2d9ef4f9d54ab68436dca72e2106e1 : SUCCESS in 1h 35m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bf38c8a3abbb42fe96e96da5fbc44149 : SUCCESS in 1h 15m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f1d8254a79c41d4a764ca2348a6fd8c : FAILURE in 1h 40m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/54141691daaf45078c195e41e9c08b93 : SUCCESS in 1h 21m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/66c06ab4ca044693ba42972e60ae0f02 : SUCCESS in 1h 24m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9e75435da9bf4000b4b84d94fed97a46 : SUCCESS in 1h 47m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d12519ba5a99421b866ae66cad2cd9b9 : SUCCESS in 1h 49m 04s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/757d7d2bb9fd4f9f930f59dd2299f60d : SUCCESS in 1h 45m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c51d6b8d6e7448368135a4524af8d9dc : SUCCESS in 1h 01m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cbe71ad7b8d94d389d27b97c6ff457f0 : SUCCESS in 2h 25m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/e15a787ce5e54ad999c12f1c0341c020 : FAILURE in 24m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aeb0b94aca9475bbfc1ae819e870392 : FAILURE in 1h 18m 21s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f1981dc2fe3846f388f6561cbfac8819 : SUCCESS in 1h 11m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b06d5aa51414f81b3503880738c30de : FAILURE in 54m 21s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/953cc55ffd784503b9a311766fb0c301 : SUCCESS in 1h 12m 29s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1018f5ca9fd14216bdc891fda05ff25d : SUCCESS in 1h 10m 58s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ded7213a4db54473a3e86bd7ce71c4d3 : FAILURE in 1h 44m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9435673b9fcc42b9a0d486f3e0cd7f83 : FAILURE in 2h 33m 26s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3ef4a094dc40424f98a1d2c242611048 : FAILURE in 1h 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8b803ec78aa2bf22f1f440573b34a4246264497d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-17 12:48:06.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ffe8533838d85ce88d270a0e6a168d58f9ef038a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 16:00:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/65508862ecf347b196e5efe2b64aec80 : SUCCESS in 1h 05m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b509f79c0614268b44adef598bc9e76 : SUCCESS in 32m 25s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/78639269e88547bc8aa14e201d086533 : FAILURE in 1h 27m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/52715cef31b747b6bc7e39ec34d96e6c : SUCCESS in 1h 15m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d9ed3b04b9704595887c9351edb73b13 : SUCCESS in 1h 03m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01af2043a47a42e7aae25fe564b6a82e : SUCCESS in 1h 11m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5acf192d0644694ace52b0eceba938c : FAILURE in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c37649ee5ed3452c8d3c0f0a5adafb7b : SUCCESS in 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca7a72e536f5444bb6376752fc600057 : FAILURE in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5c009d2da2f4a7ba00a91fa3ffb786d : FAILURE in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9d144903d9a4f7685edc16332787156 : FAILURE in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e46c863eef4609a3e64e73fb558330 : SUCCESS in 7m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b2658dbf314547ab949afe1e97309b9a : SUCCESS in 49m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1bcf97ebb09d420e9347ec76d54b846b : SUCCESS in 1h 05m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c49fb652bc4a45659d501ea52e9fedc2 : SUCCESS in 1h 59m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6f67fcb36f624dc689aeaf0666392369 : SUCCESS in 1h 39m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b761f47bff7a4d89a88e4d8bdbfe576e : SUCCESS in 1h 11m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5c8c8e950a24c31aa8574f71dea1778 : SUCCESS in 1h 05m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ee9968278b2845ee9be2c54d14ebaa2d : SUCCESS in 1h 21m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/caf7afb869824693affd65554f87b622 : SUCCESS in 1h 15m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/66c36aa9f890410790af7b64fc85dc5a : SUCCESS in 1h 36m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a2f69245f04463da6cdc65e0d689cef : SUCCESS in 2h 10m 38s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2a02ba953e6247d696a3a373c82a0586 : SUCCESS in 1h 36m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0730e008003c4e3784650957db9d7c98 : SUCCESS in 1h 09m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26481c3329ed4303964d2c82342a3bc1 : SUCCESS in 2h 11m 45s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/85cd2de640f441ea8bca0c7851630899 : FAILURE in 24m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23275b8c7cbc4fcfa60ef90a1dc950f6 : SUCCESS in 1h 20m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/129e40339547463ab344b66f70767678 : SUCCESS in 1h 30m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8c9e22b47bb4bbd83916ca8451e49ab : TIMED_OUT in 2h 20m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/359e2ed468d14780abfabc3e1044aff3 : SUCCESS in 1h 04m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/74f0b5b46e4d49e3842ab8ef9991bed4 : SUCCESS in 1h 35m 11s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b4ff1846cba14f8f965a78c7918914ab : FAILURE in 1h 47m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7d5fef1f345c4af9b5c53f15ce1bdf95 : TIMED_OUT in 3h 06m 10s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c2350b0eafa94da290cac408a96bf91a : FAILURE in 2h 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fc6bb6ab1479aa5aebff3fbc6ce6afc88f8ae747","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-12-19 06:51:03.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"8c9a17c9bebe4ae6928af9ce5ea7e5292f320602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 10:28:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3266578fe45242bab400b786961a4cce : SUCCESS in 46m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7a47c88a7d4446385d9066ea9168d8a : SUCCESS in 36m 16s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/a5471dc0127f41bbb020a09012338a70 : FAILURE in 1h 32m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8c51b9b0df8947f9b35aa20d0837d0f9 : SUCCESS in 1h 04m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cf6e680579794fcaa92760edcd6816e0 : SUCCESS in 1h 02m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a901c622c3f54b25b16614d5727214ad : SUCCESS in 1h 02m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d77e7421841647b4ad29e7b978890a98 : SUCCESS in 32m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3214e27960de45f48e1ede145ae55d94 : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e6263e0862403385e81117d4ac8e9e : SUCCESS in 15m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d407bb1955304d68b1a3b6100495e19d : SUCCESS in 25m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/611d485fcbda40088b4fb50e37639883 : SUCCESS in 30m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74fea883afc47dd867f576a57bc6893 : SUCCESS in 8m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2b0c6486f9b24fb5bec1c0aacdcc949a : SUCCESS in 52m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/39e214541f7a40e8989498dd808bfb2b : SUCCESS in 1h 12m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1bcc3f392cf24e538757c1e1442d9308 : SUCCESS in 2h 07m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/70f448ac95ea4617acbe47c76bf3a5ed : SUCCESS in 1h 16m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/08c065a220b644c294ac7263ccc3fe38 : SUCCESS in 1h 16m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f17526551fc4b1ab128e01c1746a66a : SUCCESS in 1h 08m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/98f9cd06ba0a48f08aa98336e30099c5 : SUCCESS in 1h 16m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5906978f03242a2a78021ccd9c23b86 : SUCCESS in 1h 36m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6130ed8fdcb04dc48bc9056a118be7fc : SUCCESS in 2h 09m 53s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/49b3d3021d5b442599ea8ed040fbb723 : FAILURE in 1h 40m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d143651239d04ce18eb65a08f270a90e : SUCCESS in 1h 14m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e210cb960abb4b94b8017e3b72c5fb4b : SUCCESS in 56m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aff37af1e6d34fc29cdd8a0123a0b6ea : SUCCESS in 2h 20m 16s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c0ae9ee9e36e4395bfdda512ca6c3363 : FAILURE in 20m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd9294c25a8d46249acf2cf47ff15183 : SUCCESS in 1h 00m 31s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/dcaf2476986f467897d1a93262b5e13d : SUCCESS in 1h 44m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/146c8d44c340456ba79f4678f99b3d29 : SUCCESS in 1h 04m 07s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/291ed577b9ab4eb5bcd0df9e64ff5ca1 : SUCCESS in 1h 19m 50s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/839b738ac71e49c8adcf40f8b8a9412a : SUCCESS in 1h 35m 10s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f30a42a8f647488dbc954e0d5177f443 : FAILURE in 1h 28m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb630c9c94f04d1ab01515a604db0000 : FAILURE in 2h 51m 46s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e6c95755b23c4d358974b7efcb475ed1 : FAILURE in 2h 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3de744c8f31b274963b17537ae45c5dfff9e3817","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-12 10:38:04.000000000","message":"Patch Set 6: Code-Review+2\n\nI\u0027m ok with this change. It would be good to see this job passing first but I know that Centos 7 is using python 2.7 still and Neutron can\u0027t run on it anymore so the issue is not related with this patch directly.","accounts_in_message":[],"_revision_number":6},{"id":"925899ac0300e053e94f8d0f286db53cce1474ec","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-20 01:03:00.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e075a609871f13f99721bb393c752596ec5fc8f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 01:04:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5e98f22693d1efea529fa7e88938c21dfeec9c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 07:59:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/97c8600da7444948bfae5c8606da2089 : SUCCESS in 52m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8a0db6734eee489b8e1f36f1c5bd94c6 : SUCCESS in 1h 14m 26s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f0c8aed803ec4c338658ee89a2a4ec55 : SUCCESS in 1h 11m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1c0d7cc9f84452c8463e52bb8c9d6b7 : SUCCESS in 1h 06m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbec92283947471595aef6ef9869b87c : SUCCESS in 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d2f51ef85744dabcdb8fe9084eb8f0 : SUCCESS in 14m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16602d2161f54c608729f02374a556ac : SUCCESS in 25m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f80b66dfff9c4ed999e52eed3cd626fc : SUCCESS in 32m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8824b6d730b644ba93f10b70cd9acd99 : SUCCESS in 7m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f30cc45a814d47869511f04efd6e7b1c : SUCCESS in 51m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2300ff7a90a94aa28190d888d04f9b67 : SUCCESS in 1h 17m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/992684a50a274f169d01e466bcc842a6 : SUCCESS in 1h 16m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3f8367f061884a28a8559f9dfdf0f2f4 : SUCCESS in 1h 21m 40s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4bcb7fda959f43b09419e82577879af4 : SUCCESS in 1h 19m 09s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5e11067e084a4446907722bd68b77515 : SUCCESS in 1h 14m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d53a414bd7e04f009a645ff305ecb954 : SUCCESS in 1h 27m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3ca3a9db65214c8886237fb51d9e62ba : SUCCESS in 1h 17m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf9f12fa97fb491cb605f51e392a0380 : SUCCESS in 2h 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/568eef6a0c224e60982b51d3b5a0c07f : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":6},{"id":"b841ebe5cf65277f957389aaeb24baaabf0d54a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 07:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"f7e6408e53c0db1c78f377115f615207b3f0e0c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 08:03:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8423c43371b4255aefa5aa883051a88 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"52787dd4a40d2dbc04f4df6e8683376a8aed74c1","revisions":{"0f5f050c703d9563fc4a8134cdabe762218f804d":{"kind":"REWORK","_number":1,"created":"2019-12-12 15:19:26.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/698737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/698737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/698737/1"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:16:41.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:19:16.000000000","tz":330},"subject":"Move tripleo standalone job to ostempest","message":"Move tripleo standalone job to ostempest\n\nSince validate-tempest is deprecated in favor of ostempest\nunified ansible role. All the tripleo jobs are already moved\nto that, having a unified job will help to get consistent\nresult and avoid issues.\n\nChange-Id: I472d2f07f27e9162089d08d416fb02ae8b5544e7\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f5f050c703d9563fc4a8134cdabe762218f804d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f5f050c703d9563fc4a8134cdabe762218f804d"}]},"branch":"refs/heads/master"},"7c6ae09df130c6bdfb1c68d9e38d03856ecd32b6":{"kind":"REWORK","_number":2,"created":"2019-12-13 17:47:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/698737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/698737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/698737/2"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:16:41.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-13 17:47:10.000000000","tz":0},"subject":"Move tripleo standalone job to ostempest","message":"Move tripleo standalone job to ostempest\n\nSince validate-tempest is deprecated in favor of ostempest\nunified ansible role. All the tripleo jobs are already moved\nto that, having a unified job will help to get consistent\nresult and avoid issues.\n\nDepends-On: https://review.opendev.org/#/c/698589/\n\nChange-Id: I472d2f07f27e9162089d08d416fb02ae8b5544e7\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c6ae09df130c6bdfb1c68d9e38d03856ecd32b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c6ae09df130c6bdfb1c68d9e38d03856ecd32b6"}]},"branch":"refs/heads/master"},"ba31d123b5e06abc71de9c102f99a8b5fa58c4f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-16 05:30:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/698737/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/698737/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/698737/3"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:16:41.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-16 05:30:36.000000000","tz":0},"subject":"Move tripleo standalone job to ostempest","message":"Move tripleo standalone job to ostempest\n\nSince validate-tempest is deprecated in favor of ostempest\nunified ansible role. All the tripleo jobs are already moved\nto that, having a unified job will help to get consistent\nresult and avoid issues.\n\nDepends-On: https://review.opendev.org/#/c/698589/\n\nChange-Id: I472d2f07f27e9162089d08d416fb02ae8b5544e7\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba31d123b5e06abc71de9c102f99a8b5fa58c4f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba31d123b5e06abc71de9c102f99a8b5fa58c4f1"}]},"branch":"refs/heads/master"},"baac14b556cb07c247afc6ec958f4a651805da2e":{"kind":"REWORK","_number":4,"created":"2019-12-16 11:40:38.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/698737/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/698737/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/698737/4"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:16:41.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-16 11:39:58.000000000","tz":0},"subject":"Move tripleo standalone job to ostempest","message":"Move tripleo standalone job to ostempest\n\nSince validate-tempest is deprecated in favor of ostempest\nunified ansible role. All the tripleo jobs are already moved\nto that, having a unified job will help to get consistent\nresult and avoid issues.\n\nDepends-On: https://review.opendev.org/#/c/698589/\n\nChange-Id: I472d2f07f27e9162089d08d416fb02ae8b5544e7\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/baac14b556cb07c247afc6ec958f4a651805da2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/baac14b556cb07c247afc6ec958f4a651805da2e"}]},"branch":"refs/heads/master"},"11a2b022332da0b7a1ba6ecabbaf3f517512dc89":{"kind":"REWORK","_number":5,"created":"2019-12-17 12:48:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/698737/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/698737/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/698737/5"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:16:41.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-17 12:48:02.000000000","tz":0},"subject":"Move tripleo standalone job to ostempest","message":"Move tripleo standalone job to ostempest\n\nSince validate-tempest is deprecated in favor of ostempest\nunified ansible role. All the tripleo jobs are already moved\nto that, having a unified job will help to get consistent\nresult and avoid issues.\n\nDepends-On: https://review.opendev.org/#/c/698589/\n\nChange-Id: I472d2f07f27e9162089d08d416fb02ae8b5544e7\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11a2b022332da0b7a1ba6ecabbaf3f517512dc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11a2b022332da0b7a1ba6ecabbaf3f517512dc89"}]},"branch":"refs/heads/master"},"52787dd4a40d2dbc04f4df6e8683376a8aed74c1":{"kind":"REWORK","_number":6,"created":"2019-12-19 06:51:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/37/698737/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/698737/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/698737/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/698737/6"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-12 15:16:41.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-12-19 06:50:49.000000000","tz":0},"subject":"Move tripleo standalone job to ostempest","message":"Move tripleo standalone job to ostempest\n\nSince validate-tempest is deprecated in favor of ostempest\nunified ansible role. All the tripleo jobs are already moved\nto that, having a unified job will help to get consistent\nresult and avoid issues.\n\nDepends-On: https://review.opendev.org/#/c/698589/\n\nChange-Id: I472d2f07f27e9162089d08d416fb02ae8b5544e7\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52787dd4a40d2dbc04f4df6e8683376a8aed74c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52787dd4a40d2dbc04f4df6e8683376a8aed74c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
