)]}'
{"id":"openstack%2Fneutron~698783","triplet_id":"openstack%2Fneutron~master~I4e826c8b9fa764b1820adacc3427934dc393c0bc","project":"openstack/neutron","branch":"master","topic":"bug/1856152","hashtags":[],"change_id":"I4e826c8b9fa764b1820adacc3427934dc393c0bc","subject":"Deprecate config option \"ovs_integration_bridge\"","status":"MERGED","created":"2019-12-12 18:14:54.000000000","updated":"2020-02-18 20:18:34.000000000","submitted":"2020-02-18 20:15:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698783-1582056924651-a5ae4dff","meta_rev_id":"a0913c077cda9b1ce0c37bab92025961f58c8f99","_number":698783,"virtual_id_number":698783,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 20:15:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-18 12:35:01.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":2,"date":"2020-02-18 10:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-02-18 14:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-02-18 14:35:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 19:01:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-20 19:21:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-20 23:24:46.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-02-18 10:42:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-02-18 12:35:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-18 14:35:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-18 20:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d69f5bab447f802d7c088d395a0dbe7da6e70f54","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-12 18:14:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f6f58d1318de4500b7a2beeeaea2c7695caab64","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-12 19:25:08.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/83/698783/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"10b73e34d5da37833dfd4e8a347bbea4590ea21d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-12 20:02:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698783/1 : SUCCESS in 1h 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"68c7518b5e6260e60e57e449231a47ac10c91cc2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-12 20:30:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/1/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/1/Neutron-ML2-Sriov : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3e65f279f50018139ed8b1515331385df59e458e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 21:30:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b21f2101f5784264bcc7c92083226e8d : SUCCESS in 57m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b527ad4a11f434283c991f3efad9a08 : SUCCESS in 34m 43s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/658d941474644536855d8984e6885f97 : FAILURE in 1h 48m 26s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/84f31f4292bd49e5b54bd7f5b69f115d : SUCCESS in 1h 04m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8d97305c063f4855a2048c562089a799 : SUCCESS in 1h 02m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d902de96c3f45f5a122d566bdff71e8 : SUCCESS in 1h 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7806f34dab9f44299d2c0d16fcedd491 : FAILURE in 30m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f2ac7a9b4d64c75a3e47198beb83a0d : FAILURE in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1ecea10a5044fa9b88190f6652df84 : SUCCESS in 15m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a42f7e3389a4704bf4a580a26cf6ee7 : FAILURE in 25m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75b28d50dd3049d89409e3c512f4abc9 : FAILURE in 26m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd425dd57a949d490dcc440c330e685 : SUCCESS in 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/638864e09926479fb1d805990b984900 : FAILURE in 1h 06m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/04e805e5c70648faae7af1ec81d2062a : FAILURE in 1h 13m 36s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/23b76d9ea64b4dc8bad60f48782e9cf1 : SUCCESS in 1h 51m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/57963b31616249b4a95cb29ae3475381 : SUCCESS in 1h 24m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4177a4356d9c4eb484510c93cc49bb92 : SUCCESS in 1h 14m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da678fa276db4ba3aa579205f94b7afd : SUCCESS in 1h 07m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11d37f500fac42c7ba9bbbd24a51a315 : SUCCESS in 1h 20m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e046b3c6db5f48488549bee5923adf7d : SUCCESS in 1h 36m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/985508c82f784b7cbca01d4ce33423f0 : FAILURE in 1h 40m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e826281fac9e4c53b716a9087ae1a09d : SUCCESS in 2h 22m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f11d595aab094fceb3a6f682b3b79e30 : SUCCESS in 1h 34m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/48b6ca1da10a414e9d11fa0b8b5cc677 : SUCCESS in 1h 00m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7015aa954a7c484f86c632304da2c25a : SUCCESS in 2h 52m 36s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/083bde8d69384b7a97e58935a553e655 : SUCCESS in 1h 33m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cc78ef1d3424c1998c7a2550a5745cd : SUCCESS in 1h 27m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e37185372e624a959a783e62932dc61c : SUCCESS in 1h 26m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d8d67f7f707b4d9189bf9fb0dcdac62a : FAILURE in 58m 58s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d1483b4d5384d6f880d0eceffa400e4 : SUCCESS in 1h 15m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/918d6b81ac5b43ebb8ef4de37dfe4338 : SUCCESS in 1h 25m 56s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a205def873aa439aa47867b7b6b63da0 : FAILURE in 1h 34m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/46f2391cd5a74807b1045271883f302d : FAILURE in 2h 30m 15s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1ded558993a24d7abd10358134be16e8 : FAILURE in 2h 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc7a195aada66205fb35bd365428aa7ca31b8ebc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-13 17:04:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff4384b5f0a14f369537f173c3f18e22778ebf73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-13 18:19:41.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/2/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/2/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"9011b74dc3ff20771bee7bfdb3b4d6092b3a7587","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-13 18:55:50.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/698783/2 : FAILURE in 1h 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a0b53b79a7152774c9a116eedd9bbd2604ca2598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 19:50:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a7d5eee6bce948b88ff3c7124c1ed4ee : SUCCESS in 56m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/10d6d2325cff4feb95984c9a70aa8075 : SUCCESS in 30m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3df148b99dfc47abbdce567d01ab2dd1 : FAILURE in 1h 34m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/92d46ce00dda42bdaa2f66db1d968e1c : SUCCESS in 1h 07m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2dd9549d561d4e78ae0cc0a4bce9ea3b : SUCCESS in 54m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/06762624b26e4cd8a2b90626d527db07 : SUCCESS in 59m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/866bab3cbadf4efab38ab1d12ac527af : SUCCESS in 27m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb64f69ac44c4dccad8d0eccf0dbba62 : SUCCESS in 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24310170701e47ee8376a2303f3305ae : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810164e2613b479aa2705ed0c0f8de9d : SUCCESS in 25m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/116ab2a0145648af8008cd08129bd166 : SUCCESS in 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33eed0fab79a44b6a9bf04e0ffaad56e : SUCCESS in 11m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c42fea4f17f442d80b8a08a9f0a17d9 : SUCCESS in 47m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/0742a54632d742d0983dd2ce691f0365 : SUCCESS in 1h 08m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/9f44499b49634e8ba403138e62d0a78b : SUCCESS in 1h 44m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5df636c12a2b440590ac2016c693fa80 : SUCCESS in 1h 17m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8fa0e698f1884af085b49b82c408de5f : SUCCESS in 1h 12m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3e4703548a6400dbd71772756b5c2b5 : SUCCESS in 1h 06m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c116d4c18e1642db8e9768f8a3c7eec8 : SUCCESS in 1h 09m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8bcb6c6cf36748efa119f3ac95d94c3e : SUCCESS in 1h 06m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ed653cd20e247bfaeafe140781e26da : SUCCESS in 1h 36m 52s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c49f327532f84351be4cb30733bf92c8 : SUCCESS in 1h 47m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/091673b98f2b47f182778ee9ed3b32bc : SUCCESS in 1h 20m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c7fc030f62f6467dbc1b05e7a399c607 : SUCCESS in 55m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a2009fc8c134440a8fd2b96ed009260 : SUCCESS in 2h 18m 25s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4ff54dca8d7d41af843ddc50982d6496 : SUCCESS in 1h 12m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95bbac0b1ba40078481cdd4da6a6bfa : SUCCESS in 1h 00m 23s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f71f53879dba46419a0cb0471cdf2d20 : SUCCESS in 1h 12m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/815787b579644b6baf6a2d9dac2065f2 : SUCCESS in 48m 37s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e0a62383da44a6ea2dd2487e4889acc : SUCCESS in 1h 05m 09s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/30788ff71dee495884d4a8936319d268 : SUCCESS in 1h 18m 35s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e02da07ecb34457880405e500d7e853f : FAILURE in 1h 28m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/af554b8253604ae1b6a976c8ed41990f : FAILURE in 2h 40m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9a07a23a024a4335a5fa5f98371dda73 : FAILURE in 1h 49m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d88c79e5fb0fe3b0f5c96cff3739b667e25fe780","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-16 09:25:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nwe need release note for that too","accounts_in_message":[],"_revision_number":2},{"id":"f3f65155db78bd7eb946606616c0d7251a6ba43b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-16 15:08:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e04a7563be71d952f60cc6162019eeb34e66dd70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-16 16:53:42.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/3/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/3/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c341d7124f5eab16b276dcb902766c9f5b12fe5d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-16 16:55:01.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/83/698783/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"6687eeef52e6e55d2c149dbe82c78f43648365c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 18:04:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7c0732d089da4bfa9a6524bcfe0f073e : SUCCESS in 47m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ae95935db2e4b979673f1a58cb214b3 : SUCCESS in 38m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/9f111724c462450487f324049c00fe91 : FAILURE in 1h 36m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2be422a2eeae4f8aaf5f032f3b05f845 : SUCCESS in 1h 13m 32s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5c85fa1fc73a4ab8b309ff3610d7cae7 : SUCCESS in 1h 34m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e7fb640fdd134d7194056f57a5387f5e : SUCCESS in 50m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa405bca748544a598f8ecb10867207a : FAILURE in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a814548b1eb43c9a7d0de947a25a98d : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9551c1f1b14caa83f21b60e8b57b13 : FAILURE in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d29d6271e2c412fbc6746a6d40997e2 : FAILURE in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c759cc47b954f34aad4a84a57935298 : FAILURE in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c374f691cfea4af8a26924f0e58d630a : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/157ba118be35400e8e4983091cda7f9f : SUCCESS in 15m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/326b2d89b60b47b99127fe2943fa8f40 : SUCCESS in 50m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b8c394d99d224783a4065ba1045a1c9a : SUCCESS in 1h 09m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/70439b2a2aca4921aa6f8392a33770ea : SUCCESS in 2h 20m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e85c94a2d6894394a80f968dc6eb6621 : SUCCESS in 1h 19m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cfc9a2e1b1b547fa815c259a8410510b : SUCCESS in 1h 30m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2fc24584e4c4e23955c32e6742d4b38 : SUCCESS in 1h 06m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/128b1d993402402e9ee9c2a0c285a26f : SUCCESS in 1h 05m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3279cf27d85947209017bfb9aa083c78 : SUCCESS in 1h 10m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8cd1af62d86a437b85c2d555145e84d2 : SUCCESS in 1h 53m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c9cade749c20409db8dd14450699d0f6 : FAILURE in 1h 54m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d54fefd41a4f469ca944af6290066e83 : SUCCESS in 1h 33m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e133515d039841b288972779419e2237 : SUCCESS in 1h 06m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8c76585ae0644f49f97ab4a311c6168 : SUCCESS in 2h 35m 03s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/fe541c9d894a4cc3b83c4378359c0c4b : FAILURE in 32m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbb9c1048e042e1b9c0e2cb81b7cfa4 : SUCCESS in 1h 28m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/86da507ba6b64750bb32e3f935d2ef2f : SUCCESS in 1h 19m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c7e2479b7344cad8738df2540db8734 : SUCCESS in 47m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f56c291affbb4cada156cb045b02f969 : SUCCESS in 1h 18m 40s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b463e225aa74a9da2074e57b4285618 : SUCCESS in 1h 54m 18s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f5826592cfb54753a983da3637caa7f7 : FAILURE in 2h 19m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/def9cb22490e46d1a83f72151c310351 : FAILURE in 2h 48m 13s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/61e0619c7762450c937bf555478154cf : FAILURE in 2h 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fec608c62e8515a643ca342fb8d62731a2603c14","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-18 13:39:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"48b94224a2f85fd5aa15b8d30d6afc4d2fcaec3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-18 14:37:05.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/4/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/4/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"05949c99dcf26bfdf2d669010d82952e40408bf4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-18 15:08:53.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/83/698783/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"31c7e2421d9285138ffd9650e1c8644b3fe7c8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:32:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1f0a23b10da044afb66de25548972939 : SUCCESS in 1h 04m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65a1acd40cee4959b23120043f238228 : SUCCESS in 51m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/5012de8ccc7b4ab08013dedd4d691a37 : FAILURE in 1h 55m 47s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/18159e906af24de29bea89ad6b1bcc37 : SUCCESS in 1h 13m 21s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/00d0ac2215274d499f9728c8427ea99d : SUCCESS in 1h 13m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de55752b04824f8b8791fba3aa1bda3c : SUCCESS in 59m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/212f171fb27c4df8996c8c98ddc8c168 : SUCCESS in 30m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2127c2c59e894993aded126351772d63 : SUCCESS in 24m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec54211f587417087489609660ffeac : SUCCESS in 16m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a1d23eebb5b42acbc593bf25116bee1 : SUCCESS in 22m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56e15e323af24c5b8ed9a0db5cf8b13b : SUCCESS in 24m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82316943754e463a9ef7da6457509b58 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ce3df824fcf43a59bcbdafd8b352d59 : SUCCESS in 19m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/658a32c0d2d64a3f948aab30475add52 : FAILURE in 59m 51s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6ee87490871c42119436dba63537bd1c : SUCCESS in 1h 04m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/98d03eec5667471eb0264d379e302a63 : SUCCESS in 1h 53m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dd2eef878ff94135917e9ef3725fab4a : FAILURE in 1h 11m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/202c92ab6501495e82d2c6cde710b129 : SUCCESS in 1h 15m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea3065d0ad5e4f149e9ad3cf35261976 : SUCCESS in 1h 25m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/94dee8224d804ebead0da69b4f30e096 : SUCCESS in 1h 15m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/17714881861c4deca794b2cf38f29cad : SUCCESS in 1h 25m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c62f59310564226ad7c67cd3de54869 : SUCCESS in 1h 48m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a99703b783e246e3a37817e190be5176 : SUCCESS in 1h 54m 49s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/65a15f686d98461f93dd2f344e8c9c2f : SUCCESS in 1h 45m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8a2738ec597248f792eec12331ddbe73 : SUCCESS in 1h 02m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4cc8ab58c97416dbf7459a61162f21b : SUCCESS in 2h 50m 57s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9c42de3fe45c42b8a4c23cc76f3481c6 : FAILURE in 22m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dd602e91b804bb0b39e692f6526c453 : SUCCESS in 1h 03m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e7cee48d430e40c089de52eb57d75c1b : SUCCESS in 1h 15m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e906ca883d8454b88241f992df74dea : SUCCESS in 46m 36s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2a73a4970084ae693465c10d8d8c9a1 : SUCCESS in 1h 06m 33s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7500c0108947469c8a83614685f0732c : SUCCESS in 1h 33m 22s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/035c0de21bb142ef87b7992c72465819 : FAILURE in 1h 41m 25s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9fe5b659893e40b4a5b5397d411cbb10 : FAILURE in 2h 47m 01s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5395ecb48e7f42598cea745cccdc2164 : FAILURE in 2h 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"636db22473e737cdcbd716dafa18efe0d2c5f9c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-20 18:09:49.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5b4d50a35a5a30d4315c28f4a9889b8b2c2bc926","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-20 19:01:06.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/5/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/83/698783/5/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"3436fe123f2938b86ab18610798ed9f68de43503","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-20 19:21:34.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/83/698783/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5cb9e3b035133dde20fc018f43b0456b6ab3a361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:19:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cb6c420674f24eddb18a60b73619a2aa : SUCCESS in 1h 12m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b42ab29e8d75471891c78927b90a702e : SUCCESS in 34m 14s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3128b2a003d94ef393ce2869363bcd39 : FAILURE in 1h 21m 04s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9c32828af6714228b5fb3c863cae9c62 : SUCCESS in 1h 02m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/ef01bde2459c48489bb57c6516392619 : SUCCESS in 57m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/635c02d038294da7a1f3ae3d5b006b6a : SUCCESS in 55m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4771e9a66c3b47fea1a38c3bac6141fc : SUCCESS in 27m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62ffed1dcc434493afe9b80bd73a6d15 : SUCCESS in 26m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a17331e02444a887a2a9fc9e258beb : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83a591f2956d4d7c8eb848958a69f2fe : SUCCESS in 24m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5476b9d73b3a47e0bec3a364b8533f21 : SUCCESS in 28m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b482048f6ac4a89bd5ae03f549399aa : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f27cc12b27984757a6c2657934b428c2 : SUCCESS in 18m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8289e730db1c4ef0b37111575ab2829c : SUCCESS in 46m 12s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d333981aac294202895ddc667189e973 : SUCCESS in 1h 11m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/eb0d4ba79856481dafa4c73431b92bb9 : SUCCESS in 2h 09m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/46c47108f8d240e4a789d278e22d7cf9 : SUCCESS in 1h 15m 44s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/27554b7d6c12406f929d56c15f4bfcf7 : SUCCESS in 1h 24m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e2d0e62935447a5b46e21f481907760 : SUCCESS in 1h 04m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a7b509629b5c4ba3a9d0d3f828e5d451 : SUCCESS in 1h 21m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d945807672349fe8de312f233ada281 : SUCCESS in 1h 25m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9c9c03e7b22e4d49b08dd53d6fcc9d33 : SUCCESS in 1h 33m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/73785c830a134f529e1f9822f683d433 : FAILURE in 1h 41m 42s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7df7f27be21c45ccaecb10c5a8c76fd6 : SUCCESS in 1h 33m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e289e16e76fc40cf9fd9cc5284033ecf : SUCCESS in 55m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54fda69b96284535acd8d51c7c985835 : SUCCESS in 2h 18m 22s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/624735eb1c1f4f559d0648769e9d2555 : FAILURE in 31m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1f6ef8428e4de2819143f7a96d5955 : SUCCESS in 1h 00m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4ce514aace9e403d8c4de379a9073594 : SUCCESS in 1h 09m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b84e2bcead784c8aba01f8310e6b1488 : SUCCESS in 55m 37s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de02119af17d47b6a8e3afb50ae60dce : SUCCESS in 1h 01m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/75b4f19f39a54982b8d43fc9696c83f3 : SUCCESS in 1h 19m 32s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b4f86f40d2b84cafbc0fe3e1bc80ace5 : FAILURE in 1h 41m 32s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/966a8c5fc14a41f28ffd45ddb4f852a9 : TIMED_OUT in 3h 05m 43s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/28f41eb5d4c24f24a3f22a61aa92d81a : FAILURE in 2h 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2696b2e4f24debe0309e0a8f6929460717fd7121","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-12-20 23:18:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f307f0ad0a7ca256555bd245e1da1e3ed2438040","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-12-20 23:24:46.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nA few questions inline.","accounts_in_message":[],"_revision_number":5},{"id":"a8b9047d8a366512920178c06460ea907a0fe416","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-07 11:06:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6ae22abfe5b34969d8fe5413787cbf553c20a5fc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-27 19:56:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5cdf3e8d4d3d065cdd1ec22196f58c5de51e8480","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-14 09:33:38.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"93d3393d31385405ebb2607e9e7b74ee5995d82d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-14 09:44:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4eb7b2f3bc3176a0a332aa1d558343b2ab2c29f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-14 09:53:56.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/698783/6 : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":6},{"id":"dc494c3adcd90b08796d209ed1e054f660e84f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:32:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/feeab9c73e4c4623b37f55bcd5db725e : SUCCESS in 1h 01m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c0d18929dbcb4ac9b898df7fcbf2ae4e : FAILURE in 20m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0f8bda0ba0d24478a4c4271e790678f2 : FAILURE in 1h 48m 51s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b14a0852c3e84f6d8c7df4fdaa231535 : FAILURE in 18m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/a0ca5b6f375d45b9a23653c06b44f89a : SUCCESS in 1h 10m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdc54458b42441cea3155ac6eb1b0346 : SUCCESS in 1h 03m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/346901ab8c214e5c9a6ec0781595ba8b : SUCCESS in 31m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd55b0fe952b4e8a9216de0a0c44ebb4 : SUCCESS in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99e2852a203434ca9c2d6e26ee02d34 : SUCCESS in 14m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b6fbbe57284ab7aec653f6a2563a6d : SUCCESS in 25m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f573e92d7ee3437f9fff4fbd7f4d0d1a : SUCCESS in 28m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/076e76f9dcfa42c1bc2fddea5ce9562a : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9adb2c80c64670ae8473c2dcd1978d : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/444290a6a8f742dab72e3a37cc03a0d8 : SUCCESS in 18m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4de5846f829847e78caf14c8336cd6bb : SUCCESS in 40m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/24c88fdd19014078bb029be694a4fa64 : SUCCESS in 1h 11m 54s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6d2d2a99af674fc48a292305da29b84e : SUCCESS in 2h 16m 25s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/490677bd0ef745d3bca506a2cd2bd2e8 : FAILURE in 25m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2298a90644c14e2aba2a25046dc29e96 : FAILURE in 18m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f60917ca0d274f3ca1ac7cc6ac55d3ee : SUCCESS in 1h 15m 21s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/88f08f4efbd348d1938dff6ab1bbb5c7 : FAILURE in 24m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a48dc3a0f45b4736be401bebd9d218e0 : SUCCESS in 1h 12m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/49559c6590d64b99a54f4fa48143284a : SUCCESS in 1h 41m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/acbf3a7b25fa47ef909e0d28ffdcf290 : FAILURE in 26m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b0b700e625c4aef9f2e5f43041ec443 : SUCCESS in 1h 24m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/42b9cd558a5c49b6a43861d6a56310a5 : SUCCESS in 1h 10m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a16a4647a5a48338d73379e75e64ab6 : SUCCESS in 2h 18m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5ddb77ae09e14bdaa5daf18c17d73aa8 : SUCCESS in 1h 54m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd915fc25b64c7c86f4af2657348cf0 : SUCCESS in 1h 44m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34b9bfa2ad454176ba1e3fa4ed64ba72 : SUCCESS in 1h 00m 47s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f6adb7720cfe4527b7833dacbf9d4adb : SUCCESS in 1h 33m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3795f1d1843d48539c4cf4ab9ae70622 : SUCCESS in 49m 21s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f37e4d169cd84d68a875dddcb51f012a : SUCCESS in 1h 02m 24s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3be46c6739b48f6bf1c795cb6ebbe62 : FAILURE in 19m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/42cfee95738a42998749070073bc4826 : FAILURE in 2h 40m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8ebc3a45e47a4abf974c09a14089fa94 : FAILURE in 2h 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"31a4c9d65224f475ef418cb1a3f1596bcb97a314","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-17 11:02:16.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"966fd82f3b189d9a4e874405cc2e531b9b80ba78","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-17 12:59:03.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/698783/7 : FAILURE in 1h 45m 30s","accounts_in_message":[],"_revision_number":7},{"id":"bb6276bb5bd09c5340b6ba1c18728bd228a10faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 18:08:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/de49bf16ce364381b1b2d7382eede25a : SUCCESS in 55m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51c5892115bc4347a3059cb548a3adb8 : SUCCESS in 38m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/62ea964c377c447dbe6096ecb4f0d020 : FAILURE in 1h 53m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bcf3afd0ee944bf59bec7e34093e53d8 : SUCCESS in 1h 03m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/4783f3bba8b44061849c90ae4ccf0344 : SUCCESS in 1h 07m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6a6b8729ff3d4f21a972395e0781d4ca : SUCCESS in 1h 10m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43b1a638ea594ddf84a192ad8eace434 : SUCCESS in 34m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32cf99d0ebc14ea1b7d654053767aa00 : SUCCESS in 41m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840bdd2e83fc479f90ca6fd4df78188e : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37338e4c4c2141b09eb7bb706563c8ac : SUCCESS in 31m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/765059345e9f4e74bb957bc0c5e7105b : SUCCESS in 42m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a30b74eb068045969fa873a136807221 : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093ff76619fd41888f5337805d1e4ceb : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b26243cc2ca4344b890b9278b6b00bb : SUCCESS in 14m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/93e863de77734fdc89c85d356965253b : SUCCESS in 42m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8b0f5c7084be40abb97ce9bf674fc108 : SUCCESS in 1h 08m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e2eef497ec6d499e9891e9129de3fffc : SUCCESS in 2h 06m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1cfcbd8307214b45a3829d8e84314184 : FAILURE in 1h 06m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b667b980b3be4ffb88b747b98e0523d3 : SUCCESS in 1h 05m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93942eb86e184196952807ea61fb77de : SUCCESS in 1h 16m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f626f790a4a46fc8a8c3ba881d855c7 : SUCCESS in 1h 15m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4b92da0a3d654b97b2581017ea3e8d94 : SUCCESS in 1h 18m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/383375e887c848678aa6e175305ca2f3 : SUCCESS in 1h 42m 12s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b05950fcca3c4ca0bed168967caa929b : SUCCESS in 2h 00m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/034ecf2eacfd4374a8cc6823a5ade83e : SUCCESS in 1h 19m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ff0e0bddca3842cf8a6a5f970dfc53db : SUCCESS in 1h 03m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62a2b5df7a5743abb9f1d325b93e13a7 : SUCCESS in 2h 18m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5e4ec7344e3d4b2791fd4cd53ea21b5c : SUCCESS in 2h 01m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c10320402bba496e85a934b2dff2f78a : SUCCESS in 1h 57m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc8d8270e7294b60af57ff51d2d1aa54 : SUCCESS in 1h 07m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e531ce08b5f544aab2862548c9c5379e : SUCCESS in 1h 12m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5636f45f7f764266906d0c4538b20f5a : SUCCESS in 44m 09s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa4e15aa10b4428c819e3f8152e67c7b : SUCCESS in 1h 12m 05s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/268d098da3114603b4b38c7de23efd87 : SUCCESS in 1h 12m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/559465823faa4bf6807f2979e7e2864b : FAILURE in 2h 33m 12s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2a94afeb0387447098787f41987e8ef3 : FAILURE in 2h 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e50d0981b3e7d9227cab07d983ca57a4facdbdae","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-18 10:42:56.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"64eb4f0952848fcca314e41c8944480c1b184675","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-18 12:35:01.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698783/7 : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":7},{"id":"0551dd00acb2254222c195c22ec8842fe5752b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 13:55:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/092d64d296264ec0a861f4e516e61c75 : SUCCESS in 53m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1e7d9cd9564540669aca14a3442764a6 : SUCCESS in 37m 46s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/2037bda85c194029b0306c18e2c280bb : FAILURE in 1h 43m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/24a4cb3715f348fbbaad581c7d8289fc : SUCCESS in 1h 10m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cd6b1c8c41724187b3f8e87f44ebc46a : SUCCESS in 1h 11m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ecaf07e40774569b5a51551a17ae637 : SUCCESS in 1h 10m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1f6c2ae71424e9095bee8e4f01586b6 : SUCCESS in 36m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0864db14ab413b8c5e2b5a865b28d8 : SUCCESS in 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57885d5a106475db7fe62ebadb5e00a : SUCCESS in 14m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26f63893d87f4ec1bce43230b527e9ed : SUCCESS in 27m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/627940eb07a548c9a659f3143d4cd63f : SUCCESS in 31m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e474bf52bed459389b2ab3b0ad4793d : SUCCESS in 29m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae6ab1ff33294320960acf7d3e390a25 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0894b844f3e94339aa6648fe0df48aa3 : SUCCESS in 15m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/71d840124ff1475ea9b58c1a60bab740 : SUCCESS in 47m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b76067b3e7394505afd39ca5ad055e35 : SUCCESS in 1h 05m 09s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/778e16c091e74a839481c1860c9ba8b6 : SUCCESS in 1h 42m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b9253eee59234b4ea6a350a413e41581 : SUCCESS in 1h 11m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/952c7e6752314853891d3351237110a7 : SUCCESS in 1h 05m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11d585376db74a0792ab0e085abbc003 : SUCCESS in 1h 02m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/890604ee4bf8456f9f8e13c20330fc8b : SUCCESS in 1h 15m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0a2639a74e18496994759d44b83d159e : SUCCESS in 1h 19m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c5492de91894ce1a8efdf24252f91f9 : SUCCESS in 1h 38m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d7b87d71a0840e78ba67a78538e7e14 : SUCCESS in 1h 47m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb187492820f41f18e436981da316800 : SUCCESS in 1h 18m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ad7dccb976c4f2ea9a1c4c76c9acefe : SUCCESS in 1h 03m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2764d83c87114fa1bb26dc16988b929d : SUCCESS in 2h 19m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5a51a0076ed94296b3fca6793426c310 : SUCCESS in 2h 14m 59s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5d5b16b4d84440eba9e485889f6455e3 : SUCCESS in 1h 50m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3cf2ee73e0a4b43bd2acd5593383be5 : SUCCESS in 1h 03m 51s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/32ad054793c34b04b1be9c146a354fcb : SUCCESS in 1h 11m 31s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbb3968711164ae48fdd94d3756ad5d0 : SUCCESS in 1h 07m 41s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/88fd847e22044e9fa8b616f439c9f034 : SUCCESS in 1h 17m 41s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d75c9fc03464a1c838e2bd4dc8edad1 : SUCCESS in 1h 27m 52s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d0f894ef696249589f16911eff5ff305 : FAILURE in 2h 53m 58s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/584d39c8fdcd497585435cccf53b793e : FAILURE in 2h 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e5b0409b94db8250dac43bd8ac82ff38bd2f0a29","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-18 14:35:34.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d461497f32a14228f1515b00feecd97dbdc5a66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:20:54.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b98297dc0e9a89352783bc95d96635c13c9ee454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:15:24.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ee087d4c57f04afe8b947904e6426124 : SUCCESS in 48m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c76c17cb2fb841c4a001515836ec0f22 : SUCCESS in 1h 00m 40s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7fea5adce63246b0800c32ac50a56aae : SUCCESS in 1h 09m 13s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dedf08ee4ade4078a978c49941f10a11 : SUCCESS in 1h 18m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04b7c60bb9ee4af5a5bda8fd31510b30 : SUCCESS in 47m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6598247d3d46b48e0015463fc65f6a : SUCCESS in 15m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f82f2c44ed4a8ba8a6e436b417e044 : SUCCESS in 33m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da38248cb43442b2912ad81755626c3d : SUCCESS in 47m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e339d8b0ad42829d21599950b8862c : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb877e457dc546ea9c4181d167ff0f06 : SUCCESS in 18m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/005c2baa43654f5c8ac391cb8d6102e1 : SUCCESS in 54m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6b358b4a2124433b929ffcef7cd2582f : SUCCESS in 1h 11m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f5aabf858713445a891854d332354ddd : SUCCESS in 1h 13m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/65b53f2da1f043fdbe16fe56fc61cb4f : SUCCESS in 1h 40m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3213734a60cf45f7a99a531e69dc4664 : SUCCESS in 1h 18m 57s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f87b896af225495ab34e3a8f4adcf269 : SUCCESS in 1h 14m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/97ac0ca9d64143ce9f078559b5504b46 : SUCCESS in 1h 49m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b69104e77934311b37a9210024ba950 : SUCCESS in 1h 09m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a31cd97234fc4640a4887595d5bff280 : SUCCESS in 2h 16m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f29fd6284bb948b5ae079da39fa04398 : SUCCESS in 1h 28m 58s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2be82472d41a4df2a127608e2039a70b : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":7},{"id":"22c5683e594a973a2c3b4d59daadce479777f6aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:15:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c17e750e16d49a3652578d80b24fcf42e0f28e43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:18:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6f634c958347e297ea2c48596fc24d : SUCCESS in 2m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4163a6e427494e0ba986c3714cb6262e : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"33fb446add87a3427a246444b03befc46adb385d","revisions":{"f9f3b47be8b402afb23e6b035bc518630d8c0b3c":{"kind":"REWORK","_number":1,"created":"2019-12-12 18:14:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/1"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 18:14:44.000000000","tz":0},"subject":"[WIP] Deprecate config option \"ovs_integration_bridge\"","message":"[WIP] Deprecate config option \"ovs_integration_bridge\"\n\n(To be defined, patch not ready)\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9f3b47be8b402afb23e6b035bc518630d8c0b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9f3b47be8b402afb23e6b035bc518630d8c0b3c"}]},"branch":"refs/heads/master"},"e2f328f0b7de7c6280cbb8caebb34dca03966799":{"kind":"REWORK","_number":2,"created":"2019-12-13 17:04:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/2"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-13 17:04:37.000000000","tz":0},"subject":"Deprecate config option \"ovs_integration_bridge\"","message":"Deprecate config option \"ovs_integration_bridge\"\n\nRemove this duplicated option and rely only in OVS.integration_bridge.\n\nNOTE: other projects are still using it; first we need to deprecate it\n      in those projects.\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f328f0b7de7c6280cbb8caebb34dca03966799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f328f0b7de7c6280cbb8caebb34dca03966799"}]},"branch":"refs/heads/master"},"87c6763fd85eecd2aca89c1c7514d98393f9d8d4":{"kind":"REWORK","_number":3,"created":"2019-12-16 15:08:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/3"}}},"commit":{"parents":[{"commit":"4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03","subject":"Merge \"Move zuul jobs definitions to zuul.d directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4051e0b19dc9f318c2e0dd7c60eaa2c46536ad03"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-16 15:08:10.000000000","tz":0},"subject":"Deprecate config option \"ovs_integration_bridge\"","message":"Deprecate config option \"ovs_integration_bridge\"\n\nRemove this duplicated option and rely only in OVS.integration_bridge.\n\nNOTE: other projects are still using it; first we need to deprecate it\n      in those projects.\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87c6763fd85eecd2aca89c1c7514d98393f9d8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87c6763fd85eecd2aca89c1c7514d98393f9d8d4"}]},"branch":"refs/heads/master"},"c6aeea1a1d38e471a5d9d7ce307ad5bb5d998f50":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-18 13:39:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/4"}}},"commit":{"parents":[{"commit":"33919fbd50ff7676082b4bfc8e65f270047f4248","subject":"Merge \"Use \"via\" in gateway dictionary in Linux Bridge agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33919fbd50ff7676082b4bfc8e65f270047f4248"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-18 13:39:32.000000000","tz":0},"subject":"Deprecate config option \"ovs_integration_bridge\"","message":"Deprecate config option \"ovs_integration_bridge\"\n\nRemove this duplicated option and rely only in OVS.integration_bridge.\n\nNOTE: other projects are still using it; first we need to deprecate it\n      in those projects.\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6aeea1a1d38e471a5d9d7ce307ad5bb5d998f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6aeea1a1d38e471a5d9d7ce307ad5bb5d998f50"}]},"branch":"refs/heads/master"},"fa36984d95c9fc1dc40ab253826e442369791667":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-12-20 18:09:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/5"}}},"commit":{"parents":[{"commit":"0115f71af8784dd7a180e09035c9530124e24bcc","subject":"Merge \"Bump min pyroute2 version to 0.5.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0115f71af8784dd7a180e09035c9530124e24bcc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-20 18:09:49.000000000","tz":0},"subject":"Deprecate config option \"ovs_integration_bridge\"","message":"Deprecate config option \"ovs_integration_bridge\"\n\nRemove this duplicated option and rely only in OVS.integration_bridge.\n\nNOTE: other projects are still using it; first we need to deprecate it\n      in those projects.\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa36984d95c9fc1dc40ab253826e442369791667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa36984d95c9fc1dc40ab253826e442369791667"}]},"branch":"refs/heads/master"},"49c575f7caebb0349e9d612b0ee1dffc7aa9a3bc":{"kind":"REWORK","_number":6,"created":"2020-02-14 09:44:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/6"}}},"commit":{"parents":[{"commit":"7a0e5185c6cf7b5f8bcfe50576e86798947a7ba7","subject":"Merge \"Add \"ncat\" rootwrap filter for debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a0e5185c6cf7b5f8bcfe50576e86798947a7ba7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-14 09:43:56.000000000","tz":0},"subject":"Deprecate config option \"ovs_integration_bridge\"","message":"Deprecate config option \"ovs_integration_bridge\"\n\nRemove this duplicated option and rely only in OVS.integration_bridge.\n\nNOTE: other projects are still using it; first we need to deprecate it\n      in those projects.\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49c575f7caebb0349e9d612b0ee1dffc7aa9a3bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49c575f7caebb0349e9d612b0ee1dffc7aa9a3bc"}]},"branch":"refs/heads/master"},"33fb446add87a3427a246444b03befc46adb385d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-02-17 11:02:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/698783/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/698783/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/698783/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/698783/7"}}},"commit":{"parents":[{"commit":"f71b6b361ee2d840877217b764c90ae234a09c00","subject":"Merge \"Revert \"Add \"ncat\" rootwrap filter for debug\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f71b6b361ee2d840877217b764c90ae234a09c00"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-12 17:39:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-02-17 11:02:16.000000000","tz":0},"subject":"Deprecate config option \"ovs_integration_bridge\"","message":"Deprecate config option \"ovs_integration_bridge\"\n\nRemove this duplicated option and rely only in OVS.integration_bridge.\n\nNOTE: other projects are still using it; first we need to deprecate it\n      in those projects.\n\nChange-Id: I4e826c8b9fa764b1820adacc3427934dc393c0bc\nRelated-Bug: #1856152\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33fb446add87a3427a246444b03befc46adb385d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33fb446add87a3427a246444b03befc46adb385d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
