)]}'
{"id":"openstack%2Fneutron~698968","triplet_id":"openstack%2Fneutron~master~I623f9a80541b7fd06edce05722577f7cb4d7a653","project":"openstack/neutron","branch":"master","topic":"remove_unused_args_in_tests","hashtags":[],"change_id":"I623f9a80541b7fd06edce05722577f7cb4d7a653","subject":"Remove unused arguments in test methods","status":"MERGED","created":"2019-12-13 17:20:14.000000000","updated":"2019-12-18 04:40:51.000000000","submitted":"2019-12-18 04:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698968-1576643883783-a23e576b","meta_rev_id":"de5f34e8e99d8dafbd409aac5bbabf4f5c011e37","_number":698968,"virtual_id_number":698968,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-18 01:48:24.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-13 18:18:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 04:38:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-13 19:13:39.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-16 11:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-12-16 22:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-12-16 22:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 18:18:33.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-13 19:13:39.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-16 11:14:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-16 22:23:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-18 04:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fde204bb891957b5c1870eac49fa8e85838c3563","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-13 17:20:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff1c1a6c4f093d13c31056579f28fe8ec77a8270","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-13 18:18:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/68/698968/1/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/68/698968/1/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d2fd8e56c531aad649b08fd2b76026ecbbdea031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 18:55:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e13f43cbd36149828b82405c72dfdc9a : SUCCESS in 33m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/898eec207b5c4387995215c71609ec44 : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8594742e4e054785994127a368ffb922 : SUCCESS in 12m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5181be4ef90d479c836bff8a55fd84f8 : SUCCESS in 21m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/232cb9b404ff47f8ba6d7269bb5d3341 : SUCCESS in 22m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35b866c328e4b139f446e9b30e12959 : SUCCESS in 10m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95a6195476cd431c87df562076996020 : SUCCESS in 49m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f38d37f1578a48f8a5b74c62ce5e01d0 : SUCCESS in 1h 20m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/94b4feaee4c84f7eb8cd5580effd1b92 : SUCCESS in 1h 31m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/927dcef6549f44e99207a1cfb97ef185 : SUCCESS in 50m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/218ba42325df40699ccc24cc0fa0f3c9 : SUCCESS in 1h 04m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1aa209c14281a9767a612df51f29f47475fb31f1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-13 19:13:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/698968/1 : SUCCESS in 1h 52m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b3e282d6b64ba9934d4e670d1b5b6d90ca067891","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-16 11:14:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf4d86377035fd84ee1166c4e8ae2516e34f569a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-16 22:23:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNot sure this will merge with current gate issues...","accounts_in_message":[],"_revision_number":1},{"id":"02877bb08d992b92c04853bc4aabb28e4971fb60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 22:24:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"270e723ad50d3db644cc403bcaaf67d734c1adf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 23:39:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fe33e6bb9f54c858b02faa217720a5a : SUCCESS in 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc04ed552ef4a2ab7afa6e6e2ed7252 : FAILURE in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3edcc875f0cb4ebd8c8731005a842dfd : FAILURE in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c3bba393a2e4b788318d215f8b9c6f4 : FAILURE in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07c5b1ebac14cbfb4a77d316e5fd7bc : SUCCESS in 9m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a3e7aeb960d5433294625131be133945 : SUCCESS in 44m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/28cfb3b1d3014fd7b0840f52062d8416 : SUCCESS in 1h 12m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/837892d062f147f994d6a50b7f262a79 : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eabf09a415fb9a18364a809e2b036985c330e465","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-18 01:48:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"02e20f0e309ba98227fbe692b2b361da4f090891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d491437e7742b29436c332608ed51c : SUCCESS in 31m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6cccd43f32141a7ba98a7ed9a9099a2 : SUCCESS in 24m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d8991e360a4ae78bccaf8c43600b80 : SUCCESS in 13m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c79d101f1aad4f6eb532c5c2b157b407 : SUCCESS in 26m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c18b309303334c56bba15e885a5a9c90 : SUCCESS in 34m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8be351214e433591eef0d5f72aee38 : SUCCESS in 9m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0cf13154d0e34427b4ea325fbb9d0480 : SUCCESS in 54m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f2c6a279a6b7400b8186a70f220eefd3 : SUCCESS in 1h 15m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6edcd6d9543649c286f1157c4a389739 : SUCCESS in 1h 23m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e4946e36ad1742b592a38a2d86350879 : SUCCESS in 48m 22s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6f1eac5cf2b455e94959924ada54d1c : SUCCESS in 1h 02m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d08c24dd3afafa1bd4f4e82011e868acc224383d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:19:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fdedfefa28b067eb29a230381db2d587d8fd698","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 04:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"095ceac8f5484e7bd0eb1c4782ff3197edbadfa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 04:38:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/407a21092d074fe58b28dd2c0e5372a1 : SUCCESS in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529ed871c40348eea1cb7cc42028dab6 : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb0a3291523a4f8d8ee19e79cb2fa5df : SUCCESS in 21m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bb33199989a4c0fa140e6b81fb9e8cc : SUCCESS in 30m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ae1b9d60e74b32a4dc40c87d8a5044 : SUCCESS in 8m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bc33c9ce42464fd4944660402b05c1a0 : SUCCESS in 48m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/191272c723ba4355831d41d762d5e3d8 : SUCCESS in 1h 03m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/afad2a1e110e4775b4e463bda0ed506e : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5c46108edd60d3f094139d22d143b7b2a2f87e8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 04:40:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e732286ad07e4722bd274f2556784e51 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df401ddca19eb4109a4991795bbdce81da30e5a3","revisions":{"df401ddca19eb4109a4991795bbdce81da30e5a3":{"kind":"REWORK","_number":1,"created":"2019-12-13 17:20:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/68/698968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/698968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/698968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/698968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/698968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/698968/1"}}},"commit":{"parents":[{"commit":"e04e1d80e465a04b831047ac0ffb566c80cdd029","subject":"Merge \"List SG rules which belongs to tenant\u0027s SG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e04e1d80e465a04b831047ac0ffb566c80cdd029"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-13 17:17:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-12-13 17:17:14.000000000","tz":0},"subject":"Remove unused arguments in test methods","message":"Remove unused arguments in test methods\n\n\"testtool\" library do not pass arguments to the test methods when\nexecuting the tests, except for those ones decorated with mock\npatches.\n\nTrivialFix\n\nChange-Id: I623f9a80541b7fd06edce05722577f7cb4d7a653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df401ddca19eb4109a4991795bbdce81da30e5a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df401ddca19eb4109a4991795bbdce81da30e5a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
