)]}'
{"id":"openstack%2Fneutron~699174","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I9b08a3a9c20b702b745b41d4885fb5120fd665ce","project":"openstack/neutron","branch":"stable/train","topic":"bug/1853840","hashtags":[],"change_id":"I9b08a3a9c20b702b745b41d4885fb5120fd665ce","subject":"Locate RP-tree parent by hypervisor name","status":"MERGED","created":"2019-12-16 09:47:16.000000000","updated":"2020-01-20 02:08:04.000000000","submitted":"2020-01-20 02:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699174-1579485968084-5f1f21a0","meta_rev_id":"db8d2028ae7776dc51688985218513fe5ecc5be2","_number":699174,"virtual_id_number":699174,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-14 10:24:18.000000000","_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-20 13:39:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 02:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-23 13:08:43.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-07 10:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-01-19 23:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-01-17 08:30:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-01-19 23:41:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 13:39:27.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-23 13:08:43.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-07 10:14:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-17 08:30:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-19 23:41:46.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-20 02:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e9092ae49c77b8d573572a50396616a287422b0","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-16 09:47:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284cdc180b06db469ed4af03b5275935672600d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-16 09:55:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/74/699174/1/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/74/699174/1/Neutron-ML2-Sriov : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"221e2b5603423d9abb3244e26cdccc1fb9825bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 12:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/18ffce234a68483aae2ac2d448593909 : SUCCESS in 56m 05s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/12badc9d426a4bddabe7ee9e0e3e74f7 : SUCCESS in 34m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/c70babb565944830b1b6286775bab095 : SUCCESS in 1h 41m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/4c78993abfeb49588fdab01c104c9e66 : SUCCESS in 59m 30s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/65c8ab47f7d74ff1bbd1f0e17ccef5e9 : SUCCESS in 1h 03m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/dfd09e2420d14279ab51d29eae4cf727 : SUCCESS in 1h 22m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7adff781a44d0fa6ca73b0e7a4f454 : FAILURE in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f288f7ed093409fbb884eea64bbe233 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7929c632fbe74e6ba54b6640188d0df2 : FAILURE in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de24fd947ac54289963d26cddd6f3855 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4190540a9654257ab1dc970d6ac38dd : FAILURE in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5571052a016f4849b5f0add6f52b88b0 : FAILURE in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4659ba9293b42489fe45c0b98be5eb9 : SUCCESS in 11m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/99e7f21d773a4da1abd84142d4c4f85b : SUCCESS in 1h 06m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7afb75bcf65d453ab9aed4d636d20123 : SUCCESS in 1h 38m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c7ae1cab2ce49d683ffb5492cef6760 : SUCCESS in 22m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2308f5e69fc34c2c838fab5bd3f4c849 : SUCCESS in 44m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/79c0b78daa474c53b5398ff2768596de : SUCCESS in 49m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c0c50601413145cf9072fc7f1efefceb : SUCCESS in 1h 18m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/f89b77958ed3439ba6861ffbf65ff1fb : SUCCESS in 2h 15m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c663b8c9859241a19613af13db408ccf : SUCCESS in 1h 25m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5de84d89bc8843ee9d026ee1386b4f8d : SUCCESS in 1h 20m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/848b8beab19a4120bc2920d0e07985ad : SUCCESS in 1h 05m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7f0776759f344c2490be87d499031b2f : SUCCESS in 1h 20m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8e63c1718f564ec5b1d27c69af1ee5a0 : SUCCESS in 1h 41m 38s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1bfaf65c8034f10a6a7b7fef0d916c8 : SUCCESS in 2h 11m 24s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdd9d9c3869e48b891c187061cb82b74 : SUCCESS in 1h 31m 44s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/1601311132be4cda9d8477ce173664b9 : SUCCESS in 1h 34m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/89d5b4c22dcd4050867b06276f0cc608 : FAILURE in 59m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/05ce3ee7271f4499b74e60b8ef8302ad : SUCCESS in 1h 00m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19f264ebab1f4faca872788d2a345af6 : SUCCESS in 2h 28m 29s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0a93a79792754fc89f28fee2d8da704c : SUCCESS in 2h 12m 36s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c4099ebd6b714ec4a8d311c176935011 : FAILURE in 1h 36m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15d7d141435445fca316ecc2f6a03006 : SUCCESS in 1h 02m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e057342a359846c8b62d101142fb85ef : SUCCESS in 1h 22m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc42ead618c640628ea96c34dd295f36 : SUCCESS in 47m 51s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c10a9e8b3e544028c7279a9b8ed4056 : SUCCESS in 1h 05m 29s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/28083e4a7f814347bcc5b14527a48e7c : SUCCESS in 1h 18m 52s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/21addd522ced4a139c671e14d5cb9f8e : SUCCESS in 2h 07m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4ec9746615e70207aec1ec9cdea69cf8057e6fe","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-20 12:08:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0de849559a39679ff4f07387a59fed6826c466dd","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-20 13:30:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b14ab29a5cf6fc14feb5513e4eb6c792d1ee3fbb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-20 13:39:27.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/74/699174/2/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/74/699174/2/Neutron-ML2-Sriov : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0d258dd4015d56b2ab14df31c7e0a4ac041c64ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 16:27:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/2ba3894a07e04d7fbc63f348b26ce188 : SUCCESS in 57m 21s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/bf3123fd94544aad88c15e7957373c3a : SUCCESS in 47m 40s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/bbb6e64a662e4baabdea5059849e73cc : SUCCESS in 1h 47m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/1a3d7309a1e34cdab90abc86aeeb148c : SUCCESS in 1h 04m 15s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/bff1831abba2455ca0e0161021813e00 : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/94fa43b0cb3742b8aff489b516d53b97 : SUCCESS in 56m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25f1b1e0abdc4beaa4321d0ea4ba86eb : SUCCESS in 34m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4cf57f487434230930e3fed98fc2033 : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9a8a8cc73442e688b7cb09d0c95a99 : SUCCESS in 13m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5d13a6c859946eea15d99e08d92942d : SUCCESS in 23m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4623e7eb44d44181a8995022df278edd : SUCCESS in 23m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0e0e2b82df245cfbf6dc9f1a9528132 : SUCCESS in 25m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6966efd572724b35b3f1ad50f0c82b97 : SUCCESS in 8m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b55b52912e64974af480cb2ed576675 : SUCCESS in 1h 03m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/298e5e201bd34b4a9073f51323c21b95 : SUCCESS in 1h 18m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b800bf10be4a4c2dbe8a6bb7264f9b09 : SUCCESS in 24m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aeb35bbf8d3040fbbea0c4f8d746041c : SUCCESS in 47m 36s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/c167f59e1e3d43369ff5a7bd388abd53 : SUCCESS in 55m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5ced6d914cf9498baff1188cc03677e3 : SUCCESS in 1h 02m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/658886d7d776495697ce688d11758602 : SUCCESS in 2h 24m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dc560628a7144c6ebb067af9605ee404 : SUCCESS in 1h 20m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/29fc48d348344032a470f831a6f5756c : SUCCESS in 1h 19m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1fdc790a6f9416192e29e5e7f2e76e4 : SUCCESS in 1h 06m 56s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/792ca1f97c514859bc7d95705349059f : SUCCESS in 1h 14m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6677e23ba11a4ba191141efc1c2933e4 : SUCCESS in 1h 54m 46s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5681ac3e1cee4211a89e87d6b438751e : SUCCESS in 1h 54m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a9b3eb2293d4759a456f960b6da469f : SUCCESS in 1h 35m 17s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/799ddabadb654c05a182152a1e0fefb2 : SUCCESS in 1h 29m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a228838628954e3684155981dfd5ba6e : SUCCESS in 59m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ddd312d4e70e4195985fa6b9f064c066 : SUCCESS in 56m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae64b1e647dd4239bd7fb5dbe3869ac0 : FAILURE in 2h 52m 26s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/535a99ab01dd4474baa04f228a266604 : SUCCESS in 2h 14m 39s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9d6535249b914a3f80022087c8401a94 : FAILURE in 2h 05m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7436a6ad64034820b41a2dd53d1265d3 : SUCCESS in 1h 04m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d916a9a282144de7b79d0e172830c7a6 : SUCCESS in 1h 23m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1de50e66e93b446794a0a73073524108 : SUCCESS in 45m 26s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e8b8391be824db7859eeaac706e95fb : SUCCESS in 1h 09m 37s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb24742c39584f91926eb19c7fcee29d : SUCCESS in 1h 33m 21s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aadeb660e77a4676a0fc054c4b3e2c43 : FAILURE in 1h 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0167f8ded6f62ff92334b8fc8adc2473b0032121","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-12-23 11:08:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"38e12c309e74924bbfa529b10fd7edf5e34b91b5","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2019-12-23 13:08:43.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/699174/2 : FAILURE in 1h 59m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b205132e42bdf838acc5691c278c773da1c90099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 13:29:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a9818e95fcf54df3af90c397cb26f1d3 : SUCCESS in 50m 01s\n- neutron-tempest-plugin-designate-scenario-train https://zuul.opendev.org/t/openstack/build/ca33e34749214eee91c9039ee691cd62 : SUCCESS in 35m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario-train https://zuul.opendev.org/t/openstack/build/a045d6f9a63b462a8c0cc609529fbff2 : FAILURE in 1h 27m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/284efca7c3a34dcabe74f21e699f6139 : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/ad14c3727a25476aaa4ea521727da016 : SUCCESS in 1h 03m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/a1f119d075cd433b84e4ad982db314fe : SUCCESS in 1h 00m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d861ca0032b4f07a319368e0632cfdb : SUCCESS in 31m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8630ba29b85646cc85f4862a058769ca : SUCCESS in 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9888742e6fd7408b8055d101873bb3ff : SUCCESS in 12m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44269974471c402a803b4b3422271354 : SUCCESS in 22m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81334572e48f48c0b35118d4cc89af73 : SUCCESS in 22m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ac4f980339a4288afd71b96006fd802 : SUCCESS in 22m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94206d69eab54c69aff6484ae1d871d7 : SUCCESS in 8m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1cf424a9c970437fa9f8e21086cf7a5d : SUCCESS in 57m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7bbb846add4149c6ab3f576f0dd339ad : SUCCESS in 1h 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c5ac3c25f941019e9b1ea0383ebf7f : SUCCESS in 14m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c52750b8d4944029851d94359468a629 : SUCCESS in 46m 27s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/bceb7b1c20f24aaeb4a78b55b8cb9fda : SUCCESS in 45m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1c5ec6f031374c41b03fd01b25260eda : SUCCESS in 1h 03m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6627c2068e47478fa8336996db93f70b : SUCCESS in 1h 49m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cdb82f3707184b70979cdc86a0c76e0e : SUCCESS in 1h 20m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f64ae83fff7549aab6c7c400fb0a675c : SUCCESS in 1h 15m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68aacd6d298b457b86d26350244518cc : SUCCESS in 1h 10m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fedba3f63bbc4221a9a14591e459c1e9 : SUCCESS in 1h 05m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/76e742f05bf649cd8fc8f6543ea0ffaa : SUCCESS in 1h 36m 59s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3bc7031ad244358b4aaed6692cff6ca : SUCCESS in 2h 04m 53s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09345218e2a14e37a5fd85e402f1e8dd : SUCCESS in 1h 17m 10s\n- neutron-tempest-iptables_hybrid-fedora https://zuul.opendev.org/t/openstack/build/b71c726a3be94ca4979e03215bec7036 : SUCCESS in 1h 30m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bcdbd20473414c79b1817ec3f40b8eac : SUCCESS in 59m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/47018f8d686043a6ae4c488f122a3281 : SUCCESS in 59m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f004a1a648cb4bb88434c87f2c51b00f : SUCCESS in 2h 15m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fdc8567e91b34b3096464e1e80a97e28 : SUCCESS in 2h 12m 46s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4b7168b9ff09492a8bf9eb7968aed9ce : FAILURE in 1h 40m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c295c351c0b34d2481f20a28fb0d5c2c : SUCCESS in 1h 01m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/84fec35f03ff4ba08bf5e6ee6df81e10 : SUCCESS in 1h 27m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2da87b0205264d7f9161d57a05205221 : SUCCESS in 43m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b15163d9e454384914b7e3440fbedfd : SUCCESS in 1h 04m 44s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a51811bc73542b599071696dd4dfde9 : SUCCESS in 1h 21m 06s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f8b500fd8b1742ac8f94a5b27be15f24 : FAILURE in 1h 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a32b66be19fa15772e22c5e27c4802b98ad3ed57","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-07 10:14:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nClean backport (only differences are the mentioned dropped TODOs)\nA bit large but needed to fix the mentioned bug\nDefault values for new options are identical to old behaviour, and there is code to support agents without the option\nSo looks OK to backport","accounts_in_message":[],"_revision_number":2},{"id":"5a803caec1005fb56322a35754c065635a436993","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-01-07 13:42:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"127a3e249bfda7fe1c42f83689e765f6070bf725","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-07 14:23:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d6d542db493ba9a416e3f11a698e71e4a21e284","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-01-14 10:24:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e94672d52d6d859cf8cde3c4be1d91f1c12fae9c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-17 08:30:09.000000000","message":"Patch Set 2: Code-Review+1\n\nseems to be a clean backport (except the TODOs)","accounts_in_message":[],"_revision_number":2},{"id":"3059a877ecde6c5bf50ac88a4163b0d629f52746","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-19 23:41:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5931106cd7fce6340adca21d4aadc5d5ce813668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 23:42:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca79545249fc1c3994fd0230580daa0c800daa83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 02:06:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/bfa3641d21734533a208d7814abe2297 : SUCCESS in 53m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17045a1c07834fcebcbf683b7d753d6f : SUCCESS in 22m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b447aa902a04cb0bb91f998e2e72be4 : SUCCESS in 12m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98e3999b05ee427c83a3b286415991dd : SUCCESS in 22m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/946828a3b14240adbd6ae51211f04aea : SUCCESS in 34m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0302f5e4e46c4588abfee0e8db47dc59 : SUCCESS in 23m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9781eb3a2845d1b90d69c01397d11e : SUCCESS in 7m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/585e7c5664584c73af228193597183cf : SUCCESS in 54m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa7370798c7243df9728f01cbb30f187 : SUCCESS in 1h 15m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664777caf2874375ae9c3ef14dd46c6f : SUCCESS in 12m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ace27e49eadd4442bf834440954b3105 : SUCCESS in 39m 28s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/7692c271299c4471854cfe78d2237f64 : SUCCESS in 54m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b191f8c37fe34c92bdc2aa41ad03cb9f : SUCCESS in 1h 08m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/298167f22c8243c2b5b97bba96cff04e : SUCCESS in 1h 12m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3365236ca76146629468028189aa5d63 : SUCCESS in 1h 09m 04s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98db71e57bda468cac6d08d426f57c43 : SUCCESS in 1h 03m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/423d5b3162924433b26874a92c0ba959 : SUCCESS in 1h 00m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/abe505ca07eb4d41b627210dca2b1501 : SUCCESS in 1h 07m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2816d50e89a5459c8234f1f5e7440a62 : SUCCESS in 1h 06m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48a09f62e8814723a5c07ede5256ae5f : SUCCESS in 2h 18m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/951681a8200844e18099d9fcafa771f4 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6e2cdca8996c3b8f7152aac399da7035abf5e5d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 02:06:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a26f7746f870f4c9a9e7a6e6e9760d84f3519deb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 02:08:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8722acfa240f4b55a3efe1392890fe18 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c694d21890d4d44a0d61874a5869962 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a6766470ef127ee5495a5b74b7156bd5a80f03c","revisions":{"c975b76abe3b0cd8349e4621beb97f89865f0c0e":{"kind":"REWORK","_number":1,"created":"2019-12-16 09:47:16.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/74/699174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/699174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/699174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/699174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/699174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/699174/1"}}},"commit":{"parents":[{"commit":"1afc3cbee7820820933e404833ad3acc2f7891b5","subject":"Merge \"[OVS FW] Clean port rules if port not found in ovsdb\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1afc3cbee7820820933e404833ad3acc2f7891b5"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-11-27 16:59:15.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-12-16 09:43:23.000000000","tz":60},"subject":"Locate RP-tree parent by hypervisor name","message":"Locate RP-tree parent by hypervisor name\n\nPreviously we assumed that we can look up the resource provider (created\nby nova) to be used as the parent of the agent and physical NIC resource\nprovider tree by the name set in the config option DEFAULT.host. This\nassumption was wrong.\n\nWhile nova-compute\u0027s DEFAULT.host and neutron-agent\u0027s DEFAULT.host\nmust match for port binding to work, the root resource provider created\nby nova does not belong to the compute host (where nova-compute runs)\nbut it belongs to the compute nodes (i.e. hypervisors). Actually there\nmay be multiple compute nodes managed by a single nova-compute (think\nof ironic). Plus the value of DEFAULT.host and the compute node\u0027s ID\nmay be different even when nova-compute manages a hypervisor on the\nsame host because of various deployment considerations. For example\nwhen tripleo does not manage the undercloud (so a libvirt hypervisor\nreturns the plain hostname), but the same tripleo enforces it\u0027s host\nnaming conventions in nova\u0027s and neutron\u0027s DEFAULT.host settings.\n\nThis change enables neutron to use the hypervisor name to locate the\nroot of the resource provider tree.\n\nWe introduce a new configuration option for\n\n(1) ovs-agent: resource_provider_hypervisors, for example:\n\n[ovs]\nbridge_mappings \u003d physnet0:br-physnet0,...\nresource_provider_bandwidths \u003d br-physnet0:10000000:10000000,...\nresource_provider_hypervisors \u003d br-physnet0:hypervisor0,...\n\n(2) sriov-agent: resource_provider_hypervisors, for example:\n\n[sriov_nic]\nbridge_mappings \u003d physnet1:ens5,...\nresource_provider_bandwidths \u003d ens5:10000000:10000000,...\nresource_provider_hypervisors \u003d ens5:hypervisor1,...\n\nFor both agents \u0027resource_provider_hypervisors\u0027 values default to\nsocket.gethostname() for each key in resource_provider_bandwidths.\n\nWe try to not block later developments in which one neutron\nagent may manage devices on multiple hosts. That\u0027s why we allow\nthe each physdev to be associated with a different hypervisor.\n\nBut here we do not try to solve the problem that the natural physdev\nidentifiers may not be unique accross multiple hosts. We leave solving\nthis problem to whoever wants to implement an agent handling devices of\nmultiple hosts.\n\n(3) We extend report_state message\u0027s configurations field alike:\n\n{\n\u0027bridge_mappings\u0027: {\u0027physnet0\u0027: \u0027br-physnet0\u0027},\n\u0027resource_provider_bandwidths\u0027: {\n    \u0027br-physnet0\u0027: {\u0027egress\u0027: 10000000, \u0027ingress\u0027: 10000000}},\n\u0027resource_provider_hypervisors\u0027: {\u0027br-physnet0\u0027: \u0027hypervisor0\u0027},\n...\n}\n\n(4) In neutron-server we use\nreport_state.configurations.resource_provider_hypervisors.PHYSDEV\nwhen selecting parent resource provider for agent and physdev\nRP-tree. When not available in the message we fall back to using\nreport_state.host as before.\n\nSince we only changed the free-format configurations field of the\nreport_state message rpc version is not bumped and we expect this\nchange to be backported to stein and train.\n\nRemoved unapplicable TODO note from backport.\n\nChange-Id: I9b08a3a9c20b702b745b41d4885fb5120fd665ce\nCloses-Bug: #1853840\n(cherry picked from commit 258eebea71b1cac37badf429a90d5cf57e4c455c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c975b76abe3b0cd8349e4621beb97f89865f0c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c975b76abe3b0cd8349e4621beb97f89865f0c0e"}]},"branch":"refs/heads/stable/train"},"9a6766470ef127ee5495a5b74b7156bd5a80f03c":{"kind":"REWORK","_number":2,"created":"2019-12-20 13:30:47.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/74/699174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/699174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/699174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/699174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/699174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/699174/2"}}},"commit":{"parents":[{"commit":"1afc3cbee7820820933e404833ad3acc2f7891b5","subject":"Merge \"[OVS FW] Clean port rules if port not found in ovsdb\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1afc3cbee7820820933e404833ad3acc2f7891b5"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-11-27 16:59:15.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-12-20 12:25:13.000000000","tz":60},"subject":"Locate RP-tree parent by hypervisor name","message":"Locate RP-tree parent by hypervisor name\n\nPreviously we assumed that we can look up the resource provider (created\nby nova) to be used as the parent of the agent and physical NIC resource\nprovider tree by the name set in the config option DEFAULT.host. This\nassumption was wrong.\n\nWhile nova-compute\u0027s DEFAULT.host and neutron-agent\u0027s DEFAULT.host\nmust match for port binding to work, the root resource provider created\nby nova does not belong to the compute host (where nova-compute runs)\nbut it belongs to the compute nodes (i.e. hypervisors). Actually there\nmay be multiple compute nodes managed by a single nova-compute (think\nof ironic). Plus the value of DEFAULT.host and the compute node\u0027s ID\nmay be different even when nova-compute manages a hypervisor on the\nsame host because of various deployment considerations. For example\nwhen tripleo does not manage the undercloud (so a libvirt hypervisor\nreturns the plain hostname), but the same tripleo enforces it\u0027s host\nnaming conventions in nova\u0027s and neutron\u0027s DEFAULT.host settings.\n\nThis change enables neutron to use the hypervisor name to locate the\nroot of the resource provider tree.\n\nWe introduce a new configuration option for\n\n(1) ovs-agent: resource_provider_hypervisors, for example:\n\n[ovs]\nbridge_mappings \u003d physnet0:br-physnet0,...\nresource_provider_bandwidths \u003d br-physnet0:10000000:10000000,...\nresource_provider_hypervisors \u003d br-physnet0:hypervisor0,...\n\n(2) sriov-agent: resource_provider_hypervisors, for example:\n\n[sriov_nic]\nbridge_mappings \u003d physnet1:ens5,...\nresource_provider_bandwidths \u003d ens5:10000000:10000000,...\nresource_provider_hypervisors \u003d ens5:hypervisor1,...\n\nFor both agents \u0027resource_provider_hypervisors\u0027 values default to\nsocket.gethostname() for each key in resource_provider_bandwidths.\n\nWe try to not block later developments in which one neutron\nagent may manage devices on multiple hosts. That\u0027s why we allow\nthe each physdev to be associated with a different hypervisor.\n\nBut here we do not try to solve the problem that the natural physdev\nidentifiers may not be unique accross multiple hosts. We leave solving\nthis problem to whoever wants to implement an agent handling devices of\nmultiple hosts.\n\n(3) We extend report_state message\u0027s configurations field alike:\n\n{\n\u0027bridge_mappings\u0027: {\u0027physnet0\u0027: \u0027br-physnet0\u0027},\n\u0027resource_provider_bandwidths\u0027: {\n    \u0027br-physnet0\u0027: {\u0027egress\u0027: 10000000, \u0027ingress\u0027: 10000000}},\n\u0027resource_provider_hypervisors\u0027: {\u0027br-physnet0\u0027: \u0027hypervisor0\u0027},\n...\n}\n\n(4) In neutron-server we use\nreport_state.configurations.resource_provider_hypervisors.PHYSDEV\nwhen selecting parent resource provider for agent and physdev\nRP-tree. When not available in the message we fall back to using\nreport_state.host as before.\n\nSince we only changed the free-format configurations field of the\nreport_state message rpc version is not bumped and we expect this\nchange to be backported to stein and train.\n\nRemoved unapplicable TODO notes from backport.\n\nChange-Id: I9b08a3a9c20b702b745b41d4885fb5120fd665ce\nCloses-Bug: #1853840\n(cherry picked from commit 258eebea71b1cac37badf429a90d5cf57e4c455c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a6766470ef127ee5495a5b74b7156bd5a80f03c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a6766470ef127ee5495a5b74b7156bd5a80f03c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
