)]}'
{"id":"openstack%2Fneutron~699243","triplet_id":"openstack%2Fneutron~master~I4579c575b3526a835efb475865e31aa88a76360c","project":"openstack/neutron","branch":"master","topic":"bug/1856600","hashtags":[],"change_id":"I4579c575b3526a835efb475865e31aa88a76360c","subject":"Fix reference to flake8 deprecated engine","status":"ABANDONED","created":"2019-12-16 16:49:45.000000000","updated":"2020-08-19 21:18:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa99f0d0ef2d2ad49f8ce6f4496251f5886f3108","_number":699243,"virtual_id_number":699243,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-28 02:37:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2020-03-27 19:57:44.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-27 21:12:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-27 19:59:10.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 19:34:16.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-06 14:54:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-07 02:41:04.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-01-22 17:02:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-27 19:57:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-27 19:59:10.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-27 21:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-28 02:37:50.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1df3231b0f58f7c52378d0411d4a971ca8bc711","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-16 16:49:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8c6bee6d31dceaafd6348fa4bb614a75322c667","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-16 17:03:24.000000000","message":"Patch Set 1: Workflow-1\n\nMarking this workflow -1 until we\u0027re ready to tackle the greater issues found in the update to hacking 2.0.  \n\nSee https://review.opendev.org/#/c/699021/","accounts_in_message":[],"_revision_number":1},{"id":"ec08d9690649954f238f822fa11f066f5ae663aa","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-16 17:08:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2e849d2e01f58df395481d80bfa184c9264b8150","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-16 17:09:02.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"bb11872d82286f5b3ea42c9293df4379349a6388","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-16 17:09:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/43/699243/1/Neutron-MACVTAP-ML2-Sriov : ABORTED\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/43/699243/1/Neutron-ML2-Sriov : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"0632c247df75cf939f320a76d7bfc0fb699d533e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-12-16 17:27:17.000000000","message":"Patch Set 2:\n\nModuleNotFoundError: No module named \u0027flake8.api\u0027\n\n:(\n\nI got a little bit farther Friday making sure flake8 3.0.0 was being used, but I\u0027m assuming a larger movement of the usual lint suspects - pylint, astroid, etc, will be required","accounts_in_message":[],"_revision_number":2},{"id":"9a5ba392d54729926b6fb4cac8cb7300180d5d71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-16 17:59:16.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/43/699243/2/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/43/699243/2/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e1deefc26f52e6833b862c21a30de23d11e3fceb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-16 18:16:54.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/43/699243/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"186513eec669628cf30cdba1333b22b94f0a9d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 18:21:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da61ef5b03d5458fb2e792abebbc9fdb : FAILURE in 4m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63e77c6e4e93426da86e511152de352f : FAILURE in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37eb30a019b1446686d1b7f51ef0914f : SUCCESS in 12m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/257abe2aa097448898fc64e548b98114 : FAILURE in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1794190f57c449a0a15c152f8c630288 : FAILURE in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a63f95a0d94234af3eb8ea18a5d7e3 : SUCCESS in 12m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8c8ed3744fc942e7bfc06fe1ec80f15c : SUCCESS in 51m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d6f3849212da410ea2c826c829563cbc : SUCCESS in 1h 04m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/40cd54902fa945cf9f54aaff961415a3 : SUCCESS in 1h 09m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cf019ea06db497ea74f0decceb87f58 : SUCCESS in 50m 36s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4037a36c38b6463eb36abf2407ae5b26 : SUCCESS in 1h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"934b0e19c55013efebd181727f2e6cc9b289eb87","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-16 18:22:43.000000000","message":"Patch Set 2:\n\n\u003e ModuleNotFoundError: No module named \u0027flake8.api\u0027\n \u003e \n \u003e :(\n \u003e \n \u003e I got a little bit farther Friday making sure flake8 3.0.0 was\n \u003e being used, but I\u0027m assuming a larger movement of the usual lint\n \u003e suspects - pylint, astroid, etc, will be required\n\nI was waiting for local testing to finish a couple of things... I rebased this on top of your change, so the pin needs to be removed.  PS3 does that.  The flake8.api should exist.","accounts_in_message":[],"_revision_number":2},{"id":"826b19131623ce0baa4d07914ca12d2241bb9510","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-16 18:23:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e5de212e76889abb7ed5d5dac4f9f6181121d57f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-16 18:43:08.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/43/699243/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"eb54c907f1b63b5ce2bb1020fbe3c1a55582cb6d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-16 19:34:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/43/699243/3/Neutron-MACVTAP-ML2-Sriov : FAILURE\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/43/699243/3/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"da86125856bd6d15bcd64542697b53591c910068","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-16 19:39:37.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/43/699243/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"4a6ebdd1b84f41a97eff67ed45927bcfa00aa597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:48:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e289d55673b844c1ac662636b085f06b : FAILURE in 33m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eb6eedaeadc4b5b9c4aef8c2d1bd04f : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46de56f2184d4caaab74c0d234327089 : SUCCESS in 10m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9218d90b52024ac2a93343ec797a0acb : FAILURE in 26m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f732f4b982514584a75457b551eb8e35 : FAILURE in 22m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328932da7f88416aa4f092ab2fa6543f : SUCCESS in 8m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/376f23799f4c475da3a05148ba4ffa64 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9623c30407640df91f56d419b0993bf : SUCCESS in 22m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1a720a1fe7d049bc9c7fa0d718c664ff : SUCCESS in 50m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b01084ff780143ccb14abc9c0c07c4a1 : SUCCESS in 1h 00m 59s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6f376ba5d89a42ffb18034864002876d : SUCCESS in 1h 18m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49c89d7bd29a4a86b43df13cd64cddb6 : SUCCESS in 50m 01s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/85198ac166334c40991a25a69274f0b2 : SUCCESS in 1h 05m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d63fbc4a2011d872640d16bdcda627b938b859c","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-01-07 02:41:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ce4371258d9b1654a7588c9e1bf622405664102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 04:11:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b561deb44efd448eba6145906410f88d : FAILURE in 40m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5784da157b443d7a9899c7488fc0254 : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b161971747174a5cb86861a006e71c51 : SUCCESS in 10m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ee6a3d496bb40c8aeed79f1e2b0c07e : FAILURE in 22m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef09483a9b194a09b57cf0ab33ad50bf : FAILURE in 29m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a88502f35d4b2b8427ae2f949c460e : SUCCESS in 8m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3941b877f9e54e16936c3886b8a1c348 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c952804c41a642c6955e229c54731f86 : SUCCESS in 16m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0407ced46dcb48f3a4311ed4c42066de : SUCCESS in 46m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e765a0abb7f5451f91504a590777209f : SUCCESS in 1h 29m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9912aa20e0cf4547bb34c5c0c9f0f44d : SUCCESS in 1h 19m 50s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/697d76a45dbe4decb7e2d5fef4080197 : SUCCESS in 44m 32s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92d839c6e6904ef499ebe75f4e599289 : SUCCESS in 1h 00m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e94aea61b8062a638e96cd6c5b3eb31329df6c7f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-07 04:24:42.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/699243/3 : FAILURE in 1h 43m 07s","accounts_in_message":[],"_revision_number":3},{"id":"e66594a861dc241f4e60c50138b6e95d750410b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-22 17:02:20.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI still don\u0027t know what is happening with test_flake8","accounts_in_message":[],"_revision_number":3},{"id":"7051deb6d2bbaf131e69801ebf0ceee598d74717","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 23:20:24.000000000","message":"Patch Set 3:\n\nThis needs to be rebased to https://review.opendev.org/#/c/694568/ (or wait until it merges), which will fix some of the issues.\n\nThen we need to start playing with bumping flake8, hacking, pycodestyle, etc - I stopped going down the rabbit hole there.  There was another change recently regarding un-capping pycodestyle I\u0027ll see if I can find it just can\u0027t find the tab.","accounts_in_message":[],"_revision_number":3},{"id":"75cb6d9804512f26a0bbae753c98ad5eb0cb5432","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 19:57:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5ddc9001df1a4bb722707332d52897126be8ee06","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 19:57:44.000000000","message":"Patch Set 4:\n\nI don\u0027t think a rebase will make the tests magically pass.","accounts_in_message":[],"_revision_number":4},{"id":"9aa99cfd7c8ffaffdf32258185902277fcaeca51","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-27 19:59:10.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/699243/4 : FAILURE in 1m 14s","accounts_in_message":[],"_revision_number":4},{"id":"2c512fa9452f22739d9ad34f487cfd911042d0c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 21:12:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0831902a73364d97bc7d0f6abc79ffa5 : FAILURE in 38m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44c4492d82164827a3602e1d34918251 : FAILURE in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f315a8bfc24ed990308e491ed40056 : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96b6e2ebc3f34e2ba8366ff87d8b16e5 : FAILURE in 30m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1612cf86a9f4d298efddcbc903904c4 : FAILURE in 45m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc9fdde7620f4c38aeb2051bcc2d16c4 : FAILURE in 29m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3c33ddec764477b16f28dacd1ca221 : SUCCESS in 9m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e697063f1494599815daa40243a6229 : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ed189a4d7264ae495436acec7cb6891 : SUCCESS in 15m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4c84a3c614bf497184674d2583451b97 : SUCCESS in 49m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e3d6bf301f5042eca836755cef690bd8 : SUCCESS in 1h 07m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9dfef9d7108945ac82fee43a41961234 : FAILURE in 36m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1b9fc9ef2474876a923818b8cd7357c : SUCCESS in 1h 06m 26s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ae3725a8fd841bbbefb9cd2e0769fc5 : SUCCESS in 1h 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e17020bd4ecedbced4f63220b22c0f460dd0067a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-03-28 02:37:50.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/March-2020/Arista_Tempest_Testrefs/changes/43/699243/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"e0ec80a6f51fc9cc5a51ef8e0ea8b032224a3553","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-19 21:18:00.000000000","message":"Abandoned\n\nNot required, file changed after move to hacking\u003e3","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"977bb6c3ccba090b2a816997a1231b8ab96b3f74","revisions":{"5ee193aacc386149eb651e8368e50c4edfa07011":{"kind":"REWORK","_number":1,"created":"2019-12-16 16:49:45.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/43/699243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/699243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/699243/1"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 16:48:21.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 16:49:44.000000000","tz":-300},"subject":"Fix reference to flake8 deprecated engine","message":"Fix reference to flake8 deprecated engine\n\nFlake 3 removed the Engine but kept a legacy module for compatibility.\nUse the legacy module.\n\nChange-Id: I4579c575b3526a835efb475865e31aa88a76360c\nResolves-Bug: #1856600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee193aacc386149eb651e8368e50c4edfa07011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ee193aacc386149eb651e8368e50c4edfa07011"}]},"branch":"refs/heads/master"},"18f26bdfff7aeecb00b31d324f1ff4b9d24f0479":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-16 17:08:47.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/43/699243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/699243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/699243/2"}}},"commit":{"parents":[{"commit":"36727e3463ca94d25cac1ab715097c778b3f7736","subject":"Fix pep8 errors found by hacking 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36727e3463ca94d25cac1ab715097c778b3f7736"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 16:48:21.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 17:08:47.000000000","tz":0},"subject":"Fix reference to flake8 deprecated engine","message":"Fix reference to flake8 deprecated engine\n\nFlake 3 removed the Engine but kept a legacy module for compatibility.\nUse the legacy module.\n\nChange-Id: I4579c575b3526a835efb475865e31aa88a76360c\nResolves-Bug: #1856600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18f26bdfff7aeecb00b31d324f1ff4b9d24f0479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18f26bdfff7aeecb00b31d324f1ff4b9d24f0479"}]},"branch":"refs/heads/master"},"3a6aea16f4a7b461041e924caa7d6fdcd913d5bd":{"kind":"REWORK","_number":3,"created":"2019-12-16 18:23:10.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/43/699243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/699243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/699243/3"}}},"commit":{"parents":[{"commit":"36727e3463ca94d25cac1ab715097c778b3f7736","subject":"Fix pep8 errors found by hacking 2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36727e3463ca94d25cac1ab715097c778b3f7736"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 16:48:21.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 18:21:37.000000000","tz":-300},"subject":"Fix reference to flake8 deprecated engine","message":"Fix reference to flake8 deprecated engine\n\nFlake 3 removed the Engine but kept a legacy module for compatibility.\nUse the legacy module.\n\nChange-Id: I4579c575b3526a835efb475865e31aa88a76360c\nResolves-Bug: #1856600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6aea16f4a7b461041e924caa7d6fdcd913d5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a6aea16f4a7b461041e924caa7d6fdcd913d5bd"}]},"branch":"refs/heads/master"},"977bb6c3ccba090b2a816997a1231b8ab96b3f74":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-27 19:57:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/43/699243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/699243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/699243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/699243/4"}}},"commit":{"parents":[{"commit":"8b2bda7be5dcd1ca1bd67f356cf3e2c8eb00137b","subject":"Merge \"Prioritize port create and update ready messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b2bda7be5dcd1ca1bd67f356cf3e2c8eb00137b"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-12-16 16:48:21.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-03-27 19:57:25.000000000","tz":0},"subject":"Fix reference to flake8 deprecated engine","message":"Fix reference to flake8 deprecated engine\n\nFlake 3 removed the Engine but kept a legacy module for compatibility.\nUse the legacy module.\n\nChange-Id: I4579c575b3526a835efb475865e31aa88a76360c\nResolves-Bug: #1856600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/977bb6c3ccba090b2a816997a1231b8ab96b3f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/977bb6c3ccba090b2a816997a1231b8ab96b3f74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
