)]}'
{"id":"openstack%2Fneutron~699372","triplet_id":"openstack%2Fneutron~master~Ic226b7c60292bd5b67bd4ead1048f942e1cc1028","project":"openstack/neutron","branch":"master","topic":"bug/1855919","hashtags":[],"change_id":"Ic226b7c60292bd5b67bd4ead1048f942e1cc1028","subject":"Doc improvment for rpc_workers configuration","status":"ABANDONED","created":"2019-12-17 08:49:31.000000000","updated":"2020-09-25 15:45:28.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3ebfbd3268ee0b8d80f5fc74af21475f6b7600ce","_number":699372,"virtual_id_number":699372,"owner":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-17 10:41:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-17 11:17:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"date":"2020-09-25 15:45:28.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-17 12:55:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2019-12-17 09:55:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 09:55:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-12-17 10:41:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 11:17:36.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-17 12:55:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:45:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"525fecfd7a23450de76e5068fdc43d22e7f69ba4","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-12-17 08:49:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc42996c777241da17908262f65304a55a8b37ee","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-12-17 09:39:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e6f590ca815005f45277ff83bf38ffa9b7129de","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-12-17 09:55:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5d29381aee988e94b140321b9b3e3a8f4f934bd2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-17 10:17:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/699372/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c62f2d6d78dbfa927e4c39faff9ef90c2f13818f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-17 10:30:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/72/699372/1/Neutron-ML2-Sriov : FAILURE\n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/72/699372/1/Neutron-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"34deaab21392259c9e4f8bdc2793001e65ecfbbc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-17 10:41:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Neutron-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/72/699372/2/Neutron-MACVTAP-ML2-Sriov : SUCCESS\n\n* Neutron-ML2-Sriov http://13.74.249.42/refs/changes/72/699372/2/Neutron-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e92f9d3d55e92e4a28c36fc78b070627baa7c906","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2019-12-17 11:17:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/December-2019/Arista_Tempest_Testrefs/changes/72/699372/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"ed38b8659b1322f784e202b69b34e272195084fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 12:55:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3bc82c4a0f484a62935c710807bca62b : SUCCESS in 1h 02m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/165de83ea5b748af926458d16a36dce7 : SUCCESS in 37m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0bf0259d69174ea4b17521a0d18300dd : FAILURE in 2h 15m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e9cf03e5070a420285153df4555a91ed : SUCCESS in 1h 22m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d1bf247e4e6c42878cc2109ff0663eac : SUCCESS in 1h 07m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/af0aca4d4f7b4a3db300866547002e3d : SUCCESS in 57m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb31c1c3cd7d49c781b1ef6a77ca7b53 : FAILURE in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc8023eabbce4cd3806b6f3159cddc0a : SUCCESS in 26m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd27a5c4dacb492a9f15009b9a0a6f7a : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78087b84012c44f088c7d2defb9f884f : FAILURE in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/074191a928e34bd0896c0056f41b3fba : FAILURE in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185410d33f1f4904b0fc846c566743e9 : SUCCESS in 9m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa93b666a6dc46199fe7a778372d68fe : SUCCESS in 47m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ace86c661d9441b8b959fdd6cfe350d9 : SUCCESS in 1h 04m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/91b23daf969048f9a68b1756a9aac90a : SUCCESS in 2h 17m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/828cb378526340ad8dfec4d91c31b093 : SUCCESS in 1h 15m 13s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fd503631aef04a7cbc985a2f62e3a638 : SUCCESS in 1h 37m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/15b8e1d4a34244e596209994fc49466a : SUCCESS in 1h 10m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aa20a94e54fa4e0cbae440116604507c : SUCCESS in 1h 15m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3f782b705af5448999cc6d135424edc3 : SUCCESS in 1h 19m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2288fd9fcd014aabbc62059f7b623f53 : SUCCESS in 1h 41m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9709829b12974139975589bf56ddae97 : SUCCESS in 2h 02m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0a9dc2f267984464b19c19d7cf02ac8a : SUCCESS in 1h 19m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/156da8713329421aa493c47d4090a86b : SUCCESS in 55m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b6973c3566d468287c008f6e241394d : SUCCESS in 2h 19m 02s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/a019c7fad18d4c3698955b5edb1b4d99 : FAILURE in 21m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c274e4250a8a4a0796b762964ac161c5 : SUCCESS in 1h 11m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0ab7c60eaa0148f1a734ff1da35dd2be : SUCCESS in 1h 10m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/410e06155205490ba1b4359dbd777347 : SUCCESS in 56m 56s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d632731c32c445191d43865de36ea60 : SUCCESS in 1h 18m 36s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/23258bd95e1f4a3680217a568c8c951b : SUCCESS in 1h 18m 59s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ba9713a07b854fa9abf3c988586ac32f : FAILURE in 1h 42m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a52d18aace8342138341cd5cebe7d1ed : TIMED_OUT in 3h 06m 58s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/3cb9c463d6204abe8047fc5694b9cb38 : FAILURE in 1h 59m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff6cec2d755bcbe0fb5b15857f1704040b4f4b7b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:45:28.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bd087219b88bd73399e11f3c68d05a40919ad4f","revisions":{"0482faf9274e6c3ee41465e4ef5221a0ef2b0318":{"kind":"REWORK","_number":1,"created":"2019-12-17 08:49:31.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/72/699372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/699372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/699372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/699372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/699372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/699372/1"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-12-17 08:45:17.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-12-17 08:45:17.000000000","tz":480},"subject":"Doc improvment for rpc_workers configuration","message":"Doc improvment for rpc_workers configuration\n\nChange-Id: Ic226b7c60292bd5b67bd4ead1048f942e1cc1028\nCloses-Bug: #1855919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0482faf9274e6c3ee41465e4ef5221a0ef2b0318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0482faf9274e6c3ee41465e4ef5221a0ef2b0318"}]},"branch":"refs/heads/master"},"0bd087219b88bd73399e11f3c68d05a40919ad4f":{"kind":"REWORK","_number":2,"created":"2019-12-17 09:39:09.000000000","uploader":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"ref":"refs/changes/72/699372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/699372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/699372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/699372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/699372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/699372/2"}}},"commit":{"parents":[{"commit":"3e8abb9a8f6e57d73795e615a882ed4d359b8256","subject":"Merge \"[OVN] Move trunk driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e8abb9a8f6e57d73795e615a882ed4d359b8256"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-12-17 08:45:17.000000000","tz":480},"committer":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-12-17 09:36:02.000000000","tz":480},"subject":"Doc improvment for rpc_workers configuration","message":"Doc improvment for rpc_workers configuration\n\nAlso, update the .gitignore for .idea folder and\novn.ini.sample\n\nChange-Id: Ic226b7c60292bd5b67bd4ead1048f942e1cc1028\nCloses-Bug: #1855919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bd087219b88bd73399e11f3c68d05a40919ad4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0bd087219b88bd73399e11f3c68d05a40919ad4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
