)]}'
{"id":"openstack%2Fneutron~700939","triplet_id":"openstack%2Fneutron~master~Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a","project":"openstack/neutron","branch":"master","topic":"ipv6-only-deployment-and-testing","hashtags":[],"change_id":"Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a","subject":"Define new ipv6-only job for OVN","status":"MERGED","created":"2020-01-02 20:48:41.000000000","updated":"2020-02-01 02:24:19.000000000","submitted":"2020-02-01 02:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700939-1580523706642-159bbb0d","meta_rev_id":"cd4f1abb6a75126b3667056ce79c43161cb26f56","_number":700939,"virtual_id_number":700939,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-01-28 23:29:36.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-01 02:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-29 01:16:35.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-01-28 17:29:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-01-28 15:56:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-31 22:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-01-29 14:19:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-01-31 15:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-01-31 22:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 15:56:11.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-01-28 17:29:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-29 01:16:35.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-29 14:19:16.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-01-31 15:52:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-31 22:47:52.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-02-01 02:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0121c64197d9920d7f476dc8c98b22d182f7caad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2efd1546da321d0410f98e88d733262dbca03585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 20:49:07.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  mapping values are not allowed here\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 50, column 19","accounts_in_message":[],"_revision_number":1},{"id":"b1a74f562548ad0f3b71d4e9c8f7b75c1ddf478a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:53:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4920488ca8ae6ce5f11c669058a89f47ef118b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 20:53:59.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  found undefined alias \u0027irrelevant_files\u0027\n  in \"openstack/neutron/zuul.d/tempest-singlenode.yaml@master\", line 269, column 23","accounts_in_message":[],"_revision_number":2},{"id":"9e45a85c0505782e5adbf50f60488d6470361bd9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:55:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf43a48411559e1c561eee36280d51b813658aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 20:55:36.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  found undefined alias \u0027ovn-vars\u0027\n  in \"openstack/neutron/zuul.d/tempest-singlenode.yaml@master\", line 271, column 11","accounts_in_message":[],"_revision_number":3},{"id":"26bdb086512610efa15e909d051bedaae4262924","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:56:16.000000000","message":"Patch Set 3: Workflow-1\n\nI don\u0027t expect this to work, just didn\u0027t want to lose it in move from networking-ovn","accounts_in_message":[],"_revision_number":3},{"id":"c16fce31637ea5a57c4b21b49542bd7a4d491d92","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:58:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"661390385e05ffec0df883adc011ee384d886784","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-02 20:58:13.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"1c8f8fef60a45b6ea7b2ba64d497d28a44a9191e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-02 23:05:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/700939/4 : SUCCESS in 2h 07m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6436a97e044a392700f4673670e6d34fd5f55f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 00:09:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/766281ce9e6a4891841a103c86ac34d9 : SUCCESS in 45m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a92c785f8f4f440da5f2ed53bbdce4ef : SUCCESS in 28m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/91db45ad926945acb4165ddeddcac654 : FAILURE in 1h 39m 50s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e1ee9bee30314c17956d22eb8e5346dd : FAILURE in 1h 06m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f67036399383482a95031b0a3dcedb77 : FAILURE in 58m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/addd6f4721ac4248914b2374154d2f28 : FAILURE in 1h 38m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a75cd53056f44be80dd442109761ff9 : SUCCESS in 31m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5edf6aea406941bc92570a5a6dfd1a3f : SUCCESS in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f821b8d4154e73b596ab41be764dc9 : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21db62b4d0a94b0da0e09ca01176e2e2 : SUCCESS in 25m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba04c098b43244828da903ddf505b9c3 : SUCCESS in 28m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e69ebd488794c699515f082af3df7ed : SUCCESS in 8m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f965b22dc99248b3b631510d531af748 : SUCCESS in 47m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/68f76a6b464c495d8dfa91ed6cf633cd : SUCCESS in 1h 03m 19s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/96096959273b4a84a77cecb3f94810be : SUCCESS in 2h 20m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bece3131650c495ca7d5e116b26af56d : SUCCESS in 1h 49m 34s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bb5c54a95da54acbbc98bc1605857748 : SUCCESS in 1h 13m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c158b3fc3024180a0ea6a98c39b8802 : SUCCESS in 1h 04m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/89fc516440eb4f3a9c9438d21c2cfb85 : FAILURE in 59m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/99b65316a1b049e9aaa14eac28e15836 : SUCCESS in 1h 42m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cd75b471f0c64374ba1db900bd529e0a : SUCCESS in 1h 47m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/864d1ac5ab4f4e43bf1d5c068cecbc5c : FAILURE in 1h 45m 34s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3acaa701916d4bfdafa9ddd0f7744a7a : SUCCESS in 1h 13m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a87c8203d1c74dd997d2a22034df54bb : SUCCESS in 56m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f9c918816a94b95a14cf8eec68705d3 : SUCCESS in 2h 23m 11s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/97eba91e263d43ba8a5a6a469f71bc6a : FAILURE in 33m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/b921c28c6a54416ba6000d96544a0d69 : FAILURE in 22m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/210f1a07e5d248b48521b5cab94e6597 : SUCCESS in 1h 00m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8ffed52f8eb248c3ba6bf1a194280967 : SUCCESS in 1h 09m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/df58a2eebd794aa3b912b66436040cf5 : SUCCESS in 48m 50s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c674d7efd1f545e1858626fc0a050ce0 : SUCCESS in 1h 01m 01s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d095b403b784e34b6686ddf631839ca : SUCCESS in 1h 21m 49s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2d134136208a4fecb8fa2ecf369259a1 : FAILURE in 1h 40m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9127f32be43f4f0ca928caa88de1bb8e : TIMED_OUT in 3h 05m 20s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2228322b410648ecaf9a514520cd9b26 : FAILURE in 1h 55m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"608d6b2572d2a240fe08799c22ae291c6b959cd3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-13 21:40:16.000000000","message":"Patch Set 4:\n\nNeeds a rebase to master once https://review.opendev.org/#/c/701392/ merges","accounts_in_message":[],"_revision_number":4},{"id":"266b8103d550f6abf4c063de0144da71dd273104","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 14:51:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8f442254f7f94754871d138b6a57b7611a375c68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 18:17:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4625b7d57aac41ad85f055d6ee0bd6a2 : SUCCESS in 1h 12m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e75f832c3a64f18903110cc2499ede2 : SUCCESS in 39m 44s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/d163aa2105054a4ba71b6cd462efd9fb : FAILURE in 1h 35m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/61949b7359a94cf69c4d8f202d56d40c : SUCCESS in 1h 23m 55s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/eed9d43e407b4caba081ed2605e12c80 : SUCCESS in 1h 24m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/93281e7f8c514dbb9ebc219bcf36e13a : SUCCESS in 1h 09m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2336bbf52cb4dbaa45e5f9a10170eca : SUCCESS in 31m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8d721a9b3c4b8c8661c880599761b3 : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f751c855b262413fb149fab4a723a6ca : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af170182f4684f40b4f49ff892a3dcc8 : SUCCESS in 35m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/467264c6659e4929b1685bd51ddfcc40 : SUCCESS in 38m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17847d0946724dd5861e5d0f1a94655e : SUCCESS in 9m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/30a60817dbef4b40ad884ef730d0c53e : SUCCESS in 45m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/20f59269848448dbbfdb3d74b0733594 : SUCCESS in 1h 09m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3dc6915fdc72488f9a151eb244e6f7e9 : SUCCESS in 1h 46m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51e17cc9fc244279940b25db53aa95ba : SUCCESS in 1h 29m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/98975c21edb54f22abd275e9513685b0 : SUCCESS in 1h 16m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6adbd836db064e9482ce2b1951d9aaf5 : SUCCESS in 1h 20m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a2cb8d70182d4f21984e830f5eff5916 : SUCCESS in 1h 17m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20aec3517daa43c1a52bdccf4ca02f8f : SUCCESS in 1h 20m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae189edb2ce94166ac8f38320018eeac : SUCCESS in 1h 50m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d4a47f3d9ca4587a89c16d3213cd1c2 : SUCCESS in 2h 01m 31s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8dfc87e630b2448495fa36d0fd8d0941 : SUCCESS in 1h 16m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/58bfb97c49d2484e83248c3ec6d31776 : SUCCESS in 1h 01m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/25f4efc79be04277bf0bc5e4b1909661 : SUCCESS in 2h 23m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc4142dec6d04a8d893adf56e9213b7d : SUCCESS in 2h 30m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/79fdece0a79a44b4810c6852777e3611 : FAILURE in 31m 39s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/5ce4c985f4fb42dead756d7d7e439746 : FAILURE in 16m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f39cb379725e4bf0be98d80b2ab5209d : SUCCESS in 59m 48s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bdc34e560b1647f09d771dbfbfb56b8b : SUCCESS in 1h 06m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd825b5dce594e5ea7eb580c21caf6cc : SUCCESS in 46m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7efd33d77db34e77a5a46522cd5d2d4f : SUCCESS in 1h 07m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/1abeb1e0bf5c45f6915fdb9f149b847b : SUCCESS in 1h 27m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2c4b2c6bec2441b7a5bf50b3dc4f1bdd : FAILURE in 2h 48m 22s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9ec45c3a2300407fb5e748d2ac7aac6e : FAILURE in 2h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"07aeb07e95035a08cecabdd01c7c3ed295a64925","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 19:20:40.000000000","message":"Patch Set 5:\n\nStill see OVN things on 0.0.0.0, found one change in devstack plugin that needs an update.","accounts_in_message":[],"_revision_number":5},{"id":"183a51a55ca60c80840a60e27b971e48a98d1ddd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 19:20:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bc48bc1a834110f59a9e0410420f32922f08ed06","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 21:08:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6fb02f7e7a6e4025821d9908bbcb373f00b6f9eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 22:08:02.000000000","message":"Patch Set 7: Workflow-1\n\nI am going to take the two ovn_agent changes and put them in https://review.opendev.org/#/c/703685/ to keep the bug fixes separate from the job.","accounts_in_message":[],"_revision_number":7},{"id":"5d2cca5ac5226a2e59d7cfd31c55f4f93c6ba7c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 22:22:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4133505c491b5338899117ba03a412f5faed4e22","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-22 22:24:46.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"61915e5650a269d80f9ce8c67b2a79c0a6dff79c","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-23 00:06:39.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/700939/9 : SUCCESS in 1h 41m 12s","accounts_in_message":[],"_revision_number":9},{"id":"272aa2cbb47c9f201d5b23b8a2ee70f4cf20cb72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 01:35:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4481b41955364bb686994a2e447d43cf : SUCCESS in 1h 11m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9cfa63741f944c8934b2e8c8b238182 : SUCCESS in 38m 30s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/cc1413b9ba2b4282a5b371494384df2d : FAILURE in 1h 55m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/e908efd5d0cc4db9bb00a6befd81b8b3 : SUCCESS in 1h 17m 28s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/03c477dbc52948d2b41555c4633a0d13 : SUCCESS in 1h 11m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa48ddeeff1248fea31ddd977c29c254 : SUCCESS in 56m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca683fd4e33f45a88cda5dc43c91814a : SUCCESS in 32m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e44b376518da45bab1570a6f7689c473 : SUCCESS in 23m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c214531188d435b9ecc348c28d733c5 : SUCCESS in 14m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba981c8198845f39e0215fc7ab21f85 : SUCCESS in 39m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/515950eb04bd4b4a88bec6e75ed15261 : SUCCESS in 39m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33cea30bbb44a0aa3b269808081ee41 : SUCCESS in 10m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f3b407cd91a4d9f93dee60a9133922b : SUCCESS in 57m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/771bae4de04d4779b9c9e9dea00364fb : SUCCESS in 1h 21m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/595dad9b9f9247498ad4b6474c4a1eac : SUCCESS in 2h 04m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dfc395d3d4ba44aa9cac0174b927e106 : SUCCESS in 1h 32m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/039739c7fce34553bd652c46c808d9a7 : SUCCESS in 1h 10m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd38f77313a344d5a5fad615cf483440 : SUCCESS in 1h 00m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6b767bce05cd4d6ebacbd705905581af : SUCCESS in 1h 04m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/984ce65b46de485fa4e0d251b6a0531e : SUCCESS in 1h 05m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/805f64eefc104b17a213cee04430e402 : SUCCESS in 1h 36m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d388a94231a04a9ab05f2509cf162720 : SUCCESS in 1h 59m 50s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/198e220d05db431794c3a288ea62c3a0 : SUCCESS in 1h 18m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66235d0528c34fb7a76bb125ea285b15 : SUCCESS in 58m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ca8a2b7f92b416eabe84193db4487a6 : SUCCESS in 2h 13m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/1da163d7836e4be18d4ea47a0e761923 : SUCCESS in 1h 55m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d19e219da4cf474981b2e30754d533a3 : SUCCESS in 1h 48m 24s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/0c972202308148298bce88c862fdfe0c : FAILURE in 17m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75da01f5a36a4fe392e98faa586c2196 : SUCCESS in 59m 09s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fe932ea8a5974906b397df751d9c8ba1 : SUCCESS in 1h 10m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/059fe7ee93c342dc97552b961b01050d : SUCCESS in 45m 45s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f2845f818294bc79b81faee3abcaabc : SUCCESS in 1h 04m 58s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/770ae1593d034c419d15acf95ab12941 : SUCCESS in 1h 15m 00s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/02a2fff190af4478a49ee6cbfb2ea258 : TIMED_OUT in 3h 04m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e82fa021dc2f46ef81d0139b415d5203 : FAILURE in 2h 07m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7f07d4217f5cfc519758f5798609ad1e9b6a9121","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 20:13:15.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"fbaed88a9c2fa3fd5ea36e4b5a6d466bfe92f82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 20:13:31.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"3b7fbb7c9e78711c09de94f543bdfb9b9bcc05de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-24 00:06:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f448b9c43e8ededc16faef3a2496b03b40896799","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-24 01:50:32.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/700939/10 : SUCCESS in 1h 43m 44s","accounts_in_message":[],"_revision_number":10},{"id":"854fab900f752d87e1c073e68e35aae9ea47b4ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 02:53:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd3cf1c143f24406b9726379eb942c0e : SUCCESS in 54m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58955fd732dc42a89f66fad117ed3e89 : SUCCESS in 40m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/e43981d971414e94b30f64bc5e88db94 : FAILURE in 2h 21m 21s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a1095e6353bc4610a9807a2b24632092 : SUCCESS in 2h 15m 56s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/68232316fbea4c3faae1b395e220984e : SUCCESS in 1h 43m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bca38acd38de48eab2686cecbb28a39a : SUCCESS in 1h 53m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43df5c275cde436b92d9ce9ab79fbf69 : TIMED_OUT in 51m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0510f7de5f2e4f6d94eefbc8458455f9 : TIMED_OUT in 41m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029c754a86514a668eebd9b60fd304bd : SUCCESS in 24m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f10a9f43372b49a48e9fdd3535ba2dc4 : TIMED_OUT in 41m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e03bbb1ca3ff4a76b034a71ec088ed6a : SUCCESS in 30m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd9aa8588a63481c8a53c26c87b6faad : SUCCESS in 10m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/83219910be664dc5b136b328d53bb258 : SUCCESS in 50m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aa324b8cebf34605b9e9bd25dd2a7d96 : SUCCESS in 58m 22s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b6bdf0d279f34f39876d9eed10508baf : SUCCESS in 1h 54m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/945d7007c44b484bb057dbe344f5cc2f : SUCCESS in 1h 19m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b65bdfc2d55445ed87f7e050e6841183 : SUCCESS in 1h 04m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e2edf61c6f444e6b75ba7421e566df4 : SUCCESS in 1h 13m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/74fb9baec85a486882cdc6bce57d64db : SUCCESS in 1h 08m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4666550e2cb4437092d55373e9ede5b4 : SUCCESS in 1h 09m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a61ced938e554c009d133b79da8090a6 : SUCCESS in 1h 50m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4ca3d8b47c3e4ac6b934250d26f10112 : SUCCESS in 2h 01m 29s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f3e1fb7dce004b79bc36ff6af7c1ccc2 : SUCCESS in 1h 11m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1e76152cec2247a8b654d75b1629e31a : SUCCESS in 57m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b6447e868f24feebd36b2edef20e05f : SUCCESS in 2h 14m 22s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/076c38648060468ea5b85d220b0f5b9f : SUCCESS in 1h 42m 00s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/815f5b5a34c340478001e18f2486d530 : SUCCESS in 1h 45m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21aa42d8fc5c4c2db75cddaf21ea90b8 : SUCCESS in 1h 20m 45s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2be2f73fbc0d4ac9a601a0dba5fe6694 : SUCCESS in 1h 17m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28f7e097cd4649c19acd1dafbed72593 : FAILURE in 1h 12m 00s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/932185e8935a4bd994936843b2247e14 : SUCCESS in 1h 09m 25s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a02a0b5d3c214265b68ca238c5ff0504 : SUCCESS in 1h 32m 41s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bf99a34b4b394410b6e6e726caef4208 : FAILURE in 2h 38m 59s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e861316c570d45fd896651fc304bc121 : FAILURE in 2h 18m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d1c07ab1e2b8370877e76c4de8968e9e6071b101","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-24 13:03:22.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"5c9ba2196c723203dc53fec37983849999cd3ffd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-24 14:48:21.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/700939/10 : SUCCESS in 1h 44m 30s","accounts_in_message":[],"_revision_number":10},{"id":"d1c7feebc87eeb952995ccc5366b71536d285d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 16:19:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/833c72c229c446ff9e217f8edf5b234a : SUCCESS in 58m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/327cd682092947eeae4b15a937e62e88 : SUCCESS in 34m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8e29341ef96b488890d86435011f8c23 : FAILURE in 1h 33m 56s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/301d545e94c7416480a6fb35e6985c8e : SUCCESS in 1h 34m 05s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/01a31f5548a24cf882ac652878040e01 : SUCCESS in 1h 13m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f7f861b2f23248d89b6cee782b8882ee : SUCCESS in 1h 07m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08025d67b239410ba2624a22554aaca0 : SUCCESS in 36m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73ccd93ea8e0464c8a3a112b59348248 : SUCCESS in 30m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2094b9954b594f73a3860d36c125e163 : SUCCESS in 18m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fa686d180044e99b094b7574ce1c4cf : SUCCESS in 37m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e86b126096ab44a7bf8604e798aa9224 : SUCCESS in 32m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1490428ec83841918eb6086207296a12 : SUCCESS in 27m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc1a3301e6d4df1b19be4584ef874e5 : SUCCESS in 9m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e66769a0f0204576b96ce120a89bc93b : SUCCESS in 40m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/493b823cd6284bf581b9bed60cee4278 : SUCCESS in 1h 22m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cbd5a9e333994d0c8f3cd54d9fbfec2d : SUCCESS in 2h 24m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0896b3e800444109ade56021ede3b62c : SUCCESS in 1h 26m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f08c830a6cd34915a9d7bcc06168ef50 : SUCCESS in 1h 32m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/07ffb61ba91e4c00bebb9f86b668472f : SUCCESS in 1h 02m 32s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/156db571a1974c2ea248dd296aded060 : SUCCESS in 1h 06m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/743bb3590ee04e7d9551413f59bd955c : SUCCESS in 1h 21m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e4711df900b44553bd4f967094e02a3c : SUCCESS in 1h 51m 29s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2d1f730959af4d4bbb66d422b29ec12b : SUCCESS in 2h 04m 59s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f50dfb25f7ff4ae1993568ead3e705f0 : SUCCESS in 1h 06m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ce5a608cb4db418dad7c896216f7a6bb : SUCCESS in 1h 08m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26b2fdc9daab44cf8e4d4a63c5838d58 : SUCCESS in 2h 11m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/335e29bde3ad475fa13bd4ee692dbf27 : SUCCESS in 2h 28m 20s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/026b9271d64149e2af549963c711b062 : SUCCESS in 1h 47m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a23f49482a89490eabd495bef4bfe7bd : SUCCESS in 1h 20m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/eb8e0889a5da468ea351264eb2baeaf4 : SUCCESS in 1h 13m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/690f2637588a48ff8b7b1ad3e2cf397a : SUCCESS in 46m 32s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcc05bf2eb594414a083d473c92e8d3c : SUCCESS in 1h 10m 53s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d7363959a6a4b57bdb55aa81f99b62c : SUCCESS in 1h 15m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0157378939f14b41a1ae45516c8dcec8 : TIMED_OUT in 3h 06m 36s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/183ddd85a4c1435a9e95f2767089e45a : FAILURE in 2h 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0adb897e9dfe4676cc21961e86a264d8e318f56c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-28 14:55:11.000000000","message":"Patch Set 10:\n\nhmm, I\u0027m ok with the change, but I don\u0027t see any IPv6 scenario tests runs.\n\nI think we can enable them even when those probably will be failing, because of missing connectivity between instance and devstack node for OVN.","accounts_in_message":[],"_revision_number":10},{"id":"2f11b3de18f15971f56efa6f5ede6c52eea1b350","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-28 15:56:11.000000000","message":"Patch Set 10: Code-Review+1\n\nThanks Brian!\n\nI can see in the logs things are being setup using IPv6 indeed:\n\n2020-01-24 13:36:43.309195 | controller | ++ /opt/stack/neutron/devstack/lib/ovn_agent:configure_ovn:546 :   iniset /etc/neutron/neutron_ovn_metadata_agent.ini ovn ovn_sb_connection \u0027tcp:[2001:4801:7825:103:be76:4eff:fe10:e6]:6642\u0027\n\nhttps://bfc2304b36c89dd5efde-d71f4126f88f4263fd488933444cea49.ssl.cf1.rackcdn.com/700939/10/check/neutron-ovn-tempest-ovs-release-ipv6-only/026b927/job-output.txt\n\nVoting +1 cause I don\u0027t know about the rules of adding another job in the gate (or maybe it should be a periodic? idk).\n\nI will leave it for other Neutron cores to decide","accounts_in_message":[],"_revision_number":10},{"id":"1b654372d5a87bf17db76965d662f5cb564cb7d6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 16:12:33.000000000","message":"Patch Set 10:\n\nMaciej - yes, this is only about running a job on a devstack with IPv6 API endpoints and IPv6 addresses for services.\n\nhttps://governance.openstack.org/tc/goals/selected/train/ipv6-support-and-testing.html\n\nThe one thing I found it doesn\u0027t do is use IPv6 for tunnels, eg VXLAN or GENEVE is still over IPv4.  There\u0027s a couple of things that need fixing for that to happen, I\u0027m working on it in my \"spare time\" :)","accounts_in_message":[],"_revision_number":10},{"id":"b1dcebae50cc7f52e0a4465904ea78ba262aacdd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-28 17:29:13.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4925f96e5fe67e9192fe352801794da80b062cfc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-28 23:29:36.000000000","message":"Patch Set 10:\n\nrecheck\n\nUpdated ovn_agent script in dependent patch","accounts_in_message":[],"_revision_number":10},{"id":"7c4f1f36ea918582d77ebe1d101d49576b03e08b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-29 01:16:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/700939/10 : SUCCESS in 1h 46m 31s","accounts_in_message":[],"_revision_number":10},{"id":"3951a8b32e8a6f04aa0d8dd57e8c181ce8f8caa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 02:13:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/914c51bbafcd4264b683aa2849fdcadc : SUCCESS in 1h 05m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cc4c39c22bf44352b55a0d1636ae6721 : SUCCESS in 35m 01s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c1e44b57aec5483da1b349512479b5aa : SUCCESS in 1h 23m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/df5fdc9af65c4286bd699c66a574bac2 : SUCCESS in 1h 10m 02s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d30d98b434d0497ebf2a4d7df7d2653f : SUCCESS in 1h 08m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2438f847b514407ebfbc64d2eb118965 : SUCCESS in 1h 04m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10dbd44750564f85a57935d36c618140 : SUCCESS in 46m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/753c17c07ad24c258f1e117077a45620 : SUCCESS in 37m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07b111cc2784e9a8e584f138e419ff6 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1143d816c2dc4be08b4f532764d18c48 : SUCCESS in 28m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6047c8798aba4970afc81741792bfcb8 : SUCCESS in 30m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce82224f96354c2cb4d2cb2f32d3a266 : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd3186fca8f493593724590eeae3ae5 : SUCCESS in 7m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e34de4cbc03403b981640f94df0ab30 : SUCCESS in 48m 41s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/05366731eb4548698e32de6c96598985 : SUCCESS in 1h 08m 44s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4c84f29a166d413b98eb3e60df0d9924 : SUCCESS in 1h 49m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f680d5779c7345f2a39934b617f26e06 : SUCCESS in 1h 15m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ee9c1c07a0cf47a6b41481e09058c8cb : SUCCESS in 1h 40m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/082da8a104f44a458f44dee4a20b951b : SUCCESS in 1h 05m 50s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9a7781470944b67bba51de26aae7051 : SUCCESS in 1h 06m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3f1c2ffbb95b4f138218e5ae60c5ee8c : SUCCESS in 1h 11m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a825587e390414fba06264dd7dd0842 : SUCCESS in 1h 42m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e5ca4afcc1649bbb6f960f9447c58d0 : SUCCESS in 1h 55m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b6ee0929ea55408eadd89d5cc182830b : SUCCESS in 1h 10m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d88a88eeb2984342b7135674e3fac2a8 : SUCCESS in 1h 18m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e95549918ae6458eae83e016956d1c69 : SUCCESS in 2h 15m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eeec874048cb4740a6a36fc9521d6588 : SUCCESS in 1h 53m 52s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/83c2bcae17804bfc89972fba07520ba7 : SUCCESS in 2h 05m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8a5e4f9626e4118b62fe17ff9de54d3 : SUCCESS in 59m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b355ccebd84348eaa6d8024b4e379f60 : SUCCESS in 1h 23m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a45c13dfaa294def8f26329f8ad73032 : TIMED_OUT in 2h 13m 48s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/98cdd468640b489fb43c8bf8fceb4d57 : SUCCESS in 1h 04m 17s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c654fa4206c45b1a43f762183b9ece4 : FAILURE in 1h 02m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/887a6d75e1c44e2d8f4c3d8ed87287f4 : FAILURE in 2h 27m 50s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8e5507e0fd724ab89380b41fd6f96165 : FAILURE in 2h 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"96996919e378d9289339535f651ce6e049f11e3c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-29 14:19:16.000000000","message":"Patch Set 10: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":10},{"id":"12d61ed300d5f2940ae90601a67b22319efebadf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-31 15:52:55.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ca24337d738642c55c500cea6228e7f02ba60fa5","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-31 22:47:52.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"afe66f6a437477299aff3b04f147369f73712d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 22:48:19.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"306ad43fddd8f9a36fa874fd7564fbae5a247890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:21:46.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/eeebb40df5514167901f0384c9bfeec8 : SUCCESS in 48m 23s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8858d9cad6f9438787c331651c848bd1 : SUCCESS in 1h 06m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/691071d17669469597259ffa52a16320 : SUCCESS in 57m 15s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6dd229b1ab6240aba05126e1c81a852d : SUCCESS in 54m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aec769c6f9e4fbe934728051a21e1b8 : SUCCESS in 22m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38b0bb8bbb74deea9d1f85bc4702875 : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d2df38a03dc466ab595ac18293a5955 : SUCCESS in 24m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e56f8ccd695f40c5b6e81efb4bd376e2 : SUCCESS in 27m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6826ba180feb49a9bc4a9d6d02e6df27 : SUCCESS in 8m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1aacc136874e4757ac41757558843182 : SUCCESS in 51m 36s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f313ee9d991847e587432b9445739e70 : SUCCESS in 1h 02m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77692704d6044965a3224e75160e3b1e : SUCCESS in 1h 05m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/74baaed3c4e34628804ea24dad90d986 : SUCCESS in 1h 09m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/975a35daa40c4107830244f34356a0a9 : SUCCESS in 1h 24m 16s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9ae65d1bfb7c41b48b6b58fcf7f94ee2 : SUCCESS in 1h 07m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ef1b01bb4fb48c3a0f31e88d0a92668 : SUCCESS in 1h 03m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01cad89a85cc424baeae6e7e906902e0 : SUCCESS in 59m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f71f88d2bf043b7a004d7c34bf6770b : SUCCESS in 2h 18m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00d150347c68434994c6e343277fb8ab : SUCCESS in 1h 14m 50s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/40768101ff454274b7461921f043d056 : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":10},{"id":"46ecd93607ac767926f6cf8d88faafa94485e0ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"506ce7c3e673cac92089409c1724ce3849499ecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:24:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1d1bb193d5457b9cfca51627f46284 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fc507c701c08f76be02c20a6143475f9676e3de0","revisions":{"f352d66d5b4df0345dccc1968a2305dffd2536c3":{"kind":"REWORK","_number":1,"created":"2020-01-02 20:48:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/1"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f352d66d5b4df0345dccc1968a2305dffd2536c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f352d66d5b4df0345dccc1968a2305dffd2536c3"}]},"branch":"refs/heads/master"},"c342eb33b277a25d2f0ee2faf4a5488dd3392884":{"kind":"REWORK","_number":2,"created":"2020-01-02 20:53:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/2"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:53:27.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c342eb33b277a25d2f0ee2faf4a5488dd3392884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c342eb33b277a25d2f0ee2faf4a5488dd3392884"}]},"branch":"refs/heads/master"},"cccf8d7cd431429e16a91ce92ac94695cb4a496b":{"kind":"REWORK","_number":3,"created":"2020-01-02 20:55:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/3"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:55:01.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cccf8d7cd431429e16a91ce92ac94695cb4a496b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cccf8d7cd431429e16a91ce92ac94695cb4a496b"}]},"branch":"refs/heads/master"},"2c278a270c7600823572f54e52256d318aac169d":{"kind":"REWORK","_number":4,"created":"2020-01-02 20:58:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/4"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:57:37.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c278a270c7600823572f54e52256d318aac169d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2c278a270c7600823572f54e52256d318aac169d"}]},"branch":"refs/heads/master"},"d7fca94dd787056835ebeb8e86e5e3b9fb9f9bba":{"kind":"REWORK","_number":5,"created":"2020-01-22 14:51:29.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/5"}}},"commit":{"parents":[{"commit":"4b48de8e8895cce4c5496f41be3679c1f87297fd","subject":"Merge \"Allow to kill keepalived state change monitor process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b48de8e8895cce4c5496f41be3679c1f87297fd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-22 14:51:06.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7fca94dd787056835ebeb8e86e5e3b9fb9f9bba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7fca94dd787056835ebeb8e86e5e3b9fb9f9bba"}]},"branch":"refs/heads/master"},"8bc9d7a4ed6f06dcc5dd12005e677434fe26e0de":{"kind":"REWORK","_number":6,"created":"2020-01-22 19:20:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/6"}}},"commit":{"parents":[{"commit":"4b48de8e8895cce4c5496f41be3679c1f87297fd","subject":"Merge \"Allow to kill keepalived state change monitor process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b48de8e8895cce4c5496f41be3679c1f87297fd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-22 19:19:14.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nAlso required to change default listen address to $SERVICE_HOST\ninstead of 0.0.0.0.\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bc9d7a4ed6f06dcc5dd12005e677434fe26e0de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bc9d7a4ed6f06dcc5dd12005e677434fe26e0de"}]},"branch":"refs/heads/master"},"3c4437a05e22098ae124b57ce339f478fb07f07c":{"kind":"REWORK","_number":7,"created":"2020-01-22 21:08:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/7"}}},"commit":{"parents":[{"commit":"4b48de8e8895cce4c5496f41be3679c1f87297fd","subject":"Merge \"Allow to kill keepalived state change monitor process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b48de8e8895cce4c5496f41be3679c1f87297fd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-22 20:51:41.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nAlso needed to change some default IP addresses to use\n$SERVICE_HOST/$SERVICE_LOCAL_HOST instead of 0.0.0.0/127.0.0.1.\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c4437a05e22098ae124b57ce339f478fb07f07c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c4437a05e22098ae124b57ce339f478fb07f07c"}]},"branch":"refs/heads/master"},"93355f83cd8b759a4bd2b49f00b3991101f191c5":{"kind":"REWORK","_number":8,"created":"2020-01-22 22:22:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/8"}}},"commit":{"parents":[{"commit":"4b48de8e8895cce4c5496f41be3679c1f87297fd","subject":"Merge \"Allow to kill keepalived state change monitor process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b48de8e8895cce4c5496f41be3679c1f87297fd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-22 22:22:36.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93355f83cd8b759a4bd2b49f00b3991101f191c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93355f83cd8b759a4bd2b49f00b3991101f191c5"}]},"branch":"refs/heads/master"},"100a444aea19c0c70678549c4005902f5d3dc85e":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-01-22 22:24:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/9"}}},"commit":{"parents":[{"commit":"4b48de8e8895cce4c5496f41be3679c1f87297fd","subject":"Merge \"Allow to kill keepalived state change monitor process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4b48de8e8895cce4c5496f41be3679c1f87297fd"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-22 22:24:30.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nDepends-on: https://review.opendev.org/#/c/703685/\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/100a444aea19c0c70678549c4005902f5d3dc85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/100a444aea19c0c70678549c4005902f5d3dc85e"}]},"branch":"refs/heads/master"},"fc507c701c08f76be02c20a6143475f9676e3de0":{"kind":"REWORK","_number":10,"created":"2020-01-24 00:06:07.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/39/700939/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/700939/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/700939/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/700939/10"}}},"commit":{"parents":[{"commit":"e15d1003f845a347b80fd2410b18a1ffd49baa39","subject":"Merge \"Remove networking-midonet job from Neutron\u0027s check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e15d1003f845a347b80fd2410b18a1ffd49baa39"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-02 20:42:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-24 00:05:56.000000000","tz":-300},"subject":"Define new ipv6-only job for OVN","message":"Define new ipv6-only job for OVN\n\nAs part of Train community goal \u0027Support IPv6-Only Deployments and Testing\u0027[1],\nTempest has defined the base job \u0027devstack-tempest-ipv6\u0027 (adding\nin Depends-On patch) which will deploy services on IPv6.\n\nThis commit adds the new job \u0027neutron-ovn-tempest-ovs-release-ipv6-only\u0027\nrun on check queue which is derived from \u0027devstack-tempest-ipv6\u0027.\nA future change can make this voting in the gate.\n\nVerification structure will be:\n- \u0027devstack-IPv6\u0027 deploy the service on IPv6\n- \u0027devstack-tempest-ipv6\u0027 run will verify the IPv6-only setting and listen\n  address\n- \u0027networking-ovn-tempest-ovs-release-ipv6-only\u0027 will run the tests.\n\nStory: #2005477\nTask: #35908\n\nBased on https://review.opendev.org/#/c/673488/ from networking-ovn.\n\n[1] https://governance.openstack.org/tc/goals/train/ipv6-support-and-testing.html\n\nDepends-on: https://review.opendev.org/#/c/703685/\nChange-Id: Ia0dcf47ead268aa4b265904c9e1ae1e38d77456a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc507c701c08f76be02c20a6143475f9676e3de0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc507c701c08f76be02c20a6143475f9676e3de0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
