)]}'
{"id":"openstack%2Fneutron~701409","triplet_id":"openstack%2Fneutron~master~I911469d46fccf13bc7ead7a103a7d6e4e0ede7c3","project":"openstack/neutron","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I911469d46fccf13bc7ead7a103a7d6e4e0ede7c3","subject":"Complete dropping py27 support goal","status":"MERGED","created":"2020-01-07 17:43:13.000000000","updated":"2020-01-16 16:01:16.000000000","submitted":"2020-01-16 15:57:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701409-1579190257215-1f963bb9","meta_rev_id":"13f573388a9406efe0cc88c364ef4fd534b60d35","_number":701409,"virtual_id_number":701409,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 15:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-16 14:52:11.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-01-15 08:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-01-15 08:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-01-16 12:51:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-01-15 08:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 19:10:35.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-01-08 22:18:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-09 08:09:30.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-14 13:49:55.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-01-15 08:26:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-15 08:35:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-16 12:51:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-16 14:52:11.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-16 15:57:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2c5d0ad17a579b71622ce6721b275c2faf1a434","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-07 17:43:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"010de674b3a69279f5d3aa9be147bd5c36aeeca2","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-07 19:34:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701409/1 : SUCCESS in 1h 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"df2f06ef888c428aa70675fb30b3e40bf7ac1f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 20:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8f8d1163654241b9a075dfa75a9e887a : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47a410e1de974e279f8afe379f6ac6ce : SUCCESS in 28m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0a09873abe24dadae5a68b77633ce79 : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24c37c27c064d0ab14243ba8f36632f : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94d42f0258b04d1b9c829178df186a27 : SUCCESS in 35m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a32c21cb323f4f32b5b54e930d30cb40 : SUCCESS in 24m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55fe6f541b814cf887edf6e8670a3cd5 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9168cedf3c1048299ccaaba1a6eb94f0 : SUCCESS in 14m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23d881e517e14ca3af4cc5c539328acc : SUCCESS in 41m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bf78246040f5446aace4707ad0293020 : SUCCESS in 1h 17m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2b74ebd24e4f494186de7a763160b3cc : SUCCESS in 1h 08m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b606f656643549d898d419184a17d56e : SUCCESS in 45m 06s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee9b78e4d36e45bea736492e66f47114 : TIMED_OUT in 2h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a41edfbc4d186eb41ead6b3066295ea053755bf","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-01-08 19:03:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"659d8c13e39965106bbead33c206a62a024116f4","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-01-08 19:10:35.000000000","message":"Patch Set 1: -Code-Review\n\n* In neutron/test-requirements.txt: the \u0027;python_version*\u0027 can be removed now since we have dropped py27. See these lines:\n\n  astroid\u003d\u003d1.6.5;python_version\u003c\"3.0\" # LGPLv2.1\n  astroid\u003d\u003d2.1.0;python_version\u003e\u003d\"3.0\" # LGPLv2.1\n  pylint\u003d\u003d1.9.2;python_version\u003c\"3.0\" # GPLv2\n  pylint\u003d\u003d2.2.0;python_version\u003e\u003d\"3.0\" # GPLv2 \n\n* In neutron/doc/requirements.tx, same for this line:\n\n  sphinx!\u003d1.6.6,!\u003d1.6.7,!\u003d2.1.0,\u003e\u003d1.6.2;python_version\u003e\u003d\u00273.4\u0027 # BSD\n\n* In neutron/lower-constraints.txt, same for this line:\n\n  monotonic\u003d\u003d0.6;python_version\u003c\u00273.3\u0027","accounts_in_message":[],"_revision_number":1},{"id":"3867dd4f02562566514706304bba58e2e8c8f5f2","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-08 22:10:19.000000000","message":"Patch Set 1:\n\n\u003e * In neutron/test-requirements.txt: the \u0027;python_version*\u0027 can be\n \u003e removed now since we have dropped py27. See these lines:\n \u003e \n \u003e astroid\u003d\u003d1.6.5;python_version\u003c\"3.0\" # LGPLv2.1\n \u003e astroid\u003d\u003d2.1.0;python_version\u003e\u003d\"3.0\" # LGPLv2.1\n \u003e pylint\u003d\u003d1.9.2;python_version\u003c\"3.0\" # GPLv2\n \u003e pylint\u003d\u003d2.2.0;python_version\u003e\u003d\"3.0\" # GPLv2\n \u003e \n \u003e * In neutron/doc/requirements.tx, same for this line:\n \u003e \n \u003e sphinx!\u003d1.6.6,!\u003d1.6.7,!\u003d2.1.0,\u003e\u003d1.6.2;python_version\u003e\u003d\u00273.4\u0027 # BSD\n \u003e \n \u003e * In neutron/lower-constraints.txt, same for this line:\n \u003e \n \u003e monotonic\u003d\u003d0.6;python_version\u003c\u00273.3\u0027\n\nDoing that will cause the openstack-tox-lower-constraints job to fail, because of how it checks requirements.  This can be done after the version constraints are removed from the openstack/requirements repo, which will happen in stage 3.","accounts_in_message":[],"_revision_number":1},{"id":"47c65f833e9988c75ee8df8d2547c1882f221571","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-08 22:18:49.000000000","message":"Patch Set 1:\n\nStrangely enough, I was able to remove those python_version strings in the octavia repo (today) and the jobs are passing, and I thought they\u0027d fail based on my past experience with lower-constraints and the requirements job.","accounts_in_message":[],"_revision_number":1},{"id":"e6957b875d715e0318d352ae04f96b90329be077","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-09 06:41:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ba0d0cf7b140e9fd287a61ebc18950d8b2df87c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-12 10:28:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa11fe9a064342a1375f999dbda5477110523e98","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-13 18:03:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db7fc111e57dfd410c0a24f8a5e1e277c438a42f","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-13 18:04:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6daa960c16919a9d0e87c60293847d22de58c111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 19:28:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e1d04be6f44d4c67a3181a2e770a84c9 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ade0f44fe67f4c0790c681703e754320 : SUCCESS in 30m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/152e6ec2790442fead861ae1ea75e533 : SUCCESS in 23m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42dbeb37c5224767b78ebf1cfdfb8e87 : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b341fa819d44663b3f4660ff73cad97 : SUCCESS in 24m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d71e32c913f434198a84d06752bbce5 : SUCCESS in 30m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc31ea87e9641d188d53e05c3c28d43 : SUCCESS in 11m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c318481f10674890920621cbdd6ef822 : SUCCESS in 36m 37s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e32245e853894a3882abe988b7788c04 : FAILURE in 1h 21m 08s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e819882ab788442ca6d5750d55df7fdc : SUCCESS in 1h 16m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da11e2becc17463e873d223f4b9ac26e : SUCCESS in 35m 59s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5255f9473df4988bc6d6ff179698b0b : SUCCESS in 1h 04m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"885df6dcecb5cdec0299a82e8d158ae0ac0d6648","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-13 19:49:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701409/2 : SUCCESS in 1h 45m 18s","accounts_in_message":[],"_revision_number":2},{"id":"241caf6b71d3697e1f3b18663a9f5daaef27dd6c","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-14 03:39:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b485dcdd70ad17882edb3621713f3dcc5bf844a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 04:54:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a404f155ebf4b1cb30eb548deba7be0 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6b72fbb302046abb7fc9460537b27c5 : SUCCESS in 37m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c46a615c830547499a5fd0c942bcf712 : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfda0c123c53474cbf6828d54e91979e : SUCCESS in 13m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/041dfe527c4c407989537079733e7b92 : SUCCESS in 24m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f931de8f8100429ebbd3faa6224edf08 : SUCCESS in 32m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5eb55cf775406496f473254d28b49f : SUCCESS in 7m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a9a3ca0c0940439c9fff0216a7a0df23 : SUCCESS in 43m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/215550b641d8451e8da241eee84a122e : SUCCESS in 58m 32s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/628406c5f6d44f07a6410fd27a83fe0d : SUCCESS in 1h 09m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3348dd47a064595b30c2f3c86be2444 : SUCCESS in 41m 28s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2f803642d31439083e743d92e8468d3 : SUCCESS in 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94c5e599999a0c20714564d804ffd57d1a814a49","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 05:40:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701409/2 : FAILURE in 2h 00m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d7002dbc8c397f24e7a0037e629af31c00489e00","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-14 08:02:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"813c963b5a293adc43f4f96289e069db36e5a48d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-14 09:24:41.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e7a2254884187e1a3438253a607bc763d6bb174","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-01-14 13:49:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a45fcf4ba04da0f33f3c6e5cf381ce4c4f9d278","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-15 08:25:51.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ee48e3d81ef22f5e9143ec702add2928905b9eeb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-15 08:26:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nI just updated commit message that it reflects actual change","accounts_in_message":[],"_revision_number":3},{"id":"252f1ed1e624485b6414b80c66707622bc34a792","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-15 08:35:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9bf788b1f7d0ea824c1b7fcead600d544573a814","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-15 09:14:02.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701409/3 : FAILURE in 47m 30s","accounts_in_message":[],"_revision_number":3},{"id":"627b67032f8966ecbed38bf5d6ad2180406be27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 10:03:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66599955b4dc43d098a3baf815d78a8e : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87dda85459384a8e972885164f99bdfd : SUCCESS in 35m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8d938c2ea8b433aaefd2dd88ae3a925 : SUCCESS in 27m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66775e53989f4971ac95a58e20b50eb8 : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21ae8628f60a4cfc91ff547f2298a760 : SUCCESS in 29m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13b796e580c24d45931902625e390e3c : SUCCESS in 29m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08aba940acb4c79af54267f18e60033 : SUCCESS in 8m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3a09c41b47cb4be292a5e8bbde7ca150 : SUCCESS in 40m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/71d13fc0cb8845d7856aa7b58991a380 : SUCCESS in 1h 01m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0c2c54d12a78414bbc2c5d3c3d937fb1 : SUCCESS in 1h 34m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8be6596ae40e4739bdccd4814320695d : SUCCESS in 39m 59s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/585b5c7c9f244c3ebcd7512110b92e79 : SUCCESS in 1h 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c15350651dfeb047434b4b74d624fa422ce0fe27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 10:04:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0374927ed09562c9d87be11b6aed56aa356d5e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 13:46:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/640bfbd72ea645a58e5f7d7e31af88ae : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f4d5b8ba75c4e0fb437c995f993f486 : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75e615e8fc74804b1738223607678f8 : SUCCESS in 14m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94126a0a11c441fe8fd2ace788358102 : SUCCESS in 27m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d255fd53d91450eab5b364f56dfee8e : SUCCESS in 31m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486400fda5cc49858f681a831476d86a : SUCCESS in 8m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/672d097a8e1f43c1b83b1a37d3ea4e78 : SUCCESS in 39m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6987b16d9f9f4a67b6010c6d786b6e40 : FAILURE in 1h 23m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/524a18ba57744a1c973c5627933f5040 : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9ae7be9778753e97f4576df0d5f4d4a65a8b424a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-16 12:51:39.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ffbe87e8385f3817710c428226bdcad03acaa46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 14:26:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b739225e914a4a94b3f91e1d102f825c : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32db2a6b47464ed6a32ffc9744b24621 : SUCCESS in 45m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c6bbd1fbb934af48db7147b698dc745 : SUCCESS in 34m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d13c30e4a72f4181abf5af3979b73c76 : SUCCESS in 14m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3c9fbccbc6444dc81e98a421589ffb5 : SUCCESS in 28m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/652833757ed148a5b20f5b881d79bec4 : SUCCESS in 25m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f906c4de574846518422e5248cdd5ef5 : SUCCESS in 8m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0adf6985326446359d11271e50b771e0 : SUCCESS in 46m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/8c12377e2dbf4024a2b4023f48f42c07 : SUCCESS in 1h 09m 24s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cd04e69a0bb84981aa4dc0ca8dcf5ad7 : SUCCESS in 1h 32m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c168ba8aa234d00a20d8887b268ae37 : SUCCESS in 40m 23s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/607a163ad5c943af9b5d5e64a11c5c5b : SUCCESS in 1h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7db1e6eb8ea6e0e075351482abf60bfb0464d851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 14:26:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"42760a3efdeb0d5ca94fa34c709ccba131ceb4b4","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-16 14:52:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701409/3 : SUCCESS in 2h 00m 02s","accounts_in_message":[],"_revision_number":3},{"id":"b9e260f93ddfc51de76d4b5bd09e6ff96b8c77a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 15:57:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a2c571bd99d4b4ba72cc2c16157f49f : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d5a9f9ef96641e4a992bb2e4fb8682f : SUCCESS in 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7499fb870c89437f83234db26d62e82c : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b89b2bec06248cf99bdc83a5f6ad0c2 : SUCCESS in 30m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7865037ad14e4772b581a93cad8a1020 : SUCCESS in 38m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715d2d7d6c4a4686aa5c74a5dcf728d8 : SUCCESS in 11m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dcb8d5ab01d34e4ab550ae326f2c99aa : SUCCESS in 45m 58s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/397bea5e39b94c13afd54591efda72aa : SUCCESS in 1h 12m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e949bebb35ac4e53ba1d66b84163e056 : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bae2b34a7865d58c795d891a170ff8c9f32b8842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 15:57:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4824e4da6799896508de762aced864832705d174","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:01:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679a783e646a4c39bb32f4d61dc4bf6e : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"620adfe75304c84b1e3433d0c3b1ba3ea3570172","revisions":{"7d10850ccb2f5102ff7e3fe081a806b101a47548":{"kind":"REWORK","_number":1,"created":"2020-01-07 17:43:13.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/09/701409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/701409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/701409/1"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-07 17:41:59.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-07 17:43:12.000000000","tz":-300},"subject":"Complete dropping py27 support goal","message":"Complete dropping py27 support goal\n\nThis takes care of the last details for dropping py27 support by adding\na proper min version of python in setup.cfg and removing basepython \u003d\npython3 lines in tox.ini.\n\nChange-Id: I911469d46fccf13bc7ead7a103a7d6e4e0ede7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d10850ccb2f5102ff7e3fe081a806b101a47548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d10850ccb2f5102ff7e3fe081a806b101a47548"}]},"branch":"refs/heads/master"},"efe94b7fd6a10ba70f242e6d187e2a66bf03f06b":{"kind":"REWORK","_number":2,"created":"2020-01-13 18:04:08.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/09/701409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/701409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/701409/2"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-07 17:41:59.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-13 18:02:02.000000000","tz":-300},"subject":"Complete dropping py27 support goal","message":"Complete dropping py27 support goal\n\nThis takes care of the last details for dropping py27 support by adding\na proper min version of python in setup.cfg and removing basepython \u003d\npython3 lines in tox.ini.\n\nChange-Id: I911469d46fccf13bc7ead7a103a7d6e4e0ede7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efe94b7fd6a10ba70f242e6d187e2a66bf03f06b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efe94b7fd6a10ba70f242e6d187e2a66bf03f06b"}]},"branch":"refs/heads/master"},"620adfe75304c84b1e3433d0c3b1ba3ea3570172":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-15 08:25:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/09/701409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/701409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/701409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/701409/3"}}},"commit":{"parents":[{"commit":"ac63c570a1c630ac4405e4caf3d516d069165d69","subject":"Merge \"Improve ovs cleanup utility help\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac63c570a1c630ac4405e4caf3d516d069165d69"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-07 17:41:59.000000000","tz":-300},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-01-15 08:25:47.000000000","tz":0},"subject":"Complete dropping py27 support goal","message":"Complete dropping py27 support goal\n\nThis takes care of the last details for dropping py27 support by adding\na proper min version of python in setup.cfg.\n\nChange-Id: I911469d46fccf13bc7ead7a103a7d6e4e0ede7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/620adfe75304c84b1e3433d0c3b1ba3ea3570172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/620adfe75304c84b1e3433d0c3b1ba3ea3570172"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
