)]}'
{"id":"openstack%2Fneutron~701637","triplet_id":"openstack%2Fneutron~master~I17904c996e1357f7292d25aab4d448edb052f44c","project":"openstack/neutron","branch":"master","topic":"bp/neutron-ovn-merge","hashtags":[],"change_id":"I17904c996e1357f7292d25aab4d448edb052f44c","subject":"[OVN] scripts for networking-ovn code migration","status":"MERGED","created":"2020-01-08 23:36:49.000000000","updated":"2020-01-29 09:16:31.000000000","submitted":"2020-01-29 09:14:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701637-1580289260306-5f1f21a0","meta_rev_id":"b0a014c32aeb68e4bca90445c1cd7a63083c38e6","_number":701637,"virtual_id_number":701637,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 09:14:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-26 22:06:12.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-01-28 11:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-01-29 07:53:50.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-01-29 07:53:50.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 10:55:01.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-01-10 23:26:50.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-01-21 14:33:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-26 22:06:12.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-28 11:14:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-29 07:53:50.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-29 09:14:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c76bbcbc3eb3fad370d5e77dba2a2e9a68c1397","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-08 23:36:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2164ca95dad98fbbf45a487d41905dc706c776a1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-09 00:45:18.000000000","message":"Topic set to bp/neutron-ovn-merge","accounts_in_message":[],"_revision_number":1},{"id":"58aba50df01bfd87995e66cb016791e92780fb1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 01:06:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b25182fa3a2a468a88e5eac0cff17bf4 : SUCCESS in 35m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3926d5f323d44ef08ba3c42bcd254834 : SUCCESS in 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ffd9a705234f88981affbbfca4786e : FAILURE in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd75bff1a0cb43df895c258694cee3ac : SUCCESS in 23m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2245a89ed89a40109715a0671a5c0756 : SUCCESS in 28m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ec7fe469b9423fa8018a31f85f625d : SUCCESS in 8m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/59c081ec2cb947bf8a30e683893f8fc9 : SUCCESS in 41m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/af7fb14ffb9642bbb85c23d2583297f4 : SUCCESS in 1h 04m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e4f250eee0aa449dabf1886159f4c75a : SUCCESS in 1h 07m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca0dbb9193f74154990b9d4207aeb105 : SUCCESS in 40m 03s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa1e613252304ecaab75a321c176ecca : SUCCESS in 58m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d20be32a1f759fd887e7a1e9e85fe25bad33c987","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-09 01:26:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/1 : SUCCESS in 1h 48m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7f20a9dacd310ddc97d29f859d60d006f0be6826","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-09 01:39:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6b5702814a89fd2986f2ae51db6576a3174c2f0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-09 01:43:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94b4c11d64b916d42448ea3398638c40daacfe3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 03:03:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6782b5d9aa6e4846a17ff2ea172c9c10 : SUCCESS in 36m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/998b8f6c32644d24a13dc01e5b46107a : SUCCESS in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0553c576ff641a395041ee90baa6c69 : SUCCESS in 12m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb20ea3022aa4870a6b6c94a07ab009f : SUCCESS in 23m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aadca12ef4e847d4a7d67cf3751011b6 : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb64d5671903412cb15886e5a2b9a7dd : SUCCESS in 7m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5c8e1e31a80349d69eca3b8786253450 : SUCCESS in 40m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d4002bc35254444cb2e28ba9d49e03af : SUCCESS in 59m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f02fdcc6a56549b798de2fe42df0fe73 : SUCCESS in 1h 13m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/166970f8dd574bb19823ac30790c3a44 : SUCCESS in 38m 36s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4131b7ff983448e8a556820089b4fecc : SUCCESS in 1h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"713c83399d2694fb75ca1d355f110adace9634f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-09 03:30:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/3 : SUCCESS in 1h 45m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b1af65837faa47c9fed3929c82362fc32103bc7f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-09 10:55:01.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nHaving tools that makes our work easier is always cool.\n\nI don\u0027t only know if this is the right place for it, because it looks more like a contrib tool, but I don\u0027t know recon any other place for it now, so my +1.","accounts_in_message":[],"_revision_number":3},{"id":"8585cf0e85d16405719cfdbe32a6a5396ae47426","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-09 20:38:25.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks, Maciej!","accounts_in_message":[],"_revision_number":3},{"id":"f7e953e2ad4fde04590b378944c4ea612ea32626","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-09 20:39:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c09d9d3b26435161c6ae2ca2f0961ec065a04638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:10:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0be12b15b8c748f9acb0d12ad69dcf70 : SUCCESS in 33m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed6f4e116d6943808812205d536db78e : SUCCESS in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d785f33bea40bea983e5539ac65b98 : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79921d15321b48c5981036ab033a58ed : SUCCESS in 27m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94e2fcd418d24123b6822363a4575fe2 : SUCCESS in 31m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9574644e3464ec189bce374c3f34dbd : SUCCESS in 9m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba6b4bb420a5408d92645958dc72fbd5 : SUCCESS in 42m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b278314bbf414f95b50c410424ebf2fc : SUCCESS in 1h 00m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d10c05a061f14adeae257007489246f0 : SUCCESS in 1h 10m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4eea2cfc2422460c9238a1d8d3b87ca2 : SUCCESS in 42m 44s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c053ec00b18042689c96097406c6a5f7 : SUCCESS in 1h 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0fc66abc0cb5939f10500e84b64a4a9afdef6c63","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-09 22:42:59.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701637/4 : FAILURE in 2h 02m 47s","accounts_in_message":[],"_revision_number":4},{"id":"48a474c829ea5c0e398065501801e9e8b53163e4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-10 10:23:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5e7b02d80ec00da7077cdbc3f66fd8ecd029f5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 11:35:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97754d25eee74832afe7fd4cd7106914 : SUCCESS in 31m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5782736036bb4c25981b9c5f832399b1 : SUCCESS in 25m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a008468381a40c3bf6cb8cb9c9af70d : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14368040057f48eaae838a7ebb340263 : SUCCESS in 22m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9ecf5f4f1234460b461559eea01608b : SUCCESS in 29m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c321e536447044eeb0dbddee3bae3108 : SUCCESS in 9m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cf028c04f32240e88b7b51eeafe6d45f : SUCCESS in 35m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9cebc5e2671e4fe6b23d11d0a2d892fa : SUCCESS in 1h 00m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/44e967bb5700407d9d4e738d72dac9fa : SUCCESS in 1h 09m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7fd8f9bba05426fa6d62a27f5d8b503 : SUCCESS in 37m 24s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0051c6f834e14f56a3c0d954205c7464 : SUCCESS in 57m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2a21d34817c02834a1de0742acd999f262fc689","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-10 12:14:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/4 : SUCCESS in 1h 49m 36s","accounts_in_message":[],"_revision_number":4},{"id":"f1059b33236c4b187969181fefa50b0f4525da43","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-10 21:26:28.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"13f54f65c5b7e8b03a0fd36eaa967bd98170b4b2","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-01-10 23:26:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2102115f8d025d51dc9c823c57fcd70a4e0ccfbd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-13 23:32:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eb78dd9802efa777fe2dddf42ea5a6e5af060c19","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-13 23:32:56.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"970be5532247b07bb0851b9b7d66139155a1bf75","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-13 23:38:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e0979b586c520b5a6907ec5dbdaf9e79bdcecf9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 01:18:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26ac86273dec470dafcb7f52be96ef77 : SUCCESS in 36m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc8420e30d184c19ab41a8e1c7115d18 : SUCCESS in 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c79ec021f634da8a75ad97497af3d0c : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3669253d13b14678a2325ea89d66e7f9 : SUCCESS in 30m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/421cc1bf2b844efe926f667599d52d8b : SUCCESS in 36m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a5ec7eaa8f4e66b453f0c8dc96a671 : SUCCESS in 8m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/61163664b9c947a29253964c26600ad3 : SUCCESS in 3m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/825ede2bbc1b4247a0773e89a75ab611 : SUCCESS in 39m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4ce5fc4654ef4826b78104ce9b65c958 : SUCCESS in 1h 06m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/89f446aac74d454286f383dcec88db79 : SUCCESS in 1h 06m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2883710f7984550942e1a29866007df : SUCCESS in 35m 04s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6187fd1aa6be4b0ab7e3fddc05f937e8 : SUCCESS in 1h 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"61a1a1cdd902d6080db950c67f6b5f9a74619622","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 01:22:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/6 : SUCCESS in 1h 42m 54s","accounts_in_message":[],"_revision_number":6},{"id":"2b521725b5f239e10e57519677d594e78d7a5027","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-14 11:39:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"62447c78ad6004abeb371f09e23a66adf42a469a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 14:36:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/7 : SUCCESS in 1h 54m 51s","accounts_in_message":[],"_revision_number":7},{"id":"6c5635977552e1a05d3f2fb97af384d11ede2339","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-14 14:40:22.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"446bf25e0858c12f5633602b155a94cb7b075d1a","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 17:13:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/8 : SUCCESS in 1h 46m 12s","accounts_in_message":[],"_revision_number":8},{"id":"160a1520bc3fe0b4c83d559cb28c2ee6b3a76027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 22:47:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e7fb124f656457989d15af2f7c8bb66 : SUCCESS in 31m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29a5f6615c5142bc81fd087b30c92a49 : SUCCESS in 29m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b609c1b871c745d4b393c2f722d94615 : SUCCESS in 17m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f25223559aa40e891984a8b342543bc : SUCCESS in 28m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/591a9b30092b41b5a590ca9c1237b922 : SUCCESS in 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8da29ee91d47cd80c8812c4ac7c118 : SUCCESS in 9m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0b60171f60d4c428e179d95931dadb6 : SUCCESS in 3m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/259c0a1f33ea46999e11e4ce7eec385f : SUCCESS in 43m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aee400dfc10c46a6ac8171a79121f394 : SUCCESS in 1h 07m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b222b50024ff4a52b4653859661781b3 : SUCCESS in 1h 06m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da03f7dd8e754fc7b49232723364efc7 : SUCCESS in 42m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/586e473a2c7b436d9b534781c5f6d1b1 : SUCCESS in 1h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e6b83545278f8121ffafd13c9ff3746dab47d222","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-15 20:10:13.000000000","message":"Patch Set 8:\n\n(12 comments)\n\nNice work Flavio, just a few comments.","accounts_in_message":[],"_revision_number":8},{"id":"087622af02ca7c54fe8b246faca6e0a09795130e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-16 19:00:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"649b2995200e4f89f321590617318a315ad5ef6a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-16 19:02:18.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"424c5164b2d6ee5a9c9d00f73f93a72fea5b486b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-16 19:03:52.000000000","message":"Patch Set 10:\n\n(11 comments)","accounts_in_message":[],"_revision_number":10},{"id":"26be737ce3e6a58981affc317e5f9e82ba3c6bcd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-16 20:06:10.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f24093eeb69bc400d46394010314e01f4f13c9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 21:44:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89ae3a4a96ce4ca98e6a5010bcec9e21 : SUCCESS in 33m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8195fc0021f4dc281075dc92babfa81 : SUCCESS in 24m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4df45e1ca3224a1b9772c6a6bbc67e4d : SUCCESS in 11m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef1bfb31b69a4e7d9c05515df9c14319 : SUCCESS in 27m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0dcd0ac51214ac78e4652938bd2ae01 : SUCCESS in 25m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0268d1893d74175b5457d8568e65c19 : SUCCESS in 8m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/961e88bd395e468e9a0fc38006c5dcb8 : SUCCESS in 2m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/52653b32565a47cebae77bfc93d04092 : SUCCESS in 43m 01s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1d4a0f48d52e41c08931a99de501fbd6 : SUCCESS in 1h 01m 43s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/204d2f6675574ed8a6d032b84f000664 : SUCCESS in 1h 14m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9d44884fc7140d38ad52b4ce1243ce0 : SUCCESS in 37m 52s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf5abc80d499490cbc3e55d6b4a3f86d : SUCCESS in 1h 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"310d8ebe500f5f6d126dcc8b84d5bb9683e1e2e8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-16 21:56:30.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/11 : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":11},{"id":"a02f8625edd8012589a0a6ec08a4b0138d6f2211","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-17 08:06:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d41cf71dd38481200436eb9743546d5f164f37a9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-17 09:56:28.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f19b7bdd6baff7946da75861d99c08427e43b206","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-17 10:05:04.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6a2dd267285dd3700bdfb1bd956d58ab3fd60b34","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-17 10:06:42.000000000","message":"Patch Set 11: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"10fda4dd2ac9a6da965141e908c3bca5a70bcff0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-17 19:00:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b24bfca89653449eab9304a16dbaabc45073d43f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-17 19:08:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d238d4f71f8f1e91d025d0852f346a12941ff4b0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-17 21:05:13.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701637/13 : FAILURE in 1h 56m 06s","accounts_in_message":[],"_revision_number":13},{"id":"46e0bfcdd33e45cc4251ea2cc7be4a2383aa6481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:12:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4df1cc4e480944978aa1860ef92d51df : SUCCESS in 36m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd56540524c843f2b6a470dcd2ab3c52 : SUCCESS in 27m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e0a457b2504892b582d924e47a3662 : SUCCESS in 15m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d222617a33d4566b13f41741b42a697 : SUCCESS in 25m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/587660e32635480192696330aaae12b5 : SUCCESS in 36m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9db980221a44bf5a32c3316ba903bde : SUCCESS in 9m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90b77e03197a44db822ae80e2741e45a : SUCCESS in 3m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/14a2abc1c04240538afa97463a0772a9 : SUCCESS in 45m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7a8768bcd42448d288f567603f0fedb8 : SUCCESS in 1h 11m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5eb8a581caba42c39fc2dd8354a18feb : SUCCESS in 1h 10m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab6e7c9b199640e3a3d6edba5f47fde9 : SUCCESS in 43m 54s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/81954dc20faa4620a8c7248fc61ccc14 : SUCCESS in 59m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b2facbe3668f47e37d9f1d6f345cf915f237d7bd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-18 12:30:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"ed8ac045659614f312581c1a54c30668403e1732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 13:51:55.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ad7121c97394a8dbdc88dbc8df53e2d : SUCCESS in 49m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b64b67099e564d8db49d90a47579c9eb : SUCCESS in 35m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c6d8b4095c84f34a0e7b4f9bbc949af : SUCCESS in 14m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e565377375a1496e8d91de17e0ade225 : SUCCESS in 22m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83239f13c81149ee95cd50e7b325ec1a : SUCCESS in 27m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93682e8dbcb4af9addb4df3a3c30a96 : SUCCESS in 8m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc81c6b054c34de3a9613392fbb579d8 : SUCCESS in 2m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0f0d6f149f594aba93593ec57abcfff6 : SUCCESS in 45m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9176b66154a94954993399158eaf8705 : SUCCESS in 56m 57s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4e29b5bea9684794afabc6d7970ea3ac : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/621698d44797460bb35516de5b2951fb : SUCCESS in 40m 33s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/86fab6d33cbb47cf99ec422e914ecf13 : SUCCESS in 1h 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6a7262fdb7ccd51a8e995633f7401662f7e9d496","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-18 14:17:54.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701637/14 : FAILURE in 1h 46m 48s","accounts_in_message":[],"_revision_number":14},{"id":"a39495e6b17e44f2f8d3005d751a2cb22644a6a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 10:49:45.000000000","message":"Patch Set 14: Code-Review+2\n\nThx, now it looks better IMO. And I tested it locally that it works fine :)","accounts_in_message":[],"_revision_number":14},{"id":"c2e3892b9e445d041f7b9511f2741eb1094f7072","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-21 10:54:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cbd4d9b258096428ffebd7be53a6ac8af62391d8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-21 14:33:30.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5947b4717f9a6aab7a4d79e87a65bb420a962619","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-21 16:29:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f090092bc754f562137fd94685597d2ffe2cace7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-22 08:27:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"96005f421c985f3d664f2b1b2d71cf8e23c4c691","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-22 11:35:41.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"943cbb0026cc5740f24842fc17c5d31c5db08540","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-22 11:44:04.000000000","message":"Patch Set 15:\n\n(1 comment)\n\nMoved doc into contributor dir. It looks better there, I think.\nAlso tweaked the help output to be more generic instead if having my laptop\u0027s path.","accounts_in_message":[],"_revision_number":15},{"id":"da09ae22f350a0a06fa3acccdfdd1323412f9787","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-22 12:00:52.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701637/15 : FAILURE in 24m 31s","accounts_in_message":[],"_revision_number":15},{"id":"63cf89f2e1a7e0a0e9494c7b55e16682146f4693","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-22 12:49:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"fa4aafcf9192d27e8b2048b99d95999e304f998c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 13:50:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fced360750d486fad077d809161b9c1 : SUCCESS in 45m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ed726eb4c0349f29fdea15707920d8f : SUCCESS in 28m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247b385e23e245c2a0dc421aee6ae8cb : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c579565164844cba1770a9775dc2384 : SUCCESS in 26m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e5b126199f24cd8b50a4ec5454281d6 : SUCCESS in 28m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c0a90ec5fe7499cba9bbe5c7290b6fa : SUCCESS in 8m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c2edd2514194f6ebad3fc5e5d2819af : SUCCESS in 3m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/913598b5512a47b6b6fec0927ae2b914 : SUCCESS in 46m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7fa70c3aa8aa4b12aa185c7c8b857c51 : SUCCESS in 1h 08m 25s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/20fa642bca074b729cb92a6212ea5070 : SUCCESS in 1h 09m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/172ed4a77ea3463c991675585fd6e19e : TIMED_OUT in 2h 13m 07s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba02f0f92b80435eb49abcecf7166ec5 : SUCCESS in 1h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e679c62052259ed067a11584841ee1b70808937f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-01-26 20:19:43.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"7bafc782b1a3c5e9bfbf657b30bcc23fa0031125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 21:28:11.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6111901c04944f348dac1ba5ba98a37e : SUCCESS in 34m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ea703e9d2454afa91e70076b72ef9ae : SUCCESS in 25m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227399a80c7f4e4083f4a9c6e92b5e6a : SUCCESS in 12m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f61f11c004a240b08b3db67d8b3145e8 : SUCCESS in 26m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64249b31f1f640aa8a818e415eb29c5c : SUCCESS in 28m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcec15d932a84f7e981144ddcb66455d : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37795a6c8cad4ef2b2a16d4c90c6ab78 : SUCCESS in 8m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/733a85525d8448ecbc40116a9f229baa : SUCCESS in 3m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/25d09db56f71466daf60fe4f19162344 : SUCCESS in 38m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fa568e63c5784be99956e0e190f3e608 : SUCCESS in 1h 03m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/39d8433e8d4a464383ae5f4512ed9485 : SUCCESS in 1h 05m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7976b68ff33a47c4a098ec9eff2d6ec5 : SUCCESS in 40m 19s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7023ab1aac8143b6b0b3b0096c21b575 : SUCCESS in 1h 07m 49s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ecdf9edaa685f7bbe08e75d49b4c37bb5ce996a1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-26 22:06:12.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/701637/16 : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":16},{"id":"7bf44d4bfc4aa172b34f2ea017ab5e06d91ad142","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-28 11:14:29.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"ecacb217b04418e67e4987b0b55811e49f245c74","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-29 07:53:50.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"8d11f12ccc85f2ed242d415e6fb2267fe8caf02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 07:54:05.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"d12ccd3abdc67708e9015f36473d4d10028ee3aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:14:20.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8172cf5a7d0c4f089c7a72e6645e9c47 : SUCCESS in 28m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89cd5ced77aa4862a9f26586c15adec7 : SUCCESS in 15m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/866eff0ee6b346ad87a5004138e88985 : SUCCESS in 26m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4e2ffde28bc484ab50465b719cf37c7 : SUCCESS in 29m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4782b3c3c1f245c78536108b6d47e538 : SUCCESS in 9m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f375f2af402442bbaa1b2b5ddc65df7 : SUCCESS in 5m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9494888bda754f588952544a06bd1f8a : SUCCESS in 46m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/fd90c0152605405e87e1b0e9044cd9f9 : SUCCESS in 1h 09m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/180bf7da4063494dbbd62cc6386c37fd : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":16},{"id":"034ef2db1c7ec72749d811f05cba9723d7bc5f09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:14:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"b0a014c32aeb68e4bca90445c1cd7a63083c38e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:16:31.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3844aca8d1184f12bd488ae05d9cad49 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"303937f3aec1e53d7ba49078ff399186c8cbabd0","revisions":{"cf4b3f2d490b49fa8ddf3eea8532bd4a43e1b896":{"kind":"REWORK","_number":1,"created":"2020-01-08 23:36:49.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/1"}}},"commit":{"parents":[{"commit":"331a195972b2b2aa0b2fd6ebffc0cacd8589b828","subject":"[OVN] Migrate the OVN mech driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/331a195972b2b2aa0b2fd6ebffc0cacd8589b828"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:34:23.000000000","tz":-300},"subject":"tools: Adding scripts to help with networking-ovn migration","message":"tools: Adding scripts to help with networking-ovn migration\n\nIn an effort to help on migrating changes on netowking-ovn to neutron,\nthese 3 scripts are being introduced.\n\n1) files_in_patch.py \u003d\u003d\u003e show files that are changed in a patch_file\n\n$ # make a patch to use as example\n$ git show \u003e /tmp/commit.patch\n\n$ ./tools/files_in_patch.py /tmp/commit.patch\ntools/download_gerrit_change.py\ntools/files_in_patch.py\ntools/migrate_names.py\ntools/migrate_names.txt\n$\n\n2) download_gerrit_change.py \u003d\u003d\u003e given a gerrit change id, download latest\n   patchset as a patch_file\n\n$ ./tools/download_gerrit_change.py -h\nusage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\nGet patch file for a gerrit\n\npositional arguments:\n  gerritChange  the gerrit change id\n  output        file containing the gerrit change\n  url           the url to Gerrit server\n\noptional arguments:\n  -h, --help    show this help message and exit\n$ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n$ ./tools/files_in_patch.py !$\n./tools/files_in_patch.py /tmp/change.patch\nnetworking_ovn/ml2/mech_driver.py\nnetworking_ovn/ml2/trunk_driver.py\nnetworking_ovn/tests/unit/ml2/test_mech_driver.py\nnetworking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py \u003d\u003d\u003e given a gerrit change id or a patch_file, use mapfile\n   to rename files in patch\n\n$ ./tools/migrate_names.py --help\nUsage: migrate_names.py [OPTIONS]\n\nOptions:\n  -i, --input_patch TEXT    input_patch patch file or gerrit change\n  -o, --output TEXT         Output patch file. Default: stdout\n  -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                            input\n  --reverse / --no-reverse  Map filenames from neutron to networking-ovn repo\n  --help                    Show this message and exit.\n$ ./tools/migrate_names.py -i /tmp/change.patch -o /tmp/neutron_change.patch\n$ ./tools/migrate_names.py -o /tmp/change.patch.2 -i /tmp/neutron_change.patch --reverse\n$ diff /tmp/change.patch /tmp/change.patch\n$ diff /tmp/change.patch /tmp/neutron_change.patch\n38,39c38,39\n\u003c --- a/networking_ovn/ml2/mech_driver.py\n\u003c +++ b/networking_ovn/ml2/mech_driver.py\n---\n\u003e --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\u003e +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n84,85c84,85\n\u003c%-%- snip %-%-\u003e\n\n$ ./tools/files_in_patch.py /tmp/neutron_change.patch\nneutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\nneutron/services/trunk/drivers/ovn/trunk_driver.py\nneutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\nneutron/tests/unit/services/trunk/drivers/ovn/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf4b3f2d490b49fa8ddf3eea8532bd4a43e1b896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf4b3f2d490b49fa8ddf3eea8532bd4a43e1b896"}]},"branch":"refs/heads/master"},"03355ca7f9de4eb02d0e12f85e12432ec5ab3f4e":{"kind":"REWORK","_number":2,"created":"2020-01-09 01:39:25.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/2"}}},"commit":{"parents":[{"commit":"a762292003a103ec6a7a9a8718a4f7f257e9cc1f","subject":"Merge \"Remove the condition that the mtu is nullable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a762292003a103ec6a7a9a8718a4f7f257e9cc1f"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-09 01:39:24.000000000","tz":-300},"subject":"tools: Adding scripts to help with networking-ovn migration","message":"tools: Adding scripts to help with networking-ovn migration\n\nIn an effort to help on migrating changes on netowking-ovn to neutron,\nthese 3 scripts are being introduced.\n\n1) files_in_patch.py \u003d\u003d\u003e show files that are changed in a patch_file\n\n$ # make a patch to use as example\n$ git show \u003e /tmp/commit.patch\n\n$ ./tools/files_in_patch.py /tmp/commit.patch\ntools/download_gerrit_change.py\ntools/files_in_patch.py\ntools/migrate_names.py\ntools/migrate_names.txt\n$\n\n2) download_gerrit_change.py \u003d\u003d\u003e given a gerrit change id, download latest\n   patchset as a patch_file\n\n$ ./tools/download_gerrit_change.py -h\nusage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\nGet patch file for a gerrit\n\npositional arguments:\n  gerritChange  the gerrit change id\n  output        file containing the gerrit change\n  url           the url to Gerrit server\n\noptional arguments:\n  -h, --help    show this help message and exit\n$ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n$ ./tools/files_in_patch.py !$\n./tools/files_in_patch.py /tmp/change.patch\nnetworking_ovn/ml2/mech_driver.py\nnetworking_ovn/ml2/trunk_driver.py\nnetworking_ovn/tests/unit/ml2/test_mech_driver.py\nnetworking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py \u003d\u003d\u003e given a gerrit change id or a patch_file, use mapfile\n   to rename files in patch\n\n$ ./tools/migrate_names.py --help\nUsage: migrate_names.py [OPTIONS]\n\nOptions:\n  -i, --input_patch TEXT    input_patch patch file or gerrit change\n  -o, --output TEXT         Output patch file. Default: stdout\n  -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                            input\n  --reverse / --no-reverse  Map filenames from neutron to networking-ovn repo\n  --help                    Show this message and exit.\n$ ./tools/migrate_names.py -i /tmp/change.patch -o /tmp/neutron_change.patch\n$ ./tools/migrate_names.py -o /tmp/change.patch.2 -i /tmp/neutron_change.patch --reverse\n$ diff /tmp/change.patch /tmp/change.patch\n$ diff /tmp/change.patch /tmp/neutron_change.patch\n38,39c38,39\n\u003c --- a/networking_ovn/ml2/mech_driver.py\n\u003c +++ b/networking_ovn/ml2/mech_driver.py\n---\n\u003e --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\u003e +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n84,85c84,85\n\u003c%-%- snip %-%-\u003e\n\n$ ./tools/files_in_patch.py /tmp/neutron_change.patch\nneutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\nneutron/services/trunk/drivers/ovn/trunk_driver.py\nneutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\nneutron/tests/unit/services/trunk/drivers/ovn/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03355ca7f9de4eb02d0e12f85e12432ec5ab3f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/03355ca7f9de4eb02d0e12f85e12432ec5ab3f4e"}]},"branch":"refs/heads/master"},"4dddc0bf3365ccaa3422399f9df11dae1b4a9be7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-09 01:43:54.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/3"}}},"commit":{"parents":[{"commit":"a762292003a103ec6a7a9a8718a4f7f257e9cc1f","subject":"Merge \"Remove the condition that the mtu is nullable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a762292003a103ec6a7a9a8718a4f7f257e9cc1f"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-09 01:43:26.000000000","tz":-300},"subject":"tools: Adding scripts to help with networking-ovn migration","message":"tools: Adding scripts to help with networking-ovn migration\n\nIn an effort to help on migrating changes on netowking-ovn to neutron,\nthese 3 scripts are being introduced.\n\n1) files_in_patch.py \u003d\u003d\u003e show files that are changed in a patch_file\n\n$ # make a patch to use as example\n$ git show \u003e /tmp/commit.patch\n\n$ ./tools/files_in_patch.py /tmp/commit.patch\ntools/download_gerrit_change.py\ntools/files_in_patch.py\ntools/migrate_names.py\ntools/migrate_names.txt\n$\n\n2) download_gerrit_change.py \u003d\u003d\u003e given a gerrit change id, download latest\n   patchset as a patch_file\n\n$ ./tools/download_gerrit_change.py -h\nusage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\nGet patch file for a gerrit\n\npositional arguments:\n  gerritChange  the gerrit change id\n  output        file containing the gerrit change\n  url           the url to Gerrit server\n\noptional arguments:\n  -h, --help    show this help message and exit\n$ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n$ ./tools/files_in_patch.py !$\n./tools/files_in_patch.py /tmp/change.patch\nnetworking_ovn/ml2/mech_driver.py\nnetworking_ovn/ml2/trunk_driver.py\nnetworking_ovn/tests/unit/ml2/test_mech_driver.py\nnetworking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py \u003d\u003d\u003e given a gerrit change id or a patch_file, use mapfile\n   to rename files in patch\n\n$ ./tools/migrate_names.py --help\nUsage: migrate_names.py [OPTIONS]\n\nOptions:\n  -i, --input_patch TEXT    input_patch patch file or gerrit change\n  -o, --output TEXT         Output patch file. Default: stdout\n  -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                            input\n  --reverse / --no-reverse  Map filenames from neutron to networking-ovn repo\n  --help                    Show this message and exit.\n$ ./tools/migrate_names.py -i /tmp/change.patch -o /tmp/neutron_change.patch\n$ ./tools/migrate_names.py -o /tmp/change.patch.2 -i /tmp/neutron_change.patch --reverse\n$ diff /tmp/change.patch /tmp/change.patch\n$ diff /tmp/change.patch /tmp/neutron_change.patch\n38,39c38,39\n\u003c --- a/networking_ovn/ml2/mech_driver.py\n\u003c +++ b/networking_ovn/ml2/mech_driver.py\n---\n\u003e --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\u003e +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n84,85c84,85\n\u003c%-%- snip %-%-\u003e\n\n$ ./tools/files_in_patch.py /tmp/neutron_change.patch\nneutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\nneutron/services/trunk/drivers/ovn/trunk_driver.py\nneutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\nneutron/tests/unit/services/trunk/drivers/ovn/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dddc0bf3365ccaa3422399f9df11dae1b4a9be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4dddc0bf3365ccaa3422399f9df11dae1b4a9be7"}]},"branch":"refs/heads/master"},"321931666a2f9ed7db5aa7b36aea4b5ad7e14ae4":{"kind":"REWORK","_number":4,"created":"2020-01-09 20:39:37.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/4"}}},"commit":{"parents":[{"commit":"31730e436983364a0903249dde389fff792ab373","subject":"Merge \"[L3] Switch order of processing added and removed router ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31730e436983364a0903249dde389fff792ab373"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-09 20:39:32.000000000","tz":-300},"subject":"tools: Adding scripts to help with networking-ovn migration","message":"tools: Adding scripts to help with networking-ovn migration\n\nIn an effort to help on migrating changes on netowking-ovn to neutron,\nthese 3 scripts are being introduced.\n\n1) files_in_patch.py \u003d\u003d\u003e show files that are changed in a patch_file\n\n$ # make a patch to use as example\n$ git show \u003e /tmp/commit.patch\n\n$ ./tools/files_in_patch.py /tmp/commit.patch\ntools/download_gerrit_change.py\ntools/files_in_patch.py\ntools/migrate_names.py\ntools/migrate_names.txt\n$\n\n2) download_gerrit_change.py \u003d\u003d\u003e given a gerrit change id, download latest\n   patchset as a patch_file\n\n$ ./tools/download_gerrit_change.py -h\nusage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\nGet patch file for a gerrit\n\npositional arguments:\n  gerritChange  the gerrit change id\n  output        file containing the gerrit change\n  url           the url to Gerrit server\n\noptional arguments:\n  -h, --help    show this help message and exit\n$ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n$ ./tools/files_in_patch.py !$\n./tools/files_in_patch.py /tmp/change.patch\nnetworking_ovn/ml2/mech_driver.py\nnetworking_ovn/ml2/trunk_driver.py\nnetworking_ovn/tests/unit/ml2/test_mech_driver.py\nnetworking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py \u003d\u003d\u003e given a gerrit change id or a patch_file, use mapfile\n   to rename files in patch\n\n$ ./tools/migrate_names.py --help\nUsage: migrate_names.py [OPTIONS]\n\nOptions:\n  -i, --input_patch TEXT    input_patch patch file or gerrit change\n  -o, --output TEXT         Output patch file. Default: stdout\n  -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                            input\n  --reverse / --no-reverse  Map filenames from neutron to networking-ovn repo\n  --help                    Show this message and exit.\n$ ./tools/migrate_names.py -i /tmp/change.patch -o /tmp/neutron_change.patch\n$ ./tools/migrate_names.py -o /tmp/change.patch.2 -i /tmp/neutron_change.patch --reverse\n$ diff /tmp/change.patch /tmp/change.patch\n$ diff /tmp/change.patch /tmp/neutron_change.patch\n38,39c38,39\n\u003c --- a/networking_ovn/ml2/mech_driver.py\n\u003c +++ b/networking_ovn/ml2/mech_driver.py\n---\n\u003e --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\u003e +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n84,85c84,85\n\u003c%-%- snip %-%-\u003e\n\n$ ./tools/files_in_patch.py /tmp/neutron_change.patch\nneutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\nneutron/services/trunk/drivers/ovn/trunk_driver.py\nneutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\nneutron/tests/unit/services/trunk/drivers/ovn/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321931666a2f9ed7db5aa7b36aea4b5ad7e14ae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321931666a2f9ed7db5aa7b36aea4b5ad7e14ae4"}]},"branch":"refs/heads/master"},"ea74a21c3841ca27d616ed1210ca7cffafd920ec":{"kind":"REWORK","_number":5,"created":"2020-01-13 23:32:43.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/5"}}},"commit":{"parents":[{"commit":"ae67ae824ce38da158777ad6bd9a23e1088b115a","subject":"Merge \"Check \"keepalived\" is stopped in KeepalivedManagerTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae67ae824ce38da158777ad6bd9a23e1088b115a"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-13 23:31:14.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch\n   40,41c40,41\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   ---\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea74a21c3841ca27d616ed1210ca7cffafd920ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea74a21c3841ca27d616ed1210ca7cffafd920ec"}]},"branch":"refs/heads/master"},"29e236d1d9716b87a9ce2126abf62da3b83267d8":{"kind":"REWORK","_number":6,"created":"2020-01-13 23:38:59.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/6"}}},"commit":{"parents":[{"commit":"ae67ae824ce38da158777ad6bd9a23e1088b115a","subject":"Merge \"Check \"keepalived\" is stopped in KeepalivedManagerTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae67ae824ce38da158777ad6bd9a23e1088b115a"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-13 23:38:54.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch\n   40,41c40,41\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   ---\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29e236d1d9716b87a9ce2126abf62da3b83267d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29e236d1d9716b87a9ce2126abf62da3b83267d8"}]},"branch":"refs/heads/master"},"019cbd883516a32105af139269f3877a06c9437b":{"kind":"REWORK","_number":7,"created":"2020-01-14 11:39:01.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/7"}}},"commit":{"parents":[{"commit":"ae67ae824ce38da158777ad6bd9a23e1088b115a","subject":"Merge \"Check \"keepalived\" is stopped in KeepalivedManagerTestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae67ae824ce38da158777ad6bd9a23e1088b115a"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-14 11:38:55.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch\n   40,41c40,41\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   ---\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/019cbd883516a32105af139269f3877a06c9437b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/019cbd883516a32105af139269f3877a06c9437b"}]},"branch":"refs/heads/master"},"765bf0034df65297ff043cc644203f97527a861f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-01-14 14:40:22.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/8"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-01-14 14:40:22.000000000","tz":0},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch\n   40,41c40,41\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   ---\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/765bf0034df65297ff043cc644203f97527a861f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/765bf0034df65297ff043cc644203f97527a861f"}]},"branch":"refs/heads/master"},"cbfd994e9dee86e2129c3399e190118231179198":{"kind":"REWORK","_number":9,"created":"2020-01-16 19:00:29.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/9"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-16 16:25:08.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbfd994e9dee86e2129c3399e190118231179198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbfd994e9dee86e2129c3399e190118231179198"}]},"branch":"refs/heads/master"},"6a3752e57163923653a9208da71d602b31b59034":{"kind":"REWORK","_number":10,"created":"2020-01-16 19:02:18.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/10"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-16 19:02:14.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a3752e57163923653a9208da71d602b31b59034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a3752e57163923653a9208da71d602b31b59034"}]},"branch":"refs/heads/master"},"995f3dc251b92443efd8cb25731d01f622596b69":{"kind":"REWORK","_number":11,"created":"2020-01-16 20:06:10.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/11"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-16 20:06:06.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py -h\n   usage: download_gerrit_change.py [-h] gerritChange [output] [url]\n\n   Get patch file for a gerrit\n\n   positional arguments:\n     gerritChange  the gerrit change id\n     output        file containing the gerrit change\n     url           the url to Gerrit server\n\n   $ ./tools/download_gerrit_change.py 698863 /tmp/change.patch\n   $ ./tools/files_in_patch.py !$\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) ./tools/migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output TEXT         Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: ./tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to neutron repo\n\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/change.patch.reverse -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/change.patch.reverse /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/995f3dc251b92443efd8cb25731d01f622596b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/995f3dc251b92443efd8cb25731d01f622596b69"}]},"branch":"refs/heads/master"},"d4b5f8d5c27834372ce4f2fe123d1be6dd4099aa":{"kind":"REWORK","_number":12,"created":"2020-01-17 19:00:53.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/12"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-17 18:57:40.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py --help\n   Usage: download_gerrit_change.py [OPTIONS] GERRIT_CHANGE\n\n   Options:\n     -o, --output_patch TEXT  Output patch file  [default: stdout]\n     -g, --gerrit_url TEXT    The url to Gerrit server  [default:\n                              https://review.opendev.org/]\n     -t, --timeout INTEGER    Timeout, in seconds  [default: 10]\n     --help                   Show this message and exit.\n\n   $ ./tools/download_gerrit_change.py 698863 -o /tmp/change.patch\n   $ ./tools/files_in_patch.py /tmp/change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output_patch TEXT   Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: /Users/gute/work/openstack/neutro\n                               n.git/tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to Neutron repo\n     --help                    Show this message and exit.\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/reverse.patch -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/reverse.patch /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4b5f8d5c27834372ce4f2fe123d1be6dd4099aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d4b5f8d5c27834372ce4f2fe123d1be6dd4099aa"}]},"branch":"refs/heads/master"},"b3ea1bb7a9216de95d28cebd2a35c7c85b71273b":{"kind":"REWORK","_number":13,"created":"2020-01-17 19:08:27.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/13"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-17 19:08:24.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py --help\n   Usage: download_gerrit_change.py [OPTIONS] GERRIT_CHANGE\n\n   Options:\n     -o, --output_patch TEXT  Output patch file  [default: stdout]\n     -g, --gerrit_url TEXT    The url to Gerrit server  [default:\n                              https://review.opendev.org/]\n     -t, --timeout INTEGER    Timeout, in seconds  [default: 10]\n     --help                   Show this message and exit.\n\n   $ ./tools/download_gerrit_change.py 698863 -o /tmp/change.patch\n   $ ./tools/files_in_patch.py /tmp/change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output_patch TEXT   Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: /Users/gute/work/openstack/neutro\n                               n.git/tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to Neutron repo\n     --help                    Show this message and exit.\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/reverse.patch -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/reverse.patch /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3ea1bb7a9216de95d28cebd2a35c7c85b71273b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3ea1bb7a9216de95d28cebd2a35c7c85b71273b"}]},"branch":"refs/heads/master"},"411fb39557ab47b1bbb132f9e2a91bc06cd44c53":{"kind":"REWORK","_number":14,"created":"2020-01-18 12:30:34.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/14"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-18 12:29:17.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py --help\n   Usage: download_gerrit_change.py [OPTIONS] GERRIT_CHANGE\n\n   Options:\n     -o, --output_patch TEXT  Output patch file  [default: stdout]\n     -g, --gerrit_url TEXT    The url to Gerrit server  [default:\n                              https://review.opendev.org/]\n     -t, --timeout INTEGER    Timeout, in seconds  [default: 10]\n     --help                   Show this message and exit.\n\n   $ ./tools/download_gerrit_change.py 698863 -o /tmp/change.patch\n   $ ./tools/files_in_patch.py /tmp/change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output_patch TEXT   Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: /Users/gute/work/openstack/neutro\n                               n.git/tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to Neutron repo\n     --help                    Show this message and exit.\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/reverse.patch -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/reverse.patch /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/411fb39557ab47b1bbb132f9e2a91bc06cd44c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/411fb39557ab47b1bbb132f9e2a91bc06cd44c53"}]},"branch":"refs/heads/master"},"92e2476620df2f368c1759351cfc8070e151cc4c":{"kind":"REWORK","_number":15,"created":"2020-01-22 11:35:41.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/15"}}},"commit":{"parents":[{"commit":"ab24a11f13cdfdf623a4b696f469aa621d59405b","subject":"Merge \"Set binding profile directly from OVNTrunkDriver (redo cont.)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab24a11f13cdfdf623a4b696f469aa621d59405b"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-22 11:30:38.000000000","tz":-300},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py --help\n   Usage: download_gerrit_change.py [OPTIONS] GERRIT_CHANGE\n\n   Options:\n     -o, --output_patch TEXT  Output patch file  [default: stdout]\n     -g, --gerrit_url TEXT    The url to Gerrit server  [default:\n                              https://review.opendev.org/]\n     -t, --timeout INTEGER    Timeout, in seconds  [default: 10]\n     --help                   Show this message and exit.\n\n   $ ./tools/download_gerrit_change.py 698863 -o /tmp/change.patch\n   $ ./tools/files_in_patch.py /tmp/change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output_patch TEXT   Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: /home/user/openstack/neutron.git\n                               /tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to Neutron repo\n     --help                    Show this message and exit.\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/reverse.patch -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/reverse.patch /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92e2476620df2f368c1759351cfc8070e151cc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92e2476620df2f368c1759351cfc8070e151cc4c"}]},"branch":"refs/heads/master"},"303937f3aec1e53d7ba49078ff399186c8cbabd0":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-01-26 20:19:43.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/37/701637/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/701637/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/701637/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/701637/16"}}},"commit":{"parents":[{"commit":"4ce1a65e135d83c01482d63e22ccd1e0281398c8","subject":"Merge \"[OVN] Bump ovsdbapp requirements to 1.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ce1a65e135d83c01482d63e22ccd1e0281398c8"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-01-08 23:13:22.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-01-26 20:19:43.000000000","tz":0},"subject":"[OVN] scripts for networking-ovn code migration","message":"[OVN] scripts for networking-ovn code migration\n\nIn an effort to help on migrating changes from neutron to networking-ovn,\nthese 3 scripts are being introduced under tools.\n\nAlso adding documentation about these under OVN folder.\n\n1) files_in_patch.py\nUse this to show files that are changed in a patch file.\n\n   $ # Make a patch to use as example\n   $ git show \u003e /tmp/commit.patch\n\n   $ ./tools/files_in_patch.py /tmp/commit.patch | grep .py\n   tools/download_gerrit_change.py\n   tools/files_in_patch.py\n   tools/migrate_names.py\n\n2) download_gerrit_change.py\nGiven a gerrit change id, it will fetch the latest patchset of the change\nfrom review.opendev.org as a patch file.\n\n   $ ./tools/download_gerrit_change.py --help\n   Usage: download_gerrit_change.py [OPTIONS] GERRIT_CHANGE\n\n   Options:\n     -o, --output_patch TEXT  Output patch file  [default: stdout]\n     -g, --gerrit_url TEXT    The url to Gerrit server  [default:\n                              https://review.opendev.org/]\n     -t, --timeout INTEGER    Timeout, in seconds  [default: 10]\n     --help                   Show this message and exit.\n\n   $ ./tools/download_gerrit_change.py 698863 -o /tmp/change.patch\n   $ ./tools/files_in_patch.py /tmp/change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\n3) migrate_names.py\nUse this tool to modify the name of the files in a patchfile so it can\nbe converted to/from the legacy networking-ovn and neutron repositories.\n\n   $ ./tools/migrate_names.py --help\n   Usage: migrate_names.py [OPTIONS]\n\n   Options:\n     -i, --input_patch TEXT    input_patch patch file or gerrit change\n     -o, --output_patch TEXT   Output patch file. Default: stdout\n     -m, --mapfile PATH        Data file that specifies mapping to be applied to\n                               input  [default: /home/user/openstack/neutron.git\n                               /tools/migrate_names.txt]\n     --reverse / --no-reverse  Map filenames from networking-ovn to Neutron repo\n     --help                    Show this message and exit.\n   $ ./tools/migrate_names.py -i 701646 \u003e /tmp/ovn_change.patch\n   $ ./tools/migrate_names.py -o /tmp/reverse.patch -i /tmp/ovn_change.patch --reverse\n   $ diff /tmp/reverse.patch /tmp/ovn_change.patch | grep .py\n   \u003c --- a/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003c +++ b/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n   \u003e --- a/networking_ovn/ml2/mech_driver.py\n   \u003e +++ b/networking_ovn/ml2/mech_driver.py\n   \u003c... snip ...\u003e\n\n   $ ./tools/files_in_patch.py /tmp/ovn_change.patch\n   networking_ovn/ml2/mech_driver.py\n   networking_ovn/ml2/trunk_driver.py\n   networking_ovn/tests/unit/ml2/test_mech_driver.py\n   networking_ovn/tests/unit/ml2/test_trunk_driver.py\n\nChange-Id: I17904c996e1357f7292d25aab4d448edb052f44c\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/303937f3aec1e53d7ba49078ff399186c8cbabd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/303937f3aec1e53d7ba49078ff399186c8cbabd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
