)]}'
{"id":"openstack%2Fneutron~701900","triplet_id":"openstack%2Fneutron~stable%2Fqueens~I72dc4a06a806731ec5124fa11c9f69c7dd6cbbb0","project":"openstack/neutron","branch":"stable/queens","topic":"bug/1857021-stable/queens","hashtags":[],"change_id":"I72dc4a06a806731ec5124fa11c9f69c7dd6cbbb0","subject":"[L3] Switch order of processing added and removed router ports","status":"MERGED","created":"2020-01-10 08:11:27.000000000","updated":"2020-01-29 03:53:02.000000000","submitted":"2020-01-29 03:53:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701900-1580269982185-9115562a","meta_rev_id":"024838381c1afa53ce8739ef352e968c53d4d909","_number":701900,"virtual_id_number":701900,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-01-28 18:41:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 03:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-01-28 16:44:02.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-28 16:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-01-29 00:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-01-29 00:22:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-25 17:59:50.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-28 16:12:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-01-28 16:44:02.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-28 18:41:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-29 00:22:04.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-29 03:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be83608b03b804a3faa0b57e1ff0e7f3b1edb3a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-10 08:11:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5bcecad245025307b4aa80e7a5439f2df9067ae9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-10 08:43:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/00/701900/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/00/701900/1/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0e1f0b821ced7cb60ada5ddac5b3ed7a7991255f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-10 09:16:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/1 : FAILURE in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"84890224db6c4cd3e1f14cb50c6765731ee12560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 10:38:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e3a3a4cba1a144e886bc4d48efb6aba0 : SUCCESS in 54m 00s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/fb860acc74464ad48071f417184264fb : SUCCESS in 31m 55s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/96d2d1568c6f42f89d99e62b5a5c504e : FAILURE in 1h 27m 22s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f82a2e4a584b4f34835645820a51a347 : FAILURE in 53m 24s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/88eb0ff1a0aa4c3fb6479b09736ef825 : FAILURE in 1h 04m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dcff1b2bb4647c1a500d3a4019b88e7 : SUCCESS in 12m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b2d99c419c544339bac059b02e7a447 : SUCCESS in 22m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5357dec103eb464fb79305afe35d1a5f : SUCCESS in 26m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d287d84615f54aad92eb93827dbb07e8 : SUCCESS in 6m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3a4f97ca70746159bc4556f0be19584 : SUCCESS in 1h 38m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/23005fc8cf4743ff85c15a7afa4e8e48 : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d9cd7636a744bddb6dde39f8c40d983 : SUCCESS in 1h 32m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d6aa3282663f4a47b4677d814955873a : SUCCESS in 40m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d6cba69410a54382bbf14350f967f120 : FAILURE in 26m 38s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/a71c5772a3ba482082a1d89ccd4461ec : SUCCESS in 1h 57m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/329ca13131654036997de95655e58f68 : SUCCESS in 1h 31m 59s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3a2d748884714d17ae64328f5145ceb3 : SUCCESS in 1h 17m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cc979c23a954f668bc462d17b5dad73 : SUCCESS in 1h 11m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6ed2da5c53e743f5a8cf31a83e8fdb2a : SUCCESS in 1h 01m 11s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ca3eac2f72ec4f8fa4fc4250441d22cd : SUCCESS in 1h 31m 14s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1f209d7f42fd49c5a77c5f41f34817e0 : SUCCESS in 1h 46m 58s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/7740ca5863bb4416b3360ec135bb6285 : SUCCESS in 1h 12m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/455a86c7340d49b29b9e50ac592a9741 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e4dee96740c80241dcee45890466f9cecb5fa1c0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-12 10:01:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0741640406bb38a76c2a666054d0ea9d416cfffb","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-12 10:11:03.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/1 : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"011750c7e19fb6bc235c900d844abe920ec8a59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 10:35:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/b5de7c4e4c74425198bb7018b9a48be8 : FAILURE in 8m 36s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/978f82f103ba467e94396bf20682fcc1 : FAILURE in 10m 10s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/2f16f7dcdb054bf19a65a12dd484f33c : FAILURE in 13m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/dd76b28cdaa74b45b6963cd2e23e314b : FAILURE in 8m 53s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/86f5b9b711d441a18c2f209be4a16c0d : FAILURE in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d7b1ef87634aac9d41e951434ef1f9 : SUCCESS in 10m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69273f2061234c57ac842e71e3089fb1 : SUCCESS in 23m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5656400a9f7f41e493804f743cff3307 : SUCCESS in 20m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/66d22eb772c04440a19080d1b2ed1ee0 : RETRY_LIMIT in 2m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e471d59b19649489e6048bc6d1091d1 : FAILURE in 8m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3cefcfe1616f4ce48b9547b582459ac4 : FAILURE in 4m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe25b975e6ba4dd78cad7bb83c691d2a : FAILURE in 9m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0c7f735fba9f4b0f92598dbc0f50de0b : FAILURE in 4m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/091ea3c40d9d478d88e734b478dab753 : FAILURE in 6m 09s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/37851c28a0f34370bb3169b6dd16e0e5 : FAILURE in 5m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1115c346ce7b45c0a76008ddfd4feeb5 : FAILURE in 5m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2a80f23fd4a44a49a8b19ce2b56fcf5a : FAILURE in 6m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc56cdb4090a44329e4e5112b54c9ceb : FAILURE in 6m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ca9979c69334e60a585a2888adc87bc : FAILURE in 6m 23s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1cc46e10ddb6408fa60534fa0cc14038 : FAILURE in 5m 00s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f5cfc0294b14d9cbe425b2378d1a8c9 : FAILURE in 5m 12s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/ccb452f0a70d4a378cc535936820bb72 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f44772f7d9c14dcca85ff3c3e13f79be : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dcaa6ef678e5fb00923be64be38599ee9a897b89","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-13 10:42:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0237864da72b91efd45185cbaeb2e08cabeff7bd","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-13 10:50:41.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/1 : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"92f262f1fb871ee16bb4b10eba166848f06088c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 11:38:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/159aa76225b345ad87f988e30b6b4a5b : FAILURE in 13m 49s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/ff313914a44546d08189fe2426caeb66 : FAILURE in 12m 41s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/15978604723045fc9f8de0a9fd9ccfaf : FAILURE in 18m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/1a40908ea98c447db7424ee557012a59 : FAILURE in 10m 32s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/fa79b6a00cf644448cb983440b09cbc1 : FAILURE in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e5feab556640b6a2158f8b21ef6de1 : SUCCESS in 12m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a03d2ab81494acc8545d552dfc90064 : SUCCESS in 27m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8402a2cb641f4a09818d069a81a890d9 : SUCCESS in 27m 21s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/055338d14270437cbddf86b418bbec2d : RETRY_LIMIT in 2m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/be5222d0891b42b497cdb343c1971391 : FAILURE in 13m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97694e56525047678004cf9911f7d3f3 : FAILURE in 9m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2c8412266fe43c3b96dc971f5e23142 : FAILURE in 11m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e641482720c645eb952498140a573117 : FAILURE in 5m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b3bd0724c60545549d541a250f5aaa01 : FAILURE in 4m 15s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/9d0ff4f8316f4e61bde507eacdeb94ba : FAILURE in 7m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1bf65d5cf44b4bf9b5935cb67573ecf7 : FAILURE in 4m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b93d3cdb0a784acb81242f2d7a224682 : FAILURE in 6m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99b32af337fc47308ef5503e6f8a5396 : FAILURE in 5m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/550df6729e3a4d2db43e367f9dc1259c : FAILURE in 5m 35s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/681ea1bb599f4eeebcbbf407cdbb514a : FAILURE in 6m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/afae181e03e74d72984da40c554950e0 : FAILURE in 5m 53s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/6be93a7395574156b4ad97f3f7a6512b : FAILURE in 4m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b68df49807e34d389a284a701f744dfa : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"612a47777833e580cea61dba32483642314d8d4d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-13 13:06:10.000000000","message":"Patch Set 1:\n\nrecheck queens failures should be fixed now (at least tempest-full with https://review.opendev.org/#/c/701404/ )","accounts_in_message":[],"_revision_number":1},{"id":"a9d9adffb3aaf309c02ed23d9a0aa33744d8fb49","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-13 13:14:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/1 : FAILURE in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0408eefbd5fcb76da17d3ae278ff0bbd7dc96630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 14:07:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/565e343161c1481890fb0a9ee78be741 : FAILURE in 15m 41s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/d1dd0309142f4584ad7340f1c3e47e71 : FAILURE in 12m 54s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/304a8ad189a14232a8e20070598e1e9d : FAILURE in 18m 15s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/321bcfb4f1ed4a9cbf801ca0a95e9e17 : FAILURE in 10m 34s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/b1015708af2e4b2089e48b5ef8cf69c1 : FAILURE in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50239266779d4623b66f943b35dd7801 : SUCCESS in 10m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0d4b4c383474722af7d49725c1d93c8 : SUCCESS in 25m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/316340ac62914e14b6e4b3c29d72f39b : SUCCESS in 22m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4d749ac528594faca0977535b9b7d9b2 : RETRY_LIMIT in 3m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dca63400f9ca4bc5bf5fe7db44cb2ee1 : FAILURE in 13m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/89955145bb9c473eb350ea7e526a4924 : FAILURE in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e444ea9081b84cbeb9081c11b14c204e : FAILURE in 7m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/59b3f43e796d45889cc66b3ef1647d0f : FAILURE in 4m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9e9f4f266b034414b9e49ea1fb77dfcc : FAILURE in 4m 26s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/600ebd7aaf0f4357bf04d6080c72fce8 : FAILURE in 4m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c82dc27d2aa0460eaa9d0e530e92eb8c : FAILURE in 13m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e3f5787d54d240fca93e76d43d98107d : FAILURE in 4m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d108d1a23c4b48d1b9ce60f22c36829d : FAILURE in 6m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f3db71bfaadf45c39e9fb1de6156d47f : FAILURE in 13m 31s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/848fa60c615e4c0a81ce45bc91aa12aa : FAILURE in 5m 37s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f8718f7e2594a56a4defd3256b6f109 : FAILURE in 5m 41s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/a24cd0d87394426e908cfbe5a3109cd0 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae1eaa686af04543b3e155d5cb4f2935 : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ac0f0b0484a216c7700a359e286189e0a0b14905","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-14 08:40:25.000000000","message":"Patch Set 1:\n\nrecheck setuptools fix in","accounts_in_message":[],"_revision_number":1},{"id":"33310fcbdaca971488a867d8933048f5174d37d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 08:49:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/1 : FAILURE in 8m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c9ee66dddd334d5af8e2db0f80cb91218cedfb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 13:48:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/a8e8dbe8bb7c4778a108be0b5227189a : SUCCESS in 1h 00m 26s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/0c73e9aa85524d6287787efd75c8170a : SUCCESS in 33m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/277408b94b0b47a4bc9aa07bdaeb3035 : FAILURE in 1h 45m 23s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/51e34c3d41824f919af92377dcd9e01c : FAILURE in 50m 06s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/267f68ff312d438fb452c54821f8868c : FAILURE in 57m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c37838186a4cdca551210b12109be2 : SUCCESS in 11m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89bde941f31843308c012f6ef1cefb11 : SUCCESS in 26m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7565e34621834d0b8d3eef28c005621b : SUCCESS in 28m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/68c8ad720cca4c9fb228433519fa4f1d : SUCCESS in 7m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c9e9eeb814444f67a8e59e5a52ef9ea4 : SUCCESS in 1h 37m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4b523691b02848a687650d52867f3585 : SUCCESS in 59m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/010935104e2a4ecb9bb137dd136f886b : SUCCESS in 1h 31m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e61f7878692b4417aae0f187ccfb1404 : SUCCESS in 43m 07s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/706c3fe4d61145c28253010fee4ab8a1 : FAILURE in 25m 55s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/a9219f4c6c2f494c9b785d0d312411e4 : SUCCESS in 1h 50m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ba9d97d614964e6c99a6f8caa72a8425 : SUCCESS in 1h 16m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/27e1d3860b5e4d128271c5bcb1cf748e : SUCCESS in 1h 09m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/382c054497d6492ba29e27213fe84673 : SUCCESS in 1h 04m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf9b8106ba5e4edd92cf2321e97926a3 : SUCCESS in 1h 04m 09s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/10f4c744b606436495c7b50db8e45bcc : SUCCESS in 1h 36m 03s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/95793fde9c6a4d7eb03f8f2e07386fb9 : SUCCESS in 1h 51m 59s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/0ff57afac44e4d1c8db05cd595f27853 : SUCCESS in 1h 31m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31be5d83982f4fd2b949d72c057e4be3 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9afd34c275b3f009ab69f413590ae2716baa480f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-16 14:02:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9da374b160c3f895b194cb223d13cb32fed100f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-16 14:12:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/2 : FAILURE in 9m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b9162cc64493506819e74d147a9fa7c9cf507d50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-16 14:37:59.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/00/701900/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/00/701900/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"6a342745a370e6f721a699ddb4d1741aeefd764b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 15:52:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/8346b47d1c3a48b5bc6621d0a76d6f2a : FAILURE in 32m 11s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/0d11260ce1e74c1f945c3f26fc8dc28e : FAILURE in 36m 39s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/b22c41d406814cd5b651bc817616731b : FAILURE in 46m 45s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/30e9951c98e143068cec5c86469eadc0 : FAILURE in 36m 22s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ee8021c1470a4fb88f55d64cc16ed15e : FAILURE in 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157655ff44904216947ed265a23d1089 : SUCCESS in 11m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/422979da8b1a4ed385d6b651502089fd : SUCCESS in 25m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/79504163588c456ebb8919c032eb5d6a : SUCCESS in 27m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a29c86f788b945219b76b0a0949d75f0 : SUCCESS in 7m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c9adc372b1b14f33bea77bee42aebc09 : SUCCESS in 1h 31m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d8d994c098a648c4a3ecc61ab0b40788 : SUCCESS in 1h 04m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3036273d12dd41ddb490f6858d89a36c : SUCCESS in 1h 19m 54s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/54807eb0152f487bb752a6ed184641dc : SUCCESS in 39m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/63ebf02595014903bb9cba0c19aeadec : FAILURE in 21m 28s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/053c98b385de4800b86a56cdb8bd2d1b : SUCCESS in 1h 30m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/be6d7b6e50fa48e7ac7a00b40baccf46 : SUCCESS in 1h 37m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d5cc641ff4ce4d1f860205f30f3d72e1 : SUCCESS in 1h 29m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d73b1487dd274dd2914e48f90eb86b9f : SUCCESS in 1h 00m 17s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c8d438af070a4a68ba613b623731f19c : SUCCESS in 1h 20m 20s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8107031613a14a39a63c28f482cd10cc : SUCCESS in 1h 37m 33s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f4f5abbe82041179a9c7aae615e3ec1 : SUCCESS in 1h 43m 29s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/cb3e7812effa40dfae9fab611969224e : SUCCESS in 1h 19m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3c1ac8526a94c348228cf713fd90d5d : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5563a1fb3598c0ed71a478905aa647690455d7a7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-17 09:09:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"836c8880574172a3e23e018b1cbf999dc55c3c54","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-17 09:17:48.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/2 : FAILURE in 7m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7ac7af8e0cd4624acc90314746af1f94c499e58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 13:10:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/31002d6aa3e74a15aac6c2a18fc1c8db : FAILURE in 29m 27s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/cf0bc043dfdf4286ac498e4a45e5bc69 : FAILURE in 30m 00s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/f66c70b4c132474da26079b55bf0b6d5 : FAILURE in 43m 13s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/14549138d63148c3a85091b283832ac3 : FAILURE in 28m 13s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/1c58a9732e544a629e745ef85cbfa441 : FAILURE in 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14515adf0f1a4eb0b853665adc05c5e0 : SUCCESS in 12m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f5b6604e120424e8c23b20b59539bc0 : SUCCESS in 25m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93e8ccb025f24a08b2a92a216f1d5548 : SUCCESS in 30m 02s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1c33f4a3b86d4a4994146653233f556f : SUCCESS in 8m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d10b29b9e60e46ba99dc90a66990311b : SUCCESS in 1h 18m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c72f0184f60f4e00a6fabe5b1b981a00 : SUCCESS in 1h 09m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62ac52353e7d4c2eae6622a30d29d942 : SUCCESS in 1h 17m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1370169256c74bcfab9ded71536cb9a1 : SUCCESS in 41m 46s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7a38438412974c5096facfd88b9e5186 : FAILURE in 25m 47s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/fb68cc19080842239b90ca13e94ec9d2 : SUCCESS in 1h 40m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/aad12bf90d124c00b58c2b0f1dbc6479 : SUCCESS in 1h 21m 10s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d92e2b8c4c3d4bdb8f72dfa86a0c037c : SUCCESS in 1h 29m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54201494d7e14eba84ac7c61a9d027d7 : SUCCESS in 1h 12m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/759bd938e30f43bb9cbb90e919a630e2 : SUCCESS in 1h 11m 23s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fda0cf37d0824f78b8e7093a66e72ea5 : SUCCESS in 2h 13m 58s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1966ba484db475a96ede4d3da1e030d : SUCCESS in 1h 42m 36s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/d87f7a37dd084a328c356d198c357616 : SUCCESS in 2h 02m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162835b3ca884056bbfe0bc8917938c8 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4c34cef1b46a7a903c8689a4df61b0058ba2093c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 09:35:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ff467002c31b2ed62736c77db80059f6286f46a9","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 10:40:18.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/2 : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":2},{"id":"02514b503c9de1a1fafef3e77f9d828bcbbc9273","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-21 11:10:01.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/00/701900/2/SRIOV-neutron-macvtap : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a5db12f36a9787bf5b6c87ff7dc0597a97e548f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 12:52:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/2b2f209a4700452ca0b08c255df19814 : FAILURE in 43m 11s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/4f3d45dc69ce429a814837dccb850960 : FAILURE in 31m 38s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/95f8692ed6a647a5bb646884230bca89 : FAILURE in 45m 11s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/89d55fd166b74c4095fc18d745053eab : FAILURE in 34m 44s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/ada60dff55c746d6ad5414fc0557b9a1 : FAILURE in 32m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa788d151ef41baa63cc62a0b2bb370 : SUCCESS in 11m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08296ba5cff746d5a6df6901f67da6e3 : SUCCESS in 24m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9582ecc181b64fa78e7f03f314aa12c0 : SUCCESS in 27m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/662c95da92204167886addd5f7662535 : SUCCESS in 6m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e29401151c6249cd9aa5f227ff21ff8a : POST_FAILURE in 1h 24m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/299b47c8c0e441a9a6a27953fad60964 : SUCCESS in 55m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/027a0ae9edbf40819dcfd8402282b814 : POST_FAILURE in 1h 38m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/20ba612c3b184abe81135625520cc33b : SUCCESS in 42m 28s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e313dcadc079478b8852fc518c0d11c8 : FAILURE in 28m 59s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/fa87e716b6434bbaa92d1ba6d397be7d : SUCCESS in 1h 31m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9a5d60bfe29b4ba698d8dcee39d7fdc2 : SUCCESS in 1h 38m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/25d1a1d0eb5d4f0eaea69dc3fc040571 : FAILURE in 1h 09m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d3eb4e6be9b48d8acfecf05e4af7fc9 : FAILURE in 55m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/034731d2c2bb44c486e28d8811d1bbae : FAILURE in 1h 03m 17s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4817ee2f7b754f2b8ecf53abde3c3a0f : SUCCESS in 1h 46m 32s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c561cb7ced124470ba2ebcd3c85c1e55 : SUCCESS in 1h 31m 41s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/c849563d0e70485e8f67e5ee8f0d47cd : SUCCESS in 1h 36m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebfdf45eab5c4b19812b19adc70515da : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9e5bd9d00a9d9a919ce33698545c57f1b11c9d61","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-25 17:59:45.000000000","message":"Patch Set 2:\n\nall fixes merged now.","accounts_in_message":[],"_revision_number":2},{"id":"9eb1725d29c034911136930f147fdd4c7cf77fe5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-25 17:59:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2272f4a010cb5182c347ce67733ae5622ff4469e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 18:00:02.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"4b41d8f8128b6df3259161335099f6b89a610611","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-25 18:08:30.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/2 : FAILURE in 8m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a42a3085eb9240bc7df7910df9e654693232dd1d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-26 11:47:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d3a7fcd5e6ce13409b00d850cdd8969f5a2e624b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-26 11:55:38.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/3 : FAILURE in 7m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b87f88fc62dbb09a9369b218465aa1bd172b571f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 13:46:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/54c0f18806cf4abf8ccb663bce8d731e : SUCCESS in 51m 59s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/f412d3dba67441a19ad270dd9f73fd62 : SUCCESS in 45m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/bdd64beb09f7481e81448f2f339d2aa8 : FAILURE in 1h 24m 28s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/154fced525b64a0ba9b51473dc5b6059 : SUCCESS in 1h 19m 15s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/58533a37f9424097bf728175ff6132f4 : SUCCESS in 48m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f47df8263fd4b62b11cdac4dd9dba5c : SUCCESS in 12m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72aaf5e87f644fb8bdc85717d9706c93 : SUCCESS in 21m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0bae4ae31b74d91b28d8dc53b0bbf46 : SUCCESS in 21m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ba3ee2c567954506bb8eaade00c160c3 : SUCCESS in 6m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/26ecb497fc454716abb72cc4fb5a2894 : SUCCESS in 1h 11m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a51245fc8e4642b7a86dc72822038efb : SUCCESS in 50m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ba58fa574204ee2855e31de106e13b1 : FAILURE in 16m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f36a88ffdc5e4dbcb599f90f093c0af3 : SUCCESS in 36m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b24f5d4d81b14b5da66eb0f1608e80c3 : FAILURE in 37m 55s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/150ff3c953b940d596e30bd3dcbb5746 : SUCCESS in 1h 22m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f3d4b9f9f8994d12ac1deccc27d8f6a4 : SUCCESS in 1h 17m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/86d85557cbd34edcb3f1b096e5ceac54 : SUCCESS in 1h 36m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c757943214a490ab46c694504a665c6 : SUCCESS in 1h 32m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/489df5827a87406ebe21dadfad454af2 : SUCCESS in 1h 15m 06s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a173d1ffe7cd42c7b0fa25eb5de956c8 : SUCCESS in 1h 33m 51s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ad7103b9e1c34db5b232dd072caf8dd0 : SUCCESS in 1h 44m 20s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/adc206e9ddef480ab5c6b9b7d8cd0509 : SUCCESS in 1h 50m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ab79a80e8f4f3bb912a6f2297c270b : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":3},{"id":"fa2e43d233aa2a189c556b4875c99090a66b231c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-26 17:17:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4c8316c96b7adb3804b0f05040264f673b2a3a6b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-26 17:25:56.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/3 : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a6ff0f740db7dc6855a19341a8232180760a9b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 19:22:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/66ecf1026de9425b96d39a6b6f09a3d1 : SUCCESS in 1h 19m 39s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/2e9a8a3439824fba8b7fced2a8783077 : SUCCESS in 43m 17s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/0a4f5f3774a0466bbbbabe9738691086 : FAILURE in 1h 11m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/50747c6f73b8427689f535a30b8740e1 : SUCCESS in 44m 43s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/0edf7639eba04558b6fae59aaea02192 : SUCCESS in 46m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5162ee797a5d4d68a792b5d26cc7c080 : SUCCESS in 15m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c6bb44a113c463090177ae14ab7787d : SUCCESS in 34m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7c8bc487b8348f69c82657c4de56eda : SUCCESS in 21m 31s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/626058519f5b4cd4bed8dacd4a42521a : SUCCESS in 11m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e2b544b1cf6498f9dbf9d029395e8b8 : SUCCESS in 1h 27m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/027818de56ee49dd964ebf6e32b945b4 : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f75f79955dbb45ad801d204627b887ca : FAILURE in 15m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bc7ac69c4aaa4b4b9715092647397672 : SUCCESS in 43m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/41c916ac77984bddac673cbc35e74ce8 : FAILURE in 35m 54s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/c9f7ea48fa9b49d99d8f1145f8aa5ab1 : SUCCESS in 1h 52m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/511bee3dcb64409b9608c75449a43874 : SUCCESS in 1h 49m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/abf6ef2c3f774d5abd827d158d7099f8 : SUCCESS in 1h 38m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa3672ca0d454bbebd639d35b6c40549 : FAILURE in 1h 01m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f5b57a4ab0504d1fa0719c77c94ce073 : SUCCESS in 1h 05m 57s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a7629171682845b6995aa9beacf40b8c : FAILURE in 59m 57s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b76672227aa94abf9a5653e611977cd6 : FAILURE in 1h 36m 12s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/1bc82c8ba5e34dc1a84e54f7a297f010 : SUCCESS in 1h 50m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a1fc9e87bde43dcbe5891d7458c8179 : SUCCESS in 41m 55s","accounts_in_message":[],"_revision_number":3},{"id":"d9442bec52c6e121bb91224a2891ca99e3d56243","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-26 23:40:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c95091c49450b51fa78327994cdda31c3f63d287","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-26 23:49:02.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/3 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":3},{"id":"80496c52056244e307bc5f42ed1f325c8592068d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 01:28:57.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/7560f0230d5a407db1e32d800c53ead2 : SUCCESS in 50m 03s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/b95a1bbf9bdb4065b6bdf94335af0624 : SUCCESS in 31m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/764a57778f774aea97e85f3fe76fc043 : FAILURE in 1h 27m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/b4f7a77f3c23402da1aebb18b3e290d0 : SUCCESS in 50m 48s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/2f04594911c843a5affa59375de7d399 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89631dc569394aa597892157a0fa1253 : SUCCESS in 11m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79dd0d928db6458b901e255206d137bc : SUCCESS in 26m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53f27d80c79d4fb083eafa90f3611c57 : SUCCESS in 24m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ba51e6014e01404a98112922fdbb4dca : SUCCESS in 5m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b779743365a84bf09bd1cf65e1f737e5 : SUCCESS in 1h 27m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d4fb0df84fd44372bc9094daa4d8f3df : SUCCESS in 55m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edeaf1efe770463aa40cb4fec6c1a2a2 : FAILURE in 17m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6a797556833643cea0d1270484166ea3 : SUCCESS in 38m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ebcd246c3c6244488025aba2ea38ad1a : FAILURE in 22m 41s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/398ba52a8db143d4ba9645e04cd36c48 : SUCCESS in 1h 42m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/80ea1fb299c24889aff770201f98dc3e : SUCCESS in 1h 28m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9cdaa2a1795d474987d5305c29cb11bb : SUCCESS in 1h 19m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e74edd38e8b246449760c173de327409 : SUCCESS in 1h 09m 36s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a24521204fb246f0bd410132bcfe7bb3 : SUCCESS in 1h 07m 56s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a7138814203048e7a34a333e9e442bc9 : SUCCESS in 1h 28m 56s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/11d0b615fac54d0fb5a5ca6407c6678b : SUCCESS in 1h 35m 49s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/c9e759e4a81340b9a7b07c8cb895188d : SUCCESS in 1h 12m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cca03e5804404590bbbaa2bd8b739e67 : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e907519179f6a802f6fff5dc59af62495bbc72e3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-28 16:10:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"05d45117567179d0e09b7e7e45905861a959ea37","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-01-28 16:12:02.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good and failures unrealted","accounts_in_message":[],"_revision_number":3},{"id":"2abec2f738312af4d84449367a30fdd79f3e0df1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-28 16:44:02.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/701900/3 : FAILURE in 9m 10s","accounts_in_message":[],"_revision_number":3},{"id":"879d745382d4b8161e2a07216fb184d8a71ca66e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-28 18:41:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/00/701900/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/00/701900/3/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"0a3552de34e6956dee0d27d76c27900785bd744b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:06:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/40302b7a850b4cdc9833bff3bc9f129d : SUCCESS in 50m 49s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/db30803ab0df499c833270c05819f265 : SUCCESS in 40m 08s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/82e0a65216d64f52862194c269796ed1 : SUCCESS in 1h 24m 30s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/a2631cc2167a411a9b43554d80057904 : SUCCESS in 57m 10s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6e8698973a354612976ebbb63eb6ceb2 : SUCCESS in 51m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e212e412ff4c4098884abb41031dd2cf : SUCCESS in 12m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a7506ae8bdc42fdbe5e870fc242c733 : SUCCESS in 28m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c680b26f292241c1a9cfed244a3d63be : SUCCESS in 29m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1ad5d666a023419983c1d8fbd09e1553 : SUCCESS in 8m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9dfd20e6b3374d17a094a6e3982db583 : SUCCESS in 1h 21m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0a7926e6cb3e4017846d75aae506ef51 : SUCCESS in 1h 00m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edf0073ab568435cb66221616b1c555c : SUCCESS in 1h 09m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/430dd7c8ca294ce2ab8c6740f72d467a : SUCCESS in 35m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1fe3b84b631e4442aa3dbd44160a6a01 : FAILURE in 26m 04s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/2c4169f733704b7387b5311f1039c501 : SUCCESS in 1h 30m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cbc53072ef6643068b515df8e23651ff : SUCCESS in 1h 25m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ed6c76bb646c4adab1131c4fc2f6aef2 : SUCCESS in 1h 53m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e93c0d3822c484ba4d033a5f1713ea1 : SUCCESS in 1h 00m 48s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0d8c7add9fca420d9a75f40492b00206 : SUCCESS in 1h 17m 57s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c103e4f771ec47ac8b4454d12ecb41e5 : SUCCESS in 1h 37m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94777023600a41e79d86bfd7a481cc87 : SUCCESS in 1h 26m 51s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/a4f87e421b654e3686d6be68328c2ad5 : SUCCESS in 1h 12m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a7b0fcf75aa4dc9af0630fd3f009bc8 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":3},{"id":"2ba0d42a606b0d899b19c80820fde7e641e22adc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-01-29 00:22:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b616eed68c76b97a467537e11be52f6cd1b7be0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 00:22:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"597921c1c5fa01845f6765148d6f588e7b6fca65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:53:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/e8cc17e8921a4acbaf89122de93e72d2 : SUCCESS in 51m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038b3a3429bb438b8c83e7817ac152b0 : SUCCESS in 10m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/735ea272f9264c33a9ba73c578300b28 : SUCCESS in 32m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1795ac82565b4acba14520c13406d761 : SUCCESS in 22m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b2c6281d5eab4a3eae7af8173049a090 : SUCCESS in 6m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/361a4082721c4e1aaa3434be42cd149c : SUCCESS in 1h 14m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0564c54e462340579442d3b01b5a35e6 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/762ae91cb6524a9b872e76301f5960ca : SUCCESS in 1h 10m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d24421fbc1ab446fa0a3c7ee5cb28ff3 : SUCCESS in 38m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f06892a509b401ba958bb7c8792c645 : SUCCESS in 1h 14m 26s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/89efc497b8094c14af0f190a937d9102 : SUCCESS in 1h 19m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0abe7c3b07644832899c20dcaf58e3ca : SUCCESS in 1h 00m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7bfc30d08a784c83bd1835620f24c21e : SUCCESS in 58m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da555ab233bd42b5b0dfb0b630ff29d1 : SUCCESS in 28m 15s","accounts_in_message":[],"_revision_number":3},{"id":"5769e60c769e2ca64de4ab403eb19a3aaf96420b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 03:53:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a9b491a5ed58f4ec22f9e93e7d921c8e92406544","revisions":{"839eb1a290aff9e3e5b76528ccd3fbf12d25c611":{"kind":"REWORK","_number":1,"created":"2020-01-10 08:11:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/00/701900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/701900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/701900/1"}}},"commit":{"parents":[{"commit":"2084585b81649b2afa747e24f036d54e5e16e90e","subject":"Set DB retry for quota_enforcement pecan_wsgi hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2084585b81649b2afa747e24f036d54e5e16e90e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-19 15:12:09.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-01-10 08:11:27.000000000","tz":0},"subject":"[L3] Switch order of processing added and removed router ports","message":"[L3] Switch order of processing added and removed router ports\n\nIt may happend that one router\u0027s port is going to be\nremoved and another one (same IP but new subnet) is going to be added\nto the router in short time.\nThat can lead to the problem that IP which is allocated to the new\nport is not added to keepalived\u0027s vips list because same IP address\nis already in this list (this exising IP address belongs to old port).\nBut few seconds later old port is removed and finally router ends\nup with new port configured without IP address.\n\nTo avoid such case, this patch switches order of processing new\nand deleted ports in _process_internal_ports() method in RouterInfo\nclass.\nSo now first old ports will be removed and than new ports will be\nconfigured so there will be no case when IP address is already added\nto VIPs list when it is going to be removed in few seconds.\n\nChange-Id: I72dc4a06a806731ec5124fa11c9f69c7dd6cbbb0\nCloses-Bug: #1857021\n(cherry picked from commit 3faba7cae0c5f0b8ac93c025ca057e2f533445cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/839eb1a290aff9e3e5b76528ccd3fbf12d25c611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/839eb1a290aff9e3e5b76528ccd3fbf12d25c611"}]},"branch":"refs/heads/stable/queens"},"74899bd37447e1082d14ac0e73185b7a5d234b81":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-16 14:02:14.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/00/701900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/701900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/701900/2"}}},"commit":{"parents":[{"commit":"2084585b81649b2afa747e24f036d54e5e16e90e","subject":"Set DB retry for quota_enforcement pecan_wsgi hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2084585b81649b2afa747e24f036d54e5e16e90e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-19 15:12:09.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-01-16 14:01:40.000000000","tz":0},"subject":"[L3] Switch order of processing added and removed router ports","message":"[L3] Switch order of processing added and removed router ports\n\nIt may happend that one router\u0027s port is going to be\nremoved and another one (same IP but new subnet) is going to be added\nto the router in short time.\nThat can lead to the problem that IP which is allocated to the new\nport is not added to keepalived\u0027s vips list because same IP address\nis already in this list (this exising IP address belongs to old port).\nBut few seconds later old port is removed and finally router ends\nup with new port configured without IP address.\n\nTo avoid such case, this patch switches order of processing new\nand deleted ports in _process_internal_ports() method in RouterInfo\nclass.\nSo now first old ports will be removed and than new ports will be\nconfigured so there will be no case when IP address is already added\nto VIPs list when it is going to be removed in few seconds.\n\nDepends-On: https://review.opendev.org/702868\nChange-Id: I72dc4a06a806731ec5124fa11c9f69c7dd6cbbb0\nCloses-Bug: #1857021\n(cherry picked from commit 3faba7cae0c5f0b8ac93c025ca057e2f533445cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74899bd37447e1082d14ac0e73185b7a5d234b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/74899bd37447e1082d14ac0e73185b7a5d234b81"}]},"branch":"refs/heads/stable/queens"},"a9b491a5ed58f4ec22f9e93e7d921c8e92406544":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-26 11:47:30.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/00/701900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/701900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/701900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/701900/3"}}},"commit":{"parents":[{"commit":"2084585b81649b2afa747e24f036d54e5e16e90e","subject":"Set DB retry for quota_enforcement pecan_wsgi hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2084585b81649b2afa747e24f036d54e5e16e90e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-12-19 15:12:09.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-01-26 11:47:24.000000000","tz":0},"subject":"[L3] Switch order of processing added and removed router ports","message":"[L3] Switch order of processing added and removed router ports\n\nIt may happend that one router\u0027s port is going to be\nremoved and another one (same IP but new subnet) is going to be added\nto the router in short time.\nThat can lead to the problem that IP which is allocated to the new\nport is not added to keepalived\u0027s vips list because same IP address\nis already in this list (this exising IP address belongs to old port).\nBut few seconds later old port is removed and finally router ends\nup with new port configured without IP address.\n\nTo avoid such case, this patch switches order of processing new\nand deleted ports in _process_internal_ports() method in RouterInfo\nclass.\nSo now first old ports will be removed and than new ports will be\nconfigured so there will be no case when IP address is already added\nto VIPs list when it is going to be removed in few seconds.\n\nChange-Id: I72dc4a06a806731ec5124fa11c9f69c7dd6cbbb0\nCloses-Bug: #1857021\n(cherry picked from commit 3faba7cae0c5f0b8ac93c025ca057e2f533445cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9b491a5ed58f4ec22f9e93e7d921c8e92406544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a9b491a5ed58f4ec22f9e93e7d921c8e92406544"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
