)]}'
{"id":"openstack%2Fneutron~702209","triplet_id":"openstack%2Fneutron~master~I5e5e14d5a621fc4a790810d15e1162f90898282e","project":"openstack/neutron","branch":"master","topic":"bp/neutron-ovn-merge","hashtags":[],"change_id":"I5e5e14d5a621fc4a790810d15e1162f90898282e","subject":"[OVN] Pass context while removing objects in OVN client","status":"MERGED","created":"2020-01-13 12:23:59.000000000","updated":"2020-01-22 01:30:20.000000000","submitted":"2020-01-22 01:28:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702209-1579656495465-46a14a57","meta_rev_id":"64a9b80ad764242053e79146196b25d6aadd0945","_number":702209,"virtual_id_number":702209,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-20 16:41:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-21 19:54:12.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2020-01-21 13:20:57.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 01:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-21 20:19:51.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-01-21 08:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-01-21 08:43:29.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-01-21 08:43:29.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 13:57:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-01-14 17:11:00.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-20 16:41:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-21 08:11:19.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-21 08:43:29.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-01-21 19:54:12.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-01-21 20:19:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-22 01:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3633016f2fc20d4f803acc04386283c53767d08","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-13 12:23:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0972effbe737c0061f312961176d5fffbec02e2f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-13 12:25:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b09cad6c667f423c06f5c8e04bb69940becb430f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-13 12:36:43.000000000","message":"Patch Set 1: -Workflow\n\nIn related change there is a functional test verifying this.","accounts_in_message":[],"_revision_number":1},{"id":"83af0dc6fd5de277c910f38934c86c37fc2840a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-13 13:10:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0462f7ca191bbeaa94af3a6c28fc6becd496ea5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 13:14:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5bd1a585cfa2437ca75abacd81bcd290 : FAILURE in 10m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1c077a6a33c94e3596dcfc91a8e2fa07 : FAILURE in 11m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8dd6ba8508b041b59e902a55def3924e : FAILURE in 18m 37s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/125a510dd40f42b48818b56a4285a12a : FAILURE in 12m 13s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/72d1d6a63dfb41dc82235d2200f52e20 : FAILURE in 12m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7570e6fc98b74de9b4402c7e83470328 : FAILURE in 13m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2834065336d48958d4e0d402df9da6d : SUCCESS in 48m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16fc070b90ab464e838ddb1095f934b1 : SUCCESS in 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5ae9c96214a48cabbf4032b17f0e6a4 : SUCCESS in 14m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a55336f4041e44119db7dabed904ae66 : SUCCESS in 30m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b56a5cac9c2a4c27baceab40deb39af5 : SUCCESS in 28m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddc02a00e6844f3a53c344d1ecd0df6 : SUCCESS in 15m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/713ab748f4e94e86b5c055cd95e259c9 : FAILURE in 10m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c995b095a023428b8ee9313d0f4989a5 : FAILURE in 12m 13s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c7a08aa20903489ea7dfc3e35e9d0b92 : FAILURE in 14m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6e7045a46e594ff986f4e09c22fa52c1 : FAILURE in 11m 58s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ea89ee2a6c304520b13df26bf0aee6fd : FAILURE in 16m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a123a5284c014a8c94f525473df104f2 : FAILURE in 5m 51s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae21ecd9b1b14cf0ac1403aa05a1a9de : FAILURE in 6m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52a9aa3852854984aab705cd749def14 : FAILURE in 15m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2318c367c1d7457c902d8a7bc8b1e6ee : FAILURE in 19m 42s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a47bb0751344b6c9cc4c16190300ddc : FAILURE in 19m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd106593d43348ca92b9fc8d5ccd25ae : FAILURE in 11m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f854ae2ac2a64e238919dfdc9b08fb20 : FAILURE in 14m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d97dd49903141a8b91bf50dfda7a196 : FAILURE in 18m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ee064f5a164c4a6cb2ae5a32ba1eacb6 : FAILURE in 40m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e6ee72918f5499d9a1ad1cc78813095 : FAILURE in 13m 20s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d9d998f710a247f6ba6c1d1846c809ff : FAILURE in 11m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61e4068a2e544896a5f40057aaf80bf4 : FAILURE in 10m 39s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/985bac480c9d4b63a9d06d47a5a08448 : FAILURE in 13m 29s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/84bb6c43c11642a3a99316eeb8725375 : FAILURE in 11m 09s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/00b50be8a1504b91b877a6bb95455f15 : FAILURE in 27m 18s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/67cf4292c1654ad29d5240901ea27f0b : FAILURE in 17m 54s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/713c1c3b0cef478ea8c6dbb2e1f6c6f8 : FAILURE in 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adf78a0a1bcd468c5cafd5a136cbc7b248bf4390","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-13 13:20:45.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4164dba5984e5a9872b1f0a828ab27a0a537f2a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-13 13:31:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/09/702209/1/SRIOV-neutron-direct : FAILURE\n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/09/702209/1/SRIOV-neutron-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"92d713b602a206f8149011bf1cd07db454fc4768","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-13 14:15:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/1 : SUCCESS in 1h 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e37a12b3a90d525cb7829de70cd2e8bf3796c9e9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-13 15:34:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a1f86133db4860e678f3962df39c7cc2084e6db9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-14 09:46:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12e574afb290b7985b7680f7adc21ddf12a188e0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-14 09:47:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2e9f0199e5a2310c8db539bbe57a93c741ab333","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-14 10:05:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d7c8e5b74612bba717360be1b7c95881971a7e97","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-14 10:08:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9f5c09264998ed8cc9a3991139db6264ef31492d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-14 11:40:08.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/09/702209/3/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/09/702209/3/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"665f1603ce8640db99deddc17bf60cfe4c2d4d4f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-14 11:46:49.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"64cc967f425d8052ca3b9156b7dc440dcd7e4551","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 11:57:40.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/702209/3 : FAILURE in 2h 07m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5925cafcb392782fb09823feb7450e92e921da88","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-14 12:41:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"31be39fb1d646f8f0a3d58fcbfbc68c5fb1b518b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-14 13:20:18.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"085560eb47384d17b8568e7995f38064fc63f4f6","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-14 13:38:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c9f7ba5f5c95c5216f249c6463d77e0cf770edb4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-14 13:57:12.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"951a96e798eacbb0b4f8dfea9f8dd986dbebf0b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-14 13:57:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a7471ff07f1733f4c1efa10d4846a79adc7b2451","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-14 15:12:26.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/09/702209/4/SRIOV-neutron-macvtap : SUCCESS\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/09/702209/4/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d237e0531c2834ac9c54296c385cf7169937c360","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-14 16:17:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/6 : SUCCESS in 1h 47m 48s","accounts_in_message":[],"_revision_number":6},{"id":"6e87f91b693748173b7aaef3f60e2d629bec42b7","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-14 16:23:31.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks for fixing this Maciej!","accounts_in_message":[],"_revision_number":6},{"id":"852c561f06c8e9ad18867823ea48b2ef31d4971c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-14 17:11:00.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/09/702209/6/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/09/702209/6/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"e804d8b8ce5474a1169bdfba26a3553aad69c66e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-14 22:12:04.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"2cef9b5c1ab9faa1b984212d721f863aec56650b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 23:03:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fc39b34e9ca24a22ac12dadd0b09641d : SUCCESS in 54m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b6b0dd894524427a6e948b09e2e0ceb : SUCCESS in 31m 45s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/aa2809211dee448aba5010ac7b5811ef : FAILURE in 1h 25m 43s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b9dc0a46c9144ef69ca89d4b71dd7ed1 : SUCCESS in 1h 05m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9cff849ab8cb4d8ab1ba88ac6836d146 : SUCCESS in 1h 27m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91bb77a772a843cdaf7361890661c7e1 : SUCCESS in 58m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d56e325ca3f485ebaa54badc4817d3c : FAILURE in 32m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b34c19bee73d4237b2827a154062de8b : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e0c4993d334e7f91ba1ba75c1a629d : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4a519c2cf95486cbae335e64694c98d : FAILURE in 30m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33ed7220f5d54c0ab9dff7a4a71bcf27 : FAILURE in 31m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f8d4daf21b4e768bb29f48b9591c6e : SUCCESS in 11m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/584c487684764effa0920ec010e97bd0 : RETRY_LIMIT in 7m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a6cfb61f8c9b4dc79c710bbe419a8260 : RETRY_LIMIT in 5m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c58a313882b743ff9620f5ef4d1c02a9 : SUCCESS in 1h 51m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7f839b1a500a40ef9fbaf1f79cad1c01 : SUCCESS in 1h 16m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5d40d508b6964591bf971230c1d20b4d : SUCCESS in 1h 07m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0510cad9c6b74f238aae83b41e923cbb : SUCCESS in 1h 08m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4fc0eae0b12c4b87881dbe0910430e20 : SUCCESS in 1h 31m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/51749254225d4ac3abacad20d338dea7 : SUCCESS in 1h 21m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e2a1e8160cfe4df690b99f033978b6f1 : SUCCESS in 1h 39m 41s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35d99c3df086477db381ab9b503292eb : SUCCESS in 1h 51m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/472646c2430a44a18cefdc1e7df19e51 : SUCCESS in 1h 36m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/896a8ffff52945c98c7204a0ee3849b7 : SUCCESS in 1h 00m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2bb5a9ce6c70456b86e531d370f90996 : SUCCESS in 2h 22m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/97fe77308b344d2382035596bf004546 : SUCCESS in 2h 03m 16s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/f8a2ce82e2c843a1a68915a59159a805 : FAILURE in 18m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c317cdeec1343ddbb4af14b2bdc16fe : SUCCESS in 59m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1cf6cca7f25f4f14b05eb773e4095735 : SUCCESS in 1h 10m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c6f18d902c7496b96494aad7101a260 : RETRY_LIMIT in 6m 12s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82a72c0289a04a2da7729adc02087537 : RETRY_LIMIT in 6m 50s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/593def4d613048099a3776ed57efede3 : SUCCESS in 1h 11m 48s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f00763dfd65344bb95c54df7949ec4fc : FAILURE in 32m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/482f90bb5e774c32ae1fd670158a75fa : TIMED_OUT in 3h 05m 04s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5245a3e93a2b441986f130e5834cd1c4 : FAILURE in 1h 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"da24a866c9299b251ce322bcec48174ba5208fba","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-15 02:16:57.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"3d9fc14a1baea4204fdb0a914e7c570becb0c4d6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-15 05:39:06.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"b0d20a1a316a08ca6248baa6e0820e1a59e0f5a5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-16 11:22:23.000000000","message":"Patch Set 6: Code-Review-1\n\nWe have failing unit test here. Investigating.","accounts_in_message":[],"_revision_number":6},{"id":"b75c9a55bbe73cdfe680454109d20742536f4929","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-16 11:31:31.000000000","message":"Removed Code-Review+2 by Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"f606c3343cbc5f648816034dbe85f732a695486d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-17 13:20:27.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"7d34a9e3eb9e339b153742e42f9c29973bd03a8e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-17 14:49:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"49e559fec2803d4b3aa03e6b39e5bbbd1d4253f0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-17 16:04:25.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"3872420a9c27b9f9962ab645e98b1f6db0cdc35e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-17 16:44:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/7 : SUCCESS in 1h 54m 40s","accounts_in_message":[],"_revision_number":7},{"id":"4d092ef949e3580412dc2c897931ec941f06f02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 23:04:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d335fe4b94c143e78d86ad69f3cd630b : SUCCESS in 48m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d80a275e68945759e8936e049a8eceb : SUCCESS in 33m 59s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/8571784f392646698c1b7c911672ff5a : FAILURE in 2h 52m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f8ab033297764d88b30c36fe4ba979b7 : SUCCESS in 1h 11m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f43c11af7c564a1584ccc67506114879 : SUCCESS in 1h 17m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f2f107f9727e4be39959f9473179960e : SUCCESS in 1h 00m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc90b6c517644defa6f40dd2282a66e1 : SUCCESS in 37m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f81087b619ed4073b9456db00b56aa86 : SUCCESS in 34m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5cf236c1984402bcd024ce2069de19 : SUCCESS in 17m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90797bc05fc24c2883a534ba0db547d0 : SUCCESS in 38m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6433cd28ee124d65ba6fa5a0ba1a71b1 : SUCCESS in 30m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece51396d2724c76a1172ec91dd595a8 : SUCCESS in 9m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8ce5857dda8248149a32ff065037900e : SUCCESS in 54m 38s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a65af7f04d2b4d0f9390fe82057af017 : SUCCESS in 1h 04m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4dd6527285ab4617b4b7a677cd1b07a9 : SUCCESS in 2h 15m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/58591ba71c70494f89c68d116e8ff6ac : SUCCESS in 1h 29m 33s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0dc58fbe9354effbbe3781eb3d6873c : SUCCESS in 1h 23m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1033bdd2ee7143b8ba760e5f4e602835 : FAILURE in 1h 54m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c828e1a6329a4d8699167ac9c9fc350e : SUCCESS in 1h 11m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a0c20efa03be4a28a47877289a43136f : SUCCESS in 1h 46m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e917a2d978954860bf9d4b49af376a16 : SUCCESS in 1h 39m 36s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c704556e663e4559842e8a59ed55eb0d : SUCCESS in 2h 29m 17s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/200bebf81ab349b692f2dba7f132fc49 : SUCCESS in 1h 56m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a30838c066394c91a4002b80b9834e4e : SUCCESS in 1h 01m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f166c0f5de7046bb83ed56fb762e1104 : SUCCESS in 2h 28m 19s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/686428be53704aaa83d4a605860100ec : FAILURE in 25m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1eaa31f153384a8d8bf685a40bb99a2a : SUCCESS in 1h 02m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/09fb5379439e42899541b1aa6f81125e : SUCCESS in 1h 35m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fffb1b9f5d340b1807b354f5418a2c8 : SUCCESS in 59m 03s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b980caf43884853bc4902148772a38e : SUCCESS in 1h 21m 50s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/df080a9a3dba43c48f267d68de822d50 : SUCCESS in 1h 32m 44s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9d2ea01db4fb4db181d276d9c8dddeab : FAILURE in 52m 58s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3e9b2f4bb6ae429590c271dc065d9026 : FAILURE in 2h 36m 28s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/c76e08c32fcb4270b79ab0715c134ee1 : FAILURE in 2h 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"08490cdba3dd8c4573e965f8f0ac37ba4be3716a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-20 08:05:10.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0d4c6a5e7571a93ff555339f4e3736cb9061d402","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-20 10:07:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/7 : SUCCESS in 2h 01m 27s","accounts_in_message":[],"_revision_number":7},{"id":"416065156be8a4ea51a621e6add5917f46222820","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-20 10:42:26.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"e4379011a7346403bc7e3a7180e58fc001c60619","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-20 12:35:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/8 : SUCCESS in 1h 52m 09s","accounts_in_message":[],"_revision_number":8},{"id":"5d56809ec39c8c34a93dfff4547685403ce3121a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:43:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33736faf4b514b3aa58716756c780f13 : SUCCESS in 49m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f04632ceb0241babb44fc4f6f58c98e : SUCCESS in 50m 21s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3365d31523be4df4b2eb528ca035a4d5 : FAILURE in 1h 34m 57s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/39577bd192354aa3830f419bf0ebc3e0 : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3bb415977c6748b688d5d7af848a4be3 : SUCCESS in 1h 11m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/024d01381c0c46188d5703e63fa2d202 : SUCCESS in 1h 02m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25d4e112bc4144da84f6fd2933a9c74e : SUCCESS in 33m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67bf435251374fc8a073e17814f0ef64 : SUCCESS in 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0351c4a9fd4aec813d7eca2302c94c : SUCCESS in 16m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed8a03b302b34931bb12db6f6aea6cb1 : SUCCESS in 35m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/084a1a62bbb24de08cf8a6380b798d56 : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef1f93909614daeafc28a4281f0b5c0 : SUCCESS in 9m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/51f51846085b48109fa995f119829893 : FAILURE in 45m 10s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/483e2c8ba124470ab89278f2f5828751 : SUCCESS in 1h 05m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5d1604541d10402db7ee5b35e8a7e60f : SUCCESS in 2h 13m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1948629684ae4a4d828bfc656d42d97f : SUCCESS in 1h 13m 35s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/109b63217c3348dca21b08c7a33caa84 : SUCCESS in 1h 32m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1924ef8c94045b690e484a09408c2a0 : SUCCESS in 1h 06m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6858a0b2ef8e42fe9bc85ab114328058 : SUCCESS in 1h 09m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/722f32a4b1d243a89094590a41b97cf0 : SUCCESS in 1h 06m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/891e9a7f0d8940429fc931396c119375 : SUCCESS in 1h 53m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0323905dadae4984b199644424409640 : SUCCESS in 2h 08m 19s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/03f332f4a42d464b9271ae8a0afe3a65 : SUCCESS in 1h 12m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e6b92dd653c441858ba99d558d59280e : SUCCESS in 1h 00m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd4507e4914b469fbc7ae761443f39fc : SUCCESS in 2h 15m 10s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/57dfee5c4b714d8398730eff7fcf34b4 : FAILURE in 24m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93d8b4d5c072471183a25f3e058a5e8c : SUCCESS in 1h 02m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5ac1be7c7c5b4a1cb52a5f89be639464 : SUCCESS in 1h 45m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/949ebc046a02404396a4059cac8f113c : SUCCESS in 49m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bd62dfeb65f49ce9d31065a7e57aa4d : SUCCESS in 1h 13m 43s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/17022ed914614202b956db99383b0d62 : SUCCESS in 1h 32m 23s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8942f557b1e94a40947ef61ddca7d793 : FAILURE in 32m 17s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f8ac27ee0dd4415e8d1e70b78f8d3b25 : FAILURE in 2h 43m 56s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b310df48f7564714af6bda7f5368498d : FAILURE in 1h 54m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f045247cdf8acaf90b795f2da03211cf51398d56","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-20 13:46:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"d85524010c4a900dbedff2e1b85d9bcb227bbb33","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-20 16:02:47.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/8 : SUCCESS in 2h 15m 15s","accounts_in_message":[],"_revision_number":8},{"id":"8e3e3eddb77177380374a8067d1ea07389384b66","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2020-01-20 16:41:17.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ http://arista-CI.arista.com:8000/January-2020/Arista_Tempest_Testrefs/changes/09/702209/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"82aae964d648efc8fa8d6e685b5813e2fb6e6f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 17:33:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/283c6b15a07c4a51afb4473dca928871 : SUCCESS in 1h 00m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c7b46f5a7cc4d5bb9b47c08e26620f9 : SUCCESS in 33m 50s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/b7a73194bb3440c59ecb95fcab5a176e : FAILURE in 1h 21m 19s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/ecdc1bfb826944f88738237357136e2a : SUCCESS in 1h 12m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3c3829a992f24c109e72787fcbf80076 : SUCCESS in 1h 07m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/09bffc5fcb344a5c96c60c38f285dc76 : SUCCESS in 1h 13m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f6582481277449693f540bb95813230 : SUCCESS in 35m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1abebdc97904b46920645771946bc81 : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d325eea3974dfca1821fb46da53bc5 : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/507674bef00d4d4daffa31ef409271ef : SUCCESS in 30m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbb042f371984fedb84d35364fae5a1d : SUCCESS in 32m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26cb46626efe40beae39d6394b3a9ee4 : SUCCESS in 11m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f7c10b744be64599b085aac51f60bd72 : SUCCESS in 43m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/373251559d7c4784a87083e91b2d2a91 : SUCCESS in 1h 08m 04s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/49f50fb1c7314caa969c286e415da47a : SUCCESS in 1h 57m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d425c8c22fdf4c5ab3ec4f9e075eb86d : SUCCESS in 1h 30m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c643ae7572134c99a48a10d8cda040ff : SUCCESS in 1h 22m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72a0f1f7dc6140f09421154aab400b70 : SUCCESS in 1h 14m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf37a61896a84d4eabe780e4ca945fff : SUCCESS in 1h 20m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7aa954da27c34ad1b953344f5204cb1d : SUCCESS in 1h 12m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/572e879f8b584ca09a298931ed63a54d : SUCCESS in 2h 00m 19s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/320cf94ec514477bad6bdef587568b60 : FAILURE in 1h 54m 46s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8d46537a8c7436285ac13bb8eab7a96 : SUCCESS in 1h 12m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9f83b8bc31c8444b99463867e64f263b : SUCCESS in 56m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/edf6b6822cdf4023a576da66d2103a30 : SUCCESS in 2h 26m 34s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/545438c72a394031b7bc20088b815d0f : FAILURE in 20m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7e28b72d80e4792b727ded9b541468f : SUCCESS in 56m 49s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/5855871bcc904e8f892892d57c28a346 : RETRY_LIMIT in 6m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93d80b30fde54fdd9be7a77551c4d6fa : SUCCESS in 45m 44s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0bf50e2d7544cbd9eb51405569fd6f2 : TIMED_OUT in 2h 15m 46s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6671fec811945119a0e6bcf91df8b56 : SUCCESS in 1h 10m 05s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7cb024edce8b45a9818e15c11593707a : FAILURE in 37m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e1983111a58a48f3ac1900487e8cf086 : RETRY_LIMIT in 4m 33s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2d990843410142e086dc0b3145a918a5 : FAILURE in 2h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"627b05f47b57c7f28bdfa68427328e4f7d3fcf32","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-20 22:02:17.000000000","message":"Patch Set 8:\n\nI believe we can go with this one.","accounts_in_message":[],"_revision_number":8},{"id":"c094169cdb917015aedb800726777bf62db43b88","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-21 08:11:19.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"27e4a170c0bc333783bd2c78036c96b1fa5a4f94","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-01-21 08:43:29.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"9bccfef50d28981382be277deda68c909b48daa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:43:45.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5dbf4e47c1179f47b4cb8db98d86a15292f8f23e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 11:49:43.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2e188703066b43d5927a87c470cc6621 : FAILURE in 1h 52m 22s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/40e0fd12d84f44f5a0acc57af6da8e4b : SUCCESS in 1h 23m 09s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6de3ce8e4cfa40cc90ce230a1779c1ad : FAILURE in 2h 14m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/476c8b70e6d6439bafcd1d9cba14a75d : SUCCESS in 1h 00m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a738f7fb2504ddf9c90cb7f397e4ed7 : SUCCESS in 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322aa9af21cd4ce6a71cbc167c97df5f : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d0439f80b4446aeb307d61322cdf544 : SUCCESS in 37m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89611acf3fd54feab57edc47bc1beee0 : SUCCESS in 30m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602e7fc8cdc44d4ea7e6a755e660394a : SUCCESS in 10m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b44928fb91e4905bee31b00a3634b44 : SUCCESS in 43m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/1333667f18424e6a9990d71366847c71 : SUCCESS in 1h 06m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e2bcfe0321a540cbbb8b3932eec8e8fd : SUCCESS in 1h 28m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c31f1e3b18de48d2b6e4bfa49413a367 : SUCCESS in 1h 23m 07s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1ef261b6e6c54574a7c437153599dc13 : SUCCESS in 1h 23m 53s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f08c7ef85ec143cfae38ccbf50bf0852 : SUCCESS in 1h 12m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fbd6e5a0f0c4aebbc7045dab4d292d2 : SUCCESS in 1h 26m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5bd28bbc465341d99596cdc7d356a865 : SUCCESS in 1h 06m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b7ebdb77134e4e83a503b259516c80a2 : SUCCESS in 3h 02m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3c9661ba9b847b7a445255a96949504 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":8},{"id":"b19fc071148667ce904c0bf689ff82c6f74162d9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-01-21 13:20:57.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"d0973d89c3a96e38400ff81df6aa7fa593732f94","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 15:12:35.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/702209/8 : SUCCESS in 1h 51m 06s","accounts_in_message":[],"_revision_number":8},{"id":"7f9df92297d77e9e119433a0c311c12b550e9812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 18:30:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c0d3a5f2e03a42b2b31d42bf22261800 : SUCCESS in 1h 18m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d3c5325f076408a8d1b6dcbe149a446 : SUCCESS in 40m 02s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/3c8048b0921346c2b9ea14a40ad9bb38 : SUCCESS in 2h 00m 53s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0caa9b88ff8049418375414b20483423 : SUCCESS in 1h 35m 59s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6b8032c274a14f24a179cea9d96a0d4b : SUCCESS in 1h 43m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6030f417fcad4155935a807f0d9b2947 : FAILURE in 1h 33m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7854d3eb8640438ed9e603471a4a16 : SUCCESS in 43m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bbc7ae0d0424a54a64399c9a65bb3d2 : SUCCESS in 23m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf1d2b5ad22425e88ad469729f1fb2f : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b073462b8a1341339186d71ddf474b7d : SUCCESS in 35m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d20c0e0206824092ba2686aa32459091 : SUCCESS in 26m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70efdf06304a48bb918a7048bd4b98a0 : SUCCESS in 10m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/757ebf22040e4065887f9f550566e8f3 : SUCCESS in 1h 02m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/67507a3af47f4a52837c0c4b386b303c : SUCCESS in 1h 11m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/020e44efe18041b29b43c7cf9071f805 : SUCCESS in 2h 43m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/59508741ce7a4bd6bafd9d0bf2603a47 : SUCCESS in 1h 24m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b6636e6073c646968d9840ee68ae9af7 : SUCCESS in 1h 32m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6656936e2714e359eee489ff541f759 : SUCCESS in 1h 12m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ee77433243e48d9ba68bc6b19bfb4e5 : SUCCESS in 1h 08m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8d7656ddb6734ab79ee6a36614598d17 : SUCCESS in 1h 32m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/12cd43e559894d728a9d7a0ed08950e3 : SUCCESS in 1h 54m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15b4019bf1944b1e86c4b5545d261ea7 : SUCCESS in 2h 30m 36s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1d6494ab15eb477e903db37d51095193 : SUCCESS in 1h 40m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a6009ef5b1824efb8509190b5a6f9222 : SUCCESS in 1h 02m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/664e4fcbe89647eb87a6e2fe26e30813 : SUCCESS in 2h 38m 35s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9eaf6156c8bd418e9ebc9d8300f855ae : FAILURE in 28m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ca6512af0e643f4a6a0048f211faff2 : SUCCESS in 1h 11m 06s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7be17c21e9e04d029b3e45617b9b3fe5 : SUCCESS in 1h 07m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4297f43693e940b58796f701fa69aace : SUCCESS in 51m 50s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bb45020b45f445fa69be35c1d82f0ce : SUCCESS in 1h 08m 15s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/491e7e69db5b4f1ea09d27319eaa3f3e : SUCCESS in 1h 12m 16s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ec206c5fdf349dca92f7e4fe63c5d0c : FAILURE in 30m 56s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d09b51291b4c45539f2b4941943661eb : TIMED_OUT in 3h 04m 39s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/f72a3e7f22704a2a875be2e193e8e6db : FAILURE in 1h 49m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"421a1c3d8c843943fe36872a118414e10f3a562c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-21 19:54:12.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"e82c16adb9f13cef232992baf4291ed239951fb3","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-21 20:19:51.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/702209/8 : FAILURE in 25m 10s","accounts_in_message":[],"_revision_number":8},{"id":"a33af441d41e7b0f3adcdedaab9ba0c4efc5e86c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 23:12:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/82076c132c2645639b0d62e80c416e25 : SUCCESS in 1h 25m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/083fdedb8cb74d9f92439d30b9fde162 : SUCCESS in 33m 37s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/bf83f7c50cdf468b9adf4024210c8ff0 : FAILURE in 1h 46m 38s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/82c4dce612da4e3ca6d91923cdfdcf5e : SUCCESS in 1h 13m 17s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/8e86361b097447e2b4e035303fab81dc : SUCCESS in 1h 12m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91f1e0ad7b6e42029d6a0479e3ca69b2 : SUCCESS in 1h 31m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aa14e11e6c8431dab519b30a2892d08 : SUCCESS in 43m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/854a8b9a59ad44afb0f973362c14cefe : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa3b384b04e8408da624ec1a9ad7bf3d : SUCCESS in 15m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ef52c0e38e94b78ad33d31da9e8390c : SUCCESS in 31m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78251e9138b94049b2ce7c670ce04f88 : SUCCESS in 34m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37516f5cfed54a55a2775d4f4f1cd485 : SUCCESS in 9m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a859076584bc4b5dafc2a0c4125e618a : SUCCESS in 43m 49s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6136b679aa3d45978e39719cd325360d : SUCCESS in 1h 07m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6f56d1a5175c44468fb309b0747b693b : SUCCESS in 1h 43m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2f524567863a4279a854db0cc9a66999 : SUCCESS in 1h 14m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bc4231be89244ee48deb51d578d8deac : SUCCESS in 1h 09m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5a0e6b018664d42bb03beaa0961bf59 : SUCCESS in 1h 22m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5c1792ab6ae34121884adee84bd40eb1 : SUCCESS in 1h 28m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2adfc300555748aebe358e8eda1b4eda : SUCCESS in 1h 26m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f30ded8dfaa1478aab8806c59ad94925 : SUCCESS in 2h 07m 05s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1745b1f0aca449f693be32f444e8260b : FAILURE in 2h 06m 28s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6530132b1e044de3803261c99487d084 : SUCCESS in 1h 06m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bcfb86888e164dd0ae87b3c83bcfa17b : SUCCESS in 1h 07m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9ca51ec773d4e1380505c0b724eb156 : SUCCESS in 2h 33m 17s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/9f886ab5c8cb4731a2ef8130e6a4f3b4 : FAILURE in 16m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee7ec48f084b46fcb777ba684d99c403 : SUCCESS in 1h 04m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/796e713a2e184d1ab8ad6de8488a10a6 : SUCCESS in 1h 16m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4dcf9e2a9b5f45a0937774f6d3d8957e : SUCCESS in 48m 02s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e285e3516c84f3d95735a185dcb3e28 : SUCCESS in 1h 13m 00s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6bd1102823e467ea3e9092573b94c74 : SUCCESS in 1h 17m 07s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/24e6c8f4cb7a413aae1f318f1d43534f : FAILURE in 32m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e8e96aef8f7d4123b8d7f112239478a8 : FAILURE in 2h 58m 25s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/148a5da512464ebdb001c0859db4be20 : FAILURE in 1h 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"130f66988a52c9f72fcab66891732b03e9b5ebc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 23:13:03.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"908cad670442355d4b5548af8a7263075bdf937d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:28:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"230c5fe08eda6bbb82056a4e8c500f5dcf8e1f49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:28:15.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/947b037d578b48d39f74923d99fcb70f : SUCCESS in 42m 12s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/334a1d53b854499793ae46d573b5637c : SUCCESS in 1h 01m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/2d9a0357362045b6817c8da681a6fa3e : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d2c5d93d912b4d72a07bac7c8a4e6ada : SUCCESS in 1h 13m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2f90b7a74614b0d8c0c3918d353e674 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd2b2363621647ba93f14194d66b4ffa : SUCCESS in 14m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/669bd9fd471e4a01ae1494b74903e23f : SUCCESS in 27m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ce210abb17946fe833e6d388a72c8b5 : SUCCESS in 35m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e152f00ba3493ba608195aa25ecc4b : SUCCESS in 8m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/76b1d1b1273a4572987cd370bedbf032 : SUCCESS in 45m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9de84c4307b7462caad2a61fe340a88d : SUCCESS in 1h 02m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7ef15d72ac2a44ae887d3e8f2333ec4e : SUCCESS in 1h 19m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/90cb78b801e3481a89d82f144aa9097c : SUCCESS in 1h 19m 36s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/87142edeb23b43edb20cb1412fd6b5ae : SUCCESS in 1h 15m 42s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bff21b11bde141d5965bf556da62ea06 : SUCCESS in 1h 08m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6f75216ea1a4f8ebe08ec43d2c013a2 : SUCCESS in 1h 14m 12s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0bbffe3d8c7943f6bbde86029b759e71 : SUCCESS in 1h 10m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd818e73687c4d23a27da53182f0c247 : SUCCESS in 2h 13m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b59500b2af514d2b93402bd41c34a6ff : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":8},{"id":"72d74a2246e25dfe94c0d47bb73cd441a995bf8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:30:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9527e3e4f17e4ff388f586c02877b50b : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"cc890bf88b707d867fda1c7712e51d88d590e571","revisions":{"97c5f9c3bfa6ec042c3e9244c4aef126ddf7efd9":{"kind":"REWORK","_number":1,"created":"2020-01-13 12:23:59.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/09/702209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/1"}}},"commit":{"parents":[{"commit":"6c391cada651ef0a5ae9d01b90af29a7dcf42d79","subject":"[OVN] Change DevStack to use the local OVN driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c391cada651ef0a5ae9d01b90af29a7dcf42d79"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:52:29.000000000","tz":0},"subject":"Pass context while removing OVN port, router_ports and routers","message":"Pass context while removing OVN port, router_ports and routers\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97c5f9c3bfa6ec042c3e9244c4aef126ddf7efd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97c5f9c3bfa6ec042c3e9244c4aef126ddf7efd9"}]},"branch":"refs/heads/master"},"0a3212447aa3658c312cc7305f24f821bb1bbafc":{"kind":"REWORK","_number":2,"created":"2020-01-14 09:46:11.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/09/702209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/2"}}},"commit":{"parents":[{"commit":"2d9878b3e7831cf78a035a5973f9afe01b228945","subject":"[OVN] Add missing pyOpenSSL requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2d9878b3e7831cf78a035a5973f9afe01b228945"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-14 09:41:37.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a3212447aa3658c312cc7305f24f821bb1bbafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a3212447aa3658c312cc7305f24f821bb1bbafc"}]},"branch":"refs/heads/master"},"24d66b1a1efc3037be197ea69ebbff3934b7cebf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-14 09:47:12.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/09/702209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/3"}}},"commit":{"parents":[{"commit":"32a4a7dbed8a1d368dc7bc9d7293921c5e85d564","subject":"[OVN] Add missing pyOpenSSL requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a4a7dbed8a1d368dc7bc9d7293921c5e85d564"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-14 09:47:12.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24d66b1a1efc3037be197ea69ebbff3934b7cebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24d66b1a1efc3037be197ea69ebbff3934b7cebf"}]},"branch":"refs/heads/master"},"d7fa79cc4127247137e9949426575fea22c2739c":{"kind":"REWORK","_number":4,"created":"2020-01-14 12:41:15.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/09/702209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/4"}}},"commit":{"parents":[{"commit":"32a4a7dbed8a1d368dc7bc9d7293921c5e85d564","subject":"[OVN] Add missing pyOpenSSL requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a4a7dbed8a1d368dc7bc9d7293921c5e85d564"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-14 12:40:42.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7fa79cc4127247137e9949426575fea22c2739c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7fa79cc4127247137e9949426575fea22c2739c"}]},"branch":"refs/heads/master"},"9f01a7b76d466326ed4eb376c4bd61de96e4276f":{"kind":"REWORK","_number":5,"created":"2020-01-14 13:38:09.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/09/702209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/5"}}},"commit":{"parents":[{"commit":"32a4a7dbed8a1d368dc7bc9d7293921c5e85d564","subject":"[OVN] Add missing pyOpenSSL requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a4a7dbed8a1d368dc7bc9d7293921c5e85d564"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-14 13:37:09.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f01a7b76d466326ed4eb376c4bd61de96e4276f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f01a7b76d466326ed4eb376c4bd61de96e4276f"}]},"branch":"refs/heads/master"},"21d8cd176558311eb367928bede7516ac45e8c45":{"kind":"REWORK","_number":6,"created":"2020-01-14 13:57:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/702209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/6"}}},"commit":{"parents":[{"commit":"32a4a7dbed8a1d368dc7bc9d7293921c5e85d564","subject":"[OVN] Add missing pyOpenSSL requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a4a7dbed8a1d368dc7bc9d7293921c5e85d564"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-14 13:57:10.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21d8cd176558311eb367928bede7516ac45e8c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21d8cd176558311eb367928bede7516ac45e8c45"}]},"branch":"refs/heads/master"},"43bb701d7d34bd2fafb1551cd51cba4019065306":{"kind":"REWORK","_number":7,"created":"2020-01-17 14:49:18.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/09/702209/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/7"}}},"commit":{"parents":[{"commit":"6c410b166a18362b488d210fed36c2433388f1f4","subject":"Merge \"[OVN] Add missing OvnSbIdl and OvnNbIdl unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c410b166a18362b488d210fed36c2433388f1f4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-01-17 14:48:40.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43bb701d7d34bd2fafb1551cd51cba4019065306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43bb701d7d34bd2fafb1551cd51cba4019065306"}]},"branch":"refs/heads/master"},"cc890bf88b707d867fda1c7712e51d88d590e571":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-01-20 10:42:26.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/09/702209/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/702209/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/702209/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/702209/8"}}},"commit":{"parents":[{"commit":"1d354f75776fca6bf12beae70f7e3206459c84c4","subject":"Merge \"Move tripleo standalone job to ostempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d354f75776fca6bf12beae70f7e3206459c84c4"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-01-13 11:50:04.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-01-20 10:41:43.000000000","tz":0},"subject":"[OVN] Pass context while removing objects in OVN client","message":"[OVN] Pass context while removing objects in OVN client\n\nThis context was required but not passed. This change adds it.\n\nChange-Id: I5e5e14d5a621fc4a790810d15e1162f90898282e\nRelated-Blueprint: neutron-ovn-merge\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc890bf88b707d867fda1c7712e51d88d590e571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc890bf88b707d867fda1c7712e51d88d590e571"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
