)]}'
{"id":"openstack%2Fneutron~704682","triplet_id":"openstack%2Fneutron~stable%2Fqueens~Ia05df925eccf3c9d397748f282e995203a058de9","project":"openstack/neutron","branch":"stable/queens","topic":"bp/adopt-oslo-versioned-objects-for-db-stable/queens","hashtags":[],"change_id":"Ia05df925eccf3c9d397748f282e995203a058de9","subject":"objects: automatically detect whether engine facade is used","status":"ABANDONED","created":"2020-01-28 20:50:32.000000000","updated":"2021-04-14 13:43:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"535bdbcceef3c0f0b145a51030240ec4f5a3da2f","_number":704682,"virtual_id_number":704682,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-31 20:32:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-31 18:12:13.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-02-15 13:00:53.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 18:12:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-31 20:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-15 13:00:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-14 13:43:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"c2aa737940f125415fab4a589ac9ee57cdc8d772","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-28 20:50:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3ff491a0bd2e251a30418967cc7eed6b6b987500","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-28 20:58:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/704682/1 : FAILURE in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3181527fb27cc526b80f36898ea0e0491cd1dbd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 00:05:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/d4d3eab8b1844567bcbaf2c22d1ad180 : SUCCESS in 58m 55s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/dac673d930b74d4986ffc336a76b7a17 : SUCCESS in 35m 28s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/62ecb71ef73f407c9f087ffabb977d50 : SUCCESS in 1h 19m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/f05a7223a86a4def932df87e3d5e4acb : SUCCESS in 54m 59s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/67b75fa9f57d41a89b3e9376234ab14f : SUCCESS in 56m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10644a2bb9b2498fb687b2dfc47b5731 : SUCCESS in 10m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49d2f5fe8b564913af3275c6a6ad82c8 : FAILURE in 21m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d470e5e5b70249789b0cec21b664562c : FAILURE in 31m 27s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f9e625c4909146e0a59dddb60d5409b2 : SUCCESS in 9m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6dd0b09ecbb74812a3c9891565677952 : SUCCESS in 1h 25m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/521f69fd12ff4ed596095694c11fec86 : SUCCESS in 1h 14m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/653a871c0a724dca8d8224c7ad7d6f28 : SUCCESS in 1h 27m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d5b323020a34e5aabed42bbc5b0ba93 : SUCCESS in 37m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f6fb270c10654663b10df86aa8628996 : FAILURE in 26m 35s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/368f4e9a3cda42d0b76df1c069af5345 : SUCCESS in 1h 31m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/468be1f4b44340f49f82b9cfa73f8028 : SUCCESS in 1h 45m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f1edf39046434086a919f60ff88f5bed : FAILURE in 1h 51m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4aa1f413344452d8de472f85d897b19 : SUCCESS in 1h 05m 53s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f04f23cd4b040379532300edba4e439 : SUCCESS in 1h 03m 31s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c3c918f31b064803a8c7ab27f835eeff : SUCCESS in 1h 41m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2299947932fd48128b36131f550f450d : SUCCESS in 1h 50m 44s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/5aa7f1f9320b4eaeb75ec68bcfe62d13 : SUCCESS in 1h 37m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6da3fa8966324b508c530dfccead656e : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0deef542e2294ba500a41678847cb6f32f849a68","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-01-31 18:03:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f05b1b964a1b0f7fc175b594893d0bcf26cd8bef","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 18:12:13.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/704682/2 : FAILURE in 8m 26s","accounts_in_message":[],"_revision_number":2},{"id":"491a1563fcb00c66bad9629374f1ab12fc1420be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:32:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/b383c2a394cc4b4698c5bd9395991783 : SUCCESS in 59m 02s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/012eb91e63114881900a4cf73631bacb : SUCCESS in 39m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/1bb7f2bd127c4cf8b75b7427fb20fbca : FAILURE in 1h 41m 54s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/9692f5dba6db432b98b05503d0e40044 : SUCCESS in 1h 00m 04s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/6871b5666f9a446083010cf6e36440b0 : SUCCESS in 1h 02m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7917adb9ae480fb70d67f7987325bb : SUCCESS in 10m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/90550571e1154449b35a94eac8ab6779 : FAILURE in 19m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f1dc71eb9e5a41bd914ffc88c20e171d : FAILURE in 22m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a5aa72ffd784dd2813f0461d6580154 : SUCCESS in 10m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0074fcc43a2442938065bdd21f2b5e67 : SUCCESS in 1h 30m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/40f24936cee54a5c8c43e87bbef0637d : SUCCESS in 1h 22m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89f73d0e32694dc5a3e5b6e406cac018 : SUCCESS in 1h 31m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/284967fdee574826bb2f0b23540c850b : FAILURE in 57m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/10f29dd4f1614645bff3ee0aa4f06c54 : FAILURE in 54m 05s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/550bbfb1154d4bedb84fa8ecc25ae01b : SUCCESS in 2h 03m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ed38608d808e42c9a3f54f92fd13566e : SUCCESS in 1h 25m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1290093e8838454b8e358090fbcc7d3b : SUCCESS in 1h 35m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/28a68c0d8a2f4f1c8fb7a9e251470a09 : SUCCESS in 1h 13m 24s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/94aaf3f4a8e04720b8aa2235537c7758 : SUCCESS in 1h 39m 37s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/56980181845d45819f6ead99b1f6c782 : SUCCESS in 1h 36m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7193807d15ff46e0ac2e60bad67eab68 : SUCCESS in 1h 32m 38s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/97b06c8f8ba34b3795774bf80fcd6102 : SUCCESS in 1h 19m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7efe7e9fc6624d76b42a7c20658317fd : FAILURE in 33m 07s","accounts_in_message":[],"_revision_number":2},{"id":"346a693d5f7d14d9b6238253b8602ac924a10e28","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-15 13:00:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2f05ecbbe8d4540403c368f140f60a1bc0e7c2d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 14:44:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-queens https://zuul.opendev.org/t/openstack/build/fbff4bbe1bb442a9ba719f1ab7f4a693 : SUCCESS in 54m 13s\n- neutron-tempest-plugin-designate-scenario-queens https://zuul.opendev.org/t/openstack/build/86cf1267564b459ebf10dce4ff63d224 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-queens https://zuul.opendev.org/t/openstack/build/25300ec8609d4bce8fb2d55329543c68 : SUCCESS in 1h 13m 29s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-queens https://zuul.opendev.org/t/openstack/build/73f9fcfcfee9406f90567d60e18204f4 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-scenario-openvswitch-queens https://zuul.opendev.org/t/openstack/build/46627ce4b7944d1ab53932aa19a2218a : SUCCESS in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4316a3fd86614cb19cf0302ceb0584ba : SUCCESS in 11m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6371d35bf484d57a64e2a0ff357bf45 : FAILURE in 22m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0bef7b3866984b25af5a94d00c26fc70 : FAILURE in 24m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e425903b3de94e06922ebd6a4a4c62f1 : SUCCESS in 6m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/234fd766306b4bb699f358bff12b8496 : SUCCESS in 1h 18m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ad49f5c1c02042a38eddc58ea74b2bf9 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f32c5d0a6dce4af39c58ada5c3aa1bec : SUCCESS in 1h 19m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/37e84e637c9d423db26821f5ecbbcd77 : SUCCESS in 42m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3c88e594ba6545198ab95be6d16b16b2 : FAILURE in 20m 26s (non-voting)\n- neutron-rally-neutron https://zuul.opendev.org/t/openstack/build/42b2d87bd51447698b8e12e61099a795 : SUCCESS in 1h 37m 54s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e799f7252e4a4bc9993464dc37922192 : SUCCESS in 1h 34m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/8ac89fd0316e467d9d843c05f2ed6e53 : SUCCESS in 1h 17m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0014eeb4a3284187aaf0683eba25e08b : SUCCESS in 1h 11m 54s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41241876775948c1babfae73d62494b2 : SUCCESS in 1h 05m 54s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a35066d18f57476eb04979bbeef01a32 : SUCCESS in 1h 22m 06s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99aae7458ddc45eca96b95ddd4132d3c : SUCCESS in 1h 38m 28s (non-voting)\n- neutron-tempest-ovsfw https://zuul.opendev.org/t/openstack/build/dde6c318e37e45cbad35678ab4c64621 : SUCCESS in 1h 13m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12139c8d48694415919ea1273fcfa1f1 : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"535bdbcceef3c0f0b145a51030240ec4f5a3da2f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:43:27.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6f9e245222a96a6cc8444d8f38cf8ef1dfb2c0d","revisions":{"c32833dcc7604ebdcdd6548562fcaa14423db1c6":{"kind":"REWORK","_number":1,"created":"2020-01-28 20:50:32.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/82/704682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/704682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/704682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/704682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/704682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/704682/1"}}},"commit":{"parents":[{"commit":"49e35bc4ba5c65ecfe1044fc21fcd8cb3d0346c3","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49e35bc4ba5c65ecfe1044fc21fcd8cb3d0346c3"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2018-03-15 21:25:04.000000000","tz":-420},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-28 20:50:32.000000000","tz":0},"subject":"objects: automatically detect whether engine facade is used","message":"objects: automatically detect whether engine facade is used\n\nWhile we added new_facade object attribute to framework and it has its\nniche and used in some stadium subprojects, it\u0027s not ideal because it\u0027s\nglobal to an object. Meaning that if you mark an object for new_facade \u003d\nTrue, *all* business logic using the object must also switch to new\nfacade in the same step, which is a pain and sometimes close to\nimpossible to do without changing thousands loosely related lines of\ncode in multiple modules. It would be nice to instead use objects as\nusual in different contexts - some using engine facade and some still\nusing session.begin(...) - and allow the OVO framework to pick the right\nway to nest subtransactions.\n\nThis patch does exactly that. We call an internal function from oslo.db\nand check whether it raises an exception. If it does, it means that the\nengine facade is not used; otherwise, we use the new style of nested\ntransactions management.\n\nBy default, if session is not active when OVO action is called, we stick\nto the old facade. Once we are done with switching the rest of the\nplugin code / OVO objects to the new facade, we will rip off the\ntransitionary logic.\n\nChange-Id: Ia05df925eccf3c9d397748f282e995203a058de9\nPartially-Implements: blueprint enginefacade-switch\nPartially-Implements: blueprint adopt-oslo-versioned-objects-for-db\n(cherry picked from commit 92b95815318806f90eca987145ad453782dd7041)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c32833dcc7604ebdcdd6548562fcaa14423db1c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c32833dcc7604ebdcdd6548562fcaa14423db1c6"}]},"branch":"refs/heads/stable/queens"},"c6f9e245222a96a6cc8444d8f38cf8ef1dfb2c0d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-31 18:03:14.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/82/704682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/704682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/704682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/704682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/704682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/704682/2"}}},"commit":{"parents":[{"commit":"3559d8bcc7ddae553dac59505596abbc020922c7","subject":"Merge \"Allow to kill keepalived state change monitor process\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3559d8bcc7ddae553dac59505596abbc020922c7"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2018-03-15 21:25:04.000000000","tz":-420},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-01-31 18:03:14.000000000","tz":0},"subject":"objects: automatically detect whether engine facade is used","message":"objects: automatically detect whether engine facade is used\n\nWhile we added new_facade object attribute to framework and it has its\nniche and used in some stadium subprojects, it\u0027s not ideal because it\u0027s\nglobal to an object. Meaning that if you mark an object for new_facade \u003d\nTrue, *all* business logic using the object must also switch to new\nfacade in the same step, which is a pain and sometimes close to\nimpossible to do without changing thousands loosely related lines of\ncode in multiple modules. It would be nice to instead use objects as\nusual in different contexts - some using engine facade and some still\nusing session.begin(...) - and allow the OVO framework to pick the right\nway to nest subtransactions.\n\nThis patch does exactly that. We call an internal function from oslo.db\nand check whether it raises an exception. If it does, it means that the\nengine facade is not used; otherwise, we use the new style of nested\ntransactions management.\n\nBy default, if session is not active when OVO action is called, we stick\nto the old facade. Once we are done with switching the rest of the\nplugin code / OVO objects to the new facade, we will rip off the\ntransitionary logic.\n\nChange-Id: Ia05df925eccf3c9d397748f282e995203a058de9\nPartially-Implements: blueprint enginefacade-switch\nPartially-Implements: blueprint adopt-oslo-versioned-objects-for-db\n(cherry picked from commit 92b95815318806f90eca987145ad453782dd7041)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6f9e245222a96a6cc8444d8f38cf8ef1dfb2c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6f9e245222a96a6cc8444d8f38cf8ef1dfb2c0d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
