)]}'
{"id":"openstack%2Fneutron~705117","triplet_id":"openstack%2Fneutron~stable%2Frocky~I594014d5662c731b85ee1ad814efe757aecb9d3c","project":"openstack/neutron","branch":"stable/rocky","hashtags":[],"change_id":"I594014d5662c731b85ee1ad814efe757aecb9d3c","subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","status":"ABANDONED","created":"2020-01-30 20:59:07.000000000","updated":"2020-02-09 02:31:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e0dc81578bf43ad38def33b62a3eb5da2310bcb","_number":705117,"virtual_id_number":705117,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-02-09 01:31:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-09 02:28:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2020-02-09 00:01:44.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 21:11:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-02-09 00:01:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-09 01:31:40.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-09 02:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81517a9a84e532ae485c2f27c21a58f45c1190f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-30 20:59:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a2127f0c5b21171111a467b81c9dde497aa4810","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-30 21:08:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/1 : FAILURE in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f0001e95474d96139f0a850f5d08e538ad4a9aae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-30 21:11:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a3ee50764749a7bed8340e0b1449faa86067fdae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-30 21:38:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5bf0b7a9e5eb8457e558353510bb10cac3999609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 23:12:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/c060ffd76e834e10a9a5937d607c0099 : SUCCESS in 52m 06s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/16433918feeb447ba0a6751114f70d72 : SUCCESS in 37m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/76c2e446cc49431fbd447c1f675e09c2 : RETRY_LIMIT in 6m 31s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/3f1e329dadfd4d119300d449d04a02c2 : SUCCESS in 1h 11m 43s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/6178065fac1542f4913ba2fd8bf23e23 : SUCCESS in 1h 10m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/3f38e259a9b94cdb846efa2344abe5d0 : SUCCESS in 1h 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c9b1cfddda43a893a0e4c59b4a8595 : SUCCESS in 13m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9de98c7fffb74c24ae01fe7fe02fca45 : SUCCESS in 23m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/74ab4bacff39472fb1a2e98778fe6a54 : SUCCESS in 34m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e9164e153c2d4ea98d872ad8d904436c : SUCCESS in 7m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1257f3ef9323452197c0c9685452cca2 : SUCCESS in 1h 28m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3b59cbad32864551a19a499bf9eaab3e : FAILURE in 49m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd501896ca284f84a8e265d1f1ab88db : FAILURE in 1h 09m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1c4e59d90e214290a69ef2c1181aeee8 : SUCCESS in 39m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/cf8d20051d2d4ee59f97799e7d64a57f : SUCCESS in 53m 59s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/68a5522db46d4e4180b6af72b333e364 : SUCCESS in 2h 10m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d2b55e29a3154b98a647f4e171c887bd : SUCCESS in 1h 33m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/50beffc0263b4490bba5ec9de5066716 : SUCCESS in 1h 31m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bb55e7a670644ef8db62cc2b5c0fe10 : FAILURE in 1h 00m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/30d777c30c5d4fa6b51521afd199fea8 : FAILURE in 57m 25s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de7fdbcde4924e619fc9cdf63693c6d7 : SUCCESS in 1h 46m 15s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e99ffb0c1ad746cfb3a2f0105d9c4770 : SUCCESS in 1h 43m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28ce89e32d8c42b680e20188796f55f4 : SUCCESS in 1h 23m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f99e190c423f4eeeaa4102aef849bb5f : SUCCESS in 23m 45s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/f6a33cecc3ce4f4fa3f98fffbd127964 : SUCCESS in 58m 13s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3209fbfa76cc461d965c3c5c9988d1e3 : FAILURE in 58m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e14d05bc4c34d45bc4528f5e00e07cf : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e6d1ac4f35e8ded97449fc2e4e1907d548fe20ba","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-31 16:07:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef5c089826b4ed5cb2866c0faa826e78dbbe94d6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 16:16:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/1 : FAILURE in 8m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8f80cbe24c8a72ca168c20c00eda133f667a9a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:43:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/dee243e643cb479b988ad73ddbfc5cc9 : FAILURE in 33m 35s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/645ced1df7ac4c899e4200b138bf28fb : FAILURE in 36m 03s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/9a38d208d2284721bcc3c9ad4e764027 : FAILURE in 47m 07s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/e989622fec8d4bbf95a4e47f5e2d2b65 : FAILURE in 32m 55s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/c542a671235149acac26e9529892ffea : FAILURE in 35m 18s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/b703ad13592b4ddabeebbc44cff92e44 : FAILURE in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02c2d51a96c4105ac79a7692f9bf5a4 : SUCCESS in 13m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43818ab8c94d4823ba97d9400ad85655 : SUCCESS in 22m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/615470d71298413da7f9ab1312a899a0 : SUCCESS in 24m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2c74aba35dd9411e8cea97dda1b9e448 : SUCCESS in 6m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/639225e79d864c3da3dabade1077f28e : FAILURE in 33m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f9e018fc92d448539cd798bbbb9faa53 : FAILURE in 45m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c35a2525966644558be1627b9c5c4bdd : FAILURE in 39m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2fd0da5d8835482db551a16c3c63b90d : SUCCESS in 40m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/6c5e07504bb84a0e93f1fd9e055993d7 : SUCCESS in 1h 06m 39s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c51253e109194d43bb7163c2cc78a967 : SUCCESS in 1h 33m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e4db75d803fc40d7984429efca7d4aae : FAILURE in 36m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/76b2203caccd44cfae8e0150d0bee12a : FAILURE in 35m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38ae55273f154e43981c577b70f338d5 : FAILURE in 56m 02s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb7baa0038a34cb9a52397caaab7787a : FAILURE in 59m 14s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9fc5657f2d044c5e943a7f74345b95d0 : FAILURE in 34m 39s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/28a650d58b03438ba72e67d43c4a0bd0 : FAILURE in 35m 06s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/88996f350c19489c95e3e5ce46c9d814 : FAILURE in 36m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31814359b5d84834b364c8b4907b2f39 : SUCCESS in 21m 55s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/dc47718df1ad4694bd045fcbfe0ae7ca : SUCCESS in 56m 14s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/480fde16f2304038a539cbbd66366553 : FAILURE in 56m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d8d88de39a04da6b69ca5405c5ea26f : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f34de7f2e40ed7011c21eb5e99517380f2ac136e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-31 18:49:17.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"18cc308526dbdbba1ff66a3359f1e8370bd4b25c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-02 00:02:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45440e6455e6ee4d87f02689bd3949de7a4e720d","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-02 00:11:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/1 : FAILURE in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"918dcb76cf510025957afd6cf8c94a1cb20fd5ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-02 01:59:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/1/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cab173b9a292d7190e77d5543edf4d05cd9ce8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-02 02:05:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/0c40964c3dc84dbb8fd3573c727f09a0 : SUCCESS in 51m 13s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/5dfef19f3498476d9caaa384365fa713 : SUCCESS in 34m 07s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/aa0bd54a1cfc486a8b3d5277256b52d4 : FAILURE in 1h 57m 01s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/83d0249a52204b2681694307831f42e9 : SUCCESS in 1h 13m 52s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/6c25bc2614734c53bed8cc4cbb4f8fa0 : SUCCESS in 1h 15m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/0dc40772f0634e5fbc48871cffaecf1e : SUCCESS in 1h 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85d638260554fbe86d647fbfbba2394 : SUCCESS in 11m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d3bedaad00c4015b5718ad2fc90c462 : SUCCESS in 24m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7081d3edc7a643faa5e7ff64daa4a253 : SUCCESS in 24m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b5f65edffc3a476aa9257c5a7cfbac24 : SUCCESS in 6m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b572cdae6e94393a7913387416ba6f3 : SUCCESS in 1h 29m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d97ea69bdb9d4d158bc63a5e94b7b909 : FAILURE in 44m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05fc44f1eb2d4d31b401a482bda02f44 : SUCCESS in 1h 17m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d05358738bc44a829c1b29bf60b40a29 : SUCCESS in 40m 19s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2aeaa7958ea84223a0a78049ad7fa26e : SUCCESS in 54m 40s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/5e4f1b76186e4589a4da7d1972678e3a : SUCCESS in 1h 30m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/63db30ef141141fd9e3f4e91bfe2818e : SUCCESS in 1h 29m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1f7eb9bc30604ee5b1cba25b31ac8953 : SUCCESS in 1h 19m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/02e9686158b945c383f8fab57ca413c3 : FAILURE in 54m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/15a90fedcc7e42948cb1c5243ea348df : FAILURE in 57m 09s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/20cda72a6cf54d4e8da3ec18f5683d36 : SUCCESS in 1h 31m 17s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d01740f690694f518764fb845cb8ebf0 : SUCCESS in 1h 32m 07s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1a71d6804c7a422589929cd1e1a6e36f : SUCCESS in 1h 20m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a17ba4e32e74eccaa97da479c16408c : SUCCESS in 26m 32s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/b70cc7f3cbe84da3b3a1d36ec6f1f43a : SUCCESS in 1h 00m 55s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c7f12d75fb6e40f688909af8655c265a : FAILURE in 56m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce4bb2fe22c5460ebdea4e3499ebc95f : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8c4c98cc85735b259c67e596cd85d61732363e1f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-03 17:11:08.000000000","message":"Patch Set 1:\n\nall good here as jobs are running with py2 as expected.","accounts_in_message":[],"_revision_number":1},{"id":"7c157e99e67772627015818f190208785f2666c8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-03 17:11:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8e14d81b47da367dad1df3d656a07887298856b7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-03 17:11:48.000000000","message":"Patch Set 2:\n\ntesting with Tempest 23.0.0 and stable/rocky u-c.","accounts_in_message":[],"_revision_number":2},{"id":"557ca0d57c1338e9f21dcec967cde0ae7b96f8f6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-03 17:12:57.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/2 : FAILURE in 45s","accounts_in_message":[],"_revision_number":2},{"id":"255d28edcfa0e27c648dee7f1a3ff80b4e788b97","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-03 18:19:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c4cd364babfebaad0803652b062ac9041ae33df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 22:47:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/29fbde13a54c426eaf548e6f48fc33d7 : SUCCESS in 53m 26s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/a8b410a870254e2580b0584a402f1cbd : SUCCESS in 35m 29s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/df91364f0e904d44bce00f817907538d : FAILURE in 1h 40m 39s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/c270685a2dda441ab7ccd9404ad24f46 : SUCCESS in 1h 20m 30s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/28cdec0d3c634ee480201d3159349ea1 : SUCCESS in 1h 08m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/4bc51c37270144af913d953d6c937d34 : SUCCESS in 1h 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee058c80ba404cd2bc609f8eadeaa16c : SUCCESS in 15m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5501823c0074577ad26f57e35867a79 : SUCCESS in 21m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae7a7752b7e640f0bbdbd4dbea700b4b : SUCCESS in 30m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9c43126378094c66b0d45ba2387914aa : SUCCESS in 9m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/af150d738c2440b09f55872cc5989533 : SUCCESS in 1h 30m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a87bb9ae988048c69336ac36553231a7 : SUCCESS in 1h 07m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d62a13f7132d422ca502df6b0f0e3429 : SUCCESS in 1h 23m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4089bc56ec7a49c99ba0c7e8fc155b9f : SUCCESS in 46m 03s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9f50684386d249a3b555e09d0bf28cf1 : SUCCESS in 57m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6ae7799316864161b67f02fdc2e1adae : SUCCESS in 1h 30m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c1fd81062a2d4fb2a7adc7b914951b9f : SUCCESS in 2h 09m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/96c940a193ee4a84a50a26e335ccdc7a : SUCCESS in 1h 38m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/872cd3da9b544b4c9ab5d3663fd8b059 : SUCCESS in 1h 11m 05s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d2e84c3de63249f5b3a72734d0de859e : SUCCESS in 1h 03m 52s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3354063296eb41658c9122c74743e642 : SUCCESS in 1h 53m 30s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cba139e858a0494ab2e4421ca8d643e7 : SUCCESS in 1h 57m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/72361091afb8487b8dba850add5a2335 : SUCCESS in 1h 28m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66ab3ebd7f794363acb43b6635b50e45 : SUCCESS in 31m 30s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/c1a9dfd13c954acb80590272c48227be : SUCCESS in 1h 00m 10s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c8632396950441ae8daa3de1840c4d74 : FAILURE in 56m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2533b6c6880a4108898b353e511d26be : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":2},{"id":"bd141bc497391499f6134cf44054105b94f57ee8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-04 23:00:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e9d23ef11b2c4379c6c7ba794eb52c2852a26fc5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-05 00:38:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/2/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/2/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"516ff16e52ca5ffc0cb58f487c2f30365b831e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 00:48:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/759767039703416db79ddda5a8237c0e : FAILURE in 15m 56s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/83796e4b13d6459f9d1740220083878f : FAILURE in 15m 31s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/1072a160508d441eb353437fee4b3b3c : FAILURE in 28m 16s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/84e4e9a17c97465b8c835a6ffd84b7ae : FAILURE in 15m 36s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/52864fdb5c2b4ef3bdea044283028a4c : FAILURE in 18m 54s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/01cb74623038467da8041c5a13414201 : FAILURE in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d76ced0e4ac49a884f400681824eca5 : SUCCESS in 12m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d69c2a66889f47a397acaa9646a27ecf : SUCCESS in 26m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/afc337da6ac846d9bfc756093c058772 : SUCCESS in 20m 49s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a1bb039d3f0d4fc6a5edfad12674a897 : SUCCESS in 6m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/48dc97b38d7e465eaa8005e9d2ca64ec : FAILURE in 27m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/21879608861844049b91ded829334994 : FAILURE in 26m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6dea61048654de4aab15c33145731d4 : FAILURE in 15m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/14482ea216a4443bb0b816bc0cb1c7c8 : SUCCESS in 45m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/54af598eea2f460e853cff29bcdbe240 : SUCCESS in 1h 09m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/66db83e313cb487d80262fcbf71cf1b6 : SUCCESS in 1h 43m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/31a47fb1f667426085c8b98db5b9950b : FAILURE in 25m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2707655982ad4e9dab9868e354ed2464 : FAILURE in 18m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/774900c0c01f42668fa14b0550dfb4d6 : FAILURE in 23m 19s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6402b132165b46c0b5592f33505f3665 : FAILURE in 21m 54s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/57da36c8d3314a76ac1c818e1575e7a6 : FAILURE in 28m 16s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f1714c728cd24b26a3083a1c8d314336 : FAILURE in 32m 51s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b54c13d4363440b89f87976b2d70d465 : FAILURE in 18m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eafc58c5fb6c49c182e36848e5bbcdd2 : SUCCESS in 28m 34s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/5e22bbe22ade4784a9d951534f4fc27e : SUCCESS in 52m 53s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c1140b318c78437e93923ecbd4fcacad : FAILURE in 19m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/194516d3ec48413cbffb8d30a774a3c4 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":2},{"id":"592ebf9394c5592b1bd0d062bb9844968e12276e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-05 20:42:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"05090f30acb7989f261c0197f7695e551c16270e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-05 20:51:24.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/3 : FAILURE in 8m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f566b39d80dcbfdbce06f08081a908df83b06f67","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-05 20:52:53.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/3/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/3/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"f5324c4e7c8bdf05e11864e1a193c23e555944b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:27:59.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"496c091eb6bc170ce4c720396f335e24a0b62e85","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-05 21:28:09.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/3/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/3/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9eab98f5e0a30b113605d73d9051f403888042c0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-05 21:34:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fcee88ff1f712880d64dfb0ba26e112b5e12fb6a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-05 21:37:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a23c2b8d5159ce4d0ae0913adde85739ae63e95b","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-05 22:07:43.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/5 : FAILURE in 7m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f943c14e0099369a7d1c463a2569e0d95a15956a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-05 22:08:45.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/5/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/5/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"713a7bb76cfa0d67531400ca985f59f1eb3fc8c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-05 22:58:08.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/5/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/5/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8362dcb76efe3996d02423bd1cec6df2a1e4851a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:30:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086b829e63454cfebbe8d7418014d4cc : SUCCESS in 15m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e0f734c5fc6470889bb400c57a65e67 : SUCCESS in 30m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6bec34482a5f41dc8ffa4e3374ba63e5 : SUCCESS in 26m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0d1de9cc740e4d838e9fdeeac883e5e4 : SUCCESS in 10m 11s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e6a828ec68ec4c27964289416da43136 : SUCCESS in 53m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4573d36d91d549dd8a7a905def95d4b4 : SUCCESS in 1h 41m 25s","accounts_in_message":[],"_revision_number":5},{"id":"74a88c45e95454bf6e1d3af50fad55ffe8954cb6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-06 00:13:42.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"79cd58dfa9d0ce44b0c89ea0931f4cdcaf397e41","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-06 00:22:48.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/5 : FAILURE in 8m 29s","accounts_in_message":[],"_revision_number":5},{"id":"a69d25e127a006b78a10bc09c09dbf37ad162fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 00:59:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7a7403ecd9416787179511008d286e : SUCCESS in 10m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a30b3400f244be8b183835d763c56b4 : SUCCESS in 23m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7e956586114f41ae96d1a9bd9aae8bc6 : SUCCESS in 22m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7050bc184bd4e7eaa95daf59e9853dd : SUCCESS in 10m 33s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/f9587f1fa5474b35a9a7b930582efb80 : FAILURE in 18m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ebf1b619f330474e93f76cde6379c379 : FAILURE in 21m 50s","accounts_in_message":[],"_revision_number":5},{"id":"4302a0d737190e2c254920035ba58e4222699302","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-06 01:15:46.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"fcb57124017d65cf7db992b884c0e17032f34761","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-06 01:25:25.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/6 : FAILURE in 8m 55s","accounts_in_message":[],"_revision_number":6},{"id":"ff8dd7744b4316f2046863e3699e7b79badd5ab0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-06 02:43:58.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/6/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/6/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"ca5b072c54c7ee9bfb61b33ae9d852761f27aa98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 02:49:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6283e6c47966419884438dbc23b00f0a : SUCCESS in 14m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0416a604d9aa4c8eb7107ab1c7c524ce : SUCCESS in 27m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe0efae969224aa98bb275c53636aded : SUCCESS in 27m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6ee348c78ebf4e09857d2bdfa94adf8a : SUCCESS in 7m 37s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/e5ef30acc6804689a33070022ca8849a : SUCCESS in 54m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a89fa1df66714175983517d4d88195d6 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":6},{"id":"cb74e1aba894d18bdd874f99e1709541dba28bf6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-06 16:00:39.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"fd71d0d86ba42f2eb86657e651afc12c7bf00a2f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-06 16:20:04.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/7 : FAILURE in 8m 15s","accounts_in_message":[],"_revision_number":7},{"id":"99d181334d23606e0ef8a598976c2103ac57ad91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-06 16:22:09.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/7/SRIOV-neutron-macvtap : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"d92ebb73d4b5291004a8457de678346ad948c9c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-06 17:07:51.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/7/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/7/SRIOV-neutron-direct : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"8354d3aee9a417cbfd2d201e2c12d1b63ee43593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:13:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e85d8e43c084c19bbfe637b37df0493 : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/252be7aa337c45608ffa7ecd80368b14 : SUCCESS in 28m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab5158cc5e1548afadadcf7ef35cc83f : SUCCESS in 22m 00s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a3bd58f2952b44e0ac65bfa40629bc4f : SUCCESS in 8m 26s\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/b9873b61b3724699a6278ccbd7e7621e : SUCCESS in 58m 33s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f358bb53f8df4b81a9fa2b676373fad8 : SUCCESS in 1h 59m 19s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/331c275859f14debb560209f0abb583a : SUCCESS in 1h 39m 43s","accounts_in_message":[],"_revision_number":7},{"id":"681348436a0bebc93548230ae098c6f899e9d323","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-08 23:52:34.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"54a7ded1c89086cd47fb17d4d4eb706d7b9da2d8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-09 00:01:44.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705117/8 : FAILURE in 8m 29s","accounts_in_message":[],"_revision_number":8},{"id":"cafeac9b7f23bdb47fd09feb1a8b1c8798c1664f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-09 00:03:51.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/8/SRIOV-neutron-macvtap : ABORTED\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/8/SRIOV-neutron-direct : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"74c8ecdd9443e4bc3328e30f368f5cd48254d3b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-09 01:31:40.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* SRIOV-neutron-macvtap http://13.74.249.42/refs/changes/17/705117/8/SRIOV-neutron-macvtap : FAILURE\n\n* SRIOV-neutron-direct http://13.74.249.42/refs/changes/17/705117/8/SRIOV-neutron-direct : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"29c342b1c2e566e40a3c74b976e20984a2c05915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 02:28:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/46890d3cb46c469cb4b400aae6914375 : SUCCESS in 1h 35m 49s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/255026b5c8d24215999f80f904616a99 : SUCCESS in 31m 58s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/fd9c0e6698a142c6974cdf58662d5368 : SUCCESS in 2h 32m 18s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/350e2603affb4056a0027fd83aa4d8f6 : FAILURE in 1h 48m 16s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/ec8f9a55d5fb4072b382c0aebb6b5e92 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/229b56e0f6a04492ac12750621498530 : SUCCESS in 1h 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f02bb67fce04305b412f7fb53c80213 : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f3d7591b3114150842db543697be651 : SUCCESS in 22m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c5cdd00bed2647bc93c0f913030e6712 : SUCCESS in 29m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d1c22e0ea22f467ca822bfb005e7beeb : SUCCESS in 9m 55s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c996c82bc9748479f2d9ba3da8feb79 : TIMED_OUT in 2h 27m 45s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/0bac7132b50441b58ac91071fafd5d06 : SUCCESS in 2h 06m 01s","accounts_in_message":[],"_revision_number":8},{"id":"4e0dc81578bf43ad38def33b62a3eb5da2310bcb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-09 02:31:00.000000000","message":"Abandoned\n\nall good now","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d5f2211720bac9f4d63fddd56963ce0e20dd8089","revisions":{"10a5edb517b4972adb77fa3595568ecf9f610b83":{"kind":"REWORK","_number":1,"created":"2020-01-30 20:59:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/1"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/704257/\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10a5edb517b4972adb77fa3595568ecf9f610b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10a5edb517b4972adb77fa3595568ecf9f610b83"}]},"branch":"refs/heads/stable/rocky"},"224f171d373f3fb3a03b510591897efd359f8fc2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-03 17:11:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/2"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-03 17:11:25.000000000","tz":0},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/704257/\nDepends-On: https://review.opendev.org/#/c/705098/\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/224f171d373f3fb3a03b510591897efd359f8fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/224f171d373f3fb3a03b510591897efd359f8fc2"}]},"branch":"refs/heads/stable/rocky"},"17f854d77990f44f11494e6ad80a77f5bc376e08":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-05 20:42:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/3"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-05 20:42:35.000000000","tz":0},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/704257/\nDepends-On: https://review.opendev.org/#/c/705870\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17f854d77990f44f11494e6ad80a77f5bc376e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17f854d77990f44f11494e6ad80a77f5bc376e08"}]},"branch":"refs/heads/stable/rocky"},"25dd7c039b8d4892d63438cdd76fbb390ed48dd8":{"kind":"REWORK","_number":4,"created":"2020-02-05 21:34:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/4"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-02-05 21:34:17.000000000","tz":-360},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/706119/\nDepends-On: https://review.opendev.org/#/c/705870\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25dd7c039b8d4892d63438cdd76fbb390ed48dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25dd7c039b8d4892d63438cdd76fbb390ed48dd8"}]},"branch":"refs/heads/stable/rocky"},"4bb8fc9ff67c534994ef774d41533f881690e85c":{"kind":"REWORK","_number":5,"created":"2020-02-05 21:37:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/5"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-05 21:36:58.000000000","tz":0},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/706119/\nDepends-On: https://review.opendev.org/#/c/705870\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bb8fc9ff67c534994ef774d41533f881690e85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bb8fc9ff67c534994ef774d41533f881690e85c"}]},"branch":"refs/heads/stable/rocky"},"33e371a574ea78ee26325d649daaa883554b1b84":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-02-06 01:15:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/6"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-06 01:15:44.000000000","tz":0},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/706119/\nDepends-On: https://review.opendev.org/#/c/705089/\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33e371a574ea78ee26325d649daaa883554b1b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33e371a574ea78ee26325d649daaa883554b1b84"}]},"branch":"refs/heads/stable/rocky"},"18b7e3e399cb91ef60e3955847b1f267891aeaca":{"kind":"REWORK","_number":7,"created":"2020-02-06 16:00:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/7"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-06 16:00:32.000000000","tz":0},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/706119/\nDepends-On: https://review.opendev.org/#/c/705089/\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18b7e3e399cb91ef60e3955847b1f267891aeaca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18b7e3e399cb91ef60e3955847b1f267891aeaca"}]},"branch":"refs/heads/stable/rocky"},"d5f2211720bac9f4d63fddd56963ce0e20dd8089":{"kind":"REWORK","_number":8,"created":"2020-02-08 23:52:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/705117/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/705117/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/705117/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/705117/8"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-30 20:58:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-08 23:52:31.000000000","tz":0},"subject":"DNM: testing rocky jobs running on py2 on stable/rocky gate","message":"DNM: testing rocky jobs running on py2 on stable/rocky gate\n\nDepends-On: https://review.opendev.org/#/c/705089/\nChange-Id: I594014d5662c731b85ee1ad814efe757aecb9d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5f2211720bac9f4d63fddd56963ce0e20dd8089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5f2211720bac9f4d63fddd56963ce0e20dd8089"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
