)]}'
{"id":"openstack%2Fneutron~705200","triplet_id":"openstack%2Fneutron~stable%2Frocky~I3905ea56ca0ff35bdd96c818719e6d63a3eb5a72","project":"openstack/neutron","branch":"stable/rocky","topic":"bug/1774459","hashtags":[],"change_id":"I3905ea56ca0ff35bdd96c818719e6d63a3eb5a72","subject":"DVR: Modify DVR flows to allow ARP requests to hit ARP Responder table","status":"ABANDONED","created":"2020-01-31 12:01:08.000000000","updated":"2020-02-10 10:46:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c50643dc97cb9246583b5d28c619163f5fb72c61","_number":705200,"virtual_id_number":705200,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2020-02-10 10:46:25.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-31 13:40:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-03 11:21:30.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":-1,"date":"2020-02-04 17:10:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-01-31 12:01:08.000000000","_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"value":0,"date":"2020-01-31 20:12:54.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 12:01:08.000000000","updated_by":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"reviewer":{"_account_id":7016,"name":"Swaminathan Vasudevan","email":"swvasude@cisco.com","username":"souminathan"},"state":"REVIEWER"},{"updated":"2020-01-31 13:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-31 20:12:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-03 11:21:30.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-04 17:10:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"496e11bbb68864a59ea2bf0cbc3fed2bf6febc77","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-31 12:01:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"acc0679f19e106ff2e43730114fb33c4cc6ccd18","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-01-31 12:29:44.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705200/1 : FAILURE in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"07f65123cd5692607024e7786b88ab4bac130a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:40:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/301f3097511f4907a9f58592b25a7ed5 : FAILURE in 30m 59s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/09a28d8fddd14b5bb1cf43bb24472238 : FAILURE in 37m 11s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/b18dd5e4085e4d71980c214b3e50c583 : FAILURE in 46m 55s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/0197181c0cf548398c5617852e462640 : FAILURE in 32m 03s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/c6e4be0773a64a8c8fafaff8fedc0afd : FAILURE in 32m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/16af2a894d6d4d03a4f4c25b68e485d0 : FAILURE in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d797cc50b5d148918fe8a946eddd1f6f : FAILURE in 8m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3079ca1ea5ab42549557f5ecf7ab77c4 : FAILURE in 23m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d1e0e42c4af24842b1203539bb64694e : FAILURE in 26m 36s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c5331cabfdbf4dcbb45b54fa76eed6de : SUCCESS in 6m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/444712d1728f44a286ebfaee7a818f2f : FAILURE in 32m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ef48e9112aa443f8c103f9029c51edd : FAILURE in 44m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9183950394c4abe8e9c1f52953dc34f : FAILURE in 30m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c162537f7f8d4f10b92c615cd342d55f : SUCCESS in 37m 22s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ccf79b48f74149aeb08d5b90abe93557 : SUCCESS in 51m 49s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3834e5d7dbf543f39ea40437adc84e43 : SUCCESS in 1h 33m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d26a80b43cbe4e5fbae163f73e5b64de : FAILURE in 32m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/10c241e7783f4682be3a06229cc957b6 : FAILURE in 32m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcf8a1e4acfd45378c571e26b918a4fa : FAILURE in 1h 11m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a027cc0aa4f341c5a3fcc90185e227fe : FAILURE in 1h 02m 02s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f9895e1881fb4f8fac66dc9fe4ea1618 : FAILURE in 39m 13s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e603c8522374e6ebc45a7f13034541f : FAILURE in 33m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/44a7866cefb44f379ada09a3375279e6 : FAILURE in 30m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/836a00a052ac4521a2ab829607512a3f : FAILURE in 23m 01s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/76bb641c68e64c078cef728d44c69cb1 : SUCCESS in 1h 09m 22s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/71187b52d42347b18eb59253f469208b : FAILURE in 55m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14475c1f04bf4bf1a6cc9bf62bb19b71 : FAILURE in 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2a4093b8cde83dbd1e74088ebbbadf1b8a0cd64f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-03 11:04:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8937baeeb3a899279296441b738942a7c6ac7b69","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-03 11:21:30.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/705200/1 : FAILURE in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"058d3d1f9cf92a55e0f2bc9eb0c53dcc08baf702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 14:51:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-rocky https://zuul.opendev.org/t/openstack/build/4c1216f8f1474ad386a07a18a579d0e0 : SUCCESS in 50m 45s\n- neutron-tempest-plugin-designate-scenario-rocky https://zuul.opendev.org/t/openstack/build/eab732b31479432c9d2f449951ad5a0f : SUCCESS in 33m 09s\n- neutron-tempest-plugin-dvr-multinode-scenario-rocky https://zuul.opendev.org/t/openstack/build/4c5f1d4f1d064f88924bfc585cb8d2b2 : FAILURE in 1h 50m 27s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge-rocky https://zuul.opendev.org/t/openstack/build/a509099ed20240e3b4680db970cbe12b : SUCCESS in 1h 21m 50s\n- neutron-tempest-plugin-scenario-openvswitch-rocky https://zuul.opendev.org/t/openstack/build/dc44e63c859e42a6b948c107baabdaa6 : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-rocky https://zuul.opendev.org/t/openstack/build/c3e00e8d8848453b8cdde21bc21591e3 : SUCCESS in 1h 08m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6bb892942e04b4483940468f4f2cfe6 : FAILURE in 7m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74b755f2a1604d6a9c37a9129404040e : FAILURE in 24m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d571f05492f34192a827230f8124e5fd : FAILURE in 26m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1d7f8aeb861f4cc79bef99a6fb467f9d : SUCCESS in 8m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/62ce0bda1cb64338b4a4a4b3264ea4a2 : SUCCESS in 1h 33m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/724f0f11308e4c4aa4c236460ad5a8df : FAILURE in 47m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ca6ae97f8c84e57aa5a52c47fddb872 : SUCCESS in 1h 16m 07s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7f2553228c564397aae7108edf6128bf : SUCCESS in 38m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d88b5d63816749a6808ceda8b82b3b91 : SUCCESS in 56m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2df2d7b662bd4332bd7b52b7e583874f : SUCCESS in 1h 40m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7a2ceeeb10614a3695987dd015164f49 : SUCCESS in 1h 31m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d8b8ffd38674fc58005957dd9da0eb5 : SUCCESS in 1h 29m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/540b448e306a4905bc70ebcacab1a009 : FAILURE in 1h 02m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/05ab040ecef74038b7870d4b10bd7604 : FAILURE in 1h 08m 55s\n- neutron-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a43c2500d61544878384753c85b5d95b : SUCCESS in 1h 33m 21s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c5181ca337249afa1f80b2e369802c7 : FAILURE in 1h 53m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1e470b85faf04420950e6b480c43ce82 : SUCCESS in 1h 27m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13251c093e0443a19c74f4b9850445a0 : FAILURE in 22m 59s\n- neutron-fullstack-python35 https://zuul.opendev.org/t/openstack/build/6520e17f78d84792a2227fc47374cee8 : SUCCESS in 57m 29s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9eb30d4365d840e29c6706bc66a9e17d : FAILURE in 54m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddeacf4547454257bfdd1f20db9d254d : FAILURE in 30m 22s","accounts_in_message":[],"_revision_number":1},{"id":"341f9ce92b7e40f1b6de7ba9410225e014163e9e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-04 17:10:09.000000000","message":"Patch Set 1: Code-Review-1\n\nSee latest discussion in https://review.opendev.org/#/c/700805/ this patch will need a backport of \"Force arp_responder to True when DVR and tunneling enabled\" to work properly too, I am not sure if we should change this for older branches. Opinions?","accounts_in_message":[],"_revision_number":1},{"id":"51e123633dc42bb894d255929aecb9c4cf4dcc0b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-10 10:46:25.000000000","message":"Patch Set 1:\n\n\u003e See latest discussion in https://review.opendev.org/#/c/700805/\n \u003e this patch will need a backport of \"Force arp_responder to True\n \u003e when DVR and tunneling enabled\" to work properly too, I am not sure\n \u003e if we should change this for older branches. Opinions?\n\nYou\u0027re right with that and IMO we shouldn\u0027t backport it to old branches. Lets abandon this one.","accounts_in_message":[],"_revision_number":1},{"id":"5efb0a74f27c2fbbb2bc57b7b24534e4554831d6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-10 10:46:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de5d687e4ae983ec2d9f974ede792812d0f95aff","revisions":{"de5d687e4ae983ec2d9f974ede792812d0f95aff":{"kind":"REWORK","_number":1,"created":"2020-01-31 12:01:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/00/705200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/00/705200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/705200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/705200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/00/705200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/00/705200/1"}}},"commit":{"parents":[{"commit":"54e1a6b1bc378c0745afc03987d0fea241b826ae","subject":"don\u0027t clear skb mark when ovs is hw-offload enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54e1a6b1bc378c0745afc03987d0fea241b826ae"}]}],"author":{"name":"Swaminathan Vasudevan","email":"SVasudevan@suse.com","date":"2019-04-11 18:12:24.000000000","tz":-420},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-01-31 12:00:42.000000000","tz":60},"subject":"DVR: Modify DVR flows to allow ARP requests to hit ARP Responder table","message":"DVR: Modify DVR flows to allow ARP requests to hit ARP Responder table\n\nDVR does the ARP table update through the control plane, and does not\nallow any ARP requests to get out of the node.\n\nIn order to address the allowed address pair VRRP IP issue with DVR,\nwe need to add an ARP entry into the ARP Responder table for the\nallowed address pair IP ( which is taken care by the patch in [1])\n\nThis patch adds a rule in the br-int to redirect the packet\ndestinated to the router to the actual router-port and also moves\nthe arp filtering rule to the tunnel or the physical port based on the\nconfiguration.\n\nBy adding the above rule it allows the ARP requests to reach the\nARP Responder table and filters the ARP requests before it reaches\nthe physical network or the tunnel.\n\n[1] https://review.opendev.org/#/c/601336/\nRelated-Bug: #1774459\n\nConflicts:\n    .zuul.yaml\n    neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/br_dvr_process.py\n\nChange-Id: I3905ea56ca0ff35bdd96c818719e6d63a3eb5a72\n(cherry picked from commit 52b537ca22b2d7d81a84b2f75de577d8dffee94c)\n(cherry picked from commit 09c4e0e970b02ed992aab762505d17f1decca551)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de5d687e4ae983ec2d9f974ede792812d0f95aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de5d687e4ae983ec2d9f974ede792812d0f95aff"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
