)]}'
{"id":"openstack%2Fneutron~706509","triplet_id":"openstack%2Fneutron~master~I0dcc9e91d874a9bb20754d7bf3deef2db20e83b7","project":"openstack/neutron","branch":"master","hashtags":[],"change_id":"I0dcc9e91d874a9bb20754d7bf3deef2db20e83b7","subject":"Test neutron-centos-7-tripleo-standalone job viability","status":"ABANDONED","created":"2020-02-07 13:47:48.000000000","updated":"2020-03-19 19:50:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae736ecd78b2e1649f200cf19538cdaa9d438430","_number":706509,"virtual_id_number":706509,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-19 10:42:54.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-11 20:47:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-11 14:26:08.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 15:31:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-11 14:26:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-11 20:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-19 10:42:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d5f3269bcbb3bfb586f2dc73b641d7311041b188","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-02-07 13:47:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ecd47fa4c3697b3aa1d5422c3bfd59b59f1e87b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-07 15:26:00.000000000","message":"Patch Set 1:\n\nSo looking through the logs, rpm_build.log (in DLRN log dir) had this:\n\nDEBUG: RPM build errors:\nDEBUG:     File not found: /builddir/build/BUILDROOT/openstack-neutron-15.1.0-0.20200207140911.a0ab253.el7.x86_64/usr/bin/neutron-rootwrap-xen-dom0\nDEBUG: Child return code was: 1\n\nI removed that recently from neutron as it wasn\u0027t being used.  The downstream Centos7 spec file might need updating?\n\nhttps://github.com/rdo-packages/neutron-distgit/blob/rpm-master/openstack-neutron.spec","accounts_in_message":[],"_revision_number":1},{"id":"862a5f29e224bd2eea98624f8af46739d2c111b3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-07 15:31:03.000000000","message":"Patch Set 1:\n\nhttps://review.rdoproject.org/r/24836 :)","accounts_in_message":[],"_revision_number":1},{"id":"9bf17c5933e1115e33bda7174ba90ac433a3bd71","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-07 15:32:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/706509/1 : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9910379c9f4447aa9c0bdf975a69fe91a97e1b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 16:22:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d6c754c8090f45dc9b023cae32c48d19 : SUCCESS in 51m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f35fd9432964f3184fb3aad7273e075 : SUCCESS in 41m 27s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/0b75573757c1410aab84d7d3497f2ffa : FAILURE in 1h 53m 08s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/63ad98cd45104bb2b252e18620566cbd : SUCCESS in 1h 06m 12s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5b5cc203b3ff4ed99a3c42e4b67dece9 : SUCCESS in 1h 13m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5f3268c7b856403dab0d09112c5fa25d : SUCCESS in 59m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06d15b0c815b4c6f8e8d19eacd56e422 : SUCCESS in 40m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9484f0d8fecb4be19814058aa78d3a22 : SUCCESS in 30m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397579a75fc6465e9dd4208ed5a9b055 : SUCCESS in 15m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90a58c789f84acc844b6ae7a7a880c4 : SUCCESS in 27m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16e5efce3ee24e1bb72be6dceb18c498 : SUCCESS in 38m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e053d75ec8ea4158aa83952ac70ce34c : SUCCESS in 29m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a374b2746a77401c97c0f8eb3f7b9960 : SUCCESS in 11m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1f4d217cc6ba46ce93ea5d30b0992637 : SUCCESS in 55m 56s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/abcb67c8f85949db9cb87324cc9c0eca : SUCCESS in 1h 17m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b58121c617e6499f99aab29b8884eb93 : SUCCESS in 2h 12m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c432ebcd3e6f4004b2b065cdc9082c26 : SUCCESS in 1h 15m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7bc1876a26f14803bdb56459a65c24ea : SUCCESS in 1h 16m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2d39b65b1274d0c851f9505dcfffca9 : SUCCESS in 1h 11m 42s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/56ccc8f0cee04294900f733da7b66273 : SUCCESS in 1h 22m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/85e20c2bff8d413b991e5cb161b31a18 : SUCCESS in 1h 13m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/276709b23b4746058449df556b26e18d : SUCCESS in 1h 46m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/197c0b5f92a9457eae5ec874b3b1556f : SUCCESS in 2h 09m 55s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bfa2eea590614cbe85072a95746ae365 : SUCCESS in 1h 21m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0aebd815dd464e629e6a667f81d4d5eb : SUCCESS in 1h 06m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b25edfdbf9e487692af533cc17a00f8 : SUCCESS in 2h 26m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8ea7cb8fdf984f32b474804fcf267ecc : SUCCESS in 1h 48m 34s (non-voting)\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/df50aaeb1f044f52b44eccbc43f81240 : SUCCESS in 2h 11m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5664a19ace634efc8324d9753d99f77d : TIMED_OUT in 2h 06m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/613982512aee4817b8a12f4930b3526d : SUCCESS in 1h 07m 31s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d8812ef801c4ff49c2c06406d648a73 : SUCCESS in 1h 00m 34s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ac818c5a9604149a7ad665679421cb4 : FAILURE in 2h 09m 59s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8dbbc19e6a94a46be33a2b97837d26e : SUCCESS in 1h 06m 57s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8559ba40fb3547f7836ba4776cd900fe : FAILURE in 31m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2a6b4336b57d460d82e1ef27dbf6efb2 : FAILURE in 2h 26m 23s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/0f13a8f991c046b8b4b225ddde4201aa : FAILURE in 2h 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f420d635bfb137345088b46d4b345e84019b5f5a","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-11 14:00:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cf6b52feb6606e98d05cf1fdbea7222ec40feb14","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-11 14:26:08.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/706509/2 : FAILURE in 24m 57s","accounts_in_message":[],"_revision_number":2},{"id":"cb3d5f2ddf6f75d3410c7bea120ca80415eb23e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:47:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8e315fa6234447fb8e371408bad00056 : SUCCESS in 1h 01m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/75fd4cb909804dfbb184e319c8d96b6c : SUCCESS in 39m 52s\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/afabe80074fe45fab63addde74bf8597 : FAILURE in 1h 52m 36s (non-voting)\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9b2d5ad9f27249aaa739731061d40af4 : SUCCESS in 1h 14m 44s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c0bd9d894db84546bfd6815f0279ed31 : SUCCESS in 1h 04m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fc6bc086f90b4062a3f1e1892f298e0b : SUCCESS in 1h 20m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd1df05dc7ea432e9e18b6714688231f : SUCCESS in 37m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/984f879d5f544045814be437ca48ee87 : SUCCESS in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823192752bbd4e1394c2208b2649c1fd : SUCCESS in 14m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b49a1aa8a1764c4fbfedd7a0a912dcd3 : SUCCESS in 31m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/206a017d4e20463aabbb4d664f3dfd47 : SUCCESS in 32m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29f4d206ae9d42dc8df56d6da0ca9e33 : SUCCESS in 29m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ad5d13f9c946ba98e0f80246788d6b : SUCCESS in 10m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/248bdeb5b69f4d38807df646bcfaeb3d : SUCCESS in 1h 02m 42s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d46e78c616164b1181285a9c2d1bad8a : SUCCESS in 1h 10m 10s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4059a42f0ede44688f335247287903db : SUCCESS in 1h 57m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d3b6a426f7048c5a9403a019e7df7b7 : SUCCESS in 1h 18m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/42968e8f2f954e698cc2f806c8fb4350 : SUCCESS in 1h 11m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b08f897a61504434ae36e5507ce12c27 : SUCCESS in 1h 02m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d63c18c3ad5941e399915e6a251379a2 : SUCCESS in 1h 16m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f07445a324fd4c0a9c2e0ff87b248dd1 : SUCCESS in 1h 44m 27s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/981f83d6ace546e18b7a83a55fc35600 : SUCCESS in 1h 36m 08s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8c9987fd3bb41d8846279194d772ddc : SUCCESS in 1h 17m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b1552b7b564d4fae8aceff13b6e29d4a : SUCCESS in 57m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b61df997fb3b4fda87eeee13ed593134 : SUCCESS in 2h 15m 18s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ac691d5f139147a6b7344aeadfb6f64f : SUCCESS in 1h 47m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3ea4b2cbc3694cb7a641270f655e4f7c : SUCCESS in 2h 07m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d96c686dff0742d49eff7eb889ac5065 : SUCCESS in 1h 06m 12s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/472fadae91064be682b68f208e8760ee : SUCCESS in 1h 13m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/44161feba5364b7fab0fbcecf3c35fe4 : SUCCESS in 1h 08m 21s (non-voting)\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4b424210dd74a7f9255753ecea2d2e0 : SUCCESS in 1h 13m 20s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ad46c9d519f473798423dcc6d7d8e01 : SUCCESS in 1h 22m 28s (non-voting)\n- neutron-centos-7-tripleo-standalone https://zuul.opendev.org/t/openstack/build/71144293a3764ee8a011281ce2d0908a : FAILURE in 51m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e2280a04a2c44361accfbfee09bc9ea2 : SUCCESS in 2h 05m 52s (non-voting)\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/06525c44dd054d668ff286fb1aa4d519 : SUCCESS in 2h 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"473d873acc9a8aa8893d6d21cf4733a9412d6724","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-19 10:42:54.000000000","message":"Patch Set 2:\n\nI already have patch to switch centos jobs to be run on centos 8: https://review.opendev.org/#/c/701351/","accounts_in_message":[],"_revision_number":2},{"id":"ae736ecd78b2e1649f200cf19538cdaa9d438430","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-19 19:50:40.000000000","message":"Abandoned\n\nYeah I think this can be dropped.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5426facdfbb3f0850f56fa87b20cd052e4a1caa","revisions":{"a0ab2532a6bfa8d16d7f0c69d17a5051d9887bac":{"kind":"REWORK","_number":1,"created":"2020-02-07 13:47:48.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/09/706509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/706509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/706509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/706509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/706509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/706509/1"}}},"commit":{"parents":[{"commit":"d5b33ffc775ffff500271c0f15f055ae3132a8d0","subject":"Merge \"Remove duplicate QoS backend support table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5b33ffc775ffff500271c0f15f055ae3132a8d0"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-02-07 13:47:27.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-02-07 13:47:47.000000000","tz":-300},"subject":"Test neutron-centos-7-tripleo-standalone job viability","message":"Test neutron-centos-7-tripleo-standalone job viability\n\nNow that the drop-py27 goal is nearly complete, check and see if the\neeutron-centos-7-tripleo-standalone job can be reenabled.\n\nChange-Id: I0dcc9e91d874a9bb20754d7bf3deef2db20e83b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0ab2532a6bfa8d16d7f0c69d17a5051d9887bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a0ab2532a6bfa8d16d7f0c69d17a5051d9887bac"}]},"branch":"refs/heads/master"},"a5426facdfbb3f0850f56fa87b20cd052e4a1caa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-11 14:00:21.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/09/706509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/706509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/706509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/706509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/706509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/706509/2"}}},"commit":{"parents":[{"commit":"230c25d209faa75bfba1942a1db8489e088c0a6b","subject":"Merge \"Remove leftovers of get_external_network_id for router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/230c25d209faa75bfba1942a1db8489e088c0a6b"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-02-07 13:47:27.000000000","tz":-300},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-03-11 14:00:21.000000000","tz":0},"subject":"Test neutron-centos-7-tripleo-standalone job viability","message":"Test neutron-centos-7-tripleo-standalone job viability\n\nNow that the drop-py27 goal is nearly complete, check and see if the\neeutron-centos-7-tripleo-standalone job can be reenabled.\n\nChange-Id: I0dcc9e91d874a9bb20754d7bf3deef2db20e83b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5426facdfbb3f0850f56fa87b20cd052e4a1caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5426facdfbb3f0850f56fa87b20cd052e4a1caa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
