)]}'
{"id":"openstack%2Fneutron~707486","triplet_id":"openstack%2Fneutron~master~I190afcff68c4a20f42aec34cb299bd862e7c7efd","project":"openstack/neutron","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I190afcff68c4a20f42aec34cb299bd862e7c7efd","subject":"[Community goal] Add contributor and PTL guide","status":"MERGED","created":"2020-02-12 22:19:25.000000000","updated":"2020-03-10 19:34:10.000000000","submitted":"2020-03-10 19:31:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":81,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707486-1583868683605-bd11f162","meta_rev_id":"f5cc0c31d68f0d42d149ce83c09bc55cf50fb26f","_number":707486,"virtual_id_number":707486,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 19:31:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-09 17:58:40.000000000","_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"recommended":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":1,"date":"2020-03-09 18:35:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2020-03-10 01:34:34.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":1,"date":"2020-03-10 07:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-03-10 14:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-03-10 14:35:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 18:58:23.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-02-21 17:50:19.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2020-03-06 07:44:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-03-06 13:20:36.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-03-09 17:58:40.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-09 18:35:58.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-10 01:34:34.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-03-10 07:40:57.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-10 14:35:29.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-03-10 19:31:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4927f134dcd18fee5d060ce95f0c82240b565a3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-12 22:19:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37a4883ad8613ad5757f63d245bbf4b8413107d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 22:39:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ca6797379f4e93af0ac42d23d0ca02 : SUCCESS in 14m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d16777dc71425d8b6ee3b68b13bc1d : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"01643bbd8ac2e0d37cf31ce95ed2b77ecabe40d0","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-13 00:12:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/707486/1 : SUCCESS in 1h 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1d6938d924f335bbbf254c6adab02e6f36804a73","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-15 21:45:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3234578282f1f211fc1ffc22db5e78966cd1e030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 22:02:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c340fd728d9446b90ed7ae6104f4070 : SUCCESS in 16m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072e0188b0af4ce1b20f4ba4004fdd06 : FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7e5f980042dc720ed1ae4416be83e043b5e6bc42","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-15 22:11:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"448e1d9ada74f0f135b7c0b3f1a8d8edb3f5f3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 22:29:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807c9400b0e64ecf99f732238470b7eb : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a876bf3d884a4397b96bc94fb207a246 : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"4c2b4cc733b887cc30ae09e3fbba7da728956c19","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-02-17 01:53:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b661d3985edf3c400f23338e03f2272a4a4861ff","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-17 09:38:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"857bae2649c7269c432277b0229693a2e68241d7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-20 12:10:49.000000000","message":"Patch Set 3:\n\n(10 comments)","accounts_in_message":[],"_revision_number":3},{"id":"01750369280ec1626a2b79e49a9b58d8164e8f77","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-20 12:11:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b43299ffc7de67529c35ffc81926741ba4557f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 12:31:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1f51e397fb4eea9a7cb942ec28b28d : SUCCESS in 16m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32560fc62fb4480835a3369aedc40b9 : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":4},{"id":"81998b27882c6ed53ff9b60dde418c46c6950b1e","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-02-20 13:43:14.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/707486/4 : FAILURE in 1h 31m 23s","accounts_in_message":[],"_revision_number":4},{"id":"6d74e30c4e4c5b2fb39a33e0b542b0eb17974155","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-02-21 00:24:21.000000000","message":"Patch Set 4: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2973e64942c371dd668aefa506b50000d451b91c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-21 15:28:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fa2790fd116a7c0acca9a5dcbd5f153ee74c3ec0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-21 15:28:36.000000000","message":"Patch Set 4:\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"110d31eace4c49e43465b5822baed7d93b571d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:15:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e458349c6a11433093fa654afb2ada6d : SUCCESS in 17m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3eabd5b35344a39769604e2b75765e : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":5},{"id":"0240a073d90aa24c54055cec12105b75c376e98c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2020-02-21 17:50:19.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"782aab8427236ee050a388a68911d824c8a396e5","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-02-21 20:59:41.000000000","message":"Patch Set 5:\n\nIt might be a good idea to also update CONTRIBUTING.rst [1], and make sure that we are in sync with the updates being made to the cookiecutter template [2].\n\n[1] https://opendev.org/openstack/neutron/src/branch/master/CONTRIBUTING.rst\n[2] https://review.opendev.org/708672","accounts_in_message":[],"_revision_number":5},{"id":"b06877772f5088b4e2430819318360d41d00b0ec","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-26 14:32:31.000000000","message":"Patch Set 5: Workflow-1\n\n\u003e It might be a good idea to also update CONTRIBUTING.rst [1], and\n \u003e make sure that we are in sync with the updates being made to the\n \u003e cookiecutter template [2].\n \u003e \n \u003e [1] https://opendev.org/openstack/neutron/src/branch/master/CONTRIBUTING.rst\n \u003e [2] https://review.opendev.org/708672\n\nYes, I will check that and update this patch if needed","accounts_in_message":[],"_revision_number":5},{"id":"24c3cdc7962634e5ad5527c85453cf273aa798e4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-04 21:17:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8fbeecf227a44ae27e03c7af54d825391f1a8602","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-04 21:18:36.000000000","message":"Patch Set 6:\n\nIt\u0027s now updated according to new proposal in https://review.opendev.org/#/c/708672/3\nIMO it is be ready for review now.","accounts_in_message":[],"_revision_number":6},{"id":"a8627796100c61e33ff17255dbd057ad3732c7a1","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-04 21:19:06.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/707486/6 : FAILURE in 45s","accounts_in_message":[],"_revision_number":6},{"id":"c113c867f1c80880320007a7c207c518a70494e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 23:59:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f6df60075b42279ff199ae1e2b7083 : SUCCESS in 13m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2700393b7fc84374bbe87ab5f61bfe70 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":6},{"id":"17497c7da50d0d2ad9f4e6b4ef6efd03b62aaf26","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-05 09:41:26.000000000","message":"Patch Set 6: Code-Review+1\n\nIt seems more ordered with these changes, thanks","accounts_in_message":[],"_revision_number":6},{"id":"eff82b7f5271c251ea15a38743fe6d18a4d41070","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-05 23:53:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nOne small comment. Other than that it looks good to go!","accounts_in_message":[],"_revision_number":6},{"id":"66ed1968dae938b4175f8a41639bf27d3cc55227","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-06 07:32:35.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"229ef798c4e2518b9b917267e2352d6ea0abff3b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-06 07:32:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"81ab297f4f2c15b39d5a96f8ad69e230b3dd2f09","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-03-06 07:44:36.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"f4623c9f953a1b129aeeaa00c9f267d55cd14b03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 07:47:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3c3c28943949aa8d8516eb8439bf53 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e474568ed4e745a6b051a483c194cee3 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":7},{"id":"c04a9af97be5b43b17d12a00d6ce789f612cf544","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-06 09:19:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/707486/7 : SUCCESS in 1h 46m 42s","accounts_in_message":[],"_revision_number":7},{"id":"e5d28c6561d61c1a65c0a43bd552561aaab5f896","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-03-06 13:20:36.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nAlmost there, but I think the clarification that RFEs are launchpad bugs will remove ambiguity for new contributors.","accounts_in_message":[],"_revision_number":7},{"id":"f565dedae71f19cc9a8042baf47ad56898616422","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-06 15:49:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3f910643222257610ddc8809fa7e7734fa09f184","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-06 15:49:31.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b6fda2a4a3c4b24a050b23e6520c157af670273b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:33:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa0a849418e452f91f056778ad6fa59 : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e1b0d30f6a4f30afbb1031f7ae3b65 : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":8},{"id":"66ec8c39f4319b295d86a057b685aa368dde5e32","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-06 17:03:39.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nOne build error to fix\n\nI suppose real test would be to make some external non-contributors to read these guides, but at least it looks good to me","accounts_in_message":[],"_revision_number":8},{"id":"396306f1209f367318e57826dd3cc7245e28152f","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-06 17:39:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/707486/8 : SUCCESS in 1h 49m 03s","accounts_in_message":[],"_revision_number":8},{"id":"d5c0472b8842bea72823be4b557b16b78c6f3028","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-07 18:34:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f54e3f0bc2ed9991455165a259e2c79523d42544","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-07 18:35:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8f879362b732199a8f8876cd278958df2c75f971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 18:48:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3703cecc82674ffdae04957a8330ea7d : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a54efe7e634debb32775f355fb3d94 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":9},{"id":"c0b3ed66de5987b9d2726973bd1312c05bc208b6","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-07 20:18:23.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/707486/9 : FAILURE in 1h 43m 15s","accounts_in_message":[],"_revision_number":9},{"id":"f7c67b8fe1d47d1673fbab2d925a0c8c126f7cfa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-09 09:30:16.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"92e55ce01abb1a13b959e8c518d865c1bec6a70b","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-03-09 15:14:33.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2c9e9fa6e4861152e5c42b6df54b6c7f55c59e0d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-09 16:15:00.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0c16c7d18fab6c62f1dd62189edde08f8422815a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-03-09 16:15:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8b0f3fa569c35c475b98f4de3be513126e721213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:41:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8728856935514f2393a24cc8a5e2ac80 : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf95fad1648c4982bbc44f0b0fa8544c : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":10},{"id":"95a874cde66f47949b8fb14297d85128353a08c8","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2020-03-09 17:58:40.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/707486/10 : SUCCESS in 1h 42m 33s","accounts_in_message":[],"_revision_number":10},{"id":"b87291c1b4396c45643fd5e0d2418b9e9df85d80","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-09 18:35:58.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c7d47f6664657a49ebb980504140bc25bf18f150","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-03-10 01:34:34.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"8dd94778ab5409d7eab3305b82e0b46b9ffb59cb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-03-10 07:40:57.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"ca08c60f330869233ed55f999acd44fcfd51dc8a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-03-10 14:35:29.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f224ed6cfeaf18b075ed22fb04e7277ca5b33002","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:35:54.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a0b1899f5843c3ddedbc63dc6a7b754d636e2f61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 19:31:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"8a9e1130c9502f3957cbc14ade75935270c2704d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 19:31:23.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a20c08c7b1f4d5a937c6048f44be5d8 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec145f53b2aa4bd38987671b6519c82f : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":10},{"id":"c59bbc2f8d1fa8faf99485bdb2401c7664586f8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 19:34:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d660b70f69744698ef66d9cf28fae26 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"36656ca31871dec7467d0523ae937d890e9b364b","revisions":{"1ddbf24a12c022e0f9e6cc64bed9b5d15dd0de69":{"kind":"REWORK","_number":1,"created":"2020-02-12 22:19:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/1"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"subject":"[WIP][Community goal] Add contributor and PTL guide","message":"[WIP][Community goal] Add contributor and PTL guide\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ddbf24a12c022e0f9e6cc64bed9b5d15dd0de69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ddbf24a12c022e0f9e6cc64bed9b5d15dd0de69"}]},"branch":"refs/heads/master"},"aa376ea9e19fd5699c0ec1edcb76dc1d7a5e73c1":{"kind":"REWORK","_number":2,"created":"2020-02-15 21:45:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/2"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-15 21:45:29.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa376ea9e19fd5699c0ec1edcb76dc1d7a5e73c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa376ea9e19fd5699c0ec1edcb76dc1d7a5e73c1"}]},"branch":"refs/heads/master"},"805ddbef0efeaf6ae2cdcbe2d2716b24fdb06c04":{"kind":"REWORK","_number":3,"created":"2020-02-15 22:11:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/3"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-14 22:47:52.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805ddbef0efeaf6ae2cdcbe2d2716b24fdb06c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/805ddbef0efeaf6ae2cdcbe2d2716b24fdb06c04"}]},"branch":"refs/heads/master"},"9e06fcabcd28ac603a025508484861882d3643bf":{"kind":"REWORK","_number":4,"created":"2020-02-20 12:11:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/4"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-20 02:45:20.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e06fcabcd28ac603a025508484861882d3643bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e06fcabcd28ac603a025508484861882d3643bf"}]},"branch":"refs/heads/master"},"9288b18ad85eb69b5040002ef52819594432f695":{"kind":"REWORK","_number":5,"created":"2020-02-21 15:28:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/5"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-21 15:23:07.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9288b18ad85eb69b5040002ef52819594432f695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9288b18ad85eb69b5040002ef52819594432f695"}]},"branch":"refs/heads/master"},"c3164544ed08bbf368072d56dca223e915500425":{"kind":"REWORK","_number":6,"created":"2020-03-04 21:17:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/6"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-02 02:33:52.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nThis patch also changes CONTRIBUTING.rst file in root directory\nof the repo to fit with new cookiecutter template proposed in\nthe community goal description.\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3164544ed08bbf368072d56dca223e915500425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3164544ed08bbf368072d56dca223e915500425"}]},"branch":"refs/heads/master"},"09569bcc6d48fc0b0ccad020f54149078fed6f96":{"kind":"REWORK","_number":7,"created":"2020-03-06 07:32:35.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/7"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-06 07:32:29.000000000","tz":0},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nThis patch also changes CONTRIBUTING.rst file in root directory\nof the repo to fit with new cookiecutter template proposed in\nthe community goal description.\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09569bcc6d48fc0b0ccad020f54149078fed6f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09569bcc6d48fc0b0ccad020f54149078fed6f96"}]},"branch":"refs/heads/master"},"98d2007180c2682385060e3097355ab87b0390fb":{"kind":"REWORK","_number":8,"created":"2020-03-06 15:49:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/8"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-02 14:47:57.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nThis patch also changes CONTRIBUTING.rst file in root directory\nof the repo to fit with new cookiecutter template proposed in\nthe community goal description.\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d2007180c2682385060e3097355ab87b0390fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98d2007180c2682385060e3097355ab87b0390fb"}]},"branch":"refs/heads/master"},"67cf3351ec7ffe959481b31348be9ea247c10fe5":{"kind":"REWORK","_number":9,"created":"2020-03-07 18:34:31.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/9"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-02 17:26:18.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nThis patch also changes CONTRIBUTING.rst file in root directory\nof the repo to fit with new cookiecutter template proposed in\nthe community goal description.\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67cf3351ec7ffe959481b31348be9ea247c10fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67cf3351ec7ffe959481b31348be9ea247c10fe5"}]},"branch":"refs/heads/master"},"36656ca31871dec7467d0523ae937d890e9b364b":{"kind":"REWORK","_number":10,"created":"2020-03-09 16:15:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/86/707486/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/707486/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/707486/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/707486/10"}}},"commit":{"parents":[{"commit":"a026b39617f4c2d29f2903909fb65f7adb327fa7","subject":"Merge \"Use sys.executable instead of \u0027python\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a026b39617f4c2d29f2903909fb65f7adb327fa7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-06 21:12:21.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-03-03 03:43:26.000000000","tz":60},"subject":"[Community goal] Add contributor and PTL guide","message":"[Community goal] Add contributor and PTL guide\n\nThis patch also changes CONTRIBUTING.rst file in root directory\nof the repo to fit with new cookiecutter template proposed in\nthe community goal description.\n\nChange-Id: I190afcff68c4a20f42aec34cb299bd862e7c7efd\nStory: #2007236\nTask: #38540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36656ca31871dec7467d0523ae937d890e9b364b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/36656ca31871dec7467d0523ae937d890e9b364b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
